)]}'
{"id":"openstack%2Fswift~822585","triplet_id":"openstack%2Fswift~master~I88b8cfd30292325e0870029058da6fb38026ae1a","project":"openstack/swift","branch":"master","topic":"sigs-in-logs_1_1_1","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-02-10 11:30:01.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-02-10 09:26:46.000000000","reason":"Alistair Coles replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-09 21:38:05.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I88b8cfd30292325e0870029058da6fb38026ae1a","subject":"Trim sensitive information in the logs (CVE-2017-8761)","status":"MERGED","created":"2021-12-22 05:27:40.000000000","updated":"2022-02-10 11:32:04.000000000","submitted":"2022-02-10 11:30:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":79,"unresolved_comment_count":15,"has_review_started":true,"submission_id":"822585-sigs-in-logs_1_1_1","meta_rev_id":"636bde4e89a6cd5de626f64081e19208e04a44c6","_number":822585,"virtual_id_number":822585,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 11:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-02-09 11:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-09 21:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-09 21:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-22 07:00:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 19:46:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-01-11 22:58:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-01-20 17:41:04.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"42125e5d12ca28730c1369f33e607a1549485184","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-22 05:27:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69ccbbbf4429ca689d6d43f0f1fd2c1bc52e204d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-22 05:30:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf896cc85e44d012f3871be56db6a186811453f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 07:00:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3548972f86914d4e88d3316252e2f904 : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/7650d9f197904e6a9ca0cfa5e7fc7582 : SUCCESS in 46m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62afae9e15e14cb69c4375725cd7b153 : SUCCESS in 36m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35b809b7f3d640c3b2173b0b52328387 : SUCCESS in 34m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8dd901c3f1564263a487a91fa9de4ecd : SUCCESS in 6m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/68dc3a5a5459466b88c7df7b68f9c917 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc04dbc86c18499f9d1bd2d3d5acf5e4 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74b54f9a14174fb2ac07c12c5841b9c2 : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59e4ac4988734f2894bde1e8562e9863 : TIMED_OUT in 1h 31m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23fbf402e59a46a3b301d6244342e772 : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/68148de9955e49a6b050522de3901436 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/426613e4c5954ee9b4a087c9c048c3a9 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c73a83250a52483ba3934301a0c264d8 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea663bb1191240ed9f860f49d10c9e00 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/781776c242a74a4290b6df2aaed31237 : SUCCESS in 18m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ea8cf5baebd54d1a8e02705cd4900622 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b44187f808e54089b24b1248c485ff99 : SUCCESS in 21m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/22c0665be5dc47aa9ddfb5bd71b6161a : SUCCESS in 51m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbb6aa39c3b3457081d6fcd9b52e79f7 : SUCCESS in 58m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b42a0fb83e04c749650283de78e91b4 : SUCCESS in 34m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/307b5760e83e4fbba823a8900ff823a1 : SUCCESS in 33m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b001624bd47a478baec7258b8866c039 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8533cd14464ab398537fa12bce7071 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/298d991428f04dc485900b99be4b3f6d : SUCCESS in 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb40fad371747b786b3d298160dd9eb : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e3d2489ebbf582ff80b4668765cd4ad803a657f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 07:12:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/712ba1be909b45fbba0e8e9465726038 : SUCCESS in 37m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/156a5d330c964f3da76270da2df7645b : SUCCESS in 38m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/624647c2c52d45aeb9241cbb5a111950 : SUCCESS in 1h 35m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21c56411628f4d9a8e76acc023edbf76 : SUCCESS in 39m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d59b1eb9e844b5c83524e0a558432e7 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5b8a251c900118fa1be1c43fced89802a55dbdf0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 19:46:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cceeb8e4d4a299527d26a421f04c4fadfbfdaef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 20:51:35.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"39da6b1be59121861aac217f6cbc51fa448d56cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:06:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9113d7a27d7f429fbaa4752e0e5a9520 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6059bec7006a4e6c86f889373619a5b0 : SUCCESS in 55m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86c5d3bec3be4dee96bbc9d780effecf : SUCCESS in 42m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d49f3379e4ca409d813310a9fecbf549 : SUCCESS in 1h 09m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d4766ae181b4c8a8e5f8ef107069917 : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8dc902f5a1e443888745d3813925e46c : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b0db97b3692746339e35d38f1232c261 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1366b731186041b08c96c6248b933d82 : SUCCESS in 15m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb4cc51dea1e4efb87e9b935b1a4e2ba : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5fb321705f6b4a2fb5917df575f2b4ec : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b6c6ab2269946859f2f7264fc2be592 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/55cd08da78c04e76b45c3625ae0cf225 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7715a6bd14da4f6d88b6ef70507aa4f1 : SUCCESS in 20m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f94b01d8b0074ffa8cd293416ffe4eb9 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f5a01553ee1e439b82b9ef81ebc66234 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cdbcca517c654e6ea5cf5a77cb9eca66 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a08a83e97af14a8dba2962f4b8140497 : SUCCESS in 23m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/782f0a85d7934176a45b753bf38b47bd : SUCCESS in 48m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/37d8b8db594943939ab3d1ec5368ff3c : SUCCESS in 57m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d77a7451213a441fb492e22a395597f0 : SUCCESS in 37m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eca78bc921d54d9f86b8a2227b0c7ec4 : SUCCESS in 40m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d90caeeaf454bc28031ab39f09dada4 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebf034690a14c54ad0be4a6bdfd651f : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/926e852db8a643428bc31acf4d287f79 : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a23f554987b4982a49bccf1a2472c0f : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a5d868049156dcc33bb85b15dba88a78a6f318c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 21:25:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14c81997cadf457798930c8734238e89 : SUCCESS in 42m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c9797f6b0fb4cbb8754abb0753f0876 : SUCCESS in 36m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3536260ad09f4571b833f2bde3ce69ef : SUCCESS in 1h 30m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/40f38488a3c6494ca442d0466c30048f : SUCCESS in 39m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/660235ec6d7845b9a7a7998ac6b2db44 : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"874cac0328e0b2b99538cedb6d474e9696f8df92","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 21:35:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f23eb43f9a433af6075b3cac501b243cd43bc5f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-10 19:32:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"708b11a1d0cb007ef83f59d700427c0a174398bb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-10 19:56:15.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5e32a2d2572fa17cef925e4fbcf48dfcce0a094f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:36:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b5a2a6169b43f68e9a1441a73d63f1 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/809b1b5f5800464d969f56ec1a26d819 : SUCCESS in 53m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/481e68822940479dbe171662438f1387 : SUCCESS in 53m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c6a544ebbf843a5a6e59e67e8db485e : SUCCESS in 40m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0fac992d9f444604b02521928e90fd57 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd8e3ce0a69b462b9b58a98328e10a8e : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/288f53a8eac548cdb598b9be805e7534 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff12bf2dcbc3418791157312ae88b742 : SUCCESS in 23m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d514e1b9dd784810924b04219ef251f5 : SUCCESS in 16m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d4c281ba26e4398b27225ea4a9f9424 : SUCCESS in 29m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f28b32d965a34aa68881aa089fd97621 : SUCCESS in 22m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ace0863b8133453790a232f514f7841c : SUCCESS in 28m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25878b1f1f9d41e8adc8276cd0f6e7df : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/72c99f6ead134c469cf86baea3f0dfbe : SUCCESS in 20m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/001c6fa00550435c84e5981bd233c63c : SUCCESS in 35m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4341000aa3654edfac89a92e4c4229c4 : SUCCESS in 9m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/833d0a46b53c41bfacb20f9597871fc9 : SUCCESS in 23m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f2b9d6b33dff425582f0da6481bb17a2 : SUCCESS in 56m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/03a370272e88400ea5bb7c44de71a06a : TIMED_OUT in 2h 01m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1b2afb7096941b99385a5afd0bb8e46 : SUCCESS in 43m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db24222b61e04e6486ee7499b7521ca7 : SUCCESS in 39m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc38174a5154d3f8934f2153cb6ec70 : SUCCESS in 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f4b8ee5c3344b49c545e3d0b41a844 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ef2d4bf3e2547f298b794480bbffa25 : SUCCESS in 29m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e5d71ddec98409d9ae178ddd7340650 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":2},{"id":"14513e5e71fff140a1c500afd5a11d7d9f94c4c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:36:17.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ca965255cb24566b25751eec31b408f : SUCCESS in 32m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f25d9ce814b1408e9d07b9e42c32c352 : SUCCESS in 39m 09s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a1f01862a4d14203994f2ded0f988fa7 : FAILURE in 2h 01m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/252a60af871b40d8aaa263986ed38962 : SUCCESS in 1h 06m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5759a5a5d81144d9838872174d9e44b2 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9633c54bc40b5996d1e5ede69a4ba256ca0c9bba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-11 22:56:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8c1a53105b167edaf33afc4d262524cd8b06155","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-11 22:58:07.000000000","message":"Patch Set 3: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2482d91d2bb38541ba18f21ad5ddec0bd935a3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 00:01:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c677aeb62e4019ada0704da146cce4 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9c609a34ecf54a0391e5b552e5c6b4ab : SUCCESS in 55m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38bada723c044c9c860792c88647277d : SUCCESS in 50m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22af571a3e414b1ca7ce17ae793ba465 : SUCCESS in 53m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfb7fff3e11444b0b36ab0f8d4c03e65 : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/77159068576e44b5b1b1aeb6ac36d208 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef7568a568f343c884db58553c4ec540 : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8cd1263162443e9b23f1903c381882 : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d360161e5d74177a797b1b7af20fa8e : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d741294f6ef4f128e951c9369f66aa3 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31cce6a1a1ab4013a22a0a0e739a70ee : SUCCESS in 28m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c88a961fb6bd4b50ba1ceed612b2bbec : SUCCESS in 28m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4557a2a191c24cf6b3b18275eddd80ad : SUCCESS in 25m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/97694c86021e499ba6b1b4165a7e14f2 : SUCCESS in 18m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/384b62c211144542b5dd7b7e5b18c138 : SUCCESS in 28m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08c1e135fd0141d7b085c593d2cd982c : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7048fd1459d04bed84de5ac82aaf4d74 : SUCCESS in 24m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f51776ce5e8a485a9ad659cf917a7dfd : SUCCESS in 1h 03m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4af7716a69644251812f2e0fbb314652 : SUCCESS in 56m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac309009753c4c8890fcb6fdcdca10d6 : SUCCESS in 40m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79f157d6b3324ddbb446a1fcbfbe137a : SUCCESS in 47m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6c06d07bca94451b0d68968495b7dac : SUCCESS in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61fc31b71034bcb967515c307b870d5 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b1fa1c2ed5945dda71ff40d5e220ff7 : SUCCESS in 33m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7513d51e62214046b03a2e90652caf63 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1deb17fdde06718af514bc4ca460ecd2ae7f43d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 01:29:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d90903e38eb40a486892b9f85da5a5f : SUCCESS in 45m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/260b291a2989427a9d52543c3dd88bfd : SUCCESS in 45m 31s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/708a37c1d81047f7ae910201bb1b88a7 : SUCCESS in 2h 19m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50a0c5b5f9e41b683dcc38b4a5237ac : SUCCESS in 47m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf10c59cfa134726b9e59bc6342a2025 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":3},{"id":"908976d5d9cb5c1afa649885577d2812d1f793a1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-12 05:22:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca8fd9042e2b04c6413dad2d6f30b8337d59514a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-20 17:41:04.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"edc7154ef2f21ea3b55f46445a87df0b5c0d8612","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-21 00:22:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f23559bc300fa4f9aef9a530af597d775e6d9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-28 01:08:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ee2018153f73787ab5e0477bd2b3b046d2c3f385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:44:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee7f078a88e413baaf3bc2fc9099aad : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e174daae1aa54bca912c2a3b21d87663 : SUCCESS in 58m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a678a75f84f4931aa019713084936e3 : SUCCESS in 42m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94a85f2f304d48039323ec9d44f9a0c8 : SUCCESS in 57m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ddaca0b39ed642ffa5569876e9d798d1 : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a03fb730f99f4cd18991f779a6651184 : SUCCESS in 24m 15s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0ccf792982424efda3b44af1dc9201f0 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cce765f207f49ad8d45a85f99d9f913 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4abc94b6b3ee4dfba0f6c1d62bf12f80 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78af80942a5245089467185bc7484571 : SUCCESS in 21m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5963ce5e8f8740bb83c33f2ccaace64c : TIMED_OUT in 1h 31m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5c1677024f746ccb498d8b7eb7bb3f0 : SUCCESS in 19m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f284da052f349938af49d704a270c37 : SUCCESS in 24m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f84e9e0c5e8443f7b9c9fed113a71383 : SUCCESS in 21m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89671fbc488e4a4aac19951e779eef66 : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5b38aa31dd444e4b863e7c40b076ca0e : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa80e4c706264f31a055b2b4646bd8e2 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3eb3eb19408485abeed1d9084c6d3b0 : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9380d467a8e417c8e6b24597fd17e0e : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b73f4e823ebb46bc902f0edaab1bfcef : SUCCESS in 23m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e2522594c4e4ab8812182e165b1a05d : SUCCESS in 54m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f4b989aba5a42a994831e492a5e8980 : SUCCESS in 55m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4fcbc5cce0440f49482f35dbfebec38 : SUCCESS in 41m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d6777d0f80f407f92e34311db4e7e50 : SUCCESS in 39m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73ca693def4d4afbb44ce9eefb86f28b : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81a69a131b64e9db0e84d1971133e7a : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac92beb47d5940f397506485090e838d : SUCCESS in 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3168cafa48c642f3bf5dc764fca1e720 : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":4},{"id":"3da28d17ea843cea04947dd72debdd3a9b82b740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 04:17:46.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdeb81a5a19442d6963d20bae1c6d038 : SUCCESS in 36m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2c01c1ae5fd4aaa9623de5e50b5ffc5 : SUCCESS in 36m 08s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/08bf0a0925af43f18bd4cc78f1c6bae4 : TIMED_OUT in 3h 01m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea39dffbb7434d4b83620344bf8a6879 : SUCCESS in 36m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c000eb0bd66243969bdaf1fa90bfb18f : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":4},{"id":"f0e36d1f725daf83ffde6f6f1bb2b4c7a7d1fc0f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-02 21:49:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"26331cdf05fc9b60e4a5c90cb34cd75d7c75208d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-03 06:03:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e601dc0a101939455deb5c74c35e4d0de91e81ce","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-03 06:06:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e393a110d13a3a03f3db2a1baf4c06e5d18e3bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 07:07:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fa6bb1f8d594a758c18ff006646fc9f : SUCCESS in 46m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57a46208737e45c0bfe9c57e6d0e6d0b : SUCCESS in 47m 24s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/aaf2d899b0314dd6bad89ddca09cc87d : RETRY_LIMIT in 3m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca6a3dd3ab4e40d5a6fab7e0a110bb8f : SUCCESS in 48m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cb63a198b824842862c67873dc1f0a4 : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":5},{"id":"ac5168cc74f4d441e0067dfd8176099132e97620","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 07:16:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c1eb5d02d54e789af6f0cf69e44964 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/a9e0418658ec49f79ac3b4760d5f7cbd : SUCCESS in 1h 05m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f91196da65c4382b23541b1efc28712 : SUCCESS in 1h 02m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64a2245da10b46438e7d896ecd9d6f05 : SUCCESS in 49m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3755fb31cd4a4925bb643aad5ac536da : RETRY_LIMIT in 4m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7895226839d044d9b244f37e79dcbc5b : RETRY_LIMIT in 5m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7dc581d050194f17a700436342df83e5 : RETRY_LIMIT in 6m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/856b1b08739548d3882a6a9198efb155 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a567c14772fd4bf38fac509da9cc4d04 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a46444aec1b4900b896f1a624b4d4ea : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c17442a76e2e4b4b9ea8029a2f906963 : SUCCESS in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4ee6d52db5445dc8c69ba9c7a701868 : SUCCESS in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56e0762f97174ec7b403f5a771d1219a : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e5cbaf9146d46d7aa3bd4e288384ab1 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49cad629ccdd4087ab772b695328f211 : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e6be8756d2d24a4c9a5e0ae84f56974b : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27b82d53b00242a9b00077286648cd3e : SUCCESS in 19m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d33d136121214184bdcd2e403398c499 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02149b2bb7ed4823a80b11a23d7f274b : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f4fea149cf434f73a5a5f1a861da64f7 : SUCCESS in 22m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1e5aea985e24ac2be91bbed1da6839c : SUCCESS in 57m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1317b0f0f60941f293707fb901ed862e : RETRY_LIMIT in 3m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b07dc7c5910b499fafda730060491ea5 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02bb520b7f4a4b709ebb4c6c8df983a7 : SUCCESS in 46m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00eaea9a89674638aa3902fdeeb4ccac : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5762db9bc62b43e0ae3d7c76f1dcd9bf : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3478019655ce46d2aed14b2c79d688e6 : SUCCESS in 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cec60b7303a47528a3c3e4a72dee717 : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":5},{"id":"954329ef2d1b57ac68fe1d7198205d53460bc0c3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 15:20:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"be747416bc81c2f10d3dc7789728eeb073df9e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 15:28:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5251ea6b70e024a49529d576d1c8890882b328f6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 15:30:51.000000000","message":"Patch Set 6: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b4dfd0ad5faea06469ba63d14970fae3feb87864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 17:26:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd43f3167ea141cd9f8fea793f6769bf : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/8b8fa5b6da4b41578a6d8db903303917 : SUCCESS in 1h 00m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/15f0b8d6f9f843538ef2103ecc4b247d : SUCCESS in 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73812b3a47a64b529f8d0486bc54d5b2 : SUCCESS in 49m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e3be33f416354dbeac367581cecf9300 : RETRY_LIMIT in 7m 12s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/62f2a6f425954b309b9600ebf0f79fe8 : RETRY_LIMIT in 7m 05s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/482c8761c20c496483a59d50737ca250 : RETRY_LIMIT in 7m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c51ca63ca5c840e5ade845745daa4cd8 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edba08e42fb1484ca04dd2ddbb9f7049 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf52d87226fd4c68ab0936c725fa9c2f : FAILURE in 21m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c3e5bca13eb4d4883a38714bcb38306 : FAILURE in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/597d723507924b45b856f514f0f9648f : FAILURE in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f345707a4e7480886755d9ec82088b6 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b5fa546c13347959eda1db3961d938e : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90db0f8fa2024bcbafb7e59c94901fab : SUCCESS in 23m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3a0153e4fe6b40d4b930d379e6eb9e5d : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/252c41740a054cd998804cf113999beb : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9aefd22c1ff34a2091c7d772f5b63e05 : SUCCESS in 19m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11e4acfe94d14bada331bddff1521f51 : SUCCESS in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40308d187e3f4646b227f2a4c146dc47 : SUCCESS in 18m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/916fbfe21ea944c085bcf504dc453599 : SUCCESS in 57m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4921ca949dc24fc1a38b62ff2ca74454 : RETRY_LIMIT in 4m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b67f531091c4408b740ec3fe9885b41 : SUCCESS in 46m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d265d73116844d6c92930983bb360b83 : SUCCESS in 44m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d21e14096a174ea7a632dfc1b5d2af38 : FAILURE in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d7e82e9fab4b76bb56bc9389eacb7e : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58e5aded837d4c3ea2e5ea3c3c157f1a : SUCCESS in 24m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94abd1a231cb485bb97f4051160fe484 : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":6},{"id":"d76e2943d8fea919585d6d30d502d44c90c45f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 18:42:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7ee9ad9d101515cac60dd4583400cf2008a171e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 18:42:38.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"cf5ac3fe938fafaf482848615dfa71b5899801fb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 18:48:08.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a5933a3ea6984099c442b623924743f3bbee4408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-03 18:56:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b523300b0df14d1baa2e7a0a216dbaf24c58500f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:35:09.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59878e7341df4578ac6606ff432af4f0 : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/df7d95d4f9eb4c8cba2a08d877f141f6 : SUCCESS in 1h 04m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfcd781a53d14a0f85c4d12ae15cf6dc : SUCCESS in 53m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ebfcf61efdc444a91ae493138c2ee6a : SUCCESS in 48m 26s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/50cca1b5e3d54d8a83e557706dec0faa : RETRY_LIMIT in 6m 22s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1dccdbe3d8764bfb9e1a4e4cbe9d4d6c : RETRY_LIMIT in 4m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/387d755629754774bad55fbd01728517 : RETRY_LIMIT in 5m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3324fe7838764f1bb66c74701619869c : SUCCESS in 10m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36db11dc1bc8499a95200ebe8e40fd0d : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6c924021b9644049cdd9cb46d1cb3f4 : SUCCESS in 21m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/052a4310922c41f49d2f1f5e760ca87c : SUCCESS in 20m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fcad72300504db1a39402772541af92 : TIMED_OUT in 1h 32m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81eef0bc2e5d4d1bbbaba980aa99ec85 : SUCCESS in 22m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2325a41758045629958e645088d611d : SUCCESS in 21m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a8302d6a7684bd9adede1e37ff9d7f3 : SUCCESS in 22m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c004278f823d4e9ea0dd1bd2dfb77d90 : SUCCESS in 22m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da32fb88eab9405b8d7bf2119222b88a : SUCCESS in 22m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38407aa68a6640d6ae373850df1da562 : SUCCESS in 19m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f884b0bf49054a6eaa6b3d6107f205f2 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8b6c002564645dd813cd3fbd48f39fc : SUCCESS in 26m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5ec97e1e07d543bab3f2675a2644c29b : SUCCESS in 47m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f4acfed059143789caf358fbf4ff9d5 : RETRY_LIMIT in 3m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a4ec2036be84868812c026ab22d4565 : SUCCESS in 32m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6ba75d120774abbb2006ea7d28b874d : SUCCESS in 42m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e111701702844abbb159ec34ef34f41 : SUCCESS in 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/413d662aa5e84d7b9abb173071ba8ebe : SUCCESS in 7m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/780f8b216a6b4cb8ac357075ccd8ee3c : SUCCESS in 24m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab8ec023ddae4ef68235b9d3de5678a1 : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":9},{"id":"2d651725f17a7742347d3f65ed14d3439e9f4b56","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 21:33:27.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1a85eb093cc4e68ad293bdd62ed96cbb43fb9d04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 22:33:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479269c76bb94c678c1d0e8b130baeeb : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e0b6083724524da695f7d1e613a69ddf : SUCCESS in 55m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d49855e0d96e437494a005ae5ad1d6c5 : SUCCESS in 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42a6ce6225be4b28adb80b99fe42691d : SUCCESS in 43m 51s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b1b49edf80d24087bed69b52a1c642e1 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8562ca0a09d34f1e8c15ea3e745f2aeb : SUCCESS in 24m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6b9c65aeb6594e43977112ab15a93ab3 : SUCCESS in 23m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/512ce2911bb7474a94a7d723b177b35e : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b54db400606841b28af2cd2619c6c911 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/716e805758de43fd85285c494009471b : SUCCESS in 17m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/faae3d5a39a54b7394f46a562678ee32 : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/651f84680cef42219c61f2e6305935e7 : SUCCESS in 19m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a48b472c4cb042bf898d2a99eaaa28ba : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65d62a84ffd1444bb40ce6485f578465 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8e2f70dd9bd4d18b987baccd5218d17 : SUCCESS in 22m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/41a513a60fcc4121b3b2ffa51448d713 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/28cc3407e3414aa0898156337a9b279f : SUCCESS in 19m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f08f2ec0d7c4e7387c8df5ae5459d6e : SUCCESS in 21m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23e04732b7b347da81ad432f9bda9e25 : SUCCESS in 8m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bcc5576ad60949b6932b6a5f59459c6c : SUCCESS in 25m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e6a587dc1a914416912b602d4f901214 : SUCCESS in 51m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/478a10b2190d4dddafb51c90a27aea60 : SUCCESS in 45m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e855e7274b248e5a3cb79c3acfbdbd8 : SUCCESS in 30m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a398582f2b2b4979b223aa966a22ff24 : SUCCESS in 30m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/056fcbd38d5649a882dc428c2afddc2e : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a631729a504b8491d4974435642a18 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d58d29978df041baa276efc341d41f07 : SUCCESS in 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2829bb6302d840ffb26250c0c92ef60e : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":9},{"id":"7c3994aa5394ef5fedb9256738724a38961fe30a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:46:29.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26f8c8bd37c44f5e91cb780a4a329d44 : SUCCESS in 1h 29m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0406440742c4540a5fd341438393703 : SUCCESS in 1h 12m 58s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5ee97e3eb98b48c49472ac110ed9fdfc : RETRY_LIMIT in 4m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/26e3a30318004070973d2817b0d2ca50 : SUCCESS in 1h 27m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec58deb512c94bfea4a68bf63cb937d5 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":9},{"id":"e16297ed07372156e9a5f319d126c71b9ab563a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 06:17:12.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4d58883297675932321bff846cdfdbf0ebfa1f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 20:19:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4da04ac532a19f78b3aa2474584a32ad5a010255","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 20:39:06.000000000","message":"Patch Set 10: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4b0261d7d964645b537f09f5a7f8776a0dcf390a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 21:28:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d394598b2bb4d5ba0037acddf9678f9 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/348469fc5e18456780e4e6dae2028a02 : FAILURE in 35m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f62394532dea4d8a8de1cf7246cc93ac : SUCCESS in 49m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/541d2500e9f940b3a8f418885379567e : SUCCESS in 51m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9bb0c95f1cff45c6998488272ec5b817 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f23d3ba9cb484344a34b696529552bd3 : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0cf562f5befc4e2bae7a43f9abc4d70c : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58d0ed71386c4a6fbe5e93cadeed3bab : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2bd155f77bb4bc0b6f4de31d45cb252 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9df625bc9ab42949b3012889878496b : SUCCESS in 15m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/625a01900cdf42cbbbe0977d1d18242d : SUCCESS in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d852f680d0214498b0feb61ab91c0c7c : SUCCESS in 15m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/944727cf062f4f3192fc2da0095966fd : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f9f31ed9d2d54e84b3597e436736acc1 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1bb678ade164872bffe03c1196cb2c7 : SUCCESS in 20m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/689e2745849c4e91abc6e5ccc3922aa4 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/499099b5e6ca4c3ab1c849c16926d979 : SUCCESS in 22m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/073e9bdb713a4c7f83a9b62db7567fca : SUCCESS in 16m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b824e1bde6a44742beec9f7e99479342 : SUCCESS in 7m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/990d09ebee134a80a5b42aa2d2480f98 : FAILURE in 20m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/38f48148e7374603b72c92a5981fe8f5 : SUCCESS in 48m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/664c26f224354b6ea85fe6c24a663590 : SUCCESS in 1h 03m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1e8c711ac384adc94964931d47f9c4c : SUCCESS in 39m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/030757a14fa544739929cddcf8eb6157 : SUCCESS in 30m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f9b365ac1b4ae9a9ad3a04c3e46473 : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e409aa53e3e4b52b1ae0267446185ba : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e95fa73e8fb74569aba6bfc90913a687 : SUCCESS in 27m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5011070970574f3b959a99640bbe4c98 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":10},{"id":"74b2e052112e72fa453e3e419f684af8f3335c69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 00:45:09.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48795bfd51e74675ad95f3d70ee8a2fd : SUCCESS in 56m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70d61678758b44a28494bbc0f4e8a7d9 : SUCCESS in 54m 21s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/81dc3ad7be124f9ba8c481e4e578760b : TIMED_OUT in 3h 01m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9f96723b6ff4e94be4471a2784b816b : SUCCESS in 1h 16m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d9d6db5eb294992ad856eea49681988 : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":10},{"id":"fdc0cb1f5ac50c37e99736eb9bbad4a3ce1aec5f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 14:38:16.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"0c0562b92ff0bc58c0eaf27854483a654aac83af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-07 17:52:35.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"476e72920c5664f96e49b019567d966b3a912246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:10:31.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04caacc203941fcb4b4c672732308ae : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8ae84af243fb44dab81f8753f3e7e0ce : SUCCESS in 1h 11m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/463804af8c3645299fd9077a601232d5 : SUCCESS in 49m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/933d16580c4545f699fedc4a5fe2d316 : SUCCESS in 47m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/85039de547ce4c1cb4214d67af3965d1 : SUCCESS in 23m 49s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/332e40fd203a48f89b55665b06ec9341 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f262dbb78c12451ab4cb679a52440cd9 : SUCCESS in 24m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25e51af591294c5e81c3ef5cf4556573 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1739c7586375402dacbad5fbd3a9e060 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a638bb856084be3873525a7f88b3216 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d210be1a85c44ea8935db919b082f14d : SUCCESS in 15m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92afc1ca286049ebadc96614a5ceb277 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3dcc6d78f34f456b9a8fab1142039d16 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7e26c1f0d7a46e9b97fd95cffc5ee5f : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf750ec51ec240829eca91894583f1a1 : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc8b610c9ac94ab3bebc7c3a90ccca62 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/622d374f4b194166b1c7ece73dd0289d : SUCCESS in 19m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4074a4d1b56e4d9c83a69a3110e1027b : SUCCESS in 24m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb89709643de470e8795360830a07631 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e33159e380a94ac09fc5b358ee009fcf : SUCCESS in 20m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c93461b42122466ab9f46624971e72b9 : SUCCESS in 52m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/39ce4face9bc4a5c9f8a8e74045771f3 : SUCCESS in 1h 03m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68d22eb03c8d4fa0900f7c35e055b04c : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cddd3e798f24b84a7800c4e4d0d2b21 : SUCCESS in 40m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef7b52e448f44199ba6563433b98d23d : SUCCESS in 24m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc26150eef6424a9cd192dfb6acb9a5 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4538de7da474f06a1c5c88cd872fa1e : FAILURE in 10m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2557728e295149268f5e33060bd99270 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":10},{"id":"c4a2b7e50dafba538ad27068d4e6b7e050a6ffe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 20:43:32.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0f4af648f924bfaa5da9c4a3c36154f : SUCCESS in 49m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e90732e0a2848ce92744645eba65df7 : SUCCESS in 1h 03m 42s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9720d391349c436fa5228d4cb213cc67 : FAILURE in 2h 32m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4abe2dfef09c49ea946d3cb03a50aa48 : SUCCESS in 1h 00m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6bb23da2d404ecb9761f0c282154251 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":10},{"id":"99e1139531782fc2353c08351c620e8d4c179f5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-07 23:45:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2c3b45b6c7f03ec079fc1d113ae09c9ab655efab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 00:50:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5caf6f7abe47269e5a223af23ed8d2 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/35ef594cdc834d189ea98a0d29228220 : SUCCESS in 57m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6f1736b42fb494fad9ea6d2247c5cf4 : SUCCESS in 52m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3cf8d9d39194b4a8dfca93b12496c9a : SUCCESS in 52m 15s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/53c2b7fd937a493e97fcd9ab5998b0ad : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8cf664b48cc54834bc974a50950ca550 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/abf08a899c8340cfbeeac267b0955441 : SUCCESS in 19m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9787c6f0766641a99cf8879361b8e364 : SUCCESS in 7m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c33cb264e8043eda3b8dc2cb8ec2631 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/310ba5ffee784c5d8e685be7cd721433 : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9abf8c3fa34c40faa10421b0bf5a7b39 : SUCCESS in 14m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1a0b3327be7e48829f894d3da78f8d96 : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1201a6acde1a4018be955117cc08250d : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42807162c1594fadae8138da3fc91f44 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b3df9affe234eba867cbb742f19fd8f : SUCCESS in 16m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8585c1bb9ca740ee9197671258ffa71e : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6e8dec624f084515a45091c0b98fc013 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a2690056ccd34bbdb24e5182d5005efd : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e2c4e13bf314b16a43612b1b591b4af : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97b45b5b912f4b788a5dba694193b2e8 : SUCCESS in 18m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e146fef88d447fe9f50b6fef0b1265a : SUCCESS in 53m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbc69e8d999940ca9ddfce9295916c0a : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8970cdb52c484ccc95edce0cc205afff : SUCCESS in 43m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c966e1ca28d467da4f2725f92dec853 : SUCCESS in 28m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea0e2e97e8e94d469b961389e9e69b64 : SUCCESS in 15m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ce29869d8640018cffc48cff954771 : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e58a1e76a8647bd97f5740bbd23ab4c : SUCCESS in 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87fc0dc594d64c618f88e7df11ca875a : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":10},{"id":"ba3d468ac287f8468644ab84728b15761436a2ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 02:50:50.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58ed6224910a49e1b7d94090f9f1a435 : SUCCESS in 51m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd01a89f1c574ad99f320cbef56d7b8d : SUCCESS in 55m 34s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/473947fbba8547cf8dccf5462dc1153d : TIMED_OUT in 3h 02m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c730afaede12484982ca620f3135c25e : SUCCESS in 1h 02m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e39c0208701457ab8fca9108d191307 : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":10},{"id":"eeab4e69a237af880e09d4ce987c67bb22d6bd4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-08 20:59:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"555f83ed29feb1e2995c635d259a8fcfabb6f0b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 22:54:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8684317078497cabff145e07a18c92 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/4d5aea395a3d42efbd5cefba453b7f61 : SUCCESS in 50m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70832711a2ab424097a302d4bd8d382f : SUCCESS in 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/570e98c2aac14d99baae2dda6bb77dc5 : SUCCESS in 46m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c6b5392a67d64cdb88ea7ec43421a0f4 : SUCCESS in 24m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/005a9b35612e4f46adf47dd0a0b14307 : SUCCESS in 22m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/30e6e1061ce240c99216abcabcd00749 : SUCCESS in 26m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35cbceb416a74c738690bbcd32cf27d5 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2f49aebee4b457dbec4121cfe222761 : SUCCESS in 8m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b121cad8a4f942da99f9b68ebf7c5084 : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e86e3ad66645599b855ec640fdaccd : TIMED_OUT in 1h 32m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5590430423e74845b35bbf04026c06fb : SUCCESS in 19m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b4b12226fe9e4cd1be49ad553ace0389 : SUCCESS in 22m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a875e1f0e961475a80c55836f599d6c6 : SUCCESS in 21m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cef29bd9a0b243f8a63c67d158f36450 : SUCCESS in 22m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6854b6af676543e7a5ef976a719f250f : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ce4d04a0fa04b6491ba9d30d6d2fb53 : SUCCESS in 21m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ecd24e498964977afda91356db46cb6 : SUCCESS in 21m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f631af975bc648bf911f06faf2213cbd : SUCCESS in 7m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/634776d9edda4a269d66e340b9bb3f73 : SUCCESS in 25m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3b0b6015cc7347769423ffc3123b6f12 : SUCCESS in 59m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccb4b40c9d034d1fa7d3dfcdcbdc263d : SUCCESS in 1h 04m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/daf0b71102d8497a938c5315f01ec129 : SUCCESS in 44m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f57112b0e604632967e10a500231e1c : SUCCESS in 40m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cf51ff57ceb42c981f038595d6b64e6 : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd3b16b31be4f6cbe945624af72f05d : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8560e0477b60433384956a075ef48f31 : SUCCESS in 27m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff5dc3ec38244e7f9be8dca82e5073b3 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":11},{"id":"69cff281e2a244430f6c453ab982e655f1b62bda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 00:59:22.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab5e3dbb6d5c4c0db5b586ee7be9bcab : SUCCESS in 1h 27m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc3a2111bb52498e90e25a85960c11fd : SUCCESS in 1h 02m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8d42ab7870d1484884e89a5dd460e947 : TIMED_OUT in 3h 01m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5a9789f95e14da593d0a1a3530d2bc2 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba9f19ab4bcb446ba9d739c20fe55336 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":11},{"id":"eb50e5148fc08ff46ae2b1ef9ffc65d979064085","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-09 09:59:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3d21479cad6b03315d8335e2c0191b99ca578258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-09 10:53:56.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1884d3fd24671e5d3aad5d629a705fadf1f0a539","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-09 10:54:09.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8b46d009c01d894f6f2b1f3f253922520ba0a671","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-09 10:56:54.000000000","message":"Patch Set 12: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c8426bc72eb8e27a666f259138873b3fdfba5b16","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-09 11:04:54.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"e8123cb8fa9031d6e4ff118b19ddb3d61769107a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 12:00:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a41f858826945dda3a9518011a22899 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/62333fe892c749ecb8f805b06c135cc0 : SUCCESS in 57m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83834f3d24ab4019bc52e36037ef00f8 : SUCCESS in 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d5718c8a8cc4a1192cbca96b4ab7760 : SUCCESS in 48m 07s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/82edb799034f42a09e3131c2be13fedb : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d9fcb4583b1b4491a87eea6a7f2dc56a : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0dffc7ed66934ab99aaa21fd5897f6be : SUCCESS in 22m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b334990c5bd46668ebc0a0b10ecbd67 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7daf335ad9284aa398bd468ca6a58a31 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/baab897953264a5fbeaddf26806dbc8e : SUCCESS in 16m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28e3971522d84f468236abefe5549d91 : SUCCESS in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50d121cabb584c7d9f2bdb0e1a731b8e : SUCCESS in 16m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b682de913bd24e87a30baab8e5890fc3 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b2e627f5ee4407fbc7a5c0d9e55f809 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7e4f1f9807048f1b37d58b509383388 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cf74ef97362d4b1ab29b79e24c90902d : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2936325825234efb92f6ca6373d0c435 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6ce5e0979b9f414b80adad9e2b541d7f : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/072a1f0cb1f64deeaa764554f718b97c : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7f6b5012a8a49c3bcdc190fc3620e4f : SUCCESS in 20m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4efbeb15b2474d6d945af2e8a4f90221 : SUCCESS in 49m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa7bd17b2cfb48939a7b2f85a00059e0 : SUCCESS in 52m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f382a17e3664b64be0931cd0b4eaee7 : SUCCESS in 40m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48f95f808c04449c8c048613a1c26c1a : SUCCESS in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a6350d33c5454d91c056e04e8e9958 : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/281bb2329ed74114a760eb84f08e89b5 : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9b83297ecab460dbd5e28f2bbf49c45 : SUCCESS in 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7783ca660434f47930803eb4e8b8f73 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":12},{"id":"1d68b42f16444b30b9aef24d81cdc0f6f690c7ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 14:04:55.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7593c8137594f7d951640fc424f30e5 : TIMED_OUT in 1h 31m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f45df6f0aed94b01a06844611513dd40 : SUCCESS in 49m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ac2ddaf63c8446e7bc22d51d601afd62 : TIMED_OUT in 3h 02m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/78f3b9e7f26449019d165c220e47d57a : SUCCESS in 1h 04m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c822f9e3b564efe8ccddcba64588d55 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":12},{"id":"951b4d0edf7a0186d8d11e9afc81e2af65ff0454","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-09 21:38:05.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6cbd69998280991679ed5ad39df1aeb4f7f51aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 22:55:47.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4ac1ce0f3b47caa6d32e9fe446945f : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/77284bc7b6e444db9bda9b88281a610f : SUCCESS in 1h 07m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/760a1928236e4fa086d5249459edca78 : SUCCESS in 52m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b74e45a554b9452ea310a9a29dfbdf5b : SUCCESS in 51m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/be31e76224ea4a57a14f9cc1526b5f6e : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/52250b6e758042bfbb53858441b7a748 : SUCCESS in 25m 30s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7e0c4c1bc28a41449f41da7e9c390b1d : SUCCESS in 24m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ff4386d6b404d9eb2288c2bfaff492b : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c78a749794bd49ddb7cd06ec8aced885 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2146b8c17abd4e188adf8980bd817b1b : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0970a0e09fb845ab99d6e9c5755e3f10 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f8466ca342874c62acfa95717b6074ef : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/74b8f93b08004f0488fba2197be912c7 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98cc4ed593274b5f84c6eeb0ccd27399 : SUCCESS in 21m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e67e2ec38f648548b5b3d351ca53b84 : SUCCESS in 22m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f46688bfae1c46a7a284b6c3a872deba : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6acf9000e59740b3a9bb4e6ed2088981 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d517d8be3ad249d4bc10f48bbf5b6d19 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3788aadfbf5f4fcc857d86dc53f2a189 : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87b73451617f409a9a52837c8d6da2a7 : SUCCESS in 23m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fcd26c0e0a91432c8cdb67a4780b72f7 : SUCCESS in 49m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2bf2c5351cb4ed592ae57b14858a028 : SUCCESS in 53m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0340dbb5b894d23bc8378b23efbb14f : SUCCESS in 44m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd7377f8a56840ef94ee730bc4e5834b : SUCCESS in 39m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ec50547cd6479887ea405aaffbe51d : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6fe65254e854216a8203161ae1d5be6 : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/314cd0e856324897b979abc6e6e22857 : FAILURE in 12m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d36baab876e456782ea289cc0b5c197 : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":12},{"id":"2420bbe188f260d5d28fc741577555d291d9c82f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 00:47:14.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3793b324f8a5f546e476f641b4c2feddc14858fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:01:25.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a2d19ea8834898bae7cc3f54f9312d : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3efcb5e3780d46ec87e706a95277bdc8 : SUCCESS in 1h 01m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/600b9c0b73434544812ac7b8183c8b60 : SUCCESS in 59m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40e687497ca24330a12f3dc8e0d4f52d : SUCCESS in 39m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6d4ac14c760a4720baed13db5a12b09c : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3d5ab1cb870443088f645aceb53cd5e5 : SUCCESS in 22m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d28ec998c5eb40d9b517ece3cb5c943c : SUCCESS in 26m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9493cfc0267b4ce3818a24308f5f8285 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b178b76a2306460ab687c7fd53554d39 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/289d9d1bf489471485a4a394502f5fdd : SUCCESS in 18m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7d891ecbcfc4c50a727791e53478607 : SUCCESS in 17m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2fbb538f95d64b66aed1627a54a25669 : SUCCESS in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08d8264dc7de48ae93b141f6389d009a : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0e25f39cd9c4dafafdd3aa9b7692f2c : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/416e0e8c5b5c4f47a7cf22742f418415 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c8aeea52628d434e8b8c3a0d08568253 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/97ad04cbc4a44910b7b4545f08018f2b : SUCCESS in 17m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/af22f7156a034e199532b8f814dc2c9c : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75c51afe1c014fc9b4808026f41d24e4 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f73f04d58824b709f6e780f46f3c1f1 : SUCCESS in 25m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea327229cdd3447f8d3548a256805859 : SUCCESS in 55m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/71659377fc7a485fb55d6a1856c136ad : SUCCESS in 1h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c623afaac3804751bbee0d32299bcfd7 : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb26c24867064d53ba1e5d13f30c1d44 : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00658918d2ea4772821cbbb0da36b9db : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5b2177a53646cd975eb7dfd6789634 : SUCCESS in 9m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59232f968faf41d89b37635cde2baaf1 : SUCCESS in 22m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db0d206cd592463e95a0a6724147d68d : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":12},{"id":"10973fdb0d8b6bd19249b2cfc8870c187dba6f43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:19:06.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"057b349beed32d7a8dbee16d9968035c9a3f5d4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 03:54:27.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49497340cfd9429097cf00a81ec97bc6 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e3e66eec6c8f4098b9d2a7fac859cd8f : SUCCESS in 49m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1de51eb4ec9643c0bb439fbab94f58b9 : SUCCESS in 50m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad76f19047664fe58f6d90c3140caa11 : SUCCESS in 56m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0c8189e58cf347dabfd7b69b326d709b : SUCCESS in 13m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0a0378f515524c248f98a77c335c05dd : SUCCESS in 9m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49ac32ddeb8b47408cb7a542ce262afc : SUCCESS in 19m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d50bfb6207c242d7ace0bc0182a1830f : TIMED_OUT in 1h 32m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/599d40571c4a48bb878b397a8d761e0c : SUCCESS in 19m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7611afd140742f6baaac8e9536287c3 : SUCCESS in 22m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ad910cb41e7401196bfbe437440b139 : SUCCESS in 23m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bcec03ee59a74e06b59c54402387991f : SUCCESS in 20m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d22ad7ef111042349a348eb7a196e4fd : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3d1e2a1d8df46eb91ed511898cb0993 : SUCCESS in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0ea52acf433e4110b7b4ec7a51eb42e6 : SUCCESS in 22m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bea101806b0340d98e0c0186097197da : SUCCESS in 8m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/abe8443974194a758194387171693448 : SUCCESS in 51m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d32cbb0cb36435c98ea47377590ff73 : SUCCESS in 1h 04m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcbf380bec894b008f83d581219ecf7b : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f5c253c55564bf586dba7ca34701462 : SUCCESS in 41m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b476015fdb24a17a8e9aadd446c3f1b : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6d3b71d05b46838f6f1fb9d8543302 : SUCCESS in 8m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6de378065b304cccb0725d153bd34528 : SUCCESS in 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43100415b0146698919ed9e5fd57b46 : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":12},{"id":"ede149cf7f3afb73fd5dced25835b4193d47512d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 03:58:40.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f385fbab56f4fda8ab4789a930e46a5 : SUCCESS in 57m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/339822e883c14786a30db3d0e74b3f14 : SUCCESS in 1h 20m 37s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5d5683053fa2499cab264e7e04f92f93 : FAILURE in 3h 01m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/353badd252404ef6beb605d9a1dfa187 : SUCCESS in 1h 12m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fcdb950749a44dbafd22440036950d8 : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":12},{"id":"7b2bd8f6d1b33458c8b6acf036d99e68ec0983b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 05:01:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"136323788115c5ecb027f3b543d91d1fd0404415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 06:26:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277df09a2da4413ea6ea03d34f4044ff : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/59d89655f1774682813e4a1ec10bf263 : SUCCESS in 54m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f9c7a2771014a968142090b64d1cd2e : SUCCESS in 51m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d2629e4145641a5a3240f14a0af2cda : SUCCESS in 52m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eadedc0406d84408a745a6a07e311239 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4fb981e8296a4bb7af0cfd8ebd972f6a : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f6f77e845b8c4a8f98db2731cc4d6666 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/182b386dc9474dc7bb41a582c796f071 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/55c725282f904cb084c7d8c70ddaf562 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b61da318f5b2468b9cfacd4de769e7a3 : SUCCESS in 15m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/405f59625eb04cfe9e64f4d401d09d89 : SUCCESS in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/103756b5a0e647dfa0002a35852472be : SUCCESS in 13m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71bf65d9913a493abe7ed38d33010a06 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/243296add99149078520d5b688789c79 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab10075e75b54fd69d480592d66575c3 : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d2cc995dfe74f6490988ee5418a95ba : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed8d40b138e449ffac71cf52a5a7dba2 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1cbcaad1eb3e4c5bbb5b181d2de2b3eb : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e50391df3225424ebd3e59630950a163 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25bca411d4834baf8463eb0e6b757608 : SUCCESS in 18m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a20f04d2a9547d88da90597fddf34be : SUCCESS in 54m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c0760d91ac34966bb8a0af25f371e99 : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e3ef714183443fe83d9b7872eaa4fc0 : SUCCESS in 41m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d96337cd1ca449978bd44ca8a921a7e9 : SUCCESS in 42m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cfb86d469284c1db46135c0a1403fcd : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/328132838e5241018aeeb29451cb39e7 : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4c40ef367944746bc34f8bebf7025a1 : SUCCESS in 22m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdb3842588684d5fa0cd28ce24c1430c : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":12},{"id":"248fc54c1e2b06d14608f72e89d7b94520bb640e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 06:27:37.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"482d74745e6b5258141bebff8c1cdcb6d37ac9aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 08:00:28.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce38b171f2fe4360ba8025092f13a236 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/03e4479de773487ab05314dc2d57a9ff : SUCCESS in 47m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89bee62255c34e10b0d148eddcfba91f : SUCCESS in 37m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a98c6046270940fa8c2b413a73675f76 : SUCCESS in 42m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/09f21ab28809419c9b6182057e711374 : SUCCESS in 8m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/85e9a8fb3aa443f6a47b285602dd2dba : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/98b9c910577d44f0aa32baed6346a8f6 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c31919878eb41c99762b98460be6a03 : SUCCESS in 20m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb25b835c4594793a2ac1cb1be3004a2 : TIMED_OUT in 1h 31m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd42abad2aa34de0880438d2517c87b5 : SUCCESS in 21m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e967fe976f234504a0bd91c74dc6b08c : SUCCESS in 21m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf8fc17356ea41b58a159897c7ab210e : SUCCESS in 23m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aca1b32380cb4dbe90b9e3712665bc57 : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebd7b1f21f6043739efabcc26598e785 : SUCCESS in 21m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d480b2226d31428b9da4101e56e8665b : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/483afbf73f4f488a9d3545d6ec3a8948 : SUCCESS in 11m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f94eda104ca41368bc175222112a212 : SUCCESS in 1h 12m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d08636b6cd6a485abb90825f204d4a06 : SUCCESS in 52m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89d0f9345d4544f8abc265d061642c15 : SUCCESS in 39m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0179d05423474b5ba53808e2715fafa0 : SUCCESS in 43m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcb2f93c47df4da4888b19134c828326 : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3a5cc95e9a4befba83424f20add68f : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f179be7f2001409f97a34465665a02f7 : SUCCESS in 25m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e94138a3ac5a4ba39d4fe4749e8280f0 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":12},{"id":"4f0983896cdcced558d20cd98eb79532929e5d7a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-10 09:26:46.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5890555a0df7311f57370b2c3caa995bb51f7e26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 09:43:43.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe05bcba637d47d7948ea4aefad355ca : SUCCESS in 1h 22m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82c934b1c0d1411bab88b7fd05cfe07f : SUCCESS in 1h 28m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ff0add9dfeee4f388ab4408c01001f1a : TIMED_OUT in 3h 01m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce4e997048644348b8c1296ad0158dbd : SUCCESS in 1h 15m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1c5ca72ed6844cfac10171480f1a568 : SUCCESS in 1h 30m 11s","accounts_in_message":[],"_revision_number":12},{"id":"c7cb279f9f223c44d9177b187edab6c07d02bb85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:25:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8438b7efe1314416b2938d4c3d6a1a1e : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/07dfdfa7e5214486b8f4d94ec64e4b56 : SUCCESS in 49m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58139870904d4a2280a2f76c51a4c5c0 : SUCCESS in 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18baf5c2c51144b59a4b39cdef0dc072 : SUCCESS in 46m 15s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d102395647fe49a8ba4aeeca5d5c04e3 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f1fb6e181654458d96be88934cb6de55 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/623a0e904c2348f38f046b47981e2dd8 : SUCCESS in 22m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/396cb88b0ec541ce88c848667848c9eb : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/520663daacba4ab69658fa9dad4937a1 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9978615561a3427f8408cf13e0374b8e : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/975052d2cb804376b6c7c81615498bf4 : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a5b4d04857c6475a850507e1da36c906 : SUCCESS in 19m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2c301f23e62494c876221317e24af66 : SUCCESS in 21m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db1987e0a39642848ef35bae5f0a4763 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6203226d204c4005a6a455be3cea5ef0 : SUCCESS in 23m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c4f1d40842164cbb97ba49472295d0a9 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bbc6e8b855cf4d49b1cbf170fcfb120a : SUCCESS in 20m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a2c932a232b04cf7aa4741b2f9e3ce32 : SUCCESS in 18m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb6dc9535e71402c88b5af4ae5ad5e78 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b2fde39f6354a8bbb88f6b72813eee5 : SUCCESS in 19m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e7cd868d20b41e987b581af5c7f1c75 : SUCCESS in 54m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a64595ed23e44cc81442d6d9449955b : SUCCESS in 53m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fffc0284b3ee44cf9531a9ea464d6b07 : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/117a72c3d2ae4c41a2a344dbaecbc98d : SUCCESS in 38m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec50c7414eb74eb2bf05814e78cec995 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30d7916dfa84575966384e69632bd8b : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/593cf5eeca404bf89c563c8fc519079a : SUCCESS in 27m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dda03663725644569b5519e4cc2e8fde : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":12},{"id":"01af78ed6299b3b0caba6bef709640b9544671a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 10:26:16.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"2c6475c77b592ed0d758a1c63f636e4e218ec45d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:30:00.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996aab4debec4da2923fb48495c02153 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/506e2e10cebc4c2ca7c248205c872464 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6f804681bb04cd392010e431caed2f2 : SUCCESS in 44m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c0bd4b1d4ab464d85c0598d28c1b351 : SUCCESS in 43m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/128279ab0a9741ac8e32b521fb69b04c : SUCCESS in 8m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/aeba91ef6e634228897f94c2763c6ccb : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b5e4f429d59048a8881d3db7475365de : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/831fc4ee5b194ef9bebe07a6a0213b50 : SUCCESS in 18m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51f7a192ebbf42ee9fdc123472e9c2e5 : SUCCESS in 20m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0956c762debe427995f70d74e509d7df : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22de6136050c45a0aff26292696961e6 : SUCCESS in 21m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e972fb4eec2f4c5593d820503e54c2d0 : SUCCESS in 22m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a637f395137145ab93eaa8fd6a2d0f79 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f8202d5dbce46e0b7b7b3d00bd50e0b : SUCCESS in 21m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0d40c967faeb4a8d9a8feda42ee444b6 : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b90fcf250254538a3e6c60a6ba82aed : SUCCESS in 6m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ff828ce84b4471c99e5ed177be33165 : SUCCESS in 56m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7718505ba2f44e2b9d4bf5d1fe37f397 : SUCCESS in 1h 02m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3638f95c1c54c5b869a6ae3fa711c69 : SUCCESS in 40m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bca398f751d740aabf848d85818a3e55 : SUCCESS in 47m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1ad01b195240358b3e1986452da32a : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaba633727544fb39d08d7072f2f55d7 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bf89d43a0df4809ae2127a7321542d9 : SUCCESS in 26m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12caa31b7c7c46459868bd0d39055330 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":12},{"id":"3a12a3194232783025e5a7c8dc51daec2fdc24da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:30:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"636bde4e89a6cd5de626f64081e19208e04a44c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:32:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72089dd5156f4692af3bfbb843e33a3b : SUCCESS in 1m 19s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1a0af218d8a24eabbef61499c9454251 : SUCCESS in 1m 04s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/628f8de22aa84120a908ed1f0dc051a5 : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"f2c279bae94689e2062beb6d0030d168a0b4cbdf","revisions":{"51ce5063ce8a5cd1be12688476e0404de8def9a4":{"kind":"REWORK","_number":1,"created":"2021-12-22 05:27:40.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/85/822585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/1"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-22 04:00:34.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-22 05:27:05.000000000","tz":660},"subject":"Trim tempurl signature with obscuremap in the logs (CVE-2017-8761)","message":"Trim tempurl signature with obscuremap in the logs (CVE-2017-8761)\n\nThe tempurl sig isn\u0027t the only thing that needs to be trimmed. In the\nproxy_logging middleware we also trim the \u0027X-Auth-Token\u0027 value. And I\nhave no doubt there will be more things we\u0027d need to trim in the future.\n\nThis patch adds some new request helper methods to store an obscure_map\ndictionary in the request environment. Any middleware that wants\nsomething obscured, they can either use:\n\n  obscure_map_append_param(env, param_key)\n  obscure_map_append_header(env, header_key)\n\nTo add it to the environments obscure map. The tempurl calls the former\nto add the signature param.\n\nWhen we get to the proxy_logging middleware, it simply uses the obscure\nmap to obscure params and header if present in the request before\nlogging the request.\n\nThis means if any 3rd party middleware or future obscurity we simply\nneed to use the helper methods and it\u0027ll be obscured.\n\nThere are two cases when the whole signature might be still revealed in\nthe logs:\n\n- if eventlet_debug is enabled in the proxy\n- if a tempurl sig/expires combination is invalid\n\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51ce5063ce8a5cd1be12688476e0404de8def9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51ce5063ce8a5cd1be12688476e0404de8def9a4"}]},"branch":"refs/heads/master"},"315cc467e051bdb549b3a5d5703f7017fef03520":{"kind":"REWORK","_number":2,"created":"2022-01-10 19:32:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/822585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/2"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-22 04:00:34.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-10 19:32:25.000000000","tz":-480},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/315cc467e051bdb549b3a5d5703f7017fef03520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/315cc467e051bdb549b3a5d5703f7017fef03520"}]},"branch":"refs/heads/master"},"b5fc3db3b947b34700dd9aaabb29c5a4a34b76af":{"kind":"REWORK","_number":3,"created":"2022-01-11 22:56:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/822585/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/3"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-22 04:00:34.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-11 22:54:27.000000000","tz":-480},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5fc3db3b947b34700dd9aaabb29c5a4a34b76af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5fc3db3b947b34700dd9aaabb29c5a4a34b76af"}]},"branch":"refs/heads/master"},"e7206a6871cbb1b6b4b1ccdf7046155ae242ca24":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-28 01:08:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/85/822585/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/4"}}},"commit":{"parents":[{"commit":"8ac63b760928548cd5cc1b3565ca14de3d36ce7f","subject":"Merge \"s3api: Fix types in /info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ac63b760928548cd5cc1b3565ca14de3d36ce7f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-22 04:00:34.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-28 01:05:37.000000000","tz":660},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7206a6871cbb1b6b4b1ccdf7046155ae242ca24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7206a6871cbb1b6b4b1ccdf7046155ae242ca24"}]},"branch":"refs/heads/master"},"0f140de5eaec029bf2e4d3432ae06a41f55ffc75":{"kind":"REWORK","_number":5,"created":"2022-02-03 06:03:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/85/822585/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/5"}}},"commit":{"parents":[{"commit":"b6302ab71910dba9241b293d0f5ae706d0b65e15","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6302ab71910dba9241b293d0f5ae706d0b65e15"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f140de5eaec029bf2e4d3432ae06a41f55ffc75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f140de5eaec029bf2e4d3432ae06a41f55ffc75"}]},"branch":"refs/heads/master"},"77ea2ff089a4d6c38d5b7a0c43ac4794f8bb4e5b":{"kind":"REWORK","_number":6,"created":"2022-02-03 15:28:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/822585/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/6"}}},"commit":{"parents":[{"commit":"589ac355f3ddf505983925d80b81c20107ceace0","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/589ac355f3ddf505983925d80b81c20107ceace0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-03 15:23:17.000000000","tz":0},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * X-Storage-Token header\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ea2ff089a4d6c38d5b7a0c43ac4794f8bb4e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ea2ff089a4d6c38d5b7a0c43ac4794f8bb4e5b"}]},"branch":"refs/heads/master"},"df5c5ef46934431c4704a3ce92fd427ab094bc24":{"kind":"REWORK","_number":7,"created":"2022-02-03 18:42:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/822585/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/7"}}},"commit":{"parents":[{"commit":"589ac355f3ddf505983925d80b81c20107ceace0","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/589ac355f3ddf505983925d80b81c20107ceace0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-03 18:41:57.000000000","tz":0},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * X-Storage-Token header\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df5c5ef46934431c4704a3ce92fd427ab094bc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df5c5ef46934431c4704a3ce92fd427ab094bc24"}]},"branch":"refs/heads/master"},"02e5667318afa49e9c058b465034724f51388fac":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-02-03 18:42:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/822585/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/8"}}},"commit":{"parents":[{"commit":"589ac355f3ddf505983925d80b81c20107ceace0","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/589ac355f3ddf505983925d80b81c20107ceace0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-03 18:42:19.000000000","tz":0},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02e5667318afa49e9c058b465034724f51388fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02e5667318afa49e9c058b465034724f51388fac"}]},"branch":"refs/heads/master"},"4ec366a6f6e35064a294b2f142e92fcbd10320ba":{"kind":"REWORK","_number":9,"created":"2022-02-03 18:56:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/822585/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/9"}}},"commit":{"parents":[{"commit":"589ac355f3ddf505983925d80b81c20107ceace0","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/589ac355f3ddf505983925d80b81c20107ceace0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-03 18:46:56.000000000","tz":0},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ec366a6f6e35064a294b2f142e92fcbd10320ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ec366a6f6e35064a294b2f142e92fcbd10320ba"}]},"branch":"refs/heads/master"},"70f90b51e4407d1933c36a9119440f733c119d56":{"kind":"REWORK","_number":10,"created":"2022-02-04 20:19:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/822585/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/10"}}},"commit":{"parents":[{"commit":"589ac355f3ddf505983925d80b81c20107ceace0","subject":"Move *_swift_info functions into a new registry module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/589ac355f3ddf505983925d80b81c20107ceace0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-04 20:16:25.000000000","tz":-480},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70f90b51e4407d1933c36a9119440f733c119d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70f90b51e4407d1933c36a9119440f733c119d56"}]},"branch":"refs/heads/master"},"f2cec1caab222237ad02f0c7df675664a704648a":{"kind":"REWORK","_number":11,"created":"2022-02-08 20:59:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/822585/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/11"}}},"commit":{"parents":[{"commit":"57f17ace7c911b80587a03634346478b6ca432e8","subject":"Merge \"Deprecate LogAdapter.set_statsd_prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57f17ace7c911b80587a03634346478b6ca432e8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-08 20:41:56.000000000","tz":-480},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2cec1caab222237ad02f0c7df675664a704648a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2cec1caab222237ad02f0c7df675664a704648a"}]},"branch":"refs/heads/master"},"f2c279bae94689e2062beb6d0030d168a0b4cbdf":{"kind":"REWORK","_number":12,"created":"2022-02-09 10:53:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/85/822585/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/822585/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/822585/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/822585/12"}}},"commit":{"parents":[{"commit":"57f17ace7c911b80587a03634346478b6ca432e8","subject":"Merge \"Deprecate LogAdapter.set_statsd_prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57f17ace7c911b80587a03634346478b6ca432e8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-02-03 05:29:53.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-09 10:53:46.000000000","tz":0},"subject":"Trim sensitive information in the logs (CVE-2017-8761)","message":"Trim sensitive information in the logs (CVE-2017-8761)\n\nSeveral headers and query params were previously revealed in logs but\nare now redacted:\n\n  * X-Auth-Token header (previously redacted in the {auth_token} field,\n    but not the {headers} field)\n  * temp_url_sig query param (used by tempurl middleware)\n  * Authorization header and X-Amz-Signature and Signature query\n    parameters (used by s3api middleware)\n\nThis patch adds some new middleware helper methods to track headers and\nquery parameters that should be redacted by proxy-logging. While\ninstantiating the middleware, authors can call either:\n\n   register_sensitive_header(\u0027case-insensitive-header-name\u0027)\n   register_sensitive_param(\u0027case-sensitive-query-param-name\u0027)\n\nto add items that should be redacted. The redaction uses proxy-logging\u0027s\nexisting reveal_sensitive_prefix config option to determine how much to\nreveal.\n\nNote that query params will still be logged in their entirety if\neventlet_debug is enabled.\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nThe reveal_sensitive_prefix config option now applies to more items;\noperators should review their currently-configured value to ensure it\nis appropriate for these new contexts. In particular, operators should\nconsider reducing the value if it is more than 20 or so, even if that\npreviously offered sufficient protection for auth tokens.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCloses-Bug: #1685798\nChange-Id: I88b8cfd30292325e0870029058da6fb38026ae1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2c279bae94689e2062beb6d0030d168a0b4cbdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2c279bae94689e2062beb6d0030d168a0b4cbdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
