)]}'
{"id":"openstack%2Fswift~823423","triplet_id":"openstack%2Fswift~master~I128a64c6126014774bf6218502e6bfdea63b783f","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-09-06 07:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-06 15:40:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I128a64c6126014774bf6218502e6bfdea63b783f","subject":"common: Stop translating log messages","status":"MERGED","created":"2022-01-04 18:37:27.000000000","updated":"2024-09-06 17:52:23.000000000","submitted":"2024-09-06 17:51:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"823423","meta_rev_id":"62f158162eb8c4cbb5c1b47c2d75f57f8e1a1a23","_number":823423,"virtual_id_number":823423,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 17:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-09-06 07:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-09-06 07:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 19:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 23:23:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"201752a5b52f1697b769c622e8885419d8ddec71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 18:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f4baafd5fd25a86c146e5eb4780b8d66e75683d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 18:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f80bd275161aac7e175bbb96ea015c19be7d12f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ee83bb885c48a58e7a66faefa1c650 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b6e40f277b0441c48f34c8faac8a3ed0 : SUCCESS in 58m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7a35d44d43e468086d882d0842d8fad : SUCCESS in 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eeb4144e51a4cafbe3773d725d8fe25 : SUCCESS in 45m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9fa5b4f3586441680762cb596c1efec : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02651a5098ca40cca9eb7bee402e8841 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2be0bfc52c8643b783074906df07d07d : SUCCESS in 20m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a341e56714c3440f9a51136ee93dd4d3 : SUCCESS in 17m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/057b5b6f2ed84c93972eb648d2f310e2 : SUCCESS in 20m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a1666b759264f60b4a148213bb2a58f : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3610f5a2d7e84e868062c2c9a87fc43c : SUCCESS in 24m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16542b64332e41f5baff5b4dfde88c79 : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f12f59ef1d9944a3a4efe4531c5d949b : SUCCESS in 23m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/249bd25d638f4a859c244fc12a4694eb : SUCCESS in 21m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ef050e2de7e4ecb8f890a5ec1c7dcf2 : SUCCESS in 24m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cce8f0bb18004ebc990a4a077c9692a5 : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e85003043be498abe16af8cf34ed273 : SUCCESS in 25m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a1e2317c34e844749eefae747060598c : SUCCESS in 48m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e86646c6a9614a9f9f798c17bba63224 : SUCCESS in 55m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8547a512e2764d29bcdf32c84dc16be9 : SUCCESS in 38m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45f9aa010eb24df6b89ba8e345c2bffb : SUCCESS in 41m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4e33d987b13466587c90ff079e98418 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67b9366fbbe45b9acfda45c2dc7f00e : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2d527ce385541f383694777849c6d13 : SUCCESS in 25m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6871c8b486804afcaff647d69d5b64eb : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9726972eea1a39a4b4698d20244f6e4c36257fab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 20:30:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39e950f5d8e847acb6bd100382b57bd4 : SUCCESS in 31m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abcf4231f83c4ff689670930ea7b645d : SUCCESS in 33m 19s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8d0ff048c3a342ce9f63d8d092792e15 : SUCCESS in 1h 51m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eca397c7e8a84561881efc2a5aba0be8 : SUCCESS in 39m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b9f29dbab6940478ab72016ed0858e4 : SUCCESS in 33m 45s","accounts_in_message":[],"_revision_number":1},{"id":"295edfa5bfc9e5a2b65716de55590e65f410b237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 21:44:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7cfeeebd8aa3371e4d4fceb30755fb8a30f10b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 22:46:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba575b27cbb47ce808833fc1e0092a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4e8c5e7d7849d19956664e0dbeb7ca : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/81f739710fbc4dc1b926fb8a8310060a : SUCCESS in 55m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0421abdd6e3c4fafa90cca2e5efc58b0 : SUCCESS in 40m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47e17ff1ee2b4ebda1794676b6cfb3aa : SUCCESS in 57m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/236bd00c4ec44786a072d3a8df7aebe8 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03d64c60052b46af9e7c514507b574d4 : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/badc15f0c75f440b83f77621152216c0 : SUCCESS in 22m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c3d5cba824e45d0ac8b976c8ea8a00e : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8fe3df655b647f68481b0c387b65f81 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24c167d4a4c04dd98aec56e85d55e955 : SUCCESS in 16m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fd3e7c384ee4431f8a0599710295d6bc : SUCCESS in 22m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e2d284c2a5bc44e7b58874b888f8ad60 : SUCCESS in 19m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/87378877b2d24ae38fe7bc108ec1cbe2 : SUCCESS in 25m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2c58a07d0b84556b19804941c0c2536 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fbf15e3f58314d9fa9674b315a02d571 : SUCCESS in 23m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a4d4500e57c4f8295d9da128458da32 : SUCCESS in 20m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0af8baae09824e54958b174a5981b137 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f7fc7798b274f808ce104136e0cc8cd : SUCCESS in 18m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9e0041d5dd840a68cfe509810dc9951 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9db08f06c82b482fb14027193f1c4621 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cff7662842ab446982aec45425e449e4 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d3cd77a710f4698b5434ea176c31131 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/19ec22dfd2204756940cb4fa325cf5f8 : SUCCESS in 45m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31ceca12884548e2a2d502762ff12494 : SUCCESS in 53m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79c0bb19508040348e07ecbb73c4411c : SUCCESS in 46m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ada3a9bc46445fdbfd0c3b7bfbbf30a : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc4b866e520d409f94227354c1b0a744 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d5008f9f5d640c488702e6c082aae5e : SUCCESS in 22m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5117aba7531f49a792648c4cff7866f1 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a0ab0fbf076fc5abafed2a4a1d76784f730d2641","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 02:46:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0224d478909348878992f07f6b11b480\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6514509ad764d85968dbd05cc70d5f1 : SUCCESS in 42m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/409b7f6efdf641ca922d060170797383 : SUCCESS in 42m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0b40bd798224ca0b5e0118d790f5e7a : SUCCESS in 50m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/91e241d2af1a48d1a415308208485a7c : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d6d8a46f45fa95093f682a051d8067eee7c1b1a8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-03 23:23:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ba3dca5dbd6994e33644573d8cb9356fe659369d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 15:23:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2718a32d28ef0384b3db17165880fa39d0a87055","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-06 07:00:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4f71bb4fe584fa64d6d84d8a2ae51a26a0e0cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 07:00:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f03bbb47334b03114cb1a969068fd93c841dfb72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 08:17:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1eb5dbaf7fc4d2f84f3b9a7ddbcd6f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512896df7e7d4c3daac69cc7eaadf5f6 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/da6937df28d648e7893984f0ae3b8599 : SUCCESS in 1h 05m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f671339df941472f8d04b7a4041ada77 : SUCCESS in 54m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ba1e9f776484fb6a2898f8c19ec1157 : SUCCESS in 50m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5c640bd40c094fb5a8f3c78e2291e54d : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e592729d42114669851ebda5935cc53c : SUCCESS in 17m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07d0d3272ee146608afe7f1a4d9e1aed : SUCCESS in 17m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d13d346d16254a248bb81b23dec1cc49 : SUCCESS in 18m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/af71c0cd1ce44b228686136946859a42 : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7e213799405c4dfb847b1788036b1532 : SUCCESS in 25m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0317a1daa9f48deb49275a567baf83a : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94831db4e55c4e28b88271f807e401fb : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e32aebefe0442529ce6b7e7888f8011 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91691669336b4c69b76a56c553ebb954 : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6033d42a93ef4c3eae0128c9edbc4d39 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ba90a6282cc147fbb90044eaa58067e5 : SUCCESS in 21m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30c8608493ab446db69086cbbff2465b : SUCCESS in 6m 02s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/076c187163de4de6a86181e4573f6692 : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd833f83e7354ea1acdae8719bf9b57b : FAILURE in 47m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/079e8288c0ad438399fbd248f30dc5f2 : SUCCESS in 1h 03m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dd5b745c0944dbcb7e8bb1f344ca2f9 : SUCCESS in 49m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3510474b3dc64801a4830a0eab5fe91e : SUCCESS in 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202cc7967f7e4c8e922cfc4b0005ba62 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4fcd0a0ecce4da198073098a596222b : SUCCESS in 20m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6160e56eb9f44d56be9d9585e2ca90c7 : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":2},{"id":"50d881050c8840fe8fec0e9fb6f26310987a66f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 15:40:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f42483aa0fea2f883af13be415ab6a1748bdecae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:45:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d6b81b7503b4d48a5331705cab9dee8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0305c5298724816a04ed850126b39bf : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/48c779a3561041f5ab7aff8bca8511fc : SUCCESS in 42m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0ffe74525c143f2b6033cac02323fd8 : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7708de2455d4455d9a1b6f914b003417 : SUCCESS in 1h 01m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc64d24ae6084acaa3c926541da0234c : SUCCESS in 15m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6a3f04ab7dc4814a0bb4c26a01ef13e : SUCCESS in 24m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16731d51dd544b44865afffa145e7e70 : SUCCESS in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f58ca422f5a416bb9da84e73ed81291 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e1328a2a5d945a0918c33df02287363 : SUCCESS in 20m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd1fbaec34e945ac9a887e9bd6315a04 : SUCCESS in 20m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cff662009b894df08b7aced6cd40e7fa : SUCCESS in 21m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/27437100b11b4ddb9501739145cae890 : SUCCESS in 28m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0d58e91052d7444a86bfaf4838603a5b : SUCCESS in 31m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b0c1ab5497140ac8d59d2676ad61aab : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08b2b9194fc249df9bd4c054cffa5d06 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/660f0a94754d4f7ca3061bb1f5798c96 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/216fa534b53d445288fb05dd1d0e806f : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c9440fbc93f24f4496ab451411e25ea6 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7d35eda720db413eafee475db3017f65 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f3a96527b7b44359836193ed3d0af2a : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c77c89bc3a3541b4ac8520496642ae9e : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db9f4823c3284f36bf558eb676fc15c5 : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae381d97325492cbdd5f2cc0b1f8c44 : SUCCESS in 46m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c355707ff87e44faa9272f78c0d28160 : SUCCESS in 49m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/186a126986f541019821e963785e5a6e : SUCCESS in 37m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3871275683b446d482f71463e17c819e : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addf9ca7f90a495ea284192a20361559 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a2c0aaade7941eca8f0e3cf30516b48 : SUCCESS in 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a75afa9b9d274976bead9f4f8387984b : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":2},{"id":"97361fbeec5e49a9870836cd99c1d2b4af694a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:45:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd8f0cb6d78a09885c701496e5756eb288477ea0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 17:46:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/053adce374134a52b3a7e622135e1d37\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59ea3e1f6d0549ba9454dc500ae0d4cb : SUCCESS in 1h 29m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b07d0b0dc15441699bb70540a64a55c8 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b49bfd5a513644559374167308a1765a : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3cc16e8f40e4956b9102054c9ae71d1 : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c42b5d0ddb10816602ee0c0c8b3bb7fc85218c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 17:51:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dad7d353c4a452aa891f8e40ddab39d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3c99b74d4a4936a64eb08a901c5484 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/ae58642618404c61ad11970672c06d64 : SUCCESS in 1h 01m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e37620e891e2488bad32fd90bc81071f : SUCCESS in 54m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f18e97dfc17d4069bbcc04b5031b0511 : SUCCESS in 50m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b795175032ec417380c52f912963c89d : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c416336230954a87b0561dd3c48dbd8b : SUCCESS in 11m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dadb0cd7e6a8410791548754db805554 : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e22bdcae96d404aa294e936e7ddfd30 : SUCCESS in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5925478e1dd045ce954fd28e8452ef86 : SUCCESS in 22m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d91c0318d07f4cbc8ce7c11ec2c20687 : SUCCESS in 20m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cd005037594e4b62905044d8764f2a31 : SUCCESS in 22m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4ef9d5493e9407a8be11f63e57dfff7 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d18c1a05560a485a84fbeeb5eac7a622 : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ae94432183c46f48e923dd4bd543980 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/febe50b3e4ea4f899290f4643c25ce7b : SUCCESS in 20m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1f25e08456744269623404a78bc7a5b : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f40d2625167c44c09a6e7536ca1e60b0 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ba41b86090d94330bd0d109475a931e1 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/111e6dd6396b400dbd7f81f638e4a0de : SUCCESS in 51m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/524fbf452c0c45d2af876ba0edde80eb : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd7e006a78f949f88cfe2c63bf40bc63 : SUCCESS in 52m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0a475c29e804581b47c28810817622c : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b96e8154a2d4ac389661d782d1b5fc1 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36e96580a58449cd9d95d77d4c349336 : SUCCESS in 19m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c796da3308e84d84afa413d0f78b4048 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8667abaaae4336fd24c60a2160e0e92ac38bb230","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 17:51:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"62f158162eb8c4cbb5c1b47c2d75f57f8e1a1a23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 17:52:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b908895640d1482e9969ce74808e85b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/504cf5455b524082866995eb3203b496 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f84c453f17c64d36b8e2772eea45990e : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d888fb13a6bc5094b53915fd48c947e2c6c1736","revisions":{"f6554b82671483bcca80561a1e517a363d7e7547":{"kind":"REWORK","_number":1,"created":"2022-01-04 18:37:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/823423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/823423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/823423/1"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-01 18:49:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-04 18:36:54.000000000","tz":-480},"subject":"common: Stop translating log messages","message":"common: Stop translating log messages\n\nChange-Id: I128a64c6126014774bf6218502e6bfdea63b783f\nPartial-Bug: #1674543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6554b82671483bcca80561a1e517a363d7e7547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6554b82671483bcca80561a1e517a363d7e7547"}]},"branch":"refs/heads/master"},"4d888fb13a6bc5094b53915fd48c947e2c6c1736":{"kind":"REWORK","_number":2,"created":"2024-08-27 21:44:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/823423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/823423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/823423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/823423/2"}}},"commit":{"parents":[{"commit":"769baae36e2e47b57b6390c7bc5e5b210d4fb2c3","subject":"Merge \"account info: Add --sync/-s option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769baae36e2e47b57b6390c7bc5e5b210d4fb2c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-01 18:49:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-27 21:41:15.000000000","tz":-420},"subject":"common: Stop translating log messages","message":"common: Stop translating log messages\n\nChange-Id: I128a64c6126014774bf6218502e6bfdea63b783f\nPartial-Bug: #1674543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d888fb13a6bc5094b53915fd48c947e2c6c1736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d888fb13a6bc5094b53915fd48c947e2c6c1736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
