)]}'
{"id":"openstack%2Fswift~823424","triplet_id":"openstack%2Fswift~master~I666758fd153cf36a4a2314b5bd50f6cbdb8d7f07","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-06 17:52:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I666758fd153cf36a4a2314b5bd50f6cbdb8d7f07","subject":"common: Stop translating a bunch of printed messages and exceptions","status":"MERGED","created":"2022-01-04 18:37:27.000000000","updated":"2024-09-06 21:08:20.000000000","submitted":"2024-09-06 21:06:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823424","meta_rev_id":"9b21f9ffab0095ec6eccc34413fe25537be4ee2f","_number":823424,"virtual_id_number":823424,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 21:06:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-09-03 23:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-09-06 07:00:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-04 19:56:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 23:24:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"95d33f5b523a9c5e5cf631017d2f4d5b076d152f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 18:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"714fbfd39c88a36b3b4a0960259e2c610b55ddb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 19:56:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca6621cff0a4a6aa46de0fb32bc5448 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e90fc29be8f4445f8b213e602b44b7db : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0b6461b61734beebaab16c13b05499e : SUCCESS in 56m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c528e51ebbc3496ab00e17b81aeb4667 : SUCCESS in 48m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0fcce063f094eebb92aa92f8bffa44d : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/350b0fbaabcb445fb9844062e289ade7 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64e7cd1ee3ed448d80410a67b48865d3 : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9adfc5f6a54d4401b8a369380bc8e3b9 : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d690d827a7154e1fb9f55594e3d64b26 : SUCCESS in 15m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4a5ed5150f445d38919d76a3f98904e : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5fce3c2edc054ba78a29abc5b8d28431 : SUCCESS in 21m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/acb59566791b47749daed6cad27c0a34 : SUCCESS in 16m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/028486da0cee42aabdb8b5f0acfef2e6 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b1e23ea885a412da02989b3f5c29203 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9d17f6b478d346a58254c6aa5d71a302 : SUCCESS in 25m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/425bfa6fbc5c4cd4906a12b643fd7928 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ddfa715c11c4f408d1046923fee21fa : SUCCESS in 20m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef176971bb0d43d78e86984ebfa50686 : SUCCESS in 47m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1d661b8e76c4a41bf0966608231dc59 : SUCCESS in 53m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77a6179ed2be41dbb827ef52ec11ff3b : SUCCESS in 39m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51d1094691da48ceb5bff0a9bae53f3e : SUCCESS in 58m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc703390ff5a40b4ae3e5442ecfc42a4 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e81788333c425e914bac2496d4f03e : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a418643da10a4a58a3dc33feb1a02278 : SUCCESS in 19m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/863a225f79f04103a1fabea92ba4c7a3 : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":1},{"id":"154e6eedf5c073b371b65177d7652bdb75b6e848","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 20:41:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cc0056917fa4967828277e2ddceff65 : SUCCESS in 32m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6680c47dae24a90ac7603064b25b7f4 : SUCCESS in 32m 42s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/adc8ed18af16494e929c8a1b3998ac8c : SUCCESS in 2h 01m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a111c6e9863427395a6d5b3b0055712 : SUCCESS in 1h 11m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a6e99017f6f4f23b0d028109e598d38 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1fb77bb8eda57133d9b070613f3c342dcaee92b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-27 21:48:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"54b087ec81195cf2eced5a6c215de51a7163d533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 22:50:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1934976ad89943b5b0348f9aa2037822\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7c7f17bf0b4bc9bdd2e1d27e3828cf : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/168b5a476c904a468fe0926e5d2640a6 : SUCCESS in 43m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8702e1a9a1a9493ab72a3f51e8128f68 : SUCCESS in 37m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3f346a35b844c0cb71b798279260a12 : SUCCESS in 1h 00m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e7a6d489a3f4688b04dc5bcb4738f70 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70c6116773ef470d9da58f257092759c : POST_FAILURE in 34m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dda237da97504ad987b7f2087911e24a : SUCCESS in 17m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c58c73b033de43da873b0f6888264751 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d12934c6d47c45499fcca47aae1401e7 : SUCCESS in 13m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f574da838b854c8183b2787971a140f7 : SUCCESS in 14m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3cd4f310c2d741f8825064ccc7497955 : SUCCESS in 21m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/066a9e9166324f84b7ff53a2528cbdcb : SUCCESS in 21m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d3d36549cc743fb9e144036309dbf1d : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bf76ee82a714f38acdefd16eab7d842 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/727698ba7f80460db504e8862a00eb13 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/261bc10d31664158ae84a29ddefbb2ab : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/343902d88bf74ca699381301bfb4c516 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3ee2446d6d1a482691b7ddd7aea5f08e : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd6b8525c9d34107bc6c23791e5c70ef : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7296076391f4ade95aa26450f9ae06d : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8991af55be7c4e92b0f678e3e9b68743 : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ab870c11ec944c780af25fd51b16ba2 : SUCCESS in 6m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2585694d67ba496989a87f82b7719448 : SUCCESS in 50m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4eac58ecb0bc4678844861664904741b : SUCCESS in 45m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18404f4acda8408c80f8f763336707d5 : SUCCESS in 45m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b76efceaacc74178b3e6848b3d6120e0 : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840e6234d2724943804d9f277caa017e : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f42b3c54cdb64868bd2a626542b9b4fe : SUCCESS in 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9e246f5b00d44d093acf84305b9afd6 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5b8cd1ff1ed6a3cf29b57978e4c920de85962f59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 03:09:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd8923a3a3448e8897b79eaeb01ef79\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55c84c823149453faacdb521e39d7de2 : SUCCESS in 1h 11m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bc1c45323de42adafe6479e15a2bdd5 : SUCCESS in 41m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bfaa5d9869843a5a2a86da97f2accd8 : SUCCESS in 42m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a055de6160b14ec5a84ec5121434cc91 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f6e4577b44fed2a5fb2c48ea0e2d33168190b336","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-03 23:24:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97582c5f0411c842a53160b9ce76fb766dd2c0bf","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-06 07:00:25.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be0c80ba1a817a5541c221970ac76256a9d8b4e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 07:00:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c95b28130496c1317a10fac53ca14d923d2a996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 08:17:41.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/823423 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e1770931b031e0e4686173f98e39be2f1457f9ee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-06 17:52:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55ae78223cc0e05558c47b437dd068afd96ee079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 18:57:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c863a2f5fc44c5e8801badc5e1c0a1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d5c9a383ab4dc1be028a96a9d42fb3 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/72643d4b4fdb400499dc7a828e26bddb : SUCCESS in 1h 00m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/44a91c7ba93d475e914f71e1d1c12745 : SUCCESS in 55m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a92587af31b44e8bc6b7b1d9f751dd5 : SUCCESS in 47m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c9ebc04746e4fa1b5f6bfffb6bc1f0a : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/43ed6d783cbe4f2c9fd17148fcda74fe : SUCCESS in 23m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75c5ea9f89804e01a7e10ea65c987c1d : SUCCESS in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b42b5198fd19448abca549200db20b03 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef85ad93c3e94c6a8f460d8ffe42f63f : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5696760914148249362a3e221615bbe : SUCCESS in 11m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9da45656482e491b9f49c0e0c109892e : SUCCESS in 25m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d23883c206354534b858f68847898ae0 : SUCCESS in 11m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/805c5f1328a74ea0957648e0f10a82a1 : SUCCESS in 27m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/086676a08cf447d4948db840bb430906 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7977eed86cf1486cbfb7d424b82f33c5 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1516da9cbf4b44c2b9931f99b37b8539 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d17c6fe80dc648609635950bc0f061ec : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eadac6039cb643a1b8e7f244390f3c27 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6cb8120beca947c2b6f6335be6c3c23d : SUCCESS in 19m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a5f2a69a6c2f407ab79857a667599ff0 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d5bfddda8c742d1a43f26b53fc39ef8 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d08119034b20459583ee77de76834633 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/62103aba0a914891bb40e387b6a53b18 : SUCCESS in 49m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90b47702544a417f97fdd7e7f482d134 : SUCCESS in 51m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4dbb66475d384b0a963a76cfc915367e : SUCCESS in 50m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/800333ed575c400ab65860f0cba855b5 : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/005352942caa4ce3afe26c21e5e1027c : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/40107aaf162e430caed6692a0d90fb0c : SUCCESS in 26m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d743593da4742c696ff50f0c3283340 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9c351ec69c3cc707fa2b2a9c123bd51eadae56e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 18:58:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd383558b496c4f3e3a22734ab006b71a97168ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 19:51:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f691093c074d138150e9c1a960e991\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88a39df75d594684b4bf140000c95593 : SUCCESS in 1h 00m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dacd1c0314da42dca949147aa0afd62c : SUCCESS in 42m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d7b28f123aa41f4b07227f2b4176e35 : SUCCESS in 46m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/704478f0367942308794e9cc70972929 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b58f3cea8899f13541900eb4797305a839e45276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 21:06:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e58437c2b4744d594c2bcfb73b40c9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f10fbe2128545f1aa557bfc7d132d03 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/a506eaf617e74ace95e8359f84fec3fb : SUCCESS in 47m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6945598e04e846dcbcdda214a7336daa : SUCCESS in 44m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6e4a9708e3547b79b8f10d0d4977b1c : SUCCESS in 53m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/411af315d6934baea8953a32e3be6792 : SUCCESS in 9m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77096819683844f5830d8895ae768960 : SUCCESS in 10m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd866661e1e647c798965265e384ec2a : SUCCESS in 15m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7d45a128842c48c3bd79b6fc0a95169c : SUCCESS in 12m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3d5a7f563a504269ade4556e438306ae : SUCCESS in 19m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1f2e0714016443ba878b8c82837a6e2 : SUCCESS in 25m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/052bc46f98c947609ce98ceffbf45b1e : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a20a85685f14a5b97a3feeed400f464 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56d82b1fdd304f359d5cf4794f791935 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3aa6b4e2db7c44bbbcca2142da724f9d : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/87aa3268f8a644a487f725e9be5c3794 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67744bf61b134f0fbb65195e56a9b505 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8f4d7f448e142459ed7e1daaaf22524 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4582952e5e184960b619553fd8d6ff85 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbfa952745424147b6895544698a4e7c : SUCCESS in 47m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fac4f9c1e0742439633f12c736ebb04 : SUCCESS in 39m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23f808babce24be0aad62f5bc086f8de : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6f7a7d2142d47d49b08a01c28df6e9e : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babddee3be0e484f9e7feb4fa536badb : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2505bfb0e5149a2b55cabb04ece0849 : SUCCESS in 16m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45b18285a2744d55abc9e769e4053d80 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":2},{"id":"aceb0c4174260bf86efdfcc8ce766c32c2e88185","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 21:06:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9b21f9ffab0095ec6eccc34413fe25537be4ee2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 21:08:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9afbc8a4b7240e39538313873dae119\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3afe0bb40f6f4f728de0cc8657519bed : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/969766424c4245aa8aac963d541ba8fa : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ac294b626be0c69170aa9124009f96340f2a7b1","revisions":{"2b4503dd0939406eae8d99b8e441d6979cce7d24":{"kind":"REWORK","_number":1,"created":"2022-01-04 18:37:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/823424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/823424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/823424/1"}}},"commit":{"parents":[{"commit":"f6554b82671483bcca80561a1e517a363d7e7547","subject":"common: Stop translating log messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6554b82671483bcca80561a1e517a363d7e7547"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-04 01:06:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-04 18:37:17.000000000","tz":-480},"subject":"common: Stop translating a bunch of printed messages and exceptions","message":"common: Stop translating a bunch of printed messages and exceptions\n\nChange-Id: I666758fd153cf36a4a2314b5bd50f6cbdb8d7f07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4503dd0939406eae8d99b8e441d6979cce7d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4503dd0939406eae8d99b8e441d6979cce7d24"}]},"branch":"refs/heads/master"},"6ac294b626be0c69170aa9124009f96340f2a7b1":{"kind":"REWORK","_number":2,"created":"2024-08-27 21:48:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/823424/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/823424/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/823424/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/823424/2"}}},"commit":{"parents":[{"commit":"4d888fb13a6bc5094b53915fd48c947e2c6c1736","subject":"common: Stop translating log messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d888fb13a6bc5094b53915fd48c947e2c6c1736"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-04 01:06:14.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-27 21:47:46.000000000","tz":-420},"subject":"common: Stop translating a bunch of printed messages and exceptions","message":"common: Stop translating a bunch of printed messages and exceptions\n\nChange-Id: I666758fd153cf36a4a2314b5bd50f6cbdb8d7f07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ac294b626be0c69170aa9124009f96340f2a7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ac294b626be0c69170aa9124009f96340f2a7b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
