)]}'
{"id":"openstack%2Fswift~823473","triplet_id":"openstack%2Fswift~master~I4c33ed5f5a66d89d259761b5ce12fb6652b28c40","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-01-10 17:28:33.000000000","reason":"Alistair Coles replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-28 02:15:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c33ed5f5a66d89d259761b5ce12fb6652b28c40","subject":"reconstructor: Abort just the changed policies","status":"MERGED","created":"2022-01-05 04:42:34.000000000","updated":"2022-01-28 02:31:42.000000000","submitted":"2022-01-28 02:15:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823473","meta_rev_id":"c2043dd61ab140ca4a128ddbe7be742483bb2a50","_number":823473,"virtual_id_number":823473,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 02:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-01-28 01:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2022-01-21 08:06:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-01-10 17:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-01-28 01:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-05 06:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-05 10:24:37.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-01-06 11:51:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-01-21 08:06:22.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2022-01-28 01:15:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"ba65ab48803c2c19f36b7a39eaea706f6452c45c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-05 04:42:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"728cd51016594e399b7772e9d089cbb8c5422338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 06:25:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ebc0bd3844434796e8b34c335c987c : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4f661019c50d4c27928a0b75f9e23142 : SUCCESS in 59m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7bed31ea384e473c8e65f3f167444ea2 : SUCCESS in 50m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0f2e848576a4affbd2f1a98c9b5fefd : SUCCESS in 48m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f67d01a33604cad9252e6f799916d46 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c59d1577f66c409b99d89a81dbb17941 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d66a0e4617a2451d820afa0dc91328fc : SUCCESS in 16m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f172752fd6a4037818e86b50f0b9dee : TIMED_OUT in 1h 35m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65b8de1512b24af48c7ca72de226b043 : TIMED_OUT in 1h 31m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63b44b9f0dd04f26944687d60389ed41 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5635268ef14b4e648e9c1e79a6cbc9fa : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0be7966a1534a4c80adb919c001d486 : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d150474185f477f9656c7ebb4d82697 : SUCCESS in 26m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89047387bd814ab0bf1644d37ead38a0 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adad38b214ab4995bfe6b24c0020c40d : POST_FAILURE in 28m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07610ae3b4db4effa44c207ca2036b2d : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75f5c26990004e5fa07b9fc7d927f9ed : SUCCESS in 20m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74cb17943f01419996a0f296f5d61e16 : SUCCESS in 53m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f7bb495d6a94cc9856224539e48f848 : SUCCESS in 52m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2abdcdc3dcc8485e840aeba12139883b : SUCCESS in 32m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/258a59b332774ad3bf53a701006d8313 : SUCCESS in 37m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/428f2fdcc2a64d63adfc760f5c7c58a7 : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8e8c279b774cefaa06a1288bfdc1d9 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3acb9899e14f4944991620182f746d0b : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78070ecba420457aa584271ad8ca2633 : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"82c2743db188dce834a412263a82f685b74da923","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 06:34:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35a2a101325841d98d81b028482071bb : SUCCESS in 40m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18fc0180da3c47d9b4f04eb1f72450ff : SUCCESS in 46m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4d734980a64d4c40a087f12d75eca604 : SUCCESS in 1h 48m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/66694a4a482d49b89aaf126a99d77e3b : SUCCESS in 53m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5af874ea9046d589bd3d3985f6c503 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fde7acf8aa56cd71e25d5d5a50c763b4bd4728fc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-05 10:24:37.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3a3c13d5cd45352c68154eaa4bd737308b71a159","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-05 23:55:14.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4eaac534836d4139aff60b8843b12e5e6427e8b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-05 23:55:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2125cfbea45a85489cc25ed2ac2146e26f7e6d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 01:27:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0bdb4b0e6c404e885ebac8a2ff3eb8 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/82da4c64ab3e4187b0d875e705753109 : SUCCESS in 48m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dca470e4f5564376ac1a4cfb4cad0c98 : SUCCESS in 41m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca87ab78ad2649f4849e20b19a186088 : SUCCESS in 44m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb14c5215a56402db54b86223f69fb8a : SUCCESS in 7m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b08363bbcb1a42a7a8f7fd2d69a1f44d : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c14372844594d00945e245664390bff : SUCCESS in 20m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/500ae015052a4042a374a9465e9f8e08 : SUCCESS in 20m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7ee18adc12543059f80206da256b378 : TIMED_OUT in 1h 31m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd402af6f2fa469cb7fc528e5bcd8cbd : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/19b3dad47eff4f7b96b8cd2ff762eacc : SUCCESS in 25m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb1db3ccfa164540b0e40fa14d0e8cab : SUCCESS in 33m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ea0d1355872c47f2afb2e663eb0fbeed : SUCCESS in 30m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/07b5f889faf843ffa8460bcf0a143838 : SUCCESS in 21m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d65e5b437f04715922a7eda60d3fc7d : SUCCESS in 20m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64dfd29293ec4b08a053003813bf74e5 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9f0515bbb404ff1abab24d398d936a8 : SUCCESS in 22m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5521439b9924593920d6846a21e0acc : SUCCESS in 47m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/04c8de9c40af43d187471a51c67a98b4 : SUCCESS in 1h 04m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/487f60224a7243eb93bb9e8745872b74 : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e4abb25b9dc4a38b27595204beca5ac : SUCCESS in 47m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd320719f1f54e699547ad86af94c4d5 : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515a666c3b2b4d39a5c4484b33ea36a5 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87d7dd59306b4b9e8272a8ae4c9eb671 : SUCCESS in 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cec74799cf3424fae0b7d8177c1f9f4 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e687b0615f7947396272cce032a96ac1cc4779af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 02:19:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c01b3b7b2175479eb7c0941012365478 : SUCCESS in 45m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bcc03d7631a40379f045dfb21d3df94 : TIMED_OUT in 1h 31m 11s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d374f6dbdb8d4044b2a575d96e1da89c : SUCCESS in 2h 21m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ead4ee908924ff8be8c79126179cd8c : SUCCESS in 47m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b9c7ec7c9dc418b8178c68c6bae3bdd : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":2},{"id":"1096a048f2314b3baa6d7f1696f5fa2c85c167bc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-06 11:51:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"421a9edb94091afe4eb2ed654e4a9c3f154e48eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 20:35:50.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6e535388d4c07180ae149545678beae6f7974a6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 20:36:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41b6d6a3564a5330ecda7b45596ff86301e8ae50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:43:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd8c3a92f4f487bb05788ed91ccf741 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/ddeeda2e54f64e0c99c18627e5575e4f : SUCCESS in 1h 01m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4780ff2eb71d4aa8838e0ae455d868b2 : SUCCESS in 40m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b37c0f48d9a4a528fd6017cb722fdc1 : SUCCESS in 48m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae56341837af432a945e3a35ce6bba7e : SUCCESS in 12m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42df8e432b0a4d2d87fe0e2c125795c3 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cba7492f5fa840e287b86f9a55f06d9e : SUCCESS in 19m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af7c99e89ffd4ac880b814da11f48601 : SUCCESS in 19m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1089b184d1e94fb8b0b31264c8625f30 : SUCCESS in 19m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f092452360f84bd6854324e3f0052f2a : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65a50bccf1c742f3b9c6e910c4661dcc : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c9e3907739674fe18b816cd0155a42d6 : SUCCESS in 20m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f11482b9281b4c7aa5d0640e20afcfef : SUCCESS in 22m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc6ff33095824453845659901d3c54c9 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/138c99b89512407c9a7aa63ca3a1de01 : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/de5107f1880d4754850069c9021085f8 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f312bdc8dd20413db807ee377ff7ca9e : SUCCESS in 27m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a58d466baf9444639433a6042960cb0c : SUCCESS in 54m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ae20e3e0da84811a4c99e4355b55c0c : SUCCESS in 1h 01m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4d67bc3c3444a52a4132f8c67230a67 : SUCCESS in 37m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd6fd3c96a414b37a9cbd5483534921b : SUCCESS in 44m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46f71e32ae1941779cdd46bfa64305b0 : FAILURE in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b367c3f661de4f5690bc28f2a089e494 : SUCCESS in 8m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/907544ac4db940f6a8291a52fe2ba273 : SUCCESS in 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb1e29e8380c41fa884bc4888d960c19 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":3},{"id":"36577f10e3919758a8687a1053ae8ccfffe2032f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:36:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82cd117e6d3e4ef8942aac9a14e6043d : SUCCESS in 50m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/249a8d3534634a968a25fd969dbff8f5 : SUCCESS in 53m 54s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ba019621b59d4c63881db1882dce44b4 : SUCCESS in 1h 49m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7063e874b8d4367b34ec870d99437a0 : SUCCESS in 57m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f03e3dac1cf44458a30c3d6858e5175 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":3},{"id":"b0dc1deee7a773e3b3da77952eee9c28aa5d5ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-10 17:05:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"02d3d289cde0762777239ec81a68636e3a25ceb7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-10 17:28:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"82045459117fc7a925c7a1b7dca2c5616adeca20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 18:17:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c34384ed514d879c7fbce993835311 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/76efa6ef5b9240c68719b067b98215e3 : SUCCESS in 57m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23cfc93366994dcda331846dd715fe50 : SUCCESS in 42m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19d34b87dfff4f74943b20784cf43361 : SUCCESS in 49m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08169f35067c490c99db9b90d33d2908 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7dc955f6a54b4a35b02dd0908466b0ea : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ce0658d08b8495682a217b7c6309dc3 : SUCCESS in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa1411524154a7c9336fa3b27d9c9f2 : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10a15ffbb8e94595a4e76f03e026b9ed : SUCCESS in 17m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/274eed37bf3c44ee8b0536f472dca4bc : SUCCESS in 21m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c93c41e79feb4eacbacc5c95997ee397 : SUCCESS in 21m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b45f947a51394a22b09a2655417c6bf1 : SUCCESS in 23m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f19329ff171f45a4b877935efc863106 : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/779e1ee7ea7443e786aab4b57020df8b : SUCCESS in 38m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6c2bababbfe14549827106319bdfc222 : SUCCESS in 24m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b75dec781f3f4e37896019711139fed9 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acc6259a3b394fc3ba837357ef243e03 : SUCCESS in 20m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f9c6dbe2c2ac47c89c93af74a8cc81e1 : SUCCESS in 50m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed6f9b6718ce44cfb8b81bbd1b2b6f55 : SUCCESS in 52m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3525d37aeb894fa6ab644aaa511067a6 : SUCCESS in 43m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff51dbc0a6ed445b817808b2713326ec : SUCCESS in 52m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8655a25d96ff4cd8be4930a65871360a : SUCCESS in 36m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54087899c373455e832aef2dabc75850 : SUCCESS in 6m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/408a3e6be15342c98a73ed37277bd0d6 : SUCCESS in 22m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3353912eb0f4410f825613785b60256a : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":4},{"id":"18dabb1a677a824d6ae37037ca9a371695f3df04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 19:37:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07fe581bb0ab4156aa2f904922789e3b : SUCCESS in 39m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b5e5ba93db54c2899a4d00577b36546 : SUCCESS in 46m 17s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/1766785cd4a14a8bb43c68b2950013dd : SUCCESS in 2h 23m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b43054a3ab84c3db9c3baddb93a0b29 : SUCCESS in 44m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c62a6a4b6ac940a2b5187f7f2a9d321d : SUCCESS in 48m 56s","accounts_in_message":[],"_revision_number":4},{"id":"b8474a25a3b562d9c201d82b4460f8de71f07d0e","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2022-01-21 08:06:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"752337fa717e9aff4b6be5dcf7a6a660188fc454","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-28 01:15:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d03dc7ace63079a55779da4b2eb1daab801c00b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:17:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dc7208cde7a8ffc8ef3a260f8e05ab8888207abf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:15:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d525b4e021264095ab1e831404675f69 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0cca4638f2d5477db3c7970a39d19133 : SUCCESS in 46m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac712bb8556d4fddb6859b3f0933f143 : SUCCESS in 46m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5683445b522b45f89c522d070081d97a : SUCCESS in 31m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd054e16368640db9d44077d942902e7 : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/70597ed7289f476a82ecc9242e6f56c9 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55a40ae438d449528b0732f7ca2c0a06 : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a81b3f619ed348208b1f4b79060276b5 : SUCCESS in 14m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb2f8dcd29c64c7a909e6f5649f6a604 : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21a3974fa42b405489d8aecb0426fdf8 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/298240872ab2480e870300082af89cdd : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9798301267464ba78bedd15b783e6614 : SUCCESS in 21m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc316bea34b04528bf5c084375a5fc8d : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c573460114fa4596b2d7f87c1884c7bc : SUCCESS in 12m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4bf569482ce4b31835d5cebd60b7245 : SUCCESS in 13m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db27fb929c734bd1b3fa834ea7caeaf5 : SUCCESS in 6m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/356e93063c9a4b44b21df0eed5ce7286 : SUCCESS in 43m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b70085e557ed475581922ac89f5f9be0 : SUCCESS in 56m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bde98b53902b47a69a7bcf03fbb7b2d5 : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9610cd8972a4152996be0fd234fcc2a : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0551f304fbac4de19bbf12eba7d3e084 : SUCCESS in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce3cd4664024021b19672089ea82215 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4301f906425849cd8dd0f93e2bb4f1ea : SUCCESS in 26m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/953373825d5f48058eb1a54a9163275e : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":4},{"id":"b795d6199d735b6dc560eb0a97d43d8e0763581c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:15:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c2043dd61ab140ca4a128ddbe7be742483bb2a50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 02:31:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e15e72ab34413dae455bb8b2785d36 : SUCCESS in 1m 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7fcd5e55827b43f9815df7302d93e1eb : SUCCESS in 1m 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6bd73dd61955450a850f23dc4179cba6 : SUCCESS in 2m 08s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1907594bd8b63b48c86fcb0dc95105484d97bbf8","revisions":{"69eb52a2e8689ee6cff936309ad479f7e3b6ebb0":{"kind":"REWORK","_number":1,"created":"2022-01-05 04:42:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/823473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/823473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/823473/1"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 02:45:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 04:41:26.000000000","tz":-480},"subject":"reconstructor: Abort just the changed policies","message":"reconstructor: Abort just the changed policies\n\nWe\u0027ve already walked the disks looking for work, may as well continue\nwith the work that\u0027s definitely still valid.\n\nChange-Id: I4c33ed5f5a66d89d259761b5ce12fb6652b28c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69eb52a2e8689ee6cff936309ad479f7e3b6ebb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69eb52a2e8689ee6cff936309ad479f7e3b6ebb0"}]},"branch":"refs/heads/master"},"87f52218791c3988144bd6317f2bc5709153ecfc":{"kind":"REWORK","_number":2,"created":"2022-01-05 23:55:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/823473/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/823473/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/823473/2"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 02:45:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 23:54:32.000000000","tz":-480},"subject":"reconstructor: Abort just the changed policies","message":"reconstructor: Abort just the changed policies\n\nWe\u0027ve already walked the disks looking for work, may as well continue\nwith the work that\u0027s definitely still valid.\n\nChange-Id: I4c33ed5f5a66d89d259761b5ce12fb6652b28c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87f52218791c3988144bd6317f2bc5709153ecfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87f52218791c3988144bd6317f2bc5709153ecfc"}]},"branch":"refs/heads/master"},"26b425741083146e4e923205f4d2e9420567dc1a":{"kind":"REWORK","_number":3,"created":"2022-01-06 20:36:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/823473/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/823473/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/823473/3"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 02:45:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 20:36:01.000000000","tz":-480},"subject":"reconstructor: Abort just the changed policies","message":"reconstructor: Abort just the changed policies\n\nWe\u0027ve already walked the disks looking for work, may as well continue\nwith the work that\u0027s definitely still valid.\n\nChange-Id: I4c33ed5f5a66d89d259761b5ce12fb6652b28c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b425741083146e4e923205f4d2e9420567dc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b425741083146e4e923205f4d2e9420567dc1a"}]},"branch":"refs/heads/master"},"1907594bd8b63b48c86fcb0dc95105484d97bbf8":{"kind":"REWORK","_number":4,"created":"2022-01-10 17:05:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/823473/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/823473/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/823473/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/823473/4"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-05 02:45:44.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-10 17:05:28.000000000","tz":-480},"subject":"reconstructor: Abort just the changed policies","message":"reconstructor: Abort just the changed policies\n\nWe\u0027ve already walked the disks looking for work, may as well continue\nwith the work that\u0027s definitely still valid.\n\nChange-Id: I4c33ed5f5a66d89d259761b5ce12fb6652b28c40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1907594bd8b63b48c86fcb0dc95105484d97bbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1907594bd8b63b48c86fcb0dc95105484d97bbf8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
