)]}'
{"id":"openstack%2Fswift~823725","triplet_id":"openstack%2Fswift~master~Ib29816c4626bb0d914929783bd676e8b6cb19bbf","project":"openstack/swift","branch":"master","topic":"fake-mock-or-mock-fake","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-01-11 01:46:03.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-01-10 09:53:27.000000000","reason":"Alistair Coles replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-11 01:46:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib29816c4626bb0d914929783bd676e8b6cb19bbf","subject":"tests: Improve FakeMemcache call tracking","status":"MERGED","created":"2022-01-06 18:11:07.000000000","updated":"2022-01-11 01:47:48.000000000","submitted":"2022-01-11 01:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823725-fake-mock-or-mock-fake","meta_rev_id":"69ddc81ff18b91509d899deb454f7525184f3183","_number":823725,"virtual_id_number":823725,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 01:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-01-10 09:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2022-01-10 18:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2022-01-10 18:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 20:41:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 22:02:49.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2022-01-07 16:33:11.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-01-07 18:09:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-10 09:53:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"79637504c2319702e8805c6be3ca329b197218a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 18:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c41c4b463669e4883f6bcb033f1871b0e3ec031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 19:05:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d6f9bfdc1a28f826729f61135f5b4dc9dba63f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 20:41:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50edde5dc3c64dadbc8c5443f4776549 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9299259824594912b357a21dcb163803 : SUCCESS in 46m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e295ea28c49046fe88a5a04e703d66c9 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9229a6f5478a4ce5bf5899d585b524b5 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/98dc07ed7c6748c0a6150d8bb81ae35f : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a5e3a5d17e4eae8803da9a99b945cb : TIMED_OUT in 1h 31m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/841c7f3f40bb4bdb9b8e77517a430477 : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a99adefe1e7425da5a0c0fd4758de7e : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a7b7162cd8540fc981ff1ce70085c70 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa97810ef1944fceb2eadcf314d33fd7 : SUCCESS in 22m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf1552d2c41845a4984ddc343bd3e825 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d9dd855146e48b9a602b3cb3ad55be6 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46970e560373449e8f71bf8aa35fd124 : SUCCESS in 19m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4954a5b74ef416d884016e72b844133 : SUCCESS in 40m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58f801e9579d4966bda5f5bb53ae41d2 : SUCCESS in 41m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ce1f3fa47e4ee5bf487aeb42bf12d1 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75864b9b26614bd981af4d3cf880851c : SUCCESS in 7m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69651e2a1d6140f081a85116efcc53f8 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bfbeda531c4d619431454db754bcd4c7b43142a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:41:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39d4f43ca98c43d69f757696773a08ed : SUCCESS in 46m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6a4a3a6373a4c90b46a267354479396 : SUCCESS in 1h 05m 46s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5525822c4e8f4076b4611abd3ebe6edc : SUCCESS in 2h 16m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e764c74dec264de4847585b58e7f7c90 : SUCCESS in 54m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0591101a540945d7952ae9b5ca3586ec : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":2},{"id":"9c085095a896f5241a25247f79811ce756eb2c85","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-01-06 22:02:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8a89a5fc0bf1760f89453c592f2ed15a6c1453c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-07 16:33:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8a566b5c301918c47af643273b3ee0adb41329fa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-07 18:09:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"680391a150f065af2a09cc1559aab00d6a23a6cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-07 19:35:37.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"982fc089f90d9677ffc4f1257372ead86d917d26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-07 21:09:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a12176e6eb941889cae1d96fb57ffd5eea039f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:02:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e940de002684eaaa7a8daf13a648a81 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8540b04a25604fad9b08eb7216565dd1 : SUCCESS in 49m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6373f931a8324e779aca0b6276861d0b : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce759db9b59a42d1bd025bb77372cccc : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b36bba9d6906416fa88f576aad56d5c3 : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b011aab523946199d98f6a39782397f : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0655a6926809418ea2af70263878de2b : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/137143a6a6264b308acc53dc1977b176 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef750e4e84c74d648f60d96232276b32 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/239b683620cb4957b710f8ab35a0d39f : SUCCESS in 22m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/177d720dbd9747949f420b9de70ab594 : SUCCESS in 23m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/173141b776e3405face8f7fde5e4ef9d : SUCCESS in 20m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/816e82d95afa471f8917d876b92151d7 : SUCCESS in 16m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5a829638e7d4b4794725c168a909420 : SUCCESS in 40m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c21cf4f468214e2893d262ecb40e94cf : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a839011120434ed8947cc8800b30de04 : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879d1f59d28b44c1960aa7682ec5caca : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43ab46201ddd40ac9b7924171d20280c : SUCCESS in 24m 31s","accounts_in_message":[],"_revision_number":3},{"id":"4407a40cf27b37004653f5fed31bf8377928619c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:27:37.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5efbf4331504ee2b7a7e1c6deb799e9 : TIMED_OUT in 1h 31m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a7d0d9a67b649a2a6bca628fbcd0c54 : SUCCESS in 48m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a2ea9376999e4cef95acfe81c7fdffe0 : SUCCESS in 2h 14m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1774e121700e4b33952bfe6858dbe074 : SUCCESS in 1h 02m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e79853c52f34db2b007be8fba7f40bd : SUCCESS in 48m 11s","accounts_in_message":[],"_revision_number":3},{"id":"bf8b8f09ce48f2c0930b95f0a96aa175c2ef5fb0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-10 09:53:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"33077f644948e8f094004b4fa5540c2b14387d62","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-10 18:56:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54b62f81303e6f30e4bad658839cb008ea8b3b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 21:07:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c70094ee553d10fdfcfc984e7d41de5f8cbdf6a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:46:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c2aa199b9c4f9db1d95f0b86f2e633 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c4c97e6048a4c87a684fce10fbcbbea : SUCCESS in 47m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0d1d2f4c71ce42b1982d0b5f1abead7d : SUCCESS in 10m 48s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c5746f42964043918dab049abd0118a7 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37435c078f914b7db6e224f0cf989d59 : SUCCESS in 17m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70d43743c555456d9f6b420a0ca7c8ab : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97f879ea8d66436e941065eb51f61043 : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/695bed383bc24a05865b48fec4224442 : SUCCESS in 23m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4fb1905799cb44f6b4bf7376162d6fd0 : SUCCESS in 22m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08d216f259d34ec0bc80b87914fbb96a : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71cd209f4c134352b5d6b17ee25b3ead : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed05b870ffe64426aa5f955dff2a0fb9 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fe46f40d38614f5aa6a0afd9722c771d : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab34b2d8396d43c1a89e3f400980677f : SUCCESS in 7m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5679e341aed34f81af65acc52ff5e775 : SUCCESS in 38m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3dce0819cbee408c821ae103983731d0 : SUCCESS in 54m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ead6cf94b4d45eab74273435e96803a : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27529a2e8d254817ac5db266964ae734 : SUCCESS in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe6b8ef19a13457fa89e9c1280321420 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"64914a02da086c744bfcdf32fab41ff9634eb4e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:46:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"69ddc81ff18b91509d899deb454f7525184f3183","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 01:47:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f7565515094ebc9055e754449e6caa : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d39fa1186c704325b674291de6e08075 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3cef0a7398824aff9b3f906faa4dab71 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"874a5865b8e7a1ffe253fcab5b5eb7febfaa4cbf","revisions":{"ffe1c1d95fe4baa5e2b08803de1e34639ef65fc5":{"kind":"REWORK","_number":1,"created":"2022-01-06 18:11:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/823725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/823725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/823725/1"}}},"commit":{"parents":[{"commit":"c2dfe769ff400d82a54b01f0fefc6338b41e145d","subject":"tests: Unify FakeMemcaches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2dfe769ff400d82a54b01f0fefc6338b41e145d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 18:10:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 18:10:40.000000000","tz":-480},"subject":"test: Improve FakeMemcache call tracking","message":"test: Improve FakeMemcache call tracking\n\nMake it much more like mock tracking, so we can easily add new kwargs.\n\nChange-Id: Ib29816c4626bb0d914929783bd676e8b6cb19bbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffe1c1d95fe4baa5e2b08803de1e34639ef65fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffe1c1d95fe4baa5e2b08803de1e34639ef65fc5"}]},"branch":"refs/heads/master"},"764c6c939e975f3c59b613615da329d51157f6d1":{"kind":"REWORK","_number":2,"created":"2022-01-06 19:05:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/823725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/823725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/823725/2"}}},"commit":{"parents":[{"commit":"f7101f3795edb41e8825fc87ae78021191aaa622","subject":"tests: Unify FakeMemcaches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7101f3795edb41e8825fc87ae78021191aaa622"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 18:10:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 18:18:02.000000000","tz":-480},"subject":"tests: Improve FakeMemcache call tracking","message":"tests: Improve FakeMemcache call tracking\n\nMake it much more like mock tracking, so we can easily add new kwargs.\n\nChange-Id: Ib29816c4626bb0d914929783bd676e8b6cb19bbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/764c6c939e975f3c59b613615da329d51157f6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/764c6c939e975f3c59b613615da329d51157f6d1"}]},"branch":"refs/heads/master"},"874a5865b8e7a1ffe253fcab5b5eb7febfaa4cbf":{"kind":"REWORK","_number":3,"created":"2022-01-07 21:09:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/823725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/823725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/823725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/823725/3"}}},"commit":{"parents":[{"commit":"f7101f3795edb41e8825fc87ae78021191aaa622","subject":"tests: Unify FakeMemcaches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7101f3795edb41e8825fc87ae78021191aaa622"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-06 18:10:40.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-07 21:09:43.000000000","tz":-480},"subject":"tests: Improve FakeMemcache call tracking","message":"tests: Improve FakeMemcache call tracking\n\nMake it much more like mock tracking, so we can easily add new kwargs.\n\nChange-Id: Ib29816c4626bb0d914929783bd676e8b6cb19bbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/874a5865b8e7a1ffe253fcab5b5eb7febfaa4cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/874a5865b8e7a1ffe253fcab5b5eb7febfaa4cbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
