)]}'
{"id":"openstack%2Fswift~824471","triplet_id":"openstack%2Fswift~master~I6f8354e33828f3969fe76d865fe5a8ea35866e33","project":"openstack/swift","branch":"master","topic":"bug/1883324","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-01-24 16:23:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6f8354e33828f3969fe76d865fe5a8ea35866e33","subject":"Get rid of pipeline_property (harder)","status":"ABANDONED","created":"2022-01-12 17:34:30.000000000","updated":"2022-01-24 16:23:11.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ac423ebc5867be0b4a4970bf8a7232ac661d55bd","_number":824471,"virtual_id_number":824471,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-12 20:40:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-01-12 23:23:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-12 18:01:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 18:01:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-01-12 19:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 20:40:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-12 23:21:17.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"7b08a61a36bf8964699d35accdd7905dc4953518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-12 17:34:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf2ebb47d5eda830a9682b70dd83f1ab9b12cbc0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-12 18:01:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c5506b7612e7ea68ad0762dfe3a1f762dcab7523","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 19:29:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fec212a9aaf4c9ab2fffd3da4aba9aa : FAILURE in 39m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/000817828e9c4653ad39cf8533d6b818 : TIMED_OUT in 1h 32m 26s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/46638094036c435ea8881ff9ed96e309 : SUCCESS in 1h 52m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/914c1936951247bc8d0712722f52b5e7 : SUCCESS in 40m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1f3cd4c18e043d5b2a68d93abc70ea5 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":1},{"id":"348dc05f2677a5f2f38c701b51aba2bdfb9c2498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 20:40:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9711d92dc3a4378a8dba18fab700f71 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/adcd80da11fb483d9f424f6b0744ce0d : SUCCESS in 55m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d588c08bae0485481af93b770e39cd7 : SUCCESS in 49m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1b0394992a64e4391e27a8c682d330d : SUCCESS in 48m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/276625619c9e4171a7b3ad058b05cf43 : SUCCESS in 12m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ea7666ec5bd44b08789068aa11b6da2 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ae265b6603440c1b979f9bb03dc8519 : FAILURE in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc01b13f66db46f99334dfbf5a6af347 : FAILURE in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11ec6644e3e6439ba92f7571ee4a6345 : FAILURE in 15m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3bc326c63d440d8a0e0fc17c3da7ea8 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eda1c258064b4a3a9732b3cfbbdea4d4 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51e7a934444242858e652f9f8825947c : SUCCESS in 19m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a655c4d56fd4bb9bec8f17bb67dc0d8 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa912f04805947379b5eb40bbe0f9d04 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8a5086aa8839496e88948e9e3ff3ecd5 : SUCCESS in 22m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ff4b3ce0b844dcf9c111f1544ae37c8 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8666b25f6169418ca0ca8a1c70554000 : SUCCESS in 36m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55b89fa31a0f4bfe8314f866fe72fe8a : SUCCESS in 1h 13m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2ac11b7c2164eef90ba6f54b7599aea : SUCCESS in 1h 08m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01ef5b7ba7e64e6ba77b3fcf7e596243 : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8806433718d941008f6e28a32743300b : SUCCESS in 48m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c78690c46ac4038b65b6398d6c11eaf : FAILURE in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88192d1dc9ee48b38ef5b895527b2179 : SUCCESS in 10m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f40e429ea46949c5ab55e3c31b980059 : SUCCESS in 31m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a76157e23ce44bb8c4724254ff52785 : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ecb1de14d70215c8733cabc8338d0fe5ddf96c49","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-12 23:21:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8c719d237c889a6e50c094b183b9757c3f59378","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-12 23:23:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"edd3923a5c548024462c284ec1e7937f634cd83d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-12 23:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c772167ab3ba153c36eb561957a596d406ba4b10","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-13 01:46:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac423ebc5867be0b4a4970bf8a7232ac661d55bd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-24 16:23:11.000000000","message":"Abandoned\n\nsquashed to https://review.opendev.org/c/openstack/swift/+/821920/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd5ba8e601c9cf3b1f8cd5d866b42276ae7ae713","revisions":{"bd5ba8e601c9cf3b1f8cd5d866b42276ae7ae713":{"kind":"REWORK","_number":1,"created":"2022-01-12 17:34:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/71/824471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/824471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/824471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/824471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/824471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/824471/1"}}},"commit":{"parents":[{"commit":"079abc5f71d8932ccd3b10f2c9610597c734f433","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/079abc5f71d8932ccd3b10f2c9610597c734f433"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-12 17:32:25.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-12 17:32:25.000000000","tz":-360},"subject":"Get rid of pipeline_property (harder)","message":"Get rid of pipeline_property (harder)\n\nChange-Id: I6f8354e33828f3969fe76d865fe5a8ea35866e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd5ba8e601c9cf3b1f8cd5d866b42276ae7ae713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd5ba8e601c9cf3b1f8cd5d866b42276ae7ae713"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
