)]}'
{"id":"openstack%2Fswift~824732","triplet_id":"openstack%2Fswift~master~I37260f81e86331ae666bb7e69f622cf663f725a8","project":"openstack/swift","branch":"master","topic":"bug/1883324","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-01-24 16:22:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I37260f81e86331ae666bb7e69f622cf663f725a8","subject":"Do the obvoius thing with final app","status":"ABANDONED","created":"2022-01-14 16:38:34.000000000","updated":"2022-01-24 16:22:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c65d2abc5a502df4c1e3fa7818b91e620a5ff82d","_number":824732,"virtual_id_number":824732,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-14 18:15:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-14 18:15:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b62f443cbc6d9efd024fbe8b5555e509373bdb49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-14 16:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7089509bd689bca6d4c85097e3baf01555ab269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-14 16:59:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ef2537fd5b20e2f56d90e3fba871ecdb940e6a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 18:15:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6e3d846f5d48319b21f46ee7f733bb : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/221259a76e8a41de9dfb1da066b4f784 : SUCCESS in 56m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb036c8645294c0297ebdde2ee75f40a : SUCCESS in 49m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a32dd4548af4915a383e39bb037c938 : SUCCESS in 50m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e932cdecdb58481db942bc0f611a6500 : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/923d833570c0493aabc42114ad45f341 : SUCCESS in 7m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccf9c29bad204bfc80b111fba153cd00 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ae5b5bcfe09475a9fa9d592b980da02 : SUCCESS in 15m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7daf9762427547b997ff487daa93f8c5 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e4ab05c28e647a38f0317eed31fc57f : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e080d03e3d7452898ed262a0c16dbf7 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0e30f76d52c4a87bcb4d64c4acc8bb6 : SUCCESS in 26m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a3b8a0c8c47b4abeb2db35f3ed63d3d6 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/36b3f81a5b114ef688851c3085790521 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1293f54c7bab4ba68511eadbe245af64 : SUCCESS in 32m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01163752bbbd41a6b08c949af1600775 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09c40f5ca3df47f5862245d4ab7089a9 : SUCCESS in 20m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e5f1e14a0fe46aeb1a5b1a6b69d13d6 : SUCCESS in 49m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9762f9cd18024b6f86b56c5d4187ae09 : SUCCESS in 53m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b56eb65e5c545a0a7fdf5383e862ae5 : SUCCESS in 41m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d14318d70884a31bed2e2a0a91d0742 : SUCCESS in 39m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ed50586909044509e4f8ad7fcbb09ff : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ed3c0c2999409f96cb7511c1ad9f5a : FAILURE in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c182f3860184fb894e30908befac57b : SUCCESS in 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c74cec13a17a4d8c8d8474a03221b038 : SUCCESS in 50m 38s\n\nWarning:\n  Comments left for invalid file swift/proxy/controllers/container.py","accounts_in_message":[],"_revision_number":2},{"id":"6b22a14129c6283f24873ef17864113c6bba760f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 20:09:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68994ffc6f20413693dff5a3f0f448e1 : SUCCESS in 1h 10m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f3a5419f22d405d92f9ce644d704385 : SUCCESS in 45m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c7fb4626f0e641a0a8cd767bf0431f0c : FAILURE in 2h 47m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/75e522bf6df14c7f84c0ea88f8c8f58b : FAILURE in 1h 27m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e25aa402afcd43b4b0c4fd2e97e1377a : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c65d2abc5a502df4c1e3fa7818b91e620a5ff82d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-24 16:22:19.000000000","message":"Abandoned\n\nsquashed to https://review.opendev.org/c/openstack/swift/+/821920/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92d6f07e711eb7b79079a95175ab455897ec576b","revisions":{"4d2933c178bc2ace8bb00fb0b41328a0f6d57dc2":{"kind":"REWORK","_number":1,"created":"2022-01-14 16:38:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/824732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/824732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/824732/1"}}},"commit":{"parents":[{"commit":"ab8922ff88f1405984c247c3d84f4d3c353fe13a","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab8922ff88f1405984c247c3d84f4d3c353fe13a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 16:36:49.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 16:36:49.000000000","tz":-360},"subject":"Do the obvoius thing with final app","message":"Do the obvoius thing with final app\n\n... and fix up the bad fakes with some new helpers\n... and drop mocking loadapp when passing in a fake will do\n\nChange-Id: I37260f81e86331ae666bb7e69f622cf663f725a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d2933c178bc2ace8bb00fb0b41328a0f6d57dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d2933c178bc2ace8bb00fb0b41328a0f6d57dc2"}]},"branch":"refs/heads/master"},"92d6f07e711eb7b79079a95175ab455897ec576b":{"kind":"REWORK","_number":2,"created":"2022-01-14 16:59:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/32/824732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/824732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/824732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/824732/2"}}},"commit":{"parents":[{"commit":"ab8922ff88f1405984c247c3d84f4d3c353fe13a","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab8922ff88f1405984c247c3d84f4d3c353fe13a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 16:36:49.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-14 16:59:15.000000000","tz":-360},"subject":"Do the obvoius thing with final app","message":"Do the obvoius thing with final app\n\n... and fix up the bad fakes with some new helpers\n... and drop mocking loadapp when passing in a fake will do\n\nChange-Id: I37260f81e86331ae666bb7e69f622cf663f725a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d6f07e711eb7b79079a95175ab455897ec576b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d6f07e711eb7b79079a95175ab455897ec576b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
