)]}'
{"id":"openstack%2Fswift~825172","triplet_id":"openstack%2Fswift~master~Ice17227b47871495003dab319a2be7baf1b189f1","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ice17227b47871495003dab319a2be7baf1b189f1","subject":"Handle EC disconnect better","status":"NEW","created":"2022-01-18 21:02:02.000000000","updated":"2022-01-19 00:06:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a498845d0ca792ebab1bef609853a77e688e6f80","_number":825172,"virtual_id_number":825172,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-18 22:04:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 22:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d382fd143ec866f82fa8e4808d5638b0298a7f00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-18 21:02:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ff79cb505af0cf9d4a2cfbda944c10043c7b8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 22:04:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9557e3a27f104a8e8726f5bb81a26a6c : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d0f79780c12844fd88e3dc58d0b1141f : SUCCESS in 1h 00m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e73fcc1782184e61b522577406564701 : SUCCESS in 38m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93aeafb0409543b695b1ac110788eb34 : SUCCESS in 37m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5137fdbefdc0469c8cc507bba05e6489 : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ffc4b29fa80d42e5bcd9e49f80918a9b : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9af6406b88e94e5490d71dfb51033ecc : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66dfb092dd79492aac0817645d37bdf1 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c37f2c4421b41e3bac4046ac563764d : SUCCESS in 27m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f136b21ffb0c410783c3254bd00c0e31 : SUCCESS in 22m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3caaf1f8404645029afa8596541f8689 : SUCCESS in 22m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8b2163c24eb4be28d9bab12b985e161 : SUCCESS in 16m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/efd18793148e4183bebea9ef43b5a06c : SUCCESS in 23m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8d13fa89567749e4898c0688c2a52aa6 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/68e7ae34c2e44cd199d56768cf09af44 : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bfca185f66843728b214f9b97f38d56 : SUCCESS in 8m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1338dea67d514dcd94fb111f153c6bda : SUCCESS in 32m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5622b91f17448dc8acb0d9c4f4e8b01 : SUCCESS in 45m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa6e99fa24924322abf32aaa146a68ae : SUCCESS in 57m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8e94016605d4e02bc4512d7a74cf682 : SUCCESS in 46m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f81eabc7bf1a4cba85eb472e7f820f5a : SUCCESS in 41m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/febb45ac166b464396611a0ccb7c8518 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4580ba445ebc44af957ccc09125b9cdb : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/86b0c4b851ab49bd9b63cc2980bab9cd : SUCCESS in 29m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e89a80cec4d24255a7dc5944041a2505 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a498845d0ca792ebab1bef609853a77e688e6f80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 00:06:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1278e4c83513457e804b87be1d6bcc60 : SUCCESS in 55m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02a8194bca1d4983aaee89b40a326f86 : TIMED_OUT in 1h 31m 41s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6f2b051ab7974b819d44a5d859cf2e49 : TIMED_OUT in 3h 01m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ce303a2fd484467b60661d8b309f709 : SUCCESS in 59m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2755ebce89b944c88287097e7f896506 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63e5f7d6137be8a5967bd9d0ecb5e84563b3b320","revisions":{"63e5f7d6137be8a5967bd9d0ecb5e84563b3b320":{"kind":"REWORK","_number":1,"created":"2022-01-18 21:02:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/72/825172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/72/825172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/825172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/825172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/72/825172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/72/825172/1"}}},"commit":{"parents":[{"commit":"7be5cb0ab2fd73b552ad64b72d8a07e7006ec721","subject":"Error limit the correct node on ECFragGetter resume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7be5cb0ab2fd73b552ad64b72d8a07e7006ec721"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-18 20:59:34.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-18 20:59:34.000000000","tz":-360},"subject":"Handle EC disconnect better","message":"Handle EC disconnect better\n\nWe\u0027ve observed some sockets hung in CLOSE_WAIT with data sitting in the\nrecv_q buffer.  I was able to recreate this scenario in my dev\nenvironment by overloading my proxy with aggressively disconnecting\nclients.  What seemed to happen was the eventlet hub would service a\ntimeout from a backend node (and try to resume) while it was shutting\ndown the ECFragGetters after noticing the client disconnected.\n\nChange-Id: Ice17227b47871495003dab319a2be7baf1b189f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e5f7d6137be8a5967bd9d0ecb5e84563b3b320"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e5f7d6137be8a5967bd9d0ecb5e84563b3b320"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
