)]}'
{"id":"openstack%2Fswift~825335","triplet_id":"openstack%2Fswift~master~I0f5477f383e496ec94816439d404ad6a4cfc075b","project":"openstack/swift","branch":"master","topic":"p-fix-statsd-prefix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-02-04 08:50:33.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-04 08:50:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0f5477f383e496ec94816439d404ad6a4cfc075b","subject":"Make NodeIter require a logger argument","status":"MERGED","created":"2022-01-19 15:38:51.000000000","updated":"2022-02-04 12:33:25.000000000","submitted":"2022-02-04 08:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"825335-p-fix-statsd-prefix","meta_rev_id":"41b4e9079f09e093196c843543f38e6931024b1e","_number":825335,"virtual_id_number":825335,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 08:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 19:08:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-28 19:08:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 16:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-28 19:08:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c3ab365e4a78997f8c7b046eeaebcaa6184dbfce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-19 15:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b603c392c5066277f9df701b993baec0738a385","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-19 15:46:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cba791dc35d3221dc5fe2f64312550b6ea2c17ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 16:53:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e53046fee534e2dbd9376a0183c8db2 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b0d7b3b10d8d42f8b21f8838b7a4175a : SUCCESS in 1h 01m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9d8223fde1147dba82c28de0b96e9a1 : SUCCESS in 49m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f269e80eb7a4fcf92085155d8c717b0 : SUCCESS in 1h 07m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2dceaab840c240c7850c8052a6561fdf : SUCCESS in 11m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/98431661c7524b88b2a8b47fdc1b8f38 : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f28989680d3e4c10814f72d15017b9a3 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6a47f93c285409d8854ba49165d9564 : SUCCESS in 15m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a987fd42aa4deeb3532bfd1842b781 : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cdecd178d394f50a6725d341240851d : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/339982b948874873abe8b3b5c81afcbb : SUCCESS in 21m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b89dfd34ca0d44d4acb0c625c4a30797 : SUCCESS in 17m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4928f80f4f6c40d997d35d4162eb4521 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/864d8c3f4d0141b3a3214f1c10a67a6b : SUCCESS in 22m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d351bba8ce1c4d2eaf9b7f5a8f5f3f94 : SUCCESS in 19m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47e8f4f879ae4688be88045ce703fed2 : SUCCESS in 7m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a0a4414b2974b4db48a765a786b08f8 : SUCCESS in 40m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec588cc4462348c79cfbb293c95dd51c : SUCCESS in 50m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2656a4097dcd4d2fb64e0ef00ba1508a : SUCCESS in 52m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe359b5f511f472c9644a468be7e1fb4 : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d32c38ddab8d4ce6a41fb57ee2e92f91 : SUCCESS in 34m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44f8b7885922497b9cf78242af39de17 : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9266a6b75ea481096dd3568f2ae517f : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4072708a5994721a7d606733b961f6f : SUCCESS in 24m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b864a02051e444688b680f6ef235f328 : SUCCESS in 50m 37s","accounts_in_message":[],"_revision_number":1},{"id":"55015a91aa684d5b2d2ab31e6b22f61feef5b354","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 18:43:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f23c154e998e40d6bf3498bdcada432b : TIMED_OUT in 1h 31m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/754bae1e49c7460faecaf6199d8e7c5e : SUCCESS in 38m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5055f887e46e4e82a217d7ad6a9944e1 : TIMED_OUT in 3h 01m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/17532794bbf843859ebb86a04ce3d9e3 : SUCCESS in 42m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bec114faf2a046f2afcb7781095d444e : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2335ad74cca29f318180f0b6e8b1272fa0ec0756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 13:40:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"734ffe6f4b35fe30bb7bc199039bde8e2ad0dc32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 13:55:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8ffb82376f086858c1d029941e85d46a2758cc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 15:12:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5cdf28043a495f8f16d57521a4e590 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/fd146f34ee864b8184a9fe5030019c78 : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba8e4dc6fbe445a88dbf97bae5068133 : SUCCESS in 37m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d9dd31113604f9faeb5ed4be6e312de : SUCCESS in 49m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e5f129f635234a9cb5cb7489ed19787b : SUCCESS in 22m 43s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/265e18376ccb4d8dad8a390eba527557 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/08e1c52b9cc04b1dbd5ee3201fefe726 : SUCCESS in 27m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae68b28d1eb4450d9ea528ca8da1b288 : SUCCESS in 12m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/328d8f3ac0fb45479e3b42713ce2d06c : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc2c27286bad4341811aadd771858d62 : SUCCESS in 21m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7df9f34552148a682b731606be93689 : SUCCESS in 14m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e4ea036b04e946d2b08cee112cceeec6 : SUCCESS in 18m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f0a994866514c368e28db371fdadb63 : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51655fab08974b81b860a6866cfb275d : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/651becec3f574007853e92bd45b32390 : SUCCESS in 22m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71064ff020bf442f98cdd783261c01a0 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/30f40ff769af4233b6a83ebd5e29f4c4 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4836842861c2439fba9651764d99926d : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c180efdb4fe438781f3f6e3f9b558b7 : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e7be7697c37447c8366a2fa0ca3abbf : SUCCESS in 28m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea9d71b10909400786d86826a2b4abe8 : SUCCESS in 48m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9a4fefe5ee84dcf824af30ffa7c4875 : SUCCESS in 1h 04m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a36f17898da4e338079f58f6cdd1a95 : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e08500874f0c40319e0b2552ee98966c : SUCCESS in 44m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/840f25177c19420baeff8b1b226f04b6 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68daf5308ca498c8fe55e59ef6792a9 : SUCCESS in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9570144cfb5f43b6ab62c700ee1358ce : SUCCESS in 28m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3bd780263c24b92b1d7c7b938be1586 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":3},{"id":"883662b81f688b766ae7e86d60e632d9c2121374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-27 17:06:47.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c02e30456568230fc806bfc2cbe4458ce7ac9303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:21:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c1c5647f6e44d9a9e221ca45081f395 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/7dde09052af149b6b8b2298d7e842a0b : SUCCESS in 53m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68dbac8990464e4485ce81ebdee14e5e : SUCCESS in 44m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c187e52019dc4af6a2a2562141971daf : SUCCESS in 52m 34s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/43a44febefc34682b02e1885920406ce : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/39233b358c8940cf8049c517822ed98d : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/27b232f7dfb642839f5ebb8319bc9e89 : SUCCESS in 25m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26ef797abbe848bfa4cc2a6d7db98db5 : SUCCESS in 11m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/06336e835f1347f488755760c806f1df : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33db0113e6304653aa2563a7cb1032ef : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29b1243cdd2d4a3eac8ff060f94807d5 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/665bfde8bb0e4f968fb556e8068307a8 : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/163e08d1f08f405494bd083793d8ebd0 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ecbd4e1b2c984343a3a526c6641da80d : SUCCESS in 21m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3203f03d3cef4ae091ac79ccb05dfe7f : SUCCESS in 27m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50be9c427dab4da180a5f83e39e54d61 : SUCCESS in 24m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8f20346178348acaf1bd8c9190c3e88 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5128d4b014e541e384c93e7ca18e8050 : SUCCESS in 19m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/211aaf1323c64e5eaa56b1c94e58fc7c : SUCCESS in 9m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2e71121b0144d97b571065a6123519a : SUCCESS in 29m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d52df77ecf89487cbde103195faa27b3 : SUCCESS in 59m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3703fdcbf277431cb4e9bd7702965bf3 : FAILURE in 49m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/197dc4ba1acd4dc1b7ab9b93675ed918 : SUCCESS in 34m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb589a1b001840f983c4ce1680e5a2ff : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab53d11685014e548f6ea7d4c473ca45 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce4c169685de494bba237ec6e5d4e75b : SUCCESS in 7m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7274e6e66e0a4621807b3b3a29a8e1a8 : SUCCESS in 24m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffdb530770974cb2a7ebcf7720be027c : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":4},{"id":"6ef0dacca3cb7b14c35fbbafbf6e5d9d9042ed25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 21:55:12.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f39dfa6672a74cd38d2b676e3ef30c70 : SUCCESS in 38m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8a6ab707e004ea5a26f06e1e8fe23af : SUCCESS in 1h 05m 09s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8fd7e5e4b3d64ff2a50132131822035c : TIMED_OUT in 3h 01m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c472620430fc40ce93a13e912cdf8f22 : SUCCESS in 1h 08m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/05f9040f3f0848f584bf35f43e99f359 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":4},{"id":"537a8ac43a3d113478ae645c68e86b1914644982","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-28 19:08:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6ea3280c86ec303e9b5adb98d492c92280256980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:38:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5383a95ef34a91a83d287e764f2e62 : SUCCESS in 13m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/2a2b2bdca886432bb3f958911949ded0 : SUCCESS in 52m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a6645421aea4e2abe42d7acd7d72a60 : SUCCESS in 46m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c48ff1a293db417e8cacc7f5922ebc47 : SUCCESS in 45m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d3d5a37f82bb4e4d9aa0f3be1ecc0ba1 : SUCCESS in 31m 28s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ab504bf3a0e64a2f8fda3b19d0fc9741 : SUCCESS in 31m 23s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/31113f058fe54f54846d9abd6560e2a7 : SUCCESS in 32m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/705e6384b5134ecfb0ec5c1864607cdb : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9c3e8a74320047d9b819c936aedc0ead : SUCCESS in 9m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66cc36b990cd42acadf5fecc499c4c97 : SUCCESS in 24m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e405363b008453aabbb8aab29047057 : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c7b15f6426894d7ea7f8269ab38f8577 : SUCCESS in 18m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fc0a492ca8345d393d2a9ebe8f11951 : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6ee310c679046eb80dfb32232b70546 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/413d109dcb5a4f8082f4b28d3ad12b10 : SUCCESS in 21m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d364cfe6c65e4af6a5b8173375541dec : SUCCESS in 28m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/976c3e1132e942c08cc576bbe9b4201b : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30349d078dbd494a8acd4a1cab7258e0 : SUCCESS in 28m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80686c4c39f2484d90191158695d8f6c : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a60a0b8d7874e049c132dfcf6f4069d : SUCCESS in 26m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d93cc0cfa9184a5583fc001a0f26a223 : SUCCESS in 1h 06m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/12888a97f5874640ac82b1db30d9751f : SUCCESS in 1h 03m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86a6775812a94123a15f9bf565a94799 : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dad601abaec94e22a1ddd21bbd650e68 : SUCCESS in 43m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2816d0ca7501405ebfc92a6e71d06db8 : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d79061c01e4994858152c92641dc55 : SUCCESS in 8m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/152f468ca535421aa25038069c5c114a : SUCCESS in 21m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/019f143ee4434c7fb7c0e677171d4acc : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":4},{"id":"3895f43936de1a6eb76e9e4db435c95c7a182b14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:32:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4c7cfd94b2037b267aa1d9800bdf78caa8cba4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:56:13.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"25da6ca23b6c6dfbc07496b684b8bce55661ed36","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 00:07:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0c37538d5f2b7e1761e90eee4eb7689433b096ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 01:08:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e91967042f6464095193431811f3e3b : SUCCESS in 7m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/55178cc9535e4071b9bc79bb73a2f002 : SUCCESS in 51m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a22e669152c48fb80ece75ad30b7f35 : SUCCESS in 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/840d7c6de29e4ecf93b90c1b3502dd4c : SUCCESS in 44m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4620c33d4c7948069ae18f2636cdf3b3 : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9afaee58eee04f489db2bf3c8f9211f9 : SUCCESS in 23m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b5211f61e1f04cf48b5b7a912eb904bb : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40be03d1b66f4a1d915643c4332992f9 : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f3adccc6d0b4d9991ad4bf809fd46dc : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f65547bb9f54a969dfa02b35e406210 : SUCCESS in 21m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f65ba8309a1478d847a95c8ce6db9be : SUCCESS in 19m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51e28895967640589cfd2229f2d40dd5 : SUCCESS in 16m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/401fad6f44df42c7a6ad5bf866805388 : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fa2e109b33a47359f44af06c2dafee7 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a5d8d392d664f198027442d55c46af4 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/67320217f4754275b544e80ab7f9b8e5 : SUCCESS in 21m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a11307c150324199a8cd159c5debf75b : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa9aa5ed7d654f85a67b40d79ab361b6 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/14880438e3c24bb1a596b9a30c562d06 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16cfd4e6232b4e75bd111bdd05e53dd3 : SUCCESS in 18m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/80e86d39d29143c781821acb1354b41a : SUCCESS in 57m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac3f8d230d22433fbce2114c847babe3 : SUCCESS in 49m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/854a76d45cb0466397f68e2f2d94303b : SUCCESS in 31m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9785c1718de4b028c983af7b617a33e : SUCCESS in 31m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d13c30bdec1641eaa89535110416c88e : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/761cd052c62d42b2bf835ee86a0a74c3 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0793b3ab06b84f0fb0b2eefb45d250fb : SUCCESS in 16m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9acf56164ddb4bc4b3bcd3ed82803dd0 : SUCCESS in 36m 46s","accounts_in_message":[],"_revision_number":4},{"id":"557a9a46061bd31fef0d43a2664b28e8c4c9755a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 01:53:27.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"39f04e962daa24f613057b44ac421dee96b65780","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 04:34:04.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b947452f2357453685e86cd46ad89f7d : SUCCESS in 46m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fba960cca514af680407e83cb906c09 : SUCCESS in 53m 22s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/346c6528d7574cc6adf5d68ed9c9d112 : TIMED_OUT in 3h 01m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fdea19c1a56461188cba754a594d148 : SUCCESS in 1h 17m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2d535f88f3f4514a48c8149885a4d05 : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":4},{"id":"4133944bb60aa563ea4d99b4ac7cb9c00bb6c5b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 06:37:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa35667e8912b48c874826d515851164b870c855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 07:42:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e815084938f4e3282f0a1de8fed60b2 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6adee0c2a5f441519c7d83f5e0166b38 : SUCCESS in 50m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6345302364694124a39db5bd0de6a0ff : SUCCESS in 47m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d609b00150646769ebd71d0834f6742 : SUCCESS in 47m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3bd0eeea229a4d809c36733c92277e67 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/acea549e4d514e3f8aa902d53250cd7b : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/203ee84189c44e5ab516bb5d720d9acf : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7870c86e9dfc474f9872fea232d0d5e9 : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7649b42044de4ec696cebf0ceba89540 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e137ad4519bc461cb1f620ad4fdf9d63 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf75f762edf947a18f1830177dff7932 : SUCCESS in 12m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b488936ceca40eebf903885a2e68e64 : SUCCESS in 12m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57f9be8d1bc34ffc9e0e759d895bd48e : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/898d046b81a8499cbf7b755ada23cfb3 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1c7f43e9e3646d1a8e39776ef60b255 : SUCCESS in 23m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6ae8c5517de40fba049b7fd4887bb32 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b46edc917f734ab7b6aef963457cde77 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54d6dd0d768547409f882d67764c3ef2 : SUCCESS in 18m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6eb97acf4b340cb846387a23e3d9204 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9404ea01b6c44a7853469c93c976d1a : SUCCESS in 20m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ced38b715b2244b482dedfad68399928 : SUCCESS in 48m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d64e2a9d1d1746f7ba407813fadbb36a : SUCCESS in 1h 00m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a29a9cc718a432695de14bad1cf0e3b : SUCCESS in 29m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4019a87e0f994b4295a01deb2fef9be4 : SUCCESS in 39m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdbec7fcfd434794aa8ac7713bed9d91 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385b4bbe618b4077bd7f9ae09bb7a292 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b14638f4d81408dbaa4dfdb0e19908e : SUCCESS in 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b3b97f7ba424722924afaaa74a2ad84 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":4},{"id":"c048caa9e977d8ddfc384685ace8d8828c9e371c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 07:45:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"06dd95426edb55a37b61cd873e8a9eb53e81cdda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:50:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abff8c765d14739809f414d6065df03 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3f88038ef9cc46e88d5bf2aec025dc6b : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2ffe4e0e50a48a3b37c474c162d6f50 : SUCCESS in 51m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06e2a64ce2bd47b1a1eb89583725aed4 : SUCCESS in 54m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/329314e1c4064e7ca35c471a0b040bec : SUCCESS in 8m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/edd8ac62d79d4eda8bb08f96779ee408 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/538a103cc1d14ae7ad835eeb4199e624 : SUCCESS in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb44c57b21984427a74b514f46e23bf0 : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1aeb7e1160d4c31898cec3424c3b8ba : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6279ea2900ee4e369a4ce6aeb262decb : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1614f4b7d7cd4cffbbfc4b51b366d76a : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/29c6f381867642d9a64cf0733f92ca6c : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5116b5340c9c4318bcb93dcf5af0fffe : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4afd534a638b443c8817031ac2b87354 : SUCCESS in 20m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/723873b28b2f4298b5a365a309054fa7 : SUCCESS in 18m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a5a03de15ac41a29ac844eb40dd8849 : SUCCESS in 6m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0492565a057b4c7aa898e3245024faca : SUCCESS in 48m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b3dd19023284dcfa9661568f0f9a380 : SUCCESS in 50m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a5f7a6846d7484fadbdf4333d5d1675 : SUCCESS in 29m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d74e03c062184bc0b452e442fecdc36a : SUCCESS in 30m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6713c2551d3a427eb41c34546df3fa36 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d555755bc64450fad942e9233397c74 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2d8a6b12948d4e909ed052a3d272f02f : SUCCESS in 20m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ccd2463f3dd4423854aa537ffc3d25b : SUCCESS in 44m 25s","accounts_in_message":[],"_revision_number":4},{"id":"36065c5213d992c425f56fa08d24b255e88a78e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:50:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e0704d5873a8c1ffb1a81873cd271e1c4197c7da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:53:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d409a7f84ef4f1bbb0f193994fdb9d8 : SUCCESS in 2m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/50cd5b0eafc243fc9bbbeced512c0f22 : SUCCESS in 1m 25s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a4dac506118248dabc96d6a571627ab5 : SUCCESS in 1m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"41b4e9079f09e093196c843543f38e6931024b1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:33:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ba13d892378437090402de67aa2d54a : SUCCESS in 1h 05m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53c8d5b1879a4fc2886b281bd8d1aef7 : SUCCESS in 1h 03m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3d97afadab6a4124947159155c1bc337 : TIMED_OUT in 3h 03m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f55cba44c8dc441aaa836a4244997d45 : TIMED_OUT in 1h 32m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41837e65305d4f059e57fb7c98f981b0 : TIMED_OUT in 1h 32m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a50726f27a1cf76339fb23253389bc58539979b7","revisions":{"c57f5849818fdc705e76cde625c33262aacccd0e":{"kind":"REWORK","_number":1,"created":"2022-01-19 15:38:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/825335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/825335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/825335/1"}}},"commit":{"parents":[{"commit":"091c12da42a9788fb900fbe0fcdc75ea76d4a51b","subject":"Fix statsd prefix mutation in proxy controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/091c12da42a9788fb900fbe0fcdc75ea76d4a51b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-19 15:28:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-19 15:28:18.000000000","tz":0},"subject":"Make NodeIter require a logger argument","message":"Make NodeIter require a logger argument\n\nChange-Id: I0f5477f383e496ec94816439d404ad6a4cfc075b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c57f5849818fdc705e76cde625c33262aacccd0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c57f5849818fdc705e76cde625c33262aacccd0e"}]},"branch":"refs/heads/master"},"ef6556bb5d9cd16955446d342c809066a390b607":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-27 13:40:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/825335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/825335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/825335/2"}}},"commit":{"parents":[{"commit":"e82c03df3af54996644a565d315e12ce1f3a2d7a","subject":"Fix statsd prefix mutation in proxy controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e82c03df3af54996644a565d315e12ce1f3a2d7a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-19 15:28:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 13:36:40.000000000","tz":0},"subject":"Make NodeIter require a logger argument","message":"Make NodeIter require a logger argument\n\nChange-Id: I0f5477f383e496ec94816439d404ad6a4cfc075b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef6556bb5d9cd16955446d342c809066a390b607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef6556bb5d9cd16955446d342c809066a390b607"}]},"branch":"refs/heads/master"},"0cb31487a1aebf409701b604629be2dcfa0c7b85":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-27 13:55:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/825335/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/825335/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/825335/3"}}},"commit":{"parents":[{"commit":"c3d5025090fc49dd7c9b212bf8d499e1f8d8e231","subject":"Fix statsd prefix mutation in proxy controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3d5025090fc49dd7c9b212bf8d499e1f8d8e231"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-19 15:28:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 13:52:26.000000000","tz":0},"subject":"Make NodeIter require a logger argument","message":"Make NodeIter require a logger argument\n\nChange-Id: I0f5477f383e496ec94816439d404ad6a4cfc075b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cb31487a1aebf409701b604629be2dcfa0c7b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cb31487a1aebf409701b604629be2dcfa0c7b85"}]},"branch":"refs/heads/master"},"a50726f27a1cf76339fb23253389bc58539979b7":{"kind":"NO_CHANGE","_number":4,"created":"2022-01-27 17:06:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/825335/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/825335/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/825335/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/825335/4"}}},"commit":{"parents":[{"commit":"6942b25cc109258804989c32fa3ffe2b536c6e77","subject":"Fix statsd prefix mutation in proxy controllers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6942b25cc109258804989c32fa3ffe2b536c6e77"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-19 15:28:18.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-27 17:06:22.000000000","tz":-480},"subject":"Make NodeIter require a logger argument","message":"Make NodeIter require a logger argument\n\nChange-Id: I0f5477f383e496ec94816439d404ad6a4cfc075b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a50726f27a1cf76339fb23253389bc58539979b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a50726f27a1cf76339fb23253389bc58539979b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
