)]}'
{"id":"openstack%2Fswift~826326","triplet_id":"openstack%2Fswift~master~Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f","project":"openstack/swift","branch":"master","topic":"p-fix-statsd-prefix","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-02-04 09:18:31.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-02-04 09:18:31.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-02-04 09:18:31.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-04 06:57:23.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f","subject":"proxy-server: add stats for backend shard_listing requests","status":"MERGED","created":"2022-01-25 16:53:44.000000000","updated":"2022-02-06 22:39:32.000000000","submitted":"2022-02-04 09:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"826326-p-fix-statsd-prefix","meta_rev_id":"87f642da89b0cb4c57e57ea8f301f0faf551143c","_number":826326,"virtual_id_number":826326,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 09:18:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 19:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-03 19:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 17:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-25 20:55:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-01-26 15:33:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-31 05:03:28.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"8758c7e0354b317cdbd43fdb401ae02ac5f1d6b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-25 16:53:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3c50720685904e722626046714a4e623408dc58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e4a3af085242549918678806b93ad1 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0a1d06212bec4d22990d0b5a48c5de14 : SUCCESS in 57m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/809828ad37e948bc8ce16ee5e55fefca : SUCCESS in 38m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78b9e7b2c5d5418c96ef459193fd376f : SUCCESS in 37m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71fafbbd327b4031889991c601f3cb79 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0dcf25c4e33f4f7082ac6f2fd2d5cdf5 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1adc8f073e74f37a06a210bb5fed4fe : SUCCESS in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/376715fc372c481991b863118a2ee860 : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5487865e032340609612db3198d23225 : SUCCESS in 18m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e80ce6df863049e98abc3efb5421eb20 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09769c2955f84cd58a0229778c39ee1d : SUCCESS in 20m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8f55c5bc42043129a9192d57c742567 : SUCCESS in 16m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4b19c74f937a4ba1840a561b3ab51d6c : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2835c3e8f52642a48d5585b4c3770522 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0ba4555849dd44b395d380fc69311449 : SUCCESS in 22m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8cd1c9953118486183a2871ec3457a82 : SUCCESS in 8m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5c4f85d676ab44b3949deaec89f5c0ce : SUCCESS in 18m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/400df1d1af264835ac2f6ecef4957e87 : SUCCESS in 54m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/be12d2c6c51f40fcaef897a4c5a95717 : SUCCESS in 50m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8b87ad79023c41c3a24424cc252b4885 : SUCCESS in 41m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be6e5edc52a14fccaed45661f46af832 : SUCCESS in 30m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63d867d6d0448ed9c0e1f318cb01f0b : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56197221510f444ba7d3987956865494 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6265bb3550ac444b93dff1f511405a98 : SUCCESS in 38m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b79c18dc8e4b415b91e254c61aae0d0d : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":1},{"id":"91c1c9a309533c9ad317f36714c4b1990fef826a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:00:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b18c357db0a64758a8ff13a56159d331 : SUCCESS in 33m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f4677248f8a458c99dea6056a423daf : SUCCESS in 38m 37s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/112293e6ba174edd822a74a106364123 : TIMED_OUT in 3h 01m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0a1d728884d44cfa6df038c1c95a6c5 : SUCCESS in 49m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/795b37533a424dc88c2ff38f01fdbd3c : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6dbb49a2159f7e69f5961ab5b7e974d6abb63af9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 20:55:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"870e60f4cf9cd2b736bc03b3b0756439de83baf3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 11:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68451c75a820b266c6f9eb08285e31950510b006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 11:53:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b404b511e7c130a6aada67bd1acf8d795ed68e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 13:30:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e595491f6a764bddbb7b7d4a9d882d60 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/35b25d5f838d4f0f820165c7ab33a190 : SUCCESS in 58m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26522729e7404abc9d3ebf353172de23 : SUCCESS in 48m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7951db3a9c134a67aaa26e874bcd5713 : SUCCESS in 52m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7128bffb9568496288195bc82a4b8976 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cca78bd060ec4a23bff4d6baf59ddd73 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/646945f7d4dd4134af66abcadcad777e : SUCCESS in 19m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0992cf7bd1e74c1c8277f9474ff07718 : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a499c5411eab416b92be3160c0b4ed6e : TIMED_OUT in 1h 31m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/495c6624de524e5badcf20f767cbb395 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28a9acef98384b7dbd89abcb9a360811 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a7e8626fa46447e94b0ddf4750adb15 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed55ace4a0c142b79bf41fd91c9d61da : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bc72e1cb4f6d44cd9d114762a1923888 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ce38940b9784d2cbac153f4889c6452 : SUCCESS in 19m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f62e9568be014e5d9b581404ab306f12 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/671530090954452dac7a77e91850f587 : SUCCESS in 24m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/527bb02396854bc3bb9186eea30d30ea : SUCCESS in 56m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5c4e7465aaa48c3b089bcbf533ac602 : SUCCESS in 54m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6073af76a4cb4a8bb54b4b0ead8020f3 : SUCCESS in 47m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bced13395e1c48a5b4105e1883874f0a : SUCCESS in 33m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3372dfb67d3245ba90625a3cf1f02414 : SUCCESS in 20m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c274b2674e1486599b68d4fb26629bf : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03e1215fe5af4b5a82127442e475a366 : SUCCESS in 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72ea5eed57184cebae04b12f0dbaa87e : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1862c4761930f6f854e13d56c1b5dcde24db2b7f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 13:34:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6636b6dc2592ad08a0fef329d240dd89e4a639f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 14:39:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/562dda4ffdd6487394d04424e20e15f6 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/e01c44cda0ae4f4996e3ea6aa6908c6c : SUCCESS in 46m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4122cec7ddc46aebb3cc7f3cb7786b6 : SUCCESS in 49m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3d1094939e746cea32d6259fde5a372 : SUCCESS in 40m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ce03406478944fdbed188643404d2c6 : SUCCESS in 10m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebc2b5c8b50d4a3989bae4058bbeb3b2 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd289a31614b46d49b2beff1aaec6bc1 : SUCCESS in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b3f6a060dfd469da0edff6119e4c8ab : SUCCESS in 17m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/db053c153d4744df88ca1f83c3caa937 : SUCCESS in 18m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/627c361a4d0843e88eb246f2eacc9699 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8303b6976ef4e70be8c21ca3a3698bf : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/135795cba4ef40ec826c53ff4b96cb7e : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c167ab9b5e224300a6706d13b66bd798 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a79b9ad9a0d74ee9a92d109c6e5a032f : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ecb88edc7cc64d23be2b6d6013cffe6d : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3fcbb702fa6540baada3fb88d791e4da : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/080bb245fbad4ec5b3ab129f6bacbbdb : SUCCESS in 21m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/62242f4bdfa64e868b6af752657978b1 : SUCCESS in 55m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4827843ccc874eb7a43d4cebf92a122d : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ef07bae3e7248de82e5699b93b9cab0 : SUCCESS in 45m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45cb864bcd0c496cb0f54f23a4cf18a0 : SUCCESS in 47m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/202d6c66c25847c3b949b6a58e82aa8b : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8478a41543f44e68b58460b47b84cd8 : SUCCESS in 8m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4e729af48954bfcab7beb59b9793993 : SUCCESS in 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff12fc72cc35440f99f60f67c4797aea : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8f6244ea913fa54f2795316f81caf369dc62fcdb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 15:01:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5270cc15cb824d8192be704659eaf6a6 : TIMED_OUT in 1h 31m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87bd4c0ab6d144598d19637e7dcaad46 : SUCCESS in 33m 12s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/db2b367753da43928ab20ff338ea5fe0 : TIMED_OUT in 3h 01m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37ab93b74621495b8bbf5319f3c5c3f0 : SUCCESS in 33m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba7c7dcff766483ab23aed82e17e02d7 : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c161f9d7c3b519463da9ad08f09793646c746846","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-26 15:33:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aa08a50252d2412cb752cad95f6b08aef4f34aa3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 16:26:04.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a37cb25c0710a953103fb948af79dda3f9b24cce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 16:26:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b32fbdd78bf4e5bcf7a0b092538d9e28b37dade1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 17:33:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caa316dfba04b2fa4343e92f35f98f2 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/8e29918cc718434e89243f9e482002ce : SUCCESS in 57m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f972a9aa64f4bfe966c987179a573ea : SUCCESS in 43m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19a852495db24180b1eec374393e94e9 : SUCCESS in 52m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/028a0e9658c84d77bdeecaadb0dc8668 : SUCCESS in 11m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/966d4f011f9146f2bd42bb8676a8d215 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9eda44284e4949a6ad51ee1ddcbfac69 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01c6c83b1aad482d95a97d4733a5eed8 : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d66c85f9dffd4d819264046938ac5290 : SUCCESS in 24m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6fa1f9bfcfc443a9bd705f7280413705 : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3084d813890f43d58b91a040cec06ddc : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8127f4e1dd6a412cbe741ef91871f4f5 : SUCCESS in 20m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6a7652bc51704e2888ff9d7971d51c5c : SUCCESS in 23m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a6f7a0bf34341439ec40dc038ff75a8 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eeba239b42da40e1932b1d689bc12da8 : SUCCESS in 24m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e52bfe1e5d914aa09e41ae84fb79218d : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e668dfdbc6bc4ec2a7cbcc9b5b7a83b5 : SUCCESS in 29m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/07dece481b154fafbf5dbb79d679c06d : FAILURE in 52m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/997e0d37d2264193bf3a9bd2a66556b6 : SUCCESS in 55m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcdb18067f5e4c1982f4b47d8ff7f5a5 : SUCCESS in 39m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9c2c97a36e0487d85b54f965a54e80c : SUCCESS in 43m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afb0c4e4067a43c79f5395359af892cb : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6cc3edfe3f3418a8e057b49cb5dbc3e : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ecd3fedf2df4f2584d9767feea95a14 : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ef801a178c041c49bab600459d7f242 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a7911765caf373aee4de9785aceb3ff51f5369f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 19:38:21.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42aeeb0002b04a47add8a344c06f6c20 : SUCCESS in 48m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc811dff0f794fee996e55554f8d9c5d : SUCCESS in 39m 29s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/418dd98737c74888b3d0561f4c0ddad9 : TIMED_OUT in 3h 01m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a09dbc9b9e441ebbb5d570d3f8d092a : SUCCESS in 46m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/584a9053b2fc40f0a6974a16731893ef : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ef257e1f2d9ec66218ae2db46d84e17c48829f4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-26 20:21:34.000000000","message":"Patch Set 3: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"aea8668d1ad73c99cbbd471ed69fc65229f5a2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 21:55:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98c583e50e7402b8dce572541efee36 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0a4681588b25410782cfe4c434479c65 : SUCCESS in 50m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0acdf4da5d947fe9aa54611da3542cb : SUCCESS in 47m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc48aa941678400387ba5cd68036b888 : SUCCESS in 42m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04b9628b94b64f25b7d065676e77d980 : SUCCESS in 10m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/62ce9d64409e4c0b9d37667b36c9352d : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0717b9aff394b31beb9d47fe0dcc822 : SUCCESS in 19m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30b1bc751dcf456fb1c0e48576a59be7 : SUCCESS in 23m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b2af7cd74284b6fba1aa143b95c09fb : SUCCESS in 21m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92a6f33658a04f8eb68d1252c39916e1 : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05d8d18c26c0408e9e662bf47c68058a : SUCCESS in 24m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a88c5d2c97c841feb44a034c523a0503 : SUCCESS in 24m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be97a25292e54307a19974c980f9e3e9 : SUCCESS in 19m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be54cf9a04c34b2db9b10d6d77fb3ab7 : SUCCESS in 21m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/904c5ad64b0d43ed9514b1e7c9fd2f77 : SUCCESS in 21m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec617bfa89c54b03949e65b2d49526d0 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71f00bbc24214e0f8dd8ab8d79a9295e : SUCCESS in 25m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/115094c91dbb4d069fedb9b76dbe3ff6 : SUCCESS in 51m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b80b17d75b8f438b9aa48ceecd6f791b : SUCCESS in 1h 00m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d2a4c15908e4b69829c5363f10ccc57 : SUCCESS in 45m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b5f30d9c77e4d98a9b0641639e5e4e4 : SUCCESS in 39m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f37c9b27ed384e51933bb4c65233c938 : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae6485e9187497987b05beacccc99ab : SUCCESS in 8m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6427da139ab84051b2b44fe97c385f61 : SUCCESS in 27m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4cff162b8c4408b9e5a96222bd9ca21 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bd3dd3eeb591e2d724f05e7f023480472263fb86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 00:31:17.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2536f1d030e045a492b57661f00247a6 : SUCCESS in 43m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7c51aeb87824608a9a7fc041052b6da : SUCCESS in 47m 40s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4eaf8bae37d14cd99edbc1dd23148efd : TIMED_OUT in 3h 03m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1060956ea4364a89b94006e5922a6172 : SUCCESS in 58m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf5d549eac274a0d8d102368bd79ed9b : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ee50aca678d974b4dd80dd24de707c3ba956d5c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 13:40:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"becfba6465298dacab03b49be4d2aab8265fd93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 13:55:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"feab5857ea76e2cba75da369a0ad3727a4e2e645","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 14:13:46.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ba8a623249c43dc18a6fd8b0fc11b0d728ee5460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 15:14:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9147d6ec74044ae69b95e4d63db07d09 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d6eb44b54e734299b3207068a1ad3e66 : SUCCESS in 59m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4240cfab327b41cdbf663a8d8bb6d260 : SUCCESS in 48m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0423e18861c143e9bb97f56e4c77f924 : SUCCESS in 58m 30s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/00b18d8c01344063ac954bfced8041d0 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d1de2de36f5f4fca9283e12676c9009e : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4e50d6bb290143a697b5d461c441596e : SUCCESS in 23m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b4ea391cea5461992f0c25bb86e189d : SUCCESS in 9m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2710ca9602c548c4898d868c519d039c : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed50030f796a4001a4aadd6d54e0ea82 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/60ee7e3381be4dd398193b376c8a13db : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/450ba843be9346178afcbac01b1a1c06 : SUCCESS in 19m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/183c46242678488991e1a594f4370d79 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e67bc7596a264137b5ae1b0af5fbcbd9 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10f96b5435944303a5ebfddf45447ee1 : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87d7a95922294ee3acda2861750247ef : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8df3dcb624e44b48a088ba3875da43b5 : SUCCESS in 20m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ba63f3f11294df393aa3def90bd3aad : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7e494f3f8a14a7088d8848d3f3c5c04 : SUCCESS in 8m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1e651ae5e304d3ca43b30ee6b16f1ad : SUCCESS in 18m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f9475ac73d2a45b2ad4e50a82f94abbb : SUCCESS in 1h 01m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/43585141a97648c8952efb2c36b30035 : SUCCESS in 47m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d682641bb51343759d927ac5b87066de : SUCCESS in 40m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3477abddd4c47e2a3a885f85096c33b : SUCCESS in 40m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e8f40bc0c83431ab1bcbfdefff54b81 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb550595154a4efb825c39aa394e4dd0 : SUCCESS in 7m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c16d32bd1e8c46eab4e71ee6ae9aa022 : SUCCESS in 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd8307cc9e2b4c9a9c69ed6106277d14 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":5},{"id":"90b321fd3b135f1996e5921c22a9fa79c0630403","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-27 17:06:47.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9e60be0458f8fa34425040bb7dd8bf49971a71e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:36:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b990e2e0a2a4121a4d5adb89d897f03 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/94319b28667a4389953a10f311720269 : SUCCESS in 51m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba039a5dc42347a18fde8cb2d661b425 : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76424c6c137b4e80ac3db76a34404b8e : SUCCESS in 51m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2096d8a30cbf4fa2a4bfdf913039e7b5 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e0c0ae04c8ca4d03ac658b03e0bab8dc : SUCCESS in 23m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/716173af35034ad99526fb94c9db6be8 : SUCCESS in 25m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94c5b58d694242b0ad75d95d034bfda2 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9fe5b4268dff44c1890d1ca1ac7747eb : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8f354537d9244b080c0e05142a199b6 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/576cd52c241946229a5f9b448976bda9 : SUCCESS in 30m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99775c28248c4f13b5aa852894b42567 : SUCCESS in 18m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c8539279cd14716918d2649785bca39 : SUCCESS in 23m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b76298dcfee4630ad0d88ad6d7e898f : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/afe5ada6cc6c40e5aaeafc61578765fc : SUCCESS in 29m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f8d76449e15c4eb89aeabccc19ac7dab : SUCCESS in 25m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/28bfaf96d8074714b538a5822ebeb471 : SUCCESS in 23m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2b3b705db07458ba49099c53388aeb5 : SUCCESS in 21m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1800a4330ab4e0491ba1dbe8cf30a08 : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dcd5abe01ab34ba49b0454c68239b6c9 : SUCCESS in 21m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ceac4f022d3420f85659555db31c925 : SUCCESS in 52m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3450fc7c734448b684ea94a07cd9ebe8 : SUCCESS in 1h 16m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d51cdeb5cfec47f68e918c7c6a44c7f4 : SUCCESS in 33m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/50ffc4a2dd8c4689a68c920e836b265b : SUCCESS in 35m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d0602deceda4e2a884c5502ad7353d5 : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf762fd764cc41338be49e059d0762e4 : SUCCESS in 8m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7912910137e0406a950eafb7f350c575 : SUCCESS in 29m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa405013652d43b082e11e1306ec4ba6 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":6},{"id":"9ec5f9223722c9ca4d83b9e2e4d148ecc9ff7b26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 21:55:38.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18e2f9483cdd4671a34d4e74bbdb53ac : SUCCESS in 40m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60409e0ac6c44255bc05b3f150fa411e : SUCCESS in 50m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/aa93bbe1ecae4743811488a98712f6f4 : TIMED_OUT in 3h 02m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6a7f26b1ef94e048e0539ae46a2db04 : SUCCESS in 31m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d71d1d1871d04289a068b8afccfde010 : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":6},{"id":"41f1f9d55249bf508fdd5cb819418f351f0606fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-28 19:01:27.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45e2ae79ac79ac7e6724448eeea3795565ddcb3e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-31 05:03:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b7c9bcf67aaf360ee945899697db1a041b320b8c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 19:42:38.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5df2225eb92fc32e820618039ea470edbec29ad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 19:43:41.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f838608483db4a9f5ad79da065574dc6fe24dfa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 20:56:14.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"131c58374e188d729544f4c1e71ba68d1e6a46f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 06:57:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"33a0088b6b34a8b15cc62379291164252fc6a041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:07:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd6501217124bae99cba79eada9ac23 : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/3337ab4d5ea44c6b915596a0856a2902 : SUCCESS in 42m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae7495a2384047328202a5b03eaf9a7e : SUCCESS in 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7ad835095b743c78cea1f84068923e7 : SUCCESS in 37m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/05a8199688e244dfaabca54511377d59 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/69ee105507d443078c7f7efd980394cf : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f3447ae5d7fb4f369e89b1ee2ebfb8a8 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/696ef4e69d594a0fb8b2f7cd091b92d2 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5bde2a9e998d432facafc6a53667c401 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d598d958c856466dbcf69501e176d44d : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e941cf15cf36481ab8eb85e3a4f40f91 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c07c186090e5417c8399ae8b54253c24 : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/60b1b7528bc8417c99a5eda172e7cda6 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7c7b93ac2994ca0a136f128a56d3e82 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/120fd526605b434082de8b127fbbe9ea : SUCCESS in 17m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9eae7295bbe44d5597f8d0d1cc98c8e6 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8ff9dd574ed4091b7071d6cad2d99cd : SUCCESS in 15m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/90805d09e2cf48179f20a0619d97c2bb : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/129d91cb161947299446e0977ff56567 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f0eaea9360a48deb5ef970f338189d8 : SUCCESS in 19m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d0721be80d384958afc6a2fe08198497 : SUCCESS in 44m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c40e1deecac041a7bdeb0ac2ec31c39a : SUCCESS in 1h 07m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e41ff034fe744e18eb773e3dd8046c8 : SUCCESS in 58m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/884a17ef4cf5412cba638e67b94f34da : SUCCESS in 29m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/018762c3c2b747f0be742236b56f64c3 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6876a7e7221d4701b996b46994b13f9a : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd88152ae5414deda357d63df06d23d4 : SUCCESS in 18m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f79801683b6041d99b7ced478dc18416 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":6},{"id":"f049d7f22f0625b15e5dd6e0efc3d014b331b61e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:07:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9cdb87fde1761111ca90012b1fa97d2595df3105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:18:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b0cbcb0a2044438535b4495b9d6935 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/74599d2cb4084419b6024a62a9daffb5 : SUCCESS in 59m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04fafc6a06cf4bf3bcd68f1dcd6a1bef : SUCCESS in 44m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3458e175fd4a47d2b61ec96627e6d1d0 : SUCCESS in 1h 05m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0236a381265b421499deddac8edf73c1 : SUCCESS in 9m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d8007a71f16649a397b161bf937b466b : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b5117581d2b4bde9101dd0e26ae32fe : SUCCESS in 16m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6536bbaf7bc49c2a0166bc301b8c1f1 : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0925596b2d77485dbac9cb91e570fed9 : SUCCESS in 17m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a13010ff916e4d64b0900969deb892e2 : SUCCESS in 21m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab4436d17b1f4790a184c984bd1b96a7 : SUCCESS in 21m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/48941a4d3a8a42549288ab84a09676f6 : SUCCESS in 22m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f19561398ab74798b315c5dadaa5626a : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3cf62f297a8c4f87870e7c07f6c67902 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/563de786c6934662a6441a10654806bf : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bdd050077cc845b08c7669c794fff3bc : SUCCESS in 6m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ee410f8300444460aa04dda0125c20a5 : SUCCESS in 49m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9630925a71149dc9478d79b8e02ed1b : SUCCESS in 53m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee7df6c6e8cd4a4bb1518112a83747b6 : SUCCESS in 49m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47fb483580c14b7fa21ad55bde51dd96 : SUCCESS in 41m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490e241a57de433a95ef0f3327f4352a : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3438def69fa4704aeafeaf6e9bf864c : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e9833ecf72142a288adaef13491a65f : SUCCESS in 29m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58bffae9e2e5471f9af00e9e961ecc3e : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":6},{"id":"58a60da2dce56255d34076a30031c7627207d9bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:18:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"19391d51efb1e32e8272267cf9b21c00814495cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:22:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991b2b3067204e7faf94bc9a0acebc11 : SUCCESS in 2m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a0fb9558c07c419db73a12c6ff8ad133 : SUCCESS in 1m 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cc9cec50f67442b48ba15478f54639b3 : SUCCESS in 1m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e71c33738f50558d7d4c0c6c7d6538cabb7995a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 13:39:44.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4fd58b966d64451a9273338fbfdf710 : SUCCESS in 1h 16m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7a7d42c4f5e47acb5d7e24c3ad3a34a : TIMED_OUT in 1h 32m 06s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3947cf8f7d6d4f4bbefdfe8a902f778e : TIMED_OUT in 3h 01m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e5c3119183b4120b2bf84d4602859fc : SUCCESS in 1h 16m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c132aa71ab68472b8d64a64550aca61a : SUCCESS in 1h 14m 39s","accounts_in_message":[],"_revision_number":6},{"id":"6357865c5db87f9e0922b1e77cf23aa62ff2c0c5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-06 22:38:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"87f642da89b0cb4c57e57ea8f301f0faf551143c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-06 22:39:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"114440487889ebcc58c010a4986c406adcb18f0c","revisions":{"da38fd3ccf94f44c5c270dacc81de15e26323aff":{"kind":"REWORK","_number":1,"created":"2022-01-25 16:53:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/826326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/1"}}},"commit":{"parents":[{"commit":"c57f5849818fdc705e76cde625c33262aacccd0e","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c57f5849818fdc705e76cde625c33262aacccd0e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da38fd3ccf94f44c5c270dacc81de15e26323aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da38fd3ccf94f44c5c270dacc81de15e26323aff"}]},"branch":"refs/heads/master"},"03575b6126d737c875d56284cc62929193c1a5f2":{"kind":"REWORK","_number":2,"created":"2022-01-26 11:53:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/826326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/2"}}},"commit":{"parents":[{"commit":"c57f5849818fdc705e76cde625c33262aacccd0e","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c57f5849818fdc705e76cde625c33262aacccd0e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-26 11:52:53.000000000","tz":0},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03575b6126d737c875d56284cc62929193c1a5f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03575b6126d737c875d56284cc62929193c1a5f2"}]},"branch":"refs/heads/master"},"bd209cb56c189dccdcc04472ace64c144617c66d":{"kind":"REWORK","_number":3,"created":"2022-01-26 16:26:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/826326/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/3"}}},"commit":{"parents":[{"commit":"c57f5849818fdc705e76cde625c33262aacccd0e","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c57f5849818fdc705e76cde625c33262aacccd0e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-26 16:25:10.000000000","tz":0},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd209cb56c189dccdcc04472ace64c144617c66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd209cb56c189dccdcc04472ace64c144617c66d"}]},"branch":"refs/heads/master"},"a917458533d5f7de37c3a4edf58f49e2e0f57eba":{"kind":"REWORK","_number":4,"created":"2022-01-27 13:40:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/826326/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/4"}}},"commit":{"parents":[{"commit":"ef6556bb5d9cd16955446d342c809066a390b607","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef6556bb5d9cd16955446d342c809066a390b607"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 13:36:40.000000000","tz":0},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a917458533d5f7de37c3a4edf58f49e2e0f57eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a917458533d5f7de37c3a4edf58f49e2e0f57eba"}]},"branch":"refs/heads/master"},"df6e60aaae9ba0da43e09ab9311784f841cb2769":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-01-27 13:55:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/26/826326/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/5"}}},"commit":{"parents":[{"commit":"0cb31487a1aebf409701b604629be2dcfa0c7b85","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cb31487a1aebf409701b604629be2dcfa0c7b85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 13:52:26.000000000","tz":0},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df6e60aaae9ba0da43e09ab9311784f841cb2769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df6e60aaae9ba0da43e09ab9311784f841cb2769"}]},"branch":"refs/heads/master"},"114440487889ebcc58c010a4986c406adcb18f0c":{"kind":"NO_CHANGE","_number":6,"created":"2022-01-27 17:06:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/26/826326/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/826326/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/826326/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/826326/6"}}},"commit":{"parents":[{"commit":"a50726f27a1cf76339fb23253389bc58539979b7","subject":"Make NodeIter require a logger argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a50726f27a1cf76339fb23253389bc58539979b7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-25 16:51:45.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-27 17:06:34.000000000","tz":-480},"subject":"proxy-server: add stats for backend shard_listing requests","message":"proxy-server: add stats for backend shard_listing requests\n\nAdd statsd metrics \u0027container.shard_listing.backend.\u003cstatus_int\u003e\u0027.\n\nChange-Id: Ibd98ad3bdedc6c80d275a37697de0943e3e8fb4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/114440487889ebcc58c010a4986c406adcb18f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/114440487889ebcc58c010a4986c406adcb18f0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
