)]}'
{"id":"openstack%2Fswift~826354","triplet_id":"openstack%2Fswift~master~Iaf1b3fcad079e4ebb51f007e2988dd67703566d2","project":"openstack/swift","branch":"master","topic":"fips-compatibiity","hashtags":[],"change_id":"Iaf1b3fcad079e4ebb51f007e2988dd67703566d2","subject":"Check whether configure_swap_size is strictly needed","status":"ABANDONED","created":"2022-01-25 18:39:48.000000000","updated":"2022-01-25 20:09:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"995dfa98437a76d9d1950fbf80f90facd5e9cb25","_number":826354,"virtual_id_number":826354,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-25 19:50:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 19:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cf065a6c8d34bccfeed75567922d6666404ca52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 18:39:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58fce96832c0bd66b37f7e4f8db184b2d60ce5f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 19:50:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197ad26d3bf14502b69fc09b149777ea : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/1d5997f3cd1342db9f14a011f08b8ab0 : SUCCESS in 1h 03m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/638e7d0f50734c1caab84730ae12fbd7 : SUCCESS in 47m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b77ecdd55ae04f758d03f774aef79569 : SUCCESS in 50m 14s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94ce64ced7aa4fcf9adbadeb3b34b82c : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d1e62a2bf5d74bd6abb9aee87ac9125c : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c2ccd05f7a2a40dabaf18ca9e9a6f4c8 : SUCCESS in 20m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e324ab59ac3c406d893944f05a0449ad : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3019e3fe299f406d8b58902793589f91 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d50b47160b1d4aa4b4a7d91ec4b81e32 : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17b951a248f545d0aab4b39d119029db : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7724efa72dcf4f86aab2fc5bce02cec6 : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d96d30a4556485f9d833b42fee9af29 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8676bd31e0b744f084dc5b5d540754b9 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99180662e63e49cb9d6d21c9596c22b6 : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0df7171fd9ba462e9573d6144912a9fe : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ac1895b2ac04e2786c948b84a61127d : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3e9c2682ebf54af0af68b160a7c3440d : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77f42ecfe0e74c57bf22d9fe59ae5322 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61744dce063345c2a5795e4216e30285 : SUCCESS in 22m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/01523de869fe4224836afc8cacabd1a7 : SUCCESS in 50m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea72c98ab4134e2f831ce8477e008e42 : SUCCESS in 54m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff4667f3694e4d158edbc96f61a20722 : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/538d2b4e42fd423389e6defb4064fe5d : SUCCESS in 42m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4756aa44cb0416182852eae554b6099 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d00ce473a9bd40038da5c38612c85f29 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2564655f81ac4efea9c191c7119adee6 : SUCCESS in 22m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47340e19d98445669193fdbd841f9a04 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"995dfa98437a76d9d1950fbf80f90facd5e9cb25","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 20:09:20.000000000","message":"Abandoned\n\nEncryption/FIPS job still passed -- squashed into parent.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22570ad4ea44e1c63ae4c1144c4075915ba29410","revisions":{"22570ad4ea44e1c63ae4c1144c4075915ba29410":{"kind":"REWORK","_number":1,"created":"2022-01-25 18:39:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/826354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/826354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/826354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/826354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/826354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/826354/1"}}},"commit":{"parents":[{"commit":"a35bc4b2c57d8af522cf4b11a041657f10b839a5","subject":"Add FIPS CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a35bc4b2c57d8af522cf4b11a041657f10b839a5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-25 18:38:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-25 18:38:46.000000000","tz":-480},"subject":"Check whether configure_swap_size is strictly needed","message":"Check whether configure_swap_size is strictly needed\n\nI think this was left over from debugging; real trouble was an ordering\nproblem between test-setup and fips-enablement (which requires a\nreboot)?\n\nChange-Id: Iaf1b3fcad079e4ebb51f007e2988dd67703566d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22570ad4ea44e1c63ae4c1144c4075915ba29410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22570ad4ea44e1c63ae4c1144c4075915ba29410"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
