)]}'
{"id":"openstack%2Fswift~826398","triplet_id":"openstack%2Fswift~master~Ibe2da419e0f3448f637297b607f0fc5b18b5416d","project":"openstack/swift","branch":"master","topic":"ec-short-read","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-01-26 16:34:34.000000000","reason":"Clay Gerrard replied on the change"}},"hashtags":[],"change_id":"Ibe2da419e0f3448f637297b607f0fc5b18b5416d","subject":"DNM: broken test for EC resume","status":"NEW","created":"2022-01-25 23:45:26.000000000","updated":"2022-01-26 16:34:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3b08fa9f68f6c6c47d19f5ef426c47d8108c5305","_number":826398,"virtual_id_number":826398,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-26 00:52:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2022-01-26 16:34:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2022-01-26 16:34:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 00:52:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bd66164d2b47a92769d2f3e5eedbf8d33ed7d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-25 23:45:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85fad71a3b24b327f9123d3827489cdc1a5e59b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-25 23:54:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1211e7357f67f129f93ca060c11b034d61c1b8fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 00:52:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecde6d22421400882dc9b128eef7e34 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/428d744e4276465cb16684c6d7dd044f : SUCCESS in 39m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b955f9fb9a04ea88c4e1e492b061d65 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b01591cf22e42e095281b8ac2c284a5 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e0c816b699d84abc91c9d8b3582c7a81 : FAILURE in 15m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8d985a61a13430db89f2f1ca85e5eb9 : FAILURE in 38m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1db35dbf106d4366b23a24ed5ccea862 : FAILURE in 22m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2fa9738261234a639826cc3074dc2f0a : SUCCESS in 39m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f3d9a5555cb4982becfe90dc0d3f5ac : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae1ab6366ad2444e99fc1e4df2e232a2 : SUCCESS in 39m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa09dc00b7ad400880ff812012e7948a : SUCCESS in 39m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a004c4b6054d4910bfca3c3bf0d7f0a2 : SUCCESS in 40m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e517931403c142f08a32021db3685250 : SUCCESS in 17m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7a6caf3db364c93b4d1fdbf1e2aad42 : SUCCESS in 55m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8e9682397a14f28b2bea83bb4901261 : SUCCESS in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2decfc52e7b74eca90f9cb34e6f8e49b : FAILURE in 37m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f9c09796504a5b89534c78e3d734df : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21e936c5caf44c2994ff1a785613c04d : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5ef8d495df754183c9e63cfe4b364de8f895de6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 03:01:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fbd7e51d8da4264a83cc72bc84537cf : FAILURE in 45m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a493ae1bfd6d42d3b42a789f75d31206 : FAILURE in 54m 01s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/0f40936bceb74668bd3c8f29e472c2b1 : TIMED_OUT in 3h 01m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47863a79391349608345e6993b2d693a : SUCCESS in 34m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfce830cbc954f7d91702795fd9f882f : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"3b08fa9f68f6c6c47d19f5ef426c47d8108c5305","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-26 16:34:34.000000000","message":"Patch Set 2: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51c0c4137a4f40604d492a4cfb7f5c90a6b799fc","revisions":{"8f90f38a8244ee13627cdf6ae0e3ed6f05f624ea":{"kind":"REWORK","_number":1,"created":"2022-01-25 23:45:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/98/826398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/826398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/826398/1"}}},"commit":{"parents":[{"commit":"8ef530d795df7f0300e761933c916546427a79fe","subject":"Merge \"Error limit the correct node on ECFragGetter resume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef530d795df7f0300e761933c916546427a79fe"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-25 23:45:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-25 23:45:07.000000000","tz":-360},"subject":"DNM: broken test for EC resume","message":"DNM: broken test for EC resume\n\nChange-Id: Ibe2da419e0f3448f637297b607f0fc5b18b5416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f90f38a8244ee13627cdf6ae0e3ed6f05f624ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f90f38a8244ee13627cdf6ae0e3ed6f05f624ea"}]},"branch":"refs/heads/master"},"51c0c4137a4f40604d492a4cfb7f5c90a6b799fc":{"kind":"REWORK","_number":2,"created":"2022-01-25 23:54:08.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/98/826398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/826398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/826398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/826398/2"}}},"commit":{"parents":[{"commit":"8ef530d795df7f0300e761933c916546427a79fe","subject":"Merge \"Error limit the correct node on ECFragGetter resume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef530d795df7f0300e761933c916546427a79fe"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-25 23:45:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-25 23:53:57.000000000","tz":-360},"subject":"DNM: broken test for EC resume","message":"DNM: broken test for EC resume\n\nChange-Id: Ibe2da419e0f3448f637297b607f0fc5b18b5416d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c0c4137a4f40604d492a4cfb7f5c90a6b799fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c0c4137a4f40604d492a4cfb7f5c90a6b799fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
