)]}'
{"id":"openstack%2Fswift~826434","triplet_id":"openstack%2Fswift~master~I623d13a2d4528f28dcfb6e64407372c504206074","project":"openstack/swift","branch":"master","topic":"p-internal-client-log-name","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-01-26 20:24:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I623d13a2d4528f28dcfb6e64407372c504206074","subject":"De-clutter container sync sample internal client conf","status":"MERGED","created":"2022-01-26 10:20:13.000000000","updated":"2022-01-26 20:42:45.000000000","submitted":"2022-01-26 20:24:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"826434-p-internal-client-log-name","meta_rev_id":"4d186e150195f24bc259d1c5176a9165bd3fee95","_number":826434,"virtual_id_number":826434,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-26 20:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-01-26 15:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-01-26 15:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-26 11:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 15:01:55.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-26 18:50:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c8a4e65f01bab66cf2e3326c115a79341cbb5688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-26 10:20:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5474e79af2b6c271dbc3e7243e319bff52b8eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 11:27:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8bfae2cb9a74e8e9b036be032a2dc0a : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/44baa79d58334299817084260481454d : SUCCESS in 1h 01m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf844a4aae074c6f8bfd1a8fb916e94e : SUCCESS in 44m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd0b56106677424fba5461bc9557afba : SUCCESS in 49m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da789f5c716444fda9943e82026830e4 : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/002de0a6e0db4e3294f6c9d199f440fd : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3c0855b887941cb9ff3e582dc074338 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48194bd908204e66be5025b5480ffd00 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b06416f044948329c46f142a0cb0abc : SUCCESS in 24m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34151e96a7dd4c5ab865d80f9d37e09f : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b66a09355ad74f8483636c7d7d889610 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49e896827e7a4bc785b1971b0d647735 : SUCCESS in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2d782c5f3afe4303a2d3de3c3d38912e : SUCCESS in 24m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1eeb25f6dada42e2a109eee51a533095 : SUCCESS in 11m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50b60a58c7ad41beb370c1236a7d200f : SUCCESS in 18m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/afb1b0a8c15b4f84ad08b69e9ea43d90 : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a99478059272445db48d837ab0dde4cb : SUCCESS in 30m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ddf2529d3ec444ea3a30151366b0e22 : FAILURE in 50m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f85ba4d7a6144908b6853df3a391ced6 : SUCCESS in 1h 03m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e171bd1c5cc943c2a5dd68e07714eab4 : SUCCESS in 31m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/848b3fc9c9a041e3a2041b96a5323ac7 : SUCCESS in 48m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1edb3ccc8c014461b08f657eebd1bd2b : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dbda304fdcb48a59db6ae022b382114 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea725be8ae5140a7bcd1106159bea0b6 : SUCCESS in 37m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43d3ab8c766f46bbb2f469701ec51eef : FAILURE in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f4dc6c1c2770cff5145e8eadf5d028ccc248d01f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 13:23:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46283f569ffc410e81e43d6cb6014fc3 : TIMED_OUT in 1h 31m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba49ddf865494a85bc79952702d5a571 : SUCCESS in 42m 35s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/494460f6089747b2aea108791066df90 : TIMED_OUT in 3h 01m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46b5f6b1170a498bb9dad38db564c44a : SUCCESS in 43m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/01d0d0401d2e40c09aba78105dc22899 : SUCCESS in 41m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1ee98f6be382ae53fbe32e5ddc2049482b1f41f7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-26 15:01:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4028beebc8a21d2b7f488a753fb27fa7850f761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ef94445a2243afb29b36884ff30b33 : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/476f4a689fce4d1aafc48836c4ef91a8 : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a726c6e442b54c85aba08d8fb686cd50 : SUCCESS in 58m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0632118e079f4a1b91ef90793aaa151c : SUCCESS in 51m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35b7dd71b4384f039aff91337a4f23e6 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57ac336fd69d44d796a38328a47771c3 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da63aaa2090d46ff894ffdb37cfa57b1 : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8db370097cdb4bfdb2467526b616efbe : SUCCESS in 16m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/76ce2a7479ff41949324d758661b2d25 : SUCCESS in 13m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76819025494a4e019802e1c4e80e5648 : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/098fbe9d15b44914ae3d1b0d5a8e1646 : SUCCESS in 21m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f5ddd97075447b8a3f885adefa06cf3 : SUCCESS in 20m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1f2390a01c74bdb87af542199162b6e : SUCCESS in 25m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/01bd4716036d4a2d8ce85aa636c5b89c : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/acf32ad1fedc450f9f40b9fa0877c295 : SUCCESS in 20m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/938631096cef49d9a33cace22496d1f8 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c8de42dbbe78417699fb741b0afb21a9 : SUCCESS in 24m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51864fb7107b46a0b32d5a8ca3b44990 : SUCCESS in 56m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5d2a85a33164876bff4c4d7e1953bae : SUCCESS in 54m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19aea014c7da4841b399616e9b0f8b77 : SUCCESS in 42m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3b97232d9ea46cbaa5438764dcf06eb : SUCCESS in 41m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb269baeafe149b3bc04c69eed90ccdf : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b3a4828b1a48e787d8009d1f0d803e : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d0237bc86d64871a847832a3ee8729b : SUCCESS in 29m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7ca7e8e599b47aeb3832627d4bb8999 : SUCCESS in 35m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a2d25e69f870739248c0a9329154365db59f337a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 16:09:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32bd17d5aaf79f3261d935d5a58eecf69e810512","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-26 18:50:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b644fbaf95baadb4f0bb3e6cab368c115022ea95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:24:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cbc3279e744957aa0e749c764448ca : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/4b8205b1cc604597a9bcc3ed1552cb58 : SUCCESS in 1h 14m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb9285543db443a7b7eadbd7cbc1f9b8 : SUCCESS in 48m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d05e203e7ec34736a9041b0f85a46ad6 : SUCCESS in 49m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/311d6304ddbc468188f6bf3668bd6972 : SUCCESS in 9m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/207731d32d684bdda2153f6a48fdd627 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9b418700c31741e888b1f3be2de568f8 : SUCCESS in 27m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d2727b0240744c4a93ee7667b6f69e7 : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3b310660fd24b38bc417fcee5bc14c3 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ca00ec0a21b64c06ac9b5154715cb2eb : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7dba5bcbbd4467a9793bb52f91b88bf : SUCCESS in 26m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2738b33aba7840319e5a4547dbefd242 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5c967930fb5a46e198afb4867ef95f71 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/396c0038b59b49dc97201a5b9c6a0131 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e8c719a43294d4a94d37d20503cd0c5 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/301d7c1de036430e9cd0b2dce4474315 : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/910468aa27a64b5ea6606e2afab29079 : SUCCESS in 50m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a6d8cac4e314151b0ea774a19ae84e3 : SUCCESS in 53m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/431169f711aa4b48bb5e004426753e75 : SUCCESS in 40m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72691ceaa8584502aa3f65450d12c5d4 : SUCCESS in 46m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf037b6329c44bcda7b7c2eb31277f2f : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5b621e12af4e9ab7984517bfdc1637 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/baf930ad54874136bf2d2115692f2968 : SUCCESS in 21m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8426bb010ef94d01a42b4d6d090ee140 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0c7cf8b3baeeeab642f17369f51224049c8d1b7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:24:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d186e150195f24bc259d1c5176a9165bd3fee95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:42:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d763a79bbea34c0aaee417af068a2639 : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7ec61db4f52341958a64f22740df1c37 : SUCCESS in 1m 12s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b10e67d91b4d48e38e8aeb7f289af82e : SUCCESS in 1m 12s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"876a7de6a764bcf52ba46cfba354ca16d23651e3","revisions":{"876a7de6a764bcf52ba46cfba354ca16d23651e3":{"kind":"REWORK","_number":1,"created":"2022-01-26 10:20:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/826434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/826434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/826434/1"}}},"commit":{"parents":[{"commit":"035d91dce5dfc2fdaf842e9fbeb36684bceab278","subject":"Modify log_name in internal clients\u0027 pipeline configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/035d91dce5dfc2fdaf842e9fbeb36684bceab278"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-26 10:13:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-26 10:16:57.000000000","tz":0},"subject":"De-clutter container sync sample internal client conf","message":"De-clutter container sync sample internal client conf\n\nSince the Related-Change, the inline default internal client config\nfor container-sync doesn\u0027t need to be a carbon-copy of the actual\ninternal-client.conf-sample, so all the comment lines are removed\nbecause this is not the place to be maintaining documentation of the\ninternal client config.\n\nAlso, add clarifying comment to utils.get_logger().\n\nRelated-Change: I844381fb9e1f3462043d27eb93e3fa188b206d05\nChange-Id: I623d13a2d4528f28dcfb6e64407372c504206074\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/876a7de6a764bcf52ba46cfba354ca16d23651e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/876a7de6a764bcf52ba46cfba354ca16d23651e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
