)]}'
{"id":"openstack%2Fswift~826634","triplet_id":"openstack%2Fswift~master~I0768992dda096e7ba34eb69819d7b4ec7772f76a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-01-28 14:43:10.000000000","reason":"Alistair Coles replied on the change"}},"hashtags":[],"change_id":"I0768992dda096e7ba34eb69819d7b4ec7772f76a","subject":"s3api: tighten register_info unit test","status":"MERGED","created":"2022-01-27 11:19:40.000000000","updated":"2022-01-28 17:53:55.000000000","submitted":"2022-01-28 16:28:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826634","meta_rev_id":"d5775a9ebcdce78438300685a5285c22f8277178","_number":826634,"virtual_id_number":826634,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 16:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-01-27 15:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-27 17:22:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 12:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 15:10:04.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-27 17:22:27.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"d9b733d65c7c9660602556a2e7f9f8eddd8c0fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-27 11:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4a43b5d5e428bb41db48542d1d20a9747ee361f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 12:15:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cc76da07c143a0a664a7e3b0332496 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/177ea326aa9340ad899c7564ff7fdbdf : SUCCESS in 50m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7059c6140ddb48768b1d23988cf967e2 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/29ea4a29d1e341ddbe60e0457ba67ff4 : SUCCESS in 27m 03s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f24d88f8497f403c892567c3c62c2ad1 : SUCCESS in 23m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66af911da0124abd8435ccf30a9c5a0e : SUCCESS in 12m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/81bdc48b377941c4b8b5831acaf4c0f5 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f8fcd4e6ee489d99837951850b2aec : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4cdbc9d8fb1f419e98d35ef0436c5c76 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa6a94512c5441cab56a817cba3022a3 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b78d72af1d594840aa264014389fb67b : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7bc37be370a41f69769275371cfe7ff : SUCCESS in 23m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/659946ff7efb473d83cadc649a327285 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2a5fc7c26714829adb67a33d9b4f41c : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b22725fea26489bacb2196e90354d69 : SUCCESS in 20m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/de842f564f3f4a8ea6cb7237d66a416a : SUCCESS in 25m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2e3b418a129417994cef7cf7e682eeb : SUCCESS in 45m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62a5eda4ffbf444790c94929d518c473 : SUCCESS in 43m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d341b9c2c2d04bbca8c3da0e581bb6fa : SUCCESS in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20625c42b865420db6707a9811336ba8 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce25dad7bd114aedaaccb33ce53e46db : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6a4920db4c4d5ac7cca15c5b6d9d3838e0507714","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 14:40:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be28ba5d709b45688baa1bdc41845975 : SUCCESS in 42m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0659af91ab944036888e71dd2ad7b327 : SUCCESS in 56m 51s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/0288d6565f5b42768866ab2272be99a1 : TIMED_OUT in 3h 01m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/63c30a7c6be4445bbbacaae1bbb7fce7 : SUCCESS in 35m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/568c4c09b1164d459b886e962a9d9265 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ab63e437f5c70ef398a2b793569718480426cc67","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-27 15:10:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29ce03823fea7850c3f11be84ecd21a086f5e8b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-27 17:22:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1249116a4e0f09b9effa6e97885fcd24fc3aacc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:38:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af0fa57ffe572e3dc43c03daa2094133a4335955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 21:37:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/022f71bd413b4920b077be97dcee5e3d : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6d14bc98a6d4a98b2dd2d3b07aeac5b : SUCCESS in 47m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5e5e73cd1e5647e5afc9ea551001aef6 : SUCCESS in 14m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1130113557e4493ea7a25ac7ef581a3f : SUCCESS in 13m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/876b6a5e71fc43329055159afd83882a : SUCCESS in 21m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef2bbb25cded45ec9a32da602952ac27 : SUCCESS in 22m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ddde5e8591e34d308bf948dc4228d646 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9558b55c40148c18a7ac60d396c06b8 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/076f088485dc444395dede94e3e00c66 : SUCCESS in 25m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8aab262f2294de6922780c33aec51d5 : SUCCESS in 21m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/54433be23067480aa84b1b3389dc92e7 : SUCCESS in 24m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2161ba429f6443f7b24af39c822e7372 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8073d396aeec4d75a21e873add9fc4bd : SUCCESS in 20m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fa134b05e98451a856ad036295e6c27 : SUCCESS in 8m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2476bb8e31404eb5bd3b65cbe428069f : SUCCESS in 47m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a97c4f0feebc438297a367f0c6098706 : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fe909d4259e4746893d3dea158616ff : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf27c136bea445e8246a83b22f90e3b : SUCCESS in 10m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fef1497b412e4a298f8d9f0848afdd79 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e77cd9e06bf2dba1a88813366ebb01b1619f236f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-27 22:35:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b17d942841f1de5dd705971435c341b587f08895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 23:23:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07f47b75b9d4ae9aee12fb2cd09929b : SUCCESS in 9m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49a42cda113b4ee58477506941daabf5 : SUCCESS in 41m 52s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ae8547f7c8f3495cb6dd5947911d5e31 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4b0d15b8b86d49519c143e81b4faf774 : SUCCESS in 26m 14s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ea22eb671c1c4935974f405987c0cb68 : SUCCESS in 24m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/036e8b938aed4f88bff345b8194e5b7c : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/797221d5d90a418ba172be9860da42c0 : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82e8444190de49a785c18db9cd9da423 : SUCCESS in 25m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7103c664b2b49e6aca4085613c22986 : SUCCESS in 20m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4a32ac215624d1bb3f4f98e8c2dadaa : SUCCESS in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2138c4a1ac44495494635c974a105ca0 : SUCCESS in 25m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e0e806463f6471686b1b92f155e3c78 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b3c07b4c98142f8a0c889ec26edea01 : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/df9f8b787db341cea8cdea8bbf913efe : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/761b45b0f0c945db89ddfca26ed493b8 : SUCCESS in 25m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d74d0d8c6d0c4176958c3fbce5586a0d : SUCCESS in 16m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cf275362bbe47b5a92c59c9d98064f8 : SUCCESS in 42m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd32fd7da1b5439ba40768f6cec4cdc0 : SUCCESS in 41m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b1ccf229dd3409589e78fb3e91a6899 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d397d76b7340fdba4c93c58520ff58 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b256ac2007a452f88547da8934ec028 : FAILURE in 28m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b7c89c34113360bc2c20773cbd95470575d35b17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:47:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6374b187cf8e46b0a7d904c0d9df65ae : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac03d5e7a7414cf381b08c72b1141736 : SUCCESS in 1h 28m 14s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c81f9d587454494588e1751734156cd5 : TIMED_OUT in 3h 01m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a05b8420b7d246db942566854a7afff5 : SUCCESS in 34m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1783b0c02a1c488bb1d395009b4782d3 : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"90f744718d781d37eab19d191450358968659eaa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-28 10:07:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1eadbeff74ba236ab8bd661777f20ddd73f663f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 10:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a7cca0ef284bee9f61963c41f62477 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c63ebaa2ce92469bb00f371f34288e94 : SUCCESS in 42m 25s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5e6e205084c94cfca343e54832358ddd : SUCCESS in 24m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fa6d5c7d21da4e8d91f8d7e097721e62 : SUCCESS in 23m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8508f68144a741bc819c564827bb83dc : SUCCESS in 26m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe4f25a0605c460dbd14d5b4b381efde : SUCCESS in 9m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/054ae60d32dc467e95ef95d298c59f10 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3d08c85af7948c181466f73f4e80ee0 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d37c1bef7bd6417fa7bea5d0735bc935 : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b25cefd728949a1b6ef9b7f3cfc4ca7 : SUCCESS in 21m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d5576f0ee1d4d9b8e373866afe42ecc : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a814ecab2fcc4899b201557f8b83d213 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d340e24c74843f2a4a2ef04a3577bef : SUCCESS in 14m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/836dd571fe254edc8bfa2c858c4f61f4 : SUCCESS in 13m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8724c8671fe46a582591efa5d3eb181 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eacb3db6163a4ed4b7e8526bb9f60a87 : SUCCESS in 13m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b148f16ff91a4d34be84ec000ba9d050 : SUCCESS in 40m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/143b6d1b1b4d4955a765f5db1b76cd1c : SUCCESS in 33m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8da9289d330947538faa461fcf9b881d : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9702d801064e0f836516bd665dfd65 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9baf849f83ad4e8f96dab559cd95422b : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"befcbb966b631a8e8d82dc9934532125bebc31b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 10:55:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21c537560967775a6a515a01b1621d3d70af261d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:19:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/530a148265f946b5a6855b0f55774dad : SUCCESS in 41m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b564df568dc04da0bb83157fde263ecc : SUCCESS in 33m 08s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7e6970adb14c43748181391b16932a2f : TIMED_OUT in 3h 01m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa9309440c6b4e2daf3ddddc1a61a0ae : SUCCESS in 1h 29m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0fd994c26844021b0f942307205eee8 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5975bc5b0ee1829989f5a4080933788168b93e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 13:20:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3f6ee6a382433dba1225d1eed2bc49 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2680581f7f546668d9b99bd2f859da7 : SUCCESS in 49m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/59fe7c3f79b74302b83446bfaf376c8c : SUCCESS in 10m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d94500d9a7c14ef88d66c2ce3a92fdd3 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad400c409a6f483ab02c9b853be181cf : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f108bc2170e4b6fb6e5514e75ce75b1 : TIMED_OUT in 1h 31m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e4d986a3cf43998f247462c2b223c2 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31d31d2b418c4f21a16b9b48b8eae068 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e610adaf746405d95da207bf4731d98 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e975b80ec9ce469ea6962b77e3cc8bbb : SUCCESS in 23m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b208872e2c494f9cb633e19527930ac3 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bf1df607344a4899a7c5d9f6b1f398cc : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a79afcf1656045eda313f06ff5e25ab2 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dbbdf4c13e79403695b81590e251467b : SUCCESS in 6m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4378001de3a4737bdce8ec374638c71 : SUCCESS in 39m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f78e7a50b0a4496b2be0521847d50ab : SUCCESS in 42m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48f13be9e9c64d3bbd2b38e4294c3352 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f519ec3efb41fe8c39da04a7b1e5c9 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62c74be1c58a4e718783459680400142 : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4a81d6d55c485b9fe15fb5c99e6c841b2ed5258c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-28 14:43:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"149c62487dc47ff65d8d485f0ad8beb09295765c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:30:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89354e1da9c432f887f2fabd68d19b7 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d56b9991cd404e799f3921e8685c7aa9 : SUCCESS in 44m 29s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f42fc2ed1dcf4cc49cea9b0c40587827 : SUCCESS in 26m 30s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5364cf26f42f44c6a706933fe08af90a : SUCCESS in 26m 04s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ccd6fe9f0dec416186136bba33c481ad : SUCCESS in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/252d47becb204f1590ec902df4663f6f : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/59ca7b14d8314429ac6c730b87ce2f70 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/946ae18b1da1477d929ed28f8c7792e2 : SUCCESS in 20m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96da4a7ef1e749bfa863a5ff84a5f705 : SUCCESS in 19m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/402e5fafc0bd45a69ddd2046974b119f : SUCCESS in 17m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bcfcfe6dd404f0aba900cd3bdf9cd95 : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a4a3daf92d84d95a0337765bd6e161e : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3544e4fa15d7476a951b0f81b27159a8 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/17e0fbe6ec0245d0b8fb6c28e2f12c09 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f32c8f125df345309d3077b38ccff274 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/439e0e4118e74092a633280f09655857 : SUCCESS in 15m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e20504335caf4c08aa92ffddfceb19bb : SUCCESS in 41m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e848fab3c2d4f49801aaf01ea94dd65 : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf9aa98a91de4b3694677a47bd5cb964 : SUCCESS in 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4354405f21a746d2a23876eb006d93d0 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/593a23456c8c439fafa1c957e3b0aefc : SUCCESS in 19m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7921477e0b91cbb97c5c3007d3fa80a09b254e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 15:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"714e46723117b09dab10850bd29685a48e845011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:28:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f887a3967e4600b546c408aac8a1d2 : SUCCESS in 8m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c34d7c5c69e4a14841c482479775591 : SUCCESS in 50m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9cb186e5631a4be5a336ba56c972094e : SUCCESS in 10m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d030d9bcb98c450aa9c7c38c48d160f7 : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e369ef0754354e35af649ecc93dcac1a : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac833b744d1b4ca6ab1b8983db599faa : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26b117735a22414bba6b7c8a53a89748 : SUCCESS in 17m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/821d6dfcdda948908e432b44014bb6ad : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/144a5c3541f140af9eec3be42926da01 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0cfbb7bb1e54fa7982cd46f391d1631 : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/684e7fcac653447d904f770bba0e397c : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1f1b9d8f873046fb968e09c8fd7a5520 : SUCCESS in 15m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f567802ff16463fa51e57246e64f286 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d61371a23a64eb2ba8fe89aba96b5ea : SUCCESS in 6m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ea9d9393ae44246bae4b05bb713af9c : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48c2b487977e48139255907ced805266 : SUCCESS in 42m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/592770f84e944510a2e451fadca15100 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea438253de04576aa8a40cc809ab898 : SUCCESS in 6m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00fe58d17ade4cbea59346cf9b9a504d : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"890bfa83c96f8a9f051c72effa8ce606c128605d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:28:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78ea44081ee7ba6bf20f52432254ad1ac3d0d903","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:30:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5556d312b94c4bffb66cc644e7722f8b : SUCCESS in 1m 33s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d0bb876f0df7440a9f8e80db0e8f1f5f : SUCCESS in 1m 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5e47deaefecd4606a602df70da1b573a : SUCCESS in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5775a9ebcdce78438300685a5285c22f8277178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 17:53:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d01f083322974ec5bd66394d7ff99052 : SUCCESS in 1h 10m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff78088df05b4c26bd245d4726aea876 : SUCCESS in 40m 56s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/934b67d16eea4771a1782aee5ebe24ee : TIMED_OUT in 3h 02m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/85970c6b6d3947d48431ffce9a063a51 : TIMED_OUT in 1h 31m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d99c0dac016d445aa501822f18aa243f : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a1955d88a05ebea078376a968aa14bed94234ec","revisions":{"0a1955d88a05ebea078376a968aa14bed94234ec":{"kind":"REWORK","_number":1,"created":"2022-01-27 11:19:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/826634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/826634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/826634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/826634/1"}}},"commit":{"parents":[{"commit":"4edf0cde429e1fb20865e92fd3259d862b7f00c1","subject":"s3api: Fix types in /info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4edf0cde429e1fb20865e92fd3259d862b7f00c1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 10:58:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-27 11:18:51.000000000","tz":0},"subject":"s3api: tighten register_info unit test","message":"s3api: tighten register_info unit test\n\nTest that registered info has all expected items and no more.\n\nChange-Id: I0768992dda096e7ba34eb69819d7b4ec7772f76a\nRelated-Change: I4a46bd650a53f88c642d402e697869df28bd2fd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a1955d88a05ebea078376a968aa14bed94234ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a1955d88a05ebea078376a968aa14bed94234ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
