)]}'
{"id":"openstack%2Fswift~826947","triplet_id":"openstack%2Fswift~master~If218fe19eb75c42f56d25bf9c61adceeec025b6e","project":"openstack/swift","branch":"master","topic":"release-2-29-0","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-11 20:00:16.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"If218fe19eb75c42f56d25bf9c61adceeec025b6e","subject":"AUTHORS/CHANGELOG for 2.29.0","status":"MERGED","created":"2022-01-29 01:17:11.000000000","updated":"2022-02-11 23:03:34.000000000","submitted":"2022-02-11 21:48:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826947-release-2-29-0","meta_rev_id":"4bf3ad47994548daa4fbcfd85eaa2ea67727ed15","_number":826947,"virtual_id_number":826947,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 21:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-11 21:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-11 21:03:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-29 02:30:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-31 14:48:20.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-31 17:48:04.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"cb6dc76c392543a79042ff228c0bc9ed3c3c5505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-29 01:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28a1f88ab02e84656f327f8ae0562670db4de9ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 02:30:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6625f302e9c24d4f9ce1a0f538bbdef1 : SUCCESS in 8m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad0c350ba607442390ea8b5faff4f45b : SUCCESS in 1h 07m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1781fa82ae0e4b2e869be93cd537261d : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/14d4c07a062944a3a4282a49fd312920 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a1a597c8c10456eab2c243fd4bb7c0d : SUCCESS in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9cdc1113eafa49a8a945969400df18b0 : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef77705c62524485842ea388a43f8319 : SUCCESS in 15m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3a59c716ec44174bda2b725f7cc2897 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76826a5c0774a39936afa98cb072dd3 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ae984da67e5cd992adf34ffd5cc3700fb00008e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 04:22:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24ed0c14fd8b46ef841fe319580044b6 : SUCCESS in 56m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51e30c1228784b4fbdf2c2249c7a074d : SUCCESS in 32m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e2c6e224eb6f4a94bbbe06de6572fe9d : TIMED_OUT in 3h 01m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6448e5c5df3455ebd625d4a0abb5238 : SUCCESS in 49m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d94a063124844eb2a70cfe866c070a11 : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e6d34dae43749235707983e0c28ea3287b0a1290","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-31 14:48:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf93f7d2c7aa1c452cf4fe266b407b4068b65bfe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-31 17:48:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9709a72fe4af7a41bb6329f7508f95b3473e62c8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-02 19:29:25.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12b80e3ace4b92a64c8061dd703f87e849bea399","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-02 19:29:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"662d8b369783e3f57052eafc3a4447939b563284","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 20:05:19.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08775f644b7047aaac670fc4df2d1cbf : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef01ab3dd5443da945677c0bc1afbb8 : SUCCESS in 18m 39s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f12ed5691b4141d79e6f86eb3b530608 : SUCCESS in 18m 45s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/50de438cea684262b114ded6f8f11957 : SUCCESS in 25m 08s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/02b2564897f546b28f203512f15698f3 : SUCCESS in 23m 26s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c3e89a9a64ec4ed4849499e9ed4bb440 : SUCCESS in 25m 24s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/6049e4622b2c4295a7ca714e9f1126a3 : RETRY_LIMIT in 3m 27s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/111e926a26e741628948ac1d45eff26e : RETRY_LIMIT in 3m 44s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/12d1b71818c744df99f33685ee5adb47 : RETRY_LIMIT in 4m 33s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/80abdd890f8e43218358011ff504ff74 : RETRY_LIMIT in 3m 52s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/e7f01311e4b54c5da168ef5b4a0e06a4 : FAILURE in 8m 10s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/8fff81ae32d04fc689a18eb9c0a8ed48 : FAILURE in 8m 34s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/08f5ddf2f3434341976b8f925202761a : FAILURE in 8m 55s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/12f547d36b33483dab44f5cb2621bafd : FAILURE in 8m 29s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/cfcb810ec35842039206f530d7112b27 : SUCCESS in 29m 40s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/1d7ecaed3b5540e1802203608d0cde5c : SUCCESS in 30m 10s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/0cc7a347a8044544b677abf92aad4cce : SUCCESS in 29m 20s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/639edc92cae340c58d5827340b363262 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e87c46d62a7e08e5108e8b5e9efceac8f85fbc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 01:22:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b30417a0d7794e975e5a95e828d097502b391b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 02:55:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14faf39d7634744bcd0f5efbdbdfc0e : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af611671455e426699e32bd233c5d1f5 : SUCCESS in 6m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/356e9145a257445b9bc2c5c7af39cf0b : SUCCESS in 1h 01m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58fe1b20b2da40c592b4541c33ff2d51 : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/510eac836e3047e6a91304d12f6f4fc3 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54eadb2a411c4b638bb122ffa9bf7107 : SUCCESS in 20m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c117f37e02444999f32cd3709e0ef74 : TIMED_OUT in 1h 32m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e423752283a24cec9dfc7566d7ed6bc3 : SUCCESS in 17m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e641217f16d402db3ca44a8cbc9466b : SUCCESS in 23m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe32c2e8a74048f3b4dbd0c224fbfbd7 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8ff801492fd6bf17d7c0b79497686e918afa967c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 02:56:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c77189b80ca4c9abeead045e2b37076 : TIMED_OUT in 1h 31m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5b94121aab44644bd01a9ad159f86db : SUCCESS in 1h 20m 17s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d197e220d4b84244b3d58b9a9367a673 : RETRY_LIMIT in 5m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/81d1873e01844262aa25ad4a9bce8987 : FAILURE in 1h 04m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/09d71ede4a444643bfdae1302473aede : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":2},{"id":"a2d3aa5cfe43f8f9a8009ab1c507dc8393807cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 19:27:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d653daca5d3d844e470302cccd28544ea1b6ceef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 20:28:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73ee32f572a7437c875e5db2352b7f8b : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb176f838cc6498eb718aa4f195fefe6 : SUCCESS in 5m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eac93a22efc46ceb5e83518f53ad0d2 : SUCCESS in 56m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a7bb479fffb48fb90a537ce8d071bfd : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ca9e78b67004264ac54070eff5dfa40 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81bd27aacb0a4b1794a7652b233d7d15 : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa2594fd70e43dc8f68b343ee429e89 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7594c167cc614bc1a4398d2e6985e0c7 : SUCCESS in 15m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2edd487591fb414dae6d94865e01738b : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55cf4f0f33b4e32be67e9a7bcb0b278 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1a33d0b48b44429df4f1c417f1c972ff663a5a12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 23:00:18.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c96e9643354e4b2cb37f467c148b75a5 : TIMED_OUT in 1h 31m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d23f495447f348f7b0ac6d66bfb473d2 : SUCCESS in 56m 14s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5615f0a50a2b4d86879b143abad25a0e : TIMED_OUT in 3h 01m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de33e91fb87c4447a320d0453db42053 : SUCCESS in 1h 20m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fe39297d79347499123e80c592902dc : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":3},{"id":"93a4b25de58d648318cc333ebd1bb46a3b92852c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 23:17:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c888ea941ee65b07be59d10b40bbe470f956590a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-05 00:53:49.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4fd897ac4eef445fb559f9cf105ebbcf : TIMED_OUT in 1h 31m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48f04ab02db74ecdaac70cf09f86b8fa : TIMED_OUT in 1h 31m 10s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/4bb1637db3ed46dda4893fc896305946 : SUCCESS in 19m 21s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/ec17eff16db44de59503b1a139771208 : SUCCESS in 26m 18s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/6f14bf7f50a649d6acaa09c0848cb9da : SUCCESS in 24m 51s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/80b8ef2ab4d44db49df5e83ac3e98b16 : SUCCESS in 26m 58s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/aeb10db1163b472fa677d047d69fb72e : SUCCESS in 17m 45s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb8be6fc3b044e009b9b3d172833dfd2 : SUCCESS in 21m 23s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cd3f760a20f43a1bd441693b5c82902 : SUCCESS in 23m 40s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0d45bd0e9e14871bad2311d6b5cd152 : SUCCESS in 26m 02s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/0f24f5eb0beb499c9fc13ae30261432b : FAILURE in 7m 59s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/32a70226de7a483b89f061eee2b55d39 : FAILURE in 8m 20s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/923494c5c1554da4a40970873709694b : FAILURE in 7m 08s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/5b9ffa252a1b42568e82c3ddfaad860d : FAILURE in 7m 53s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/f98e9fd263804a65a0f1002f70457015 : FAILURE in 20m 45s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/4506148ff232450790bdaadf807fa870 : SUCCESS in 18m 52s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/0c911be7ef1746f49a03df9b99b72cf7 : FAILURE in 10m 45s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/01cb7d159de74b8f984eaaae8c53188f : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4c500472e2449a564225d9c6d816c8f99da326db","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 15:01:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c4a3af851e018906eb72d05daeb0965a0ca4c162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 16:41:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d634b2f8ff54a7bc1a73dc88fdec9ef1c520ff8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:27:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b209f262c6e4a8185f8c9da22b4541d : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef6a177545a64d0ca86905ada18c6c88 : SUCCESS in 6m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c1b728e7076482cb50250894589880c : SUCCESS in 43m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13fb9f191fe748c2b14cac920c70340a : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d8e319cbacd142a9bd2976f77f1cfa02 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a8c9610ff2e4aeeb1ea5936fa207901 : SUCCESS in 21m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed890baaddbc4f948f98aa65a06be319 : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5eda83aec956429b80feb899d7bcff98 : SUCCESS in 19m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed9047c380c24b7a88db464e55369f08 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7236544b6ef4bf4b16622278c72c327 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":4},{"id":"dfd214e5b080edd8ee60ce9db2a151bf0ec61212","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 20:51:05.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8d0063a6b80428b9acba398d41b6f95 : TIMED_OUT in 1h 31m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e13c8a3cca88421dac18a0298233b98f : SUCCESS in 1h 12m 06s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/983cd6c066524258aed5ce50a6a45b95 : TIMED_OUT in 3h 01m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d55b81de4c7544cb9dd65f7e8781f230 : SUCCESS in 57m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7626463a3b8d41cc8f9eb7c2d00c7755 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":4},{"id":"9de6eca6efefdc3878e7e9a11fb342d3fe6992e2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-11 09:14:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"400c13ffe88b850ad5e29a5e16eb7c114210e271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-11 19:59:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af15a9cadab516ab0f66be4e74867d97f937f670","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-11 20:00:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"269e36391442154c7c0e2fd6ea66004250359fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 20:40:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa8138390be4ebe82f01a7a0c1e7f94 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b20b183a45fa4ec4923db565dd03cd0f : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c2082563ce248b3b85f2e48f4152e05 : SUCCESS in 39m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/997a47fbb3f44265b34785a203199004 : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9f930d64c3bf4581a100240917c361ec : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e51c6b7691ca412faf6a53af61f0ce2d : SUCCESS in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ce3e42ed46ba4cbfac68b1d2fd9473be : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/273bd7efd5c145378c2874f07273655a : SUCCESS in 19m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16aeaaa263c849f6bf49c8d0b570925d : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641306d134e24b19ba8d8dcb0b5e8a6b : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":5},{"id":"3b1c4a148052af96f413f577772b9c9e4c5aa82d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 20:41:06.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4dc911e078c14977bccee5b9a59b6638 : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e64485a9f5314dfab89cfee688651387 : SUCCESS in 19m 33s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/febf677868134dbebc9ec70eceb4ce7a : SUCCESS in 24m 17s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/5ce386cc182b4006a32fdcaf09aeaf56 : SUCCESS in 25m 40s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/4e79defbe67c4019a8cf176ecba1e49e : SUCCESS in 27m 48s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/40ad2f60aa54477781b99fd7266150b2 : SUCCESS in 26m 38s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9c7c3411fd946b9bf29e3e57ad66604 : SUCCESS in 22m 50s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae3e7a4f48d04f41af6bc9857a963ace : SUCCESS in 22m 06s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9dc90cc1b0543478e1c156343ffdece : SUCCESS in 23m 21s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/25bdf15a14704d9abb2a0711ef7934ab : SUCCESS in 18m 49s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/803426d67e3946b9a4fca4e8eb86556d : FAILURE in 7m 50s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/65ad90ea7d9449729d1e54b12099d864 : FAILURE in 8m 44s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/0c8ecb414e45427497f66db2373318eb : FAILURE in 10m 53s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/ae14f2892d104ec9aedc3db76c18e7ea : FAILURE in 6m 43s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/2eb7fb04fc7049d9903c9a2bcf562599 : SUCCESS in 25m 38s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/694fe776fc6c41bd88c42667d6ab42fe : FAILURE in 20m 12s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/d79475618e834ce4846a4b9d0e62060d : SUCCESS in 20m 42s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/255f3a8958a3419db780f5c2302b885c : SUCCESS in 28m 34s","accounts_in_message":[],"_revision_number":5},{"id":"9908185eb6416855161f606c43173394181abe60","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-11 21:03:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"553e06611ce2ae7aa3f8c483d70087667d75f89e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 21:04:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"02f607faf6a582ceec7579cfc80abcc25f2d1b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 21:48:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ef3f04df4846a9b0ca60ff960e49fc : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/323e10a50d614dcc98490125377cedef : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/950056a0ffa64fbfb1c7ceb57ec3214a : SUCCESS in 43m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a1060d75c89e45d88faa3e19837825d9 : SUCCESS in 8m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/761f300dcb8b494e893f0e9bc00d09eb : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d04846b679c46899c32d4514f71ce2f : SUCCESS in 20m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b209c9b64442e7adbbb754ee126a90 : SUCCESS in 20m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27c7431c2fd34a2f8fc83fe2403122c4 : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5ccecd777bb4e3587a3a1e874b31343 : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb55ce61f91d4323b68e6ba601b42f75 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3dce5f5ff3904d25972faa1d41dc16fd : SUCCESS in 19m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5380420fb28d4b1dbb1e503665622ff5 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed879a31edce45d89de2de70583b0f0f : SUCCESS in 19m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61d81bf13be14b1ab46d2083075e0511 : SUCCESS in 20m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0e42bffdf1e47f19764460607c5044c : SUCCESS in 6m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0896b9f4b43748919f24a2a82aebea24 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5a4b23e4ec4faeb688b5662ffdee22 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":5},{"id":"5217c6482b4b74b65b3895e034fd6cfade6f7703","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 21:48:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9df1a89762c9e1e4d0a92b46e09ff78ce301a847","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 21:50:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d59969533540c59f1318588af8413e : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/667eaa182a984b47936ff8523c42827d : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/11db84c69a644153b3c471ca5aeb6447 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/74766bf896f44dd0a7e5f2ccbb9caa09 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4bf3ad47994548daa4fbcfd85eaa2ea67727ed15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 23:03:34.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c3b4c7f4bec404c923d75778c616e41 : SUCCESS in 44m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1db1f98276204c439aefed8dd1167c3a : SUCCESS in 1h 01m 48s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6567b4c27be54b49b6d5386e7e960531 : TIMED_OUT in 3h 01m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dea6816b6c840f5bb33b7dbcdfc7cdd : SUCCESS in 1h 05m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a453eb0301504c5baf22528993db338b : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e","revisions":{"268231dc13b95f2c5abfb7a08fa74e2571198fa1":{"kind":"REWORK","_number":1,"created":"2022-01-29 01:17:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/826947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/826947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/826947/1"}}},"commit":{"parents":[{"commit":"c1d2e661b1a448d5e4d00a8e41e12803a93b364c","subject":"Merge \"s3api: Allow multiple storage domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d2e661b1a448d5e4d00a8e41e12803a93b364c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"subject":"AUTHORS/CHANGELOG for 2.29.0","message":"AUTHORS/CHANGELOG for 2.29.0\n\nChange-Id: If218fe19eb75c42f56d25bf9c61adceeec025b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/268231dc13b95f2c5abfb7a08fa74e2571198fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/268231dc13b95f2c5abfb7a08fa74e2571198fa1"}]},"branch":"refs/heads/master"},"bdfa8be7755fbf9f295e8db2dd16569fa5dc2e1a":{"kind":"REWORK","_number":2,"created":"2022-02-03 01:22:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/826947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/826947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/826947/2"}}},"commit":{"parents":[{"commit":"6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c","subject":"Merge \"s3api: Fix non-ascii MPUs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-03 01:22:25.000000000","tz":-480},"subject":"AUTHORS/CHANGELOG for 2.29.0","message":"AUTHORS/CHANGELOG for 2.29.0\n\nChange-Id: If218fe19eb75c42f56d25bf9c61adceeec025b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdfa8be7755fbf9f295e8db2dd16569fa5dc2e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdfa8be7755fbf9f295e8db2dd16569fa5dc2e1a"}]},"branch":"refs/heads/master"},"48dd71255d4db3f369593d670aba9798d710f288":{"kind":"REWORK","_number":3,"created":"2022-02-04 19:27:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/826947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/826947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/826947/3"}}},"commit":{"parents":[{"commit":"6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c","subject":"Merge \"s3api: Fix non-ascii MPUs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-04 19:27:02.000000000","tz":-480},"subject":"AUTHORS/CHANGELOG for 2.29.0","message":"AUTHORS/CHANGELOG for 2.29.0\n\nChange-Id: If218fe19eb75c42f56d25bf9c61adceeec025b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48dd71255d4db3f369593d670aba9798d710f288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48dd71255d4db3f369593d670aba9798d710f288"}]},"branch":"refs/heads/master"},"205a19b50485d2541076d17fd7c7703a1ea2ed38":{"kind":"REWORK","_number":4,"created":"2022-02-10 16:41:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/826947/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/826947/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/826947/4"}}},"commit":{"parents":[{"commit":"dc061dd95e34b807e4a7da2e6d44b5a3061b3c54","subject":"Merge \"Trim sensitive information in the logs (CVE-2017-8761)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc061dd95e34b807e4a7da2e6d44b5a3061b3c54"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-10 16:40:58.000000000","tz":-480},"subject":"AUTHORS/CHANGELOG for 2.29.0","message":"AUTHORS/CHANGELOG for 2.29.0\n\nChange-Id: If218fe19eb75c42f56d25bf9c61adceeec025b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/205a19b50485d2541076d17fd7c7703a1ea2ed38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/205a19b50485d2541076d17fd7c7703a1ea2ed38"}]},"branch":"refs/heads/master"},"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e":{"kind":"REWORK","_number":5,"created":"2022-02-11 19:59:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/826947/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/826947/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/826947/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/826947/5"}}},"commit":{"parents":[{"commit":"a846225c398922275965c1f64c9d443cffef5d39","subject":"Merge \"Quiet more BadStatusLine tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a846225c398922275965c1f64c9d443cffef5d39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 01:16:52.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-11 19:53:49.000000000","tz":-480},"subject":"AUTHORS/CHANGELOG for 2.29.0","message":"AUTHORS/CHANGELOG for 2.29.0\n\nChange-Id: If218fe19eb75c42f56d25bf9c61adceeec025b6e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
