)]}'
{"id":"openstack%2Fswift~826950","triplet_id":"openstack%2Fswift~master~Ia62456be136f4e32288dd19d8061708e9e7b7cb2","project":"openstack/swift","branch":"master","topic":"ci-lockup","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-04 20:56:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia62456be136f4e32288dd19d8061708e9e7b7cb2","subject":"DNM: try to repro CI test lockup","status":"ABANDONED","created":"2022-01-29 05:11:39.000000000","updated":"2022-02-04 20:56:04.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"81df1152eb215c2794f6379b5afbd0604f50f0e1","_number":826950,"virtual_id_number":826950,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-02-04 04:59:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-29 05:28:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"164c77f31ffbd58becdd1bd00157402dbd0d0610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-29 05:11:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e8feed4be509eef8ab2bb268212f355f0ccb6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 05:28:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8fe9459ee58b4b788c4a3f57688a9260 : SUCCESS in 4m 57s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e297ce373b054f9e9e4f32043a9badba : SUCCESS in 4m 48s\n- swift-tox-py27-2 https://zuul.opendev.org/t/openstack/build/74ad93f68fca49ee8f718b688f21b373 : SUCCESS in 4m 35s\n- swift-tox-py27-3 https://zuul.opendev.org/t/openstack/build/b2b196e1221040e09b4cb3bc093fb58c : SUCCESS in 4m 34s\n- swift-tox-py27-4 https://zuul.opendev.org/t/openstack/build/f1b0a69ce7c0432ba0c38d6e6f4bbc5b : SUCCESS in 4m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c33a223c55e4a10955db5ee01ae112e : SUCCESS in 4m 19s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/e8658b99606747f4b4fae95a5fca14ab : SUCCESS in 4m 23s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/68564001b3634b8481cac1abbbb812b0 : SUCCESS in 4m 55s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/f3b57169770c4882a3ce114ce8890354 : SUCCESS in 4m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a3a6fb85160e42fb901c9ff6fa6059b1 : SUCCESS in 4m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f84821ea337d4efcb7c7664e60fcafbb : SUCCESS in 4m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d98d1bec99a47be89998846351ce048 : TIMED_OUT in 11m 09s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/98049c5c40aa42708849df0d6629b6d9 : TIMED_OUT in 11m 24s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/6e4f31456b714c97a770f8483f1cf87c : TIMED_OUT in 11m 30s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/94cd51f841a14fc7bf2132e4307b124d : TIMED_OUT in 11m 01s","accounts_in_message":[],"_revision_number":1},{"id":"026361b86929c4b182462ee60bc3979894e233bc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-29 05:30:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4ad60859c8ff58b4855405e5297f37db3fc6bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-29 05:33:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"34d67c68c3947b2fe8db73445d9073ecebae7c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 05:55:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fea29e16fe184ad2b4524223c7135029 : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5bba3f0fe0459a9a6dc9903fbbaaf6 : TIMED_OUT in 21m 04s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/05825ba58ee4415fa657f1fcea43d2bf : SUCCESS in 18m 10s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/c15975f98036463ba44e04fc939cc65f : SUCCESS in 19m 44s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/95448fc5e79d437ca2b14116d006d6f8 : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/173d5f51759844c5952066aff1efe3fa : SUCCESS in 17m 44s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/b2f4c4e3d6d6476ea602587b9c195ffe : SUCCESS in 10m 46s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/51d463fd3f634185a76715c59d76eed4 : SUCCESS in 17m 25s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/8eab47d771f1491cafd1ffe13db7a73a : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79eb3ff97a8342b8a78f5ff15728793d : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e39ac8e76e1d4cd8be857c3ef33f8574 : SUCCESS in 18m 29s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/420c06fe65ad4ee4bfd4d0ba57859d4c : SUCCESS in 18m 10s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/1b3d09653e0c4068994fa5890065e196 : SUCCESS in 20m 04s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/4dd7f6cc0a9d4fa39c0b0dbaa8060f82 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":2},{"id":"816138fe22125965c17f7534e795d56b992ea114","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-29 05:57:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ea7e340ad6a15fe6678d18dfb72409fa743f1f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-29 06:19:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77f950c0c1fc4d8881c2a92c43357db0 : SUCCESS in 16m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0488b72580b64bc4b873db5f98c9a95b : SUCCESS in 11m 18s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/6754b17bea3847d0a50f5d84697ea634 : SUCCESS in 20m 09s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/591d0c5cd4554a0395665d048a49d6ff : SUCCESS in 16m 48s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/e8b1b9425911468a84fcac69bcbf1ea3 : SUCCESS in 14m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55c23f37f79948659f496bcab6a589d8 : SUCCESS in 16m 08s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/396d188edd0d49e08f69ba6b7a3fc600 : SUCCESS in 14m 34s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/2d74bd89b0bb4c6f94cdc711e20f5ec4 : SUCCESS in 14m 24s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/43977020ca1d4c5f980a42232d6df8fe : SUCCESS in 16m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/91b25c5ae10944538397ab28c26aadb4 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17aae2a4c246499dba365a702b4d9bbe : SUCCESS in 19m 35s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/85e768bceee948e4b4f6caee0bdc83fc : SUCCESS in 20m 27s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/e9e6171ccebb42cc8667ca4a3bf3df80 : SUCCESS in 19m 39s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/9ecffc8f3f454f3092336ba2ef332f71 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4bfab0214b376b2eceb15a672481a713ac04b7b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 21:28:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dfa4d149437a9aff3b2577e7145c23c323178c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 21:55:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fed70dedcbd64b08860ec9e339c88fc4 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/31adff556f26401ab6beda8e8dd7a909 : TIMED_OUT in 21m 22s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/1c9d992b21d44128867b8befe3c50926 : SUCCESS in 15m 48s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/45b9accde2bb4b45973c665227478a31 : TIMED_OUT in 21m 37s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/7639bbb5ff9c4fc5af65bd7d4c1c34eb : TIMED_OUT in 21m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b7483d0e3814f4285aec51cff4ff9f9 : SUCCESS in 15m 37s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/007d8b7176b34d25be25e5529c3a47ea : TIMED_OUT in 21m 35s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/2c8ea08b1f994a5898927618b98c8e5e : TIMED_OUT in 21m 11s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/32167671bbfd40dfb86992ff2b236297 : TIMED_OUT in 21m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/933a4de8122e4e1daf393cf79ef33d87 : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c195da7bedf54e129903e30a239c0f3f : TIMED_OUT in 21m 26s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/f8d2aba3a59644a497f4c8f05e86350a : SUCCESS in 16m 36s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/3f5acb75b30848548c251a16fd3f62c0 : TIMED_OUT in 21m 24s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/50c174b6911d4934bfe5e3daa37297df : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2ad107768b9c8a3cc472f306965c63c2e612d6d0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 23:45:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04c6234a150870949b1488b1b82c726a671588a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:12:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e83eff64570b4649b5c83f9f1835a3a6 : TIMED_OUT in 21m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1940c5de33b4461b935e484ada253b3a : SUCCESS in 11m 40s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/107e519b213441dbb931429db733979d : SUCCESS in 15m 55s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/5c65ba0899204c5389edde2a9ea646b3 : SUCCESS in 21m 06s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/127388b5e1a34f28bd2ffda4382ab146 : SUCCESS in 14m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/089a02e6b6db4d18897016d6f51bfb89 : SUCCESS in 16m 15s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/99bffd6b1be54766bf411b38bed8f493 : SUCCESS in 17m 30s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/08366f7fe5954a18be672c75dd2c6b6b : SUCCESS in 14m 15s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/1ba77065d5c249fe8e97524ecf976d63 : SUCCESS in 11m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/575a0927232f45308bce2fc945d7e388 : SUCCESS in 15m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe7bd02a95942508a9aff196fb75532 : SUCCESS in 11m 28s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/b82717a2075d4bf3a95bb5803bc6ab81 : SUCCESS in 17m 24s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/a8dd7cac7227498486c28dbcbfa6e451 : SUCCESS in 12m 47s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/0093f2d1801941eca82759068cee7c18 : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"849de68547b49b9fa3fdff0924c73f6b09827021","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 00:32:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c026e2fcf36fec136122cd184c7c352c2da918d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 00:54:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ea583cafb734e9baa5be2d92491988f : SUCCESS in 21m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a5aaebe6834044a1540a057a1b1c88 : SUCCESS in 20m 52s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/c4e1e55b40924c2295e6522f79d1c26d : TIMED_OUT in 21m 25s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/b1373816aa314d38beb2e96d45ba5b7b : SUCCESS in 16m 15s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/b2b2b67aeaab4d97b768ddfa97939d94 : SUCCESS in 20m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d86cc839c6d472fadeb83ea2782ed1d : SUCCESS in 16m 05s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/60fb67776ad140a9a208c7eca0c17a0e : TIMED_OUT in 21m 21s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/749cfe7485364a00978cb0610337be3b : SUCCESS in 16m 41s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/4c9ad43912e74ec2959027260534d3d1 : SUCCESS in 20m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44384aa1945d4f95a4f6ccfe086af3b4 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4ccb9eb6d41a4b658b70d8d3afa1280b : SUCCESS in 15m 29s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/9693f591e608452fa8516515385c2ea7 : SUCCESS in 20m 54s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/f07441aa38eb4152af05492472da3526 : SUCCESS in 20m 20s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/7e9b354b678b48eb8415e32786d02bff : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":2},{"id":"018982d767858da71be3c99ff6cc82cae5c14baa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 01:03:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"253591a8d90741a3079cf0a5e081367787f42dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 01:05:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec97d5defd9b3d76d026126383cf8f5a463ca535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 01:32:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4cacf27b89e44fd788d9c3a3482141ac : SUCCESS in 4m 28s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad8bc9145278470e8135ca22696a8b5a : SUCCESS in 6m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04620df212314ec58f6dca8c7e31547c : SUCCESS in 6m 04s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/285cbaae97e04d8a823c252abd94c7cd : SUCCESS in 7m 18s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/390da68323454675847690a7a32b8c90 : SUCCESS in 6m 02s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/e6ba863025b345b7966fcdefa378f3e9 : SUCCESS in 5m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c352a3d534f461aba76f04ab6878853 : SUCCESS in 7m 20s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/781a85cf31ea48a2a9f48d89f52d6816 : SUCCESS in 6m 16s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/71e4c856ae9342aea4551cf6dbbf145c : SUCCESS in 6m 26s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/f6bca2ec02ae43c088be714cfd52fd8a : TIMED_OUT in 21m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d866b4ed33e142e49c43f3546f9d3741 : TIMED_OUT in 21m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74bb0c95372045aa8b5482dba2693f2c : SUCCESS in 6m 15s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/f4477f4336424f258b2032775fcd1fc8 : SUCCESS in 6m 44s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/103533d3458d485ea71db6da76b8418d : TIMED_OUT in 21m 08s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/6af2d8462b034cefbb842bb6cf1a14a0 : TIMED_OUT in 21m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0a7b915cd5aa88673b7e3c7d1ca2d1a3438b5eab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 01:34:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1fd67871e2c7e2a019e2f967c161ca92e6bc5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 01:56:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f05a1eff949246e9a281dcb8cb10a25e : SUCCESS in 6m 08s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cface12910c54dfda69d1cabf3db15ca : SUCCESS in 5m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c2bf16018e240fab835d4d3017a367f : SUCCESS in 6m 02s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/2cdb88dbfeca4c9085c3cee8d9c7f571 : SUCCESS in 6m 01s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/e8e7470366374f1ea1a042471b490732 : SUCCESS in 6m 38s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/0735c1cd04344005924379e76e6f1576 : SUCCESS in 5m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85b680c7710c4bc9a387040142b76596 : SUCCESS in 6m 05s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/2dab0e46dca94d56b179aa6c3735a481 : TIMED_OUT in 21m 15s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/835601a06b4846099e67d5b24195e4bc : SUCCESS in 6m 01s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/29faba529eb84c8c90dea6983800546e : SUCCESS in 7m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a9543f13ae1a4397b3cc161ae4621796 : TIMED_OUT in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96318940a6884dc69a90908a6f00c784 : SUCCESS in 5m 58s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/5eb8418d4a0149efa360a59109d71741 : TIMED_OUT in 21m 12s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/75fced94fa2445d09131ecfebb8edbb9 : TIMED_OUT in 21m 08s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/12c2096baeb443a888737cfd0e20f06e : TIMED_OUT in 21m 11s","accounts_in_message":[],"_revision_number":3},{"id":"3a4d105a7c243603e64e45eadf6f10d365b7dc06","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 04:23:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"faa990b4251f89c96ec5caba809215716a63a283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 04:46:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64896d152e844146b85f8e7c94ac4d1f : SUCCESS in 4m 15s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00d3a72a2dc545e18074372205f8ae54 : SUCCESS in 6m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6e692d6ccc241e3bb1fd15293b40e14 : SUCCESS in 6m 02s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/63b0318ad17347fbb3cd59621d4ebe9e : SUCCESS in 6m 00s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/b995e8451922490ea0bfc6fe60bb7ef6 : SUCCESS in 5m 58s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/68ec41032709497ba4f2b6f2e9f5b6cd : SUCCESS in 6m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c37bedc71c754bc096fc3249f4b136b9 : SUCCESS in 6m 04s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/485461d01a504ba7b1557942dd0f487c : SUCCESS in 5m 56s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/ad21720c55de47bdba71ed58ad27b625 : TIMED_OUT in 21m 28s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/902fc79b8827470993d9cb422207baeb : TIMED_OUT in 21m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd04a6c90831421d9adf155ac1195379 : SUCCESS in 6m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d08bd518d954b6986c8ebc616e42030 : SUCCESS in 6m 04s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/c3e1318abb6f4c99a1d2777bce3db87c : TIMED_OUT in 21m 09s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/2c0fbd14c9f14b23b388b7489ee2bdcf : SUCCESS in 6m 53s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/c17b78136d5047cab1d2c71bdb8715c6 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"b6c32cd165d8e47d5fad57c5298278b1d0016e6f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 04:48:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8dbb9b183af9a50aff7c59f716a32208abe9df20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 04:48:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"04c59ca0da631a231b4acfbfb662b46c8f10859c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 04:59:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/509b796823a043558563efca00208fed : SUCCESS in 5m 24s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f79ae7b78f0c48fcabd91800c448d799 : FAILURE in 4m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc8ee84b21fa418aa1e89622b05bf776 : SUCCESS in 6m 58s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/7dc5c21c4dc74e7a9f526806f77c8206 : SUCCESS in 7m 05s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/ac3d91ad31be446cb7f3d6952c15ee2d : SUCCESS in 6m 22s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/e6e0603aaf6a4b9d9aa3c0816ffe40f6 : SUCCESS in 6m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a293bb10a5a4b5296fce0eb67d24f47 : SUCCESS in 6m 33s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/a4c569f807554c5487aea8bc66dd4db2 : SUCCESS in 5m 16s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/07ba0c2d8f2347ac8e562736db7e56d0 : SUCCESS in 6m 03s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/2450511dffc24e7ba14ff4bb4247e9ac : SUCCESS in 6m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd66ed343aa4251be06f43d2550ab0b : SUCCESS in 6m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1de0ac3d4bf147438263c2307b50357e : SUCCESS in 7m 04s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/68541dc0763a4184987261f69281c42c : SUCCESS in 5m 58s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/3a62cab20e7f414494af62d48774b7cf : SUCCESS in 6m 33s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/49e2be55bcd44aa99e695535579b71e4 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"3a324c9917e4215d21127b08b1a5d3b38a708ae2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 05:02:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aed4ef26f5433f5d0cf5a66111135eeae3e68f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 05:13:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/332de1fabecc4e89bc6c59e1bca3247c : SUCCESS in 5m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/565287d66fb34a1281988082d5ec1533 : FAILURE in 3m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4857444fe0a44f5ab7e79b564c683ba : SUCCESS in 7m 21s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/a0f10327acf04f05bd1ab50615990309 : SUCCESS in 7m 18s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/46f83b24238f4035b054c57b948a8c7a : SUCCESS in 6m 59s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/434d43f4f86b4ec79a07f5d0f6aa8f26 : SUCCESS in 7m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3a991d078c514297ad07ecfbf27b09de : SUCCESS in 6m 40s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/d47576354b5648c8906e710359fa7c4d : SUCCESS in 7m 03s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/7e18f6fe8be14bfeacb3992152758f96 : SUCCESS in 6m 31s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/2bdd66a95e134a48a03646a9d35ca00f : SUCCESS in 7m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/279235d806bf4d22b36d69a7de79d3d9 : SUCCESS in 5m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5da1486125ba4a45bac416004dac71e7 : SUCCESS in 5m 58s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/a7f64b48b0444e2ab1338aae71378a06 : SUCCESS in 6m 30s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/47883b55f68a46e882dee42b0793cd8c : SUCCESS in 5m 49s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/ce5b47e3288843f0b497db668a03d99e : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4},{"id":"d6755de301eb4cd9f44c326f9882266fbb21c909","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 05:13:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d3f73764bc1a0088b2da71735c579937d1a857b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 05:25:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2bc93c35bdeb468f8a899bcb798146de : SUCCESS in 4m 43s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15f0d92165364eeab85c124bcbec0fad : FAILURE in 3m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9982270c3c3749a2b6f868975513035b : SUCCESS in 7m 34s\n- swift-tox-py36-2 https://zuul.opendev.org/t/openstack/build/187353105864477a953251fa68f94875 : SUCCESS in 6m 30s\n- swift-tox-py36-3 https://zuul.opendev.org/t/openstack/build/3529f88d03c142ef92f26119426b12a8 : SUCCESS in 6m 19s\n- swift-tox-py36-4 https://zuul.opendev.org/t/openstack/build/37f0012a599b42cc81ec0e463fe9d1f0 : SUCCESS in 6m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b6abcb0df5f423495ae9a23b3a8e9d1 : SUCCESS in 6m 36s\n- swift-tox-py37-2 https://zuul.opendev.org/t/openstack/build/baff8d54763b45a8a6d7fbbd1cf3b0f8 : SUCCESS in 6m 23s\n- swift-tox-py37-3 https://zuul.opendev.org/t/openstack/build/0e50c1c88a8c43c690a9d73b2acae87e : SUCCESS in 6m 09s\n- swift-tox-py37-4 https://zuul.opendev.org/t/openstack/build/c9f700dfb4e04a7cbdd5c8951c67aa28 : SUCCESS in 6m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3f2286e9903845f28f13fd899b0b40d0 : SUCCESS in 6m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84fce4fd2bd541e38bd96dc7e4e75134 : SUCCESS in 8m 59s\n- swift-tox-py39-2 https://zuul.opendev.org/t/openstack/build/0ebcf8abf07f4fe091e92793e85ef92b : SUCCESS in 6m 52s\n- swift-tox-py39-3 https://zuul.opendev.org/t/openstack/build/55368102427e45f7a52a5c1419bba6d9 : SUCCESS in 8m 05s\n- swift-tox-py39-4 https://zuul.opendev.org/t/openstack/build/21432393f1924a34ac0055afda919692 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":4},{"id":"81df1152eb215c2794f6379b5afbd0604f50f0e1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 20:56:04.000000000","message":"Abandoned\n\nWe\u0027ve got a plan for a fix! https://github.com/eventlet/eventlet/pull/743 merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfc71b5e7c0c3969f1aa7faef16eb77c5429a362","revisions":{"27e4b4fb2a3f44bd2eb1e9579f094ed6fe7b5326":{"kind":"REWORK","_number":1,"created":"2022-01-29 05:11:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/826950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/826950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/826950/1"}}},"commit":{"parents":[{"commit":"c1d2e661b1a448d5e4d00a8e41e12803a93b364c","subject":"Merge \"s3api: Allow multiple storage domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d2e661b1a448d5e4d00a8e41e12803a93b364c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:11:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:11:34.000000000","tz":-480},"subject":"DNM: try to repro CI test lockup","message":"DNM: try to repro CI test lockup\n\nChange-Id: Ia62456be136f4e32288dd19d8061708e9e7b7cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e4b4fb2a3f44bd2eb1e9579f094ed6fe7b5326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e4b4fb2a3f44bd2eb1e9579f094ed6fe7b5326"}]},"branch":"refs/heads/master"},"3502e4266158e1c7adf65c3fe80cb1716b2e5add":{"kind":"REWORK","_number":2,"created":"2022-01-29 05:33:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/826950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/826950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/826950/2"}}},"commit":{"parents":[{"commit":"c1d2e661b1a448d5e4d00a8e41e12803a93b364c","subject":"Merge \"s3api: Allow multiple storage domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d2e661b1a448d5e4d00a8e41e12803a93b364c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:11:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:33:14.000000000","tz":-480},"subject":"DNM: try to repro CI test lockup","message":"DNM: try to repro CI test lockup\n\nChange-Id: Ia62456be136f4e32288dd19d8061708e9e7b7cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3502e4266158e1c7adf65c3fe80cb1716b2e5add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3502e4266158e1c7adf65c3fe80cb1716b2e5add"}]},"branch":"refs/heads/master"},"c0bdb5ca954cbe5ef17d7d8f5661edb8a1d28411":{"kind":"REWORK","_number":3,"created":"2022-02-04 01:05:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/826950/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/826950/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/826950/3"}}},"commit":{"parents":[{"commit":"c1d2e661b1a448d5e4d00a8e41e12803a93b364c","subject":"Merge \"s3api: Allow multiple storage domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d2e661b1a448d5e4d00a8e41e12803a93b364c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:11:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-04 01:04:59.000000000","tz":-480},"subject":"DNM: try to repro CI test lockup","message":"DNM: try to repro CI test lockup\n\nChange-Id: Ia62456be136f4e32288dd19d8061708e9e7b7cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0bdb5ca954cbe5ef17d7d8f5661edb8a1d28411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0bdb5ca954cbe5ef17d7d8f5661edb8a1d28411"}]},"branch":"refs/heads/master"},"bfc71b5e7c0c3969f1aa7faef16eb77c5429a362":{"kind":"REWORK","_number":4,"created":"2022-02-04 04:48:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/826950/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/826950/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/826950/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/826950/4"}}},"commit":{"parents":[{"commit":"c1d2e661b1a448d5e4d00a8e41e12803a93b364c","subject":"Merge \"s3api: Allow multiple storage domains\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1d2e661b1a448d5e4d00a8e41e12803a93b364c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-29 05:11:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-04 04:48:47.000000000","tz":-480},"subject":"DNM: try to repro CI test lockup","message":"DNM: try to repro CI test lockup\n\nChange-Id: Ia62456be136f4e32288dd19d8061708e9e7b7cb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc71b5e7c0c3969f1aa7faef16eb77c5429a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc71b5e7c0c3969f1aa7faef16eb77c5429a362"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
