)]}'
{"id":"openstack%2Fswift~827058","triplet_id":"openstack%2Fswift~master~I1843d67190d0d798f31b7450fc36d3e149dcc956","project":"openstack/swift","branch":"master","topic":"bucket-ratelimiter","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-01-31 17:48:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1843d67190d0d798f31b7450fc36d3e149dcc956","subject":"object-updater: count deferrals even when no deferral queue","status":"ABANDONED","created":"2022-01-31 11:22:26.000000000","updated":"2022-01-31 17:48:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"acf3922f9094cd1d917dc7e30f273146a1106b3b","_number":827058,"virtual_id_number":827058,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-31 12:29:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-31 12:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e83d25d76091646860587a38a57b59e18085902d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-31 11:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc315a6bc28b4c846c8cb6926c36592c965ff136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 12:29:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c7cdd47d3f4a6e9aefb79d1110be02 : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/4ef006eb4121407f9e7b12ef518d8137 : SUCCESS in 52m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f0372c306134f6695915b1a5c3cf2c0 : SUCCESS in 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1df2c9ceb59e4b4893efff7b7c3e9a66 : SUCCESS in 44m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ec3d0821c7cc443e96caf60f12a1b15c : RETRY_LIMIT in 5m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1e85c9ab0ae442c5a7172922416f873c : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d7d44ac270f84d10a7c227fd46d4c625 : RETRY_LIMIT in 6m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67633a59306646bda563a5e12cd967b1 : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3aaf34d42f204c3d988ab5d73043e897 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77ace755bd2c456d8f1e7471c9708e50 : SUCCESS in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a10e922c814746efbc56fd1a38bcfab9 : SUCCESS in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f48dbf5364b94256a120071c49f054a2 : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b11799ec57b84df4adecd7c1e933e5f6 : SUCCESS in 19m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0dafefc73c546419f801ef5246bb85c : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b694f608174142999e23d20b94ab06cc : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c34b97fe8fcd405ba8f176a7e763c6ac : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3df43e9b6c94fa082b0423fd7c7184f : SUCCESS in 19m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c328b65cfcb74205bc5177bffd5d5648 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb440df351764ea595065e3e34eb25b0 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/287f47e653594ce5b8823ea2d7acfa69 : SUCCESS in 20m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/758d5b415b7440e6ab792aaae150cf04 : SUCCESS in 1h 02m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/96a00478974e457baad1bca0eaf69142 : RETRY_LIMIT in 4m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a7d89c6fcd64916a2d0bf96585438c5 : SUCCESS in 41m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6a124a049b848f99744befb3101048e : SUCCESS in 55m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/981190c811f948ee980e06696429c741 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744fe3d401d24b6498839c717e3be9ce : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4801f3aee1e4785813e8a89ab4355c2 : SUCCESS in 25m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed1345b5a9eb4e7796beb7b2108d5465 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d34732ccde3e9da226407e5bdd97646d14d4a0b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 12:32:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af312ca5a1c9445c8559b315a784f528 : SUCCESS in 55m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6acbb7fb4d8448c7a204d04bafa73bd7 : SUCCESS in 59m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/1226f8cd3234472bbd37be3b06988d59 : RETRY_LIMIT in 2m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8803e1e7193c4feba7cf29121e1e916f : SUCCESS in 1h 05m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2c570bcfff74c33aca4919c471588e2 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":1},{"id":"acf3922f9094cd1d917dc7e30f273146a1106b3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-31 17:48:33.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4c59c63b93eef4c8eca1c3f119c9c74f2f811a6","revisions":{"d4c59c63b93eef4c8eca1c3f119c9c74f2f811a6":{"kind":"REWORK","_number":1,"created":"2022-01-31 11:22:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/58/827058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/827058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/827058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/827058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/827058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/827058/1"}}},"commit":{"parents":[{"commit":"4572c4c5f8cb281ceb465fa3811d5b9f9992c895","subject":"object-updater: defer ratelimited updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4572c4c5f8cb281ceb465fa3811d5b9f9992c895"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-31 11:21:51.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-01-31 11:21:51.000000000","tz":0},"subject":"object-updater: count deferrals even when no deferral queue","message":"object-updater: count deferrals even when no deferral queue\n\nChange-Id: I1843d67190d0d798f31b7450fc36d3e149dcc956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c59c63b93eef4c8eca1c3f119c9c74f2f811a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4c59c63b93eef4c8eca1c3f119c9c74f2f811a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
