)]}'
{"id":"openstack%2Fswift~827845","triplet_id":"openstack%2Fswift~master~I7692860e3b741e1bc10626e26bb7b27399c325ab","project":"openstack/swift","branch":"master","topic":"p-deprecate-set-statsd-prefix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-02-08 02:07:25.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-07 23:56:47.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I7692860e3b741e1bc10626e26bb7b27399c325ab","subject":"Deprecate LogAdapter.set_statsd_prefix","status":"MERGED","created":"2022-02-04 11:18:55.000000000","updated":"2022-02-08 03:01:37.000000000","submitted":"2022-02-08 02:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"827845-p-deprecate-set-statsd-prefix","meta_rev_id":"5474fbce6af97136fd6842716511cc63ab850b87","_number":827845,"virtual_id_number":827845,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-08 02:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-07 23:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-07 23:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 12:31:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-04 20:33:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-02-07 23:56:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f1b5755990e6f8200b4c65316f627bd7bf0596f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-04 11:18:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"830da67b38ab2d02aa6502a24e70fe25f572d7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 12:31:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6019716d57444425b3e8952686493ff3 : SUCCESS in 10m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/be96d45e9a4b4f5f985b42e2864afa5f : SUCCESS in 56m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f88d785006b84b3f9b15361b14e74b89 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f1872c6199a4abbaf493badba2bcb5f : SUCCESS in 50m 12s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dbfa62fc208d4a32a42d1c3f00edd30b : SUCCESS in 25m 09s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/36a9185c7d554791a31023ed22bc8afc : SUCCESS in 29m 02s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5b32f6eb4a7e4861b7343e7bb23f9c6c : SUCCESS in 24m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17baedf07c8c4d90ad89a3f14daa8276 : SUCCESS in 12m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc2b2fbe8a7c45e4ba42aabd0c11d20d : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/766c3010a4304ef8bfedddbde0cbd2a6 : SUCCESS in 24m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7673ed1ec0bf4197812b161c15f0d14a : SUCCESS in 12m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6f410c8f494dc1b668dfe797c3b6d0 : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06901b5798ae4feea22c5ee501908660 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e7d4d17c1d742b4b977e0b18fc9f61a : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/411099a8a1724cad8fc8c6a7083e510a : SUCCESS in 23m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e6b834d8318b4d87b4334eb1cd82d88a : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6bfe8d941b2a47dfb0302c2978d515b3 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cc7fabf0ecd443e98e97b3a3c4680c89 : SUCCESS in 18m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ddcc94cc71054a3eb947749c8a46c186 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3071cc9855c0424d8123f7dee0838585 : SUCCESS in 21m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d3f8ddbd6c14bbaa5f31f62a45ed7a2 : SUCCESS in 51m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7dabc955c2524262b3a070f2b1ebeacf : SUCCESS in 53m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bcf38c671604c02aaad012921aa4125 : SUCCESS in 41m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfff5664d9a24a4cabd43459cfffed6f : SUCCESS in 41m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fd927210ca9414ea9f7172bce06f87e : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ed42aca1d44b6d81537713582aa176 : SUCCESS in 7m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/616f595d3b96440091da020c8d08e3a6 : SUCCESS in 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a426cb5ff31477d9c55d906010d49f4 : SUCCESS in 48m 27s","accounts_in_message":[],"_revision_number":1},{"id":"10eee6b241d4c785f4b3edae17f190f63b08879f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 16:48:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d059500c6ca4e94a2357c6bed66f918 : SUCCESS in 1h 03m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6683b3acf1c4547bde5db888f90d480 : SUCCESS in 1h 06m 03s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/89ee952bdd0346a4b0d70917cb85e804 : TIMED_OUT in 3h 02m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c97d5acca9bc47d29f757e62d5c95009 : SUCCESS in 58m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/075392ab0c0f4a38a936fd87d2e715a4 : TIMED_OUT in 1h 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bbf518f653f2547f738af0a717e049a38da0b99a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 20:33:07.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0ce121b53856488d82e6837e7e91e8f226b37eb5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 14:08:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3fa4d4f60833cfc16abf60100f27a0a3562f4e59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 14:10:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8cd6ba0014751c141d61b7111d4de4067378ce13","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 14:41:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e1439e80b01045e568c46d5d61cdeaefff544f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 15:25:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c5b611c988408fb7c79222a3d2b05a : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/94c7d288eed44f938bb02302efac56c6 : SUCCESS in 52m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86730b795b3745a080b524cfee0985c5 : SUCCESS in 42m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52d85d3ac11b4166b65df09865320c48 : SUCCESS in 41m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a38abc92ef0346949f3ffc1519891785 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d42e4b80eda141cfaac7121175568164 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4bfb1be25f1b4a6c8d1b9edebdd708ad : SUCCESS in 22m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0eca04e0737451c81ece869f90250c1 : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6cfd95cfd39946d8a8e2e563d6b64cfb : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/990bb4d38dff4b428bd3e63ef89fc3ac : FAILURE in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24a60f23d8654e5aba27f79b03300f78 : SUCCESS in 19m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f9080ce5a544c069803e1efa0a27836 : SUCCESS in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a42223f57ae42d8857f30f68609d74c : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4ad1803d631431bbc785ab3fa92d912 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c217e19045074203b39810f95a26319c : SUCCESS in 16m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c48224a3c191431483f3025fdefbad13 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e91d20f19ae64b229089154fe40eb17b : SUCCESS in 17m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d1c254f3b8f545d188ef2b5cdec3104d : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf53e3e0c23c4304814b73fe801f0b75 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2926e0c3ea654fdbb0a27ed9f9bc1c84 : SUCCESS in 20m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6bbfd052a47e4729b1015be33768d0b6 : SUCCESS in 51m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/74c9cc13c3d64cca97a570523061824c : SUCCESS in 54m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/353019e2c0b54c939eb0a41bd0f65e48 : SUCCESS in 37m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9da36796f1ee4b68a99530c76b5c88ab : SUCCESS in 39m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959caf9473174f6b93457525bc39d34c : FAILURE in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42764c2a0379409bac626f9436077722 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e68afd75d30649ab8c09afa7abca3fff : SUCCESS in 26m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/433bd7c7d999463e96f3f9d9705c04e2 : SUCCESS in 1h 08m 05s","accounts_in_message":[],"_revision_number":2},{"id":"d9083a819d27eea79c551e676a812b7eccf7dc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:41:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0db7440844646a7ad4529a0ba089438 : SUCCESS in 1h 06m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61c7d849c2d14a43ae90b43c9bac2fb2 : SUCCESS in 1h 10m 04s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/903c42770b7741a7b3bcfb0094aec50b : TIMED_OUT in 3h 01m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f9a0a5f9b6845fbbc320d0bc551d161 : SUCCESS in 1h 16m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e97902a2fb134781987bf2e50bfb37d0 : FAILURE in 1h 03m 26s","accounts_in_message":[],"_revision_number":2},{"id":"82b1f93e7e1e80e6651566b30a2baf907aa36637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-02-07 17:46:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"881b3bb9e26086ec71ebbf5289991e1e5cf0a5fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 19:22:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06645e0289914ac3840137ffafaba0c9 : SUCCESS in 17m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f0b43077807e453faf730225212f29b4 : SUCCESS in 38m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/622e57add74244a08619249a0d9bfa96 : SUCCESS in 47m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08ef4e79c8004ac1899cf7ccb1dfe45c : SUCCESS in 49m 52s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3c4cff8b8fb54423a0fc2c4eff0a80b5 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2fdf85711bec45108a787234b901b39f : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/342a1159a8234a2fbf91ac9c253b2f09 : SUCCESS in 16m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77b10e5bc3024650b29166c2f51fba55 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82c8ebdbefb646cc9ad19db9eb67eb99 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ccbffc3eda7429a9a8090d447a54ec0 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5372c370b34441cdac799ed4ffedf934 : SUCCESS in 12m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f0bd6c273af4df49acde2b61fedf2ef : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d15caea2df304c2eab6381b677495be9 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ccbef705f4ef4b7ba5cdfad65e6f27da : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0499d538552d46be8a82d4ef0cc5603c : SUCCESS in 14m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cad209d631eb4ed8ab7840c29d7f0f42 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e85202259de46f680993e3b8c5396dd : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e599ab6eaccf4ed798d9ad57146a40a8 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b2a5ec96a71846079519556e670663c7 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ee4c3e14fe04b8aaea8550ad4a03dec : SUCCESS in 21m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/23d7b092fa874d3888da4cbc4a188abf : SUCCESS in 48m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cba8f7c7970e4416b9ab5938deaa655b : SUCCESS in 1h 03m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1236d317a7fb4e5b91afdbd3025b93e5 : SUCCESS in 38m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63b0fd5421ab422298ff4dcbd6c2ea0f : SUCCESS in 38m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ed14bd1964a4e358fd83c810b73029a : SUCCESS in 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6edeb60a984963a32883b8c5e42a55 : SUCCESS in 6m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/adcbfadf5c764c40ae52bd9058c51427 : SUCCESS in 19m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b0c1e002a8f4660aa8681e96a7f0e90 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":3},{"id":"814fe1f1dad0d988441f46e769761cf9e1538312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 20:51:15.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ea24e6991c7455280aba309ec6c6722 : SUCCESS in 1h 12m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d3df32848e644f48282c1ee95b0858f : SUCCESS in 52m 57s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/eda9177ebb874bdfb4d90b56f97b8e74 : TIMED_OUT in 3h 01m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/18e2c977408847daa12113f75202a435 : TIMED_OUT in 1h 31m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/635360e4f3f5458199d44afd138fbe4c : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":3},{"id":"90eeec75d1b3edaa29f453d5f0ca5f95e078bbd1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-07 23:56:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0963d194fe2be0cd70f57b06fda8e06d9e3f75c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 01:08:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce52ff44f964dd282844ae26c1e9feb : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a22a7f02f64749ab9503c72d6a863795 : SUCCESS in 37m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09435a9f6e354d84a1d388cd64fcb8cc : SUCCESS in 34m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63b099bde8b7416baac8a77cf307cc2b : SUCCESS in 33m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/823a8b19de4c4e968071a7008b651819 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aec87f54c38e425d90e53ffeef6b375a : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6854beb182834cf1abaeb94061ea327d : SUCCESS in 16m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b350023e3004b9f9028477be3ef0ade : SUCCESS in 6m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5ec5ac81a83f43289395c98a977a9254 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4ddaa0d24c04caa9184a6ab629eb20f : SUCCESS in 12m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52af9ab08c1643f9b8e62f998c7b0116 : SUCCESS in 11m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/396a1512e73c43589484af619f7639dd : SUCCESS in 12m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13083f32b5624ee2beff9607ee6d3559 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85ad6c760e4f4bdbbfdb4989e4dfa8e1 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/290ff0ffda7d4213a99ee4419bcc8453 : SUCCESS in 14m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6684b4695874075a750c911ba0056ba : SUCCESS in 12m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eada1615c106438e8f160b74d5f38ff7 : SUCCESS in 12m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e488cabe5044d1e8344b0f9cf636913 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1a826c63606425b9b85455cb405b6e0 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/54c7c2343d8f432093380eadf3b088f5 : SUCCESS in 14m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6b326b6d52b74d0db7a127d2b185e458 : SUCCESS in 51m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/69ad93e41c644e0b951b6da1dcd63a98 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1095cc5ccdf8447fb22a8cac5df91fbe : SUCCESS in 45m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d14ba5c3f0a408d8a7654897b5929c8 : SUCCESS in 26m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d990d3d7190c4cf1996b1b9924303e40 : SUCCESS in 22m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29827089e7e7493aa9d4c4060a091e15 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90344837b1df45708d8d0c8ca3944bd0 : SUCCESS in 27m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c534a4b02424a2099ff424cef6727b9 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":3},{"id":"54a460463c2edfe4e16fb5a7239f470cab0d779d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 01:09:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dcb42274ec0eeb47b06cca24688335dde64155c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 02:07:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e321901a51f4e55a1ca998059f3358c : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/58142abba6d643a2bd162ddd82c59e2c : SUCCESS in 48m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef2dba11c7f94e1c85ca0a8b4dbab5d5 : SUCCESS in 43m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8e68b517f0b4ad6888f0e0f1e005b47 : SUCCESS in 44m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9d2c90cc94774b479b3585c3ae224363 : SUCCESS in 8m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/83b251de96e445c4a70b15ccfe3f5586 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/72a27d563c354f218f7dae7859c50bab : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3bde37ef298d47d8a4965129671ee967 : SUCCESS in 20m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5bcd3fe6eb7c4e14af872d58266e133b : SUCCESS in 20m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e4026b23c6d24ccbbb85699c6514524d : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5905c3a700e49c9b86d23f0a9d8e375 : SUCCESS in 24m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e66c360f75944af9a4c7d913740db10 : SUCCESS in 25m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98beaf04dfd0438489c5dd64a7c8d0b2 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e79f2e9faebc45a4962d3cf7425bea80 : SUCCESS in 24m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a59fad1ea7bd47b4bb9cdaecc04ab06e : SUCCESS in 12m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3d361163c834f26a8147eb22d635181 : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8248911daa0746ff8a958e995d1f697e : SUCCESS in 40m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/438724ddb4c04b918d28d0f1b5fe1d64 : SUCCESS in 56m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64908d00277f40c69f5c55a51e5df6e0 : SUCCESS in 26m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce2b04ba0af64153b1acd2f6d23f1b2e : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/264e580dea2e4fd48ce93e743b41a55a : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511bea7b049d4194b72bc6738b4b5dd7 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7650636a1b7847f7813d682596efd8f9 : SUCCESS in 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ceb953ee1e145ae88263544068d3e74 : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":3},{"id":"27b881c2a0052290d935a4c5bea9a2055c781fc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 02:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"129fe33781df53bb7bbbf2229749c14e5b109d0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 02:39:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7732592624a148dba4adfa9746ae9525 : SUCCESS in 1m 28s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c0a8a759897c4c269a4df40445884c4e : SUCCESS in 1m 56s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2acb6af2f91847389a966ff03e255929 : SUCCESS in 1m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5474fbce6af97136fd6842716511cc63ab850b87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 03:01:37.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cf2e89155cd4b21a718b7caa0bed0c2 : SUCCESS in 51m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f7a9faf763c4919b9b0d7a022f45fdb : SUCCESS in 51m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9445aa55cffc4d01aa15f0aa5f8db302 : TIMED_OUT in 3h 02m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/71da64a326d14c01aa0ebc4fec8ad426 : SUCCESS in 1h 06m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8bab264d4f74595b02e7751abd42d22 : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eda7d5fe3c9527ae20850b38f8b626782dd5efec","revisions":{"0b7b09ff4ff3a67adf251c6f83481ef3c5058901":{"kind":"REWORK","_number":1,"created":"2022-02-04 11:18:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/45/827845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/827845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/827845/1"}}},"commit":{"parents":[{"commit":"2a60c91f6fcf9ed165a008c8800e475e5f2fc777","subject":"Merge \"Make NodeIter require a logger argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a60c91f6fcf9ed165a008c8800e475e5f2fc777"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-04 10:58:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-04 11:13:11.000000000","tz":0},"subject":"Deprecate LogAdapter.set_statsd_prefix","message":"Deprecate LogAdapter.set_statsd_prefix\n\nPreviously, the set_statsd_prefix method was used to mutate a logger\u0027s\nStatsdClient tail prefix after a logger was instantiated. This pattern\nhad led to unexpected mutations (see Related-Change). The tail_prefix\ncan now be passed as an argument to get_logger(), and is then\nforwarded to the StatsdClient constructor, for a more explicit\nassignment pattern.\n\nThe set_statsd_prefix method is left in place for backwards\ncompatibility.\n\nChange-Id: I7692860e3b741e1bc10626e26bb7b27399c325ab\nRelated-Change: I0522b1953722ca96021a0002cf93432b973ce626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b7b09ff4ff3a67adf251c6f83481ef3c5058901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b7b09ff4ff3a67adf251c6f83481ef3c5058901"}]},"branch":"refs/heads/master"},"8072295869df773d1c66a8c826a7ff206ad73a45":{"kind":"REWORK","_number":2,"created":"2022-02-07 14:10:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/45/827845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/827845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/827845/2"}}},"commit":{"parents":[{"commit":"2a60c91f6fcf9ed165a008c8800e475e5f2fc777","subject":"Merge \"Make NodeIter require a logger argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a60c91f6fcf9ed165a008c8800e475e5f2fc777"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-04 10:58:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-07 14:09:45.000000000","tz":0},"subject":"Deprecate LogAdapter.set_statsd_prefix","message":"Deprecate LogAdapter.set_statsd_prefix\n\nPreviously, the set_statsd_prefix method was used to mutate a logger\u0027s\nStatsdClient tail prefix after a logger was instantiated. This pattern\nhad led to unexpected mutations (see Related-Change). The tail_prefix\ncan now be passed as an argument to get_logger(), and is then\nforwarded to the StatsdClient constructor, for a more explicit\nassignment pattern.\n\nThe set_statsd_prefix method is left in place for backwards\ncompatibility. A DeprecationWarning will be raised if it is used\nto mutate the StatsdClient tail prefix.\n\nChange-Id: I7692860e3b741e1bc10626e26bb7b27399c325ab\nRelated-Change: I0522b1953722ca96021a0002cf93432b973ce626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8072295869df773d1c66a8c826a7ff206ad73a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8072295869df773d1c66a8c826a7ff206ad73a45"}]},"branch":"refs/heads/master"},"eda7d5fe3c9527ae20850b38f8b626782dd5efec":{"kind":"REWORK","_number":3,"created":"2022-02-07 17:46:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/45/827845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/827845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/827845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/827845/3"}}},"commit":{"parents":[{"commit":"2a60c91f6fcf9ed165a008c8800e475e5f2fc777","subject":"Merge \"Make NodeIter require a logger argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a60c91f6fcf9ed165a008c8800e475e5f2fc777"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-04 10:58:46.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-02-07 17:46:06.000000000","tz":0},"subject":"Deprecate LogAdapter.set_statsd_prefix","message":"Deprecate LogAdapter.set_statsd_prefix\n\nPreviously, the set_statsd_prefix method was used to mutate a logger\u0027s\nStatsdClient tail prefix after a logger was instantiated. This pattern\nhad led to unexpected mutations (see Related-Change). The tail_prefix\ncan now be passed as an argument to get_logger(), and is then\nforwarded to the StatsdClient constructor, for a more explicit\nassignment pattern.\n\nThe set_statsd_prefix method is left in place for backwards\ncompatibility. A DeprecationWarning will be raised if it is used\nto mutate the StatsdClient tail prefix.\n\nChange-Id: I7692860e3b741e1bc10626e26bb7b27399c325ab\nRelated-Change: I0522b1953722ca96021a0002cf93432b973ce626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eda7d5fe3c9527ae20850b38f8b626782dd5efec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eda7d5fe3c9527ae20850b38f8b626782dd5efec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
