)]}'
{"id":"openstack%2Fswift~827902","triplet_id":"openstack%2Fswift~stable%2Fwallaby~If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2","project":"openstack/swift","branch":"stable/wallaby","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"34120":{"account":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"last_update":"2022-02-22 15:59:54.000000000","reason":"Tim Burke replied on the change"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-02-22 18:22:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2","subject":"Add FIPS CI jobs","status":"MERGED","created":"2022-02-04 21:39:13.000000000","updated":"2022-02-22 18:29:05.000000000","submitted":"2022-02-22 18:22:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"827902-fips-compatibility","cherry_pick_of_change":827901,"cherry_pick_of_patch_set":1,"meta_rev_id":"d229bd905bd394bf979c4099adcf4d1ad0256006","_number":827902,"virtual_id_number":827902,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-22 18:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-22 15:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-22 15:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-04 21:39:38.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-04 21:39:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-10 19:45:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-02-22 15:59:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"0ee8e3e39003a32b641f329ce14cc67316c44045","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 21:39:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"132eca8afd98ad1bb80d0abf5feb4f7e0694dd28","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-02-04 21:39:38.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch stable/wallaby.  The error was:\n\n  while scanning a simple key\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 514, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 515, column 1","accounts_in_message":[],"_revision_number":1},{"id":"e0480bedca650654e042b5968fd76fab106322a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 21:39:52.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch stable/wallaby.  The error was:\n\n  while scanning a simple key\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 514, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 515, column 1","accounts_in_message":[],"_revision_number":1},{"id":"e0c712c72dd6e638b4b92878f11ffb3311eb86e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 21:40:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e83bb765ebf2a7bc44050360f952ae060aa07a1c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 21:40:32.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6834abfe25a0655c0ddc8f3aa579e52d522b035e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:07:30.000000000","message":"Topic fips-compatibiity-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":2},{"id":"22201e9fae95f08c74dc35af4c86095e3c834231","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-04 22:07:37.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":2},{"id":"616a5e60b87a6d8ec922cdd1b7a85eddfb098f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 22:43:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d8aba9c02946bf93b6cb85f1efd5c2 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/301e44be1bf64c28a680a4a6f43c80bf : SUCCESS in 55m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f65c6e65a2204e73b7b982bd838f9e34 : SUCCESS in 45m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd222b589ee249ccb0924c88e7df1e43 : SUCCESS in 59m 00s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/077f36f0d45549a498be76103116b574 : RETRY_LIMIT in 4m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35cf5939fcc04c69ab042eddcf004ba3 : FAILURE in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d9e9516ef4d472484141c82817d62c1 : FAILURE in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a67c78c7bc34ca68a76902668300af6 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0e1188117f42cbb278f726edbc4f61 : SUCCESS in 16m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63c65413d7a241278d8549b1c913c716 : SUCCESS in 15m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3083c70ff6ce4b3a97656c4af693ec78 : SUCCESS in 14m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2df2ae0070ab470aa59309954cc38ab4 : SUCCESS in 19m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77cfd41fc2d541f8b9150cf679695bbe : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70caa8be364c4bbfbb7285fa5efb05f0 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89365ff4ddea43ea8e73d41ecc8b0b72 : SUCCESS in 19m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c4c3d4fcc091446e8d6564e7dd36ea09 : SUCCESS in 20m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9aeb44f0f7944afe87f8837a61fc9ee9 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a41aec7b5fb24153ab3286a418792805 : SUCCESS in 22m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02de624414714c3baf3931c6395e6f03 : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88b430c7ae494beab79deb09ece16381 : SUCCESS in 26m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5c999d314a34de28ba8c50720a8c7f9 : SUCCESS in 54m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b1c6329e2ec44f1e9445b64686cfbd68 : RETRY_LIMIT in 4m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/371ba4fe48994cac8adb5bc9343677d1 : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/398e59c6316e4db195377b4e0c49649c : SUCCESS in 40m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ca13ea89d4b4ec998c6c031eb5f0697 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9193df3de45741abbb0d5d48164ffce9 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d64c158598db4602820c0bbb96197662 : SUCCESS in 27m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9ca9291ca8d407ab5bb60b78b969a37 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3acacdf7b5d5f47d7bb67ee2c47bc4c3be579882","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-09 17:01:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89fef8c028cb20ea90ecc9d61987f164fce7423f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:28:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1899df3d67e4b7398ebfce69b95bd3f : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/08e59374b7044644b928cc5138ac4595 : SUCCESS in 1h 18m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e4c70b18da441c4bbe7c328f9243c1c : SUCCESS in 58m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9668a70e63b84b399eb46fcf663a0ac9 : SUCCESS in 53m 51s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/e044c92f3514451a9189191a49745cf2 : SUCCESS in 21m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c32de810ed4546d4962c435c0a059eef : FAILURE in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/919f05c76e7c4e459fd3579bebca4508 : FAILURE in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65728483766941d98f1e512bfeb23c3d : SUCCESS in 23m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c478f6af379473794f33ccdd85375be : SUCCESS in 23m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/491ae012544346749457e274db7c6bad : SUCCESS in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ab5442de2d7c41fbb5d7c55653ceef44 : SUCCESS in 24m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c433b27b20e424b8e0e91f46ad67f79 : SUCCESS in 25m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/099b378665b74eeda0dea67a5b60aec4 : SUCCESS in 33m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0aa3dab62a4409a88c59774d2367520 : SUCCESS in 21m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b0b7d75e5294d3a86f6b94ba2319c6e : SUCCESS in 19m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ed6070963fed43558cf4cefe65d5e74f : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f9e7604ad8994086826003b184c8638f : SUCCESS in 30m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5ad71da3f7f74858972e03cf8e923d85 : SUCCESS in 22m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2f473daf5a64e7ab7c6f7049ff1ae52 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3602d25045df472192f5d34e591be720 : SUCCESS in 21m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2ae4387a6e64654926ea330a2a1dab4 : SUCCESS in 52m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0a640d1da833494b97df605f2a59814d : SUCCESS in 56m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2ed77efc7bc45a89683c0aede6aebac : SUCCESS in 41m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/082c13c967eb4fd3b42f08dcb5e164e0 : SUCCESS in 45m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64db31c69bb74454806e20a1fbcd2fd7 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75189a5b0272470eb637da3975d7cfb8 : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4ef8dceccb94771bbb3fbf2213244e4 : SUCCESS in 36m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e50095f084e493f830e6856cbdd3066 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7245efc7dbf372aa46d67202d848e0c27bce8891","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 19:45:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b500f8cc7f78657a9e293961f3ebf41fcd867637","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-02-14 20:14:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"637e620bab5d954511d712b0774e0c9f53b049df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 21:14:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4974e3cda3cf4e1984c575c9bd7c42b8 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/63f38e4faa0843f88d1f9490944662d9 : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66d14802ce7444d89df95cf85d8db16c : SUCCESS in 42m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb3476cf9b3c46f88a0ee08c0bf38317 : SUCCESS in 50m 19s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/bb18fadfb3a8453fac45fbdb87171a48 : RETRY_LIMIT in 1m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26c21d9c87594dc4aac4e5795b2290a1 : FAILURE in 7m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d6885936f7c4d02bcf45524082fcc2c : FAILURE in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e004131327da43f8b0c7c5df70314ca6 : SUCCESS in 17m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd401ad5345a4ffdaec5f5abc4023aba : SUCCESS in 15m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1646f573c6d4dbcb95fe4807dfd4c2e : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/567487f5b5fc4a2bb74b78f44da4847d : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce1f8e4175754b628b2c4d4b62d3bf9e : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0f12cbb271d41c49dd3c6bd778871d1 : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5bb31936e9941c98edf1e8e0c091efb : SUCCESS in 20m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/163d5bbd52c5454cb6a5128e9be74ab9 : SUCCESS in 24m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/844e1f561d644844bb3b45e4a5834c34 : SUCCESS in 20m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/30ad9d0e2e854466aa3e3323bafa08c5 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/eb4e0ebd4184494d85e46ca54388c5e7 : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8100d0476d214451a50ca713ba35779c : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ebff043d6e24d63bb3bc86e681232a2 : FAILURE in 24m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3da1e4e863004efd96443c178d602296 : SUCCESS in 48m 29s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/98a0d6c050f8438f8367369a00b739d6 : RETRY_LIMIT in 1m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ce53f91ae43462b830805b5b5bad2fa : SUCCESS in 39m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa1b012af9ee4808805b8d82d3eedb4b : SUCCESS in 40m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/549693916f7f49e691023ccb74e95d2b : SUCCESS in 15m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42914ffb372a469598ede75a08d26068 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61f1464c791f49db807d5ccf25aab13e : SUCCESS in 28m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/389f5c3146e44888b180b717ec7b081c : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ec1369de712b3b5d6591175280e3a54e7dba25f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-19 18:31:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4a7770d9e71bdb58f93cda325376d1f0ba5ea8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-19 18:32:02.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch stable/wallaby.  The error was:\n\n  Job swift-tox-func-encryption-py36-centos-8 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n      check:\n        jobs:\n          - swift-tox-func-encryption-py36-centos-8:\n              irrelevant-files: \u0026functest-irrelevant-files\n                - ^(api-ref|doc|releasenotes)/.*$\n                - ^test/(cors|probe)/.*$\n                - ^(.gitreview|.mailmap|AUTHORS|CHANGELOG|.*\\.rst)$\n          - swift-tox-func-py36-centos-8-stream-fips:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-encryption-py36-centos-8-stream-fips:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-ec-py36-centos-8-stream-fips:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-build-image:\n              irrelevant-files: \u0026docker-irrelevant-files\n                - ^(api-ref|doc|releasenotes)/.*$\n                - ^test/(functional|probe)/.*$\n          - swift-build-image-py3:\n              irrelevant-files: *docker-irrelevant-files\n  \n          # Unit tests\n          - swift-tox-py27:\n              irrelevant-files: \u0026unittest-irrelevant-files\n                - ^(api-ref|doc|releasenotes)/.*$\n                - ^test/(cors|functional|probe)/.*$\n          - swift-tox-py36:\n              irrelevant-files: *unittest-irrelevant-files\n          - swift-tox-py37:\n              irrelevant-files: *unittest-irrelevant-files\n          - swift-tox-py38:\n              irrelevant-files: *unittest-irrelevant-files\n          - swift-tox-py39:\n              irrelevant-files: *unittest-irrelevant-files\n  \n          # Functional tests\n          - swift-tox-func-py27:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-encryption-py27:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-ec-py27:\n              irrelevant-files: *functest-irrelevant-files\n  \n          # py3 functional tests\n          - swift-tox-func-py37:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-encryption-py37:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-func-ec-py37:\n              irrelevant-files: *functest-irrelevant-files\n  \n          # Other tests\n          - swift-func-cors:\n              irrelevant-files:\n                - ^(api-ref|releasenotes)/.*$\n                # Keep doc/saio -- we use those sample configs in the saio playbooks\n                - ^doc/(requirements.txt|(manpages|s3api|source)/.*)$\n                - ^test/(unit|functional|probe)/.*$\n                - ^(.gitreview|.mailmap|AUTHORS|CHANGELOG)$\n          - swift-tox-func-s3api-ceph-s3tests-tempauth:\n              irrelevant-files:\n                - ^(api-ref|releasenotes)/.*$\n                # Keep doc/saio -- we use those sample configs in the saio playbooks\n                # Also keep doc/s3api -- it holds known failures for these tests\n                - ^doc/(requirements.txt|(manpages|source)/.*)$\n                - ^test/(cors|unit|probe)/.*$\n                - ^(.gitreview|.mailmap|AUTHORS|CHANGELOG|.*\\.rst)$\n          - swift-probetests-centos-7:\n              irrelevant-files: \u0026probetest-irrelevant-files\n                - ^(api-ref|releasenotes)/.*$\n                # Keep doc/saio -- we use those sample configs in the saio playbooks\n                - ^doc/(requirements.txt|(manpages|s3api|source)/.*)$\n                - ^test/(cors|unit|functional)/.*$\n                - ^(.gitreview|.mailmap|AUTHORS|CHANGELOG|.*\\.rst)$\n          - swift-probetests-centos-8-stream:\n              irrelevant-files: *probetest-irrelevant-files\n          - swift-dsvm-functional:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-dsvm-functional-ipv6:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-lower-constraints:\n              irrelevant-files: *unittest-irrelevant-files\n          - openstack-tox-pep8:\n              irrelevant-files: \u0026pep8-irrelevant-files\n                - ^(api-ref|etc|examples|releasenotes)/.*$\n                # Keep doc/manpages -- we want to syntax check them\n                - ^doc/(requirements.txt|(saio|s3api|source)/.*)$\n          - swift-multinode-rolling-upgrade:\n              irrelevant-files: *functest-irrelevant-files\n          - tempest-integrated-object-storage:\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(api-ref|doc|releasenotes)/.*$\n                - ^test/.*$\n                - ^(.gitreview|.mailmap|AUTHORS|CHANGELOG|.*\\.rst)$\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n      gate:\n        jobs:\n          # For gate jobs, err towards running more jobs (so, generally avoid\n          # using irrelevant-files). Exceptions should mainly be made for\n          # long-running jobs, like probetests or (once they move to\n          # in-tree definitions) dsvm jobs.\n          - swift-upload-image:\n              irrelevant-files: *docker-irrelevant-files\n          - swift-upload-image-py3:\n              irrelevant-files: *docker-irrelevant-files\n          - swift-tox-py27\n          - swift-tox-py36\n          - swift-tox-py37\n          - swift-tox-py38\n          - swift-tox-py39\n          - swift-tox-func-py27\n          - swift-tox-func-encryption-py27\n          - swift-tox-func-ec-py27\n          - swift-tox-func-py37\n          - swift-tox-func-encryption-py37\n          - swift-tox-func-ec-py37\n          - swift-func-cors\n          - swift-probetests-centos-7:\n              irrelevant-files: *probetest-irrelevant-files\n          - swift-probetests-centos-8-stream:\n              irrelevant-files: *probetest-irrelevant-files\n          - swift-dsvm-functional:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-dsvm-functional-ipv6:\n              irrelevant-files: *functest-irrelevant-files\n          - swift-tox-lower-constraints:\n              irrelevant-files: *unittest-irrelevant-files\n          - openstack-tox-pep8:\n              irrelevant-files: *pep8-irrelevant-files\n          - swift-multinode-rolling-upgrade:\n              irrelevant-files: *functest-irrelevant-files\n          - tempest-integrated-object-storage:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - swift-tox-py27-centos-7\n          - swift-tox-func-py27-centos-7\n          - swift-tox-func-encryption-py27-centos-7\n          - swift-tox-func-ec-py27-centos-7\n          - swift-tox-py36-centos-8-stream\n          - swift-tox-func-py36-centos-8-stream\n          - swift-tox-func-encryption-py36-centos-8-stream\n          - swift-tox-func-ec-py36-centos-8-stream\n          - swift-multinode-rolling-upgrade-rocky\n          - swift-multinode-rolling-upgrade-stein\n          - swift-multinode-rolling-upgrade-train\n          - swift-multinode-rolling-upgrade-ussuri\n          - swift-multinode-rolling-upgrade-victoria\n          - swift-multinode-rolling-upgrade-master:\n              branches: master\n  \n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball\n      promote:\n        jobs:\n          - swift-promote-image\n          - swift-promote-image-py3\n\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 542, column 3","accounts_in_message":[],"_revision_number":2},{"id":"0bd67f3c1355171d6db904a0273afd7311bb93cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"date":"2022-02-21 16:02:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6a5089d202621abefcf30ee10183cbd267bf9e4c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-02-21 16:02:55.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch stable/wallaby.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 552, column 31","accounts_in_message":[],"_revision_number":3},{"id":"2ae6f88a66c8e235f646991f3c1fc6a9c9b48662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:03:39.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch stable/wallaby.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@stable/wallaby\", line 552, column 31","accounts_in_message":[],"_revision_number":3},{"id":"6775a296afa9bced0709978093198ea463588dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-22 06:28:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"192d98a4c8f466c49a32399ac53edf52da359484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 07:46:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b2f62f40464ef791820c646235b53c : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/5fab64b32db546b597e915eb2ba837cf : SUCCESS in 56m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/592ebf06226f4ff89c80be0d3bbc4816 : SUCCESS in 42m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ff1af15d2e848848131781452beb2b6 : SUCCESS in 50m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8960a78570eb4f5a9865c224feac50f2 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ae04261531f945d4ba6824b617255dcb : SUCCESS in 24m 58s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8ace03a90382467f94dc8e8f8d8841bb : SUCCESS in 25m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d86b79c840be4fcdac7086831ed8d925 : FAILURE in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b7cbcc7a183491497c5f1cdc7835985 : FAILURE in 4m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94e3839aa258427aae5b22791170fa87 : SUCCESS in 15m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d53de299fa604236abcc396d0bf55ce2 : SUCCESS in 14m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ced9e490496646868a9b57238e55dd3b : SUCCESS in 16m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f793a15687eb4125b1e1258ffebe8401 : SUCCESS in 23m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97fa097336c2409ba920ad32244bfa7e : SUCCESS in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37053aae74c44bca947ee3861771aaed : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/046353c0194c49168dd1e7550fb20ff7 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97368e52d53b4c21914241edb233c194 : SUCCESS in 28m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/de44f6d093d94dea99574aecf2653984 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e2268465fcb04d6a91f1393780db70aa : SUCCESS in 20m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/10e231a4fa1a4b8a89bd65bc43623eb9 : SUCCESS in 23m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/27d463315b91471fad8bc72bbac5c57c : SUCCESS in 9m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa5b0c6fd5e3448485ba666ae5cbf38f : SUCCESS in 19m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2bbb842041844710b846c88cdb7f1f3e : SUCCESS in 49m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4045dfdbc804d668c0016f45179d2d5 : SUCCESS in 59m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ca9490ec50a47de81f46b59e7e15afd : SUCCESS in 44m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c76d30a25c8f4ccabcd18d08593e3f61 : SUCCESS in 44m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a997f5843fa45e6aae63ddfdbcd6f9d : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bf4aff548a4d29b4d7fc8f7b96013f : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0317ef85f224c10bde707411087eaed : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3adfa8ee198742468c96d109b4928aaf : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"2d3d73ee75c7b6a99c7f519e2994f64683348b56","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-22 15:59:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f361260c985bca5782ae1a62cd1ac6944d3bbcdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:01:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"39f5aaed80a60a6a9b524966888eb9373211a89d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:22:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa892b66de334a8fa104567812bcd877 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/c192652b02c043debbd7ebb09edab581 : SUCCESS in 1h 08m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c0a092e4c4a4945aff6ed84da1e59d1 : SUCCESS in 46m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f2557508ce44c7a94644e44aee0d43c : SUCCESS in 54m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6d37a093f34c4cd487ca6a95af982804 : POST_FAILURE in 10m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fe6b8711961241acb945f233c2ba4f4f : POST_FAILURE in 7m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5a492973acf40098bde163b3d9a4ed1 : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7de0aa0c6a2c4db6923dd00ba925745c : SUCCESS in 19m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fae6e1eb2764546bab848f159929910 : SUCCESS in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d26f354d98b940d4b81f8c5ed16337a6 : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8cc9e32d39394646a4e53b5b975122ba : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a944e9f3c2274ba8aac7a8981c3d6f94 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21e659691ea6436b9dbcfd947fc3fcf3 : SUCCESS in 25m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6efed72672c444148c104d47a29f559d : SUCCESS in 18m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/77e2d90f0f1842f1ae0fff60cf083879 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aa0e0913595d464ebf8bd6c84600ae20 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a27eccc53c7a4a4baed426a75d435850 : SUCCESS in 18m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d44ccf2b5684f3ebaa804011e66020b : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e20ec00eaab249a293f7363d91cc21cb : SUCCESS in 46m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/35958061f25745d3bfb5793d43f00996 : SUCCESS in 49m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/056a2209637645a9a900d93f0e41c6bc : SUCCESS in 30m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/734d1014b5ce40d6b6908c71f29b1181 : SUCCESS in 40m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0282eb6a86f547ea8bfd7756dd1c8663 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2472c2f252fb4396bb4d5b9604a92e4d : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe7cabc5906647c8a91f78ed4d5b07e1 : SUCCESS in 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b66d5cd546fe4a86b6ad8d63e44f0876 : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":4},{"id":"f4f034e8bf34d1107d67ceba5b821c388c0e88af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:22:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d229bd905bd394bf979c4099adcf4d1ad0256006","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 18:29:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d7f3bc7554452aa69587f01308736f : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aa69ebf2a7044efea3873d807087c265 : FAILURE in 1m 02s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/318e4797b53f4f3a972f59a7c5c7020d : FAILURE in 55s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d256c76ef31a406ad1d8d4b46f5f0ab92d5425d","revisions":{"dd864bfe264df0110dc90f883e3cac55263c28e6":{"kind":"REWORK","_number":1,"created":"2022-02-04 21:39:13.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/02/827902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/827902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/827902/1"}}},"commit":{"parents":[{"commit":"a25a936c4fd4486d66c434715f0cb8a6f404885d","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-02-04 21:39:13.000000000","tz":0},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n(cherry picked from commit 03be71c44aee283fa940f9a43c31072cb6082301)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd864bfe264df0110dc90f883e3cac55263c28e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd864bfe264df0110dc90f883e3cac55263c28e6"}]},"branch":"refs/heads/stable/wallaby"},"87fbe26def303a415be59e772a99855a1934d927":{"kind":"REWORK","_number":2,"created":"2022-02-04 21:40:27.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/02/827902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/827902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/827902/2"}}},"commit":{"parents":[{"commit":"a25a936c4fd4486d66c434715f0cb8a6f404885d","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2022-02-04 21:40:27.000000000","tz":0},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n(cherry picked from commit 03be71c44aee283fa940f9a43c31072cb6082301)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87fbe26def303a415be59e772a99855a1934d927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87fbe26def303a415be59e772a99855a1934d927"}]},"branch":"refs/heads/stable/wallaby"},"ddd5a112b12c77fc60b1e9c2024692e50d24e44e":{"kind":"REWORK","_number":3,"created":"2022-02-21 16:02:35.000000000","uploader":{"_account_id":34120,"name":"Andre Aranha","display_name":"afariasa","email":"afariasa@redhat.com","username":"afariasa"},"ref":"refs/changes/02/827902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/827902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/827902/3"}}},"commit":{"parents":[{"commit":"a25a936c4fd4486d66c434715f0cb8a6f404885d","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Andre Aranha","email":"afariasa@redhat.com","date":"2022-02-21 16:02:26.000000000","tz":0},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n(cherry picked from commit 03be71c44aee283fa940f9a43c31072cb6082301)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddd5a112b12c77fc60b1e9c2024692e50d24e44e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddd5a112b12c77fc60b1e9c2024692e50d24e44e"}]},"branch":"refs/heads/stable/wallaby"},"6d256c76ef31a406ad1d8d4b46f5f0ab92d5425d":{"kind":"REWORK","_number":4,"created":"2022-02-22 06:28:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/02/827902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/827902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/827902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/827902/4"}}},"commit":{"parents":[{"commit":"a25a936c4fd4486d66c434715f0cb8a6f404885d","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-06-11 19:24:54.000000000","tz":-240},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-22 06:28:28.000000000","tz":-480},"subject":"Add FIPS CI jobs","message":"Add FIPS CI jobs\n\nAdded jobs to run the swift functional tests when FIPS is enabled.\nWe\u0027ll set these jobs to be non-voting initially, so that we can\nensure that they are stable.\n\nChange-Id: If0b4cfc1ac9a8c66085eb4dc95366d43806d5ae2\n(cherry picked from commit 03be71c44aee283fa940f9a43c31072cb6082301)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d256c76ef31a406ad1d8d4b46f5f0ab92d5425d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d256c76ef31a406ad1d8d4b46f5f0ab92d5425d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
