)]}'
{"id":"openstack%2Fswift~828588","triplet_id":"openstack%2Fswift~stable%2Fxena~Idb3efa94288fd2b7480057cb16f2f0b9c2911e25","project":"openstack/swift","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-10 00:47:55.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Idb3efa94288fd2b7480057cb16f2f0b9c2911e25","subject":"Fix arm64 jobs on stable/xena","status":"MERGED","created":"2022-02-09 16:38:20.000000000","updated":"2022-02-10 06:28:01.000000000","submitted":"2022-02-10 06:23:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828588","meta_rev_id":"a14b3287f6f68d216008f7e28f3b0dabe04fc7c2","_number":828588,"virtual_id_number":828588,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-10 06:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-10 05:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-10 05:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 16:38:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-02-09 18:15:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5db5d786312a230bb50622086dc36752d9c97d6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-09 16:38:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b81f49f188e94e5723a4687c95c761ec60f3dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 18:15:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa6646c8c8949e4b5c0535131aa14bb : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/a759f35ce4204288a49f4a64f7248157 : SUCCESS in 1h 14m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41f04f4ef21c48ac8fb0164aedd7c236 : SUCCESS in 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470bb6ca2b4e4345a6d17e6e917bb727 : SUCCESS in 48m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e187caced8e492e9f227f1b6516fe75 : FAILURE in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1932f653a21446ffb2423f588a3bb0d2 : FAILURE in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/544eff37ffe942769e17a7731b6c3878 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f298faacf7654308b368e6200e13af56 : SUCCESS in 18m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05c24c25fc144d62a46ccef39175c099 : SUCCESS in 21m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/59de74a1466f4ef1b9917e6bf4659d70 : SUCCESS in 21m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/442035473262468d939ec53669bce020 : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d852db9ffbfb483899049641fd17027b : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbaf4397552b4fb0b05b544f991165c9 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/19903f90bbc945d08677d8a4d2a703c0 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e6dd1c360e50411aac3a5b137c76212b : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b7db99fdca45436197c5877a09616314 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/34f1e7b442164657a53c9aa3f554ae29 : SUCCESS in 18m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c844c58e90849aba409c97950569bd7 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a531aebf3a464e9bbce56f5eb70eb5c7 : SUCCESS in 21m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf92253938b7414fba0748401470b018 : SUCCESS in 56m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6384ff79ba64b889da8620c42ea673d : SUCCESS in 1h 00m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8a0faf139664d9992c886652d68cbff : SUCCESS in 38m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80b17118d7994f3f8a012e9e05ad029d : SUCCESS in 39m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc1e36d788c34784bad7e43e87e90d52 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ae6c468bb14b80ae94b11b4ef5a7e9 : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3943d5babf34c3ebabcf5f9deec6bd2 : SUCCESS in 25m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87d936e1fcbd40e28fe90e7b1869d174 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"de055c657913238263f05ea766b28eb4dd551937","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:25:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aec2a195f9354351a7af5c1b8c7fb14f : SUCCESS in 1h 08m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/602eb5081e794f3db4c6a9e5fb4b4783 : SUCCESS in 1h 06m 06s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e34ddd2ae1bc48d783e935488e610c64 : TIMED_OUT in 2h 01m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/807d1af43bbb43898d36100544138d94 : SUCCESS in 1h 14m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3598a8828034a1990151dd474817f6f : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"25737056f68e289046ae07a916b87f74cc611031","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 00:47:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99b10430f16e09b6401712b890eecb7448aef54d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:08:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45bc6f013c604e849f260bfe140213ee : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/332f66f2ef8246f48efab9747f3598a0 : SUCCESS in 1h 09m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96d42939599c42fe99168b1cfeebae31 : SUCCESS in 55m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07aaaa1bd7604d52b913992caceeb7eb : SUCCESS in 1h 04m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52d7ca6ac0b3482180e28cf032e006d4 : FAILURE in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0c16573cf274890bf71b6ee913dd9c9 : FAILURE in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e6c70a2d9e24a9d9622e99b522d1e6d : SUCCESS in 19m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0071269848504265a66d663a2a92a143 : SUCCESS in 19m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68899f826fdc43fb92f3f08030127c57 : SUCCESS in 17m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e65229bd5af74f43ac62cba3f7da7896 : SUCCESS in 20m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74f6f6ed2d3d4f73864c33c4735f4336 : SUCCESS in 20m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b1f357596c247c39542223559093783 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/01020e088b0f4a70af26fa6507dc3279 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec799663d76d41db8cc3c51ccbaad87c : SUCCESS in 17m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/96b0b7eb64be42ef8f7878818715009a : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb85f3f16c2245969b3a92289e1d0660 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/576cafeabe1042d880f2099fb59ff2ec : SUCCESS in 19m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d29bde324c64ccea9a7e3354a843b01 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65e000c19b64489f9ebffc613e99628e : SUCCESS in 25m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/289125090441400592e7179ef5a10ef6 : SUCCESS in 50m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5eb39fd283c4b01a840db42a3a041df : SUCCESS in 1h 02m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c05ba6cc9c0463d8290e65cf75174bb : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1495bb0d2b50446cbc924040b15eebe1 : SUCCESS in 41m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/065f2e617eb746388e2e46203aa4e30c : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ebbfca5f0eb469ab3c488f613558d8e : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/179b6111f8e94df389067222eb14b5d9 : SUCCESS in 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b4b635306e14586b039718614dccc4c : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2a299cc63ef31f92220a81c39a461a53c97b737f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 03:16:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53cc59733d974a52960a441a7511f6c2 : SUCCESS in 47m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35e319f8eae644f1bf909c4074a8bebd : TIMED_OUT in 1h 32m 39s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e5a8e33259564df3bd53b3ae10197810 : TIMED_OUT in 2h 02m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/caf7b0d28e7640f5b75c0abcd0c222bb : SUCCESS in 1h 12m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e53f8d1bd5784637907dd43523956cfe : SUCCESS in 1h 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"60f26638f2a01b685e4c37a456c54a7a7c09c8b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 05:01:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a965d9668fd02f5420992212315cd13187da133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 05:03:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e837194d21a1a8b5596c36c79392e900af612b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 06:23:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddb816f2b23142369fdf988fb45b9b4d : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/498e545254b94875bcc61798fc771438 : SUCCESS in 50m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c70abb1a027472484bf9339c4b4faa8 : SUCCESS in 35m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb0febb7db0746e1883e00cd06780e7d : SUCCESS in 45m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd01dd0d83114f559db8dcb40469ac5a : POST_FAILURE in 9m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c2f0638db7f7466eb81cd81ac0f237e0 : POST_FAILURE in 7m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f5716e09de4435da1b9f000ca408f62 : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1552d7607d0e4d85a9001d068f42a50f : SUCCESS in 22m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b01adefe090746be94215daf54d23eb4 : SUCCESS in 12m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac5b5512d1624249bc8bcfd397cc73cb : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d470851967543b3a92fdee6a8c4f2d5 : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d1ce18e935a4a87ac91f60dad92003e : SUCCESS in 23m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8b0cd8e47084facadcaf436e8497543 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f17ff1395e7e44bfbfd4f89fb79aed95 : SUCCESS in 20m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf1edcbb13014f1f9bed0b070a6c25b3 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ef1d20b1a2c486cb2dba80800a5532b : SUCCESS in 19m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55662b5472a2437d89e5419d643f2d7b : SUCCESS in 20m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c23067ca16546f4aa89ac9fd1fceb26 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed9e9ec8bd334fae8db63b66d1ea634d : SUCCESS in 44m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5010667d3a164f3e81eb7540c9d1d333 : SUCCESS in 42m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f54dafa2c294367837c147f907a1319 : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81e54b0a4e204d669cf9e1dd5f33ec50 : SUCCESS in 49m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69ca6b174fe34884ba2383f68698d6a4 : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b8c374c5aa489ebe7179254470d4df : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10213fc794af42e28c2f44e255a56e5a : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65e3aa12043742a4a77d7da8b1dd57d2 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"609e91beb0530e31b647bac5b6218129a5f149bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 06:23:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a14b3287f6f68d216008f7e28f3b0dabe04fc7c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 06:28:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c6cfdb71e4434d92e00b66474d20db : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/678e1b3f8b714034a5bbdbd5af045489 : FAILURE in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f24ade499d124d76bfa7b34f37a204cb : FAILURE in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71b491b057f23ac9dc1bb280780a4a1c5cf64cb9","revisions":{"71b491b057f23ac9dc1bb280780a4a1c5cf64cb9":{"kind":"REWORK","_number":1,"created":"2022-02-09 16:38:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/828588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/828588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/828588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/828588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/828588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/828588/1"}}},"commit":{"parents":[{"commit":"68bd795a6439baa869535944380996ca365ae4c2","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68bd795a6439baa869535944380996ca365ae4c2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-09 16:37:26.000000000","tz":-480},"subject":"Fix arm64 jobs on stable/xena","message":"Fix arm64 jobs on stable/xena\n\nThis is a combination of 2 commits:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nMake arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nPreviously zuul would report \"Build succeeded (ARM64 pipeline).\"\neven when individual job(s) had failed. Now zuul will report\n\"Build failed (ARM64 pipeline).\" but the pipeline still does\nvote on the patchset.\n\nRelated-Change: I84f59256df3aa41338df9829a565ae830ee8e847\n(cherry picked from commit 346f518d6cd10b62c411fa2b5fb480b81905568a)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nMove CI from CentOS 8 to CentOS 8 Stream\n\nSee also: http://lists.opendev.org/pipermail/service-announce/2021-December/000029.html\n\nRelated-Change: Id34eea3f20621c2b25cc7753b1323d4ffb88f11b\n(cherry picked from commit 546c9629ebf1831f7dca55ccd3f7b0e424a788b8)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nChange-Id: Idb3efa94288fd2b7480057cb16f2f0b9c2911e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71b491b057f23ac9dc1bb280780a4a1c5cf64cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/71b491b057f23ac9dc1bb280780a4a1c5cf64cb9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
