)]}'
{"id":"openstack%2Fswift~828612","triplet_id":"openstack%2Fswift~master~I632d58697a1026a6a581db27786ab9503c62f97e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-11 05:24:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I632d58697a1026a6a581db27786ab9503c62f97e","subject":"read-only: Only act on Swift paths","status":"MERGED","created":"2022-02-09 22:01:50.000000000","updated":"2022-02-11 05:24:26.000000000","submitted":"2022-02-11 05:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828612","meta_rev_id":"8dc655e00ae352ab05da60481937daa85b8f7efc","_number":828612,"virtual_id_number":828612,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-11 05:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-02-09 22:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-02-11 04:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-09 22:35:03.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-02-09 23:12:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfcd0b337ebfc07984fe36bbbc049e59f7ebc59f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-09 22:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d8cfd8c5df53a5690ddabd3ffc1b9f4bd734854","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-09 22:35:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e822c82dd3cc3b4128cf859fe1c1b238721de566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 23:12:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6899ba8d404dc4827f3373997427af : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/eab4024144f947acaf6f194c8e583425 : SUCCESS in 1h 00m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1cd0fe8360904be291fe5db00888f395 : SUCCESS in 49m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6842b26d142a4ab193f69daff46080fa : SUCCESS in 51m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/336026f31f124b939b69222b9af1041b : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ea26b728471443518959e77de6768d96 : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7dc248b1c08248679249d77ecb36ffc8 : SUCCESS in 26m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8de5b9ede3a24947a029b8a1f39038e1 : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fcace67ce10e4abf82ec41771e7f28af : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4df40bd9c8aa41108f9025d8e0b90ef0 : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3be0a3d153e648f7a2b63ea0265edb3a : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e5b43bf2f6f4e74a0ca2cecf5cdf9d1 : SUCCESS in 14m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/186395bf63ad4ac3b004f5903531d7dd : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d39d32996a346f085033c0d0716e1ad : SUCCESS in 14m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6bef309bfc584fd1a3ee5a59639eff74 : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/683544395b6d46d19d3e7fe12eb507be : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3cb11d6f02c4d5f88280b474a6dba10 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e36bcb057e81456b9ef42937eb3e6d4e : SUCCESS in 20m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e09f3b5be3f4b1ebfe83ed4d9d6f04f : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f4d72c4643344a5fb26b3d83db325feb : SUCCESS in 21m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8c9cc052ca2b473aabf187d89088c567 : SUCCESS in 36m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/30e74766c3b249e5827e473a02324703 : FAILURE in 59m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/996bc09928c54decbe5876d6b24569a5 : SUCCESS in 42m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ce42a0c25064ed29a92b00cd2964f5a : SUCCESS in 44m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57d1bad9af8b4027b413fae5e4283628 : SUCCESS in 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a8598a5df14bb6a15b9c73811f1d44 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d7f9a92d04b446080391a4b410ce336 : SUCCESS in 26m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08fd7cc5f1644872aba5247497000428 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8f61a967971c5bcdf0d2f5f1ea462037e8f2abb5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-10 00:47:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e331c0ff552e59360eb1f603fb9810fac325b699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f6587d339a42f595f2b78764824556 : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/17bcd968b3614c60b42efa9b203257d9 : SUCCESS in 52m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f84b820c32d4e04a53dabf13ba37e4b : SUCCESS in 46m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c49fb153f85e4a1f94847a1ed6109d0d : SUCCESS in 58m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fddef2e871b449d19a6f3d0c0982d3e2 : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1e4a9d99d87d4673b03faeec33c910ab : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6dc3790fa9b341498ce097c7917442c6 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ce23f60cf8e42d093218c51867d62fd : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c15c9545f9e945a2920444c740cd4a46 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa191eba74674668917185c328d7a705 : SUCCESS in 15m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45cfd816e9b945aca1be0a8c54ef7ef7 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dfc5c1bc7605456c82ccf7cf18ee0fc2 : SUCCESS in 19m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b1a7ed8836243a2aa82de954c6d7eae : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bc7969b311484fb2a23c97cb4d70dcd7 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90528fcb11a54f7290eee70aff8f8351 : SUCCESS in 22m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/213fe74e7be141a58ced95862aea8c31 : SUCCESS in 22m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd68ca2027bf42bc94433d3f2699a1bf : SUCCESS in 22m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b77e80071ad04926ae9d6c5d4fb8a5ec : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b7ad31a8b964171921d6a9ca2c91c60 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78e053a1d11c49a1ae266749c7bc9e5b : SUCCESS in 21m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/055f0b7485b0433a9e14a4f8df9e9a5f : SUCCESS in 53m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e845504a219432a8137dea145847d0a : SUCCESS in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26890567731b4f2a96d423ed6826f659 : SUCCESS in 44m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2fc3abe98bb42e1bc8647bd73045a06 : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/434506b5dc1b4aa89351775357987c7b : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeab6af599224045a2a41b6c3f681550 : SUCCESS in 8m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bac79a5dcf774328afa3377215d72e47 : SUCCESS in 23m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8483974f2fe94077bef1757d37b1c3b0 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ae6ec586ba98a13375c013126ec11826904bb4f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:59:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a071aee793f94c74ba35708ba242b88f : TIMED_OUT in 1h 31m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce9c3291768543b4b9c74ad1004502bb : SUCCESS in 1h 08m 09s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/dbe8e3907e794a48975b988cd5f70eae : TIMED_OUT in 3h 01m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/163456f504b443f39090e9d3d9a6bd65 : SUCCESS in 1h 24m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b8eb3c0bc4b45ff9342ae82d2e1e8f6 : FAILURE in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"341bae78ae3e59b0a1427cf36f6740835cdfef90","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-11 04:07:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e277ead4cd8363b3829b59cd067eb0f76ba5f4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 04:08:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15b3472abc7edeaa782b33a522bc071ff6b57753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 05:24:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c50378a11ee416c8038eb4b2798731c : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/9e8b40a1ed4946db9e9324d9f1703800 : SUCCESS in 49m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd48a320671d4a8e89311e3ad5289391 : SUCCESS in 35m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef2fb5855e6c420b965f5fb976896201 : SUCCESS in 40m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4033f3636dee4cd19da214639daae1a3 : SUCCESS in 10m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca925ab49e9c41588e509d5091f94b9e : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b4d99b0f8314c4fb8482c191cf4ed4f : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a24370fe2b5d49aa9ed0c6061b59cc57 : SUCCESS in 15m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/030d335878cc4c0980eee6046c9de32e : SUCCESS in 16m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c98b84f8460648269b320761b1bce212 : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d6450615a7d4ebb8afe107257b628e4 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/303694f5808e4e7c8d3883793c285eee : SUCCESS in 21m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d55e5e79c4ae4c6a8beca48ab4043cbe : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2259b015a499423bb58a86bdb4ae7012 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b841839b3e8748b7bb80592868afd22d : SUCCESS in 20m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b18bc9bc23645d388e6b1c6afb0753f : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c4cac3d98cba4ec89024353558967d3b : SUCCESS in 51m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dea24ac89ad94bd199b4dbd3d4b26982 : SUCCESS in 54m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2569895bc1354e2fbfbb9c03c732764a : SUCCESS in 39m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a2ec05fe4e60457398c78926428401ff : SUCCESS in 33m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75ea287eac6f4ca68c850ec578484dac : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833af2698651476bbcc8aef1dbf6de3e : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c172c27252714e88ac1eb7ee6ef08330 : SUCCESS in 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6f5168b22d54321b08cf1c2edaab90a : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8dc655e00ae352ab05da60481937daa85b8f7efc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 05:24:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00bb0f8ba68f2108dcf2f3d486217de298114044","revisions":{"00bb0f8ba68f2108dcf2f3d486217de298114044":{"kind":"REWORK","_number":1,"created":"2022-02-09 22:01:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/828612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/828612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/828612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/828612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/828612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/828612/1"}}},"commit":{"parents":[{"commit":"57f17ace7c911b80587a03634346478b6ca432e8","subject":"Merge \"Deprecate LogAdapter.set_statsd_prefix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57f17ace7c911b80587a03634346478b6ca432e8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-09 22:01:42.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-09 22:01:42.000000000","tz":-480},"subject":"read-only: Only act on Swift paths","message":"read-only: Only act on Swift paths\n\n... and they must include an account.\n\nChange-Id: I632d58697a1026a6a581db27786ab9503c62f97e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00bb0f8ba68f2108dcf2f3d486217de298114044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00bb0f8ba68f2108dcf2f3d486217de298114044"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
