)]}'
{"id":"openstack%2Fswift~829053","triplet_id":"openstack%2Fswift~master~I7cfeff4f436ade319cf21d0d29853931aef6d20f","project":"openstack/swift","branch":"master","topic":"srb_exit_status","attention_set":{},"removed_from_attention_set":{"34175":{"account":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"last_update":"2022-02-15 21:41:44.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-15 18:36:29.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I7cfeff4f436ade319cf21d0d29853931aef6d20f","subject":"swift-ring-builder: exit ERROR (2) on uncaught exceptions","status":"MERGED","created":"2022-02-14 14:21:24.000000000","updated":"2022-02-15 21:45:12.000000000","submitted":"2022-02-15 21:41:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829053-srb_exit_status","meta_rev_id":"964882861071c8a46ed81c9846ce85aee551a3c7","_number":829053,"virtual_id_number":829053,"owner":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 21:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 18:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-15 18:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-14 15:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-14 19:26:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"72069230dec0189ff15311a359b3431e56e5e310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-02-14 14:21:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28df414731be714cfecf27c21fd8db0d48fefcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:39:40.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a225b0c21b4fcd9c930b2e45e2a2ca : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6a4f4b7abeb842f39b4f60dfee84abbc : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cfb34c14fbce46fb86dbd29316dba3e8 : SUCCESS in 56m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1a761f49fe04c91b9c4ccd78e586899 : SUCCESS in 57m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6e6fcec6825b4c41ba38046d4b1c104a : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ab010eb3c5894dbab5a7cb5a6a1e781c : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e50f80ef07ac47089e7c9375c81278dc : SUCCESS in 26m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed277fc628d14eaf8f8d93ec9806480b : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/11c495bce864417980570beaa2b6646b : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5267b362dfcd428f9e0491e77bbd40d7 : SUCCESS in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ffa6f8bdf89943d49daf2596599f2c88 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/593e947151894c90b3e275d5d9cb269f : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6948ab2ed044408b095eba6b616c8b0 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ca7f5286f3143dfad1b08b94a8f2fa3 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98b9fb58c26542e984354794d6a8da0c : SUCCESS in 16m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/22acae15f0564a78a896e8fe2e4a1121 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/141720558cff40c1a4746f86241e1fd8 : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/35532a595473440eb8973abf23aa9122 : SUCCESS in 39m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5925618cc5ca4300a9c5b194eaab02e3 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc7ed6bf790840dbb6ad0a808ee6c112 : SUCCESS in 23m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a3ba27f52d6a450c92b303fa5f43166d : SUCCESS in 49m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e8f8ac428764e318c7e83a228c5dcb0 : SUCCESS in 1h 05m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3f7c3ce052c45219d58bb2c54555641 : SUCCESS in 43m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8adca874220c48c099b5dd776a1c5611 : SUCCESS in 41m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/639c5f038045483e905e4d8a99d8efc6 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617785988f184aa6ba9c917e0ad2be72 : FAILURE in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1be0fc3b8c5e4865850b5aa597477e79 : SUCCESS in 24m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/154ff811e09c44319e87f0248356b2ba : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":1},{"id":"371163fbc715d78b2d7c71c89ae40872f1c065c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-02-14 16:05:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a0fd514b7f8d00c7cb8e232ed92a9a444f9a73e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:04:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a16d75c5184256b57545926e8ce9b0 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e5f6ffc49f034e83873f99573a2dd6c7 : SUCCESS in 51m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a45fcb825294d70a0aa173e04469b5e : SUCCESS in 41m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff12ea1121be4eacb338c21eef195d25 : SUCCESS in 40m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ec57d57c9fe24ee08bad57a813aecf5a : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2399330035c740df8363c2ee77efad9b : SUCCESS in 24m 44s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c0e657dd1e3f4df3a6358753fa05fc99 : SUCCESS in 21m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70fd99a0ffd04174805e2ef8335683f5 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb4f93c94fd74f81bd81e8487a0677a2 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c92e9a3800d48e98aceacf0089ec657 : SUCCESS in 20m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e4ff743c304499a848cd822c3a2f59d : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b0b74c7f119245ff8ea8414728d31827 : SUCCESS in 22m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ec899d9bfba4a5fa2ecf5089f6689af : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/32662465924248208b37fb88a695380c : SUCCESS in 15m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6e89fbfdb95644b299d608630b886b07 : SUCCESS in 15m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ee7ddd4e62f74ca3ab9fdce1275700c0 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a182e9d70cb404bb3c270c66153bc66 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/62df21a0e8e340a195be038b1d605da3 : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7bd1cd03b194a42a84c89f80895216f : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62606feb59af4d1d9b3078b4730cdb91 : SUCCESS in 23m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9cefb4d02438413aa94200d6d55a4bc3 : SUCCESS in 50m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/44c36410fe0f4a2b86c8a9fce018ab8f : SUCCESS in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/192177409b4c4866ad2ba80851461f9a : SUCCESS in 27m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1f4ae86a39048e5bfe22df594b0b1cf : SUCCESS in 34m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36ead38ac22541238ef716cc319b0137 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c79e518ccd4c7eb670158b14175fc1 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16321019bdd64845a6824bc83b805704 : SUCCESS in 29m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81c8101693cc4c5381a44f5e6f08efd0 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0cadf416323fcae652bf6979cd60c139e7ffa8bf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-14 19:26:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80aec5b206f28c7ebd8083f7079215595d0636a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 19:54:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/794a981f6e1d4cbfa33986aff14040a1 : SUCCESS in 1h 15m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e71540d112344fdaf73aaf609eb6017 : SUCCESS in 30m 36s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e770ba62e8814b1393f77f533b3e88ea : TIMED_OUT in 3h 01m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c660fa090a424707ac9b0c2ccdde8bcf : SUCCESS in 1h 20m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a20333560d44ba29a697f0747c17bd5 : SUCCESS in 1h 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"71bc61cb253996feeb928b43a950d6d6024182e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-02-15 11:24:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48a292716c3d1fccbcfbf01677d541d94a0ba077","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-02-15 11:25:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e2dcbb9354c605d6079d7189d529e35424dea73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 12:31:05.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62066f43f82745c9af77e71cd46305d3 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/be5ed0fd9b1a457b937ea011a58a2949 : SUCCESS in 45m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5962f18ae9de4309832f5b32a1df62e7 : SUCCESS in 45m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68de8071ccd14f0dab6a1f73c6012bff : SUCCESS in 47m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e23e669a584c4b7bb62255b84cfd3361 : SUCCESS in 22m 51s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/783e52a18d96425295a30c7878ae5b82 : SUCCESS in 23m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/edf2d691aba74dd3a3b1965f36d1a4bb : SUCCESS in 22m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c4dfbe0dba547838c49f552f3e35ecc : SUCCESS in 9m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/044b781cfb7e42dfa1fbfb6248679cb5 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c71d259f6a16491ba879a79e29021348 : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba8b9c7fb65b49d3bfd9fa11b284e4ba : SUCCESS in 18m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37c5cc2c5fe949579d33fcc6941de362 : SUCCESS in 25m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e18f388105f4fad95016a681c4ec765 : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d54d2f5c3a04838a565598c7295fa32 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b1e354d30974f0c90bb13cfba3f0939 : SUCCESS in 33m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f0ade3f9589842788dbcf04ea4f3696e : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fccc8f4cb912487f9aecbdd11ce9ee8b : SUCCESS in 28m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7abde22743f844ee9f48e457f54256f2 : SUCCESS in 31m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50c8bed2a9164cefb27f608e15b04b4c : SUCCESS in 13m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ed7863db692474483e56914feb794c5 : SUCCESS in 33m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca111ed2dfa24061b1fe8b424225fad8 : SUCCESS in 1h 04m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fecea21da40f4fd881f31b63bb65ab48 : SUCCESS in 1h 00m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8e1311180af487091660c0ccf178796 : SUCCESS in 48m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d49b59c485b9448a87184920164c9a9b : SUCCESS in 45m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69de826c26774800ab47b5da3e682c2b : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9297a4981d1b4a508dca94f9e5999f64 : FAILURE in 7m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/198e806b790c4888820240ce1004a517 : SUCCESS in 30m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5958d46fc8f4607987f9bd687f41148 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":3},{"id":"46e87c9049584f6b5514ffb9dc502dac17c8f9ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"date":"2022-02-15 14:29:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"74e68d606dfd928f0146bca409680a9ab5c4f5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 15:48:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2771b2456974387aa8680cddd4654c0 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/73d6db4318574623bcf09ba3295c2948 : SUCCESS in 1h 05m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e89d25bda1c54b17b8a647dedf715114 : SUCCESS in 51m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac8b2844f6f14bdfa7e55fd907daec17 : SUCCESS in 51m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/41d133e210a04dd8a54ccbc0c12a7697 : SUCCESS in 27m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/101946bb3e8f4ef08c4dd97ffa1d65a5 : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0d4d4194cf6044ff81ef7e6d7d6c6dca : SUCCESS in 22m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4a6cf7734ab49edac121a93fa9df2e3 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ea8a9e64c81471295ba878ba5ac49a5 : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f846bdad14cb44a68c469bd55dc6b564 : SUCCESS in 20m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d3efd7c8820409b88117c6048677048 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/982055a21a5d4ff091f462f2785a5690 : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f4edecfa5ab3471bb028d45d091941d5 : SUCCESS in 22m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c393bb6749b4c73a375b80981eb31c1 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7553e3d69d8547ae8976813bcce8c13c : SUCCESS in 23m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/edf24055c81d42bcba574a3c7cfe2c38 : SUCCESS in 21m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a025e00387044d0583b0c5e0808b3162 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ef40897239440f494c2b1facf6f5396 : SUCCESS in 18m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee4ee74723934ce792da74a4371de230 : SUCCESS in 10m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7abb8bdfe084468781aea7f59ef99fc6 : SUCCESS in 25m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8619b8f26004109a0145c2f0899869e : SUCCESS in 58m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d02986778b341b791c79c5952243c21 : SUCCESS in 59m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49ffa079e5b64ab9a5dedd50767da59d : SUCCESS in 44m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30cd06364ae94db7af486a67806af15c : SUCCESS in 41m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40f2de06a7f1481e8d4f263af6e7168b : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f5e07839ed4248b30e450b2910370f : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e72ef5fa3cb542d48c129bcdece53a3e : SUCCESS in 32m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aa5a65ffe754694aee24c458eb901ab : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c35b78a8267079ff882a355420cafdcc3e841a23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 17:59:33.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33bd02d7e09046e989485f57200bef81 : TIMED_OUT in 1h 32m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7320f112b0704164a3925e3021bfe703 : SUCCESS in 1h 07m 54s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8f11eaa782684097869f8271a29bed85 : TIMED_OUT in 3h 01m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f22662a2ccde4d2187bf2315a0783d0a : TIMED_OUT in 1h 31m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5dd8b1e1cd964b53b87a0d5dee942c2d : SUCCESS in 55m 20s","accounts_in_message":[],"_revision_number":4},{"id":"33743230b659ecc56278d4975533d50b304c720c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-15 18:36:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7b90a36affd4102c9238841d77067886cdceae12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 18:37:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c7a0e99538f880dd73bb3b730a10b06bfe66803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:41:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4308518e1f3f447e9c928bb25b1684a1 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/01641c865c1e444aa5b395763f5794d1 : SUCCESS in 49m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58d67582ca124cdea430dcbb4b4c3f4d : SUCCESS in 54m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eaba52383db64726a70b6cf381da3643 : SUCCESS in 58m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/095acff00aa1457680e3edfda2b0d670 : SUCCESS in 8m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/846b410bd53244b987622b275a6d226d : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f17a585ded743c99180258f75baad93 : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a90e9be25814b0393996446aa4d832d : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/467bf0c3d8ef43e6be816b4964c1e2fa : SUCCESS in 17m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/095a0ea747c94379a0c0662b64c76026 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6fdb8892ec8445129802c60576fecb30 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9303838fa3a47f2843f9413918b94c2 : SUCCESS in 17m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/54a429deef0c48568ae7bf27ec48eab8 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb7bea1ccd5946b8ad04cf85ca3f2b9b : SUCCESS in 23m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a576839e50364529b09966e0182bf895 : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f8e06952dda04711a082ac4efcbcc7e5 : SUCCESS in 6m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ef005d237f724ceda90ea1024efb627b : SUCCESS in 54m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7383767980b40d4b8ad4ef4840485ae : SUCCESS in 59m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/daa4e4e83af8494aa73cfd8f0d5b8d95 : SUCCESS in 25m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fa379f2fedd4cd281ae049f8e054171 : SUCCESS in 1h 07m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99e16c5a0825491f9e701316199aff3e : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c7d11ce4544e96814ce9d989a5fb1f : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2de5a2d9ea1e4c04b4adac826a3ab53e : SUCCESS in 29m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/542deaacf01a4740af80158f9a29dda4 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":4},{"id":"37f0f33655b8b0d70ee691c77676fb9422d871d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:41:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"964882861071c8a46ed81c9846ce85aee551a3c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 21:45:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2590b1e9e86744b2bf8e2dd9d5dc90da : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/06c3726806ca4256ac479e1a5733c96f : SUCCESS in 1m 08s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/327e1aa6cdb94ccfb26d77e358a74088 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"89ee843080b5e4d6f142b8199414720dc01be116","revisions":{"9daa503dde52b13e56b6763f5016093ee0877020":{"kind":"REWORK","_number":1,"created":"2022-02-14 14:21:24.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/53/829053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/829053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/829053/1"}}},"commit":{"parents":[{"commit":"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e","subject":"AUTHORS/CHANGELOG for 2.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 14:14:27.000000000","tz":0},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 14:18:16.000000000","tz":0},"subject":"swift-ring-builder: exit ERROR (2) on uncaught exceptions","message":"swift-ring-builder: exit ERROR (2) on uncaught exceptions\n\nswift-ring-builder has three exit statuses: 0 (OK), 1 (WARNING),\n2 (ERROR). Uncaught exceptions in python result in an exit code of 1,\nso for example problems writing a builder file to disk will result in\nan exit of 1 (warning) rather than 2 (error).\n\nThis commit addresses that by wrapping as much as possible inside an\nexception handler that will instead produce the usual backtrace and\nexit 2 (error) if an exception is thrown and not caught.\n\nCloses-Bug: 1960657\nChange-Id: I7cfeff4f436ade319cf21d0d29853931aef6d20f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9daa503dde52b13e56b6763f5016093ee0877020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9daa503dde52b13e56b6763f5016093ee0877020"}]},"branch":"refs/heads/master"},"a0aae46d5972d58121b18225c26a84c7722026ef":{"kind":"REWORK","_number":2,"created":"2022-02-14 16:05:06.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/53/829053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/829053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/829053/2"}}},"commit":{"parents":[{"commit":"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e","subject":"AUTHORS/CHANGELOG for 2.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 14:14:27.000000000","tz":0},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 16:04:45.000000000","tz":0},"subject":"swift-ring-builder: exit ERROR (2) on uncaught exceptions","message":"swift-ring-builder: exit ERROR (2) on uncaught exceptions\n\nswift-ring-builder has three exit statuses: 0 (OK), 1 (WARNING),\n2 (ERROR). Uncaught exceptions in python result in an exit code of 1,\nso for example problems writing a builder file to disk will result in\nan exit of 1 (warning) rather than 2 (error).\n\nThis commit addresses that by wrapping as much as possible inside an\nexception handler that will instead produce the usual backtrace and\nexit 2 (error) if an exception is thrown and not caught.\n\nCloses-Bug: 1960657\nChange-Id: I7cfeff4f436ade319cf21d0d29853931aef6d20f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0aae46d5972d58121b18225c26a84c7722026ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0aae46d5972d58121b18225c26a84c7722026ef"}]},"branch":"refs/heads/master"},"d0334acfcff73511f34fa05a78298e095586cd45":{"kind":"REWORK","_number":3,"created":"2022-02-15 11:24:08.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/53/829053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/829053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/829053/3"}}},"commit":{"parents":[{"commit":"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e","subject":"AUTHORS/CHANGELOG for 2.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 14:14:27.000000000","tz":0},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-15 11:22:41.000000000","tz":0},"subject":"swift-ring-builder: exit ERROR (2) on uncaught exceptions","message":"swift-ring-builder: exit ERROR (2) on uncaught exceptions\n\nswift-ring-builder has three exit statuses: 0 (OK), 1 (WARNING),\n2 (ERROR). Uncaught exceptions in python result in an exit code of 1,\nso for example problems writing a builder file to disk will result in\nan exit of 1 (warning) rather than 2 (error).\n\nThis addresses that by overriding sys.excepthook to produce the usual\nbacktrace and then exit 2 (error); excepthook is called when an\nexception is unhandled, unless that is SystemExit.\n\nCloses-Bug: 1960657\nChange-Id: I7cfeff4f436ade319cf21d0d29853931aef6d20f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0334acfcff73511f34fa05a78298e095586cd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0334acfcff73511f34fa05a78298e095586cd45"}]},"branch":"refs/heads/master"},"89ee843080b5e4d6f142b8199414720dc01be116":{"kind":"REWORK","_number":4,"created":"2022-02-15 14:29:32.000000000","uploader":{"_account_id":34175,"name":"Matthew Vernon","email":"mvernon@wikimedia.org","username":"mvernon"},"ref":"refs/changes/53/829053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/829053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/829053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/829053/4"}}},"commit":{"parents":[{"commit":"975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e","subject":"AUTHORS/CHANGELOG for 2.29.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975d3dbcfe8ef7e8007444e5bb5c0b3f890e534e"}]}],"author":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-14 14:14:27.000000000","tz":0},"committer":{"name":"Matthew Vernon","email":"mvernon@wikimedia.org","date":"2022-02-15 14:29:14.000000000","tz":0},"subject":"swift-ring-builder: exit ERROR (2) on uncaught exceptions","message":"swift-ring-builder: exit ERROR (2) on uncaught exceptions\n\nswift-ring-builder has three exit statuses: 0 (OK), 1 (WARNING),\n2 (ERROR). Uncaught exceptions in python result in an exit code of 1,\nso for example problems writing a builder file to disk will result in\nan exit of 1 (warning) rather than 2 (error).\n\nThis addresses that by overriding sys.excepthook to produce the usual\nbacktrace and then exit 2 (error); excepthook is called when an\nexception is unhandled, unless that is SystemExit.\n\nCloses-Bug: 1960657\nChange-Id: I7cfeff4f436ade319cf21d0d29853931aef6d20f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89ee843080b5e4d6f142b8199414720dc01be116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89ee843080b5e4d6f142b8199414720dc01be116"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
