)]}'
{"id":"openstack%2Fswift~830535","triplet_id":"openstack%2Fswift~master~Id844f47ead7fab47915b6db76c06155b22586efe","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-03-01 00:03:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id844f47ead7fab47915b6db76c06155b22586efe","subject":"db: Attempt to clean up part dir post replication","status":"MERGED","created":"2022-02-23 00:07:34.000000000","updated":"2022-03-01 00:05:16.000000000","submitted":"2022-03-01 00:03:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"830535","meta_rev_id":"d6d82cee48b95dc42bcf2ae5a071cd8e72b97c6b","_number":830535,"virtual_id_number":830535,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-01 00:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-02-28 22:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2022-02-23 15:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-02-28 22:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 01:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-23 15:19:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-02-28 22:34:23.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"35829b67abe1a79f2a8d3c07ba98901aa889525d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-23 00:07:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3539c115c36a00ce1293ab1bda2c1617e5a446c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 01:26:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9394e581b7a4facbfacc32c16704ec5 : SUCCESS in 12m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/360666beaa7d4c249dcf231ad9253233 : SUCCESS in 43m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c80f4a0326643658e484c4ecde375e1 : SUCCESS in 37m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49b13e5ced1f4ee1995f900643b1ea29 : SUCCESS in 37m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/22032bc02b454fbea0a9f69957c3c02a : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/abd9a022a9c04ce3bf310eedf84d07e8 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/20c7710ac89b47b786df0237b13e7e20 : SUCCESS in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14536d38c15d46c7b90cb6ff1ff1e61e : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/23df4289876848cc90ed526682acc6ee : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac2866fa22184939bdeaf2e22deb9bb4 : SUCCESS in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/542720f4406a465188001f5f44f020e0 : SUCCESS in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35c557c9f5f34e93b5e5f3a488123b39 : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fd4962ab3684065ba929133b6e0e593 : SUCCESS in 19m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eec79b34fa6c49b99c6c83c3b708be9a : SUCCESS in 20m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1b6d8a4af9b47719c63bfe54b5c6de1 : SUCCESS in 20m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/414c24fe3fbb497ba63c6aaeeb36c08b : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d81d6d5b62b34dc8a79628ae1757197b : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30957c50bd2543088991bd857f0d2ed9 : SUCCESS in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0f277400c9e4dc7aeddaa6d9342c892 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e0862f3b51841b0a3e6f2d29e13b0d4 : SUCCESS in 22m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed9bd1f59d1a4593bd01b4d2e8bc866b : SUCCESS in 49m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/43840ad917774df1882e2a39f64627dd : SUCCESS in 1h 07m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7dd5f7c5e5d04e26a689134eab39c3a0 : SUCCESS in 42m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8603f81778334f6fba83e7738043ba7b : SUCCESS in 1h 16m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6494dfcb0ea4d1b881e0d82c6ad5143 : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859b8f9513d54a2fa94ba965f6c15e2e : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cc40ef91d414c0a8ce6e5ffaab395c2 : SUCCESS in 22m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faad5326553941e880869a536bdf82d3 : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"739291ebf61077042d9db3e3a037d5c83dea916a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 03:11:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccbde3810ff4410a920ce8aac5af0a3f : SUCCESS in 58m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/149831dc7b5e43078b53d4ddc8f0e5a1 : SUCCESS in 1h 14m 39s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6aded2c088334680bec4b4d70ffec334 : TIMED_OUT in 3h 01m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38e477ef9613499ab35d517c94213b09 : SUCCESS in 1h 15m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcefa1c610784dbfa8f57303cc283935 : TIMED_OUT in 1h 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8cd528d1e64bf6e808b6ae85e0fdb71883ac31e8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-02-23 15:19:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5848b6568708cfb4aaec121ec34c1e468f4dbc6b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-02-28 22:34:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ee6a1665e4cea1208741398543a04b41f7df917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 22:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f22b76637d9834a0cbc87a5139602e1612c7fcfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 00:03:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb1676cb20484f7bb4514e817d0a9e6d : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7cbd48ba60654cc3b1730fbf16125c03 : SUCCESS in 1h 04m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25006cc5e112437bae9f473d11d9b166 : SUCCESS in 48m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85c7b37b246d49609873fc4c35566f6d : SUCCESS in 46m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/027a03ab603349a9a727ccc63d73876c : SUCCESS in 9m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/62cc7f29899d4e159a47d01f4c72aa3f : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ceb1553de7c948e2b49f5738d19cad10 : SUCCESS in 19m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0435aa393b6482a954f97bb8eeaf28d : SUCCESS in 15m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/012628a0c6134f89bab45ad797361b2e : SUCCESS in 24m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5d36094ae4d4b5c9c21e95614ff039a : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a2b4d8d66edf40a3b4a98245be8bd4b1 : SUCCESS in 27m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0548387a7484490b81a9eb4e9ced0143 : SUCCESS in 26m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe6057e54e194763bb18d3d46944dd70 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1fd7c132063d4e5bbf899287705b9114 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/75a0b8019f6d4ab3ba0e68730f140409 : SUCCESS in 21m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/820c4233986c474e9762d85767419170 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b49a57ca22f43d5bce71ea97be233d0 : SUCCESS in 1h 23m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/762cda2373024a3daa0b15498352f7dc : SUCCESS in 1h 05m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/586e8cf008954d8497580a187366e092 : SUCCESS in 43m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df3a9a3daf2943b89a2b87a64d8f6cb3 : SUCCESS in 46m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/653e388d18394e528117355d13797f6d : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbe6377f1228446980fad4c8c74ae85f : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/772e68bb817644a3abe3a298ae94ba42 : SUCCESS in 25m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e225f717ca04523befb25945abcd8b2 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a579ceb6575b8b7d250ea9fa2c9fcd1315d586c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 00:03:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6d82cee48b95dc42bcf2ae5a071cd8e72b97c6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 00:05:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d4faa4f3644e12a0e90c9f3e3e0124 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/562aa67cc17d4d299c8830d34d74ba3b : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6494b728436d425cac06787dae93ad63 : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd7159c69cc8d3de135298c1cb5613bea1397215","revisions":{"cd7159c69cc8d3de135298c1cb5613bea1397215":{"kind":"REWORK","_number":1,"created":"2022-02-23 00:07:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/830535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/830535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/830535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/830535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/830535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/830535/1"}}},"commit":{"parents":[{"commit":"c7774d960c4186b166e95b1122b1652d090de64e","subject":"Merge \"object-updater: defer ratelimited updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7774d960c4186b166e95b1122b1652d090de64e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-22 23:54:24.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-23 00:05:28.000000000","tz":-480},"subject":"db: Attempt to clean up part dir post replication","message":"db: Attempt to clean up part dir post replication\n\nChange-Id: Id844f47ead7fab47915b6db76c06155b22586efe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd7159c69cc8d3de135298c1cb5613bea1397215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd7159c69cc8d3de135298c1cb5613bea1397215"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
