)]}'
{"id":"openstack%2Fswift~832920","triplet_id":"openstack%2Fswift~master~I6ec8b8ceb8c6a58e74772e57e6fe5700f6ff8db1","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-21 04:47:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I6ec8b8ceb8c6a58e74772e57e6fe5700f6ff8db1","subject":"cors: Include `Vary: Origin` when using the request\u0027s Origin","status":"MERGED","created":"2022-03-10 06:22:46.000000000","updated":"2022-04-21 05:55:13.000000000","submitted":"2022-04-21 05:53:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832920","meta_rev_id":"92601d34d44bfb76c6454da8d3856f5b9e53864f","_number":832920,"virtual_id_number":832920,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 05:53:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-03-21 00:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-21 04:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 07:33:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 00:43:38.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"a3defc1dd77dcdd187c4259ec9e15406e0698e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-10 06:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f75c8b449d9ca341d0a9f930bb5d29f58edd4d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 07:33:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72ae85b1ac44ff6bfca2174a94a8bde : SUCCESS in 11m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/dbd49b00efa749cd882546a087a3ebdf : SUCCESS in 57m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b28fba24824a47ac9e77578f4ee974da : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ae402d4b6e44f1698876e8a735c2b76 : SUCCESS in 54m 42s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3444f646a5cc4c2d9b2f9d5adc5d4053 : SUCCESS in 26m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/477068764a8e4e589746a1e5eff5d67a : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eeb23de8541042288628b3bf8c49531b : SUCCESS in 24m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f64a89ccfb54c5c934b41ff37536c86 : SUCCESS in 12m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dbd2a72485914969829033996331890f : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b17e76bbbdcd4e6991b47f85b070a41f : SUCCESS in 18m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/226d75854aef4733953ccf4a11880eda : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce7284e5d73417db6b6ebb3d695dead : SUCCESS in 16m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0a80ec985dd4474bb35fd10df8a7c03 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b21292999d184b8aa65b49e91da6dfd6 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0eaeed634b7147978e927ae6db3fa89b : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/303dd60a29764791b7c8652239671715 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b90f3a976d92436d968ba57ac522e7e4 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6fb2e32fac25485fbdd0e8d987b9ea88 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0c6ce10165b41c7ab955eee13357721 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de212f943f7d4481a77e577eeb6fa5aa : SUCCESS in 20m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f95c8603d8934434ab44f52669e9707b : SUCCESS in 56m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4616747c68aa4734bc747a8c7d57bb2e : SUCCESS in 1h 02m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf32e9ffe1904f44854b8307ad7ac7a5 : SUCCESS in 43m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ace603c075446c9af5fdd3e70aa5793 : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a23380d5541481183037408371addcf : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b1ea42143741dc96aac5d38763cc28 : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/144d33922b6f4f139afe41fe3978639e : SUCCESS in 22m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb8a819838a46549cfb1a5e5734ea94 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ad6d7edea735c9d69b63322d8c0dc51c38298ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 09:01:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6926aeb626ea4815b89ce0f4216a6b34 : SUCCESS in 38m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7b6f20d384a496eb0156aec6dec4dae : SUCCESS in 44m 12s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/84cc03e223974defa5fe6fd4fa6a4737 : FAILURE in 2h 35m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c54d6f7df1184976be179e6e7cba1071 : SUCCESS in 1h 10m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/877c960858964f61b0317070624bcc45 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4e8b998f9c1b29578e15bde83766c3e510fe504d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-21 00:43:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d00bd0e429287159ffaeedc8dc286a0c40796274","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-21 04:47:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"246cdd3511973f03eb58886c185f98ad7a9ff86d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 04:47:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bcd34a2ef98e0eb977ab0b87e706a2748450baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 05:53:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406ce50d0bf44db29c0f3743ae45e29b : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/a7d788d8cd1049aba9fd6d3a571b580f : SUCCESS in 45m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/915cdc72f4c84d69a89e8238d79334ad : SUCCESS in 52m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c20c29597b446caa1e36bd3893bdff4 : SUCCESS in 39m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/90ba68eb5b3f4952a4dfdee932137833 : SUCCESS in 8m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ea7814e6afce421996b85852ac4dc81b : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9dfde712c22a46888873235874b311da : SUCCESS in 16m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9401b80e3e2446b5a7fcdd83630edf7e : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/370c11f03f524576b05a2d0ef4e1768e : SUCCESS in 22m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0648c98c50443cfa80f06058cebe850 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6165f5801c614146a8bfe3eeda80a06a : SUCCESS in 14m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3dd787eaf6d64c2db36a6ea23169462a : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7d68e7d0b104226897617d278883ff1 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/642bc01a2e694264ba763ebcde71f14b : SUCCESS in 13m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46e237da42b149c49f6aa35948f7b05c : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dac4e5d0d28b4465b9026f93203ae9cc : SUCCESS in 6m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94eb37273e50495eb24cad64f6aa1450 : SUCCESS in 48m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3152fa3e0e24be99137fe901827070e : SUCCESS in 1h 00m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b9e350fc6d4445283444a7164245373 : SUCCESS in 30m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24c2f45987b4487c8535eb2ed62256f8 : SUCCESS in 40m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/805e10ced49e413d923289c6c2178b5b : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1b3682a1d64a4884f2084cacd1ac99 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a846b7bab9b4e10a2a4ab0d5ebee9c9 : SUCCESS in 27m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e00d40dc2e084daa8bce42ceabd20ae8 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":1},{"id":"34f811f3f1d97be2c55fdbb872edba317adf8720","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 05:53:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92601d34d44bfb76c6454da8d3856f5b9e53864f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 05:55:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eeb04dfd501420aac6307e73a464d8d : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9264f5be1bef4fd19db2b087cc8890bd : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/70f0cb9f8f9f47e794b73ad018438102 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bab7f93223ae5c4721ba8a224cc12ec05bcac8bf","revisions":{"bab7f93223ae5c4721ba8a224cc12ec05bcac8bf":{"kind":"REWORK","_number":1,"created":"2022-03-10 06:22:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/20/832920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/832920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/832920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/832920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/832920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/832920/1"}}},"commit":{"parents":[{"commit":"3ff3076ce6648937993e90334b7a9f532b06806c","subject":"Merge \"db: Attempt to clean up part dir post replication\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ff3076ce6648937993e90334b7a9f532b06806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-10 06:18:49.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-10 06:18:55.000000000","tz":-480},"subject":"cors: Include `Vary: Origin` when using the request\u0027s Origin","message":"cors: Include `Vary: Origin` when using the request\u0027s Origin\n\nOtherwise, multiple frontends attempting to use the same data may get\ndenials because the browser served a cached response from when it used a\ndifferent origin.\n\nChange-Id: I6ec8b8ceb8c6a58e74772e57e6fe5700f6ff8db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab7f93223ae5c4721ba8a224cc12ec05bcac8bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bab7f93223ae5c4721ba8a224cc12ec05bcac8bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
