)]}'
{"id":"openstack%2Fswift~833654","triplet_id":"openstack%2Fswift~master~Idd07c97b5db9e120bbd3c535b3754ec211bcb444","project":"openstack/swift","branch":"master","topic":"p-sharder-docstring","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-07-22 11:46:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd07c97b5db9e120bbd3c535b3754ec211bcb444","subject":"sharder: fix and expand CleavingContext docstrings","status":"MERGED","created":"2022-03-14 16:34:15.000000000","updated":"2022-07-22 11:47:38.000000000","submitted":"2022-07-22 11:46:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"833654-p-sharder-docstring","meta_rev_id":"e97f39b6d6959294bea2f19e69bd48a9f50e98e2","_number":833654,"virtual_id_number":833654,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-22 11:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-07-21 15:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-22 10:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-07-21 15:50:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 17:48:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-14 18:33:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-07-22 10:35:31.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"836ed011a7d9886344d660f680a10ec5fb63524d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-14 16:34:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4e09790326e10dd60cb65255bc8cf17321cefcc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-14 16:35:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ee882476b6c80eff4baab59a656d1af91c89eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 17:48:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea94454d1f614fbc9432225d5c6aa0cc : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/3e60e5bfa4984af0b83990216054a0d0 : SUCCESS in 1h 04m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38ace883640d4856a7b38d73fd24f02a : SUCCESS in 56m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8b41e07e619420591aae00b5dcc3f24 : SUCCESS in 54m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/294227b51632437ab8aaf08c91fd0295 : SUCCESS in 28m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8995e1d2471f49b6a2bc7e1d14f4fec2 : SUCCESS in 22m 02s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e8b3e852a77e4dfebce5e6b7cca9de2c : SUCCESS in 27m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44dc1387cb8943638c9d8cd804c732f5 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/93e8c0d013634578b5c285b83fb2e75c : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d007cb78f9fe43bc86970a790b9f62bb : SUCCESS in 42m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84d81d6fd2274df3b736eff8db13c223 : SUCCESS in 17m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33bcc3fbdf50494085f4630282825030 : SUCCESS in 43m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0cd1f71e6bee41c59d38bc3b389bab4d : SUCCESS in 43m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b73cad338fe24775bd8a20b3a8c764de : SUCCESS in 43m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/79088db5b271497d88f74bfdbd216174 : SUCCESS in 23m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db73fd894d4e42efae477754ed39d29e : SUCCESS in 41m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de3cdab779614058a792785e03e04199 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7fbb331975624832b039a8e8bb3959f2 : SUCCESS in 43m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2bc320d9d0b9491b91edd55186f34fc7 : SUCCESS in 9m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fe19f51d4d343f4a3e32d28b9a794fa : SUCCESS in 48m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/069dccc9aaf843b0914ce849aabbf44d : SUCCESS in 1h 08m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb705001735341d6b97ea3449cdad2a6 : SUCCESS in 53m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6b794944e4c40b0a84a1f954711a703 : SUCCESS in 44m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe7cc3a95acd477485f7045a5ee7df1e : SUCCESS in 47m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8db68d24288c4012ae210a06bc02b82a : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d2153055e240f3a73ce4fb6a29cc69 : SUCCESS in 8m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17fd0fcfca614d7685547d97670540f5 : SUCCESS in 25m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37dd160f05d24ccd8ac4660ef88e0e0b : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c1cc007858f835a661ed9f030f5285edf83fd276","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-14 18:33:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"79dd26154fa7ab696f5f50d5ae2b03bf07d6a837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 20:02:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67e5d08c19964af4a73f6f6f5f1e00a9 : SUCCESS in 57m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a834acb7794a4ffca18fea2f2f8b96e6 : SUCCESS in 1h 06m 09s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/42f4dc73e10146f09dd9018c6c9d4ce8 : TIMED_OUT in 3h 03m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2563e3593c84cdb8115078a00d3ac41 : SUCCESS in 1h 20m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1121e62b092497eaf541b712d42d822 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a06ac0158c0f81949154f33dc4410d719b4ecebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-16 14:58:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3bf8600679b2b1ffaf9369ab7e7042e6df24875","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-16 14:58:24.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"64e05fd9fde5c37b4ec3776cc2206a6b8592682b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 16:08:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02189496f0741bd905754ae7b8837c5 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/987836c13af0423f820fb74937651ab5 : SUCCESS in 58m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a445d38c112b438f9b0bffb3de77e079 : SUCCESS in 54m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11e60a40e5ac447b8ede476863e79a75 : SUCCESS in 1h 03m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ff45c1d22c5645bbb71e14bf84f32440 : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3091308c3bfb44b182828ecc1a47c587 : SUCCESS in 26m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4091231f7e484a9d9be50eee087aca41 : SUCCESS in 26m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/92f15684dee442ba937c06baed392a80 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8afde1a54c7c4ccbab105b1f8315a892 : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b46d847ff4d47b4b824aa68f5177522 : SUCCESS in 19m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f2f58c2dd5c47db9e3130937544e8b6 : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9639eb85a6344ddfac0c893dfc331a25 : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/452f5bb2bd6d405784834ce73105ca2e : SUCCESS in 21m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13a0bfc941ee43a5ae4491af46a39101 : SUCCESS in 21m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/777c4e10960f4bf1bcbcfcf7c93f6403 : SUCCESS in 24m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d462dd9672c4f659fc432076f077e37 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4e2a55837e624bdf81aa23efda04c8bd : SUCCESS in 22m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aff7bd5a350b48518188f9ddfba2410c : SUCCESS in 24m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ab4e1d52a1a4dd1a6348e690a0b458c : SUCCESS in 8m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba36f995858b42d69b9cff217349f9ac : SUCCESS in 21m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa12b930191d49f482371e17fd9acbd7 : SUCCESS in 51m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3000d6315a6e4881833d4811bc7cde22 : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af71d2b0c5fc44db83e13035d66ce8bd : SUCCESS in 1h 00m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25478dc2eb3141c995916ef66e94db1b : SUCCESS in 43m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c07963b47344d62b6286bbfc78c2fe7 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12edfd4edff24e049ed314ffdea0fcd8 : SUCCESS in 8m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98a6783b9b174fccb8656ba3c345e25c : SUCCESS in 21m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9343cfa3ba64c24a17f92c1226ea817 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a343a8ddd08c2a73df83de053f45f2cf413535ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:30:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/964e69b81db1437186df40925b5849f7 : SUCCESS in 49m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e99807333e39474e8c5ff9eb95173685 : SUCCESS in 45m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/61d8525677b14082a3d9d4da3e461ac6 : TIMED_OUT in 3h 03m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c283139bc0b42828b93fab2d8325b96 : SUCCESS in 1h 08m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0f58933939a440e99ecac9227ce15ef : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":2},{"id":"3500ce524d0ed4660940cba715414194c031aec7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-21 15:50:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"623a3e986af6e6a943db79040f5ed9ae24156a1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:51:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce66a11902bf7edeb54da3583f538e48870b97e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:34:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93ae87a73ac546e6a5040c61f9f17340\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96f86e1da2e4cc4b281217fc25562f0 : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0d774afc78ef42049beef175753f099e : SUCCESS in 1h 12m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf5197e82a944972800cc7ac6f9b6bab : SUCCESS in 58m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2a46cf45d36437ca35919df0ed99bd0 : SUCCESS in 1h 03m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d5276eb9b6474baa87f44b7a8f8fc82f : POST_FAILURE in 10m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d3d7ef69e52d48cc80a826b0e466deb7 : POST_FAILURE in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82e9a700ddff42b697e2fd7fd6cb73a8 : SUCCESS in 17m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba55cace08204c55af375595a3750808 : SUCCESS in 19m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/76ff450d580547e4b4bc8c4fba656c57 : SUCCESS in 17m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e7b5986990644cabe2c680cfea367cb : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b8e5d6b1d3134c4883a0e174503baa7e : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da30136ad5104d569abaa60c0765f22d : SUCCESS in 20m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95f1e13d8442407fb00e646ed377cdf0 : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6842a656933e47a1b700137389b53ae8 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ba72fe2407b4400681878e88b17ef161 : SUCCESS in 21m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f755317ec5444c7987e4616d8f2977e5 : SUCCESS in 14m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/57626c4a59934c33a75382a3336f3e3b : SUCCESS in 9m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8307b2b537b84c23934e781959e1b4a8 : SUCCESS in 1h 00m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ecbb6aad8754c02a300b8070eaea842 : SUCCESS in 58m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30599fe874024a63b21dc945c9de6163 : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78796c559b5b45bf9bc913f5424d6095 : SUCCESS in 47m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95bbfe27187b43e19ec6ec25b52adc5d : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dc7c5afbcf416e809296351c9b7a65 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11d5d5ff39ed410f9afec6b3dfe3444f : SUCCESS in 35m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0c4a7b95eb54180b9f642d199d1bdea : FAILURE in 50m 03s","accounts_in_message":[],"_revision_number":2},{"id":"89135e0d3a3856643f0a152345155527b666fb90","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-22 09:09:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c71bec605aadb6f327b96afb1de1bc58d8ca0ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:17:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f366a1b1eee344df828ca51d95d85a26\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/804dc27f683c43aba9e0125a475c5f41 : SUCCESS in 39m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3657508157964149bf38ea9772ae2dd3 : SUCCESS in 36m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2988df8a999f40a9bd8eca9e78407793 : SUCCESS in 1h 05m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54308d3778ee4432bb7c55b3153a0b8f : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5f716268b17cdb01551d1022c96111aa83fd348c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:22:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/401eb3e76c424dd387a5d7fc6a2d846d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64fbed505504c44837a6d938070b223 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d149c4d38d7c492895ab231a7810b4cc : SUCCESS in 1h 12m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c39728d7ccd47d087be5c522c0e8458 : SUCCESS in 43m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6f7c3f82ed8464cbd3fdd503f16d30a : SUCCESS in 52m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cd693c95a52848a597257a7b88564b90 : RETRY_LIMIT in 9m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/20dbcbff782c404c8db5b21be7f5ae08 : RETRY_LIMIT in 4m 04s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fe94c7821b2d4fd686a7d0bfe0c06c89 : RETRY_LIMIT in 6m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0647a112a056428bb9db0a1cfd896638 : FAILURE in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/79202507f1e64648b465b03309783caf : FAILURE in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/410bbef17db044529f1b31cd5070f7c1 : SUCCESS in 18m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/729d804ee4324a0d8029abfc5be3cfe2 : SUCCESS in 21m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eddbf8f6ebe045758d039bbe0d7dd6e3 : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1b97147ca1f405f83e5d9193ca5fd88 : SUCCESS in 20m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef73b978a9f540918696ed279eb3f3f9 : SUCCESS in 19m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cfbcdb5e25364bf49fab5f2b95e4b1ac : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/768a18310063416a85b35cb02aeba532 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12d9763fc8f64227b4ca6fb3b1f39f22 : SUCCESS in 20m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aac5f698802b4b01aa66561fef394ba8 : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e060fe378f444cd1a1447039fa93e22c : SUCCESS in 8m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89d068da48bf44ca8422eddf6073f990 : SUCCESS in 24m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b71fbb6b01b44968c5ed875bd054e36 : SUCCESS in 9m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c34f156e9f1740b5b2c0d61784d21a67 : SUCCESS in 53m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/00db75d6a0ae419387eb2dc5e7fbc734 : SUCCESS in 59m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0aea3dbca3204984a1bc98fab6a5c07f : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44aff28da27d4df5857a2433bb28b8a5 : SUCCESS in 48m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a794774c764428eb3789b1e3cf5ba6b : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014f666ed7d4453aab5ea05fce23fdce : SUCCESS in 8m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4cf7a47d12f941af86cfc0e2c09a5114 : SUCCESS in 25m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9645d0315c904c5c9a1ad4427a90d479 : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4eee100d6cfba125dc45677330486754202ba26a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:23:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c1f9c1d36be2c3484619229dd5a581437001494","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-07-22 10:35:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dda39e54c3c57c4fc67d139336512b115cdcca51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 11:46:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56b65805e8804748916ff82022df5ca3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d3876fc7894a6794dd3ae4b1dd0dc4 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ee0ba8098a364e279533a9f676b00707 : SUCCESS in 53m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6856291512b4325b0362eb7da39cb7e : SUCCESS in 44m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/723c937a7b144437b81ddf77c98b36a0 : SUCCESS in 45m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f5643c7bf60b43d891aa13c6da0b733a : POST_FAILURE in 5m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/51998bc5231c4cb8a18ab6f6e001e93d : POST_FAILURE in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35572897925b426881ae0f05f9f02e2c : SUCCESS in 22m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f41cf976e21431c862ed37bda2f092b : SUCCESS in 22m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/895e19910095426f83123422c9e99829 : SUCCESS in 23m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a9445ae83e24e53bedce7393b28a895 : SUCCESS in 25m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6be88a0fee284c1e96d48898fdf33077 : SUCCESS in 26m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b45fca528ae47b8914d381a46d77523 : SUCCESS in 27m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d172adbe87a74ac6849a6684c37acb8c : SUCCESS in 23m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebc11a22735a4476beb2be7d8eaf2211 : SUCCESS in 25m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37c5a093db2247bab306943634949def : SUCCESS in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e340e5748aef41dd8390eddaa354505a : SUCCESS in 7m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a103a21f2f564c259fbd356cd4429b99 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1692140fced743ef9dbd37b5dbab52d3 : SUCCESS in 1h 22m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb392a967b9841dc85d685a89ba0fd09 : SUCCESS in 1h 13m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f093d93e17243f7803ec94fbbe4b5c9 : SUCCESS in 45m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fff33ef4141f40619bf7b80d740b3aa6 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd166fc178314e4e88a28299509ade3f : SUCCESS in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33613add16194812a4848a0f1cf6351d : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a744ad583b1c41c0b0afa81915e7006a : SUCCESS in 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54812adac54a4981a44709f8a335ff1a : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7e8f649bc2fa69ee6aa9c4d5a4df57ada3424219","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 11:46:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e97f39b6d6959294bea2f19e69bd48a9f50e98e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 11:47:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb7ee9d16cc4cf7895a8a2839fe97a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca299875cd324707b4a44bf9ea8e098f : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6a9401485fb04e9e8b262d674e176ba6 : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bca693ae47624546842998bff7bd8b68 : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8155e69b6a6f6eaaf30ad88d59d0c6812e56e76c","revisions":{"27681e6009f1c2ee459b6b4e4b98c5d741c5155c":{"kind":"REWORK","_number":1,"created":"2022-03-14 16:34:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/833654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/833654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/833654/1"}}},"commit":{"parents":[{"commit":"014c98e853b86fc44312d879bec12099be6e22d0","subject":"Merge \"s3api: Fix multi_delete with object names using non-ASCII characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/014c98e853b86fc44312d879bec12099be6e22d0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-14 16:29:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-14 16:29:24.000000000","tz":0},"subject":"sharder: fix and expand CleavingContext docstrings","message":"sharder: fix and expand CleavingContext docstrings\n\nChange-Id: Idd07c97b5db9e120bbd3c535b3754ec211bcb444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27681e6009f1c2ee459b6b4e4b98c5d741c5155c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27681e6009f1c2ee459b6b4e4b98c5d741c5155c"}]},"branch":"refs/heads/master"},"8155e69b6a6f6eaaf30ad88d59d0c6812e56e76c":{"kind":"REWORK","_number":2,"created":"2022-03-16 14:58:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/54/833654/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/833654/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/833654/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/833654/2"}}},"commit":{"parents":[{"commit":"014c98e853b86fc44312d879bec12099be6e22d0","subject":"Merge \"s3api: Fix multi_delete with object names using non-ASCII characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/014c98e853b86fc44312d879bec12099be6e22d0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-14 16:29:24.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-16 14:58:03.000000000","tz":0},"subject":"sharder: fix and expand CleavingContext docstrings","message":"sharder: fix and expand CleavingContext docstrings\n\nChange-Id: Idd07c97b5db9e120bbd3c535b3754ec211bcb444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8155e69b6a6f6eaaf30ad88d59d0c6812e56e76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8155e69b6a6f6eaaf30ad88d59d0c6812e56e76c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
