)]}'
{"id":"openstack%2Fswift~834193","triplet_id":"openstack%2Fswift~master~I1ca7bf42ee03a169261d8c6feffc38b53226c97f","project":"openstack/swift","branch":"master","topic":"p-sharder-hasty-active","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-07-14 05:09:29.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-07-18 23:57:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ca7bf42ee03a169261d8c6feffc38b53226c97f","subject":"sharder: always set state to CLEAVED after cleaving","status":"MERGED","created":"2022-03-17 15:57:24.000000000","updated":"2022-07-18 23:59:05.000000000","submitted":"2022-07-18 23:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"834193-p-sharder-hasty-active","meta_rev_id":"0f70f5ce9745c70a1eb2517bf4381a12a5b377af","_number":834193,"virtual_id_number":834193,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 23:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-07-18 03:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-18 22:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-18 22:40:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 16:58:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 04:18:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-07-18 22:40:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b1c02758862ff4d491475639850fba79898ebc58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-17 15:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a5064751b5ad68762493e3f436607a21ed5bdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 16:58:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d3d11dee114907889a8a631b98d586 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a37eeeab62647b9b55ffa7b388f4bdb : SUCCESS in 52m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a79c04b422934112bf6e7a434c34f5aa : SUCCESS in 53m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c0e86b76c7a4349be52bb71b81ba3eb : FAILURE in 56m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b68e48b8954d3c8f55baa240b971eb : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"54f5c788de060db363d678756ab7a581e0533e56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 18:46:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5003e34435c241d2aba496408d6d0c94 : SUCCESS in 39m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbac15659f454b52a2390f8b6e4ec1c4 : SUCCESS in 42m 55s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/22b46f074f274f0ba7f9ab7c44ad058a : FAILURE in 2h 46m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5fffb5cf98548d393692a165850d5c2 : SUCCESS in 49m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/508eab0d2e104f85b193088ccbef7d51 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d1acdf09e6aa3e473a9fb3fa2cc909f9ce9cbfcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-18 20:48:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"759a18231fa2e99c1ab5d7c7f056001d41144ff1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-18 20:49:12.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f496595594b60892e1096849b4899692fce33ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 22:15:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342d5e93eb4d419b8c99f4a51adbaee7 : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/7901ffebec094da7860e611fd2c947e0 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2c470df8d0148cc8f0de1117e1c5f63 : SUCCESS in 1h 16m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e77afe80cd74ba8bdce88a102401942 : SUCCESS in 1h 20m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/35b15a2010e74ec9af51a7773212b342 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/63d8e54dd91f4803a5d8a73e61cc0f74 : SUCCESS in 24m 02s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e23cd7c93a20415aab72abc21619b5e7 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ae2fc788aaa41b5995ebe0f2215a14d : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/48cf86c287f64c92b308211fbb9783fa : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95a1122cd6b4467b95f485e86fe0caf2 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fab7c34200d643e9b5ff1d31b5eeeff5 : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42eea113532a44dd835f90c2bd48c558 : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0f535703e28f4c59b05c4a6c4fb7b676 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99a3838b089e4cd8b23af9f227491af8 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1503673885e84eacb31ece5e7c5d20c3 : SUCCESS in 22m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65a5603210ef4050b81686e198ff89f3 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34c49f0b08f0472ba584591fb61d62b2 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/210680763feb4b8fa09c8b169c1dcecd : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1c9433eed794f27beba526bd419dbb6 : SUCCESS in 10m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d5100bb9e6d4d5ea12ca23fce5ebf0c : SUCCESS in 22m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fec2bf51430f409da572441e39385535 : SUCCESS in 50m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/df344baaea474919ae18c7a11afaade5 : SUCCESS in 56m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7040263614564ac989c881d0fa223eee : SUCCESS in 41m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d64d5cbd741417688b195daf3c90ace : SUCCESS in 41m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/081877fcd1b043ac956f9a081d88ae82 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f01d878a54443581b45340c50bf029 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/925a720fdf084c4fa43bae550f563b2e : SUCCESS in 25m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53e7a97f205449078fb3821a92678f25 : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f5fc04ace8624dd844d0d32ffaee1a7998f4edd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 23:53:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35cd9e968c7143369baa5fa4805c7df3 : SUCCESS in 37m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be6c7149197749e99e741175efa3b7bd : SUCCESS in 47m 03s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ca8d07da9bae4615908408dcdd77cc4a : TIMED_OUT in 3h 02m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ecbbdada87846dc99f0f65a1e7c4cfa : SUCCESS in 42m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cb5dff942d7469199a9be51051d2040 : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bbaca257b55ed5c73134c9fdffe13a2a321ae032","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-21 12:09:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66f4f441ba6cdeb0332a12e474809e72b92c5ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:10:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59003e72d9948e6af27610c01a93815 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3b5e9f299c7d4083a0a80421e3be826d : SUCCESS in 53m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fc29cd919d04e28b7f913fd36d7bb40 : SUCCESS in 54m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad0560464494483481ba62cf190e5d53 : SUCCESS in 50m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/01f57a49399842ea8f34b31a93386dd7 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a1666f87a9b64f8bb644fb3d6a1113c4 : SUCCESS in 24m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0d02ccee7dac4fa990b93fa43334e478 : SUCCESS in 22m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aeece59f57334263b55f013297a88924 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2bd45314bbb49d0b72244f7b532e7e1 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba9c41d697f94d5fb96f11d4db4f9f14 : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2dd3c2020a384f5ba877de6f627e44cd : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4c065232dc2406f8b564d395fa39af2 : SUCCESS in 19m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8f82e264902c4266b61ebcd8f60a2ba6 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7008a631bc74890b15594073c14484a : POST_FAILURE in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/be9478097cd44a6a83cdc882c69ce142 : SUCCESS in 23m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de1412b11e5b4ea09a0852703313a2b0 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7b599ae13134bcfbe538a59e84aacb4 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ba944beed67435d8cda731ad842f20f : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2853767063234f5398bfcd419ff6fb50 : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ec8fbe832644079ba43cd67bfe1eef7 : SUCCESS in 23m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bbdfa2367b5c470a860e578e49559eac : SUCCESS in 52m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/755432c2d4f2425c8fab736e4797a26b : SUCCESS in 48m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7aab1d5d74de41f8b10c6e215cb91993 : SUCCESS in 46m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9f7645fc13640129806b22646a8899d : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccedcc4eeefa41fcb95fe84b11d4ffd4 : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f374a48a1c40858226f7f728204b8d : SUCCESS in 7m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da085b9cd3dc4d3b9bc902b01111983f : SUCCESS in 23m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8949f6f0e23741579dc17a518a3b3408 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":3},{"id":"faf6accc4d2600559d172b65d78b1296cee598d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 16:41:15.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f6e1efe4b5446c79913cc87d31e2235 : SUCCESS in 1h 18m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5e35dc6cbcb47d68a29bd08b55774eb : SUCCESS in 57m 47s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e4612c0c29f64371ab02b2a8b12c14c3 : TIMED_OUT in 3h 03m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0c01f3371c74d83926db7499317933d : SUCCESS in 1h 02m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/546d3dd5cb5a45f3aed53dce71d50cf2 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":3},{"id":"6bf2d1f8e2171776e1b8ba1c45af923983bfc64b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-23 04:18:39.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5998bd31983b60ceed50d75737239d526a4ef6b7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-28 05:44:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d194ec90987bfa4cf3953a861b571d91eb012ab1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-28 06:19:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8c1c595fc3b5ccff77613ffa698297e0bfc16cdf","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-28 06:26:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41fb48967b33f56d75497181033bfb6eefde1038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 09:59:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d6b1511a9d4cc79ea9eb32b54f59a7 : SUCCESS in 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2420f495856a44d38d89fe0f2d06f42e : SUCCESS in 56m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/918faa48077a443d9763776a0438d80a : SUCCESS in 58m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e83e4f737ceb463aac8d3199fa01d275 : SUCCESS in 1h 00m 57s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5a1f812ea472422a92bf304d02fc1401 : SUCCESS in 28m 52s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/600d76e273264ae28b922666b1e14411 : SUCCESS in 23m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9fea3f6aa4574afaafef7a041f84ab34 : SUCCESS in 25m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77436b2cce3246ad8fd8f0d386eef32d : SUCCESS in 12m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1bae78f8694c47e0a469c4dbdcb44fef : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6999018bbe9746779a092916004d5c94 : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89b3d328d4ff4f55a69028477da58b80 : SUCCESS in 19m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d50e5f0f6934852a18faf5913bd53b9 : SUCCESS in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d4ad7d168c14272b0eaca797b5dbcae : SUCCESS in 20m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e4f7667fa264f20af1973cbb74337ef : SUCCESS in 23m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0be1beae63044417897afb7c3238ddd7 : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fbd69724c97f426593442213e84934df : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8fa959d0dc32413b8e46e7bdc77b49c3 : SUCCESS in 27m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8fa348104eb418d84e583e188c9f02e : SUCCESS in 23m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/700358b5c44d40a19e91b5a3585e71b0 : SUCCESS in 11m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dc0426f003e43d6b35845e09899dccf : SUCCESS in 24m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6abe06f168a44cb5a96b8c263aef49cf : SUCCESS in 1h 12m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1776fced231469788ce068bde8ba18a : POST_FAILURE in 1h 09m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38cac3bffff3427bb806035d4c5d6594 : SUCCESS in 42m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81ce0335419744bca7a489ef12e54116 : SUCCESS in 50m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e597321b8456436ca39c5287c0ef4086 : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c2b3e130644e0aafe623b8b9254214 : SUCCESS in 8m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce79ac24c65c4f2eaefb567d59499fba : SUCCESS in 26m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4adb1df1f6f45a1a5d45415fbc7abd8 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":3},{"id":"d969bf1a39f7509ce0a289bb4cca1f113ea5e78e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:37:40.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a2e13c5a41047f6a257a21efdd5c1e4 : FAILURE in 1h 13m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96c4b361b57a4311a50d6c063720ff56 : SUCCESS in 1h 06m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c7a2d45f29254305bd3fd8bd38b285ab : TIMED_OUT in 3h 03m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/97911caef70844b1931fecb08ff6ae36 : SUCCESS in 1h 27m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6997e7c0352412ea11d02480aa9ace2 : TIMED_OUT in 1h 33m 09s","accounts_in_message":[],"_revision_number":3},{"id":"cccf95e22bcf238bcd3744db9054cdf45a4a9a5e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 14:17:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e220a48785e7814266f2bd1ad592db03ae31fac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:55:59.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c08e386d61eb4e25bc6571e24d4af903b1c697cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 17:44:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad0279a3d83049b7b75ad28356f637ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be41b754385a465bbdf8a09f3581ee19 : SUCCESS in 33m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b44579015b004e4ba498104ff70180c5 : SUCCESS in 46m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1173bc0b9a484936b56349e20ca868df : SUCCESS in 33m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a45c799ba64c4119af8e3d57062c959a : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":4},{"id":"22fda694c0c639f413532ccadd3ed91662f7c8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:05:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7106fca4311e4331b4488ee551f68a41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7ce72c6a9049a4b81e170c2df00964 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/e9dbbb8b7bc845888a695c21fa4c8447 : SUCCESS in 56m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26fb96dff2144f8e8ec9c890e7f464c9 : SUCCESS in 40m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc47f42ba3b54f0590cb1d14adeb1213 : SUCCESS in 48m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b69469086b2c4f129b4fea25dc611f10 : RETRY_LIMIT in 4m 16s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bdd64426e4424ad7ac66f9e038a1ef30 : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/28690105f820421fa216a3de843e190c : RETRY_LIMIT in 4m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d641c3d2c7994a8bb1195bdfe107b707 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cc7ec6bde1a849d9a533dfdb9f91bb68 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/748f6e71fd2d454ba81961888084c357 : SUCCESS in 13m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7df079fa8af449dd965da5a9424ef3c8 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3cac0aeb9aa3407eae206c9eef6e3d9e : SUCCESS in 12m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/034b2a0315bd439c98a68eeab65bae2d : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fbc43129f7ca45babd8688cb76f95592 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae2f344a30c44d44a9faaebb5239afa6 : SUCCESS in 15m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d77f053d98524b5b86c80a41f48c4f11 : SUCCESS in 13m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/965d715fe1e04fbd844fe55d63c2dc0b : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f53d01b940f4873865b85f13fc613a4 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d653fcb58d24911a29d4bfb48926800 : SUCCESS in 9m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1a064e4f7734ca0a139cbdfb8dd97e9 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8525ec07f9c4c42971316fda25a93c4 : SUCCESS in 8m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4daf5141e2304cb5a0fb22e0ef2c034d : SUCCESS in 57m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4143c8b4fa4541258aed51094106c873 : FAILURE in 1h 04m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55f86a81dac8474fb53db82b2236302e : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1863594a602941e7bfa480953a6bc949 : SUCCESS in 46m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5621b294c41240599fa330ae917787ce : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad275b951d8247beb8f04a05e38043fc : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9710ef9970eb4c299360166fc42ca392 : SUCCESS in 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d4eb730ef84b13b4064fbf83cc92e1 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":4},{"id":"fe7a232316b621a4d4c0dc1e7380f928dff3e23e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 05:09:29.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"59516f1a62d56b5efd6041bc22c586082c92d160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:08:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8779852c25143b88a2a58756da00dac\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e48d099000594ed6969d5260559b32a7 : SUCCESS in 27m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdb09fe3bc7944ffa1a28c861a9c19b2 : SUCCESS in 56m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4033945924af47d39f296ed717fde7e2 : SUCCESS in 27m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a51e25f7bf9f40d8b99adce9f83cc81c : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":4},{"id":"9375481c40b8f130eedf55c1507de55e556740c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 06:22:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61986652777d4c7890d7159cd490ab0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed00d83b6c047dca6978c773b7aac2e : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/fc79009f7fd14a0b997adc025dc8423d : SUCCESS in 58m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0601f8772b6d4633aee34a67b7a9ebab : SUCCESS in 37m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e33ec735f9f94e9ca944aa02e48f5600 : SUCCESS in 1h 03m 27s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2eabc0a7e3c346e0bef9d9b67a9f8591 : RETRY_LIMIT in 5m 16s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/48302916a7a742d58e7e83a50fb57cc3 : RETRY_LIMIT in 5m 11s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/71cd591294f344e999a7be00ae9dfd45 : RETRY_LIMIT in 4m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec5b992002114ec7974589fcfd3e393b : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/65fea29e7557448497f4a1741d20dbaf : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b52244155fc64ab099ae206cc3de1a19 : SUCCESS in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a44e7917e39b40cfb88fc954359483b1 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3dac17c96631453eaa9115de1c8e9881 : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f214e2300dc94127b87842bbeb778eb9 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f19060b810394a879e9caf7cbdf8c96a : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c89636c8dad49309b7bce4f487e5e4f : SUCCESS in 22m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87c5bb9a16af4bbeba9647976ff3d4ca : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c953eeedd82478e9371523c08d333e5 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/761604680e164fe8b7fb65fe2a281ffe : SUCCESS in 20m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac575556e0084a9691be340a18d5de40 : SUCCESS in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1dcd9f7b209e4a5280b863a6b9e91de7 : SUCCESS in 22m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dde1bdb952554d5395b862c1634b41a2 : SUCCESS in 9m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f58064b4b5a4caab578399b67edc653 : SUCCESS in 53m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/45a4b69a907943a9a276e9f8007fa4b0 : SUCCESS in 59m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15d5e12ff88a4d57ab8bd7bc93c8174d : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5881970ca714fda87a6e54206385ad9 : SUCCESS in 50m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d64f79d3e28408d972065b31026b3d5 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d67bc47be84354a3a8458620b33079 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75b36af8f2f84c52ba72e62007a63a44 : SUCCESS in 23m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7918b236e84d4cb3bce0c3ea3cbdcb48 : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b7bf07a0eee6a07d72f163f10e9a639862cb43bb","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-18 03:55:24.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e255ba99d96baa70e82eed2d1d6c50582d389075","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 04:37:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e21fe1496986496b92c813e89c559460\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec9c7ffcbb0e4b4d8d404a896f36efe2 : SUCCESS in 31m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd73f405d2884514b019e2f165db8462 : SUCCESS in 31m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f39e9c7d76874662a157bf0b5887ad4e : SUCCESS in 33m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5257c717133d4db2b852cd920c672d93 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":4},{"id":"6a62d63c4cf717c2ecacc16bad3678cffa5aa8d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 05:04:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d65c52e2cdd146b9bbe8e0e27538f3c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba00e58541374156854adbf40715f73c : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/a2279fb592b54d279ac1e5a75842395e : SUCCESS in 1h 07m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a4aa2c97c34443a8b67ff155fe39a5f : SUCCESS in 42m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/381482c198bb40b2966de62cc5bf856a : SUCCESS in 45m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/322d2c4aa496410b996853203f9b121b : RETRY_LIMIT in 5m 36s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c7f48e92c55f4cbdb202781f988a69d0 : RETRY_LIMIT in 5m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fcf003fcae774f4ea55c365a426b9ccd : RETRY_LIMIT in 5m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/895589df9c3b407ca74281512070cf2c : FAILURE in 5m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3f658d1c8664dee9d0f7de17d20108a : FAILURE in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf232de8e84e4d65a150ba43ac749df5 : SUCCESS in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9418219f2b6449608491423ae3ce7a6f : SUCCESS in 19m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc355ca8708f47df9cf482d6fb4e03fc : SUCCESS in 18m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3747788dd6dc4bf2bbcd42b8969b5c6b : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/751c0a6c94b84a9dadff2e91fb1896eb : SUCCESS in 20m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04f8e89a447348abb35c112c78a7c421 : SUCCESS in 21m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8f493f465a6b421bbde15983a1200bc3 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/659a1679dd0444618743185ae3bcb06e : SUCCESS in 21m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c30ade5425ae420597267bb4ad975381 : SUCCESS in 20m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f4c7fe782024801b789686d61142847 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c967f869d4244bba037673cfa9a3ec6 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f7a03887e0004381af393f413335bdac : SUCCESS in 9m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ba36335f27414eeda24af6b2d8e567cc : SUCCESS in 59m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c10ff9da0fa2453486acef046cf9206d : SUCCESS in 1h 00m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eec6ec44961b4dc59e27f397f3fa20ed : SUCCESS in 38m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0de71dc3bf5f411598fd344678da2bf1 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eb13554b36441de928dfffca79f92c0 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419447a2d4774870a9a00892ba4bb97a : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f66d4e0246f48c49c46d51cdd76e469 : SUCCESS in 27m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2453d7154ff430a9035d7a2ec2a9314 : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":4},{"id":"812771080d8ef811319e074135bd865f6d275e31","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-18 22:40:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"948e807b961918fc7117ef2352fb4396afb278a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 22:41:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ef2b2ce0360f182515beae3cf1deb23afba4978c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 23:57:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f294405a460b4211942d8aef11126f1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28937ac888f742b48364f2e028f8d429 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/20b7c0f9d7c04423a50d1fe92a163a21 : SUCCESS in 1h 00m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ee4b13dcd2141c3af372cdd501e15a9 : SUCCESS in 51m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c9162742e88447d816f5b77cd2ab23a : SUCCESS in 59m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7045790e95c747fbb1d80398e0ce745c : POST_FAILURE in 6m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/01da3e8f6075480c82a613d49e827115 : POST_FAILURE in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f89faa6e7609453491608391523ae546 : SUCCESS in 16m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/249df3f2cc334efb9243deae53b0043b : SUCCESS in 18m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/edb340836ad44a18b235b69326e74556 : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b9dd5aa74af493284a9073ffd23ec0d : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a757a2966a74729a7d660f8a5fc198b : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9618dd52c1949f2bae2ae3f98b95380 : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9059cef3d24e4965a38afea21dd3508c : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88eb1a66d3ae40d2a1cf44111dc8608e : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6d3561d738a44728eb02536dc57552b : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0dd8a82bfa184a4c957f2a32bf1635c0 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a1655143f9f44b4abfc27b81ba1d9971 : SUCCESS in 8m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/16b3b2d3244a4a11908b444491bbc3e0 : SUCCESS in 59m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b32d0dc5919145dbb3e17d11f4a97b3f : SUCCESS in 59m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea2725e7da3a43a5bd7005d304e7f4be : SUCCESS in 51m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bca333c01488488ba17e0f9872ba13eb : SUCCESS in 51m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/070ec5a9703844c5a653b32dc9ebd505 : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1fb6d99a604cd49591c117bc5db938 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e4957dae0144dd5b7df89eef35c496f : SUCCESS in 24m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/501687982d3e44638ed0df9fff5c19c3 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":4},{"id":"17109ca630d8744f7be67342585821e0e110ada3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 23:57:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0f70f5ce9745c70a1eb2517bf4381a12a5b377af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 23:59:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6659db30724c45acbdcbbd45412173ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9276814026b74dc9a7c142c8b73860d3 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b0f72d7f35c345bf9ce61f15986d0f2c : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8aabcb5bade94798b93866dbe1a43bc5 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"57f7145f7379de1f736ff2d904e85918c8166536","revisions":{"92f50ba5712e305905347b52f34da538fe75b091":{"kind":"REWORK","_number":1,"created":"2022-03-17 15:57:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/834193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/834193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/834193/1"}}},"commit":{"parents":[{"commit":"cb8b3cdab262af1d223c0536220400b13c1d0a9a","subject":"Comply with AWS signature calculation (s3v4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8b3cdab262af1d223c0536220400b13c1d0a9a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-17 15:50:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-17 15:57:07.000000000","tz":0},"subject":"DNM probe test: too hasty CLEAVED shards leads to listing gaps","message":"DNM probe test: too hasty CLEAVED shards leads to listing gaps\n\nChange-Id: I1ca7bf42ee03a169261d8c6feffc38b53226c97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92f50ba5712e305905347b52f34da538fe75b091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92f50ba5712e305905347b52f34da538fe75b091"}]},"branch":"refs/heads/master"},"2e7984e55be97bcc657d63f837319db5919737a0":{"kind":"REWORK","_number":2,"created":"2022-03-18 20:48:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/834193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/834193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/834193/2"}}},"commit":{"parents":[{"commit":"cb8b3cdab262af1d223c0536220400b13c1d0a9a","subject":"Comply with AWS signature calculation (s3v4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8b3cdab262af1d223c0536220400b13c1d0a9a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-17 15:50:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-18 20:32:36.000000000","tz":0},"subject":"WIP: sharder: always set state CLEAVED after cleaving","message":"WIP: sharder: always set state CLEAVED after cleaving\n\nTODO: unit test\n\nDuring cleaving, if the sharder finds that zero object rows are copied\nfrom the parent retiring DB to a cleaved shard DB, and if that shard\nDB appears have been freshly created by the cleaving process, then the\nsharder skips replicating that shard DB and does not count the shard\nrange as part of the batch (see Related-Change).\n\nPreviously, any shard range treated in this way would not have its\nstate moved to CLEAVED and would remian in the CREATED state. However,\ncleaving does continue, leading to anomalous sets of shard range\nstates, including all other shard ranges moving to ACTIVE but the\nskipped range remaining in CREATED (until another sharder visitation\nfinds object rows and actually replicates the cleaved shard DB).\n\nThese anomalies can be avoided by moving the skipped shard range to\nthe CLEAVED state. This is exactly what would happen anyway if the\ncleaved DB had only one object row copied to it, or if the cleaved DB\nhad zero object rows copied to it but happened to already exist on\ndisk.\n\nRelated-Change: Id338f6c3187f93454bcdf025a32a073284a4a159\nChange-Id: I1ca7bf42ee03a169261d8c6feffc38b53226c97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e7984e55be97bcc657d63f837319db5919737a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e7984e55be97bcc657d63f837319db5919737a0"}]},"branch":"refs/heads/master"},"628e406c27602318dfb97bfbdeaba9748b95f9e2":{"kind":"REWORK","_number":3,"created":"2022-03-21 12:09:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/834193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/834193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/834193/3"}}},"commit":{"parents":[{"commit":"cb8b3cdab262af1d223c0536220400b13c1d0a9a","subject":"Comply with AWS signature calculation (s3v4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8b3cdab262af1d223c0536220400b13c1d0a9a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-17 15:50:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-21 12:09:06.000000000","tz":0},"subject":"sharder: always set state to CLEAVED after cleaving","message":"sharder: always set state to CLEAVED after cleaving\n\nDuring cleaving, if the sharder finds that zero object rows are copied\nfrom the parent retiring DB to a cleaved shard DB, and if that shard\nDB appears have been freshly created by the cleaving process, then the\nsharder skips replicating that shard DB and does not count the shard\nrange as part of the batch (see Related-Change).\n\nPreviously, any shard range treated in this way would not have its\nstate moved to CLEAVED but would remain in the CREATED state. However,\ncleaving of following shard ranges does continue, leading to anomalous\nsets of shard range states, including all other shard ranges moving to\nACTIVE but the skipped range remaining in CREATED (until another\nsharder visitation finds object rows and actually replicates the\ncleaved shard DB).\n\nThese anomalies can be avoided by moving the skipped shard range to\nthe CLEAVED state. This is exactly what would happen anyway if the\ncleaved DB had only one object row copied to it, or if the cleaved DB\nhad zero object rows copied to it but happened to already exist on\ndisk.\n\nRelated-Change: Id338f6c3187f93454bcdf025a32a073284a4a159\nChange-Id: I1ca7bf42ee03a169261d8c6feffc38b53226c97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/628e406c27602318dfb97bfbdeaba9748b95f9e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/628e406c27602318dfb97bfbdeaba9748b95f9e2"}]},"branch":"refs/heads/master"},"57f7145f7379de1f736ff2d904e85918c8166536":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-13 16:55:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/834193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/834193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/834193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/834193/4"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-17 15:50:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-13 16:54:06.000000000","tz":60},"subject":"sharder: always set state to CLEAVED after cleaving","message":"sharder: always set state to CLEAVED after cleaving\n\nDuring cleaving, if the sharder finds that zero object rows are copied\nfrom the parent retiring DB to a cleaved shard DB, and if that shard\nDB appears have been freshly created by the cleaving process, then the\nsharder skips replicating that shard DB and does not count the shard\nrange as part of the batch (see Related-Change).\n\nPreviously, any shard range treated in this way would not have its\nstate moved to CLEAVED but would remain in the CREATED state. However,\ncleaving of following shard ranges does continue, leading to anomalous\nsets of shard range states, including all other shard ranges moving to\nACTIVE but the skipped range remaining in CREATED (until another\nsharder visitation finds object rows and actually replicates the\ncleaved shard DB).\n\nThese anomalies can be avoided by moving the skipped shard range to\nthe CLEAVED state. This is exactly what would happen anyway if the\ncleaved DB had only one object row copied to it, or if the cleaved DB\nhad zero object rows copied to it but happened to already exist on\ndisk.\n\nRelated-Change: Id338f6c3187f93454bcdf025a32a073284a4a159\nChange-Id: I1ca7bf42ee03a169261d8c6feffc38b53226c97f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57f7145f7379de1f736ff2d904e85918c8166536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57f7145f7379de1f736ff2d904e85918c8166536"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
