)]}'
{"id":"openstack%2Fswift~834363","triplet_id":"openstack%2Fswift~master~I49d8480fd0c775249d8b586f9dea026448b8910d","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-03-23 17:01:59.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I49d8480fd0c775249d8b586f9dea026448b8910d","subject":"CI: Run CORS tests under py3","status":"MERGED","created":"2022-03-18 23:35:02.000000000","updated":"2022-03-23 19:59:23.000000000","submitted":"2022-03-23 19:15:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834363","meta_rev_id":"24b846d3a76e2af809dda49c9650a70b51010427","_number":834363,"virtual_id_number":834363,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 19:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-03-23 16:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-23 17:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-23 17:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 23:35:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-18 23:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 16:39:46.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"cbd26d7b6f57abac1891b73ddc9d6f9687559f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-18 23:35:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a3590347eaa6642e3dcc389bf01bf42c0df2227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 23:35:27.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Job swift-probetests-centos-8 not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-func-cors\n      parent: swift-probetests-centos-8\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s CORS functional tests\n      timeout: 1200\n      vars:\n        s3_acl: no\n      pre-run:\n        - tools/playbooks/saio_single_node_setup/add_s3api.yaml\n        - tools/playbooks/cors/install_selenium.yaml\n      run: tools/playbooks/cors/run.yaml\n      post-run: tools/playbooks/cors/post.yaml\n\n  in \"openstack/swift/.zuul.yaml@master\", line 363, column 3","accounts_in_message":[],"_revision_number":1},{"id":"457fb7cdb484b5aec00a9f1bb34d57a543e6b5fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 23:35:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  Job swift-probetests-centos-8 not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-func-cors\n      parent: swift-probetests-centos-8\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s CORS functional tests\n      timeout: 1200\n      vars:\n        s3_acl: no\n      pre-run:\n        - tools/playbooks/saio_single_node_setup/add_s3api.yaml\n        - tools/playbooks/cors/install_selenium.yaml\n      run: tools/playbooks/cors/run.yaml\n      post-run: tools/playbooks/cors/post.yaml\n\n  in \"openstack/swift/.zuul.yaml@master\", line 363, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0ae5a6b41922e8489aa2d832f46a4edd635ccf50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-22 20:50:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"121f18f84fe27e0b657c8a99e68fd6cacd2a3c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 22:14:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce2d14a04a244d38d69234da67f5cca : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/d5f9a36dbf9046758b9ea1b2a0fceffb : SUCCESS in 1h 03m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d96f678c70f14b909ecd17d7894f6a69 : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/437c2cc3e06d4ffaa490065fd323d89b : SUCCESS in 58m 23s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ef6178b656b34068bf9438fe21bc39d6 : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d62e2a844f86471a87a8cee5f0d3485f : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/847ac1ff1f3548deafb103390d3f2eca : SUCCESS in 26m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6013435c0f624da1a17f8dcef07987a7 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/358da1df2a24457bb3c10241170320e7 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19f1503a701541af8aeb02f20fc9709c : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2eadb1353c440048996fcda55e51ea5 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8e6cf01f8174d7bb1445ccfac393e68 : SUCCESS in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ffcb09ef86c40b0837b1bf878129fbb : SUCCESS in 21m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3a0446d47e44b41a37802905f6030cc : SUCCESS in 28m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeb6a815db6c4264aeeb919a423e1024 : SUCCESS in 28m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cd6aac4f3a944c22b25d5629ab39db6d : SUCCESS in 28m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a69ced0cc4f14925ab5cbf21e236ee97 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/71eec983c792428e92a2d3c793b54b86 : SUCCESS in 20m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93d1ad8f23ac4be39f3bd09c203ee03a : FAILURE in 9m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/50a1a6afcfd64969a6b4d5692b7e9559 : SUCCESS in 33m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a560ed8000d4e0cb88a5610a546962d : SUCCESS in 52m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/41d9bc8c6fb347d0a420b7925364368b : SUCCESS in 1h 22m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/823f5f722489453597c4e23b39eded88 : SUCCESS in 42m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81db383a61b0485d9dddb679c951a292 : SUCCESS in 49m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e367edca18d4fdaa53309e39f1040d7 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6890e7ad0a4347e8af0c93f76f0d24ab : SUCCESS in 7m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f68b45d417194af2b918b6511e9aaaf1 : SUCCESS in 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be7e8d94630c4180975c54751d502676 : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0035bf5718cdb5485c8996ee90e877c774651776","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 23:28:44.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f1194ed1f7b42798301993f12fb5992 : SUCCESS in 57m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/737b48f0699e4301b02823da881c1d7b : SUCCESS in 49m 15s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/213b7d5af57246c0b86e64e2f40c10c3 : FAILURE in 2h 35m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bea4f15fc45e43aa8fa626bea4c77cb8 : SUCCESS in 53m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1454faf5710e43ff99848554a85d7234 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"426fd443a767b7441d3a36eee53fac104f43d6d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-23 06:12:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14885564e192176527759429da8c0c20f3348484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 07:48:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08576230a1b24e76a330a8d7ac545865 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/7be9a941b6504245bc839354a1f8fc06 : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af13bf35da4f42c0b34ed98d5f743cc6 : SUCCESS in 48m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bcd2ccb878040788bd62806125ec359 : SUCCESS in 49m 34s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/98667fb186184008be56e96145970509 : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fc5e8b0eb79b4761b5e4d6f09c2d818c : SUCCESS in 24m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2e11a8089a9841fa8aa43b665277ecdc : SUCCESS in 26m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/daea060ef67d4443b2ff0c1e1cac0ccb : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e79855c140a4af48033044f40cbd26a : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e4a8347e53c411a8d42b5fe44a8ffbf : SUCCESS in 22m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad52c54dad5c4921a7facbcafc2192cd : SUCCESS in 21m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc007ee427b04221bfd026bb871cb3fa : TIMED_OUT in 1h 32m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da1d8e3196fe4f01b9a37e4c94b4d982 : SUCCESS in 22m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bf450aad8c124f95945230090b0d5b1a : SUCCESS in 22m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/378483901a524c5b8875c1cef834d722 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/009248c3046948579138ca5ff558d0e4 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2cce8272979642f19ee4f0d07922bbd7 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b94140573f04de185db13013dd0c6f4 : SUCCESS in 20m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8448d757d84048a2a2244dd869de6f75 : SUCCESS in 10m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91438f9a55c2406a85855aeea97003e8 : SUCCESS in 25m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30ba87232ac445efbf4ce0046af268a1 : SUCCESS in 53m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed546678b78e40c4bd911c87147cb9ae : SUCCESS in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8338424215c4def86bee86cc54cbd72 : SUCCESS in 43m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c14c3589d1846d6a804abf7cb5a0a59 : SUCCESS in 40m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/689e358683b84bd6b2245bcab44f4f93 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dca29267cf34808b2818722b0eaacc5 : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cbc8281e2cb4370a6ba72e3dbd3ff0e : SUCCESS in 26m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d63a213de7f84f4fad47e2344c0241db : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":3},{"id":"18fdd3cbea9f69a4298d8b16971a7a0db9371b6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 09:17:51.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca3f1cb2cd654d25a3975011df09a95e : SUCCESS in 1h 14m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2596e2ab51804b7a8ba82a8ac41adb78 : SUCCESS in 55m 20s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7346f14061fa401fae6d251d8cb25250 : TIMED_OUT in 3h 02m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/00ee52a92aa4455498845ee2529a4266 : SUCCESS in 1h 04m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d53f7ff4c3e456d942454c91e740146 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":3},{"id":"72cdc95599a593376cf8bc49049b4dfdca6be599","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-03-23 16:39:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b5aa8f06acd0487ce9357ed72c32a3330f02d67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-23 17:01:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9299f991812b571067be9420bb025cfe8e80711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 18:15:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3029b81ff0d4e01b7b506bfc0c23e99 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/61750cd797b543f2bb817defd9d62bdd : SUCCESS in 1h 06m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f883496ec84a4a58bc67f4c03b155926 : SUCCESS in 48m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ffd593a9dfd49b29aa60fcb88b81430 : SUCCESS in 48m 24s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d02f6c66d7af45bd816ad09cab1ac2b4 : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/19dff85ccab44a8b8cf0b5b6687a0bb2 : SUCCESS in 26m 55s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7442fb8ce6424c708f063f7dd11af18c : SUCCESS in 24m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/704db62dc6134f88a82dd58daffacaf6 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6733cb27519b4db2a6deee23b090a0b7 : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0cce3e49d5cc4e498598d5ce03e21e99 : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/671f0aa2abb548b297046e3bbd439801 : SUCCESS in 18m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e9a6724601a41e1a84ddadfbfbc5341 : SUCCESS in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac8d594ddc2b4c02b2d5656fa617a331 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af3492cc4952422997b99db3c3c4e133 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d194b9d9f5b4815a12fcb813e1ef323 : SUCCESS in 17m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc8f91473c364020939f7c6cf4ccc128 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7991c42d5c840cb9125fa156864da5f : SUCCESS in 34m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e94424d00e074f36a10e5d6394791c57 : SUCCESS in 18m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8927d11112ff4743a50be5a3af33c31d : SUCCESS in 8m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/154ecf5356664bde854a7cfd54b21130 : SUCCESS in 21m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83b4d7f1f384467b9009a2157726f73a : SUCCESS in 52m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb46af8430a24667a0ff2ad88bf31b6d : SUCCESS in 1h 06m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fc412c992d34743874f95a61d0d2472 : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/840b14d91d234406afef26d5dc65adef : SUCCESS in 47m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f95346adeb7f4823856b718979d6f61e : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df8b5b7bfa446b5b7a49d19c752e119 : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62cbd52cfe8b4ddbbed8b6c0fb87e048 : SUCCESS in 24m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ba1c64d71ce4824a404a1bb77f8dbb0 : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":3},{"id":"4f29217dc7a145d39190e4d5c6fc66f28a495abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 18:16:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"926482f20e9f9dfeb9f64d5028040c1ee4f39da4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 19:15:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3cd8f27de74ff9a4822b3d80bebf49 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/cb601943b2c14926a2ec840ba3fc5551 : SUCCESS in 55m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e338afade3364742bcf62f59dc1afca1 : SUCCESS in 47m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f880e1bfc9354c60874bbde6aad69a0c : SUCCESS in 48m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0913020623204af098a922fc0fdaa053 : SUCCESS in 12m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/88f5526c78d149b78593d8b7117b4140 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fdeca2d4299649db84b150ef3054d07c : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1526abf5e1b541a8aec763bec0c220ee : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3034f6da7f0c4d3680ea702c5a74de89 : SUCCESS in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aede3f99e04d4c11ad578f1e3cf2a369 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c773089e56d450d9bdf8f1e74906206 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3394d7d753b54e92bafdb20eaa29ff5d : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7a973e5c5f1b4d37954711d1aba1de09 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/61fdc92f48f9412baab22ef13852955a : SUCCESS in 24m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/718e45699a734d3d94058447542e08ed : SUCCESS in 23m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e60a86f5b49f4f8fb44b8329faacf390 : SUCCESS in 9m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fe96e4d694274ae6916c52076f774ca6 : SUCCESS in 54m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fde2d98355264c92b15ab5f6883f7c92 : SUCCESS in 58m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da3fbffc794b4bfea7ee0fb1733b8fd2 : SUCCESS in 41m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6af9c74a9c447709d304aa9586d5077 : SUCCESS in 41m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2a69a0fff324f18808bc80d3ca850fc : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9335e8dc69b4887ba581c684ca92a92 : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b71f7922b4e44d69d252e2fcf1b81ca : SUCCESS in 30m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413f5ec536674ce582751accbe28b3ca : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":3},{"id":"c898995fa92cd912c6e4f1edae78a7e1c85bd17d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 19:15:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f8b48b7accecb55db8a1429a1b41a9bcd2642b84","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 19:20:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29c65a770f74357baf4d6bd7b2eea14 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/746632aead4e41ab9576c0ac2dcf6bab : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/533201100c35471e9971a1d19a6e7ad3 : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"24b846d3a76e2af809dda49c9650a70b51010427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 19:59:23.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3477757b2ab4e8b94456e9e5a08286b : SUCCESS in 42m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2209839ffdb64d5c839d122cb5aecfe9 : SUCCESS in 42m 23s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/07ca363950c743159513af7903d88a50 : FAILURE in 2h 54m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/55ccf22561f74a7994b46272d2cbf671 : SUCCESS in 1h 00m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/66037db88c8044b5aba65aaf535f1a8c : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffb173f8ad1e01397f50664551d45afe4a1794d0","revisions":{"56970d3e200e0ae375ed97746d775ffbe11cdb0d":{"kind":"REWORK","_number":1,"created":"2022-03-18 23:35:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/834363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/834363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/834363/1"}}},"commit":{"parents":[{"commit":"954032d5d245411e676a5722b7e581c97a0aaaf7","subject":"CHANGELOG for 2.29.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/954032d5d245411e676a5722b7e581c97a0aaaf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:34:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:34:49.000000000","tz":-420},"subject":"CI: Run CORS tests under py3","message":"CI: Run CORS tests under py3\n\nChange-Id: I49d8480fd0c775249d8b586f9dea026448b8910d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56970d3e200e0ae375ed97746d775ffbe11cdb0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56970d3e200e0ae375ed97746d775ffbe11cdb0d"}]},"branch":"refs/heads/master"},"dd34020f2f887598efc25c13fcea2b2bc206b266":{"kind":"REWORK","_number":2,"created":"2022-03-22 20:50:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/834363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/834363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/834363/2"}}},"commit":{"parents":[{"commit":"954032d5d245411e676a5722b7e581c97a0aaaf7","subject":"CHANGELOG for 2.29.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/954032d5d245411e676a5722b7e581c97a0aaaf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:34:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-22 20:50:13.000000000","tz":-420},"subject":"CI: Run CORS tests under py3","message":"CI: Run CORS tests under py3\n\nChange-Id: I49d8480fd0c775249d8b586f9dea026448b8910d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd34020f2f887598efc25c13fcea2b2bc206b266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd34020f2f887598efc25c13fcea2b2bc206b266"}]},"branch":"refs/heads/master"},"ffb173f8ad1e01397f50664551d45afe4a1794d0":{"kind":"REWORK","_number":3,"created":"2022-03-23 06:12:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/834363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/834363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/834363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/834363/3"}}},"commit":{"parents":[{"commit":"954032d5d245411e676a5722b7e581c97a0aaaf7","subject":"CHANGELOG for 2.29.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/954032d5d245411e676a5722b7e581c97a0aaaf7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:34:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-23 06:12:01.000000000","tz":-420},"subject":"CI: Run CORS tests under py3","message":"CI: Run CORS tests under py3\n\nChange-Id: I49d8480fd0c775249d8b586f9dea026448b8910d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffb173f8ad1e01397f50664551d45afe4a1794d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffb173f8ad1e01397f50664551d45afe4a1794d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
