)]}'
{"id":"openstack%2Fswift~834364","triplet_id":"openstack%2Fswift~master~I253d8e3e8678fad3fde43259ed3225df4048a458","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-03-27 21:16:46.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-03-27 20:22:03.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I253d8e3e8678fad3fde43259ed3225df4048a458","subject":"s3api: Copy more headers from MPU marker to final object","status":"MERGED","created":"2022-03-18 23:58:03.000000000","updated":"2022-03-27 21:18:07.000000000","submitted":"2022-03-27 21:16:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"834364","meta_rev_id":"8e2b09500ad351e9cc530f86165de6dc47c437cc","_number":834364,"virtual_id_number":834364,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-27 21:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-03-25 09:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-27 20:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-27 20:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-19 01:13:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 16:37:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-03-25 09:50:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"06c8e8f9c065e14407753cccae613a23fb5b6455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-18 23:58:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00f75bce299d4526c3bfe78b130dbf2638ed8e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 01:13:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60c53c086944200966620f900578151 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/005e59ef910f4233b6a589cfd7c0b406 : SUCCESS in 1h 09m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b3970e8d0fc4c7695b2132a6270ba8f : SUCCESS in 58m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53fc0d21673f498c9976017958f06e51 : SUCCESS in 1h 04m 24s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9ca05e9cc69a40d1980b294b118a684e : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aa6fc857df3b4195ac5f1c5e6ab17f7f : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/356d0bb10f784c1e99f4d9137d6ec283 : SUCCESS in 22m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7022cc5d7a4409fbfdaa8d64d11c1e6 : SUCCESS in 9m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/caeef2c1416d42ea8b251defe7acd0dc : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ccb764748254906b20e207d062e715f : SUCCESS in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e63b42eea7a64480bf5cc0caeffc4343 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/151004dabec840f681d711a92b951d41 : SUCCESS in 17m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bbb2fc7da0e4e668c9a0094d65fd1ec : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0fd3fc58a42746809ce72c2d2debbf42 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6a3fefa6c0c489c8bc703054337f13f : SUCCESS in 18m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b78f9e1be484a289f250bdc7f9d52ba : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8144806d8b824f178c9b2e0eda3c01c6 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c3da9b8f3c99498baf93e1553b39ad2f : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7e3eabffe1947d3aced636982a60b55 : SUCCESS in 8m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fdb674b6c8f44d99857b58dc78faacd : SUCCESS in 20m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1dee235db98c47ccb6809b5a485a6f47 : SUCCESS in 49m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e05d69ff3685450dba78b1dccea3b56d : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/509cd21070c14c4f8750daed74e84f28 : SUCCESS in 44m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5dd0f86d5ef84a0eb23b3dd37655e6aa : SUCCESS in 40m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99eb5ee4f29043cca38e26900f8d47cf : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f06d55ed54480e8da4bbad9fc99d75 : SUCCESS in 6m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d1803960d39481bab6c9f4f55269ad8 : SUCCESS in 20m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27eee365983e406eafeb3b491cba4ec3 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cde92b3152cff78769d2a44ded49cd55e33d9da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-19 03:04:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfc4062c0a5d4f53884208ff6dc42f3d : SUCCESS in 1h 04m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7292381f2fae491c8d82cf3eb0930e09 : SUCCESS in 55m 33s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/567e7cecea2847f6b99077803c5daad5 : TIMED_OUT in 3h 03m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b510006a5c0c4ecfab0f6d1a501498e0 : SUCCESS in 1h 01m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfd72ac686984558ac9e66526ecfc4d2 : SUCCESS in 51m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a6a00326838cae34ceb7bfe82d39eff49328548e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-21 00:40:07.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7cb311d18933a970e0930b3db8828d066aee54c4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-03-23 16:37:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62a3b2ecc2d59f7eca4ebf7cff0726d573ddff73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-25 09:49:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2fd37cee02941982a752d97fb9e3d3c5748e50ab","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-03-25 09:50:51.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc2c39dd1b56b858cdcd24ad033d0c6f16c1195e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 10:54:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a42a24fd3564e6dbd4eca15392c91b2 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a0149569c85b4528b25a8ac6a58980ba : SUCCESS in 55m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dd56ea775e5448509c84a93ccc514b49 : SUCCESS in 50m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eeca616096454d3eafebf0c66477f6e3 : SUCCESS in 51m 25s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1750cc7674b547a8b070537e406510ec : SUCCESS in 23m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ec35f5373696472cac4e5a378e63629b : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6a3ebbd8e4044e0b99383250492e6658 : SUCCESS in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ded6ab7a0f24c1fae193c6fecf4d7e1 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6348de5f58d14cb3b493780ec8e2d498 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64baa4e41c054041b8a0431571fcb0b5 : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf37d804eaf84a55ab1c03d477465def : SUCCESS in 21m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4ec1bc88b6c4c3fa8cc562cb7447dd9 : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ca366bb2efc465f8dbeb7d2dc8fa151 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/945a032d9d674ed290c632851a5256a5 : SUCCESS in 24m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d59236c595ee4f9886d36d31990ef141 : SUCCESS in 24m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c197d21848f14fd19c24ed73bae913ec : SUCCESS in 21m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a38718565fe84479a483367488052168 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b1b71123e16e464a807f10760810078e : SUCCESS in 22m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84cfac8de5df493590c2f13bc4ce7d6d : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c47f35a100a5419684b9a59c0cc7920c : SUCCESS in 27m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/463f0a4c4e9340a891b2eca649bd662e : SUCCESS in 56m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d688bf7420548639632e4fc8236c317 : SUCCESS in 1h 00m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb7d66cba64b4fb393944cb5c513487a : SUCCESS in 41m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc6c6d40e884438b923b4a7bf835dd1b : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a33e0a2a199a4e0c90024feda24e3ce2 : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62e2d2b430f40358d481424ede205a9 : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c027a3741a584aa78d8f7a22e8f05414 : SUCCESS in 24m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9167ce4147334a57bd888bd69c228da8 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":2},{"id":"22c4924bedf3d3aa3bd80bfd10594a575bb157dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 12:56:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/606c5b7251864ea69ebacd8d8064a1cb : SUCCESS in 42m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1bd04a62e50545fc8fb443ffaea9a99a : SUCCESS in 41m 52s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8f84aaff39ff4b4f8df15a08dda83208 : FAILURE in 3h 00m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/94a9842a78aa4ff9a7c16cf4b1b2d257 : SUCCESS in 59m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be0af6437f3a4482908ea9320ecec6ff : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":2},{"id":"39563dc77a629e13af4aad3a318f16fa8c542daf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-27 20:22:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"883841b058eddef1d1ff6b7fba132ff2e8835e42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 20:22:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"746a3f7a409b6a1ccc9b4c9d887672b0d127b669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 21:16:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ed3d5a4917462994e42ac5daa19bb3 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/7b67a65c047946e4bea970767fda0e21 : SUCCESS in 44m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/604de4a8b43046cbb9f4bbd5a0472d57 : SUCCESS in 35m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/774864d581574a3c9f5d4d8817eb7550 : SUCCESS in 41m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3448926e93347338d3b7819ebc501ca : SUCCESS in 8m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/24ed7cb7e5e44ed58155c0fa9ad428d7 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94df2b6e2c9946d7ae44d2e51ce51aba : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf154329255e4ecf95a6afcceb0f5b5a : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84f7d20b44984898a7822df2098ab3a5 : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b0fe30c41084fbbb882e3b6c28cc429 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e43336bd2a8d4196ba01e630338363dd : SUCCESS in 19m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e649da98ff84a04b3b7851a9092466f : SUCCESS in 20m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a14b8c1905d14f4dac5f02589a2dc783 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2562d8885204d238d548c22d1fb9d6c : SUCCESS in 18m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/63a3602edd4641aca35e3d8388876de7 : SUCCESS in 19m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/302674a62d9f4f75846d3d306632c808 : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6737bd807f8349f997cb98bba6d9fc96 : SUCCESS in 50m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d613bd3c5de047b19419f7b13a45926a : SUCCESS in 52m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00a50111e4de4462baec43b177298dff : SUCCESS in 32m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d49061b3ac134f35b4ddea047aa2486c : SUCCESS in 33m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76c95817def44307b2e5e3182e569ebb : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21de2a8ff3146e8936c9b8891339dd8 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dab88e3f1db4be38a6338ae87cad46a : SUCCESS in 21m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac035ea296ed4d73a30beae418fbb6fe : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d883761fd978970ce9ad5ac023af733a9c2685e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 21:16:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8e2b09500ad351e9cc530f86165de6dc47c437cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-27 21:18:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8842b225faf04678a73f5e505eabd010 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b12733b2790d42a08b29f707aaddfd15 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/abbf11244da8452b9674132221d1b4ea : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c4acf2d8f7ab5a3170ab958290c5124920a6d1f","revisions":{"d60886b85f4e2bca591bc5f54dd4136550b4a141":{"kind":"REWORK","_number":1,"created":"2022-03-18 23:58:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/834364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/834364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/834364/1"}}},"commit":{"parents":[{"commit":"cb8b3cdab262af1d223c0536220400b13c1d0a9a","subject":"Comply with AWS signature calculation (s3v4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8b3cdab262af1d223c0536220400b13c1d0a9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:57:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:57:46.000000000","tz":-420},"subject":"s3api: Copy more headers from MPU marker to final object","message":"s3api: Copy more headers from MPU marker to final object\n\nChange-Id: I253d8e3e8678fad3fde43259ed3225df4048a458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d60886b85f4e2bca591bc5f54dd4136550b4a141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d60886b85f4e2bca591bc5f54dd4136550b4a141"}]},"branch":"refs/heads/master"},"1c4acf2d8f7ab5a3170ab958290c5124920a6d1f":{"kind":"REWORK","_number":2,"created":"2022-03-25 09:49:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/64/834364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/834364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/834364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/834364/2"}}},"commit":{"parents":[{"commit":"cb8b3cdab262af1d223c0536220400b13c1d0a9a","subject":"Comply with AWS signature calculation (s3v4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb8b3cdab262af1d223c0536220400b13c1d0a9a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-03-18 23:57:46.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-03-25 09:49:11.000000000","tz":0},"subject":"s3api: Copy more headers from MPU marker to final object","message":"s3api: Copy more headers from MPU marker to final object\n\nCloses-Bug: 1966396\nChange-Id: I253d8e3e8678fad3fde43259ed3225df4048a458\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c4acf2d8f7ab5a3170ab958290c5124920a6d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c4acf2d8f7ab5a3170ab958290c5124920a6d1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
