)]}'
{"id":"openstack%2Fswift~834440","triplet_id":"openstack%2Fswift~stable%2Fyoga~I5359b89212bfc6fd8ab03050436d255250d3a5d3","project":"openstack/swift","branch":"stable/yoga","topic":"create-yoga","hashtags":[],"change_id":"I5359b89212bfc6fd8ab03050436d255250d3a5d3","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","status":"MERGED","created":"2022-03-21 10:09:25.000000000","updated":"2022-03-21 13:20:20.000000000","submitted":"2022-03-21 12:51:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834440-create-yoga","meta_rev_id":"83bba0d1ba3164deb99b094b12d31837b8a2926a","_number":834440,"virtual_id_number":834440,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-21 12:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2022-03-21 11:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2022-03-21 11:50:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 11:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 11:50:49.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4b7c88dee81e4e0721914b04926871f214ac7b98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-03-21 10:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"842fe68ec3151c1e310041b504877e7afb5e2981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06326a37847f40789f9c91cbd76cbb87 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed5cd614076640fd915496007d645dec : SUCCESS in 7m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/367d526741994e978cd57bb80daf2eaf : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f41a81bf186a47df9449af262c512aab : SUCCESS in 29m 42s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7873cb39113c4afab8595ba3ee2cbccb : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e38f310bb414b6ba12ff2f74862db77 : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e0c550ffd5d43c9a25738b3dd7b140c : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/579cc1291a174139b0e410cb3239bcca : SUCCESS in 20m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1da02b7fc90b437a8647078eb2c9b92f : SUCCESS in 14m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/295f998da4f243baaa27ac30822a9f56 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d46cc269a5c4c849146b8811bb9da77 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f4341114468f4e758d2689d98abe21b2 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/00a37eaf02304c3ea0c3730b628d974b : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7751215446b64847aae59f252b9fbc28 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/adffb85ee452417ab3c620b8775bb4d7 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/93f7cc97185349b4b28cdb9fee3bab3e : SUCCESS in 22m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e706b4936b242ed8af3d8fd770f5fda : SUCCESS in 8m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa9dbe8ac60c4552a8d989084fef818a : SUCCESS in 21m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b6f0ad1c7a234415ba9d8c434218f400 : SUCCESS in 52m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/74cf00ebf2634411a0a1c1bd10201d64 : SUCCESS in 1h 05m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7e6c8535f514be4a99422a9adf99bf8 : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e6cb756ee841e9bb23bb195dec29c0 : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce5c0148dd8b409cb097041d6c920d0f : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dd304c9278d8a5d39d7be0824156bafa167e6da6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-03-21 11:50:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c770e6844fa69b817429a324d043c9635a2832ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:51:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"585111c09af39d9f8b9342c88a5dc140df7c6995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:51:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ae05d28cfd49c289f4a41dd5249201 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12492f8f063746d4898184006db02b89 : SUCCESS in 6m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/386c081cfee34ed4aaf73d92ae0ad606 : SUCCESS in 10m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b7a27c03c62f405da7f7b9bafcbb7c99 : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c52791e7cde4aa0a2eb8cc7970e2d45 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8eae72bab9184b14a4e0b9a26c83ff59 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a855e0681284fc1bd4e15bb47088c81 : SUCCESS in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f32c36e6e6c144689de26f2f34b389ea : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d3c8d81072f48b3ad2452854e2d69cc : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9212bc224fb6454893ebdffc3b303ddb : SUCCESS in 18m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f94c3c8c6ea045fbb053ecec595c08a8 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00934e48411b45a68b2639378b5f4e03 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27bfcda360c54a3983156944ce909c34 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1488c9d5bd7a479d93899a48c348c6c2 : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df2348b9e74b4ce9a309ce61e9e5ff74 : SUCCESS in 54m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a21b262dafe5488396a19ed15bcfa4b0 : SUCCESS in 59m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d15fc62839f43928f6e8b0570c6d223 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a15665d44984caaa91469bf714de854 : SUCCESS in 7m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/96f2bd6be6e648eb9635ced0ce1c8702 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d7b290c1e6118bf8ed84ddc296ff8e2b7c35bbf8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:51:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0bb6db508bbd0c259cdff878c1256bfc2f86071","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 12:53:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4e57504f2d94a8e9a59b6f6a298796f : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b22c5e182f7a45f9b78a0d9c2b2136a0 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2f03dbee585046dcb5cac1db5ca4743a : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/141b2b8043c6417fac59d6808a69a992 : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83bba0d1ba3164deb99b094b12d31837b8a2926a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:20:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/047fc67adf454cc0bad45ccf1a5ddd42 : SUCCESS in 56m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a4d023acbe3468ca3ad83172f0c738d : SUCCESS in 58m 20s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/f3da722bbe224713b39497aefd1eeece : TIMED_OUT in 3h 02m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9c669bd545f4093b1db3fff32788c2f : SUCCESS in 1h 12m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6134bdd03aa8461197da9578145b600b : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"687515c064fd4747a03c3df42553369bb9585656","revisions":{"687515c064fd4747a03c3df42553369bb9585656":{"kind":"REWORK","_number":1,"created":"2022-03-21 10:09:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/834440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/834440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/834440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/834440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/834440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/834440/1"}}},"commit":{"parents":[{"commit":"59e7ddd1bd09077b616c31efa06de1ea5f8df2ab","subject":"Update .gitreview for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59e7ddd1bd09077b616c31efa06de1ea5f8df2ab"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-21 10:09:24.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-03-21 10:09:24.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","message":"Update TOX_CONSTRAINTS_FILE for stable/yoga\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/yoga branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I5359b89212bfc6fd8ab03050436d255250d3a5d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/687515c064fd4747a03c3df42553369bb9585656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/687515c064fd4747a03c3df42553369bb9585656"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
