)]}'
{"id":"openstack%2Fswift~835001","triplet_id":"openstack%2Fswift~master~Ie78fa3f3b4403689fa7689d4319157993302176f","project":"openstack/swift","branch":"master","topic":"ringv2_history_chain","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-03-24 07:23:41.000000000","reason":"A robot voted negatively on a label"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-01 10:33:31.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie78fa3f3b4403689fa7689d4319157993302176f","subject":"reconstructor: Use last_part_node when building SYNC jobs","status":"NEW","created":"2022-03-24 05:48:06.000000000","updated":"2026-06-01 10:33:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3737b30c3d22a134568e0e4b1aa95bad13ea5b0","_number":835001,"virtual_id_number":835001,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-01 10:33:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 07:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 07:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27ef4ca575768a1b604272a77096c26f684aeda0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-24 05:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de194a3f76f5717085155b7f91d36ea3b48ed556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:23:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eb3ae7658a140bdadc7375a2866b484 : TIMED_OUT in 1h 32m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a43fc45fbbcc4e0fb5f6fa1cc1a6fd82 : FAILURE in 47m 44s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/88ceb4c0544d4db0b5df3f34645dee14 : RETRY_LIMIT in 8m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b5c827728bd4f669f43d8b39739c88d : SUCCESS in 55m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e77e7eb7c574b778377c2e411b05918 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"646806e1d1143b6b935811c72dba133bde96d031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 07:23:41.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab9a3ca18534bf9ab623c6286979dc7 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/95ba4fa6e8ac4bb7bd4a61bcbfa038c8 : SUCCESS in 57m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e843f2dea7a84e658f418e8e3d3e8f11 : FAILURE in 33m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64388d49f74b49df861c362ab1d2f1b1 : POST_FAILURE in 30m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4d6d6a7e89184ac38316b7ac6cce5a1c : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cfe09613156e404e9cabc84e6fd56e22 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/098b39b42664445ca4f4543ec552d3ca : SUCCESS in 24m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34bf5ac7ec704511a2404d6142b734ba : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e7b952a2faec498baf25f732760bcd1a : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3886c80d8cac4e149ddf7695c80294b3 : FAILURE in 6m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff6edf50d1fc4a2c8ddb16b8b446053d : FAILURE in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0563686dafd4e12a1dc732414ee1e76 : FAILURE in 17m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a80f54668fc140da8a14b935331999b3 : FAILURE in 5m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7943f4359843452a8e705dbb813c5e14 : FAILURE in 6m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b40703f8487441d9efb2cea2ca183bb : FAILURE in 5m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1cc64b2d65f44d4698e0a60c6b94133f : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00f98d3487ca4ea5b07cff63e825514c : SUCCESS in 18m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa1f882dd17d47c6a5c555db26e3ecec : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64ea70a1825644398899657a8adf4a09 : RETRY_LIMIT in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da6e33a1d03e42dbb80ed3a73c99f10e : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8ddbdcabb6db4c65ae186e45cfeeb5d1 : RETRY_LIMIT in 5m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e77b6763a403434fb40d1386f138c301 : RETRY_LIMIT in 6m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/80cfc5dc28b141419cf519cf940aba9e : RETRY_LIMIT in 23m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd66b7a7931c4fb8aa3c3e47cfbba9fa : RETRY_LIMIT in 14m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcf35272c51a4fea8262e50e51f6c18e : FAILURE in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6205998ee8a04d688901379b90615b1a : FAILURE in 7m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85fc807799504111bcc31248a1a283bd : TIMED_OUT in 1h 33m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6f58f2e65434a0faab4d6b70cb8f756 : FAILURE in 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"551eb3cd6cf994e5f2bfea124788654fe0635bdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-29 05:14:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18639059f6df8926782b6e47db30084eb6093304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 05:55:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02e6c998a5af484a9fd54be57654ac6c : FAILURE in 34m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9bca188ff68428495198a6b9fc08c3d : FAILURE in 36m 32s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c2492870817d4b25bb758a8cb75aae1c : RETRY_LIMIT in 7m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9acaa9646e814449a0ea5d53ed9254df : SUCCESS in 37m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/11510d1aa4ad48898e8ce607fc5bbd8b : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":2},{"id":"84c597baf8a4e580eeff68adbca0dab11f7223f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 06:51:12.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee80c020cc54cdb92a5dcb25abc4cb3 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b899464cfe09438fb4231f09370aa608 : SUCCESS in 40m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4cc429b124964376aff2acbc317ac9eb : FAILURE in 28m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a4c909b146b4125a7a3fad003fb769a : POST_FAILURE in 23m 15s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0f62e7751d124df690111512e0fbe88e : SUCCESS in 26m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0206c43201b34c14a9eb672c4de94576 : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/840fdbc91e6d436293df2d8cd04b2490 : SUCCESS in 23m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b3ffa7ce6004ee5a20165f79d5adb86 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3770f701dc9a4ee698ad67812c214105 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1a2d5309a734602b881e061e3243902 : FAILURE in 7m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45bd0035ec6c49a8bafc21da75c6bb0f : FAILURE in 21m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b83b7feab74846eda4d49d526439a477 : FAILURE in 18m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ab6c5d3a9014a50b3a78699d1f70096 : FAILURE in 6m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65593d88366f4c6397920af64eecbd1a : FAILURE in 6m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/597fbe24891d4d91be39a7f84f514667 : FAILURE in 6m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63cac22354ee422fb0eacca1a07d1b1a : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/16e071c542984922a51d2319f2e437a7 : SUCCESS in 22m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/088ee0a95d5f4f249fca69df3a3fe6bf : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77ac8aac5db548efae1e524ab9df05a2 : RETRY_LIMIT in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b1f73494f9e54d9ab9e58a4188654ec5 : RETRY_LIMIT in 6m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/077ae67fddc0455ebf4b94c70b11da53 : RETRY_LIMIT in 6m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e0c17562fa94b78990b8261c4b4ef4f : RETRY_LIMIT in 6m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/211753bba90f49f09704ad25204cbcb6 : RETRY_LIMIT in 22m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bb1f02e45af410d9f3313d0cd809823 : RETRY_LIMIT in 21m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63bc410cbe4b408db2a5936aeb1f1b90 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b00aeddea5f24a66bb822243943ae06e : FAILURE in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de8585d660094934a8fb41b6f4820aec : TIMED_OUT in 1h 32m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22817f4cdeea4919b7dc1430950deae3 : FAILURE in 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5f0e755d72af57163f8e2e17cba6e86e4ca58ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-01 06:08:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"84ffaca30b2bfb613ddfcb633a3a8ff3964d2a8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 07:19:57.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10eb923e36834079ab7c06e7fee7d5ee : FAILURE in 37m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/490e6718077743a6a2ecc51a7e96b247 : FAILURE in 40m 29s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d3396f0c9e5a43ddab7e1bc6b71f4d99 : RETRY_LIMIT in 6m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdb1d1a7e98b45b297a77a6e55f97f7b : SUCCESS in 1h 09m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7b9c8ce863445c0acd4fd3ade5a61c5 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":3},{"id":"80b64f5be7e7cd7bb4feff9c8af6ec95103ea622","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 07:48:06.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c1b5024e524aef952277c582461446 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/6a98e9926dc54f8ba5af35b7ab4fda6b : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb676658e4b64d59a7dea45e1017ab43 : FAILURE in 25m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/831121f6d504442597d1e574164599e2 : POST_FAILURE in 27m 45s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bc50531fa2db4f1a8e9c3d6742fbe42f : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8522b9e7e56340a398160341374d088a : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d2ed059ebf9a4002b68e87959060d7da : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b8f0912f681464fb0d4d699ca3aaf0c : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/15a3b0cb970d4268bc85c476823c664e : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/692174b9128540abbb8868d64045a78d : FAILURE in 3m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3fc4539013e4b9885c618119c3342e6 : FAILURE in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b394e0e81a68415e86fe8a02b8724536 : FAILURE in 18m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25afde0979fd4b9d9d0093b0d1c92ebf : FAILURE in 3m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/10037671d8824bf7acd71d508ef4cbee : FAILURE in 4m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a2b6718f4435443f92d56ce80aa42236 : FAILURE in 4m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/74395d1e710a4a1bbe4e41f332baa7a4 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/55871a60d6d74d0a91811a20b06e3f17 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae4e7b5ae03f4890af60d64d80a736b3 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d97c89fd67b479cbcf314fa0f3c2b14 : RETRY_LIMIT in 4m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/435f66235679494fb7e74e438cc2c6fa : RETRY_LIMIT in 4m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e582746653774c8ebd7703c469cdfd7b : RETRY_LIMIT in 3m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c7b9d213bd04630b04e53a9911695e4 : RETRY_LIMIT in 4m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16d9daa6d1624b1493db6aed290dabdc : RETRY_LIMIT in 25m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd909ebd3a2c4caeab53ef23437b68bc : RETRY_LIMIT in 10m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5b8706fe43d4f2e8ae95f1bc28d983e : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b1d17c46d744918721a0b4977fce8c : FAILURE in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/998d68f99eb7434084418c2dfbaa8699 : TIMED_OUT in 1h 31m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/266f9fe4ab514440b413a66ccd667b63 : FAILURE in 29m 57s","accounts_in_message":[],"_revision_number":3},{"id":"24ccb42d189837c58b69958c6df1282f086d6689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-03 23:37:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ab3e7cd317275a711bd88af87885cbe048faa4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 00:27:15.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7d145984f914cef9747d3df9138bf8b : SUCCESS in 40m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ef7f167aa83403683d2763113c027d0 : SUCCESS in 38m 20s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/96253d2b29564cb09063f6b4df4f3d73 : RETRY_LIMIT in 6m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9edb709869ad40bd9558192847f29898 : SUCCESS in 47m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cd8a19693064ce89198559e54ce2afc : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":4},{"id":"8083cbf1909d24ddeb8ee4bd19540bacb0aaebaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 01:24:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a136947921c24b78a37fdc73b888153f : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/df2eb600e39b42c7b10019be24a49661 : SUCCESS in 54m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81b4a9e0532c40848d7efc57abea59a9 : FAILURE in 25m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/991e7484f9bf44b1be3d495681ee3e75 : POST_FAILURE in 26m 48s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e9892de868104b3ba0b7457cd62933b4 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fc3b0a2b811d4775a0aeb6f740220931 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/af1cc8124319467888754845c7910263 : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7575bb3c88324c49932c42acb69ed365 : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/186c018edb8a47c5be3eeb62b4c3d565 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec12befa28514ac786ddc67831a3be9e : FAILURE in 4m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/415955f0e81643adaa03f9b3ba78f863 : SUCCESS in 13m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59fad008747d423d9f4e4f1a1d56f8d0 : SUCCESS in 17m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25103347803f47859e1db6fb286b86f6 : FAILURE in 4m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/817c1c86410544f2bcc9149310189f15 : FAILURE in 4m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1d544c1c78e4dfaac33fc31b7c19d29 : FAILURE in 4m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4b73101e71794083bcdd0b08d657218c : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/72b9a8e2943b47f29a09423b88311a45 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/48e7d444d33a49e596384a571a9bcc17 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac9d769d11eb4f8ab7dcbdfff1351a34 : RETRY_LIMIT in 4m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72662ebb715740b78bc54d9011da2d00 : RETRY_LIMIT in 4m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2cb9474f3cc44b4a984d280941a4b8fe : RETRY_LIMIT in 3m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9944813ff6149e6becefcf69573fe92 : RETRY_LIMIT in 4m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85191586d92a45cc8e7b3ff07be79134 : RETRY_LIMIT in 18m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb90b645007a4fa480d64c347cf4667f : RETRY_LIMIT in 10m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52e908e1c8b14d94ab3b02359a228bc5 : FAILURE in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a22062f6fd941f39b8c8bdcd0aaa098 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cae8150c1d2f477b9408ce7eab2a2916 : TIMED_OUT in 1h 31m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac701ff5d47f48fca8bdd8a4d841246c : FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":4},{"id":"86ad000503b9227766f5abeea1712c5bb9df4f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-19 07:28:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"cc380f718700d9b05fce60b73ca4e8a7b588fff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 08:32:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727c2883e3964980927a05c88033f1e1 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/f92f0e2f3e1f4ecd9bc260d76efbe49e : SUCCESS in 47m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33c1343d2c134443bcca92d4321911a8 : FAILURE in 31m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5236f466bfd54e5aac0f6a5ceb711927 : POST_FAILURE in 27m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f69699ae576346f09c95d657164c8250 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5566863a18754898ac96e49f2387bfba : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b0a47df2cecc474687232f551736524e : SUCCESS in 18m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/688191c3e37a46ce992ead24d01eac5c : SUCCESS in 7m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ebfc0d0a79f4d368e348b1318aeba76 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83ff0f9cba8443d6b7bb1553519dbcbd : FAILURE in 4m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/404619c1ac0341c08651968c345d3d36 : SUCCESS in 14m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95e73c8785874894804daecc1a2b6936 : SUCCESS in 14m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/57abcb5f801d4f0dbf6ef063856d0552 : FAILURE in 4m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df9a47a46b2e455a807051dccd089959 : FAILURE in 4m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e98ff56053f4179b42c06b118c1c839 : FAILURE in 4m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d586a9a639bf4ec7b2131e361806ef7a : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/202b9dcc5fe94926af79179d2bbd69a5 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d24c6651af594c838044ffccde172502 : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b607ec5ee3e04280815a60ddb583a644 : RETRY_LIMIT in 4m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/883f14438f2b4ae782d2d9d79a00abdc : RETRY_LIMIT in 4m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/824764a215734c45a53d3f683cbb4ad1 : RETRY_LIMIT in 4m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/69a1846979824d26b8a2d6c57de6521d : RETRY_LIMIT in 5m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8b661bc6d364375b2607aee2a8519c5 : RETRY_LIMIT in 13m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bec57c4fcf44577a6dbb618721af0b0 : RETRY_LIMIT in 20m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69c0bc775974440e9bf2ca15da863845 : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41104ec8eee48f2897354d2b931a26e : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5e5f3f75c1a4e2d8c6cd63cbb78b750 : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d11db1ad7e7e4817bfb64b27d98219f1 : FAILURE in 22m 26s","accounts_in_message":[],"_revision_number":5},{"id":"4641d801449146d772e63a0d917a754a7b6257f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:42:54.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1214ce517cc4c1ab012b04961ce62aa : SUCCESS in 44m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/afcfabf64dca4c2d873930789bd1879d : SUCCESS in 1h 18m 36s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b4fc49df9eca486c82cc3d289eb3cca2 : RETRY_LIMIT in 7m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0911c44dc7b45c9b34d569497a7fc25 : TIMED_OUT in 1h 31m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a5dcec3f5094438b398b4362345abd3 : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1209628412e7178b5b75d2274a13188ccb119cd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-20 05:45:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"0bf6a9ebbfd6749bf04b01cb9ea2d6d8e6fa44e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 06:49:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279ec22f61cd486fa4b969014f54bfb0 : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/20142bd810f64f408079eb1216e0caf2 : SUCCESS in 54m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b09bf088d38a443681d112f773444334 : FAILURE in 23m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4c7efe039c94da3b19ce9736e1a420a : POST_FAILURE in 24m 25s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/213b5a92a0564bc3983316988adba648 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a7a01985f36b48b3b9a0d52806d1c65e : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e6276208d5624fe880d00c43c8b0b6c8 : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/104cc54e42d1461fb22f9a88bbd952d0 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b70eea60a2d0401c86186ef02b63f472 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb4934120504a9cafb9be347d6a61c4 : FAILURE in 4m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4287f67ed61a45bda4bc92f51b2ab57b : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f4c4a1ea3164c69a39b39967d644e35 : SUCCESS in 21m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/389454646b15484f8faf9a0496ffa9a6 : FAILURE in 5m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eefcc692a405437da78c970e166fee21 : FAILURE in 4m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8f717ed95144adf9e42cd1bb00e09b6 : FAILURE in 4m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2460b50b82541a8869c4ea8b56d14a9 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/864fc72866944030b3354cb1815854a6 : SUCCESS in 19m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff6c7209e9074c7aa7831f9806b4aade : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/027f5def3c864afa8823f0054e3a5df5 : RETRY_LIMIT in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/061636e732e5429f93ed76c9c0cfc2bb : RETRY_LIMIT in 4m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25277e44f22f480db99bf6a1f3eb2c05 : RETRY_LIMIT in 4m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0084297106b7432bb11d6d8e76d2a644 : RETRY_LIMIT in 4m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfbf0ea8f9bc4db89bc47de7282249e8 : RETRY_LIMIT in 12m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/478b368ae0f244288f824d18fb3cf91d : RETRY_LIMIT in 11m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba48d0e902bd465bbb6c0636d3c42bbe : FAILURE in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79b9d1f3d5e462691f4155a173eac54 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98068abf8aa546d783867d33d74d60ee : SUCCESS in 23m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3869a6884f0421a81a24ee4ecb1c037 : FAILURE in 22m 15s","accounts_in_message":[],"_revision_number":6},{"id":"683da898bb5790dcfec705fae992ea82eb02517e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 07:31:56.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cf6fb084ba44de7a8c2ce20273afe0f : SUCCESS in 46m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/578a67370dd64d30a142607275382a66 : SUCCESS in 45m 47s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4504a9651124496c85527b33a9ddebe1 : RETRY_LIMIT in 7m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f5bef7e39e541de9a888b9e014808b5 : SUCCESS in 54m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b034843a6c794779899636e50e8ba614 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":6},{"id":"62b2c776986ac9742a447d9cbf9df2c78b5a2c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-28 04:20:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"4b7bb988e9db35fb2ecb218bf7b5ae39e4cd8e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 06:34:23.000000000","message":"Patch Set 7: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e01ba036ca4155abdaca092c74c4ce : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/fb34bcc757944a7db30470cca07d9898 : SUCCESS in 58m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53eb1333d8b44fd5ae741783b57bf3f3 : SUCCESS in 35m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/beb99e12c2ac4eeab1298d8a2b552212 : SUCCESS in 42m 45s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d1e3ecab9fd041aca76c52aa09222d1d : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/909e4254baed4ab1b4f924f2906e58b7 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aed1364c134a4bffa2c903df3f2e5d62 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07731649c4a9498cb6132e7e4861ffd3 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/556e1bf85061454ea10b4ff59a4b2df8 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd94f064d3b4449ba76c9304053f9e48 : FAILURE in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8634dc68925e4dfa80d8ca72be5b77ed : FAILURE in 13m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ea1eafe738e446ab383c0e849e4be23 : FAILURE in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e4db1b1ded924b39b09d9f1429bceb7a : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e3e61b0d0b024ab4a0782d4a273c7090 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a76d85ad9543477c8f3468f064e6d695 : SUCCESS in 16m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/efaed93f992b49a9abe6c377f12c982d : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1a23101a2434179b8b444afc34a7d00 : SUCCESS in 13m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3ab2f5a30bb4357aa06e79de921a55a : SUCCESS in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c765d0b353d1425fb8bc577a39be2573 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6aa7787d72b4e01b892a022839d84c0 : SUCCESS in 15m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/19f55fdd8bfa4582b9d6156485eaccb3 : SUCCESS in 42m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/86ce0b38ca1f4a5d92662d30b7787837 : TIMED_OUT in 2h 01m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a32a2b17571d49758e90d6292c9fea58 : SUCCESS in 30m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a92cd1f20fce4c40b9c31f99d2d05330 : SUCCESS in 28m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eaea4c93159402e9d4f910c258f6703 : FAILURE in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a11e2546c847f79e729f1340d67af0 : FAILURE in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4dd4e900b4d4b9db3d7cb726c22f717 : SUCCESS in 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15a5b96b7ea4423e8022b5057fcdfa6a : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":7},{"id":"991751aeecddc2236dc13dabbb4bab430751d987","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 08:07:54.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/84d10dda5a5d425b8eef94cd0fcf1ed7 : FAILURE in 49m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/587175ba4e574f23ac7f7780890a4d15 : FAILURE in 46m 04s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ffa0b33bf01d49b7aa56a0960d651f99 : TIMED_OUT in 3h 01m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e83c8b8650548a6ba76409e23b8097e : SUCCESS in 54m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dc0e07efe784130b779045fa158d2a2 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":7},{"id":"7eb504c87069ba1e046632f0915c2489568ef96a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 19:41:17.000000000","message":"Topic set to ring-v2","accounts_in_message":[],"_revision_number":7},{"id":"4182f8254dcb0d135c62104ab1738bc9c92bdc43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-16 10:05:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"97a8f82b6d275917d5c0e49aa7d7334104f5eff8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 11:54:44.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e7221e37bc44d5fbf1cfd34e57855ad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/529f4ada9fd9420f9e9916eed35e27ae : FAILURE in 1h 47m 03s","accounts_in_message":[],"_revision_number":8},{"id":"e02f78b3b68f1fcad7b4f5c602000c8d1f2ae2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 12:09:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3bde3c83fca4b1dad3c5f402f68d33d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cefc8025c154ee6a385b6be34cbd574 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/0541be9500cc486ea382b0ffecca619a : SUCCESS in 40m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67a4cbcf4f2f465c8ee4720ce4d7a4cb : SUCCESS in 59m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8c5db250be0448f896d5bf0dc49b181 : SUCCESS in 56m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75bbde8f9fde411a8a910ba780a0c1df : SUCCESS in 1h 06m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/032db33b98c6432a82e3a8ab2456e0a2 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5707942771674bb5bbf1c5eae8f26e9d : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29666e57a0cf4cce96bb7f3e1b72928e : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d620bbb932a461a80dd97ebc27b3c4f : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/edbd096900bb47ca97cbc10bd93a2f23 : FAILURE in 54m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/59ecbfe94a5b4d099d873fbfe41409e9 : FAILURE in 45m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f55c29120614818be110aaab7f81788 : FAILURE in 1h 06m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5aa766637d4a49d9878ef915ac484297 : FAILURE in 1h 14m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/612ff85cd6a242a8a348fd011c8b4f23 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/98d3fe2ee1b242078436883d112f25fa : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2834f647a4be4652850e5b655397d220 : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f847cb5ccdc41489999918117f47b29 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d53fd537c70b4078932f9edc450cf2bf : SUCCESS in 12m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a1b65fa51e6b4affba90b272d6a8b8f6 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/140dcda09a634bfd9c1e099578861097 : TIMED_OUT in 2h 02m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e914133bd0244f448d1eb802e9a26cbd : SUCCESS in 50m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66b8765086b048059d4e51eed1200965 : SUCCESS in 35m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ca7d48135e44b76a3e61c98129e318f : TIMED_OUT in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ef1a700cd94ae29773a7f3f882bcab : FAILURE in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b580f597a984aa98ab308017e784836 : SUCCESS in 25m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e6a196fdeb046448b8cabdfc9b32e14 : SUCCESS in 46m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/537388f79ae64eaeb9a1deedd5a6fa7e : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":8},{"id":"e60875ebedfb22f04f16356881a1507a1b8f238a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-24 06:39:04.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"1fcee91f28234e89a4a1c057d8953dad9a1bb2b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:43:30.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b8eaa0c05ea41ccab2d890b346b02a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24396057cd924ba48652f2e3388b1896 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b3bf285f7b444f80a0b255a9ad3ffff9 : SUCCESS in 41m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9930ba2574184d029fadb0c462cd355c : SUCCESS in 53m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b612d71fd73d4c8f9255bc7074833ba1 : SUCCESS in 49m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d6df6d4253f4e798c180b9548fe4302 : SUCCESS in 1h 08m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abc853357b04468ab6c088a119c41dfe : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a0291a95dd245b3b0458dff37b34774 : SUCCESS in 12m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f65ca0fd04f84f2ca910265cd6aa1494 : SUCCESS in 10m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa2d3e12df8c445d99444ccb13e3e47b : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b6c91605f6c6425f8da11210e03a2fe7 : SUCCESS in 1h 12m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc6a31bcfeb84eb8aae1389794104361 : SUCCESS in 26m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c2d46a40b83d4301a4a4d285881ca8d6 : SUCCESS in 24m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a4ecff3f57d347cc86d3eb5c2b07ed3b : SUCCESS in 36m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/95adc790d5cc465988b4624edf6cbae5 : SUCCESS in 8m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/11c3650e5ceb49a4b264dff5a4560b45 : SUCCESS in 7m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3cd0e1e7973c463ea90c1211b1a9204e : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65858248f9f047748586518a4f5a26d0 : SUCCESS in 6m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97ff4537515b43c3b947a3a411837bda : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/98c5a49e3b384b57af8f8dba027d5959 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4bf91e587bc4e689917749bbc546ef9 : TIMED_OUT in 2h 00m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0424975aabb4b6eb4362b35654fd2ee : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51d30618be554c07a3f321aa6cdf2ebb : SUCCESS in 51m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aac7ba98caa4a37b3b91aec3837cdbb : TIMED_OUT in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681cc2c873214f398d8c125a70ae29d1 : FAILURE in 3m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d371219bd1df420cb2defebe70053903 : SUCCESS in 16m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a22861ac88f24932b7d9196302d13431 : SUCCESS in 50m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/583f461abfe24f76af856ee07b6edbcb : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":9},{"id":"c6dbebb7b69b0cfd83065f5a67bfcd6ea147759f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:54:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40634c2ea6bc45a3b2b3c95c1ebc4d84\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/837397f27fdf4650b978693b98336173 : SUCCESS in 2h 12m 55s","accounts_in_message":[],"_revision_number":9},{"id":"cdde8dd1b1bcca34092f8f2a46965b0f4db67880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-07 03:46:16.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"2a06342b99ff2cdab8aa026cb1ba8af3293601a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 05:25:49.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe775f57053c4e9cb21cf9694d0c6236\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f45b6bf3044b4d239b6c3bda5865c419 : FAILURE in 1h 37m 17s","accounts_in_message":[],"_revision_number":10},{"id":"106469b14f8af0fb0cfdd8038e18e250cbd33dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 05:51:52.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d562f065b8c64d529dfc86ce32f275cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e038d3b89944448c56bde0008da449 : SUCCESS in 5m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/0d5916a2a00d45e5bc013b9ef7a68099 : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac0c319fa8234997810cf9ed68e54929 : SUCCESS in 57m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/402edf29b37e45d18e28f66b084772b8 : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6b8908572614ca29755486ecb184ff7 : SUCCESS in 36m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bdfc43f292f4fc8aae0988ad9dac323 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/109062db6559413fa4bb527c74b99ae7 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc7fc22daba7419cae3b80e1781880ea : SUCCESS in 16m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfa1c291c4ff49ea8c5ed3471a70a781 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfd54f317e204bb48e28731b7eb3ff84 : FAILURE in 27m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04d56cbf7f3f423fb0e1c67dcf06a9d8 : FAILURE in 1h 04m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1733e94a91ae4528a9c1c89cad3eb448 : FAILURE in 1h 09m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ecf779a698074161a390b515757b370c : FAILURE in 52m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/847037b0748047c58ac4f10740906348 : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f1bd4dbe1dde498eb6d178253b34d87f : SUCCESS in 9m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/75c07aee54fb4054a05565c77a7d7771 : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/163e54cc5cb24fc781913d678ab33aa7 : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dad0161a2604d408e309e69121d56ba : SUCCESS in 9m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2733f70417ef45789e63683a98144607 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa4ecc63f46642ceb696f7a4a292daa2 : TIMED_OUT in 2h 00m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/238d3bc8c73e4b9495fc376565ab83e8 : SUCCESS in 34m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99ebaa2d996f448eb61581ec940f3ce3 : SUCCESS in 53m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70b117d0f51a4f3e9fd0391f53c37a45 : TIMED_OUT in 40m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1acf91baa24d4e0eb40d840f2031882b : FAILURE in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f8a3e85b5b344d68e135e0239232a9a : SUCCESS in 8m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34766b34f08d49fa8dfc386a4825cf4b : SUCCESS in 48m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/721b436a6df744ee8b4e2191a31dede2 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":10},{"id":"924389b62717bd929490ca45e352bbd5bf7c9fee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-18 01:45:36.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"db5941a315c25bf8b3382c4fdc8e313ef70a727f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 04:01:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88bb474593174192bc5543f5784678b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cc53c69d1d40a7b122bd8e2ba65426 : SUCCESS in 11m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9a7fcb98775b448c97eaa8b71dc12f20 : SUCCESS in 1h 04m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4dedf47462d243e88ff5f4ca8f3a9abd : SUCCESS in 1h 00m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8be4f775441479cbd666383dcbe85a1 : SUCCESS in 53m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbd6c6fbc9464836b9667fbc82a9ade6 : SUCCESS in 1h 12m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dce4cb06773a4df5add6d6293c1b4328 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1dd0de3252a847b2a38d929563a893d4 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/788806d989d440cd9d04a46ca084c728 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2af78ae4d1f844d39182988db04ac6b5 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ba6dca5b4a84edd84ab35c9d0e01390 : FAILURE in 1h 13m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2173c85a5c429a817d1d4accd455b6 : FAILURE in 24m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c8d2f341eb544aae88b4944437624b5d : FAILURE in 30m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f228112dc26f4b50b4702c14dde7bc30 : FAILURE in 1h 16m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/621bbac010dc42119237aeffc13cb937 : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3781fd01c41949eeb591a2a671729652 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34f8e43a79244a0599cb40919072f2b2 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11465c53d7664205a68940ecf53813a8 : SUCCESS in 8m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63c722cf64bf4e11a3bfa2f714609edf : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4dfd5833edd47cb84b1e8f4a180f0a6 : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f7c4a20a67049ac99f89d1daeeca6ed : TIMED_OUT in 2h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4588db43121146b9a6cd8fd9226008c3 : SUCCESS in 49m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ea4aed749634548af4dfa495f36c47a : SUCCESS in 1h 02m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6955e6dee96941d19308a3231c289213 : TIMED_OUT in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767bb5572b3a45a4a1f661afbb4347c1 : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/341477b5d5174820976c43a60e5b2764 : SUCCESS in 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/999a1e7c1d7a4e7ea61d7a9fdd226916 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":11},{"id":"99c3fdb1a00b582c41f317b0d91d2e670777fa74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 04:31:26.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c49d72605e42d5b5cd9dcb1039a345\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cc6b08abc4f4bb3a61aa62d8721444c : FAILURE in 2h 43m 38s","accounts_in_message":[],"_revision_number":11},{"id":"800285309dd4b888cb3de70150d2895ca41805bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-07 05:01:51.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"cc000cd43252b987825a30a913abfff387d1da65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 05:12:01.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e943f669c942808ad5936e79be8f5c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e6691b5e20148c696a166d75a1271d2 : FAILURE in 7m 43s","accounts_in_message":[],"_revision_number":12},{"id":"bc2d5f6bd61f43d1861e259486f1185c473f65cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 06:36:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63a8947e435f433f9b0498687871f0cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b308b504824631b22241dbe75fdd9b : FAILURE in 6m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/a766347ee506400abaaa96d8512d8bb6 : FAILURE in 48m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a6c8b2981f84bdf9ff8e4f0bfd6dc91 : FAILURE in 51m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ba081f80e774bb3b7042a7f63d64102 : FAILURE in 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2346268956c745f29bb167f404bbb284 : POST_FAILURE in 24m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b84c1631d2094c1abda58ecfbff476f9 : FAILURE in 5m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/523177b46c2646929acf5f73ae674573 : FAILURE in 7m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b4aee80d64241c3ac4452afe4f0609d : FAILURE in 4m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05a69c0d550b42cb872ef02fe6e58d87 : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e96895db76a6436785597c5f631d95aa : FAILURE in 10m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3229a0aa21144059b33370829911461d : FAILURE in 6m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/877500f5678f42b099bb9d863ca350c8 : FAILURE in 6m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/69b3367afe364fc19cd7be467d20ea8b : FAILURE in 6m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6838d5fcff004edc88071e03524f1477 : FAILURE in 2m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9582673f98764260934a0ed36ed9d2a7 : FAILURE in 4m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4e39f3a824df4009b34c1f16f485d7c5 : FAILURE in 6m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05beb0cf6a94c0d82e8a8061417f3f7 : RETRY_LIMIT in 2m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8bd0683359e041319706655e67ae0d1e : RETRY_LIMIT in 3m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b15e16a523e4bbb8795147798c5533f : RETRY_LIMIT in 4m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ff834232de949d28536ca89b5192e42 : RETRY_LIMIT in 2m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc089b8785f2403c8acb7b52d77550c2 : RETRY_LIMIT in 16m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f98bcc3ca9574a4bb7fc13c5e9459d0a : RETRY_LIMIT in 9m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b494677a1cf4d159c6da11769b6a655 : FAILURE in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32743add99e44941955772cd85b7c2e9 : FAILURE in 2m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80cb487c33bc4124950631239364205f : TIMED_OUT in 1h 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9048bcea350348a7879950125a125c5f : FAILURE in 21m 17s\n\nWarning:\n  Comments left for invalid file swift/common/ring/builder.py","accounts_in_message":[],"_revision_number":12},{"id":"3f909cbf9271733e00f247d7abe86df65acdc5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-01 08:21:52.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"8d9cd43991a68fb262455f072d13de8b0214cccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 10:20:15.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2579d92957d74553b725e1c80c0321ab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17dab4bdc5094576a7caac11d32a33ab : FAILURE in 1h 46m 51s","accounts_in_message":[],"_revision_number":13},{"id":"b3737b30c3d22a134568e0e4b1aa95bad13ea5b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 10:33:31.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1929c1fb576f4bdda37821752be2f9d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3fcd1dd3ce47deacf55fc1f16817a0 : FAILURE in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1afa13728dd4004bf0d94e50a546c3f : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b5114e3ba96340e18d0423ae92db9c5a : SUCCESS in 1h 01m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f359db0bc785410c80ff34629382bef0 : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7836c42b02d24b13b9223c3ce9cc6d72 : SUCCESS in 56m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d76a308978bd445ca3dbf7cdda8c5726 : SUCCESS in 59m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc4e08bd0033412583c58cf56f5eb941 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1a8249a38ae478ba8812cdd105c49b9 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fafd9b05a8cf455eaa4b2c9de227b1f3 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1437cebd27a4dbf9f3732588b3a261d : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/927fb64cb76043f2b864aee707fd4deb : FAILURE in 57m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/93f48ec3b96440d1936bee9cb2832796 : FAILURE in 47m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/21050f8b6460453592a7dcca81ce645c : FAILURE in 54m 02s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/072c2d5a2da04f08ae5109106ed15069 : FAILURE in 1h 03m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/de1a3511b2e24a4792955ebaa3390691 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54837581c9e548218678fb943623e64c : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21a56922069241dea74dd39c134bc401 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b73e6168c2c4d12b06b59cf01a4febc : SUCCESS in 6m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92823c14efa74b1c9237583ff0df834f : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a4bbfa1204548bf8019650947dc2ddb : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c15fa3a3b4e2404bbaa59ba1ee6e6422 : TIMED_OUT in 2h 01m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2963f6bb2c641ec9ebd0a873d43ccfa : SUCCESS in 56m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f91e30cfed14a8a85e6944d2cd99cde : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b679d8e3d944599a4083957d969074a : TIMED_OUT in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76e00516911406c840a683787467f32 : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89eb03f2abbc4c849f93c3f549df9744 : SUCCESS in 9m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6eb664d2b404a729104604f178e0f0e : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"8554a2752e3fcac0f55dc7442bedbb4667b14ba7","revisions":{"9b42a895ae7628e189134182d20ca7b942dee8a2":{"kind":"REWORK","_number":1,"created":"2022-03-24 05:48:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/1"}}},"commit":{"parents":[{"commit":"659623616c0f359472eda9d28c2f28e0e5da9e44","subject":"ring: Last part table","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/659623616c0f359472eda9d28c2f28e0e5da9e44"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 05:47:47.000000000","tz":660},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last_primary_node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b42a895ae7628e189134182d20ca7b942dee8a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b42a895ae7628e189134182d20ca7b942dee8a2"}]},"branch":"refs/heads/master"},"7ef7437b14a8fa6fe61862d2dab940ccaf062865":{"kind":"REWORK","_number":2,"created":"2022-03-29 05:14:35.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/2"}}},"commit":{"parents":[{"commit":"067f9e69dffda8ac236d6973084d0fc42705ae47","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/067f9e69dffda8ac236d6973084d0fc42705ae47"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-29 05:10:44.000000000","tz":660},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ef7437b14a8fa6fe61862d2dab940ccaf062865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ef7437b14a8fa6fe61862d2dab940ccaf062865"}]},"branch":"refs/heads/master"},"a5cdc6ec9c7f07c548824ec9dbe057ebf79cf823":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-01 06:08:39.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/3"}}},"commit":{"parents":[{"commit":"984210ff1b9b935d784e97d06dfb8a4136bd7ec8","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/984210ff1b9b935d784e97d06dfb8a4136bd7ec8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-01 06:08:17.000000000","tz":660},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5cdc6ec9c7f07c548824ec9dbe057ebf79cf823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5cdc6ec9c7f07c548824ec9dbe057ebf79cf823"}]},"branch":"refs/heads/master"},"3a5337223d89251c4277c4ddbbaa012bada7ad12":{"kind":"REWORK","_number":4,"created":"2022-04-03 23:37:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/4"}}},"commit":{"parents":[{"commit":"9ebedb11b0d61b25f2fc2f4ecd492dac80f58589","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ebedb11b0d61b25f2fc2f4ecd492dac80f58589"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-03 23:37:13.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5337223d89251c4277c4ddbbaa012bada7ad12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a5337223d89251c4277c4ddbbaa012bada7ad12"}]},"branch":"refs/heads/master"},"de66039130460d90fd60531038782a506aaf181c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-04-19 07:28:25.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/5"}}},"commit":{"parents":[{"commit":"3490b506b565854d3e161dc4b4e3e9c8390eceaa","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3490b506b565854d3e161dc4b4e3e9c8390eceaa"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-19 07:25:56.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de66039130460d90fd60531038782a506aaf181c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de66039130460d90fd60531038782a506aaf181c"}]},"branch":"refs/heads/master"},"b168a662d2918b4c769863ffd4b04ee0a236ad04":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-20 05:45:07.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/6"}}},"commit":{"parents":[{"commit":"dae9cb4aeabae83b10108774058e1082dae16d90","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dae9cb4aeabae83b10108774058e1082dae16d90"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-20 05:24:29.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b168a662d2918b4c769863ffd4b04ee0a236ad04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b168a662d2918b4c769863ffd4b04ee0a236ad04"}]},"branch":"refs/heads/master"},"27e1fd27af08d2db3895cce211b32fa06f5a5325":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-04-28 04:20:33.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/7"}}},"commit":{"parents":[{"commit":"e373bd1bda1a67d3476e79850845240edc7466f3","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e373bd1bda1a67d3476e79850845240edc7466f3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-28 03:12:04.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e1fd27af08d2db3895cce211b32fa06f5a5325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e1fd27af08d2db3895cce211b32fa06f5a5325"}]},"branch":"refs/heads/master"},"07cf957667cf4f58bb3f04060f766fbd9f4476b0":{"kind":"REWORK","_number":8,"created":"2025-09-16 10:05:33.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/8"}}},"commit":{"parents":[{"commit":"059dfc11093ee765d083066a7f6f7f99082127d0","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059dfc11093ee765d083066a7f6f7f99082127d0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-09-16 09:57:04.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07cf957667cf4f58bb3f04060f766fbd9f4476b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/07cf957667cf4f58bb3f04060f766fbd9f4476b0"}]},"branch":"refs/heads/master"},"191343d8f7b1c7567ad99b7a65e6cb6bf27fa7fc":{"kind":"REWORK","_number":9,"created":"2025-09-24 06:39:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/9"}}},"commit":{"parents":[{"commit":"059dfc11093ee765d083066a7f6f7f99082127d0","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059dfc11093ee765d083066a7f6f7f99082127d0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-09-24 06:38:35.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191343d8f7b1c7567ad99b7a65e6cb6bf27fa7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/191343d8f7b1c7567ad99b7a65e6cb6bf27fa7fc"}]},"branch":"refs/heads/master"},"9520016ce3a59a5b660bfeaabbb8e111f64a9a4f":{"kind":"REWORK","_number":10,"created":"2025-10-07 03:46:16.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/10"}}},"commit":{"parents":[{"commit":"9a17bd5318fa1e350674b3be5263af3d6c24d836","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a17bd5318fa1e350674b3be5263af3d6c24d836"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-09-25 02:00:38.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9520016ce3a59a5b660bfeaabbb8e111f64a9a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9520016ce3a59a5b660bfeaabbb8e111f64a9a4f"}]},"branch":"refs/heads/master"},"d9dac85feb05e14b376c42d97ca8d8d5069a8e2b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-02-18 01:45:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/11"}}},"commit":{"parents":[{"commit":"7e7b8b95fc6c1ef456810cb9184037e798a95165","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e7b8b95fc6c1ef456810cb9184037e798a95165"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-18 01:37:58.000000000","tz":660},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9dac85feb05e14b376c42d97ca8d8d5069a8e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9dac85feb05e14b376c42d97ca8d8d5069a8e2b"}]},"branch":"refs/heads/master"},"55c26cdedd2f06682d91510e60e7184b174d7ae9":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-04-07 05:01:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/01/835001/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/12"}}},"commit":{"parents":[{"commit":"90c16321caa315b7e3a51c8b6e69149f703ea455","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c16321caa315b7e3a51c8b6e69149f703ea455"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-04-07 05:01:28.000000000","tz":600},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55c26cdedd2f06682d91510e60e7184b174d7ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55c26cdedd2f06682d91510e60e7184b174d7ae9"}]},"branch":"refs/heads/master"},"8554a2752e3fcac0f55dc7442bedbb4667b14ba7":{"kind":"REWORK","_number":13,"created":"2026-06-01 08:21:52.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/01/835001/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/835001/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/835001/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/835001/13"}}},"commit":{"parents":[{"commit":"fdb55a8ba5b6f798ade5b4e7acc90a29cb94c058","subject":"Move last_part_moves to ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fdb55a8ba5b6f798ade5b4e7acc90a29cb94c058"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-24 00:50:55.000000000","tz":660},"committer":{"name":"Christian Schwede","email":"cschwede@mailbox.org","date":"2026-06-01 08:14:50.000000000","tz":120},"subject":"reconstructor: Use last_part_node when building SYNC jobs","message":"reconstructor: Use last_part_node when building SYNC jobs\n\nNow that we can query the ring for the last_part_node for a parition and\nknow what index it was. We can use this information when building the\nreconstructors SYNC jobs. If the job contains a partner with the same\nindex of the last_part_node _AND_ the last_part_node hasn\u0027t shed it\u0027s\npartition yet, then we log and don\u0027t add the SYNC job to the list of\njobs gathered.\n\nWe do a REPLICATION check on the partition for the last primary node and\nif we get back suffixes then there is still data on there. This way we\ncan make a decision, not per object in ssync, but per partition. If\nthere is newer data on the newer primary, then worst case we\u0027ll SYNC on\nthe next reconstruction cycle.\n\nAdd a new option for the reconstructor `past_node_timeout_hours`. Which\nis how long to look for past primaries.\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: Ie78fa3f3b4403689fa7689d4319157993302176f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8554a2752e3fcac0f55dc7442bedbb4667b14ba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8554a2752e3fcac0f55dc7442bedbb4667b14ba7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
