)]}'
{"id":"openstack%2Fswift~835123","triplet_id":"openstack%2Fswift~master~Id9d1da2017b8d13242ae1f410347febb013e9ce1","project":"openstack/swift","branch":"master","topic":"s3api-make-the-quiet-key-value-case-insensitive","attention_set":{},"removed_from_attention_set":{"32333":{"account":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"last_update":"2022-03-25 05:01:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9d1da2017b8d13242ae1f410347febb013e9ce1","subject":"s3api: Make the \u0027Quiet\u0027 key value case insensitive","status":"MERGED","created":"2022-03-24 21:01:23.000000000","updated":"2022-03-25 05:02:47.000000000","submitted":"2022-03-25 05:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"835123-s3api-make-the-quiet-key-value-case-insensitive","meta_rev_id":"00a7132b18e2fde5fbf07e6c73bcaa92ebd5c116","_number":835123,"virtual_id_number":835123,"owner":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-25 05:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-25 03:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-25 03:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 22:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-25 03:43:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"0ae43124857d887ffc3a135a001ea8d7e78509c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-03-24 21:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4da45a234495ba5c913f0b1f384da34371f3ba9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 22:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc40412153e74c7eb330c5a31a69f2b3 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/40fb48eccb9944409e3e4e673d492c01 : SUCCESS in 53m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31fe1aaa08ef412d97e0b6e369fc9db2 : SUCCESS in 1h 01m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80d75c21cff540faa8b5eb499b6de0ca : SUCCESS in 49m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/da04e30cb7ea4411b3cefee7092cea5a : SUCCESS in 24m 46s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/efe0b51bfe624927affbe261df68f14c : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7ea75dd23d4f4cb385c5c8cba376753b : SUCCESS in 28m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d9ca451584f47f99281ebd403105f5b : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd166cb05dbf4c4195bcc39dd0f77a98 : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b443fb1632d34a39889d88d466d5ec4b : SUCCESS in 21m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a304b62a5ed24fb1b545b69ca94b1157 : SUCCESS in 20m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d78840c5d30244a2be3257d48917a272 : SUCCESS in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ffc96a7d49bc4034a818e72074c18a2c : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27544da2ef7d46269184919131fc3ecb : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c3a11ce0fed479e9e58bf1a7fe0a332 : SUCCESS in 25m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7a6839a687b46fe98e5711ab23f38cc : SUCCESS in 23m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03f28aeb36e4490286e09598c6a06c7c : SUCCESS in 20m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/224fdbd887864ccc9b1b99242141899c : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f9752197e9945d6b880f82a1fca3c12 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9edfc4b9e9924127859e169caaf0325d : SUCCESS in 22m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6bc73d51061446368e8cd1b4ee899a10 : SUCCESS in 57m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d37f57e5e8fd4729bce9facabbb8d768 : SUCCESS in 1h 05m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e615f9f895084f5a937a97414a3bb9e9 : SUCCESS in 44m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc224fc2a36b4cee8a8033bd3ade4a09 : SUCCESS in 42m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00b2bae604cc4b69bd8f1ba7f8526c6e : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16241630d7194e87a8e1f92990c51789 : SUCCESS in 7m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8abdb98127eb432c86f34844e1aec8b2 : SUCCESS in 29m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c800f33073884d6fbf05888751a070a2 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":1},{"id":"340e4fe847ceb569442a8f3a5cf461a35d495bb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 00:14:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e3c5ebf3ec844d199d20cada0a25a5e : SUCCESS in 48m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a3798b8404841e4bdcb4501ae2d1a29 : SUCCESS in 37m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/db390395611e4b5fb539328849b91c60 : FAILURE in 2h 41m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a287f9e9123463eaad043d5ea1f9f93 : SUCCESS in 44m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0a4faf32bf74542a730c858566b84af : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2b27c8ace3c915f09850f675fc6119f9cfe5ab4b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-25 03:43:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bb1eac9c3dcb26167fab848ca744f2ebf9c8f31b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 03:44:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a880d0e623564d1839d14017ede8d01d4397e819","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 05:01:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ffbb8d157a45faba8fd7989b64f6db : SUCCESS in 11m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d77b90dcd27443c0835b531fae11589f : SUCCESS in 55m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ec6ad581102479d93f879732e1986cd : SUCCESS in 56m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6494360cc4844ec3815dbe1fe7b5c253 : SUCCESS in 1h 01m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a0551ae76bdf4ead9afec6ba3a11ecca : SUCCESS in 10m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/78e454180a604073a481c2b7cbb2714a : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/468ccbefd8234a7a896c94ee96439053 : SUCCESS in 24m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b57d07cc91f949fabcbf78a64f569f94 : SUCCESS in 24m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8709bc02dbb4ff5934a6e5fc725edb8 : SUCCESS in 23m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d0aab25f1d545aba4b6e34c977b64b2 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c56a7ec9ad364449a556f26c5f6ddba8 : SUCCESS in 26m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e52e76dbffc343a89f7b2fee27408926 : SUCCESS in 27m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e12dd126e7be475c800b2f95dd5f6afa : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54b039aa1a6a4928a83b44715cd97a7b : SUCCESS in 25m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d491bfbe413942acbcaa3242b20e0c9a : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/36dd04e54cf44216b00be41633b76f6e : SUCCESS in 12m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a3364dfe378f45b1bb9718561a55d249 : SUCCESS in 45m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7f798215ccd4bc5a2d31329056d4d2e : SUCCESS in 1h 12m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4958e44da0e044eb815b2b275d85132e : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80008d37740b43de9a6f2105f8696011 : SUCCESS in 57m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc68094bbd54ac78ef1887d2b839b87 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db299fe549046f8b2e3fdb1da5d6358 : SUCCESS in 8m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02709f9b671f423886e2e66de8015b65 : SUCCESS in 22m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66d1dd9bf20440739a4ed5a2ec98c667 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ae7fb75d2047df7e104ff4ac4f9880fa0428340f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 05:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"00a7132b18e2fde5fbf07e6c73bcaa92ebd5c116","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 05:02:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a6f6955ed84153beed9e4e5be9d6f3 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a417230320dc4a83b206d8eb806eac88 : SUCCESS in 57s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/89776e85c1db47378c42a7f102905e60 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd2dd11562be842db518b751a505c7e8f915ca5f","revisions":{"fd2dd11562be842db518b751a505c7e8f915ca5f":{"kind":"REWORK","_number":1,"created":"2022-03-24 21:01:23.000000000","uploader":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"ref":"refs/changes/23/835123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/835123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/835123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/835123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/835123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/835123/1"}}},"commit":{"parents":[{"commit":"0651d8175d627841051190d76b8fdf5f9c48b4c0","subject":"Merge \"trivial: Replace assertRegexpMatches with assertRegex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0651d8175d627841051190d76b8fdf5f9c48b4c0"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-03-23 16:48:51.000000000","tz":60},"committer":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-03-24 19:22:21.000000000","tz":60},"subject":"s3api: Make the \u0027Quiet\u0027 key value case insensitive","message":"s3api: Make the \u0027Quiet\u0027 key value case insensitive\n\nWhen deleting multiple objects, S3 allows to enable\na quiet mode with the \u0027Quiet\u0027 key.\n\nAt AWS S3, the value of this key is case-insensitive.\n- Quiet mode is enabled if the value is \u0027true\u0027\n  (regardless of case).\n- Otherwise, in all other cases (even a non-boolean value),\n  this mode will be disabled.\n\nAlso, some tools (like Minio\u0027s python API) send the value \u0027True\u0027\n(and not \u0027true\u0027).\n\nChange-Id: Id9d1da2017b8d13242ae1f410347febb013e9ce1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd2dd11562be842db518b751a505c7e8f915ca5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd2dd11562be842db518b751a505c7e8f915ca5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
