)]}'
{"id":"openstack%2Fswift~835602","triplet_id":"openstack%2Fswift~master~I729e3ba5f19160aa09dd8eb983d9594cb102e85b","project":"openstack/swift","branch":"master","topic":"quarantine_enodata","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-04-05 21:16:08.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2022-03-30 21:39:52.000000000","reason":"Pete Zaitcev replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-04 05:37:52.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I729e3ba5f19160aa09dd8eb983d9594cb102e85b","subject":"Obj Auditor: Quarantine ENODATA","status":"MERGED","created":"2022-03-29 09:53:50.000000000","updated":"2022-04-05 22:39:16.000000000","submitted":"2022-04-05 21:16:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835602-quarantine_enodata","meta_rev_id":"a29cdb065c1d08a0f32d1bf9da1c33b65c570a1b","_number":835602,"virtual_id_number":835602,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 21:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2022-03-30 21:39:52.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-04 05:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-04 05:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 10:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-30 00:16:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-03-30 18:16:25.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2022-03-30 21:39:52.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"a9e23fbeaaa79f988006d06379952062995eddc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-29 09:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff798d4f15da72a8b9d54874677e51df535f0603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 10:57:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13416a789d14651a420e4ba82f903a5 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ac7e65c8ed0543789e6fe250ffc32752 : SUCCESS in 51m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbae37a38e314e8b9716454e2e9fbf40 : SUCCESS in 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5d7d011029b430f93b2583e581aea23 : SUCCESS in 48m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3911e9732f2445e7bf0ec1cc7726d89f : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/71aaf08650c2428a8ad812c4fec4c0cf : SUCCESS in 24m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3fc9fcf15e1341efa9b35008fc3056ee : SUCCESS in 27m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/507ca2380ec541ebbfb2e25369172ad2 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0434f75fa56a461f97943b8821454514 : SUCCESS in 7m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/401d5f2d527e4476adbb76863a755beb : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e53c87b0293842858e556c3b429eb283 : SUCCESS in 22m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3016aed7c6a49b69ecf28f9c1b681ed : SUCCESS in 20m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75b6f0945d4b4d6aa112714d66598a8b : SUCCESS in 22m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08f411e4c7774f1f812e555351b8035e : SUCCESS in 16m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83ccbeb12014454fa620acc5dcfc34b2 : SUCCESS in 24m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e31501a180d4e3c971b1f2c7f6713f2 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0a69ccedd4734ff3b8361ef2fa0f1a78 : SUCCESS in 20m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/29de95fa04e040558a8c49852cc02ec8 : SUCCESS in 23m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51e3beaa5914437da6a94bd137b17109 : SUCCESS in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88001a88bd714adcbe7958929a41c31a : SUCCESS in 24m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7ac753005474f9ea820dcf1f07ca049 : SUCCESS in 54m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f3f5791cc9f46609d35016991de0013 : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66f7f65430154aeeac823e6f2e8109f3 : SUCCESS in 45m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2fbd0e21ee14bd2a5caf236c5160faf : SUCCESS in 42m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/425c146f3f1f4d8da78d7dc3cd548fd4 : SUCCESS in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b15ca56f4aa041f9adc31ad164ce0bcd : SUCCESS in 7m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0f3d11952f947259385e830e4e5c6f5 : SUCCESS in 30m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8517f0abd3b846238c3b1b36c1f42cea : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d2b9dda04864c54d2e8ae8194b75e88788a18b48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 13:08:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73d25554809c4ef89e49293c98fc0711 : SUCCESS in 55m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5f5f00c3208423d8c6462f376de42ac : SUCCESS in 52m 42s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c25df8e4b3fd4fe08fc5fdb41e01ce12 : TIMED_OUT in 3h 02m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f60e3114adf64627a924cb43b2688a90 : SUCCESS in 55m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/18b94e320c304fd89e56ed39ce0c394f : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0a1d9822b6219e6818bde68736619da2d9ffc42b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-30 00:16:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cf53f4b6d46d28b0a77605dc1143887c7f38b2e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-30 18:16:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ad51112ed54a0c6a4002cfa412cc3d110650037","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-30 21:14:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"548a905afaaef947c86447070f9d291b0ef0ac39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-30 21:17:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0cacbf1baafbe35608ab3a852d311b91220880f1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-03-30 21:39:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bfb07db0ed758294df5712fef99edd67231481fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 22:22:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051278b7d3eb479ea4e305e691289747 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/36da044b73b7495d824347365cbdccd6 : SUCCESS in 51m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7227b8f4cbab43bd81c0b59ada7a7717 : SUCCESS in 38m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43b70a0080994f028cea63c851cf8e43 : SUCCESS in 46m 12s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1966747be181403e96a90a6c2926e439 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c91c53f9ac804faeb5116ae943e6c6db : SUCCESS in 26m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dfb3d8f8f4d74721bfe6c19c503b6078 : SUCCESS in 25m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69917d7ad8c04ae1ba12fa38c2f8dd3b : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/954db607fcd64fe7b7bda2ebdcd86faa : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bcfdef57fcb449019ce655ac75a0270b : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc06ce8f8a2b4a4ca17b6ec3504703d3 : SUCCESS in 20m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3fd35c3f0dbc46a3a482ea34c72c23be : SUCCESS in 20m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e8ff71fb9864369b6499d1afb5ebbdf : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6b41fa4e6ed41dea2e41fe2de589623 : SUCCESS in 23m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/712bd513050e4335ba2004fb7c78e129 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52bde7e323ee46a4a3f7c234df7b456c : SUCCESS in 21m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68be90a5bc83495ea53d99ac8dfe8d28 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac4aa610d70a4018ac46330c901791a9 : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff14dc0555f34baea30bc9c30a6dc42c : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61c43f0e7da8490a9e5a0d981411fd0b : SUCCESS in 23m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/caf2968ae11842869e78b8d787f8d77c : SUCCESS in 1h 00m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd2d2b812e7644f29b201a86068d7718 : SUCCESS in 57m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ed19369126d491b8e644c4768402e2c : SUCCESS in 51m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a88cb307e87043b6a0c7b053c7d2a4f9 : SUCCESS in 37m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/933d9780b4124a8ca655a6db4e90fc29 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc623b620f1b4ca4a3c690ef08b76768 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c70dfdf6057640afa9876b2928d6bd75 : SUCCESS in 24m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17379b7b7ed847eeb466fa902dbabd31 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":2},{"id":"fb59d7bfd920f45d6eda54797b0c2bc4089aba59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 00:07:25.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b4e0278c5444e74ae8b0506c6df5dc5 : SUCCESS in 59m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9664243463b148cfaf337a284e48143a : SUCCESS in 49m 53s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/dc6b6efa822b446a85f2aad61728f35f : FAILURE in 2h 48m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1830828546a44e0adbd6c560053f76c : SUCCESS in 52m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b6c6f90314143028b8b473cab88fee7 : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8df3583c1723b43532eb94f9fd0ee8e59ce7dd72","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-04 05:37:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"32523661597cef608f6937b88166a07ed18d577f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 05:38:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d52e609c7edee471ba8b6a3e95c11e1ba8ef73dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 06:51:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61a9477877840a4b9508b6784e8e04f : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/363ea7ab4458488baa72e45cfd8cc4fc : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e75d8187b634a2b9097dcac83fc963e : SUCCESS in 44m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e400714095f490aa8019401cff48664 : SUCCESS in 46m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d60e93daa9094a329c7555ea5e13a96c : SUCCESS in 7m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dd08c3e8a5094aee9a2a97bfbf1fb4f2 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/44e8361c5e2e4b4f9416b93be823d79b : SUCCESS in 20m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/465d1354806f470eaa828b7444334046 : SUCCESS in 21m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0be8233c139b4dc3941361ca16d46e44 : SUCCESS in 20m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e86c76cf9404ba29b9318cbca0b0f47 : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9565bd3f809247aeb3a26925a8433ad7 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e55291c60204dc2b410044f47bc283a : SUCCESS in 18m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/35da118fca3f410d92e53c569d07eefc : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de4ef969dac84613a42cb7ec4b376277 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/81e938aa42664aa59682111609b88a8e : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f9a860bce5549fba4e6a0f65e2b3d67 : SUCCESS in 6m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7f36e8a9be04a8790557be83a59ea23 : SUCCESS in 51m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9dc1a08e40141098554ed3ef8f2429c : SUCCESS in 1h 02m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5871ed1e56bc4f1d9de914dae06d29f7 : SUCCESS in 39m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de9e49474ec04e20a30a7ad207ad939c : SUCCESS in 40m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9354ab51596244d6b0b9fb05cbee8ac3 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81321b4d41474dfbaa5ef3a6b6ccbe02 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b2b1141052c041a7aa13a61f709a01cb : FAILURE in 7m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec764383e0d747bb98b53d5c4a21b130 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4c94790136276df6846802a5084b199b0f915c3d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-05 18:48:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39dd4d20ad3de58ec106379d89450467f0a73b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:04:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b98f244d3f4bf1b4ba6ae3d347fcde : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f35807698c634804ae52b7c549c6fc06 : SUCCESS in 51m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f80a66a43504f4c94eaf3c001be341a : SUCCESS in 48m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cfc75433fcd42abbad3dd90e38184c6 : SUCCESS in 47m 39s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/362dd26fd8f14b82b38762db4bcb7728 : SUCCESS in 23m 22s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ccae31d7ac1f4a6b8151c55dae6ee4f2 : SUCCESS in 23m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/da06d9b586924175a33a4ad3bb1bc0f1 : SUCCESS in 21m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af7a60a9b2a74000a702e253e637c1c7 : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97a9e4af8bd04b13b87596c24f6ced6b : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4337672d50d44b3a9afa96b5a7f488f0 : SUCCESS in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13a50943d8d44e5aa403c79f8573c66d : SUCCESS in 22m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1926b30e47401089b218f8129d2ab1 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18ac50cfde1843f9a01cb7120e3b4c87 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/14433d0f868045b1a574b60a79e054d4 : SUCCESS in 21m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84bac44d8f0841df9e9242b042a31206 : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/39c140df88844ff98b5635e45c6f5397 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c6012bb979614412817874ebaf697c12 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/816e0387ee644956a41012111db921eb : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/af8022b24e9f451f8c3b8dce43f897e2 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4347d336c13480cb9f4b2fb19d0733d : SUCCESS in 16m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4da7d5edc7844033b32c1c65abc9a41c : SUCCESS in 52m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdbf32acc1c047bd9270791f9f84dbdf : SUCCESS in 1h 09m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db1cd4ba8190419e8f0751bba9ed0d1a : SUCCESS in 46m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/66f6883bb21c4f70bd6a7f2c5eb82fc1 : SUCCESS in 40m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7877386b88894ad4a495bb227f28fbde : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8c80e8badc43708a2a83ebfca1fbae : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b307b7124cc4d2984fe6f608bca7bb4 : SUCCESS in 28m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1cea1265e354445bc14a6561b0d8b6d : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2da5cdecf6dff3498a5d58483d904bdb8717a0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 20:05:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe813bccad160cbd5682646db88428c1b94fce96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 21:16:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6b338b5db44eb1bf3bfd00272ca436 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/e9b80e704354481988d40b9ac049c2fa : SUCCESS in 1h 01m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8589bfd20b704fdd8ac2e2a61a5db072 : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e2b80aaaad6459e939ca9d1316d6ba9 : SUCCESS in 54m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d6e7c2b6a585443f93b382499147e2c5 : SUCCESS in 7m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/33fe8b8445114caaa8c05c8fcec1dd07 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/534b284aa67e453f8def959fe3a97d87 : SUCCESS in 17m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/862ecfa1d552421aa83e52e6f06862c9 : SUCCESS in 13m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb66ef052054f85a7814cf2dc40d780 : SUCCESS in 12m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dca07cd9d0804039a724d71e91dc7b0f : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2e72a6878324854b147148000a76443 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad5608287008488da646a56a419584c2 : SUCCESS in 15m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5481269f71d54168a6472f125702912f : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1dcae020ec3e43b786babe0be3bb6dcb : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4192125efb34c5a8b9b09c1be494590 : SUCCESS in 13m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8d8930ccd2e4cd48b2432c66e1823ea : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ee833d6260d47c48417cf62e6d23445 : SUCCESS in 55m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2134617edf349f29240b9d1c19839d0 : SUCCESS in 1h 05m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39e06b9d63784c9ab084eca08ef22473 : SUCCESS in 44m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/622f786847964c5190919295a1b9ae82 : SUCCESS in 27m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e573355992564f6c9a8cf665be7d8f61 : SUCCESS in 16m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57683ff12f56472782e55cbf18f8ef10 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3aff66c0ad14bba8d17a98084711b9d : SUCCESS in 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5267a25ebea42febaad3609dec6d0e0 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"969c89d3c2535c135b4a1cc4117a5e3ea6cc65b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 21:16:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9f328b64b7bd5127e95d3b148924a841ae519d3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 21:17:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c42e622bdb9c453380801384396c3650 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8f8257f96ffd4051be15409383223703 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c0e52ff5bc994c169d19e623b3615ea8 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a29cdb065c1d08a0f32d1bf9da1c33b65c570a1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 22:39:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cefd30c8685e42efbcc8648a4dee442b : SUCCESS in 57m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fdf611a1e34946938f8a77de0c2f09e6 : SUCCESS in 38m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/89d225037b1144adb6c913ea14bcc3a4 : TIMED_OUT in 3h 01m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/726f8ffcc4d74207a307db13eb991e88 : SUCCESS in 58m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9d37a3dba804dd7b5f18fc400b71cc5 : SUCCESS in 54m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f92be1bdd011388b2cf763c9ff62b330119362c3","revisions":{"2ea222c3648a5a88ae221d9037c94216351443f3":{"kind":"REWORK","_number":1,"created":"2022-03-29 09:53:50.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/02/835602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/835602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/835602/1"}}},"commit":{"parents":[{"commit":"ec964b23bb0025737c12ba2addafc94b69d671eb","subject":"Merge \"s3api: Copy more headers from MPU marker to final object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec964b23bb0025737c12ba2addafc94b69d671eb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-29 09:47:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-29 09:53:29.000000000","tz":660},"subject":"Obj Auditor: Quarantine ENODATA","message":"Obj Auditor: Quarantine ENODATA\n\nWe\u0027ve seen a disk corruption take place that throws -ENODATA when a diskfile\nit opened. We currently don\u0027t quarantine on this IOError.\n\nThis patch catches and quarantines the datafile at the place we\u0027ve seen it\nhappen in production.\n\nChange-Id: I729e3ba5f19160aa09dd8eb983d9594cb102e85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ea222c3648a5a88ae221d9037c94216351443f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ea222c3648a5a88ae221d9037c94216351443f3"}]},"branch":"refs/heads/master"},"f92be1bdd011388b2cf763c9ff62b330119362c3":{"kind":"REWORK","_number":2,"created":"2022-03-30 21:17:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/02/835602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/835602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/835602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/835602/2"}}},"commit":{"parents":[{"commit":"ec964b23bb0025737c12ba2addafc94b69d671eb","subject":"Merge \"s3api: Copy more headers from MPU marker to final object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec964b23bb0025737c12ba2addafc94b69d671eb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-29 09:47:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-30 21:14:52.000000000","tz":660},"subject":"Obj Auditor: Quarantine ENODATA","message":"Obj Auditor: Quarantine ENODATA\n\nWe\u0027ve seen a disk corruption take place that throws -ENODATA when a diskfile\nit opened. We currently don\u0027t quarantine on this IOError.\n\nThis patch catches and quarantines the datafile. The catch happens in\ndiskfile.open in order to group similar quarantines in the same place\nfor easier maintenance.\n\nChange-Id: I729e3ba5f19160aa09dd8eb983d9594cb102e85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f92be1bdd011388b2cf763c9ff62b330119362c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f92be1bdd011388b2cf763c9ff62b330119362c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
