)]}'
{"id":"openstack%2Fswift~836408","triplet_id":"openstack%2Fswift~master~I95b262739b395129198a05939c2dd6045dab34f3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-04 23:45:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I95b262739b395129198a05939c2dd6045dab34f3","subject":"CI: pin smmap version for py27","status":"ABANDONED","created":"2022-04-04 21:16:44.000000000","updated":"2022-04-04 23:45:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06347a8f694454787804174a048c8b3145cefd85","_number":836408,"virtual_id_number":836408,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-04 22:32:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 22:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc9e41e23a70f7a1a4acd34473735c3bf430632d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-04 21:16:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d767b9a1aee1d62e41e0afa9c111559e97aedfc5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-04 21:38:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48be8894bba26fe092cdf64508198a1fa04b5ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 22:32:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/158aa3e0b9fa4b00957aac2ccc9e78ae : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/ed00064722da4480a16daea5047a9238 : SUCCESS in 1h 00m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad0b0a52a44c42469f4b89d1cf937965 : SUCCESS in 59m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88fe0aad4f01445e9c18b5bf4c6002c5 : SUCCESS in 1h 03m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/025321d3a3c44921a72215fa3a45e142 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0372b26b25f944debdabe5a4b1d41cb3 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94f59d360dcc4fb29fdec595a85f9fdf : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f77a5f9cc05477cba8655a9bbd24e69 : SUCCESS in 6m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2cf84018a8b247e1a5022cef6b4fc804 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e414fc2311b1400fb0102ce5169238f3 : SUCCESS in 17m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc7b2da12485412a81767b7a2521cc0e : SUCCESS in 15m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/12780d89b4f84325986de6548a64b0be : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9236378c02124cfb99bff7a9ca8d9d8b : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ab158ec872084264b21a050f7a5836e7 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74cb130318a74ae2b575d27f5097e67c : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/11e9fe6a77464772959d6ba8af910cfa : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d075cde64e04b9bbb0f9257ae359426 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9bd96d22a24045f98a23048752d1bbef : SUCCESS in 18m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac832db3a2284710ae6f5de0e1273a00 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44e325a1f0ec4ac4be7e43cb72d55a44 : SUCCESS in 21m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/274674d6e73f41aaa360bb388202e12e : SUCCESS in 45m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/27aeb39911944df6a047d5357e1781d5 : SUCCESS in 1h 10m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02b5076db6994af68b00b746d452adaf : SUCCESS in 33m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27361f1899624ba6b39f653fd7ea7823 : SUCCESS in 33m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4616fa5793474c23a438e267a4d136a3 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb285401884f4d6a87e9bae0f51a6f87 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ed3f1d9e80741efaddc6b8f4f5a7d06 : FAILURE in 7m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c12d523f55d74d09ad7296904274de7f : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"674868df2499af608f33093d56f390fd1d31eab0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 23:11:20.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b39c903c2a040c693faa94672222032 : SUCCESS in 15m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/856f26ebb401412d9a1a7e6b5116f600 : TIMED_OUT in 1h 31m 18s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c79275852e5f499cbde0e360e7a03e91 : SUCCESS in 27m 00s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/3acbb34662244b25afdb04c8dbc46a9c : SUCCESS in 28m 18s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/3ccfb88249134924b4720f5a5a936b64 : SUCCESS in 23m 42s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8cbe2a7010344ecb9f129b24a7a663a6 : SUCCESS in 28m 49s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe8bc1279cbb4f3585ac9ae4dac23a31 : SUCCESS in 23m 02s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/e01e6f6e86f04a569d48b6dd0e454a96 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c3da4c63cc94214a1e9eefddb80f5a5 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0ff084a7eda49fdaf32c5b054c65447 : SUCCESS in 22m 20s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/6d86fcb86a2a4d96af3c400f2ec19dde : FAILURE in 7m 34s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/c3c0ac3af9734e87ac4cd27c9517d08d : FAILURE in 7m 51s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/f4d0b2eb37f540a7ab35c3ba10898a49 : FAILURE in 7m 27s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/151871c5e7c241fca3c380614b798024 : FAILURE in 7m 44s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/06badbc01b924216a5d9f11c8686ccac : FAILURE in 8m 07s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/826a9762088e4d4fbb86fa6c72a3bad3 : FAILURE in 8m 10s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/ff54c7d13e7842768b9c78079d80ba6e : FAILURE in 7m 51s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a0786dfacb3941fd86fdb99dbf5cc91f : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"06347a8f694454787804174a048c8b3145cefd85","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-04 23:45:41.000000000","message":"Abandoned\n\nNope, didn\u0027t help -- moving to py3 in https://review.opendev.org/c/openstack/swift/+/836269 should, though.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce30dba8cd30394076efce4ddfe898c0bcd54946","revisions":{"ce30dba8cd30394076efce4ddfe898c0bcd54946":{"kind":"REWORK","_number":1,"created":"2022-04-04 21:16:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/08/836408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/836408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/836408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/836408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/836408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/836408/1"}}},"commit":{"parents":[{"commit":"ec964b23bb0025737c12ba2addafc94b69d671eb","subject":"Merge \"s3api: Copy more headers from MPU marker to final object\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec964b23bb0025737c12ba2addafc94b69d671eb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-04 21:16:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-04 21:16:35.000000000","tz":-420},"subject":"CI: pin smmap version for py27","message":"CI: pin smmap version for py27\n\nHopefully this will fix the rolling-upgrade job?\n\nChange-Id: I95b262739b395129198a05939c2dd6045dab34f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce30dba8cd30394076efce4ddfe898c0bcd54946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce30dba8cd30394076efce4ddfe898c0bcd54946"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
