)]}'
{"id":"openstack%2Fswift~836411","triplet_id":"openstack%2Fswift~master~I37a58c50c0ca86e4250bb2471b28317f3550c26c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I37a58c50c0ca86e4250bb2471b28317f3550c26c","subject":"Update ceph/s3-tests known failures","status":"ABANDONED","created":"2022-04-04 23:20:11.000000000","updated":"2022-04-05 18:09:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e656b5a17da7abb3a3fe9f1861b9d66ddc10dbb3","_number":836411,"virtual_id_number":836411,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-04 23:42:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 23:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47b1bb26c6e493733346a7a60f6f9b91343e3792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-04 23:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c55f764d94525acbe369bc190dd5a9228d0f9ca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 23:42:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daa3992d882c489e97339decd54cf588 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a109e254536b49d4b92581985a63b558 : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea0d271fb76ab13dc59425a85161bc99cdef57af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 02:07:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4743931c349249d78baff593cc7964e4 : FAILURE in 2h 44m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e656b5a17da7abb3a3fe9f1861b9d66ddc10dbb3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-05 18:09:12.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15a0eff7781ffbef6c15d653568e65d50b2ac412","revisions":{"15a0eff7781ffbef6c15d653568e65d50b2ac412":{"kind":"REWORK","_number":1,"created":"2022-04-04 23:20:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/836411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/836411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/836411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/836411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/836411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/836411/1"}}},"commit":{"parents":[{"commit":"9a4ec435e29d3be4cb64cccbf5c2fa4a79f5a4a7","subject":"CI: Run ceph and rolling upgrade tests under py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a4ec435e29d3be4cb64cccbf5c2fa4a79f5a4a7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-04 23:06:28.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-04 23:15:23.000000000","tz":-420},"subject":"Update ceph/s3-tests known failures","message":"Update ceph/s3-tests known failures\n\nI up-rev\u0027ed the test suite, so there was a decent bit of churn.\n\nChange-Id: I37a58c50c0ca86e4250bb2471b28317f3550c26c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15a0eff7781ffbef6c15d653568e65d50b2ac412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15a0eff7781ffbef6c15d653568e65d50b2ac412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
