)]}'
{"id":"openstack%2Fswift~836755","triplet_id":"openstack%2Fswift~master~I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0","project":"openstack/swift","branch":"master","topic":"aws-chunked","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-05-14 04:13:55.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"34643":{"account":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"last_update":"2025-05-14 05:23:15.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-14 05:23:15.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-05-14 05:23:15.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-05-14 05:23:15.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-14 05:23:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0","subject":"Add support of Sigv4-streaming","status":"MERGED","created":"2022-04-06 09:20:01.000000000","updated":"2025-05-14 05:24:05.000000000","submitted":"2025-05-14 05:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":324,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"836755-aws-chunked","meta_rev_id":"c68105a6b967c465049b7179c9f5f2d405423984","_number":836755,"virtual_id_number":836755,"owner":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 05:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-05-14 04:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-05-13 19:37:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-05-14 04:13:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21482,"name":"kirubakaran kaliannan","email":"kirubak@zadarastorage.com","username":"kirubak"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 10:29:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 16:34:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-11-10 20:42:20.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2024-02-01 10:30:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-02-15 05:49:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-08-08 12:09:31.000000000","updated_by":{"_account_id":21482,"name":"kirubakaran kaliannan","email":"kirubak@zadarastorage.com","username":"kirubak"},"reviewer":{"_account_id":21482,"name":"kirubakaran kaliannan","email":"kirubak@zadarastorage.com","username":"kirubak"},"state":"CC"},{"updated":"2025-01-21 23:31:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-01-28 16:22:50.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-01-30 20:00:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c39815cf31c804acdde1b2459753c80382bd4cc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-06 09:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6483e4a35572912f4f18d27dd2351089f9297aa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:29:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1291a3892ed547d3afe2b28d87aa9b3f : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/70a3a247b9094e1d8db4bb2f9501805c : SUCCESS in 49m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edf99ddf532d4c00853edde1ed6302fc : SUCCESS in 42m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/311fcb66e7744b32b42269ec850e9270 : SUCCESS in 41m 29s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/01936bfd099b490ab6c53d2ea5d6c6f8 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/899115cf532b45d3a77f7e829eecd062 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cdaa3521bdaa484694fd7082d99e1f3e : SUCCESS in 20m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d74a75e2122a471fa0f2fa975df03b76 : SUCCESS in 6m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89a9d6905d55406aac055eb52662a0fb : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/de9ed7715a69483abeda95a5813935c9 : FAILURE in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32d71a16d8264202bfaad419c50f3762 : FAILURE in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60e79d4cc967434faa32fb95433d37ff : FAILURE in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae5fade1f833451fbc5e6d9292ce7c63 : FAILURE in 4m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94a95385802f41089c8f4dbe7a845169 : FAILURE in 4m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c9febec0e624cb78fc24264b6a6e219 : FAILURE in 4m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c509903e4e04c5d98c12abf4893282e : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/efb13242b85249feb68c875b10e27744 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/741edfdcba6749df8010c13ee0266974 : SUCCESS in 13m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d4800c7a3a54f4c84a78ac5708af179 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7859506b1e7445b0b1f319fe0f31b768 : FAILURE in 15m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8784d85d81b84ec89c9b7d460b9ab65f : SUCCESS in 47m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f13ab7a5722143189544bc14099f3e34 : SUCCESS in 1h 02m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86e96af2ed0e4c60a7dad696284299cc : SUCCESS in 32m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbf079f082fe4d77a31ef9df5651829e : SUCCESS in 28m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8f5b37083b945269c0cfe2ae58274e1 : FAILURE in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd692b0e91ff42e481158261c5c41aa1 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60e81c600e5d4c05976ce7b4d98e6477 : SUCCESS in 22m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25394005bfe84da58a1fd9bd6393f791 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":1},{"id":"23ed4c78a8e31d392240c53d34e582781cb7fe76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 12:41:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c883ddf62e0146cc985b4e1ba2f3760b : FAILURE in 50m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/720048fdb7c04982ad3f72582089cd42 : FAILURE in 38m 33s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/76e0b41a64cf48d2978552b96c20545d : FAILURE in 2h 45m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/52cd376f0cee40a98f35e6eca0061919 : SUCCESS in 47m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c92a5000c5e430c976f368d5cf7fb7a : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7513e52eec94af533ab11c1a6c04db1a73c4c694","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-06 13:17:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b0efaef1437b743c5a4fe8c07d3da1d1c899627e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-06 13:40:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea7461ac1c02134c942ae2f17887cbb09530c5b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 14:50:12.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34e42b4560847da831615caf78668e2 : SUCCESS in 11m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/4b9f592285764bbebbb009766717a1c4 : SUCCESS in 1h 07m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9490d138fff406eac4a291df2d55b99 : SUCCESS in 57m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0f46a6251794cd09b529c39062d9f4e : SUCCESS in 1h 02m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bd3e3f8358bf44208c6010a78597a459 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/190bc87417e74fa0bd633a1ba325568f : SUCCESS in 49m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3e57b130e122465f9bb861247454096b : SUCCESS in 22m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69a18f3031044737a9afbcc5fa78b814 : SUCCESS in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/582a03b37e4b4c9db0809f66891616c8 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/781586417a114e588506e56fa4092c6a : SUCCESS in 17m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9e68b85b60594fb3889c3b11f9c068f5 : SUCCESS in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/280ff220e7df4143ad57a0e9b2bcc899 : SUCCESS in 19m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b551a8537ede488c903d84e38d0d8558 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b1c892a252444c2b2b6bbabd85c0d6e : SUCCESS in 14m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7698a0520841443a8f59234dfa172d2e : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f8f99031803541f99cd4df2c7abd450c : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/853ac59de86e45fe8fd4dee734c99231 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6dd6e308d3694141ad5b2a51b0fa8435 : SUCCESS in 21m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77f5031355cd47b983508b9195291b7d : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00133a403fd14c6a88032f530c1fd84b : FAILURE in 21m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/19eb39947b0f4eeeabfe89f5742ae664 : SUCCESS in 55m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/294cc5ef250643d7815141f8861038c7 : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d525026fae3c403c830b15c0d3ea9098 : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf95e2150b4345e19d475fb5aa32091c : SUCCESS in 49m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be1ea413dabd4ec6b1d3f6a919bc703c : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8add84b0af43f8888b760862d59b52 : FAILURE in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb86c8f4c3c24ba6ae401a57c975ed35 : SUCCESS in 25m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af20aa4be6454694863fc601a443eedc : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e07fa283958b8ffcb8d3de5f84ec907738075a9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 17:04:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeee9a88d32d450eb38d5951eb3174ce : SUCCESS in 57m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28dfa22ffc66461ca095efabd9a3a3c6 : SUCCESS in 41m 15s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/468c768797b441508dcd5af6c492d8b9 : FAILURE in 2h 45m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb88e55600cd4cb38a3ef770388cc801 : SUCCESS in 45m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21cf0792dfbb47758461bdb231ac8633 : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":3},{"id":"f19857563a73c9c2b66cdc996e445a6af78474ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-13 10:17:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f236bf89a81d7b69b2472bbeb04d6efaffb4ea7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 11:15:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39301638519463ca9f53851d5d765fd : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a0ba637ddd3145fb9297557f24f4451e : FAILURE in 10m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e917177e061f4ff5beda075ce68006c1 : FAILURE in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7044886e5f8f49c199a625774a7e0149 : POST_FAILURE in 10m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b31b083e7f7e4686b3b29ab9e1ca4d20 : SUCCESS in 23m 43s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a68c6a6b438649dbaa351af683d012f5 : SUCCESS in 23m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8dbc2029f273470db4a6443bf5cd4b9e : SUCCESS in 23m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d7f26f266854656b6f96d1e7ef28c65 : SUCCESS in 7m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a2fadfb70674c249000e9480fe138d0 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a88b9668b3bd4386ad134b5985dcc23c : FAILURE in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/588abea509664b4aa733eb6f408a8302 : FAILURE in 36m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2cedcba46dae4836aed22e3469662498 : FAILURE in 20m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ab106eb936f49b2b654d6dc6cffa1d1 : SUCCESS in 24m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1052428d9964978bb5f5e2ce32535e0 : SUCCESS in 23m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac029e50496348178ec8d98c8474ce5f : SUCCESS in 23m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/770426f7dbee4a0c9b2173dec76799dc : SUCCESS in 23m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c28ea4d23fbe479781c7c2948e0518cd : SUCCESS in 19m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8b1e958dcb0d4559822dbd720b05a6c3 : SUCCESS in 27m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b6d8984abf284fddae6e21be5b1b1154 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f86057885ae24a3f90974eab229d630d : FAILURE in 21m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/814c108788734dd7a83bfd61fd126720 : SUCCESS in 51m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/22b581983ee343f3b99c9615c0756366 : SUCCESS in 55m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2fa3ca6e97945c28d3e5dfb354322e0 : RETRY_LIMIT in 10m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f2c7872c83d491fb34c2db073bf9e15 : RETRY_LIMIT in 8m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87c347a18f0c4c029901a80492a70b3d : FAILURE in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43f10938847c4cb28d55bb044ef89a59 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/612c790188b74d56bc40cbd03d4f50d7 : SUCCESS in 27m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4af2de9823e640bc802f001869d948bc : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":4},{"id":"cbfc24cde59e44352934fd7dee9f99e0ca252d5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-13 12:50:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f920e3a6c6d498dca626d1be1bac150579eda0d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 13:53:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83c6f1ce1b24eebbca40c015236ac38 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f200dc8471ef40d896711d1ea5db1cef : FAILURE in 12m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23e14d94cbe74793a64b00b8d2357705 : FAILURE in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/240005aea047409bb836cdc31a8bcbcf : POST_FAILURE in 18m 13s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f28e5ce3c24f41019fdb26cbad6ca956 : SUCCESS in 23m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5ced0775cf3842e9b594c16153f1c863 : SUCCESS in 30m 02s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ce1f10419df44183b9ab40addcd8e4c2 : SUCCESS in 23m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85868180568b406ba802a747433a576c : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c62383383e5e4b8885594268341e59b5 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3019398c61cd4e09bfe6b845a5add23e : SUCCESS in 16m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5a126b705349d5bb4c0fbb65768908 : SUCCESS in 24m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04746029d419483baa2e32dc1a645b1b : SUCCESS in 17m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dde07b3ee50042cf8f3350d5df75e0f8 : SUCCESS in 29m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1899d1a84d5e4baa825c37e459f72e02 : SUCCESS in 31m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/786f144348b040a3bb2765e7867d6a4a : SUCCESS in 30m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e24d115214e42028cad270757519685 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d16672b0457d45ba8574abd1a2ffe7d6 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/66d9fd39f27f4ace98aaeb05ec484f4a : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c361d24d9624bf8b16039bbf583e0b3 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fe0e64ff12d44e6acea70fb2ad17425 : FAILURE in 15m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1e854e3991245c3a23f7d4c684c3b6f : SUCCESS in 1h 01m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/43ee725b83e344cd82fb92961186a670 : SUCCESS in 56m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/559c5c3013a440ce9925cf59cd96a19f : RETRY_LIMIT in 8m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17343c22f7e74ca8ab1f3b5b4eaf15a7 : RETRY_LIMIT in 7m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8efd86e71a74edca5f44cd5910f12db : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7237a7856864df9afb52ceeac60206f : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d2b4c9ca6bb4e219620b4b12458b81e : SUCCESS in 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea76f403451a4a0a8052201abec83932 : FAILURE in 17m 40s","accounts_in_message":[],"_revision_number":5},{"id":"57c0b5e66304a40c857871c0d59b0b3906bca2b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:56:40.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68c2441910134a6eaa74cea6cc60a067 : SUCCESS in 43m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e11cc3fcbed4b7fbe3b0781ba9299a7 : SUCCESS in 40m 08s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/2ab15357bfa64967acfde4cd64ab39a4 : FAILURE in 2h 55m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4ca4bb692584fa49ea0b7badfedf930 : SUCCESS in 52m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd29239cba4149f3bdfffa2f560c1fca : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":5},{"id":"cadc3b3a01ac2d7bb791cd6f14c2bcbb6b6a5866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-04-14 07:50:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e70c4d788dd539a357f1c519d9a019fc03e9295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 08:58:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5c526f7ccc4330a7e1b93f6bd1cd9c : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/ef3041317162401683be9894d67d85e6 : SUCCESS in 1h 00m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ad0161328f44e0cb382860049893e34 : SUCCESS in 50m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b302acb0b6749448a57b5ada17d1388 : SUCCESS in 1h 00m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3e55b6e3aaf64233b6798bfef61086d5 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/04880c5bc1db46d9ae202fa2ee85ae1a : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cb43a3d66ad745b7895317d8fce0c6ff : SUCCESS in 25m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/668960181f2049f286b1f08cfaa068fd : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17583e0078cc4ff68072346bf782455d : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e4f8385f106445f9b60cb0f562e98c1 : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8554dd33154a8c8aca0581cf1f8da5 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5cadfe3302bf4759bd6b05be4dee490b : SUCCESS in 21m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c0db45634e5418c8855b86df6958632 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dff081abcff6498a8e47c11c6a81662f : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76aff82da6124ed489c03e8d16355887 : SUCCESS in 22m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/679a2658605544d184b95524f6b877b0 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3c6d5b94aee043a69c8410a08651ae5f : SUCCESS in 17m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1eea60daffcb4bda972b8afd97b84812 : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db9c4b88d8e645a38a744a39407694a4 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2aa7a902a8ba41b0beb6da0ec765b363 : FAILURE in 14m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed46a1a941014fd4bb0bb62cc382e04c : SUCCESS in 59m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfca6e71ad8442e58bc194df05587443 : SUCCESS in 51m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0729dc0bc2f34f91b67ff354fb2455ab : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f842b4234414062861167c14fc84eca : SUCCESS in 38m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/326ca847e4d7429fa8640ef5c2d9d9ee : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c235afb4a83148c7b1b9ec63fb1f773f : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa551faf2101460bb8163421152e883c : SUCCESS in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35a86d73b0ce4bf8b551285c8ba4f408 : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":6},{"id":"604085f38c97911cfd644a8baa54846b15fc709f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:29:32.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e4da49d7312499dbd0dd8c3ae9e33e4 : SUCCESS in 36m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/651c4b22313646b9bc604560942e115c : SUCCESS in 45m 19s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8a5e0611f0154886807e4e1ea209ceb8 : FAILURE in 2h 36m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/256201ba159249249d356a28544ef700 : SUCCESS in 42m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c156cf639a143349a800aa95ad9aca5 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":6},{"id":"b659e7822cdfd85032d167e5b8ceedc3bb40c02b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"date":"2022-07-01 14:00:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"eb25c20df59be0dbb028a83cdc25055fea8a3480","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:53:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/273e065215274809b055957ea60c1c90 : SUCCESS in 27m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d43096e3bda4193811e52c78c4e8560 : SUCCESS in 50m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a0df9538a2546bc99d07dbea875bbd6 : SUCCESS in 27m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/659768eaac2649b687bcd0b45dbc5d78 : SUCCESS in 46m 52s","accounts_in_message":[],"_revision_number":7},{"id":"91712c02ef70f4a5bf21e17472a89f38170d1bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 15:07:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee85a3a3ffbd484480a943ac5015a1e4 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/6fef42589c9a44dba0299542989681c2 : SUCCESS in 53m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c101f253d41749198ebe252dccc75b42 : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30fc7c09fe084b13b60a3c692046a37d : SUCCESS in 56m 31s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d564e396f4324cd1b3445cb9098108d1 : RETRY_LIMIT in 4m 20s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ab4653e8ac1c44d28a2b0c0f40b1e99e : RETRY_LIMIT in 5m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cb0c38a866a9444abe3d3f62ba4c4c69 : RETRY_LIMIT in 4m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5a9dec1558b4da997290e8a9bd9ff2b : SUCCESS in 7m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1185c971621470aa8825f7a34a66e85 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ca9395ea99a4f5595df69b230a39879 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ee55d71d52a4b0b884dfc1222128da8 : SUCCESS in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b21223bddc5b4599b4dc2027125cf07d : SUCCESS in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1b243725d494cd08a154a4465c4b686 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1672faaff61445f9a6688a5d007232a9 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4f30ece2492246c2986faf6617e48e97 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ce94cd21bc3e4a91920f8d01feb1a492 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c9418e8ba2a54d6997e7a8a09bbedceb : SUCCESS in 14m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38f9e10864ea422dbe1b49dca4bb5566 : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd74e76b1e6141f7ba07513cf8c795aa : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8814b748e7054a5fb5b0aa09c6f70d8a : FAILURE in 16m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/267f0d95131e43d7a4f934b317b22332 : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a0023006607e4e70a8c505768292255f : SUCCESS in 46m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9db9f846e26046d7907f42ded8f7aab6 : SUCCESS in 59m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddaa9d038356461f99324614121f681f : SUCCESS in 45m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/766b37c64cee48188037a2c63469f299 : SUCCESS in 40m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/420ea1ba5af241698e8ad4bc779d0196 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6703cf1cff8a4f3eb2370d3eea18bd74 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1b95fc29d9e4576b1627a344a2aafca : SUCCESS in 25m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2b99a773c7d47aca769f6a12d157c20 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":7},{"id":"647ea81d48a62dbec3c949c4717fa6edc290fc11","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-04 16:34:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dedb45e7dfd59abec3fea868700cdaa63037a231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-04 20:59:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6821b057d2c1ec728a87a75ad91d8fa03828106d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 21:36:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d858794ccc74c0a8d18aa5dfb6c7bc9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b591208e7ac647fd8d348d048146aa75 : SUCCESS in 32m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75fb17fb965a475597f48bd633e5856c : SUCCESS in 30m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/787380699b94444ba35d1ae2cdba9cda : SUCCESS in 31m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/714dbfc79131497a83c63b4353b4f2ef : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":8},{"id":"ce9a632a8d763aba096499afc49f0aaaaef1f58d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 22:05:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2dc1da85979489ab8f9fceafa7e61e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd417ab58704f7ea39592ac3f1ae698 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/15694ab890a4413cb7187685ab915ee0 : SUCCESS in 44m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a181f7bd4104b69b3ee8f1c356c135a : SUCCESS in 42m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ebeb9f5565547658247f3bb222a99c8 : SUCCESS in 41m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b331177916f44ab2b0039e0806ef912a : FAILURE in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ca1dc2a3d464b4597b09cb250e8d759 : FAILURE in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89e857f48c4f461798c96ae3cf64e972 : FAILURE in 18m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6202fb3ea757430492e158e84355e4f7 : SUCCESS in 7m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c1fbd27b9bd54e28ba522f3aa1ea5517 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/184b242ae0994dbcbd38ea3475757ac3 : SUCCESS in 15m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a9a492d30414b44bfa7202a1568be1f : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c51c3936388420096a8f9c49846a228 : SUCCESS in 17m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dee202af2079459e9ac1a45012c6b7fc : SUCCESS in 19m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdf99d5b9bdd467abb8eea71f920e720 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/48b8b4c4a1bd4f2cb1aaa5664524ba3a : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/64e4c89a058a4dba8bc3463590d80c17 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fe866ae0a0314f98a3f7d00069b8d7cd : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7051883f31a34301b9c05867e968f377 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f1f5ab9269e4595a727697d0a71ffb3 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/29c74d4bb97540839f1971bce071ae00 : FAILURE in 18m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/141c7af072a04fe38a31c9afa9fa8518 : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a66188824cd945debd36c3158cc3bcd4 : SUCCESS in 48m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8066e86efc78477f9c5dfb45d57805a4 : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21b2e857aaad46e49158839782d1ab56 : SUCCESS in 44m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2c4e97791f44a78a855927e4b381dcb : SUCCESS in 40m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b58e50043134deda833235091674e12 : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6ee55ed7a34e85865de92ccbdf699a : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b952c5fa2c22435e9c6fec11224b0584 : SUCCESS in 26m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a28098168745e6beee29f66ff7ccdd : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":8},{"id":"ecd49a0efee5013758c3d93555e73922cf8248dc","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-10 20:42:20.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"32fbf6577cb578bafa676632995791e50b6587a4","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-11-11 00:00:01.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_34892\u003e\n","accounts_in_message":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"_revision_number":8},{"id":"1753bc556300d18f94bacb62e29ba9bca0e728b4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 16:15:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4a222b904d90a5fad746357e579e84fd0c289641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-09 00:22:43.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"734994b11460ee23093bc2479a9b825a3775f160","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-09 00:23:56.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7a60018a246fa363ed35d74cb07f88e2275c6210","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 01:17:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e693bb32d380435d893ef652b269a13c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0bdb50c0abd47c0a01ddec42abf6fed : SUCCESS in 15m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d75beaa9fdef4bc7b638ee0587ccbbe6 : SUCCESS in 53m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38adbe0e633c47dba4c30ad843137e7f : SUCCESS in 23m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a9d26f20f2746ab8d8b6b56a5af271e : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":9},{"id":"830b687969f7542ee45a3cffdd594daba90c5dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 01:49:49.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eed7a814d97444897a2de9d37a5e58a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5a0db15d214940ab4f3b739fe27a03 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/ed8eaf9c55024e72b0350116169d74b2 : SUCCESS in 1h 14m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/451056f5acf0433a8a08d5e339a01db3 : SUCCESS in 39m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/841e60c88e6e49bbab8c53afb36e9d0a : SUCCESS in 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8d2b01e186642ada5e6deb7e57dfc2f : FAILURE in 36m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20f3e775a7f142cabdf1ae0374e2dd13 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46c4871e54c74e0ab22d171d0f03513d : SUCCESS in 23m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf62ff1603b54d2eb0ce7cd6131528fe : SUCCESS in 25m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/044d29b6aff5440eabdb6f9c0dd1ea46 : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d025054bf8c4fcfbc9e9d4d6cb36933 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a300f749e7a412a9d9fc09f5bbe352f : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9c952193e7841e892cbe63fa73b610a : SUCCESS in 16m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9752f29dbddb4f8f84bb2960d2e97ec2 : SUCCESS in 22m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7e32eca1f67d457e820d9ecf8627b9f6 : SUCCESS in 18m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6b0b82426abc403ba13d43c3bf4738e5 : SUCCESS in 19m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1f267c2c63b4061afe1d224165f89d8 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dccb3c7a028e4ce7a6e0ac7041df38eb : SUCCESS in 19m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/823fc9884df84acf93d0e8f86b11bba0 : SUCCESS in 19m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/31005cc5e9f94873b7f2cefc1eccba42 : SUCCESS in 20m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/423f83fa8d484e59b9e28a569aa81133 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c3ac285aaf6c4f6f884f54939d5ec37a : SUCCESS in 22m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62aa6388b3fa4ab19bc940a998626430 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/68366cadd446410da9ee5dd543557db4 : FAILURE in 19m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5994d03ed32344c8b6cf88d7503adfbc : SUCCESS in 6m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9104e0df8bed45d59e7feca7d5f2ddbe : SUCCESS in 1h 10m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfcdfcc107284e50949c45647c5b93cd : SUCCESS in 1h 01m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89a2cb9e790940b396b8fc9181229115 : SUCCESS in 35m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/043ca403bba948d295a9e3ba9c8af551 : SUCCESS in 51m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23c09d3500f94f3aad6ec1c1b54ceefe : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec24edb009384e5a968d5b8fa4c27c59 : FAILURE in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80bfbc8f2061411ea9908a9404aa9b26 : SUCCESS in 24m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb1096c0dcf841bc85846725e9e2b9c2 : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":9},{"id":"35c09a47007dd27efc973a8afbcbb609273928f7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-15 05:49:13.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9419ec8f32942c96e37a54554233cc97b0c6ebe7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-17 07:05:34.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7d6087a2232f40d380966fec0959313e22ca2e33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-27 23:42:55.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"51b5e3343e90a5b28c7292eac000f0db6a648a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-28 00:13:36.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"e44d485dd6dc891170b5cfae862f6c42a7ab04f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:06:59.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5d0ec715d44c45aef885a1ae7c4007\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/343a3cb4ad5d408fa1e3d7c2909f6933 : SUCCESS in 17m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db537c9b4ee1420cb81bcc8b260462dc : SUCCESS in 50m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bb798e2e8a4459c9aac98dd5a4d1eb6 : SUCCESS in 20m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/51d683be46e541339e096e2cada056b1 : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":11},{"id":"66a4565201bf909c4ac2fa66b49f78b8d5c27034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 01:22:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83e602a8361d4b9f96af0f1be414b86a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c30b1a89d44613b4f9b28b2af3f72e : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e51d5f804e7746f8b981ff3e37c4d059 : SUCCESS in 57m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d4b22b934a1744dab8fc3c878d4a4f6d : SUCCESS in 45m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1892e4ed52a4ea2bcbc872f2591325d : SUCCESS in 44m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b84745d576f849e2b0983f096e122efb : SUCCESS in 48m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80095a2802964806bad790136088cec4 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e4f75d1b7064de9adda21917bf5668d : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ec795718b954e04a7c93185da1c06b0 : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54ff475deefb4ea6bbcc30050f7a40ec : SUCCESS in 10m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dabd6740d0544104b9f7047c4115d208 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d9a793a0c3f4183b88128c74fc0ec7c : FAILURE in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0bbee06b81c443218a3e482642b0b4c3 : SUCCESS in 18m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22c5d8c04d8c44bbb1155c190ac205a1 : SUCCESS in 14m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/38dce2b0b4974dbc963da68b9479d720 : SUCCESS in 18m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ac2ebf4a31d14cd8b1d008635b2b0adc : SUCCESS in 16m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb3e907900d94cf78ce49fb11308fb80 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1f8a1ad488a49fe8dc67f90bfc9b955 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e36603dbeae548859c7b737b24d71362 : SUCCESS in 19m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0f4f71747e5d4807a3fdaf949221adac : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2f4d6426c40c4f2a91dd7ad773bf9f35 : SUCCESS in 14m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b939cbcac86b477c8b9391666c8da8a7 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2652c0a9dd6f4875881a9bc2848774a0 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d11920805e045b7b73c361912f05771 : FAILURE in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24a4c0f9c11845c0b9b0256cc5236da3 : FAILURE in 6m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6cf6b5dcb6ce45de8a92cadf499a79ab : SUCCESS in 58m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/96d14df4ddf74e458282838c3a6354fb : FAILURE in 1h 00m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93e839155dfe4d969153b6ef83f47b58 : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7b58d3b22434a85950247ab2c818091 : SUCCESS in 52m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5afe23f25c4f4a42bcb99d622d889a54 : FAILURE in 18m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a1f4d5064c47f38e5b52ff35cfde6c : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1a48a5f32b44f0fae68d12adde9cac2 : SUCCESS in 17m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebd9782c753b4aaba95f546af14bdf1f : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":11},{"id":"97f21bdb6205645bd98a9030100b8e4ac80017db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-01 05:18:04.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"cc64171f5914f61b0c433272df7ef44124e77ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:43:40.000000000","message":"Patch Set 12: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5a19fceaa9c401b9932076d4024327a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520b14b436a044c08c4337d97555790a : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/f0e7014e23b04e91bfdcc9be6942e747 : SUCCESS in 56m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/108a3d50607b4909a73423c3ae0c2451 : SUCCESS in 49m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb1cd0911ca8476e9755088e2c566349 : SUCCESS in 47m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fd4509aa39c4f8e9b95cfb562d996d7 : SUCCESS in 1h 03m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d17c5c3d75541a0b4be7d57edb21575 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3473b7ec49b948d0839c8ef40037b355 : SUCCESS in 25m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15abac6753734011bd993c5e4532b756 : SUCCESS in 18m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75e28e91e19b46498523e0039c4b5bf4 : SUCCESS in 11m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c53f8b8fa8e34e28bbb1495fff3ae72f : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd6884025aff43d9ad11d4fdd7c7ff8f : FAILURE in 4m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/993968cd36de44799f5a3ce61efde755 : SUCCESS in 23m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24793d9080ff4c7a9adb64add420e6b8 : SUCCESS in 16m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f0d78e6146bb43309a4d1050ad274647 : SUCCESS in 16m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/eb53d18a84114242989f7c38cb0a3c1d : SUCCESS in 23m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6246df53578d4a0cbebc56f4af4ffbe7 : FAILURE in 12m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52e919a871984d8098033838ef4d6ff8 : FAILURE in 12m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a93a3ecea5df44d1a4785962d9488c34 : FAILURE in 12m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/551b30cc5e06402ea47cd4c232f99a85 : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4efe916c1a77488ab9e77ee16f710c65 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4aa0f502dd2848eca1ecb24b853b6774 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/545f5c2b13774dc2bd887e8793174f7f : SUCCESS in 8m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac825386bee640bc9a42d9d4326e4588 : FAILURE in 18m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c41599178204447282de489c421e1ff5 : FAILURE in 10m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb73f35d6d6349178203e911912885a7 : FAILURE in 7m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9153b3f3e5634e20b6e77a1412c68fe5 : FAILURE in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ef8c2bc878f4f1e87efda9859863f0b : SUCCESS in 46m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62dffed35734482ea1a2644cfe3f0bba : SUCCESS in 45m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff8052e3bed04c45bb0d59c588f1e2f0 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b497a3fe42f14a16a53a683adb740047 : FAILURE in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44bda6d82c9048cda9c1a018c307df27 : SUCCESS in 25m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c93c34d0b22498e9530ce2c7160c2a0 : SUCCESS in 53m 47s\n\nWarning:\n  Comments left for invalid file test/s3api/test_object_checksums.py","accounts_in_message":[],"_revision_number":12},{"id":"f90ab064185e31f0ed7abddb288117b8f7f4e474","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 06:51:25.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be02424f481042c28ca17c33e8554c92\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/672aa262fbc546299c224239d859dfcb : SUCCESS in 18m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acdfb1bb68b04595a8e4053545cd684b : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7983e9fc73904703a10740246f0e862d : SUCCESS in 22m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/babb4431ba7141feacac8112340830f6 : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":12},{"id":"955fc55240de5d576b1930192f979ae480217ea4","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-03-01 17:20:47.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a8fed7ccbf7d2939e19f0ac90d9d309c6126f42c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-02 01:02:16.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8b0b23f835b615944bbb83d6f0a2917c03b1ba94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 01:11:27.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d04b03d426c14aa7a72d6ab27685b2da\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f56942cc21a40f6904ba5859a3e0ebf : FAILURE in 5m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88c402c530214098ba8a2d6c16ce3e39 : FAILURE in 6m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd74268e495842aabfc997ce6869bae8 : FAILURE in 4m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5347cc15dcf49389a36bfbd47e81252 : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":13},{"id":"dacfddc705fef76785737901e52b6cef4fd2924a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 03:13:11.000000000","message":"Patch Set 13: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/caa52de2a55b4e84a0ec74713ecb31fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d5cab188d445d38411fa7d449a715f : FAILURE in 5m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/08f776f6bb0c4be29b7e36c398f5b08c : SUCCESS in 1h 04m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6c4b0d5fedc74743ac720a73739b3a55 : SUCCESS in 59m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8dc971ad78f4bb5a8ece1f79b94322a : SUCCESS in 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5698f1a21d664e8c8eec4b496df39ec7 : SUCCESS in 1h 06m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3362a1ecd2224024a345aadd8bc1a2a3 : FAILURE in 5m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1814c49b078a4856a9dcf3d74446fee2 : FAILURE in 5m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f21652ce854b4ce0aeaeef17a58abd40 : FAILURE in 10m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52ebf363abf34a18a146f678f77ed12d : SUCCESS in 9m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8708cccb222c4f6882a682de04b3ab93 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/881a25fed15b444d963bd124d43b18d3 : FAILURE in 5m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4b0867ee341403fbb2e1f3ba3060074 : FAILURE in 4m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c505bbee9050409d9999d8eb294a77ed : FAILURE in 5m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a51f81c176b04b6189e22c5b359764cd : FAILURE in 4m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3117bc85e6d74f70b90370b64f8e4755 : FAILURE in 5m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb3466a23b4546e190deafeace6fb69c : FAILURE in 4m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b798f0f383a4a5682d6de657b4065b6 : FAILURE in 4m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/549fb18ed6f746fbb3c20f5d8b3b143d : FAILURE in 3m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8f1f29c7e887441e9cc82dd70dad8af2 : FAILURE in 3m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9f04d0f05cdb415ead2b19bbe73a537e : FAILURE in 3m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eb17fb9fc8a74b79b32b08705b5cc909 : FAILURE in 3m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/52dd9056715b40ee8bdffa531326184a : FAILURE in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9c59241064343d186c612daffd3ac2f : RETRY_LIMIT in 5m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd3298707dfb4917b1a0784cc6620507 : RETRY_LIMIT in 5m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34950a8641824925991e08e9bdba8db4 : TIMED_OUT in 2h 00m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0c04f1c066a4cc095199d5912bc073d : TIMED_OUT in 2h 01m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec68d3b7bd944e5d8fb85355f73be8df : RETRY_LIMIT in 22m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/226d84d3bbd74c68a82c31f3a954727f : RETRY_LIMIT in 23m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40b336c5531a404c955be49ae1f8a6b2 : FAILURE in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca12c2f74b644260bae0ab640ccc7f3f : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e32c160f591b429fa532cb1716e3f133 : SUCCESS in 31m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77941512e3f04980b5a28e5ec95db634 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":13},{"id":"60a7637adbf550cfa290f070fae299f0d6fdea09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-02 15:47:51.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"d2c439144a91dc7e34c14306ea2c791fc74f9378","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 17:05:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b21eb94fe2e425eb1e7a721a778f25c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c860c647181e407bb725d1e5ca9cb507 : SUCCESS in 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e00a76453ca4353ab734333f491a6d5 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/739121f2e2344814b54a044ddc4b74d5 : SUCCESS in 22m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7ba482a6597428a94c4d9ea1d1c5da1 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":14},{"id":"cf5407d376d931f2246967966f9cc3bd753d0888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-02 17:09:21.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8ecb0f5533f4e799001fc98d778edb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56fcb38a88fe48389eda53436d7b3b9c : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b635101ca0a74fea89880f64c095a943 : SUCCESS in 47m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bee0fbe85b3f431ba37ef03fe46c0a78 : SUCCESS in 44m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9f030d66bd646d8af189d927d1e2dd0 : SUCCESS in 36m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f82b3a96cf54813a0b66eb41b74f62a : SUCCESS in 47m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c604f6fe95304ecf801e8d432c19ec80 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf12aca5989f47d8809aa89d02c5b8b3 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9d94716aeda4c84a252be755579bf68 : SUCCESS in 17m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5017f306ff174598b24ce23e1c98e2ca : SUCCESS in 10m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9f3e6824f2764869a3805fa2ad551f60 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ddf95aa8f57342f2ade3737281d6b48d : SUCCESS in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9def2b79d20049bf839d8c3ee32ad581 : SUCCESS in 18m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ede2093eddf24be6a8b4a4856aceabd1 : SUCCESS in 17m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e156bba9b0b845a3b859a0f4f90b7d01 : SUCCESS in 18m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/efa85139c8b5427bbc87bc0d01fd9e3e : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bc644f80750461586c4efe68cd7eee1 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d22f833c2d584ff38ebc70df5cca64f4 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8a627204461465c8e1c4c5bfb439b82 : SUCCESS in 22m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56ecd3f5cb8b4f3099d582a510b78a58 : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/44f335d2e48746d0b7a861fc36f1844f : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/32b6248e88e248dfa7723afbb0b177d8 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7217e5028f7e4ae9ac9315d258a7a01a : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5127b9f78e3b46a3a051aff06fe708f1 : FAILURE in 18m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/802ff2a41d29423d8ad62f5fb9dc8cd6 : SUCCESS in 10m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8daeff898931428b901ea44a9ce9b6c6 : SUCCESS in 58m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6e9b8eec8e3476b8f09c58202fb6bfc : FAILURE in 1h 03m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c41335596e6443d780025e8102349bbb : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/457e66514cf4439a88950df2ed054b31 : SUCCESS in 36m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96f4f15ec69d4372879ad281f341c7f1 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a53bfe9f31b4d1fbb0e21a14ee53e11 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ffa606073494f82835102592185e35e : SUCCESS in 17m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01c92abe448c49c6bd8fccca82623c2c : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":14},{"id":"7fbe72f8073fb25dd068601508b14b0f1e3bc9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-15 00:39:19.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"116b2f25b6537803a929d990921cbea920acd130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 01:54:41.000000000","message":"Patch Set 15: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44b419f58f3e44238882294ce193ae43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184d469971044d3ebb8bb84609b52c4e : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/c510a3c739ee49728468baf5bf4af4b2 : SUCCESS in 1h 05m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5fe057d124ff450f94711f65a0508043 : SUCCESS in 52m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b898b55782c7402bae6a5339819d3b35 : SUCCESS in 44m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf8c6bfa672c45c9a9def015d3593499 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e77e6ce17a1b4094ba75e35d61942155 : FAILURE in 25m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f23cae8d604e477d964d792e1ae0329e : FAILURE in 24m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c37b06d68e3e46a7a683285718797740 : FAILURE in 27m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78573b37e58b4c509064ea180e0e744a : SUCCESS in 12m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b7ba31a37f2d4023ad23aedda01661f7 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07a756c51dd8498ba77afd27f55a41a0 : FAILURE in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/816012fe2c3c4aee909d4a10bb791da5 : FAILURE in 19m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d470ee922ff4c2ebd770e7e329dace4 : FAILURE in 16m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/08b398fe405247bb84593596fd18234a : FAILURE in 17m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f097c7320c3d4273bb920a7906727b78 : FAILURE in 24m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/568440db1fe74a54a59b1f59610a3ee4 : FAILURE in 24m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef11b730a9644c31ba6a556d9389bbe1 : FAILURE in 24m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/affab37f418a4e99b8674e0a9e4ebba9 : FAILURE in 21m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7c9827373c264ec7a6d46d60a9de85a3 : FAILURE in 21m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2f2369a9ac7c415abf24c99f09b0e1d2 : FAILURE in 23m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15fb6428ca9e485db828f4ec64be814a : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b0fcee6c08d453bb1187426aa87fbef : FAILURE in 20m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/30f303855a4346499ffb919ac2fe30ce : FAILURE in 11m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/91f0cb6d994c454b8f8d11a28259bc30 : FAILURE in 59m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57baf941f139486f9f3f20a29eae0d84 : FAILURE in 44m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e089fdaee1824f41ad062426e1a24ee7 : FAILURE in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/714103b5a592492c99cc94a8791c8f87 : FAILURE in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559c1539ab834fe48f72dd9df738d4b3 : FAILURE in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4564f33d742242948770d9d696db3d43 : SUCCESS in 26m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/357693d1e7a74bbe838d4c7438b1b79f : SUCCESS in 58m 01s\n\nWarning:\n  Comments left for invalid file test/s3api/test_mpu.py","accounts_in_message":[],"_revision_number":15},{"id":"1967951656a2d6d124443e93a333901372a1f0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 02:12:53.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43118be8e8f74eb898ec7c3d56327aa8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e2e4d047f5b4c29a451de7e5c099a55 : TIMED_OUT in 1h 31m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fedcc69e8b8c453785edbd8498cb515d : FAILURE in 19m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37a3129cc92842c8a92bd7696e006e32 : TIMED_OUT in 1h 31m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e9ad1e833074f62ae575033b4154446 : FAILURE in 29m 51s","accounts_in_message":[],"_revision_number":15},{"id":"2b38670dac8cad848eb2da322cdb144e8789bd73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-21 05:46:48.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"8c41927399db87e1b8be056cdb70834312fd3c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 07:12:58.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0d81e287ce422a9b1558b52bd03adb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bea79a2df07c469090a749e5e2d53ad0 : FAILURE in 17m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc30c559fe76432ebeb70f7c7ed64e18 : FAILURE in 1h 23m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e147033ce22c43b7bd6b08f48a866195 : SUCCESS in 19m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac7bf667a4bd4baebe7a0d259598827c : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":16},{"id":"c7cbdeb232263130ce54b8c959d1b0930ba66b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 07:50:00.000000000","message":"Patch Set 16: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aea182097f644407a65b1a5f3aef4c49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f38290b2f54b28bcade3f1c8c8d44a : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/07330f3e8968474484f0072f033aac5a : SUCCESS in 51m 05s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/528c24d3e5e244a89f077cc8c8e48015 : SUCCESS in 44m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/550d2295ed244063bc7fbdf605578ca0 : SUCCESS in 37m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f50beb64e0be4e6ab4d3c9e03dee844e : SUCCESS in 1h 03m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b3b5527f0184301903407b286e1301a : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c535cbdcf2246a3b15b31c7bf4eee48 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e83b1f0867b2487e805c84641b34c285 : SUCCESS in 16m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0de82c06561442399ca7df1a7fea5b38 : RETRY_LIMIT in 2m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9a9ddbef88264151ae34bd5fc889e3e7 : RETRY_LIMIT in 2m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b4927aa9d4c4b1385f34683e45c0982 : FAILURE in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c398d67ded1143b4ad1d1658dad6caa6 : FAILURE in 15m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b48cf6e6de934db7ac8d4d9b84eac1ba : FAILURE in 16m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7b4311290c424145b87d868d23849309 : FAILURE in 22m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ffd62b46249b451a8e70786912993da4 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b054768b40d3414781ef944ff6f19ebb : SUCCESS in 24m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d155c41312b6438e8f937f58d11b42dc : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e091f698e3d41e4af604fdc8ce4855d : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9c31f3c11fb446ea9c70c2b8aa11efa4 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5d022d46f2f1438c9dc09e213356201e : SUCCESS in 21m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0735b501a72d4d2390de18bbe661b939 : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/30e3513acfa145159e8f7e1040095fdd : FAILURE in 18m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e943df8700d146c08c780cee5de28078 : SUCCESS in 11m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2376f33321f47c285d9dad0b1a9058c : TIMED_OUT in 2h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7dfa45c74144b67a289e6e86a1d21a7 : SUCCESS in 39m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0271cd2e77846b5b4ab4c5b41895954 : SUCCESS in 59m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb19a43d27af45918b476235daa9d49a : FAILURE in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50539a78b0b646649254ccef0d945882 : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa286e150a774b8999a4e93746a23e51 : SUCCESS in 16m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fa34b627a874c2ab1a2f8cadf320561 : SUCCESS in 44m 46s\n\nWarning:\n  Comments left for invalid file test/s3api/test_mpu.py","accounts_in_message":[],"_revision_number":16},{"id":"ed892e061a5b71262250bc7a9a2ca9de5654b325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-22 22:49:43.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"0d0944a3fdbfbe6d33cfd51e9b7f5e4d6e41d3b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 22:58:49.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ec650590f9405088182c97c44593d1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25ce23f2fa3646ed87c34509968e07d6 : FAILURE in 5m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea1a006bf92643c8990a9bbe82defcae : FAILURE in 6m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef1bcb3c69534f08be098176f2666c70 : FAILURE in 5m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0049f52b38d4f4088f3799795f6bd70 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":17},{"id":"487827880ddffd074ac0e1ecfed3dd315dd92bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 00:57:00.000000000","message":"Patch Set 17: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51440c5dc3394932a3d682e0fb1c9a5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b19f81dbc4143ab88135251db93993e : FAILURE in 5m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/7b82629f902a40dbb64dc0880001c18c : SUCCESS in 48m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b31d01091c454f4e860af3eee70ee7ff : SUCCESS in 41m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91d90b15362e483d9d1607c99fe40cda : SUCCESS in 52m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a5e8c1853b943d5963d6198f86a7dc5 : SUCCESS in 59m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b6702380a084cc6979e086efff8eefc : FAILURE in 6m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e649fa53d9164afa999a572f1aaf2837 : FAILURE in 10m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6eadb170a224483299e564f7d1f9ec31 : FAILURE in 6m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bc65942d20c42888b73eb21257af48e : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7318d76b98434ce39079f3f5badc5a09 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40e4e24b881b44cb8f13848f76e0e39f : FAILURE in 4m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7de78f747dd74148b37dee03821d3c43 : FAILURE in 4m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b9b6bbc9dda44ce4b3dce40df65ee480 : FAILURE in 4m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/58df39bdb37b40db93179cfdec09deb3 : FAILURE in 4m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/498b8ee55830479088324475e128226d : FAILURE in 3m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/162eb323172d42ed8b0661cac05e6e77 : FAILURE in 6m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c68ed6b89764cfdbe4eb19c303336cf : FAILURE in 3m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb44c4d2e437450397eb0876232a45d1 : FAILURE in 3m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e95718fbc0fb4ae988142704edc88a62 : FAILURE in 3m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a8fcd79272fa4e5db8d182d63f8d3d60 : FAILURE in 3m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/daadea6e35b0414ea9cf833fdbd87105 : FAILURE in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bf5b3c979b04201af3476d89d0cf4a2 : RETRY_LIMIT in 5m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/56ff3303b41240b2b304f7daf50c31e4 : RETRY_LIMIT in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d2d8c94eff440d5bfee1116dd59bb5f : TIMED_OUT in 2h 01m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13383d042b8a4731aa96eb733ba6ebf9 : RETRY_LIMIT in 21m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d843090b6eec418caca1208d9bf041b8 : RETRY_LIMIT in 20m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d00d83aa837c4df093b2495ac0dfaf21 : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033acf22a3194d0899a2acb683d1911d : FAILURE in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3322c2a044e9426f8bc343c8439f7aae : SUCCESS in 22m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6bae58f3ef44168a111765ffed839c0 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":17},{"id":"0aeca424aba9d50d9df6bcf6c0b9a1c12bb7c723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 18:47:56.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"394108f4e3950ef26f8dd6de0afafe0d7a75beae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 19:50:40.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2416dec891014b94a2680ef9180840d6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6369da2e762e48369e889547b25e940e : FAILURE in 1h 00m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64ca480c09304a1287545c61d0fb89e2 : FAILURE in 21m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9c89b06caa44d52a6b7132cbe43618d : SUCCESS in 22m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b4fa6e208aa44a1a0f3314b939860c0 : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":18},{"id":"d3cbe6b8e87c531720732a6389f469af21c272f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 20:03:31.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03558d63e55d4df68cccdd0c4dad4b90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6346caf00f7487d92cc00a5e231cd3b : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/0bcfe5492a094fdb9456d570c2b4dc7f : SUCCESS in 56m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3e15487c619849b3bc091d0d4885030f : SUCCESS in 56m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/985a1d304c804b70a5e5ccdf85f75a4e : SUCCESS in 44m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/631bff9066ec4ea5bc266eed5b388cf9 : SUCCESS in 1h 08m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05bbc88dec7649a2b4d7ad3290e43c92 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d8677b724fe4c4bbc1d0d2beee9373f : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9897372338ca4811af3c475e61cb3902 : SUCCESS in 22m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/095e85d31f5f4504ade5a05168a6acd8 : SUCCESS in 12m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3d57096c8fd1431eb1a164350738e9b7 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e82312dbb83541d996a58ba9c587d001 : FAILURE in 22m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de96b77497f84703bd36736e2209c33f : FAILURE in 17m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5788b2b2a9a41039060fc3587219617 : FAILURE in 18m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/61515d4b4678457aa69aeb6477e5da44 : FAILURE in 15m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a87d6516676d4d9a8c7fff4b2e621ae0 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28882daee0ad42c7a8190ae03b5acd28 : SUCCESS in 19m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0d96b1ce78f4f498040a878e4a2c2a8 : SUCCESS in 22m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14eb1f7997c14078b89db5127747a6fd : SUCCESS in 12m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/314cc650baaf4f2bb85ce44138587f2d : SUCCESS in 13m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/65f6365d3916405aba0fbda66e14ec77 : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18571adda15a4ff19ebdf82d17ec21c2 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a734c7dc0ee42a29915215cac159de3 : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1fac8e0f3cb548ec96202042f399f8f3 : FAILURE in 8m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c95204743ec2499a866315333d4cbeb5 : FAILURE in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7566db93c082455c9a4426b7a58bd771 : SUCCESS in 51m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/760bfbd420754e17b2d3b88f0637c81f : SUCCESS in 48m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fd1c50c83b54a75a25a028a7aa66fe3 : FAILURE in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808130337b5947b392f15eae57092b4d : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e246767495144864a6741b669f6b8d06 : SUCCESS in 18m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48e425cf46f445e695f83d099d5e9f65 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":18},{"id":"fd6b95ad4f6d759a42c53b1344cc02a675d4d609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-25 22:06:43.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"784b91e7bf6542d26d1825ecd822920b5a4f50c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 23:01:55.000000000","message":"Patch Set 19:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909802 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"a8df6ffeee3a345cc26e9e0ec46cd3a8143f3a3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 23:01:58.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/909802 is needed.","accounts_in_message":[],"_revision_number":19},{"id":"c1f86116ee06667048256af51406b2dee9ed21f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-28 05:02:57.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"9c28176d6faeabb52c63ba07871408f6acdf002e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 06:04:53.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5d25da7d2b940d59e9bf43e0058ac61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a390ac3b8ce94c7280ad8f4421d1b60a : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f6b6c3e198604ee8b6ecc29235ac7bef : SUCCESS in 49m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/541898360dfc4964ae9c9c42f30b09ba : SUCCESS in 55m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe67ed462d844afa9908332b00193b58 : SUCCESS in 37m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/201606f6ad60414c8ad019a3e5e2d1c2 : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/269cd2cb58a04bb29230769465411326 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/306c0c6f027a4cba9385fe29d4d582f9 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54fda932df5e465eb69c120380415020 : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ddfdaf340974b859f64092a36e4c1bc : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c2a312447f544cc8cd6cfa45653af73 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b58bb8c36864210a3be74ed7f7392dc : FAILURE in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4832753678346a4998e33f5c7060b1b : FAILURE in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed592ff23954a29b7226c49e5d507ef : FAILURE in 16m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/590c74d2e1f44a849ee8249db59e3407 : FAILURE in 12m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e88016fa30c64673bf53a833b76b0413 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1548b8a992445fcb779a8657f1bb8d4 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/441a6cbd96304cc59bbe10be341b3233 : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72af6b3fb86e4b14b2675020590e7797 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/240f8d2cc57547cd877860e88027d945 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80b16b366b644511a646848e435064e3 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b639d551b3f04dbe9aa9cae7c21cf1c3 : SUCCESS in 5m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75692bd09b834741a331baf8c77b4bcc : FAILURE in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9052eb90d0364375a40ab6e486caba19 : FAILURE in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2365fd7931934342b59c1d440f21d8d8 : FAILURE in 52m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ebdb24a851f457d804521ae73bf157d : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52fd7ef764424a28bc1c6ca254f5bf8b : SUCCESS in 52m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a58e3ec3fd7845809060a2f6cb78db10 : FAILURE in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faee4b0ee099479e9a8e2e2251b9b129 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/212ee98f31d7421083a8c72eb5aab796 : SUCCESS in 22m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e0fdc78436a4c609a61283b116f9a00 : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":20},{"id":"3db913abfe7447fd0a8a47eba55b3b4b72f50b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 06:26:29.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098917e49eff4c71a9337ca38dbfdca3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bde7aa2a6a14f88b88535975b312a55 : FAILURE in 15m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6d887bf64d742239032039739f99956 : FAILURE in 16m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/331a86915f694a1ea1195009fce38972 : SUCCESS in 23m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3ad9a9cf0e04cd2bb96b90fc15ae837 : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":20},{"id":"679e6926961ed5a03d03dab420546bd98664e2ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-04-01 20:22:21.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":21},{"id":"db938c2a4e4530cdf3ed7a9c2ec5e8950f2060bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:40:55.000000000","message":"Patch Set 21:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1399f5fa466847018a5394109f1ade1c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b79fbaf35eb5413f994d7a172d3cd8b9 : FAILURE in 23m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfbe414b914042f3a5f701a574d8d5f3 : FAILURE in 1h 10m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9cf6a6932f24b208aa20ed53d702d4c : SUCCESS in 22m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c065419fbfbd4b09b4cf078bea3265d7 : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":21},{"id":"e50651f36209d192309ce175da8d7a1c34ffdd28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:50:48.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a786e507b06452d9425efcc791860c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fd0892bf8f469c87aba293f4976eee : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/bd717db55658495f8ea5fb8ba8757b61 : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8054774ee354497af7345c2d7c2f75c : SUCCESS in 55m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bf7385abe494a49bea900822334d0ee : SUCCESS in 57m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17c78da8dbca4744a484dc556b698b17 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77fb9aecf0a14a62bd7a65fa151ff1bb : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99851db774984e5497a4736c21f7955e : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65813a386c3f4c78b71c0e187da10d0b : SUCCESS in 11m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/31e73507a15e42879adf22dd0e7a5ba7 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1989da8ca69c4409b5d994fe171c52c4 : FAILURE in 13m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af3ac5576f00404b9874a93c2e73465b : FAILURE in 14m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f828d53c84436986126403ba802d4a : FAILURE in 14m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8fbaf29e5d914f4184a18a4f93736566 : FAILURE in 16m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7868a76719924222b37f1237edc39d19 : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ec8ac1db34474d28bec6b3156e48358c : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/732daea3784e40c5af79ddc065b089a6 : SUCCESS in 19m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/07e19baf7066447b8720a77a0abb88c2 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4ce8b25525c14c658d48193b21c905f2 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3313b9de2b144250858a9f47dfc73a1f : SUCCESS in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74bacf7061a7489188362d3c5b4e9cd4 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f648daa64e544b2a05aa1ed3ede3243 : FAILURE in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e3b29d298f84c1d9a9fb57508b584b5 : FAILURE in 13m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2efd806c45e74cb6a9e90d043acbeaad : SUCCESS in 59m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a60d9c7f289947699e5719ffa0c736e0 : SUCCESS in 52m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01255494a788431fad2b319824c1efd1 : SUCCESS in 52m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e114fe82d8543359ef4e19bb829cc08 : FAILURE in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c797f3ede34cd58a38151d3a946f42 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9090ea70044740f19e16b3c6494b705f : SUCCESS in 18m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42ac650058174d5387fa193bdb0771d7 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":21},{"id":"1568cc3c184b8db8bafab1acf0b2606bbcb2fbc7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 15:18:21.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"2967096d3cd0738580df49eb05b8be8f8a2d1d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-17 23:18:51.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"4cdfb94a5a66a92ea35dd7b6720f82f642198492","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 23:40:45.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f803ccecf1d424c82d0f6636d903b9c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e4f7ad1a80d4986a2838f74bf977921 : FAILURE in 15m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8029fc44af764cdc8547917e133b9a31 : FAILURE in 19m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4050fc0b31e4feb99a3cf2e6958661c : SUCCESS in 20m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/74bab8ade12644be969d0af2a65062a1 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":22},{"id":"06d5487e00c505b290677132455c84c1ad3077c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 00:30:27.000000000","message":"Patch Set 22: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eee22a6cb8664c1f8c19b844e0c58dc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d086852d32b4936a29c9f8317129f74 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/a2a31479828c4071907ba8e1e6d2eb60 : SUCCESS in 1h 04m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e725cc7680847ad8711bed2255b7b46 : SUCCESS in 53m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158cf47bdfee40949ea409707ff7b83f : SUCCESS in 1h 05m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bee1f647b50043cbb4ab10335eecf72e : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ae82cb47c4a49428606c2dc70774b8b : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05d8ed02c49f4f8cb361efab94c8814e : SUCCESS in 15m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e595966fcb7447084209479e4ef1a5c : SUCCESS in 10m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3a0947c2695e4f61b3aba7529854a769 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/67c1889cbe1540e4891198eeac0cb47a : FAILURE in 20m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1b080a85ea24b03b1a65f2270bc56c2 : FAILURE in 13m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3b466fa6f528446b90b4b770d03ef1e1 : FAILURE in 15m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/23d4e3a73a0e46419430c99e5a9688de : FAILURE in 24m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6436ed48d3e7498790f8d8b07bf4926e : FAILURE in 22m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/420df33351fe4105becf89de620470a9 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d8a10689c9442bca6580b87d132e070 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1595dae7612545bda3c49b58bc49d2f3 : SUCCESS in 17m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/321d1b4c78864f42ba7d9eca1bd7d816 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/399cff1fcd804951b0c827594312a458 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30efd0f804b34483b815c42f0d10cf8e : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f2f86c9290d48f59b933ca2b6b605aa : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0044deef61247719592f5a8e925b57b : FAILURE in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4e4b029a415a453980c80d3fc258da6a : FAILURE in 9m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72dcc05bb78945a7bff3420495655f5c : SUCCESS in 57m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d632600136e4c4da09ae3e2adfa16fb : SUCCESS in 55m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5a8034f6bd4d454e823c7835941662a2 : SUCCESS in 50m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/565966a8bf8b4f1e8e0f09cbde859a91 : FAILURE in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd517d4a1ee54e10bdee5d8f69eeb271 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03d156c9370143b89f6f570119322e8f : SUCCESS in 15m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da1deedfb5f4b4f98b2cf1a7afde3cf : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":22},{"id":"e655ffbc668758647b353d3fefd70bea3451a365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-02 00:25:00.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"2a3f214858747adf154844e91ee62595bd19a095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 01:37:16.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ffa25c220d74c6e83f2e1c8b7f1fc8e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76b18470078c4c3199df9458d8267233 : SUCCESS in 18m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09310d01682143fc9c87904d7a9f96f3 : SUCCESS in 1h 10m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0592f0ab22f44976b5b6ed48dbd198b7 : SUCCESS in 23m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6752ffb94b4f4158887e36e0d0b81da7 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":23},{"id":"2cf310ab466b77e14ff86c51df75e0712f0ba833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 01:45:41.000000000","message":"Patch Set 23: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e837e37e82b24cc48321776010611a51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d65b77c1f0d46c0a116deefd789f650 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/cdb0ebd2fe954d00bf084f3a62a9c955 : SUCCESS in 1h 13m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1aa036d341084079aaccee9f2184fc20 : SUCCESS in 55m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0798a65c9c4c4f8b7d43f511824db8 : FAILURE in 45m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/645942cd7ee14bd0819da85f37f3a9a1 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d4c588810224330b4445bb727b661bb : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1d9dbdf1867471f9d96c6307d8b77c2 : SUCCESS in 17m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cff7fba8f5884b35b046b07cd672e2c3 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f15ac3087ffe4fec99271bf94e01e182 : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80d37b2cca2e483e946fef2ee28adf91 : SUCCESS in 18m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/28c65217ef4749a982d8b178a1aa1a97 : SUCCESS in 25m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/421ee46bd26c46af9823e29d0b8ca614 : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9b1cb7207a1040f99562d88f7e2226b9 : SUCCESS in 25m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/808741380f78437a961056251336a06a : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66cc7248db4c4a51bd48a07744624e89 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/11b076b193954c83ab0b7cbc804827a3 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7ef6a81571f47fcb071998a2e882c94 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbe3f52c7bc140e9b50d83818526fc6f : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aaf7b8b9881847d19f513e6de80d947d : SUCCESS in 19m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/00216c1aa551431aa895d474b9eb8e30 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc312d6d18544ddabf7c7c4b6b23258b : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7dc0b1426cf04d9cb6e6a798c9db0c57 : SUCCESS in 7m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36afa0f47690496db8c78a60614b8378 : SUCCESS in 54m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0eb5682fe8546e486d23f0c62c1975c : SUCCESS in 37m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/576b9269e3364d5dbfd0b0cf30857148 : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959cae495f774388947caf9252bdc4fe : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e8d8479b934c79aece71004901c4ec : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d79d9c5aca564724bcef2d021ed95c10 : SUCCESS in 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed53846e5fd24676b7ca84130f694109 : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":23},{"id":"29c691f303c34a96dc2292b175cdc724b60a818f","author":{"_account_id":21482,"name":"kirubakaran kaliannan","email":"kirubak@zadarastorage.com","username":"kirubak"},"date":"2024-08-08 12:09:31.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"5fbf79d54ad3344b4de2ad18e1dbf8586e7174af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-16 00:11:36.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"339eeb86593f1a1cf82051f13d97df5fd9ea6add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:54:10.000000000","message":"Patch Set 24: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/433585b37d984174b26f9867caa6bd6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3c271a1cb5458f9319e8577679b8c5 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d17d445f8fed44cab5c829dcd18cbad9 : SUCCESS in 1h 11m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb0ed1964164490bafbef8585af770b4 : SUCCESS in 50m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f8881d699754a6786714e2f5308cae8 : SUCCESS in 43m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d39097a18b5146f488aac8368ae076b3 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffe0a7fe207f4e47beb007f737fad813 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adeb1410012143d5bfeda81eb9931c1b : SUCCESS in 18m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66eaa1e0348e4040b564d4dc6d934891 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e8cfc9aefe164ac1a0eca686825c8de4 : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbbaf048d4640d1a1a4b2400f765278 : SUCCESS in 23m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/abd05aedaf424dd982ea7cff4f8f9714 : SUCCESS in 20m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8cb4158830224516b13a09f4676e081c : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2df161f9c13c4bf8ab6617805029889e : SUCCESS in 20m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c807d408fb22407d9da9b25d5da0c92d : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac356a9f39124844a724605fae746595 : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f438118bb30a40caa687bb2a9ce4c5e8 : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/828ce5eb5571412581201f3c4e6f3ab6 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/17b9de4bc1c24cea9fbd3b72bd02a853 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6007c10a2b4945059923e3adad208c76 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfc9d6d2c88c42bdb69ab754bba92688 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f02032a3baed443ba9687f9acf73626c : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a6967064f6747f7aacf7617935421f3 : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ebdcfd74fefb42fda19a3d2d91fce5ce : SUCCESS in 51m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9578b51b03794bfb9e88eb71510cd859 : SUCCESS in 45m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3761c86aaf7545268ce85412bf242147 : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abae10060e3c43f98d63b4c148057bd1 : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c69670291449ff9a2bf3f56f633fd7 : FAILURE in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16df8a5dece148be9cf9b7d37fda7127 : SUCCESS in 16m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95e8a4c55ca84e9f83bb395439121137 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":24},{"id":"a55dc38f555dfdce377c2ea15ebf9808dd7c9f88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 02:46:07.000000000","message":"Patch Set 24:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b97b8fa36048bcb9f2f136f888a275\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d0589edf3d742e9b3a8a793c952ad55 : SUCCESS in 1h 18m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05859fd5ae09413d931485f9152bc221 : SUCCESS in 1h 18m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdba4e8ddb404705bff213c88b594de1 : FAILURE in 1h 26m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7383ed2669dd455f800b5ac08e9b148f : SUCCESS in 44m 39s","accounts_in_message":[],"_revision_number":24},{"id":"b4ef13fe53b507e684c9425fd2886e72cf1fcf76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 00:06:36.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"6e51c60daeef252b613eea12193181146e2700cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 01:27:58.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3bef4722d04974920deaeb8ee5aa4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bd181da9d124a69933fa346a5705ae9 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/e4c277810d134000856def9329851d13 : SUCCESS in 1h 14m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41564376b8a0405886ee32e93bf8d54d : SUCCESS in 45m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b945f7b4dde1474581dc397713dd661a : SUCCESS in 54m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0375346574bc46358a509e3b33ec57b9 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e385f0edbcac418897d3d30f4a5c1bdd : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/823d57a8bca9419bb81dfb4e81c07440 : SUCCESS in 18m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3795830c950b4abdb040b29580773d88 : RETRY_LIMIT in 2m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/91f6a284ab924ee89a1771cb2835b8aa : SUCCESS in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/252406ce3798434c9da3beb863a05fa0 : SUCCESS in 18m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eafa813505b64893aadceae74f4c3069 : SUCCESS in 20m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0989fbfa7f074c01a14a8c93d66103ab : SUCCESS in 26m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a056a4628ee0413c94c4b5dac11aef64 : SUCCESS in 25m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/679ea9cfe8e74f089f64cde14bdd06de : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c4ab97521cf48f0b37d9f4ba098a9cf : SUCCESS in 20m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9db61e832aa44d9caf2f0d2148d40797 : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/00dbab220e8b45d2a0cb22292729dcb7 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f157666869047e09d2c3ed02594923b : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ef7fe89320c4f6b8e19477b98e08977 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/88a6647e745b4e1490dd2bc8a9626bf2 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31d6dcd4858b4ece862a5c2afce3d97a : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/551f7ed7b1c54060b3948724c9fad38a : SUCCESS in 7m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ab50a2effad4425bc755f08054e6732 : SUCCESS in 46m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e27c64902dc43ceaef5f77eba045cd8 : SUCCESS in 45m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39e687ab85c94252833669e034f24ce9 : SUCCESS in 48m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d437da3fd9e048e49134e74b901a37f7 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2031e7467e4774a3c37d5572a3ad53 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c85b150d2e834fc8b5241b8d996795d7 : SUCCESS in 22m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b54a0dd5fcc04d9db93773f2be1d68ef : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":25},{"id":"386513300db759204ae09ff233915ddd809d4396","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 04:08:36.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f77e6121230e446e9bd959880980daf6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f613bbb42453434d95cfc53f9bc17fc1 : SUCCESS in 1h 17m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ccaf4c5d5274c4fbcc2f27c629f95a6 : SUCCESS in 1h 20m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba20308e2d604f518d69937f49af507e : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/513f6a7d32624c5cb72a1ee511334f80 : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":25},{"id":"be69156f73c7af65af81d8bb16c7dd0c44f4fc69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 22:20:50.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"60543b46ca72c1877ec6598ed58ed87ea6077300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:28:33.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/251af7d9e49b48a9af9980e65266b718\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46333b2feb1442d38c25bb9b56545f33 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/81e5ee25ab8b4250b6d47f4870017486 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/406cfef35c2241c5bbe421750d1069e0 : SUCCESS in 50m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f5aa6e6b7247699d4a885092cea25b : SUCCESS in 57m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32a2b8613fd5460dacaa13eab04ad43a : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e37361fcbf4341a9989afd652a7282fd : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/939464ddee404934ba31098b713c4235 : SUCCESS in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15da2413c915407e90a6374eb97492ac : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/867fb487bb0240158054ab96af581cc2 : SUCCESS in 19m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/728c613ecc6b4b11b4e121f37e234738 : SUCCESS in 18m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb4471eed5d4ca4b67d30b3a9caf521 : SUCCESS in 22m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/452c33c750824f8b9c57d08011ab2276 : SUCCESS in 14m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aaa216265d654aec96b41f1b104f4c27 : SUCCESS in 34m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/edfc3e8fe29141419a298e9bf60e0393 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9338d00742d44b04b6bd6abcae5babd7 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc60090c74d54fc688a8be909524fe95 : SUCCESS in 15m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1eec7e18001042e395785169468aa8d5 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a6dd034335dc4147b2429790714bc994 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2040ce52310d40c59461bc88f931ec2f : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7757d3bf05c44648a02ef8538103174d : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87bdec673e804d2196f36a125f1b1c13 : SUCCESS in 12m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3af5281215d742e0a026fa12dfbed742 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b38a85d466354acfa076e7c89da319ca : SUCCESS in 47m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e909891b0fbc41648d2e465f6fbf3671 : SUCCESS in 37m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5467a92a9eaa4189bd3e749e1bba55b8 : SUCCESS in 38m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00593b01a534b6b908f77da8f0b5bfc : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eabeb11e7854839ba1b54a1c82a276b : FAILURE in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2272bba07e45420a94da983c5afcd804 : SUCCESS in 18m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29811e11c0474b6da34909ddc78a4f73 : SUCCESS in 40m 54s\n\nWarning:\n  Comments left for invalid file test/s3api/test_input_errors.py","accounts_in_message":[],"_revision_number":26},{"id":"f300ceeafb3c847e276f952d6aa60b7e4f9ac778","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 03:52:13.000000000","message":"Patch Set 26:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe2262fb76a4bb3abf7a3a3142cc108\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a69e939b2dad484bb8e54a04419269b1 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66134c96f8a14068afdc23dd8e81b7b8 : SUCCESS in 55m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9b7fc47456144f2b721216e24622559 : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d384f4f12d14497ac840d4e9e8e557a : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":26},{"id":"5654453fcaf4a4d4feeccbf9c3d09b60c6383b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 19:22:17.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"735c8504c285f8cf2a5ea702e6c2a27d51e6627d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 20:34:46.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f6e583942ae4278b9c52d158cc67719\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51650969beee4a90b8fe5bd3b1f89d72 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/9bc8f066e751468aaa452f7a895f38d2 : SUCCESS in 47m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c945d42bfef486cb218d8c458b00370 : SUCCESS in 54m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba2029d2b21f4783aa973c0d85f2c2c5 : SUCCESS in 56m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5f12ed6df9542bc8a0075b4f9f70393 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2908d611bc89406482a077e789bf47c4 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a488f05af8440e58d01af2a72a49e8a : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f387f9e121324625ae27b96ba8f02c67 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c1463cc4f4a24cd8b81366bb2232e769 : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46c7160908044988828286b2f6789758 : SUCCESS in 14m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d10c270ead464f5db741efe9fd5f7893 : SUCCESS in 18m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/acced3a9738e46b2aeeda9f79538b53e : SUCCESS in 19m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0683cdf46c5e4ab881ced3b12dc34cfa : SUCCESS in 33m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c8782f8678b46f8a99cb308209503e4 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9880c281b743422b9e8d82f5d369e7ec : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e5c0728eb414b9da27078cd3a877b18 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99c29ecbb17540598ada36c59ba99d4f : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f26cc2e72201485893ffc7ba20b5ea68 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/076ea8b1197242f1a17ae82afa5c3ae9 : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a3c65932a8a4aa595b19caeb04b8e72 : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5508e0053f2a4d9fb5a8123e2062c2e3 : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/58f6061f97b845d6b65050bfcb61ecf4 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f569efcb63e84aca9bf1a1b0e1e56e79 : FAILURE in 47m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0cd763fedb2348a8a33b283ad57c7c66 : SUCCESS in 36m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3a9ee144fdc420b86e2f82455a831c9 : SUCCESS in 47m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00ebf2a7f3f047b2b934dc1b648aa906 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1be472668f4bd5bcdea8713c82e307 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/258678dd68d147f5beed3d9b29606218 : SUCCESS in 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2663fd69e6f549a7bc2eb77afa436a39 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":27},{"id":"642e5433081db87c51ad4b505c8e35781ef49f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 23:42:54.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"43887d88da1a16c96e4c9a666ec39f3d74fd9c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 00:58:35.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24d81e1154149e1805006e8b3cc6112\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57de959ab2c84a05ad419fbf3c038c62 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0245db5deb134daa923acb0e9391ba05 : SUCCESS in 53m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5f42ae42d804c90802b7320c3713e34 : SUCCESS in 36m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9193f0c19ebb454a8c197c0924a570f1 : SUCCESS in 54m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0386287ff7b43c1be261a69c20e964e : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31ff38a7515449e09d6e00f71cd5918f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cfb4a1603b1475a9846b9743a6b5121 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2628b924bcfe4215a9f62b96d0df2ee1 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f29868e42f9d46da861556370ad71a9e : SUCCESS in 18m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a7558f944cc42c9925effc83a243ce4 : SUCCESS in 14m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/991eaf4841b34d40ace318ed5b572c22 : SUCCESS in 18m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0c20051e0a8640479e09d103b5528370 : SUCCESS in 17m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a943a8b65ee746cdb00e2da1b0800302 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/50bdb563cd1e49d79779b121a7f45dca : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae26717055924f3fb863e12082eef648 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f08322528ae447a5910b6ff20be65b8c : SUCCESS in 17m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1b68f244879498db194d90db7f74208 : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/827904f20cc64e0ea5557a9f43083d1c : SUCCESS in 14m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c28ccd697454fbaaecf6ed5a19e812f : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc0cb037478d4b48aefea0d15b3dfe6f : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6aa1b474cbfb4e5180e5981074e9c719 : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ffaee34e48be414888044e7ef1697252 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/69ab64f5ead5458baf2c67c5d1ce2591 : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32b799a21b0a42f5a4feb5b9a03d9d82 : SUCCESS in 44m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5a367c6a0490408196e46d1417658ba9 : SUCCESS in 48m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53995996353b4d77b85c017926c755fc : SUCCESS in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c44598b4fda45f7846f182c464fda6c : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5928a4abb9b14c9e802e891e8b9cd44d : SUCCESS in 17m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6acc74ef0104dc088a639890afd9919 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":28},{"id":"7e496911899bc486f304b0a3031471a078977f5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 06:24:42.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25410fb129e4957a6201642a9d100bf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/860d80a572ee443f85919d2d818951f1 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8cda0d4af604beaa7debc27829d2ae9 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/94229da5acbf4dbd89a7c79146353352 : TIMED_OUT in 1h 30m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/613e847254c34507abb60512bde47f54 : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":28},{"id":"e9285ee9eadf5a527b6dcc1b1ee9ecfd412105ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-20 23:29:09.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"bf9bc6b003ce443c50d70e6060d6df81d1e40a29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 23:48:40.000000000","message":"Patch Set 29:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a76b75816ce54f1383960aafee059d7b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c5b6c7bc90f44aea6cb93e47c0125e4 : FAILURE in 7m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bcba4d7ce72475fbfda4ff7b9aff308 : FAILURE in 9m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6199476ed8ca4d69a9420462539a575c : FAILURE in 7m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d10b9b2933804c859bec3b2465e570b0 : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":29},{"id":"baf0404cceaf911730f1276cb51ed12fb2949e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 01:34:59.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cef6be649964421b12bb677764aa674\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582c92666c1745dba3fde9786b4cbc89 : FAILURE in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/03294cb00a6b4f268f24346eb29d0678 : SUCCESS in 50m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f41ab49f78842b7a738624d0b30a8a5 : SUCCESS in 36m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3bb6eea0f034432934c1c09bf59935c : SUCCESS in 17m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0791182c7ae846daab6ad0032d6672cb : FAILURE in 5m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49793f4f779745a49bb877d4c9d581a1 : FAILURE in 5m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb5768fcc0824525b6d84d44f85fef03 : FAILURE in 6m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0e2acb644f14c849c44be93bb378af9 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0539233c64a940929631d771a8203b53 : FAILURE in 19m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4afd9a80f8044559ab3158d17b35a0d2 : FAILURE in 18m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/37b9a693a28f45c5b53766ceb5e2425c : FAILURE in 6m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/34ea9ca691a740b88297a3e0cb36ace5 : FAILURE in 5m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5c2a33c3932942ebb8fc663f97348b3b : FAILURE in 6m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/036735652ed744fb9164472b9af99124 : SUCCESS in 10m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e053f3ab11a48acad7fe4c06ef6e072 : SUCCESS in 22m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bc04c30ef694134a54a122578517bab : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7718f97bdc4a44c4b4f7cef57b6b8234 : FAILURE in 3m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb752cd6fbad4d5483439a97ab67ab47 : FAILURE in 2m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a15a51b186c146ae9d98f0fa53b9d68d : FAILURE in 4m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e769b435dad14557af6bcd869f982a18 : FAILURE in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b322cbc50795453d8aadd2e1699902c8 : RETRY_LIMIT in 4m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/176bf2f0f89a42ccb45115a26ef08f5d : RETRY_LIMIT in 4m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f6845395be14a96899baab937ac97cd : TIMED_OUT in 2h 00m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b226807c7a7e47d78fa543d900988f2b : RETRY_LIMIT in 11m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c554cef59800468a862ad7db191985f2 : RETRY_LIMIT in 8m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c32bc3709f6483697bc1c83cded4132 : FAILURE in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09bdfd55d22c4b2aac020bf9340f31bd : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd50807d6c564946b8cdf2374e22b5ce : SUCCESS in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35a9a18e9d5947ddb54e6c43e6d11b32 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":29},{"id":"4bea1078a988d992b21306de3d8f6349e69be426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-23 00:06:46.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"58ad8635e89b3a79dd8e94948bd707dfed5b22d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 00:20:43.000000000","message":"Patch Set 30:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af7f29d045d43baa6f8e94a8e4804e9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3565c533d3264da5a75bb88e13e5cc24 : FAILURE in 7m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e8eaef4d770414fa1b4637c66646f87 : FAILURE in 7m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6221b1593b804484bf4efa9bb44a1a09 : FAILURE in 5m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7626057f0d0b49f094593abcadb4865d : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":30},{"id":"7b2e9ec7f409b670be19066706f6d41bc34601d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 02:20:44.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b756e5202bb040ce9f2e96047da98d79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b6070ebb5549009f689ce0d409c5ba : FAILURE in 4m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/f5c6631057b245118b17087f81f0c69a : SUCCESS in 40m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b865285037a4e6c993182a34b9814dc : SUCCESS in 46m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e697d9f1c08e4fac8ae3c0576f00af6c : SUCCESS in 29m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ec952d2625d4c28a5fb0476d905c5cb : FAILURE in 5m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae923db447f947e882ac13f1046d07e0 : FAILURE in 6m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cd6f71b906f4413b2b11e368fc92bb0 : FAILURE in 6m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4cb17b0312849d2adbaefeae0d596c1 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b718ca3566d1467db15b6829259f3943 : FAILURE in 4m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69b2f9af257447cf8968f89269f3a310 : FAILURE in 6m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd52da7b0d4649faa5aceaab4cd361f0 : FAILURE in 4m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d308eb3a06e240e0ac00fa8a8a34d890 : FAILURE in 4m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e5ec63e0ed44faf99f720ee4f692f2a : FAILURE in 9m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d93a35a457c745388ec987d8ee3e9b53 : FAILURE in 3m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d39058ab9ba34c15819798b470e33f1f : FAILURE in 3m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d4195a0128944d3966b8ee8993564c5 : FAILURE in 3m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/327e168cd49f4a8b9875334f41db4895 : FAILURE in 4m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/014be5470df64c15ae6c1fd11b36dcab : FAILURE in 3m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5af6dd0b7351433dbe11245c1fc3f895 : FAILURE in 3m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99ec7ba1188741fc9a63dbb798b9a19a : FAILURE in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4957396d954b4c7180dc8027f147b9c4 : RETRY_LIMIT in 6m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0be51d0d1944649b4c5acde74c232b3 : RETRY_LIMIT in 5m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa6ad61cd8bd460f81f678e5068e98ec : TIMED_OUT in 2h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76c0d3e018e6498cab622e1edb378f43 : RETRY_LIMIT in 19m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b12da77c38d4ac78099a52771b2fb63 : RETRY_LIMIT in 8m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f35d0fc79b904ef2b38e48b09716acd2 : FAILURE in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c5d0aff7de4f97946b761fce9437bc : FAILURE in 3m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7da7685cd424eeea0399c1d9835ecdc : SUCCESS in 15m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47cc3a37db3147678f7ee882c5c994e4 : SUCCESS in 43m 32s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/utils.py","accounts_in_message":[],"_revision_number":30},{"id":"d8347f2c9d766cb4880970612f0554592a595474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-23 23:21:32.000000000","message":"Patch Set 31: Patch Set 30 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":31},{"id":"aeae0fd3b7c8bbef167dd11a7003e2df617e7267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 01:28:56.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77ed1cfa178345b3a82fc941e167f510\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d0e95b8a2a480f99d0a6d36ed65491 : FAILURE in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6ecb3626495e4556bff1353e9f2c9460 : SUCCESS in 1h 02m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2844ca64da9b46cba00919d0170b9529 : SUCCESS in 36m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2927c6b450c4e43bfd994182fc5536b : SUCCESS in 30m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5daa3584295b4dbb94e65a4ca848862d : FAILURE in 4m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/325a158a04f04130bb0892e7ac163f3b : FAILURE in 6m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c798466b1ac40f5b7d07cca1fbb9adf : FAILURE in 6m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e40e6cd7ba7c4cebb4fa3ed4f7cfaa14 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62ea68baa02342b1889cd0833ea96990 : FAILURE in 4m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f9a9023d0411422083d13a93efbdbc60 : FAILURE in 4m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fec30ccf21442087c771533078ead7 : FAILURE in 4m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aaf8e6995bee4764ab8d379ccdba510e : FAILURE in 5m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c68a49543f124e2a99ff681d50f6bcd8 : FAILURE in 8m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/291efea308184e37939f61d0acf169ac : FAILURE in 3m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/00dc54a99b6b4d63ae633286b401f570 : FAILURE in 5m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f973357e609640eb8444491bf02dda5c : FAILURE in 3m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4c326eb67b7e4ff7b49e3a886a116f23 : FAILURE in 5m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dec2fcf17fb542d8a0904bff69c171d2 : FAILURE in 3m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1724b51283394d28ab78ebfc8725ea69 : FAILURE in 4m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f91f90e46b784a4f95c8086881a3942c : FAILURE in 5m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d70d3ef663144b587c7bc2f43294637 : RETRY_LIMIT in 4m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/773b47f73c25459bb9fab8f0cc9bba47 : RETRY_LIMIT in 4m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ceb3dabeb5e46778e267ef91632bb90 : TIMED_OUT in 2h 00m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cd3eb8de183428cac05b412f2c05c3b : RETRY_LIMIT in 20m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2df8b0a52f414594a24aebc89367e170 : RETRY_LIMIT in 19m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94ea32d2f58e47dba845bc9982c5cedf : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b873098050444926afb4a65a99f4c9e2 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e0437533cd73419caf4328865f8258e1 : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/873b13fa0ec944dfa7d90f4622347e29 : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":31},{"id":"a9fbaea2ffe4721aa09df1109a86fd92748b569d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 02:59:59.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c7f92c1b454fd58ad5fb60d5466af9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/698cab5428da44018a1f888059bdefb7 : FAILURE in 7m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/915a865b1a7149dda60065ad5c279d85 : FAILURE in 7m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbbac6e1238549f686d2929f647b7f87 : FAILURE in 4m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a0b5b437f17434c8e698a61a37ee1af : FAILURE in 7m 21s","accounts_in_message":[],"_revision_number":31},{"id":"d231be6ff98bb135e7c64cb8001f2347fdd50816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-24 15:49:22.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"1293444a360f91be64a847ec4f9f9e4b3f893023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-27 23:23:07.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"b9cefa232655f9adfedb7a9ae22fccaf5c1c69f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-27 23:33:17.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"d3db869829e8f43e5a7db3cf7559615e9da4997e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 00:41:19.000000000","message":"Patch Set 34: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ea35772a8d64666b8e3aabc1027c383\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af50a36f03e34350aeb57eb4cbc5733e : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/953cb695ba3c4bae800940b68e65f079 : SUCCESS in 1h 02m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2f305b4a7d564d07b83bc47d581ac7cb : SUCCESS in 1h 00m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8147119edaf4073ad6e50506b5e01d1 : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83070d977ca24595809b8909dc882c6c : SUCCESS in 30m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0e64a4df5fd404fbdb890f751fe67fd : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b81e0c8f5c8d4c2da8effa52e858c9cd : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9104b78e1a842b5b76c9ee899646ad5 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0da8c26ca3244c59927a77746a6fc6fc : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85290b589e2549bc9d8eade26977f408 : FAILURE in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2deeb82c76a4e6bb6392d3ace4a02c9 : FAILURE in 14m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0bfac4a31042c3be8a643b065b2213 : FAILURE in 10m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/77d479e349634afe94126949a9b9387c : FAILURE in 27m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e715435c248644c8a5b67c58e7391b4b : FAILURE in 20m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eeef9daf5ef54c4c83fb2856a08f9759 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1c2c4ce516f84b33a1544bce20702637 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/317510b2574d47e1ab7e0d6cb377c035 : SUCCESS in 10m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2b805c2f8f3c4e83a6415dee5c34ea32 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1bf53d5a5b6a45c0a2d25bad564df580 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e77c230baec94b9ebce6ef28384bd959 : SUCCESS in 8m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6dea078035044888b9bd01eb3d5512e5 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c28d2ca855f4b92abc932ce20f27521 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/21eed7af9ef247e2b207c69b5b95fd03 : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e60b87403e14f16a1be5db934f8fe58 : SUCCESS in 50m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afc33bdecfbf4bb99fe2ed526bf73ff6 : SUCCESS in 46m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4394c493d3b487b8914ca58a2c06b13 : SUCCESS in 48m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2da6fb18179b4d4781c0e5bcf0c42daa : FAILURE in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592d1754ae974c338c17e5ff281331ab : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e0fe24be3134a5187d49a8a499106ca : SUCCESS in 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a81d9c78ea8948788f45218c365fb149 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":34},{"id":"6b0f3c35f5ea8367a56acbb3312ff82e3b17b8ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 02:22:00.000000000","message":"Patch Set 34:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96053fc48cf6429aa0eab45025d12580\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081526d46901457e9bedf5d6ba21a5cb : FAILURE in 53m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fe139277222414dbf3b6393277723d8 : FAILURE in 45m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74f0124dcbc544e4bed7b85852c1088a : SUCCESS in 1h 19m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7fca55b7f3742aab8259f348a67fdc2 : SUCCESS in 1h 25m 27s","accounts_in_message":[],"_revision_number":34},{"id":"e76228caafc0e9b6626799cce9e34114a8506e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-01 23:18:11.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"fbaeda25ee9c0aa141e9a552f27c46e326792e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 00:29:04.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68425783fcd74f4691f5aea06b835e0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51aa515ea2344466a07fae1f0f9910d8 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/241cc8bb62124e57bffa1c17ec8dd90a : SUCCESS in 1h 04m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44c159ba29bc4cddb9a1bb556d9172a4 : SUCCESS in 52m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6dc37d9bc15e4d4f984c2cf49eb8577d : SUCCESS in 42m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0ba76a7afe5405e8858c49b35c280c1 : SUCCESS in 29m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c03ece79014f433a9106c9dfbd4d4e1f : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb5959ae0f7144b1a142aaeaa39f26c1 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfa5022b6eda4aed98c33470f75df15d : SUCCESS in 15m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f72b8674604944fdb37dccf737606ab2 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d6ba8c4b7f43d99efed41a904aadd4 : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a871ff2a3934ecfb150be7d4cac71b1 : SUCCESS in 18m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd237c9b7a574bed828c23e3cee0eefb : SUCCESS in 17m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f7a2b1aba35c4df68d50c4a661f2bc10 : SUCCESS in 19m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9109972cc494fc2bc516627b516757c : SUCCESS in 20m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/305e8d28024f4c698476aabe9e82d2d0 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/364dc8ce66e74d81a30eb2e0806fc040 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b47fcf865bf948b1a167dc57b8a1816a : SUCCESS in 12m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/36a45be71cac45d9bc6bcd6460732f5e : SUCCESS in 11m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a22175eab62d4b278560a832522d52c9 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74204f9631d04c03b9669f23fa604a34 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a96eaaaf38a448a695f588abb42567bf : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/628854e1737242b5a54f2ac6cd486aff : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6020daaef16c44acb44bd9a399965649 : SUCCESS in 8m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b302960addff4b7385f9ee1a45d496a2 : SUCCESS in 35m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a35315cee2de48cfa4a673f36864aeb4 : SUCCESS in 52m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4cfd2a4ccd5b487e86eb4181f4c451e8 : SUCCESS in 48m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4979cfae81594d129500098c196a2706 : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7516390d6b794b8c8a70f3acdb3a1288 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed2c95beb5434e29946c4977572858e9 : SUCCESS in 17m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a92e4d6f2d420fb6d37ac5660b403f : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":35},{"id":"c05b1e27dcb5af4b87cebbf58c8020ce217ec25b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 05:32:20.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89990ff73de74dc29bb591cf25444542\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/001e716936aa4c45a4659b3ff78d2f52 : SUCCESS in 1h 19m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aabd12b9fbf041cbbfc637444d13377e : SUCCESS in 1h 23m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea41eb3ffe7b4a3b9f18e7d23252b4ed : SUCCESS in 1h 26m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e3adbd35d504bfd8dca0d6c2f9a801f : TIMED_OUT in 2h 00m 52s","accounts_in_message":[],"_revision_number":35},{"id":"6704d3362b027b30e55bbf6e3292e04fb1e34651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 20:10:30.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"817f6a60cd4a0a47b1429684ebe43e740f49cd2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 21:34:00.000000000","message":"Patch Set 36: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/880a0e9539654b86ad7a24cdb360db48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2093c764b6a14737a3a279367e7076ec : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/165ce72014ce42a6bc892708f2327eaf : SUCCESS in 59m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d7e2fe1f9cd44ca801f4dcbe63e9e25 : SUCCESS in 41m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fefdd8521bb84d71b842d9a43fa49d86 : SUCCESS in 47m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eb9ec369fdb4b85aed312347cda7a2a : SUCCESS in 1h 04m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bf7e19d169d4eba836e2d3cca99d385 : FAILURE in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bf722ef0a7e456d9ec0433c894da98d : FAILURE in 20m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3875cb3197c740389facb8fe74bacb5b : FAILURE in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b4a5118668f452e9a1d3b0c42486017 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bda067225e041c7a72cd6cfe217ec44 : SUCCESS in 20m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4eb9b88db274531a258a35c2eea0fb0 : FAILURE in 10m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bda28644c92e4916962c974293470daa : FAILURE in 17m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ab90a3142f524a9ca0759723a7a128fa : FAILURE in 20m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5825da3d9b3547c4994803c098c76e68 : FAILURE in 17m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9222fde068c448fabe20c8a42a36a7b3 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adb960d0a19440548354eb0fd8d83f77 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4761e0538d32420c86349dc2d5ea3680 : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/326b93da2d7b43f0affa6dcce642cbac : FAILURE in 14m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e0a71c76a20c4185bdeeeb5e3891a0c2 : FAILURE in 12m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/abf358fe45f64224adf4c838c8af80b9 : FAILURE in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ce4315d57fa4c199e1ce0b91584b015 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36f0e9acb8894244add4bb9429c3e453 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/98b01c57771a42d7923442c6f21b0e3b : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4941114a7d6e465a9b613b157c9b2356 : SUCCESS in 46m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ffb3d27a4fd4127af3b538e530c538d : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aeb62dc8052f4b2287a08025ee99e732 : SUCCESS in 38m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6abacd6bd64a4c0bbcf3cee07241b1e8 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea0dc9c5dfb46bd9078f65d5e8bbb57 : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac08080424b0453899773edd6bc8f871 : SUCCESS in 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a9046cf342e4da1b14bec0aa796c883 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":36},{"id":"5774ac10feee4a94f4da19a31d4f66a01d2ab48e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:19:59.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea95c2abde945bbadd13863d8a66f0a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b0e40f49f3649f28d7920e14309135d : FAILURE in 1h 22m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0d81d2c7bf74e10b00c82f7f9f90c23 : FAILURE in 1h 13m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4afad1acffc4d59a77567f60cb75b4f : FAILURE in 1h 32m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6522a13d1ac44b98a246a0908f16080b : FAILURE in 1h 15m 10s","accounts_in_message":[],"_revision_number":36},{"id":"97a57737c24abed366a7db95724b78b0187deba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-10 21:54:21.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"4d20f39df521fc4afe767010951d814eb3750e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 22:56:35.000000000","message":"Patch Set 37: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed94cccea00b41a8bf8f441e947f7a2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee702eda9c445b3ab455c507c39414a : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/cb4500733b0f4c32a5cb05dc1ce0dcac : SUCCESS in 54m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e159c0b4301416a81a047afd25b39fc : SUCCESS in 45m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7342d5e1f234493bb04a6357ac818a6 : SUCCESS in 52m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e91684d60614c1398047c1d5fc0664a : SUCCESS in 56m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54c8889c8c1a40a089fc9326a8be3931 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35ef2601b89d44bf89ab400c33bb3879 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24d6b856b07c4628892cc8c7ae4fc7bd : SUCCESS in 16m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06939c5dab284943b8e3713eb280c48d : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1fb2fb1649d94385816c0316a5e63a5b : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9483212ea004b00b1cb39188bc51e1a : FAILURE in 16m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fdf889bad3d1496aa067621748c7a2bb : FAILURE in 22m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/df1ff5271b394385ba1d166454037cae : FAILURE in 27m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/47a75137a22c495ab32721863b732495 : FAILURE in 26m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89e9a4dbdadc46878e9fe0c85e9a73b8 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f35fb37613c4bce876801b0747a8575 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52df5c6c36ba47058ef695e757b87e3c : SUCCESS in 16m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d48ca43412c84acdb5be59841df67b5e : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/39392b27f92a43d0820cfe4222de1452 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5462cdb378234fafad77b757e91fec02 : SUCCESS in 19m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d6198038dfc4cf18d06ca22d4383e7e : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6a2748b354044e58b35859c8fe283cb9 : SUCCESS in 13m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/86a52b055026447cb996ab5edac93544 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d094c17f63d458789ebc8e704be6ff0 : SUCCESS in 46m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66c6fd83189440ddbd786427457b1b75 : SUCCESS in 49m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a1435047b194040bb7c7b30c3ef601c : SUCCESS in 45m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a01ec8f4d7f4efa9d6ddbe4edd3406d : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2a96a7a8a44aaaa318411588ca00f1 : FAILURE in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e93637d396264951984cb834c3a13fa6 : SUCCESS in 22m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d8bb6266e0d4636a0678f3c40d6e705 : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":37},{"id":"38dac06ab374b228d5b431bf8feae4ffa0adfcf5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 01:20:30.000000000","message":"Patch Set 37:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e3bcfebdb54b20989552e039aad092\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe72f42887d14464bf26a364714d9993 : FAILURE in 1h 55m 12s","accounts_in_message":[],"_revision_number":37},{"id":"6c4520126de35fe9ec5441f8a1e4120ead025772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 23:34:45.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"6289d88c4d9614fb499f84cf3366d5b89e80667b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 23:35:26.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"28f77e21b3f55871e3a5849cacc995596a666b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 00:46:10.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e88ea60728874181a3f17d26d6fa03a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ff04e2cb174866919c4055ca139dfe : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/095093da435e4e4780ba82f350ec9c8e : SUCCESS in 1h 02m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6fefa83ff924f11a487a696feaf5b60 : SUCCESS in 1h 05m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/425cdea70c9142809efa23b58ab418ea : SUCCESS in 51m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b19e7ad1f58b46d5ac20f2ebbf5615b3 : SUCCESS in 1h 04m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55b7096db2c94d5b9207d5d842742a27 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e9979cb15f0437183b6498f94b5dc40 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8441f0088ef54fb78d1b495d0f6fb769 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20ee0b65db4c43f8ae8b7b9b0403f533 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b7e599b5d4d4fa4be5973a00dad7be6 : SUCCESS in 20m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de107d1e0620413c8a2d1fba0bcc4edf : FAILURE in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2dcbf38119c4041ae39cc6c810add93 : FAILURE in 21m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a1e305c5b0a94efb99eaf696bae22043 : FAILURE in 14m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/334b565dbcba47949365a7fdb7a0edd2 : FAILURE in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02b7ae1ab78647b59a59372dcc302b56 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11b1256d399143a0a87f39a5eebace92 : SUCCESS in 11m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f96f2a823d0648768d40a7da484d72c6 : SUCCESS in 15m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb8d2dc3240e49ffb049df00dcbab5d1 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cdd716ad2a1345a19cf6e89139c01a7b : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7eb74dbb8e9d4287b6c27e3ab337ef67 : SUCCESS in 12m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6d2ba47a28134fb1ba8f3eeafc820671 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f14b003f4d74ff086a456f9a4c9f062 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9af0f2124b5745cb8135a18483f51a07 : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ab2569974c940519c1ea68a99a7189f : SUCCESS in 37m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e465385245414e7b935a017a65f6be89 : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41075f8074984d20a247ce3edae7adbd : SUCCESS in 46m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9828dce50a934b33ab0e7dcb57b56895 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f219c8783edc4c949a8a4e10f4e9dee4 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/060c39bc4e8e4b5e8b1ef18dc55fbb9d : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/829c367977ae4a599f2b67885d7213b5 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":38},{"id":"b09e81d225dde82a8d18e02f09a3ad302e22a0c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 00:50:54.000000000","message":"Patch Set 38:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd6468b9e2741e0b28dbcd97efda072\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbe4f81d4b98432fb252c2efe6934c58 : FAILURE in 1h 14m 25s","accounts_in_message":[],"_revision_number":38},{"id":"c0bde0ad89c24d3e85c097b1c35931eb61798aab","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-21 16:08:01.000000000","message":"Patch Set 38: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"23fa1d9fa1af40bbbb9e83abeb4374d6494fd864","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-24 20:58:58.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"cbff6f8c9c7aa3ebde17bbbb82dd2a1794173c08","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-25 21:31:15.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"eb21b01e1f48ae8453a33332dbbda4cc7712c166","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-29 23:28:07.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"6186cf03900ad3729782f95596428d016a2378ad","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 16:12:56.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"d4c0bef01cbbd0f5a587a3734ede5f82b86aefbc","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 16:15:18.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"fd249845d6b9e0465b6d3d233ff2f8d7fad9ba2c","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-30 20:04:37.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"1f91dcc34018cbeb50978a6639710e188a2fe949","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-31 16:05:43.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"c28c36b84a06cbc16570fdf921d27032884cc1b1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 17:01:17.000000000","message":"Patch Set 38:\n\n(4 comments)","accounts_in_message":[],"_revision_number":38},{"id":"32067eb9f0ec4b5609daa60aa7702cd80c2e4c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 19:28:55.000000000","message":"Patch Set 39: Patch Set 38 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"8e13ea5518f4fe9e6fe09df8f5d95948f29fe91f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 20:32:30.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6a9f2ac410b4bde9b9943660dfabc31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c09d3cfbde4b79b586b1ebef5a4b39 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/1ac6e49067884a9a8fb7c0cae0e3319f : SUCCESS in 42m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64b641b609f049758c42d7813d13dad0 : SUCCESS in 57m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d6c91b20e1f4564b962210c1d35f73d : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9c9d99a6507495b9e6b6c5c9c3cd392 : SUCCESS in 48m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a081659903740b1b7823d593e793d00 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f20f037fa574f6ba22ff55b704e5f0c : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58cfecf441914f91b5bc1c03970d53c3 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be6d39eb8aa84633a919cd5ae7530c99 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c7a36110f5140b9a2703b04b836b3b5 : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49ebbe85d46f408e9620816c462b0b59 : FAILURE in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a24cf54e53a947ed87c9145c11dfa375 : FAILURE in 20m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9841024aa9704f8caa1acea295cc2d2b : FAILURE in 27m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/89ff3de9d57e46dba817589d6bceb715 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a2d8947a7d34bb8a08c76be1c5ddf29 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/797521b245144b01b10eb3bb5cf7ba0f : SUCCESS in 18m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0cbd3d4e2d4b48f899fb6a6a901719ae : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87f6102d3b6c497d9067dbc763e82c52 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d9e411fb32a54820a5c575e616c1ec47 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c2fc9d7d94d4beca8887f28c14bf65a : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08aeb6b391b249dd91c053e5559f633d : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6c308026461d4725a14d08bc672d43ce : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e14aa4135b1841cbabd75f4503589649 : SUCCESS in 46m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92f43a1f9dd747289859d15aa0c03792 : SUCCESS in 45m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d115bd1940e483da92c9d9c72923667 : SUCCESS in 53m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/901b70feddfd465ea51e53981eaa46cb : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b432a7ed1c6845afa8a627c9699d6a49 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ea69b88149f4a69a2232798f08b486a : SUCCESS in 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f48fa6b9be94f2aa83fbdcd51c5be9c : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":39},{"id":"92056ae4af8181498f2ae6291d4fe7903132b6ed","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-04 21:44:40.000000000","message":"Patch Set 39: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":39},{"id":"6a250395bc0db30e76a6cae92885e9f7c468c6b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:49:30.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f64e32e17dc84cae97f7ff13e911629d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/68f0c6691e784eefb3f78489fb39502d : FAILURE in 2h 01m 48s","accounts_in_message":[],"_revision_number":39},{"id":"9e20c078d7bdb4234fe6aac7c4f16847bb765681","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-08 17:17:47.000000000","message":"Patch Set 39:\n\n(1 comment)","accounts_in_message":[],"_revision_number":39},{"id":"d7603498212fef55976a430c599c257d18e9f925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-15 23:29:47.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"f9694e9b105df969c9288afffe661fdf76cb8c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 00:39:18.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a4689cdb89b410dbf924671c47b124d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25878b94d8e413bbdb34c78fb6e306e : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/de9f26a6516c4a89bc78f30c0e15ef55 : SUCCESS in 34m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d803b59fdefa4e2d89df5c927e4b81e1 : SUCCESS in 1h 00m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d451a320a5c947f1a93cb02fa8bb8adf : SUCCESS in 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/586b9fded5024b3ca2c449314095dd11 : SUCCESS in 50m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a046f8e6a4941d29809fef0d3014b57 : SUCCESS in 12m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0a400ec8b88450fa3c37bbfced93972 : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49656f3ae03e45e7bd5cd9569f58dfea : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/149bc6608a9d453590788aab2842666e : RETRY_LIMIT in 3m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51097dd4991a45b7b621b34ea83871af : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23c0885d0ad5431a94e058862d2344fb : FAILURE in 20m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25799bbc880a4bc0a3a19fcd2697b8bb : FAILURE in 17m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0b087c6b7b6847cb80606ba931cc5296 : FAILURE in 24m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39c5c3b3b8b5455597e88e2cfbe8e618 : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8152d373caa47fc9e80da7e7a57f8a2 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6bc086ebd22f48c8b4f98c6c856b8fdd : SUCCESS in 14m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8ed0c3cee4824ea6955f88098bd45848 : SUCCESS in 12m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ab99a1a4adc46f78f95c8c3d2ad1ff8 : SUCCESS in 13m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e5ffb1e5e7b47f6b42e8eaa5950b161 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e9c6e8bbed284b77b25f20563b96fa2e : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a62201bb9e484d09b6c1cb8032e1a74a : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/794276f5e3014a3483b0863fd6d4f09d : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8e996515d0a4f26bc9495c67729fbc6 : SUCCESS in 33m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1800ae6efe53422e8031cdee8e5a6d54 : SUCCESS in 50m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d619862e02704bd2a01176ce75c1714b : SUCCESS in 52m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70aed8bd08d74046bb63bee15376636e : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf8add3d36c4d0eac8094a77d9109cd : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b290ce69ab2b45b5bad2a1286302dd57 : SUCCESS in 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3631e90b75504ad3af88434772ac989f : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":40},{"id":"b1f0ed468326929083879f03d23eff2511180e33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 00:54:17.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b48580eea41a4364af784b468945ce98\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3151466901e34f7d92020a66346cf1e1 : FAILURE in 1h 22m 33s","accounts_in_message":[],"_revision_number":40},{"id":"95c38e9075c7e49d32ef719d60149dbaf9496dcf","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-12-02 22:57:27.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"a9cf234a1f9236b003aed8027ba7bde06a28f69d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-21 23:31:59.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"92905c9bd3e0fa9906fe488c87be85fd434fcdee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-21 23:42:50.000000000","message":"Patch Set 41: Patch Set 40 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/s3api/controllers/multi_upload.py\n* swift/common/middleware/s3api/s3request.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"042d4315aee3be8154a12187aee3aa56fcee245b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 00:00:15.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b230efb5e047129a01cf73cb3b2ce3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5764e568b47e44248ab9ecb7ddd4cce4 : FAILURE in 13m 05s","accounts_in_message":[],"_revision_number":41},{"id":"9fad9ef10531bfbcc7305604475e5cbe239ad352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 01:50:33.000000000","message":"Patch Set 41: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d3d657f312d497281cd6ecfd4c318d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a31244e2474078bc6392866acffd1b : FAILURE in 4m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/34005af763a34b0bb902af301ce67073 : SUCCESS in 37m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/59d2baf47830442f9675b087ddd57908 : SUCCESS in 59m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7dfaf32c40af495cb08969da87b45b49 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/924d3ab429154fc29e00ffda33373aa0 : SUCCESS in 1h 03m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97343e8d60b24714b786e5dcecc22773 : FAILURE in 6m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5af4b6e832da45e0a93a3732542ca605 : FAILURE in 6m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4275ef781cd3429394c39d8bd7c4489a : FAILURE in 7m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54c7f0380462430aaddc9ed0dd79ab87 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08fa4dadedec4c4998c52ec6433bede1 : FAILURE in 5m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bfbb1514a64a498f8589dc0297ebcd2c : FAILURE in 5m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/43466fc12ec84bf4b40b5a0170a6de53 : FAILURE in 8m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/695fc4b235314e43bb97650e6a6818a0 : FAILURE in 3m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/47ce8bfb1da14e0dbc48baa7757d1fbd : FAILURE in 4m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04945a73e8d3490cb8773f08d2b5ddde : FAILURE in 3m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/026e3193ccb84461a8d840b4b2f96428 : FAILURE in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3c964829c9b4d43b39c2744fbcb3cbb : RETRY_LIMIT in 3m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fd15672e0cef4074a54a3a211bafba5e : RETRY_LIMIT in 5m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00a5da90541d493da22822f7158d269a : TIMED_OUT in 2h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c90d3b5f1af44a1993969c691e52d1c : RETRY_LIMIT in 8m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4256e89f5eb54567b44c5f8a149f4de6 : RETRY_LIMIT in 21m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/847ae3bd91964f8180b1bf296bc50e06 : FAILURE in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f9bf5c10504399ba8a9ae3883f3732 : FAILURE in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a0b911110b24759b768f4cbab0449ae : SUCCESS in 19m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebc697a943e34d0fa339bb836dbfb32e : SUCCESS in 41m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99cbf39c312e4c678072366131a960d9 : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":41},{"id":"c7ad0a765d55ed4b4c22ee299018460cc9c1ecc3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:43:26.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"c9aca47854cd36abbbe5717fedf4587d8de61b61","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:45:52.000000000","message":"Patch Set 42:\n\n(6 comments)","accounts_in_message":[],"_revision_number":42},{"id":"2ded0eb1f528c32ce330668453bbaec6bf82171d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-22 14:56:40.000000000","message":"Patch Set 42: Code-Review-1","accounts_in_message":[],"_revision_number":42},{"id":"dbc6025ad9cec7e2e03de4cfae39a7cf4e8617b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 16:15:29.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5158d99860d74afabc83d96e8277dfe2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c5f47888a24001ada83d0f75aad80a : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f0eb9deb7074460db7e84d60ddb204c0 : SUCCESS in 1h 04m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/503c9fc9ba90438282cac6c2fc863ecc : SUCCESS in 1h 09m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df9c875435c147bfbc7cdba7ba043cc2 : SUCCESS in 1h 07m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33ff53b3a6db47c784b362fcd41fe866 : SUCCESS in 1h 07m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7570a68ae73640c59e42c1151bf29742 : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51c5b8f97fca48aeb5b6e69b4a33f109 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eaa9ad12474450ca52ddf49c43cbfd7 : SUCCESS in 21m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1057f58459f8495f9b654c8acceb3c20 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b1efe47809941d8beccfac6450e1d28 : FAILURE in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eec9b2d3e3ed4ace89223ecb2549387a : FAILURE in 18m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5688ae3f32d843289905e1d4386a7643 : FAILURE in 32m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/00b45efccd774d34b34a1c5afef06b24 : SUCCESS in 11m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/34298a5bac584dc797a39708b6b814db : SUCCESS in 14m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f6168cb6ef1417eb524545e7c16cfc8 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38c2c9492ce74b51a1c660dfd171b736 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c9e0c4023bd49128a2de9e3d2413da8 : FAILURE in 9m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9657994a37614ea7bbadd3e58af01685 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24ce5fdd50d04f0bb71383824251ffef : FAILURE in 56m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/83dd6d8222024450ae1485b326a4c76b : SUCCESS in 43m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae0f551c52914b28bbecff6502019c4d : SUCCESS in 46m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d20c966d6394f83be21623e4e625e26 : FAILURE in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7893bae56dc241a7910a0a7788745f7f : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f93cfe6b969b42da9c04d6a2b690c54d : SUCCESS in 26m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a27db44d347b4f32b5e492b10923525c : SUCCESS in 39m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f5710f5314948dabb49f09cb6449d89 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":42},{"id":"0df9c3ea713c3ea7e3c41b13270284493424f501","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 17:09:14.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c487d74c88e43289485adfcff6d5a8c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d56938733e13457db580067a1ec96087 : FAILURE in 2h 22m 45s","accounts_in_message":[],"_revision_number":42},{"id":"32a994e3bdcc3537085a81880dcc2d21cdca033b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-24 21:41:01.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":43},{"id":"195762627bfda078784c4c10dd1162f1c4f89828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:53:06.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f1b96aad904eceb10c35df8b67d3ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11425155f2c54f83a0a90dbf812ee6d0 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/edbb1b8d88424aa39a3024a037fc97c2 : SUCCESS in 52m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/647025da80374d4da3d03146ed130ad4 : SUCCESS in 48m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6d7af3901884cd09d8019bd4e44f7bc : SUCCESS in 1h 00m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2038b479dd744fb94c3735773dee84d : SUCCESS in 1h 04m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acaae6bd69bf442fb6083e31e3bd0556 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19800a6f97344597b7817863550b9e22 : SUCCESS in 22m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8cc7af236e649fc80b77a42c9e7ddce : SUCCESS in 22m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a48bb35b54cf48449815d91cf8f3509e : FAILURE in 6m 28s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa0cb554f2ef4369a9530896d754e518 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f4063b5267841a28f52adc6072bd538 : SUCCESS in 21m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/28acd5b9aeeb47649ac278ffee222bd1 : SUCCESS in 31m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3976c733e2047278d7c1ed5a45c6b09 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/79fe19b1a997499da03d75186dbb9ac7 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e6999e6250f4b76a8a89e94afb20768 : SUCCESS in 22m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e50f4a07d1d4b60a05e43f42021c82c : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88688d60ef234415baa2bbe84947adb4 : FAILURE in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9658f53d1c0d4893a7990c618e9f37a6 : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db2a9a4db8f6424998f9140bf91a091f : SUCCESS in 43m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78319aa6c5c842afb8126fc392c1969a : SUCCESS in 35m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2992837cbbf42f190a64ed2e1842b76 : SUCCESS in 30m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5436c56aeb7a41759164887ffa13a434 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22e9dfb19e9d44238b4acd42c1496295 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79dc0daaf5e34c76a9cbbd35bb76ecf6 : SUCCESS in 27m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2c4df059de649639ef402286f63262f : SUCCESS in 52m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc2bfca9c3024456b4a1da48caf1a59d : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":43},{"id":"26aca5b64346e82a863c46d104e5eff251061ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 00:21:19.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350427a1c1ad4724a2c6f136f26fff8a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5054c778e288412dad78bb5847870997 : SUCCESS in 1h 24m 56s","accounts_in_message":[],"_revision_number":43},{"id":"2ebc49912be3a0eb74c5d379e6e0a0bb92c59df8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-28 16:22:50.000000000","message":"Patch Set 43: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"59bde3779f6c5b0b950f49b538aa0b3e8380df59","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-01-28 20:15:41.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"b521d1d8926484ede00b1cd682f679b64e00cdfe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-30 19:57:49.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"609e0a2864d76254e7e2b2190b664a04cc8d85fa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-30 20:00:09.000000000","message":"Patch Set 44: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":44},{"id":"3a1eb63c5d87a84d016429202dfefd4e0f2b4269","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 21:16:52.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c466a0eb47f143fb91074f7fda2eb201\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1fb01d90dba34238bc969ae47af9b415 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":44},{"id":"985d99f5a12b8b35a07353fcbafd9d48a69c1275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 21:30:28.000000000","message":"Patch Set 44: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13c57fb7ab1649b9b53dcf7c69dda433\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dad14c9dfa145daa688e332e09ae105 : SUCCESS in 5m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/9496c997d30047b4b6243058919ef677 : SUCCESS in 1h 04m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ad736609f534253968c74b17671fba8 : SUCCESS in 38m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0dec5280ee40468f8f710555c5730d50 : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3a898724eb94260b85a46a8cd6208bf : SUCCESS in 1h 25m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b7d5b9d682b46478e14bbd284b135f3 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6664481934ed45f7a2ab1bea6ef9fda6 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57c2568a010647a5ad74d72d718c7f7b : SUCCESS in 28m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1a3af1279dd47be863af6c65779cacc : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc738ae2432461ab3f46051035eb890 : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/47a52237ddd741f693710bb7a2eba45d : SUCCESS in 19m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/514f2875b62a4d8695897ac023eccb66 : SUCCESS in 25m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dfd46a83874b4a11af7181f9b10b5fae : SUCCESS in 13m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9525773d410341b696a27a1f60ee3c30 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed3e8dc51c8d4ae9b8b2a7af747f0b75 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4c27ea45aa2481fb1668e4d5a6b32df : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97838405926d4921a1b88b472bf197bf : FAILURE in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c59df09c524c4c64b7d7010de68fd1b2 : FAILURE in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c6115e7d9f143c1a384736203a780b3 : SUCCESS in 58m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c06ceadfb3194e53ae22c3a721af58e6 : SUCCESS in 37m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2669a2c89d5945e18735b5258119c866 : SUCCESS in 55m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea22a8d38c0c4251b399c598f6c6e510 : SUCCESS in 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a95bf0a91ec4fd1b325ae0290762ed2 : FAILURE in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/598b583971b74ca884af529186a7994e : SUCCESS in 20m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/250b8f74ee1d459f8ed14fdde33a68b5 : SUCCESS in 59m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7145ef1c773b41609d5124dc5ec819de : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":44},{"id":"9534d0ff11a84bd78f5dc6f8b9008ef750036624","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-31 02:51:28.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"907939a787c3c7ea89b1b36548f59f244a667654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 04:04:17.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da72475dd7d49308f1b9071f75591a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e266b773614b4d4dbb101c8aac6e06a3 : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/bf4a55fb0c32453cb36df32b9361f73c : SUCCESS in 1h 00m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d57ef14ba8a49ddb5bf1b8c214c29f1 : SUCCESS in 50m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18e5ddde896c480e9af147fa4e453bca : SUCCESS in 49m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/481812907e5a42cc92719abb4de7b459 : SUCCESS in 1h 06m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2631ad21c9743e19a6f0ce088c0e5d2 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51b85e380b2d478098d85d0b10b65819 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05c3dc47d0034c7397a76ba4379052c2 : SUCCESS in 20m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebf6cfb8708647d58ecbbe6d7d7cd8ec : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1675ac055bcc4c0ca4928aab57476414 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11188c0e1cdd4a258bc6d9876893659d : SUCCESS in 17m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a53eb017bcd3495f8e26553c33f95a99 : SUCCESS in 38m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f2dc4687770547dc8755ef6ec2a96240 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0b045fe2de654f288fcd067cbd07f13a : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/07ddf4b9349043dd811128515c09f422 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0439734d5c24fa38b7e5f11e5399a6b : SUCCESS in 10m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6f5cf2fc9b14c399323bab010f97e43 : FAILURE in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e00f9d9fb8e455889db383455f4cf44 : SUCCESS in 10m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/73a9f801de3f48c69a81bc9bbfa40400 : SUCCESS in 35m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec9323513a7c4aba8cfc4b07db1f9bc7 : SUCCESS in 56m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e5dd0a8f4d84adcb5c434f8e86e62e7 : SUCCESS in 30m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf7d614409f04651a42e48d0e8eddc09 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9483ae67ffd840509d023e74369f7307 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54da9a8f98b74330832a2713e7d300bd : SUCCESS in 18m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4675726b69c4144bda6bf26aae6f404 : SUCCESS in 58m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f85c4609b5f4719a32f577c430498f5 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":45},{"id":"c3a976607cfc77a802fe3f8b045cdf786a6fa4a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 05:39:31.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da3be0aa504e4fa88c99c25d616494a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d36b722cc15b43c28a06aefcef7e7493 : SUCCESS in 2h 45m 17s","accounts_in_message":[],"_revision_number":45},{"id":"bea747187137f7a678beb5e1c0b6ad243527a31a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 19:46:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":45},{"id":"b4baa74f4520868352d47f731eeb31ca364e219f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-31 21:09:24.000000000","message":"Patch Set 45: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"9d49fdf2ef7cd6714ee6c25dbb628acf8dc82865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-01 02:38:20.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"7f14c3f67abb826157e0a2ff5bdd2b96f350b205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 04:04:47.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00b0e53bdc974439a6bc130b59b1ad15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db4368661be4da4bc90fc55bcd0c436 : SUCCESS in 22m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a340d58944464fb09d263b6eacf70645 : SUCCESS in 1h 04m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee5c3b1e129f4c7980b67b47fe198806 : SUCCESS in 1h 09m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a4efe580b94f478c9eed7c6cc972168f : SUCCESS in 1h 14m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69f44b38607d4be29c93b7fec897f428 : SUCCESS in 1h 11m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e75b5ea1de1d4138bd4ca07f0c418d43 : SUCCESS in 25m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/329e5c2e792747c88a31b05cf8842788 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/304d51479ba7476cbb5c83ecebc0976b : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9a38b29bf9b443769701c2fad646bc5f : SUCCESS in 8m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbb1eb729a62406ab2a4c2c6be6beebe : FAILURE in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55b00ea54f1e40058f4e7a5ae31c9b20 : FAILURE in 30m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e43292b15f44290ad0eddf23862ba90 : FAILURE in 31m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e7761dd444c457a965e57cbf098fe92 : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c03264b36d244585bdd0a66604f6f045 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/664ab55cb6be4815af5f8338d9e571af : SUCCESS in 22m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57a362edcf0e4753bda5694c94ea55e8 : SUCCESS in 17m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de53613bf6e34d8d9823a27fefa7f1a1 : FAILURE in 16m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b5ca093b80564fd5a431cbe0f206a255 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dfb48d2bfba472695df9f8c5421cf49 : SUCCESS in 55m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/189f2afbbd42440d8d15dc8022e68f3d : SUCCESS in 45m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/006aafb739064d8492b11d587fbd460e : SUCCESS in 45m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54d134bbca7545ef83f2d3e533d6fcf1 : FAILURE in 28m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22267ca487248c3bda567e508a545af : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97c96128bd254bd792d1a6ec7c38face : SUCCESS in 19m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a45fa9396c134bd0bdf44fca9556f0bd : SUCCESS in 50m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd8a62dfdedf49f1bc27dd24e0cf57f1 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":46},{"id":"81aacd48616612822940c5834cc2e552246d4add","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 04:37:08.000000000","message":"Patch Set 46:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba835e2137d450894d7faccea652d01\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72dec3fbfd324934ae0c89984a1e6880 : FAILURE in 1h 56m 10s","accounts_in_message":[],"_revision_number":46},{"id":"3d69d6254c2408da05e59820c0b7487724f9fe45","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-03 14:44:18.000000000","message":"Patch Set 46:\n\n(5 comments)","accounts_in_message":[],"_revision_number":46},{"id":"7d2d0b02da49e84c94b54e44f4ff88b48fa9482c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-03 21:38:44.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"dae266972e1bf0ab7fdea7e24efdf78c67ea5101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-03 22:07:43.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"1b29f2fd795061c81eee915cb8b2af126d181244","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 22:17:42.000000000","message":"Patch Set 48:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8627d55638864a778cd23f407a40e21d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84b637fbe1d5408ab967eceede3208e2 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":48},{"id":"57322ceae0139fa782484b1b2c9ee059aea46aa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-03 22:39:09.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"94f0814fe2f6b9826a0f1f88e8bccd39ff4f452a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 23:40:45.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96a6d9532acc4f979393a78c5ee3ec07\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f1b9a6c8490490d9b856f2f92db2dbe : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":49},{"id":"1296352ea7b683e663fc1038a0b5d58a4dca91b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 23:45:33.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f118e7dc084412a71976f16db3184e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfafdc3f84b41be9d28e8537e35400d : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/325bd7601c954ed6b42a115ef4621c4c : SUCCESS in 1h 00m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce1b005c77dc4e988ae59b1d47137599 : SUCCESS in 56m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1486ed713ed41ff903744439c31630e : SUCCESS in 56m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0b9f6f2ee844d2196549daff22e22a4 : SUCCESS in 1h 05m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/207e535ee8994495a5572c2809d136fb : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47c8c490eb784b1892fd845495e9771f : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b5253df28bf407883ded73e9c6f6653 : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59007b4728b14a8abcc5d3f044050927 : SUCCESS in 4m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b8ce2e1651944f5eb9a1fd415d36126d : SUCCESS in 19m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/691e4069413b41bfb97eed282eceaa1d : SUCCESS in 9m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/41c6b9e379cc4afeb080070f0cab5cf5 : SUCCESS in 17m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da3c1085ccce409d9f48b0f1bfb3eb14 : SUCCESS in 8m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d765560fb9a4888856d030aa4e378df : SUCCESS in 13m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9fe6c82ee21b407db02a07727814f516 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23899366e6e945d281c42812acd5a181 : SUCCESS in 4m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd321a41defe460fb34380d358699ac2 : FAILURE in 13m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/54e692a3fac54099a5d8e4a0abacc5be : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cebf2ccbd4ca445ab821411fe55fb267 : SUCCESS in 54m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a9cd367b64f49d2bd6757ccd8f5d280 : SUCCESS in 29m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1e72340f71dd40e380fd271dbba9e0b8 : SUCCESS in 28m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d04044646ff14b749b94433c3289a1f0 : SUCCESS in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc0e3b56741447d92472422d8fbb448 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65d88c3c47444623853b01647f5e21ec : SUCCESS in 15m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898c9af69ceb493cbc8aa3e3315cfa36 : SUCCESS in 35m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/331c16c502c14203b3fb934883a309e5 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":49},{"id":"93730bc76a9a2c7418b860779d228fb8906dbb13","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-05 20:26:54.000000000","message":"Patch Set 49:\n\n(10 comments)","accounts_in_message":[],"_revision_number":49},{"id":"f62f54216d84c2543e744a573b8921addd67968c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-06 09:54:28.000000000","message":"Patch Set 49:\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"43bda1845c17bcfdb0cbc9ff7631c6c9d89f1628","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-02-07 06:07:48.000000000","message":"Patch Set 49: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"a0dfeb1aada36ea5128d8733eedede238aa27c95","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-07 16:51:17.000000000","message":"Patch Set 49: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":49},{"id":"a883dbd0bd95517b216c06bcfc24c0f21f4e291a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-07 18:31:47.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":50},{"id":"6e83b55837a69866b08ef2cb94b70a0c02016f38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-07 18:32:40.000000000","message":"Patch Set 50:\n\n(2 comments)","accounts_in_message":[],"_revision_number":50},{"id":"f0c329e138625ce21dc0006b4e9344c76b9f6b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 19:55:06.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba60eba436a0429da45996d42b85b044\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a463749be8614733af59c4b36fbcd73c : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a0fa1b2f23ff487385feae4164c4f5e2 : SUCCESS in 47m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e39445c2f7c3410da8d90fadb3e26fe0 : SUCCESS in 55m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70319bcb0d084efda2d84eb68afb1f02 : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/682a4e434d334ef7a718ac0cbd68df29 : SUCCESS in 1h 07m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e32ca30318e84b04a16755d2bd37b014 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07cde1fa719f49e9bce70883c179fb1c : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/060ae0626a9d434b9779c00cb5ce49a9 : SUCCESS in 22m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76bcf372b3e446e19c6dce0cf343ad8b : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8822578ff3de46b98801f37596458721 : SUCCESS in 15m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b42055ce8d844cc5999ba3137cc6e0e5 : SUCCESS in 22m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba86d1b01d94a549de219b802810d3a : SUCCESS in 28m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a6e0fa3fcaa84f0f99f8de64cee9d0e9 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b0149fb2fb03497dbf9257bbce21140c : SUCCESS in 17m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/843107b7d6bb4aa3aeac1a721d9efd3a : SUCCESS in 19m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d41745ddc8c74cd792b8dadc1de0b6e1 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37eafa9fdb414e7c943122b3daafc7d4 : FAILURE in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/299ba230186a4a2abfe79bd8f3e3ebbb : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/877121be13ed438b96724a0dab0a660b : SUCCESS in 48m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b06f187900a54bd78fc274c75e10e27a : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb47c10e036f412ca5612e997a220f62 : SUCCESS in 30m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2b4ebb40f41495187faab6f463d7eaf : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b76f2f17664f549f0a26b31b1d774f : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/727df7573efe4db6ae45f6ead5fc99bb : SUCCESS in 17m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32eda8e965484ee1a6123f7a1c188514 : SUCCESS in 52m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be8c17f7034b4144aec01614ddf2d7b8 : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":50},{"id":"7567a89dea4d39492bcf2ef7fe41d87d19ad28a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 20:10:06.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20b250390ff04fc6892bed639807de58\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/321098bc2d5b4dd4b3304800fb8277d7 : SUCCESS in 1h 36m 01s","accounts_in_message":[],"_revision_number":50},{"id":"3d12638d1815d137225ef3120158edcbba6f697a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-10 22:41:42.000000000","message":"Patch Set 50:\n\n(5 comments)","accounts_in_message":[],"_revision_number":50},{"id":"bf344e63a9d27e6a936e67a0617f72faa657b856","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-12 17:04:40.000000000","message":"Patch Set 50: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"8815de2ba46ab74e15459bb550bb6b1e786c80df","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-12 17:05:36.000000000","message":"Patch Set 50: -Code-Review","accounts_in_message":[],"_revision_number":50},{"id":"6c20ae0449594c661628b9c3bd842fc1c898f24d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-12 18:58:38.000000000","message":"Patch Set 50: Code-Review-1\n\n(33 comments)","accounts_in_message":[],"_revision_number":50},{"id":"c8e21801da3d8cd8db2dcff35cd9df6755df2e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 01:03:49.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":51},{"id":"f07af3c2fbe3cbec9a50eeae50ef99f1aed737c9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 01:04:12.000000000","message":"Patch Set 50:\n\n(35 comments)","accounts_in_message":[],"_revision_number":50},{"id":"4beab02dc86c409a4074d44ee817ba47e3c74f46","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 01:42:43.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"0ea8078d8b1eb069aea23b174b23eefb6cba1f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 02:13:32.000000000","message":"Patch Set 51: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0d4938ab7454a0fa9c23190d2d01103\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaacaa24ba994dc395a6136703a82fd9 : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b2d5d7e38d0d492c815dec93a0674953 : SUCCESS in 49m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc1b77972f9341e38601291c2bbd7831 : SUCCESS in 35m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/287394f12eb34484873deb7f1fd758b8 : SUCCESS in 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87e9680fd2ae4ee5a2599a9e456819a1 : SUCCESS in 1h 03m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e73beff469a54ae88719a0f648c17ec0 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db86276240f94721b63f0c5e420ab281 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4299e7d2bb514b328a86a08a34e219fe : SUCCESS in 21m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07788160c7b743ef9dfbaa1a7f3a9625 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7aa7e51991cc44fab68200fb40f72b69 : SUCCESS in 19m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c56cd641f5646c1807f8bd42b3bb1ff : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c10f2d0d67342d094cb803b3168aaed : SUCCESS in 28m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ffb554638fbe40218a55824ca30eb895 : SUCCESS in 9m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e6621e26573e4388b6b2e7b06a211f2c : SUCCESS in 16m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/350fd3ff98204c109506e71394c5724e : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9060eb3ee8984eccbe7dcd61de97542a : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc94b99bdb7943d9b5b63ede6473fd1f : FAILURE in 10m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f3226570a44c4a6baf39a4db1beb042a : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ee9b49c23394fbd8ebce50bc78c4dbe : SUCCESS in 51m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58c0a836287b42b1bea5cb7fc2927e6a : SUCCESS in 51m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6f8e57db29a4d61a308d75550f34241 : SUCCESS in 35m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdbc73dec68d459484e71acfe33ae886 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd2ac46d5414d1faf7a65f681974594 : FAILURE in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74363632186143d8bc407708d2968686 : SUCCESS in 19m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca3c9782192544ed9ceaae7e4699a7c5 : SUCCESS in 40m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db562afe5a1d4a7b9f80e54d1c3583f1 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":51},{"id":"818a32496dc13ed8977b2305c3bd37f9d83570ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 02:28:38.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73340bc35c9643f1a0d76bb487da98fb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/998bad48624141c783fa1e589e5c7755 : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":51},{"id":"d277b63b1bfc20737e3d26c661d1b784bf9173af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 03:22:19.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"ec33f3362eae905cf4088c001d7d0610f0750bf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-13 15:00:54.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":52},{"id":"fb124d2b95ab9568e893f958d87c0cd9b3faef69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:24:45.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67461d6799704043a68b2b058952cf3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be4cf9d7a2964e968f58725a064dc8c9 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/0ebaab9513714c059e727f2ffb85dd18 : SUCCESS in 1h 17m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0567b063c2f64f5c8cb10815160db77e : SUCCESS in 1h 10m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a43593fc737846b3addb9ec70bb32706 : SUCCESS in 59m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/600f90b8691f47baae331005b4aa839f : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/092e42a9710447e9ad8447caca05376e : SUCCESS in 23m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aefc185ff2e54763b49820e954d9d314 : SUCCESS in 22m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64748154430f4a03880d791371e57c8f : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d947c50b66a46e0b9c2c249a7da2d70 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5baf5db2994a98858a45d86427cbff : SUCCESS in 22m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6b075cdb2b404790ba2b787dd1d55e : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4279dabd054e48ecaaeb8633c4e7785e : SUCCESS in 13m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6b51b4f7ec5041af86b7aa6ff178af76 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d85d90e47f29479985ee4f976fc0bef0 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d11b62ea06e41b3980d0dce719cd0cf : SUCCESS in 9m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10f9cb6ae063438194e3f49b7645a877 : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7a1fa3be9be4c2bb7485c731e8a6f68 : FAILURE in 18m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4ae341cfbe8549e7b3572a1501f05f9e : SUCCESS in 5m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f40c9b271476476e8fb3c6e8696b757c : FAILURE in 55m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ee858261bb943dc98b401abc2b869c8 : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cde56f21504241d5b496b6ce4453fded : SUCCESS in 35m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/299935e3abd341288974ecd023d9b20e : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b54e5e99808a47a9addb53616e9d1a84 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68a534ae258b400ba9692a14d7247289 : SUCCESS in 19m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f801a0d4e6f14f2ba09aece5da9027ae : SUCCESS in 58m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cf62efde68644ea9011b915e2202052 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":52},{"id":"25e277f7d1ff844f75cbbf42a6bdaf3f74976d06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 16:29:22.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae93bded30d422e9122a921167d5354\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/feba2086ef35460e988264e571262ff7 : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":52},{"id":"d86316e467cbb08b9327136d0f6c1daeebca0bd0","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 17:32:15.000000000","message":"Patch Set 52: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":52},{"id":"3d40533b4d257c306593abe76c84a0dd33008c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-13 18:27:37.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"26c9fd1fc6317316cfbbf93a35fa5b2487814419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:23:50.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b481b2b736459eb4820f3fe18a556a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305a2efb216d4892809b6dec6e515daa : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c0f4941dda264630bc4a4e412bbbbd18 : SUCCESS in 1h 01m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8bc2d7d10ff45049c67ec29ae054e1a : SUCCESS in 1h 20m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67c2ed7553c840d0aac5bebd3cd8e94a : SUCCESS in 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/113d8b4b9fe6439d8a10690f273531fd : SUCCESS in 1h 55m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73ddaf9da376439296196db6790ef927 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1b88a22759b41a9934ebf43765efd00 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/707d6931ed9e42b1808ce4cc574f861e : SUCCESS in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f85b94abaa84c1c91ceab2b73a55ee6 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bfd88c9e524eed9b1a282fff02d4c1 : SUCCESS in 19m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7dcb4c55f3634c27b5d64bae89b96eb1 : SUCCESS in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/395ec3d69d1547a48c10d654a429b7ed : SUCCESS in 20m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cb5ee5a0f082432195d16ffb4d7e131f : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e39ab0d259184632beb5fb339d5a9853 : SUCCESS in 10m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b9d77bb9ec3e43b08a1a688df18960ab : SUCCESS in 19m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1a14321ceb4426b8ca2f13d4fdb00f1 : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86097e244325487eae1fdb724fa73a21 : FAILURE in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/89b32bede3024b3dbec7f9d40ddac0ce : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e95b5bb49c20461f9f4f7894c59a7750 : SUCCESS in 42m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0adb91d8fde741358675e067297763e9 : SUCCESS in 41m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4234a81b0504cb1a89ebe784fd6af60 : SUCCESS in 29m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cb952e62f0248849baca3b7a4f0e74e : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fae6e25d3c48249b0d389793fe5139 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f5dbd6451034f69b110d018f6a7e5f5 : SUCCESS in 12m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00a8492b1df4452a99fa6397b5e70fa4 : SUCCESS in 59m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/003ae42286f74d27a9ca160a71caee44 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":53},{"id":"fa0f6133528ab7ee304ecf8aab63df1395c1e50a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:26:44.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d895e6f5a3814570af5827dd043b7c97\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b35179c9289a4b6fa9d16c51422fa7c9 : SUCCESS in 1h 57m 26s","accounts_in_message":[],"_revision_number":53},{"id":"c4a1851e4861306a2e0a168d3dc920abc6ac54ff","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-02-13 23:29:58.000000000","message":"Patch Set 53:\n\n(2 comments)","accounts_in_message":[],"_revision_number":53},{"id":"bccd3e4a71355f50d6003b5d15bd754f992c43fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 02:34:33.000000000","message":"Uploaded patch set 54.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"c8ff744efa9b56c095f4f8d22724f0e5989f9943","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 02:37:44.000000000","message":"Patch Set 54:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240057e3cd4d4c86bb323c68aebf99bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dd57983ee3814c7ca208645a2a453760 : NODE_FAILURE Node(set) request 300-0026336614 failed in 0s","accounts_in_message":[],"_revision_number":54},{"id":"16724a10ac203750eea2d30bfb1be8b5510c1767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 04:18:12.000000000","message":"Patch Set 54: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60a19d652e594320aaaf1856ee7cfb8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc986a0417e4d6fbe4c0bf77d31ecc0 : SUCCESS in 17m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/db57c82aaaaa4acf9843acb9e4367d30 : SUCCESS in 1h 01m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9981b5a05dc44d0cae589f42e7f7a72f : SUCCESS in 54m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dee19e46cf094fd196c0005e8048ce6d : SUCCESS in 41m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e214d9649cb745bd919292aadfd572fb : SUCCESS in 43m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4bcb1ae86124bd1b73b1c0f1e6bcc7a : SUCCESS in 14m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f738a46cf8d483c895259a53b514b6a : SUCCESS in 14m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dc7bac5878d4213bd6f756d08da205e : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c538c38088484cf0b2a15e01be5fb09f : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5cf1644a995b4f68a546d39d5024b7c8 : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6bf0b252eeb47c1a66fc7708a1fc1e5 : SUCCESS in 21m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8d9a57640dda43be99e89767ffc44b6e : SUCCESS in 28m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ccfe3206f9774d36aee2e69b043f3504 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/08cd3e87a4c04eab8596d34470f5dfb7 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/01ecc2aef1d04b3ba55911f120ce1eb3 : SUCCESS in 17m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc7282299b6b46a193125773118242fa : SUCCESS in 10m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7154723e7164af9bfcd1c75655ad6e6 : FAILURE in 17m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8576db1325f843bb8e8bca7887dd048f : SUCCESS in 10m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14339eb61e6b4d4f90e0ef80b434133b : SUCCESS in 59m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba10bace5dd64f54bccb85016ffb6f89 : SUCCESS in 55m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e23abc7bef94cd9b2d6362d61ef2d00 : SUCCESS in 33m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73533b758b24422ebcfb59712553b94d : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbb7d6465bc4b86b746193a7f36edab : FAILURE in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f34e7b1780264157a6fd188dfe2bbaf3 : SUCCESS in 17m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/090144e2de914cffb38ffad23b947b0f : SUCCESS in 54m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cefcd20a42d34b7fbfe7bb32e94e61ae : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":54},{"id":"5ccebf510bc7647e2800b852235452ab6b5dcf7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 04:22:48.000000000","message":"Uploaded patch set 55.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"7c7432c200742c75ec983c3bb9461efefa2b3fa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 04:23:10.000000000","message":"Patch Set 55:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a558e1a37f2417ca44b03b9518aff94\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8881424785ef4146b5169b250f92daf8 : NODE_FAILURE Node(set) request 300-0026336881 failed in 0s","accounts_in_message":[],"_revision_number":55},{"id":"2bec749cd138f206318ef88d6c177e39a2a6e74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 04:32:08.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"e01b9d37fa53c176609da84aff38121b9cc9a5ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 04:32:28.000000000","message":"Patch Set 56:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee57a9fdf8ae4c87826f7347c5f18f9b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f9bd9185920b4086a2fe62ee4770b482 : NODE_FAILURE Node(set) request 300-0026336928 failed in 0s","accounts_in_message":[],"_revision_number":56},{"id":"2b433203b1e4a64a4a87ed9f42d3918455d0f72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 05:39:41.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc79b361f9740759b0a19ac404fccf1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3e3e29a9ce40f8a7c2099b126c10f9 : SUCCESS in 6m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/187a68eb940c4efaa7e03004be7cac75 : SUCCESS in 38m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25244ac42ba9444180a618d925fe9b5c : SUCCESS in 1h 00m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48d6ad89dfd2452bad00575449f28eac : SUCCESS in 56m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b08a22d5464620954f9a418ac9e40d : SUCCESS in 40m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08a720fc631d42bdb190f990e90c1550 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3a29928b31d42f383fe595d4f32d3d2 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a421a7289764c8786b5f3940d95d65f : SUCCESS in 19m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/934691fc20ed4ad2901852467263e3d8 : SUCCESS in 3m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf250028610d442abcfae41ae87da9e4 : SUCCESS in 19m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f1bb9ff16de4489853a13053848ab75 : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/58b73e4b7a8245d7bc2ec51ad7246e97 : SUCCESS in 28m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/90eef28b31a84f018241d204f2cddeab : SUCCESS in 13m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90d6133a4d4a42f784e999a7750944c4 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c171939383744acbff9d04ed6943713 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efa7ad46664a4985abaa5adaa2c31943 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/113d57e0462a49c5b8cbbf780c4b81bc : FAILURE in 15m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d07ca00554cb4d3099a8feb985541571 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7343be2223a942c7a2c09cbeae29bfea : SUCCESS in 52m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5ad1cf95f5d49f69a3767a8cb27a9f6 : SUCCESS in 47m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f263c94a5854ad1ba17556c088d3c7c : SUCCESS in 32m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eee34cc0d5bd4f0b9bcb9dff8ab4fe99 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4a644ef550491f986b6b3914b01059 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7653e43b44c142bc9efd1e3e1f981198 : SUCCESS in 17m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e41793374c6043bfb20516929234010d : SUCCESS in 37m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab022bb166be4178882b94d484a3eba5 : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":56},{"id":"78d2c539e2f444e7a2519ee7f2f87ac9cf207f0f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-14 15:48:55.000000000","message":"Patch Set 56:\n\n(3 comments)","accounts_in_message":[],"_revision_number":56},{"id":"9162e5c67f279a4fd73adc16230f5b4ec7cb9cea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-14 17:04:50.000000000","message":"Patch Set 56: Code-Review-1\n\n(32 comments)","accounts_in_message":[],"_revision_number":56},{"id":"18ffb027c5cdb0124f9d4ccad31c47a508b6d6c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-15 16:43:38.000000000","message":"Uploaded patch set 57.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":57},{"id":"f9cf3fa330c5d60c9dd6afac62be80d2b92f43c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 16:44:14.000000000","message":"Patch Set 57:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8828cdb56164ca0bb09f9bcfe5d781f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f7763f2d5ba4563825f5358138fa289 : NODE_FAILURE Node(set) request 300-0026348666 failed in 0s","accounts_in_message":[],"_revision_number":57},{"id":"d602aacdef44121278a23a38fbfeb2b5cff3a061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 17:55:58.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3a5ccbb6759409ab5d4bad8235de8f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5103cadd4330435ba1c38b4ad9ddb570 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/75019723dec34bd684dd60edb9f628cd : SUCCESS in 53m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87f73089e8e047b4ae1ce5f52d82cb38 : SUCCESS in 53m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/46ddfff1ea504f68b95b5913727041f5 : SUCCESS in 52m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d2968fca7404961a1c070de45923404 : SUCCESS in 46m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c464f8eb56dc4ac082f69b640657a6b2 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d32146474b3a4b6dbbaa8efe34ac5f63 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6ad9cd045a045669f65b8c4665a4599 : SUCCESS in 21m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/24c24309a6394ab7a89f069451355e8b : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5604a068a054e1086c512ac49c97f2b : SUCCESS in 20m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30ca2424068f4585a39f0f433eba3b90 : SUCCESS in 17m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e4712474a084a8ea1b6f52fa353293f : SUCCESS in 29m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/45c39a120baa4766878f9cc403dc51bd : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/66771b2cd4bc40f0b45af5f43f96b711 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0b6120b61f024efbabb0fcbf7160c5d1 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5cf30d180ea48c7af7569f897acd61d : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9fc76578daae432f85303acc445c87d6 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/353575b331964805b52f515a0ccce428 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b06fc9f406364f018a1f5563d6dc9f7a : SUCCESS in 36m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23d31a4e7b5142fb946a9064bfade659 : SUCCESS in 48m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af101c9cf0794b4aab208c453956ef7e : SUCCESS in 57m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/610bbc08234a4dfea38bba760b9922e1 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39807ce1cadc436bb2d12f1ff6c7edb5 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b22a4a38836483a9d282544ab3193a7 : SUCCESS in 17m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d11182a8347c4398a684fd892cba170f : SUCCESS in 1h 00m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c733945148614528b8625fde106ac267 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":57},{"id":"6a2209ec4f95c072aaf821aef1dce87cb897a02e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-18 16:56:25.000000000","message":"Patch Set 57:\n\n(7 comments)","accounts_in_message":[],"_revision_number":57},{"id":"c54820f7a793c9e89ee05e00bcb354b65cf38c79","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-18 17:01:31.000000000","message":"Patch Set 57:\n\n(13 comments)","accounts_in_message":[],"_revision_number":57},{"id":"d7c05f160c9d300ff45e2267aea8f3488ef0268e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-18 21:14:19.000000000","message":"Patch Set 57:\n\n(2 comments)","accounts_in_message":[],"_revision_number":57},{"id":"3a9898f680777895b320cc807fc8daf97778f9a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-20 20:16:51.000000000","message":"Patch Set 57:\n\n(1 comment)","accounts_in_message":[],"_revision_number":57},{"id":"9bf17070f1744bead4b282b4a40ac95c31d7d91a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-21 04:53:04.000000000","message":"Uploaded patch set 58.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"0af90cdccc46c5fe6ce79757ef9294084fa7eaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 06:03:36.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e7eb4a6e3d445295bbf89453f9b494\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e12dbbf651940be86b387ee1c421afd : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/aa14e070598849398c3de49914315f39 : SUCCESS in 1h 03m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eae025c7245b4d7fb0a3c93e3e2ab0dd : SUCCESS in 57m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f6741258dad451cbfdf0833da5023db : SUCCESS in 41m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/358507abe5d2454f985dedbfa1ed7553 : SUCCESS in 49m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/032d2842fc6842bd80c87a9fa904dfeb : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6173724d6e92456787c194fae9595816 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d83ca2a11f84568b89d4b0607c3ae9e : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45763d109dd24ea3a02470d2f765f0cb : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f254a8a75d2d424ea5f484d11ad608ee : SUCCESS in 18m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82284609102d4ff596c7302403037940 : SUCCESS in 10m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/911074af27b641b1affce0650d8bfe7e : SUCCESS in 36m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d824d3913674bf7a9804d19fe48ef28 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3cd3f6994b9f484991c6dd1580be0fc3 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/900859642ff840f0b297b1aa76eb449a : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9abed7ff8bd4466faf8ec39459ef60cb : SUCCESS in 8m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da20be5abed64461b86ff6eee1601a82 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/978ffe347eff435a834cc5cb3cdc68cb : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/830151a0a45b4f7dab7a9a6f049de9df : SUCCESS in 57m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e5343db399f4ebbaf0fa3602d0b0a51 : SUCCESS in 52m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26eb809bc6b84c02836d3665d665e863 : SUCCESS in 52m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16b3cc28e48a4b938a5407eade713ec8 : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d79ecb7786b42dba7d186549c0bf0b0 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38a741da03ec4e019da38d9c1393a57b : SUCCESS in 18m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2953036991894caca822297edbe5760b : SUCCESS in 35m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e02e5ecfbd74ee0aa050a2b2809d690 : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":58},{"id":"4ecd44b6232d0c688fa71c4a35204813b877e9c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 07:57:08.000000000","message":"Patch Set 58:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09cc6797fe25457b9ac081377fdf404b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43eb8cabd2164d45ace59af27dddb21c : TIMED_OUT in 3h 00m 51s","accounts_in_message":[],"_revision_number":58},{"id":"293cb45ded0910e41a842ed8fb5a23a20aabf424","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-02-26 16:32:07.000000000","message":"Patch Set 58:\n\n(1 comment)","accounts_in_message":[],"_revision_number":58},{"id":"629d9c7e19c050085e5d94c764a2f14ddbfe8c28","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:24:09.000000000","message":"Patch Set 58:\n\n(16 comments)","accounts_in_message":[],"_revision_number":58},{"id":"edf377df635c3524b7c9395b65eb9672e139c1ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:24:22.000000000","message":"Uploaded patch set 59.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":59},{"id":"c34bf9d8b4e472c8fa4f4cf5dc24b0939da0fbbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-26 18:40:51.000000000","message":"Uploaded patch set 60: New patch set was added with same tree, parent tree, and commit message as Patch Set 59.","accounts_in_message":[],"_revision_number":60},{"id":"289544263a238c95ed35c1d9f5c567f0f038f619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:04:48.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9560f940e84545d684033fa2ef3f08fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475e2dc0ce4041c284500e4e59a27e7d : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/0947f186b68c47c7b629759abd0d9a98 : SUCCESS in 59m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c3a0efe579af4a899227950247241903 : SUCCESS in 51m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/521c52447fb04001a57aa40afd32a20a : SUCCESS in 1h 08m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1fc420245224ae199e15fa58d267a3b : SUCCESS in 1h 15m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11ffa6dae74049c3955d591f4125eccc : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfadeb0de4204db8b52237f5764cd0f8 : SUCCESS in 23m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5705db93f1c94b91bd11c027d516f406 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f42905934f064771b31164d44a5e56ac : FAILURE in 4m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ae5e99b3058456eb3f1e48ad27a612b : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/713e65513e0c4407a2f6510a3ab84b72 : SUCCESS in 19m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b3ae424abdca42c0a7d416b73626e13c : SUCCESS in 22m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/78812419d2b84dac90c2cbbc282d5d1a : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee47eb8172584bb195ee1e80262c58d1 : SUCCESS in 24m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/193c4919fb094bad882e0c5aa53acde4 : SUCCESS in 14m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d60758003fe340d9a3383453332ad5c2 : SUCCESS in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/abd8e4aa42184f2ebc026e69da0471db : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c63e7aadd6af438fba938a899e0c9b38 : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d66f473454394942885b99b2dbd032a5 : SUCCESS in 55m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/631dbe50b70a48be87fecab47116393b : SUCCESS in 39m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/24b22cd6e4164968bc983c0ef02b17cf : SUCCESS in 42m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afb6a8fedb42492c928221faa362ecc2 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ec7910fdde45fcb6bb8242ad068644 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9eef02a90d15406e920a05cc77ddd1d2 : SUCCESS in 18m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca3625fa983543e9a04ebab4b2a5a520 : SUCCESS in 55m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a3c7465f10d4675816265e5b3e423d5 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":60},{"id":"f06d5d442b06289cb1a0dbdb7cc527491fc8f4ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 20:48:01.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3074f01f605e4736adb6d439e5612b71\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fcd9da2cba4d45968e925b721d27a09c : SUCCESS in 2h 03m 36s","accounts_in_message":[],"_revision_number":60},{"id":"3b42077ebeef0c5f6b66e4145325a4f5e8ff947a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-03-01 17:24:17.000000000","message":"Patch Set 60:\n\n(1 comment)","accounts_in_message":[],"_revision_number":60},{"id":"57230a488bf582e08774a3e4430ea192309742f3","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-03-11 15:48:17.000000000","message":"Patch Set 60:\n\n(2 comments)","accounts_in_message":[],"_revision_number":60},{"id":"802d366b42c95d0371a55205f38c60bec6838d35","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-03-11 15:48:57.000000000","message":"Patch Set 60:\n\n(3 comments)","accounts_in_message":[],"_revision_number":60},{"id":"5b2e605a85346f4ef1408bff769ebdad154df69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-11 17:41:24.000000000","message":"Uploaded patch set 61.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":61},{"id":"893eae3b07e1987a97c3aec17cc5e3c6cefd86b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:26:57.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f99b4b89dfa44ccf88855db340aba0d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f40bd4844ee843d392679f0081f63dc2 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":61},{"id":"a6f4c5fc831d8d2e046ba2271bd63515d5c1b769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 18:43:03.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938479fcb0bf45d5a44ee6984bc1f35e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e00f2487f546e2a94cfc0890e1174e : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/7d454735a51f428993d8865b5a5a68c4 : SUCCESS in 41m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84bf7ba03d0f4a6a9ea84a1607eadb1a : SUCCESS in 32m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1db474544f1a43409b47fa1508e01264 : SUCCESS in 46m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee925cb0d1ac4bcf942e681b7860fdab : SUCCESS in 48m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/661f016eea204d2db67e823093d86d0f : SUCCESS in 13m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83ded0fd9aef48ef9714b827d3c74594 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3363ea08e68543b19804120c08597197 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f51e0e5d09bd4d2486586143448cfb19 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c82ef013a1f24091bd85e6310f055ce8 : SUCCESS in 19m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f5d897332fbd4cfc8de86ea1010c25af : SUCCESS in 13m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7b2828f2ff1242ba865847830442bcc5 : SUCCESS in 26m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e02938a6b434e26b0cbd5bcbf600b65 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/461cfd0e954140768c50a1c2f8b23e84 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7385f0a4dc034fb7b2a9e6fb73d913db : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1ef84cab9eb413488b4175c651dc0f7 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a5ccb6b984645188062a68710ccd63f : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/48f95df065e9480bbbf5c1b304e9b658 : SUCCESS in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/97749c1b984e499f90f1b6d1715286aa : SUCCESS in 33m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb0f1de096f54965814db3091722aa86 : SUCCESS in 36m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eafac9a03b3448979c8e35a75c48dddf : SUCCESS in 32m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2418cd65d3d438b95a47ba62250880a : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fe41b5834b4fc6856819faa7dd29a3 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/76134d61be404e5587b49a66208c5a26 : SUCCESS in 18m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff3d33efae3c4caebd2aa1b1ac04053c : SUCCESS in 33m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea1924b485e4d98bfc4d884ebf31417 : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":61},{"id":"8ac49e53700921b6fa03bb3d97c433acee0db62f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-14 01:43:53.000000000","message":"Patch Set 61:\n\n(2 comments)","accounts_in_message":[],"_revision_number":61},{"id":"3e20913df091bacc7426675979eae5dcee881a06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-17 14:22:03.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":62},{"id":"50ffa5610b17979100880e8ed8d1c08159a6206e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 15:05:41.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3497f177cf408bbc290d140f7848bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d281aa734c84ce0932fb015f01881d9 : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":62},{"id":"73cef070dad02edf9e49e46a14d37f04aef06f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 15:30:35.000000000","message":"Patch Set 62: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6863aabf0849bdad3e4bea4f1a0276\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c85fee8da354a048f0dfbd7c3a3c83f : SUCCESS in 5m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1cac1332c7294942a6f82343487c7132 : SUCCESS in 1h 07m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e7bfe4249694c238186f73823bcf55f : SUCCESS in 52m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9721f2d4b4714e829badc58378138abb : SUCCESS in 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d22f9dfb3c74f62906e49e8b5c71b08 : SUCCESS in 1h 05m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6526bba3f8ec40dc814e552081dd3003 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5da1c890b87e4e039b59af141ce95a3f : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fe05300c3e4457db395ac794ed9e283 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c3fd226f22c498396147b6479c43c4d : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/252da9fc44454f6dbce1694ad5eaa6d8 : SUCCESS in 9m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/945676f857ef4508979af9944059a601 : SUCCESS in 10m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1c3c68e7fc0481abdd701645db23ed0 : SUCCESS in 23m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/629e14846322461cb0631c26411d8303 : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce1c165e41b543d195f7a2f4a9c15408 : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ccc94d7a29344897ae163d05dcbecebd : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/827be449d4ba499394eaf5316e8f80a1 : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffd879ec8cbc4d52937a222c3dcf83b7 : SUCCESS in 7m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/307d622cdf1f42cea78e523cc9a90f07 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a1d8002cf6140a4a231159dda8dcfa2 : SUCCESS in 59m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02e80781f62049a88cba2d72001d84d0 : SUCCESS in 43m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/875eed3e9f634276aedd23be42a33e0d : SUCCESS in 52m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f015ec5d287a4b71ac9374100cce3214 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47a3fc918d344ba8d53da37f9befcfd : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49c608c6e1b34325996c9ad6b7e969d5 : SUCCESS in 19m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db741aa783904598a9c81ebe575a80d2 : SUCCESS in 42m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52a3a0885db0455b8666dab45e88f562 : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":62},{"id":"63563b4bb3814825a7141537388085ac2b060be0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-03-27 02:17:19.000000000","message":"Uploaded patch set 63: Patch Set 62 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":63},{"id":"63e076374f39669b9f1ea30778064059ec51d18b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 03:05:13.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b540532090041909805f7f21a64caa3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc5654c2a0c94fd4b55ba78339cd6e4c : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":63},{"id":"15c41269b85cdbf14cdf2d55383cc1681e486e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 03:54:51.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe5f7bfa6ac34d21810cda73fd33a1fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64893a9ab1ec40a38fe991e1f612d9ca : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/6d4d3768d59b4e6aafb59b7458bd1266 : SUCCESS in 55m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9c6f9b6fbbe44a39eef964aeb1cfecd : SUCCESS in 47m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5be365eb1e084c17972ccfcf1e13e9f2 : SUCCESS in 49m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2058de0cb1a4e449331c772c80daf41 : SUCCESS in 1h 03m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3345c8cf4c114dc6ab7df8b66472b06d : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38d390727399453aac1007c595a53e06 : SUCCESS in 23m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/889fbcc6dff949a493e17bb98a65cf15 : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/958f6f1bcbe14908b388879a940f898d : FAILURE in 4m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6afd3f6337b24a41b018b9a64f2af35b : SUCCESS in 16m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36d8f854a6374681a6a57b4dc35a67a1 : SUCCESS in 8m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/23142fe242524f059ac5f91ca40f1008 : SUCCESS in 27m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/19d127d579684a3c81d344bdcde77958 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a18ab6d6b0447558dc50f62b18379ec : SUCCESS in 13m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac2a9012ba764849a0ea95664a807453 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffa2118f28154f279c83c74130801c98 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cf26fdba7814ce992957393afd6541f : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/36446b3402db4900856c6b77561af3fc : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/479273cad99c49c98f2c60d2442e48ad : SUCCESS in 1h 02m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52817f3218f34d2c8a817b050c1cd32b : SUCCESS in 44m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5e059615e9b4983b1b7e055cced6ca7 : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c08f1be88d2458cbf5b11d5b71eca14 : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957ec1979b044570acffc72125d9da73 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/550f50f30b044949b6491bf623123398 : SUCCESS in 17m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57a48a2fa8eb4385b286c428bd7da1b1 : SUCCESS in 53m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be384e59eba7464cb90bf1b8c7798243 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":63},{"id":"3fe02cde53d3dc36e4dcf9854c6972718f8a1468","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-04-02 04:16:58.000000000","message":"Patch Set 63:\n\n(5 comments)","accounts_in_message":[],"_revision_number":63},{"id":"cd41d8bf4e1d863bd6c63907eb4c6c5109b539f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-02 12:11:08.000000000","message":"Uploaded patch set 64.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":64},{"id":"3831d764703a4dafe854a7807a0740c524861ae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-02 13:32:20.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.","accounts_in_message":[],"_revision_number":65},{"id":"d49793da0a20acc5c7df09f7c7fdc7c7067d16b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:17:56.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b30560546de443928cc46f08639c74c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6d01473624a40f6bb78392b21c6a880 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":65},{"id":"887eefcae54163f4395b4badc305120555b2a6bb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-02 14:19:50.000000000","message":"Patch Set 65:\n\n(15 comments)","accounts_in_message":[],"_revision_number":65},{"id":"fa6e668ee3ee5c99889f5be22e19f141695cdfbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 15:54:18.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/247bd3f5c3e846f48b4b6346a7683b13\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f9c70eaa494b4eb59e7d0ddc9aaef6 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/a0307beb684346f5a0e4b225de697eb7 : SUCCESS in 52m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a983718557484a7393eba367d8223ab1 : SUCCESS in 51m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3aa2729224304ffb9675df1096731841 : SUCCESS in 39m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2de52192d40c45d4999836a9c88ff304 : SUCCESS in 50m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1676d586859d413f83c154c5e883b803 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ce91537e34749ca9f7eb8285f043e92 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97eccea44d4c4a84a03702288080bc6e : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0103600f07d4c11855b4ee11916977f : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50226837814e4cfe963ed6b8d6cff565 : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/931a06f648fb4ff5bf8adace5af5b5ae : SUCCESS in 18m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0c2c0ebc31624a378596a6e29b8adfdb : SUCCESS in 29m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/27b9f86b7dca44e2895930d710951497 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a88a264c63f24a26b151c78c4b09b532 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/14b7b5aa9dd4455bb1f624ce32d51907 : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25207297ca9c459aa4ccddc9a2bd6f74 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0b58305872e4a1eb7993a053e27f330 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe6f1a9350664584836bc6a7f82a1ed4 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16f2b51e7c174437b65d67268925e69b : SUCCESS in 52m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/73f6bd35683943799cbdffdd02e5e2af : SUCCESS in 43m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/516f7e31e41542fc84878f25c5d44b8e : SUCCESS in 33m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010c6d4f82e548a7bbfc3067cbdc3eb6 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5571446afa4e919d612a6c9c5da216 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf114efc88cc4c048d0ffd2d7959bbab : SUCCESS in 20m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01e7681f07d4480fab1cdd4830afb5a8 : SUCCESS in 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca6b2c95eed4c39b64e18a03f6501af : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":65},{"id":"e25ce597711538c70422549630533b8539155e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-02 20:12:50.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":66},{"id":"ca37a5658a1beeb61134b319b3525348d34b3ddb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 21:05:14.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174e3ff32d0b4f95a9a78e2167320558\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/af5a5da2960149bb9f718b95a09766df : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":66},{"id":"55e80a02d688eb877b37edb75f3d5e88df637f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 21:26:31.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab22e2b6a0ac4421beb96d6018ae935e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba612e238557490c80e0ed29b86b3f0e : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/1806c83d38874749821237be480d7062 : SUCCESS in 44m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0f687c581884473a013ef27fe7ae3ee : SUCCESS in 41m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69746238668e4c86af09c7bea0ea52cf : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/146705836fdc452daff415781e390964 : SUCCESS in 31m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c36f745ab7f943e6854b233eb9e1abbd : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f11d753f68343d68c80a1928c53732d : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0ef1bad31a147ea9f5050a6aa467a3a : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6017ffd8987843f1add05891ce5971cd : SUCCESS in 3m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4205a93b126e49428251f6f5eb2c6535 : SUCCESS in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/653a52276cbf4a5fb6711e9fbbfcf87e : SUCCESS in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6c4d7e7a900a444d96266340fe438cb2 : SUCCESS in 27m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57f5317f0cce4fbcab6c1936590c6bb0 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae71630e978340f2a32590964018a516 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9ea2cdca46be4208878f92c84f7fb1b6 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/41642192b0e447d6b37b6e5a05547acd : SUCCESS in 4m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b562541612c54424a8fc1c1ccf614d7f : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f56169086118416dae8d05d5d9c3e81d : SUCCESS in 10m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cba8556bd34d4e29b0d450cf8d2746e5 : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09815a1e8d1349a5a9e40aa9d59e3954 : SUCCESS in 32m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c5f63fed24914450b70af0c63437d60c : SUCCESS in 45m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6b923c384df42dea1cce37b2a04137e : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ce00db7983460eb12a530a50ef4848 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7fa2b3e26ecb4937bf357d4c86c29024 : SUCCESS in 21m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b93d4069d9b149039fc2a3579695732a : SUCCESS in 52m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1429c820e6df4a908b4bb380dd0c1d23 : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":66},{"id":"5ff44928e650682640f0480e1522b5b3c982057f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-03 18:46:03.000000000","message":"Patch Set 67: Patch Set 66 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":67},{"id":"68a775e9a6f1d2dd97a7ec16b8f6801db826dfbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 19:28:17.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc76013832614561b9a7f8964db47c99\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/432fc39d177a420391b138442242138d : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":67},{"id":"3599e4d39d810a6cc51d71a9fa2c265585a2e419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 19:59:12.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e361748e3134aa1a52328bf36e07eab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0fdcc1736847639d66bd2f047fa981 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/fbf0c42cfe0845c6bb05e2994225f513 : SUCCESS in 53m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1b7f99b360546eb84e3d055c052210a : FAILURE in 1h 07m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5af2e61c3e94191a6c035f5b3ee0f31 : SUCCESS in 56m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/961d5a6f2bff43c593ac9838bc2cf389 : SUCCESS in 59m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d486197ebe9a47a9b7fd18ebaa66a902 : SUCCESS in 11m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6023ff16130341e38a41bbcecec2a8a6 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8767b16c7c404d88a79dc7acdfcd7e15 : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb962db69e2d48768d5cb72f789255c9 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/59b04dfad343493f83c9a84946b13a59 : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4722168217d4f5c857d829e42c6ffb4 : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb4dec1c08bd4c56a73e6503914d7c41 : SUCCESS in 27m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1eec82d8d2bc44639c4a47d88cb74514 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5e7db391f22742b4bf43971f22bb8574 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a89154e1f294c8c88001b01b9ee43d1 : SUCCESS in 21m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/995dda4de5d243d4a68f4f4a3d7ed481 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a5e5a041bcf4ca597f4b5590566b47a : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb229086417541398400bc458180d593 : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b463f4207ae74fc6bf538de06e59a07c : SUCCESS in 51m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f16f2f87f6c4dfc9d5b9b17b4b1f915 : SUCCESS in 49m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/080d3790de0b4758a22df44f16d904bd : SUCCESS in 44m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beef8e8f43ee44cb88ddb511bb563642 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0422ca351f9e447c82d8372bbf956cc7 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8032cdb16a14423484780cbb2f073f19 : SUCCESS in 22m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c24c14fa0e45479795d98865a549e0a2 : SUCCESS in 53m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b1a30625614db0badd8a34b4136654 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":67},{"id":"575bb53ca8c9c025c94733fd83228164f538975c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 15:06:56.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":68},{"id":"6f392db64e2a948e5514ac2d38954e5789f3ddb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 15:50:24.000000000","message":"Uploaded patch set 69: Patch Set 68 was rebased.","accounts_in_message":[],"_revision_number":69},{"id":"9c89b676ef8f87527de6c7884c510a3e80f4e65d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 16:36:25.000000000","message":"Patch Set 69:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2847defe43144d4490196548d998926f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c632147050b94134b60bad4ea0c25bf9 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":69},{"id":"e97f1ea3f24fc30789b4f2a37f0ac4b2f8d5dfc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 16:55:37.000000000","message":"Patch Set 69: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4671db7886a64732b32ccd106435e5e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb640405d67b4045aca091c228aef1e6 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8bac4495ead64780b39a6181e448cbfd : SUCCESS in 56m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/979032b2028e4e749a18c881f0b14de0 : SUCCESS in 56m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5969b5556f5e4bbaae12404754e4d417 : SUCCESS in 51m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21deffdd839a407c9c0809fc8e94900d : SUCCESS in 48m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92349c3daab1426ba12110daa312c02d : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a18ad9a72374f82969af87e3ebb2570 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e6c106a37954748a7ece39eb2d0b1a0 : SUCCESS in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70bda1385d4149d8a5ed6a14f70d4197 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a667c4180244b219864162436fbba55 : SUCCESS in 17m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/453ae6acb894482784c56b2bb6552161 : SUCCESS in 20m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bf42475d813a4901a5db5dbb2bbb35b3 : SUCCESS in 18m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/54f54d720ef94dc89d4f477557930741 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7aa262f2a1844b9c819948e79d026ee2 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/50f2f385037a4d9e8e77691a142f26fb : SUCCESS in 10m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d47782b5aedf4aef962b47c745b7990b : SUCCESS in 4m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26920f6aec62441383a90e7a6712641c : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/46661e95d39e4f6a87a87f1cf348c65d : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81d89aa6244b4aaea5990c4b0dce5157 : SUCCESS in 54m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad249f6b0a9d400cae102c1e16d3a03e : SUCCESS in 34m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ba0851a9d734feab5e2ac7f3f82a4a4 : SUCCESS in 37m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74aa4b7b4682483faf33a191f27e988e : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cfe286ef444b31ad477c2a8a7449a3 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e78b1897c62c4ceebcbe7c3c4aee2565 : SUCCESS in 17m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97bc6a0bebf4701a9fb47b866747d05 : SUCCESS in 44m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a56df4794240ae9481db4d373cd670 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":69},{"id":"4c134de0a5ba32c7ad5c9a8fd9fab78ce63e65aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 17:18:02.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":70},{"id":"0e03f4aaed9d35b06a1e4a97bf337d2a84e93d1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 17:54:00.000000000","message":"Patch Set 70:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a5d7f2f11a4477a2de7bc794584710\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/30215bd74a2940d39e17353ced8ab1a4 : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":70},{"id":"909505ad89a3d3dd64f9a69562ccbd75d173d645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 18:25:07.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/892f8521f9294c559416f685cff0b6f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8a07db1da34aa389523bc5f8fda4de : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f4caf66db0b34c8183d412246d121fee : SUCCESS in 45m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7d52ed34c0f4855a7a15b371089ed46 : SUCCESS in 44m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62734e953c9b469590d921595a7100fe : SUCCESS in 43m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcb392ba4cd3437184fa90894439fc8f : SUCCESS in 31m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb56cb0fa08744a1848099efd7a50ebb : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6d5a421198047e2bae749e1f53e2a7a : SUCCESS in 23m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96e378bab43e439d9c3bd06c887f8a97 : SUCCESS in 11m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/026b37fa186e4e0abba0fb4a240098b8 : SUCCESS in 3m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03cdca511dd24d5886de429f3c42ffbf : SUCCESS in 7m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/062c389818a04c76aa8d82ec824cb11e : SUCCESS in 18m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bdf4b7a1cfd441e3b9832e90acd5333c : SUCCESS in 23m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1ccb4a99560747eab1be7033eaccbeb2 : SUCCESS in 9m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/371be126ac3a432ba3328b0a561d15f6 : SUCCESS in 9m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5e5d01575e4e48808d3060a72cc6c786 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d601639413046b8b8930bf55eea86f9 : SUCCESS in 4m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c07dcad7afb64c039e4316e94c565e42 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83e0efd8a2b040aa86b5fa2ec9909fde : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/61ea9ba2a3d345cb972e1057c090842d : SUCCESS in 1h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0332848ae3c14aff8c867f205ecd1f82 : SUCCESS in 50m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1e95ec6ffe74b9b98b955dec125b769 : SUCCESS in 58m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ccc26271174411884ab25c868d47559 : SUCCESS in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a21bb781704849981405d4590cc3e1 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7948db1228441b79bf91811f6ec4222 : SUCCESS in 15m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20e579a1448c4760ad40430dd39e954e : SUCCESS in 32m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dc4ec4373b54caa8040cd92596cbb7c : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":70},{"id":"0d636061be3a93cc22fc9f9e19c5ca6db2c6225a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-10 15:40:31.000000000","message":"Uploaded patch set 71: Patch Set 70 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":71},{"id":"e1cd4ec20269818a13aad7b4b6affc2f70a0ae0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 16:31:33.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58760e377be34052a20c61500c8419a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d4ab43dcf7b4aeba31a074c794296fe : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":71},{"id":"3cc4dc19608593ff26f2b7d8c0220bb2abad1b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 16:48:41.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c7575d9c0d4b1aafe173ee289e28db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26920d60dfe2457ca46ad9323bbdaf8a : SUCCESS in 5m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/22987afea2344388ad94e8484a35e8db : SUCCESS in 26m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3eb0eb188254919a158c120e0dcc940 : SUCCESS in 52m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/851088cacb8e4b6cbcab62af4006d660 : SUCCESS in 58m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30d5f654cffd42d78dcb732452cdb86a : SUCCESS in 50m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c21b198dd4b4b37b8964e6c6773e55e : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ed8c1d0169a4eab95e99791b2d480d4 : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a789ea5d3c1404bb06541df4b2ffcee : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7d937fd281044fab9ecab576cacd67a : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ea956249ccd4d64b7a0dd181709d2e3 : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/58eb253a4f5b49e78b6711f15f906785 : SUCCESS in 14m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c7d042b8d2244017a2a8b0467cadfd19 : SUCCESS in 23m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4986b6bfaed9461c8b04a5a4d827cce1 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/146b0cef2bf7429eb77d565b54e94413 : SUCCESS in 13m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cec8e0055fd54d5593eb3151c4f938d0 : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67e757416652476f934c73bccb35b7d5 : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e05bacd1dfe497f9fcba024d6995ed5 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0a6ff45184254a86be925c65ce04272e : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3968509de96141f29e2d2409e7b3fb10 : SUCCESS in 54m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed62579b27c043edba22ce5cdcff6249 : SUCCESS in 43m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cefdaa57695048358d1323ab41c5b6ae : SUCCESS in 43m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fde9e9be3deb4fbc9cc857817d40eaae : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44794c88399e44a7ac74edd847bebbba : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/89d4abf21ec6450e9f19c12a841bf43d : SUCCESS in 21m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11929865008d49b8bdf3e4eba59ece18 : SUCCESS in 52m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fc169aa5c6d4422a2b02204592d0bc3 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":71},{"id":"7c2b1b1768caac221d1797a74458893e8f9954b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-12 00:10:31.000000000","message":"Uploaded patch set 72.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":72},{"id":"353a060316a752217451f9da5990e95dedcb56e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 01:03:51.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/889b76468f34416587c84e0468bad646\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/15ad754f704647598ad97fc999e4e5c9 : SUCCESS in 51m 05s","accounts_in_message":[],"_revision_number":72},{"id":"1ad49ae337deeee8666b12a612ea57a58f3d0b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 01:23:52.000000000","message":"Patch Set 72: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8caa67d2c984077a381df8e14765493\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc33c8949f047c6b6eb564fd60afac0 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2182e6e817e140d6a5b6c26891cf6cd0 : SUCCESS in 55m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d1e76cd99ca4e26aea24ae2a38962ad : SUCCESS in 59m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f61e7016cae4cdc9601b6578007ec0c : SUCCESS in 48m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38dece9b70f5432194b9552a5ea6d35e : SUCCESS in 1h 06m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eeaa9ef88e04cda9b6157c0456912ea : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b059e1f05f8942daab17d55109119e17 : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad1606cf096943ed9a90236414679e0b : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/211dee884df34a578a88d59cc0b10cff : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9cd5131c13b14524a74b072c067b72bb : SUCCESS in 16m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5aef5dd9ee64407b8b943271e0c98189 : SUCCESS in 14m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e96771030f7a45e983bba8806b0dfcd5 : SUCCESS in 17m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a76a6d1b0fc44dd2ad89edd2ebf9615e : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/99ba35c4d3e7489cb9367a8a56871af4 : SUCCESS in 14m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d502b3384c8a47d3881b5273718a721b : SUCCESS in 14m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4e170adced44fab89858edd60a979b0 : SUCCESS in 5m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2138771286be4460ac8f4a03d1bd152f : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d59b53424a84aac8cca3b039a8f8b42 : FAILURE in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ab4e97374b149aab653f9f8df1d7ab5 : SUCCESS in 52m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb1ff3e7459343d89358a9d508dad35a : SUCCESS in 34m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d2dd87ff72644baa54287995016a187 : SUCCESS in 34m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e7faa1664e04b39973ce7234cdaeb78 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11095569afce47b9bf74f2b9606ff77d : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06e6a26e9d7e4eec81ca117593e9b815 : SUCCESS in 14m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0313640095f24424a7274c10cf2e1cdd : SUCCESS in 38m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a37b59bda5534bcfb1c1bb0ddb040727 : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":72},{"id":"4ba1d224dc518af0c7a4dd666967bd9a0edf0955","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-12 17:12:53.000000000","message":"Patch Set 72:\n\n(3 comments)","accounts_in_message":[],"_revision_number":72},{"id":"5ce572522cf53552a350578385cbfd106bcdaedf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-14 00:05:30.000000000","message":"Uploaded patch set 73.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":73},{"id":"5a0eae0ba25e0f5c29db13c3009de94ee38eae7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 00:36:04.000000000","message":"Patch Set 73:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a04e6428e0b4691b8bfa9f77a1f8069\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/443e2a9da73340b5a84c26fc4f9c838a : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":73},{"id":"7250026010c20795364a050f2b7f3fe075dff1c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 00:53:34.000000000","message":"Patch Set 73: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf9da16179445d78938f92e56fc3b61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba82dab3c814a6184ea43d119cbcb13 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/e18d0c220d1f4b64b0ea111e1329d8b9 : SUCCESS in 40m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e3f5e2e092341aba161951372dc76bb : SUCCESS in 41m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e672f60b8b9a424ba2b8cce3d0188afd : SUCCESS in 38m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e87f427f222d4f628024564e24d475b7 : SUCCESS in 46m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a509c12b3cf4894bcecb450ce1c1d0e : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f88a916ba8e47bb82cfed04b2d53f2d : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/330e79c520374aa58ad70254749d318f : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd1b25b640b9479ebec6c27f8e39c9d0 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02ddc2b50df5405fbba16dd2942e2403 : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e96c46a31adb4ca9894c6c6b266074a1 : SUCCESS in 8m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed66b0d667c54852a23514f6a4160341 : SUCCESS in 30m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8551d3da99c04ef1b288e05099e42a60 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d208ca776a3147e7a8d0e56ec539be97 : SUCCESS in 7m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b874a6088304bb0bb7ee4e8da6f3b6a : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a06a717464cf4eab8b8c2641362928b6 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f739b5013f34acb819563771d88876c : SUCCESS in 7m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6cae942c5a564f9280006c9f59115477 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/402c74fbee7441d38bb739e69058363d : SUCCESS in 34m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97ab17a808a24d598fcecf4bd84d560b : SUCCESS in 22m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fb7bc175287460ea6dee0b1a42df43e : SUCCESS in 30m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/588a79f0497045dab14e3b26e132db02 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c046809e8f1a439c90c41344a7d56326 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c153e9dc6cad4946a2344fd1150e71f5 : SUCCESS in 15m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e2875d6b8240eba60615598c80169d : SUCCESS in 34m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96e3492362334ca88107604ca61ddd21 : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":73},{"id":"5f56ce513927701e01c33f1c46956a8914e8d8bc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-16 18:09:08.000000000","message":"Patch Set 73:\n\n(8 comments)","accounts_in_message":[],"_revision_number":73},{"id":"6b934c8d30a0f733fa202e3f7cfb3c867596478c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 13:33:56.000000000","message":"Uploaded patch set 74.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":74},{"id":"2f89c331119bbce4ad144f968721e3511bdce00b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:06:52.000000000","message":"Patch Set 74:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7086368503a146c0b7a4d394bd692a5c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f3bbc51b05a4dd38bcc161db7e65429 : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":74},{"id":"29515e7b2a7dc1b1374c22c70d39169c08a55e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:51:53.000000000","message":"Patch Set 74: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faff1e3ad9b04a62a9084da30d22bf93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8c726889b44a23852333abc2aa1fe4 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c2b7fc9fb8bd4e38b9954730ce2f3dd2 : SUCCESS in 58m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b7b9e30461e498c9c6aef95db372706 : SUCCESS in 48m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0883ebef1f3c4c33b92f8f27657c62fe : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/437f76cc071642e68b055ed45f1617a6 : SUCCESS in 1h 05m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5446167c6d5042eb8588b9784e23b1cd : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/784048ec2463442ebc0fc577582dde41 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13070d11acae4099bbfddc224b3d5134 : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbb3822427cf44dc9560405196c40224 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e163f6fa4d5146cda36b415f48a91906 : SUCCESS in 14m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e973d66142bc4232b9f7f403eac1dc06 : SUCCESS in 17m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a4f436e2ef040ed913365573a614d2f : SUCCESS in 23m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d8c75ab08124834bdaccf3ade3ae00a : SUCCESS in 8m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d21dcf922ceb43598bde447201ddc38f : SUCCESS in 12m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e940f2d53f9c45e8a39d7c1bfcb2cb62 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c226fb7235541f0a032e117be75872f : SUCCESS in 8m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/275d319a2fd54c4eac8846a14f1e55ce : SUCCESS in 7m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e7caf57c76ef4d5c88d9839dbd1be16d : SUCCESS in 8m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/04d511ef9ea74188bf5336140169eca3 : SUCCESS in 49m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84ab0d6f23984bf0b51aba28b4d76141 : SUCCESS in 43m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9121d6a7d01a4dcfb14282cf4f1bac0b : SUCCESS in 18m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc9764b93ca04277a3eb80c9b8759811 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a343b5f8c3d74f9d8068c2bb4cf7f935 : SUCCESS in 2m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83f79b04464f4a2b9de620031748afb9 : SUCCESS in 20m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8af11ba102624a3bbfa57581f9e95dd0 : SUCCESS in 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/503943984ff94fe1b18136f59edc4916 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":74},{"id":"9280a5aba89f56fba97fa6bc378a8e570ba9a8c4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 15:37:01.000000000","message":"Patch Set 74:\n\n(7 comments)","accounts_in_message":[],"_revision_number":74},{"id":"12e0a770522e1cb462b8019cdefb9e9b4dccf2b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-21 22:24:59.000000000","message":"Uploaded patch set 75.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":75},{"id":"4a78c6dcc524a8f7b8c3c4993b973002f508cbfc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-21 22:36:30.000000000","message":"Patch Set 75:\n\n(5 comments)","accounts_in_message":[],"_revision_number":75},{"id":"237f544412d84624bfec4809556678bd4d30cfe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:16:13.000000000","message":"Patch Set 75:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888463fa096d400ab8fd5562e8181504\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19e7406ea03140128d2c72e89390eeaa : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":75},{"id":"d884ee016402b544c33719edf536d3444b92ca20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 23:34:27.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d433948bdab4b99ad562ca042a406db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f7ef994ef6416293852675e0c33163 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0fdec16ef0dc4cd29649d994cb79e5d8 : SUCCESS in 50m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6455dcdc81d4080b772c983f0983994 : SUCCESS in 55m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4dfe13df40634c289b11ad8e636d373a : SUCCESS in 49m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eddafac0e9f49ec93f3c976d9cf7015 : SUCCESS in 1h 02m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/733afd097ca848dea0bb095f967c1935 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5426cea1461423d8d6a0b5c9bbe5fa4 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1e7496638d34c61bc2891b29bca90ba : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6f9284170c91423b8910afb20831e7cf : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/822479ce14cc4067ab19ab31a794fc71 : SUCCESS in 18m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/356b5e49b7a047f59a1d9489efc16488 : SUCCESS in 18m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8e907d31822343efaa0071b1e7b7c767 : SUCCESS in 26m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/137573c8388749b28b096f499257e37a : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/28953475f24940dfb12cd5f17aef298d : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62e75fa87c1e4b03b6866a506673869f : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3d2c0e899dc4ec8bf21d4305856e519 : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2595455cc5af499fb847fa9f9179fb2c : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/331d2f7a550e4b289d1793f8470cbc5d : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25cbd9b2c304477c87187d7f236c574b : SUCCESS in 56m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3bad12eeb15f41f189e193f8eccbb3ae : SUCCESS in 18m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2ca12d1789f44615a24a56cae76b07bc : SUCCESS in 41m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a685c09fd10a443b94ac8c258287fe2b : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18feaedc05e4e599100b3a4d0bd45be : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02e6173fbb824cea9894a7fed94be8d7 : SUCCESS in 15m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb19e6a686c4a5c9244a1e952ec76e0 : SUCCESS in 43m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c52a3444b2743deacba3e17885dfd2a : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":75},{"id":"92eb42bec2f1130720fee8e0afa7308515c087f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-22 10:17:52.000000000","message":"Patch Set 75:\n\n(13 comments)","accounts_in_message":[],"_revision_number":75},{"id":"f53c3c4434014378ed5617884735b1629db430bf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-22 16:08:19.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"88b632cb872542cbbf0ca8ee294ffe8b1981c26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-22 19:31:35.000000000","message":"Uploaded patch set 76.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":76},{"id":"767ec3aeadbde7f8a81bf7cfa1131a005fe0d9f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 20:08:51.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77a806e927d492cb383c644d8f27ef6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4196e85f38d140e6b73d9f9c4bd1bc61 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":76},{"id":"78c8a09ea68b29c18762dbb773da242d6f8a7e83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 20:40:48.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222134b462274299b58c7f01e002ce9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b031d375590449c59bce2496e9c7d666 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/1384f3f8eae34299ae82186f33abd980 : SUCCESS in 44m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d5b080254b824623b957c85fd75929df : SUCCESS in 27m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a9d7b5fb58c44aaa8604aa7efadca00 : SUCCESS in 42m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/484e7943d56d4cd693d24e53d8d2ab1b : SUCCESS in 32m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7740f316e5e34be98979615be3117b08 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bc0c5cb4dac4e52a050c959d4b32b07 : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/438863ac89f9416cbb84228137f1e1bb : SUCCESS in 10m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32498f3e465a4171a0742ae1faa859cd : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b0b698acd284d28a10aea227b794b1b : SUCCESS in 7m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af6cf809c98c487cbf0b319d3039a938 : SUCCESS in 12m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d88e71862174649b34cd291e6acfdf7 : SUCCESS in 10m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a9dbc5e6636b4d5dbf50e7cb2ee37665 : SUCCESS in 7m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ba232cc62be64a1690426a5a165e7a3b : SUCCESS in 15m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0efb54092a244871b61c587daa8daff3 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df99b290933c47a8982c0e2cfe4b22a9 : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed0e192bd3e140ae8d2e0a2c5183bdc1 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bde75a26b64b49bb9f00e3a5fce415d4 : SUCCESS in 7m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93ae8df390894ca5a7b1cebd549ce2a5 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6027aa834ee14d979435507db1355663 : SUCCESS in 31m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e19b40690684d5b9dd4639b97332a9a : SUCCESS in 31m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f6347e77d8243fcbe35b7da7f2bfce4 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a044bdcb78124d7da84fc17892d8de26 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7723aa37c214dd4b74c171893600b1c : SUCCESS in 17m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4ad80a0e7cb41b6bc3b72afce98d7d1 : SUCCESS in 36m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6db9e6fbc984019919ea60f84de71ca : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":76},{"id":"6e04eaf69adac7a4b88f0cdb8f2cb5b6f515614f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-23 08:57:19.000000000","message":"Patch Set 76:\n\n(2 comments)","accounts_in_message":[],"_revision_number":76},{"id":"ef0118261af370511bfd995d2856e3af986f436d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-28 19:52:41.000000000","message":"Uploaded patch set 77.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":77},{"id":"01b77abda3199fe2354a2d2c06fa3bfd493acdd7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-28 19:52:48.000000000","message":"Patch Set 76:\n\n(6 comments)","accounts_in_message":[],"_revision_number":76},{"id":"87af1522d52c3a6b4f901b080022fbf504ed0fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 21:00:21.000000000","message":"Patch Set 77:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a323760d3e14985a00c2530628349c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/746c8e7c66344256883949eb3ff2200e : SUCCESS in 1h 05m 19s","accounts_in_message":[],"_revision_number":77},{"id":"eeb56310b3bdf0beacd4c97132fe38c3cd619c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 21:40:28.000000000","message":"Patch Set 77: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16315afb61224e19ab1454828bb593df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7ff3209d014f57a3015ed1fc8590cb : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/6681e434f5804e3eb22522ae52feb7ca : SUCCESS in 55m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f111b43b8904445db407bc4a317ceb32 : SUCCESS in 55m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98edec4d52c640f3be428ea3590309f6 : SUCCESS in 45m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc57de790a5b4240ad8348a05aba6c4f : SUCCESS in 1h 35m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae3f9881806949fba1377a6c8b113c94 : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92cbd78d3f364c318104073320beaba6 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73778ee053574f64844c19795326db90 : SUCCESS in 16m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86f2a6d92c9a48e0a2ec3c1afd3f5e98 : SUCCESS in 3m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/769bc6775df74639839cd6d4fa7abdf7 : SUCCESS in 18m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1937e4e72f1845c082151b864835f0f0 : SUCCESS in 19m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ec70e49884e549d1bfb7d6d43deb7312 : SUCCESS in 31m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0b705fef14cb4b4f8923724af8a29923 : SUCCESS in 27m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eb1db27a7373440f8c37efe6200fbc38 : SUCCESS in 7m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d55fce0fcd124c1db124f14baf50b5f4 : SUCCESS in 13m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c4148191922e474c934f0bd2b05aea31 : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ac4ef1393e644c4bd054dae6e66b907 : SUCCESS in 5m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4427faefc06341f393699351caab135f : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb1782ea69cc4df1b6f942d594664545 : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a593d93a9024b8392b7c89a354a8c30 : SUCCESS in 56m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/675723b92100485ea6ccfeebef5bffd9 : SUCCESS in 45m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0096b80ba69742e4869d863a695d605f : SUCCESS in 43m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/283d4730cced4fd299fe54abc06497d5 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414b93c98d8b47c9b3132163287cd7aa : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc5b1fa6a1454cc8b09ae3ec3ee4fae9 : SUCCESS in 20m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b9d4a7203444ea1967229aa7da1c6d3 : SUCCESS in 40m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e96f23eee88a4a1cb60f9acc64069626 : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":77},{"id":"3baf06d82ea8891fb4d5cc61f646bb4c5ea3e91b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-12 15:30:59.000000000","message":"Patch Set 77:\n\n(2 comments)","accounts_in_message":[],"_revision_number":77},{"id":"75c3c1c85fe3e1ae7137ef794a7c5c8758629846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 18:43:31.000000000","message":"Uploaded patch set 78.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":78},{"id":"2567a2c48e216cd6cc35d9228698df323670b6fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:32:15.000000000","message":"Patch Set 78:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd761fc1e2a346e6a892ab8660f94164\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab9950d0723b4ce29f0234773eed1397 : FAILURE in 43m 53s","accounts_in_message":[],"_revision_number":78},{"id":"3c41b92ebf57836016c1461335df8f74c72fa390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:50:55.000000000","message":"Patch Set 78: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/479d22eb1e0f4d858f05475a831e1ac3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3a44f952614d54b767986bd02bf463 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/278037ade6524e0a913ec6411548a579 : SUCCESS in 27m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/493d6d0f3ad34ef48991f994df04f655 : SUCCESS in 1h 00m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2161c0096bbc4c078c6c62255679ddb9 : SUCCESS in 39m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63325f3609774346b5f96d438ba32de1 : SUCCESS in 56m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fff506851c9a42149e249f052b49a4b8 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1bdbdbf840ed46ac99bd64a67938e0a5 : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c230a71f57346d29951668fe1c6d901 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63ac9bb86ef24342a4f72d04f69362c6 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc509280401e4ba48aa12eefe489571b : FAILURE in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e63a2f07b24a0b9651d724f9e128b0 : FAILURE in 15m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/950801cc32eb44e28db40e53047d940f : FAILURE in 31m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/89ce749a88d845e2a5d2e133dd3f8042 : FAILURE in 26m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d164200047f94d48b8bc0991b9794197 : SUCCESS in 9m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85e3cf80f2894b598c786a98f29bd3f9 : SUCCESS in 12m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ecd5140f84a467b90f177aac7a6199f : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/917a9e48e1e34bbbb9a945c047a08f81 : SUCCESS in 8m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9cdc2a4646d346988e0c1077b9b04a99 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16dfdf7ee8a5460b9ed5418ad87dc50a : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/efd49f50efba4d7aa42ee80a4f418711 : SUCCESS in 45m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e37071a17b474d88a35103636bc25c30 : SUCCESS in 46m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe94be2ca4e14837bbeb690d15926199 : SUCCESS in 41m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06d2ce22bddc48e4a9022b16cc78a27e : FAILURE in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e192e6b05654061a76b635335666a6a : SUCCESS in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c02c733d9bcc4444bc1ba146b93f97da : SUCCESS in 20m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7ed51ebfe824c6cb9d877b055752fcc : SUCCESS in 49m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5f7a30a2a7e46a3acf28d6dfd12b25c : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":78},{"id":"60e78b647a65067c55be8bdee146421f8996a3cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 20:42:35.000000000","message":"Uploaded patch set 79.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":79},{"id":"4005ed5396600681b4acb6d0187954cd976b7764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 21:36:37.000000000","message":"Patch Set 79: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc50fac38f943beb8a0924249be6de1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ad2d93f0604d61b1fbc292e683b762 : SUCCESS in 4m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0badbb3819f34aa09d5e617c4b836fa6 : SUCCESS in 27m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a5c9c28f31a428e9c1efbd2f2e5b2ea : SUCCESS in 44m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/340f9f51da634c5b9fba2a55302bdf71 : SUCCESS in 42m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95a4704deb184f89a769aa6cfdebef79 : SUCCESS in 52m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/275a485a2e6f4b04b0f26e36248feb98 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b22fba17a53b4327a3b0653a9027dd26 : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc5b7aa922e463eb96450ef215be036 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0a6adb4c9b44137a48f81bfd493e4d8 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a342088fcaa64b23a33d515e3ed77041 : SUCCESS in 20m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd42d7ddf544fa7b81c5452db70d039 : SUCCESS in 20m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81c48a20096244fca55e5beb3bb8194f : SUCCESS in 18m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4d6384b9b3db4c4eaf08fa8f72bf48f0 : SUCCESS in 24m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/77ef88a84d994faeab2a25d6d9f12f61 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8d03830f61b4f7bb8ec25a40e33f7a7 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d7505852b74745daa9ba43a515e847aa : SUCCESS in 9m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13af6e7c420d42038fb82219b08ea078 : SUCCESS in 7m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6d4b00e122345c2a7d620cecdb4a631 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7b146388b054212867fac2b21b62e41 : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27363031cfa24069a92894e0c61ebe19 : SUCCESS in 32m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8383ec19fe534bc4bd5a56d50362ba7e : SUCCESS in 23m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e615930ff0d44318afbccf8fef37663 : SUCCESS in 38m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a27178f3a26492ab1f9264657778f17 : SUCCESS in 21m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fd0aa13fe14abda5cb180ee958c9cd : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/058645e7671542a296dab7ffbcb171db : SUCCESS in 11m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e6cf9c7ee384599b26826f1079a3aa8 : SUCCESS in 48m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9ee53ab94354d49adbb92bf73897aac : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":79},{"id":"a442d092c359175cd9100f15642eff3de76df782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 21:40:27.000000000","message":"Patch Set 79:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407e29a144144a03a753af9deaa09c0c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a85c4d60eb0f4c68ba3bb979dd0c3450 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":79},{"id":"01c155094918e746cd14bc223442a277ba692088","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-13 09:06:37.000000000","message":"Patch Set 79: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":79},{"id":"384257cb3f1660c189ea5aac23665978d5a996c2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-13 12:53:03.000000000","message":"Patch Set 79: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"d20f988a5b404f3e380825d5691ab94858df8193","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-13 18:37:41.000000000","message":"Patch Set 79:\n\n(1 comment)","accounts_in_message":[],"_revision_number":79},{"id":"61dab533ed6147d9f1b92b0c8918a4d3685207b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-13 19:37:47.000000000","message":"Uploaded patch set 80: Patch Set 79 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":80},{"id":"1f5609d33378e0af16eb2ad741eff6ea848931dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-13 19:44:12.000000000","message":"Patch Set 80:\n\n(2 comments)","accounts_in_message":[],"_revision_number":80},{"id":"d73a1a25e65b7c982e701781bf605da1bdde8d36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:32:49.000000000","message":"Patch Set 80:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050e09c7527a401999de3892cd8ec8e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f80545c58a7e42e981f7913cd5a050c6 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":80},{"id":"8fa7f2e7c91fdcad1ae4b2d9a6a3b0b5852a819a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:47:26.000000000","message":"Patch Set 80: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cd6c0d993c4efab378c8deb86261ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8457e629dea48f89e4bf7de91a39756 : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/f25b663012ef4bf9b7f4ecf55a39b7e3 : SUCCESS in 30m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0358edaebbd54166aaa1f65779517311 : SUCCESS in 30m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e6c767e7fde483d8309fa1518961445 : SUCCESS in 31m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52aaaa1fe0d143d5974ee778bdf670a5 : SUCCESS in 58m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e13bed934a442038fce364e4d315a26 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eca9218be3d4487abe01a4940ed10961 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9efc4f81c53e44d6940d405945bcaed5 : SUCCESS in 19m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18f7fc1a4ea94bacb602cc15a3a67ff2 : SUCCESS in 3m 18s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb36957b4264a23a4b2ad9b99f43931 : SUCCESS in 9m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1e8f08096f4426e94c95678dc20eb26 : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5bd9be816b3b44569c2fbe8682637342 : SUCCESS in 33m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3e77966670e14362abfb3a4ab97b678e : SUCCESS in 19m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/85436549ec4c4d72ab23808206a9041b : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9db3ab250d2c4168a418e7b4a6407de5 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6e5be0b4ec44fd8bbef6de7d5e2ddd2 : SUCCESS in 9m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/000028f5190d46858691b905404b050b : SUCCESS in 7m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26e7f48d7c1b4b24be85971e53dec72b : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e477973a7ad4c50b7f112c0d086d949 : SUCCESS in 8m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc5d593694aa467095b0578096672649 : SUCCESS in 57m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aaaf490ba45e4a3aae89c26f93b13d81 : SUCCESS in 45m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3277bbe977b4c9f8f213e456674f468 : SUCCESS in 45m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b9405c9bd2f467986060884629e7317 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812783b4b3b746c79f3471f2a6840169 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7af757c4c4ad4202adbd33f4ab44a920 : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d033c50c6ae54ebe98561e01b34d747e : SUCCESS in 55m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0c361042b744c42a554aaaf72884f6c : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":80},{"id":"22106fb2094ee08dc2ffeeb168a8035a535d921b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-14 04:13:55.000000000","message":"Patch Set 80: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":80},{"id":"06a9814428cfd75c7a7247c6abfede84f3802b4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 04:15:02.000000000","message":"Patch Set 80: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":80},{"id":"33d979e8b3e7d9260c2497881939a1cbbc7b9f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:23:13.000000000","message":"Patch Set 80: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84fb94aa80064cfe855351c22b2a4a2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a3ccaefb0b455b8d90a7802af64ce2 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/e0fcefdfb06d4f5ead4ff47d57a4b79c : SUCCESS in 52m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70ba816da31a48e9a73b1dbad724934d : SUCCESS in 55m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/235978c8b6ed4eb4818b77652a46a348 : SUCCESS in 52m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e326894975c4447eaea9e1b885e4b285 : SUCCESS in 1h 04m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a57dd6ec6bdb475c8cd64fe1672404aa : POST_FAILURE in 6m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/217aaafe28c240cbae21c34ca2613f60 : SUCCESS in 18m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/101694ea0bf643a2b66b4aae340419e3 : SUCCESS in 7m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8c484084cc6543c697b271de97db6a62 : SUCCESS in 26m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/593f2f5c658341aa915960ee0d9bcd58 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/76a442a21ef74b4aba989efe06a22a06 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b72cf996e9ed4475af6ea83542b9ea14 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ce1ab808aa74958a0b052bfe3edf98a : SUCCESS in 5m 03s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e33054093554a56acb3f1ae2c1c8b2b : SUCCESS in 3m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2185f99564d47869d2ae50318baed62 : SUCCESS in 58m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c33f533027c24045a7490c415a1d9967 : SUCCESS in 44m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d493d2e9c2af4305830cfa04687d5a51 : SUCCESS in 28m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36837079812d4e3eba41c1bb03407d53 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b14132f3dd43ce91098a0bd5bba68a : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3247005e5f26429699752141a039bdde : SUCCESS in 14m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a02a8472a6594544a97c59959f1a5f9d : SUCCESS in 43m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b1f4be73ac54799bec0b928f15484a7 : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":80},{"id":"58d53a9c36f416aacf6e9dca557f383b8afcab73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:23:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":80},{"id":"c68105a6b967c465049b7179c9f5f2d405423984","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:24:05.000000000","message":"Patch Set 80:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/809352a3ecd14f60b733922a5c49a6cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052bb701faee400ca6ecb1052d52eff6 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/364dac89e08c4eaca85e20cce29c9d3f : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":80}],"current_revision_number":80,"current_revision":"f9ac22971f6b08878b8b038855cd0ee7d93f7fd5","revisions":{"164a1604143ffb28ed1f7e57ce8a151b56f4f838":{"kind":"REWORK","_number":1,"created":"2022-04-06 09:20:01.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/1"}}},"commit":{"parents":[{"commit":"014c98e853b86fc44312d879bec12099be6e22d0","subject":"Merge \"s3api: Fix multi_delete with object names using non-ASCII characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/014c98e853b86fc44312d879bec12099be6e22d0"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-06 09:13:56.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in the\nAmazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/164a1604143ffb28ed1f7e57ce8a151b56f4f838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/164a1604143ffb28ed1f7e57ce8a151b56f4f838"}]},"branch":"refs/heads/master"},"fb55d5fa0d1b5dec0758c19097214006a06bca47":{"kind":"REWORK","_number":2,"created":"2022-04-06 13:17:51.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/2"}}},"commit":{"parents":[{"commit":"3d7ff12064709fd77eb95cd2e2c3276707d5b5d6","subject":"Merge \"Obj Auditor: Quarantine ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7ff12064709fd77eb95cd2e2c3276707d5b5d6"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-06 13:17:38.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in the\nAmazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb55d5fa0d1b5dec0758c19097214006a06bca47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb55d5fa0d1b5dec0758c19097214006a06bca47"}]},"branch":"refs/heads/master"},"f32dbb66920bb02c095c7983fd6eb0d23a09818f":{"kind":"REWORK","_number":3,"created":"2022-04-06 13:40:14.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/3"}}},"commit":{"parents":[{"commit":"3d7ff12064709fd77eb95cd2e2c3276707d5b5d6","subject":"Merge \"Obj Auditor: Quarantine ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7ff12064709fd77eb95cd2e2c3276707d5b5d6"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-06 13:39:56.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f32dbb66920bb02c095c7983fd6eb0d23a09818f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f32dbb66920bb02c095c7983fd6eb0d23a09818f"}]},"branch":"refs/heads/master"},"90242b6429ce0dc85de665892de1fac2610f3755":{"kind":"REWORK","_number":4,"created":"2022-04-13 10:17:21.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/4"}}},"commit":{"parents":[{"commit":"1cee51d5269c2b04f13623593710d2be0045781d","subject":"doc: also add reverse option to pagination doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cee51d5269c2b04f13623593710d2be0045781d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-13 10:16:31.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90242b6429ce0dc85de665892de1fac2610f3755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90242b6429ce0dc85de665892de1fac2610f3755"}]},"branch":"refs/heads/master"},"8aeb6ec4b347755cca139e58e72668b0a83c4dd3":{"kind":"REWORK","_number":5,"created":"2022-04-13 12:50:38.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/5"}}},"commit":{"parents":[{"commit":"1cee51d5269c2b04f13623593710d2be0045781d","subject":"doc: also add reverse option to pagination doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cee51d5269c2b04f13623593710d2be0045781d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-13 12:50:26.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aeb6ec4b347755cca139e58e72668b0a83c4dd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aeb6ec4b347755cca139e58e72668b0a83c4dd3"}]},"branch":"refs/heads/master"},"e150265039b867afacf716b8a4b050264348f35a":{"kind":"REWORK","_number":6,"created":"2022-04-14 07:50:38.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/6"}}},"commit":{"parents":[{"commit":"1cee51d5269c2b04f13623593710d2be0045781d","subject":"doc: also add reverse option to pagination doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cee51d5269c2b04f13623593710d2be0045781d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-14 07:50:25.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e150265039b867afacf716b8a4b050264348f35a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e150265039b867afacf716b8a4b050264348f35a"}]},"branch":"refs/heads/master"},"dfaaaf671097ab09f3c2b00cd1ca8da466c486cc":{"kind":"REWORK","_number":7,"created":"2022-07-01 14:00:44.000000000","uploader":{"_account_id":34643,"name":"Thibault Person","email":"thibaultperson@gmail.com","username":"thibaultp"},"ref":"refs/changes/55/836755/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/7"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-07-01 13:56:00.000000000","tz":120},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfaaaf671097ab09f3c2b00cd1ca8da466c486cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dfaaaf671097ab09f3c2b00cd1ca8da466c486cc"}]},"branch":"refs/heads/master"},"a272f57105d6eec7eee356e56daeaa18183f97ec":{"kind":"REWORK","_number":8,"created":"2022-11-04 20:59:05.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/8"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-11-04 20:58:55.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nFixes LP Bug #1810026\n\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a272f57105d6eec7eee356e56daeaa18183f97ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a272f57105d6eec7eee356e56daeaa18183f97ec"}]},"branch":"refs/heads/master"},"2236731d7bc66fde44a63684a6a977fb08a9609a":{"kind":"REWORK","_number":9,"created":"2024-02-09 00:22:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/9"}}},"commit":{"parents":[{"commit":"51ae9b00c917d963d0477120b4a5a20373ea79e8","subject":"Merge \"lint: Consistently use assertIsInstance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51ae9b00c917d963d0477120b4a5a20373ea79e8"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-09 00:19:32.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2236731d7bc66fde44a63684a6a977fb08a9609a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2236731d7bc66fde44a63684a6a977fb08a9609a"}]},"branch":"refs/heads/master"},"6bd7b51edcf27c447345ea8312a9fe7407ce1e98":{"kind":"REWORK","_number":10,"created":"2024-02-27 23:42:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/10"}}},"commit":{"parents":[{"commit":"d3d6bdbbbec51ee238f0599f537f2ce030a88b03","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3d6bdbbbec51ee238f0599f537f2ce030a88b03"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-27 23:01:38.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bd7b51edcf27c447345ea8312a9fe7407ce1e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6bd7b51edcf27c447345ea8312a9fe7407ce1e98"}]},"branch":"refs/heads/master"},"2a4c81da02c1ff1c96ab5d6abf7dcf1a69cba366":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-02-28 00:13:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/11"}}},"commit":{"parents":[{"commit":"41761a6c66a2ce682fee902f2354c7efa6abf949","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41761a6c66a2ce682fee902f2354c7efa6abf949"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-28 00:13:30.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a4c81da02c1ff1c96ab5d6abf7dcf1a69cba366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a4c81da02c1ff1c96ab5d6abf7dcf1a69cba366"}]},"branch":"refs/heads/master"},"3ebf7cf1d7c76bcfe0eaf05bfec74593c4c4adf7":{"kind":"REWORK","_number":12,"created":"2024-03-01 05:18:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/12"}}},"commit":{"parents":[{"commit":"499f829be3e3ea5ec95695feada0ff244876b4ed","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/499f829be3e3ea5ec95695feada0ff244876b4ed"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-01 05:17:18.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ebf7cf1d7c76bcfe0eaf05bfec74593c4c4adf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ebf7cf1d7c76bcfe0eaf05bfec74593c4c4adf7"}]},"branch":"refs/heads/master"},"98a1732ee14c2ea41e2f0eb20d505a1a842ebf4f":{"kind":"REWORK","_number":13,"created":"2024-03-02 01:02:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/13"}}},"commit":{"parents":[{"commit":"0cc26df11e5d139120ca045c77e9d6a09a52bc24","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cc26df11e5d139120ca045c77e9d6a09a52bc24"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-02 01:02:07.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98a1732ee14c2ea41e2f0eb20d505a1a842ebf4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98a1732ee14c2ea41e2f0eb20d505a1a842ebf4f"}]},"branch":"refs/heads/master"},"847f0d3a651f3288dafb1c6dc69fa7e72f03c8de":{"kind":"REWORK","_number":14,"created":"2024-03-02 15:47:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/14"}}},"commit":{"parents":[{"commit":"0cc26df11e5d139120ca045c77e9d6a09a52bc24","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0cc26df11e5d139120ca045c77e9d6a09a52bc24"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-02 15:46:41.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/847f0d3a651f3288dafb1c6dc69fa7e72f03c8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/847f0d3a651f3288dafb1c6dc69fa7e72f03c8de"}]},"branch":"refs/heads/master"},"21bcec68ee162857d2208dfca93609acbb0eb1e9":{"kind":"REWORK","_number":15,"created":"2024-03-15 00:39:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/15"}}},"commit":{"parents":[{"commit":"3d1e2f73b071d82c393a421d989fd061eaa48b65","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d1e2f73b071d82c393a421d989fd061eaa48b65"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-15 00:31:28.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21bcec68ee162857d2208dfca93609acbb0eb1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21bcec68ee162857d2208dfca93609acbb0eb1e9"}]},"branch":"refs/heads/master"},"5c7d2bfad33df30aa79881029a19b55ef3bf49d1":{"kind":"REWORK","_number":16,"created":"2024-03-21 05:46:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/16"}}},"commit":{"parents":[{"commit":"0c79a39813983ff9ed342b9363fc16ff5a95ca8e","subject":"WIP: s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c79a39813983ff9ed342b9363fc16ff5a95ca8e"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-21 05:46:38.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c7d2bfad33df30aa79881029a19b55ef3bf49d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c7d2bfad33df30aa79881029a19b55ef3bf49d1"}]},"branch":"refs/heads/master"},"73a0866f131f2bde6a6cad5a2f8a0f143e22b17a":{"kind":"REWORK","_number":17,"created":"2024-03-22 22:49:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/17"}}},"commit":{"parents":[{"commit":"0db628280a3183d78651294638142e5dd773b2a0","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0db628280a3183d78651294638142e5dd773b2a0"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-22 22:48:05.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73a0866f131f2bde6a6cad5a2f8a0f143e22b17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73a0866f131f2bde6a6cad5a2f8a0f143e22b17a"}]},"branch":"refs/heads/master"},"33f958b4d02b6aa9fb636622985107b2e5c2d62c":{"kind":"REWORK","_number":18,"created":"2024-03-25 18:47:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/18"}}},"commit":{"parents":[{"commit":"d355b1e87e3401462c2b1e712c46339341d2a814","subject":"DNM: Try centos8 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d355b1e87e3401462c2b1e712c46339341d2a814"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 18:47:29.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33f958b4d02b6aa9fb636622985107b2e5c2d62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33f958b4d02b6aa9fb636622985107b2e5c2d62c"}]},"branch":"refs/heads/master"},"1150d489ce4bad144328bd4279e208140a525a96":{"kind":"REWORK","_number":19,"created":"2024-03-25 22:06:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/19"}}},"commit":{"parents":[{"commit":"9808fb14c6d27c46839d0698c6000a9067a38bbc","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9808fb14c6d27c46839d0698c6000a9067a38bbc"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-25 22:05:52.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1150d489ce4bad144328bd4279e208140a525a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1150d489ce4bad144328bd4279e208140a525a96"}]},"branch":"refs/heads/master"},"d21036fcb628c4676402c7b45fc132d2947ba717":{"kind":"REWORK","_number":20,"created":"2024-03-28 05:02:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/20"}}},"commit":{"parents":[{"commit":"4de37abf6747f01caafb13e2a173abd89addc538","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4de37abf6747f01caafb13e2a173abd89addc538"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-03-27 18:10:56.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d21036fcb628c4676402c7b45fc132d2947ba717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d21036fcb628c4676402c7b45fc132d2947ba717"}]},"branch":"refs/heads/master"},"ab14dd41af254171ac7c79923089add04dd67b13":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-04-01 20:22:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/21"}}},"commit":{"parents":[{"commit":"7295a792b02a09aa518a46d7cc401fda099f8c7c","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7295a792b02a09aa518a46d7cc401fda099f8c7c"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2024-04-01 20:22:21.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab14dd41af254171ac7c79923089add04dd67b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab14dd41af254171ac7c79923089add04dd67b13"}]},"branch":"refs/heads/master","description":"Rebase"},"c96f19bf2c300372b457709bae9de26c8472a586":{"kind":"REWORK","_number":22,"created":"2024-06-17 23:18:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/22"}}},"commit":{"parents":[{"commit":"8205fd5ded0d2189bde5a7f6aeeb4e45529c8cb8","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8205fd5ded0d2189bde5a7f6aeeb4e45529c8cb8"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 23:17:18.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c96f19bf2c300372b457709bae9de26c8472a586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c96f19bf2c300372b457709bae9de26c8472a586"}]},"branch":"refs/heads/master"},"300a083363e6a101f169072cb654673090e9c022":{"kind":"REWORK","_number":23,"created":"2024-07-02 00:25:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/23"}}},"commit":{"parents":[{"commit":"dd7e69b4b75d338c00c3332645feaaefdd822dec","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd7e69b4b75d338c00c3332645feaaefdd822dec"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-02 00:24:47.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/300a083363e6a101f169072cb654673090e9c022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/300a083363e6a101f169072cb654673090e9c022"}]},"branch":"refs/heads/master"},"d2226fb61eb188bf2e2d7d49554619a587510df9":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-08-16 00:11:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/24"}}},"commit":{"parents":[{"commit":"f9eaac6fcb118a050e9d6799b28bc566729fa6a4","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9eaac6fcb118a050e9d6799b28bc566729fa6a4"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-16 00:09:40.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2226fb61eb188bf2e2d7d49554619a587510df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2226fb61eb188bf2e2d7d49554619a587510df9"}]},"branch":"refs/heads/master"},"20a53d89c44504faa30a80b90fa29ed8d23b6ce5":{"kind":"REWORK","_number":25,"created":"2024-08-22 00:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/25"}}},"commit":{"parents":[{"commit":"7301f9dca42ccb02674e65c88f12604688a2e316","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7301f9dca42ccb02674e65c88f12604688a2e316"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-21 23:52:12.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a53d89c44504faa30a80b90fa29ed8d23b6ce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20a53d89c44504faa30a80b90fa29ed8d23b6ce5"}]},"branch":"refs/heads/master"},"58a3e83a3bb6a03ad2b035f551a53fb274bf7e2e":{"kind":"REWORK","_number":26,"created":"2024-08-22 22:20:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/26"}}},"commit":{"parents":[{"commit":"64d047d4cb8b37d3c0f344e6c2b495d88b69350a","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64d047d4cb8b37d3c0f344e6c2b495d88b69350a"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2022-04-05 20:55:56.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-22 22:20:35.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58a3e83a3bb6a03ad2b035f551a53fb274bf7e2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58a3e83a3bb6a03ad2b035f551a53fb274bf7e2e"}]},"branch":"refs/heads/master"},"e786ea229d537d61425c7bf170c7203fb697abbe":{"kind":"REWORK","_number":27,"created":"2024-08-23 19:22:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/27"}}},"commit":{"parents":[{"commit":"760003f51a4ae5be94ceb97b1aa4b10bf0961297","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760003f51a4ae5be94ceb97b1aa4b10bf0961297"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:21:46.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e786ea229d537d61425c7bf170c7203fb697abbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e786ea229d537d61425c7bf170c7203fb697abbe"}]},"branch":"refs/heads/master"},"a683db748e917a631d1f0d4d7273a70341c572b2":{"kind":"REWORK","_number":28,"created":"2024-08-23 23:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/28"}}},"commit":{"parents":[{"commit":"bf87b1d073c1efa9fe049587672ea6b0a747c34d","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf87b1d073c1efa9fe049587672ea6b0a747c34d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 23:41:18.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a683db748e917a631d1f0d4d7273a70341c572b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a683db748e917a631d1f0d4d7273a70341c572b2"}]},"branch":"refs/heads/master"},"1fca1fcd7df462b4682e8221cb4d362ad5319f46":{"kind":"REWORK","_number":29,"created":"2024-09-20 23:29:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/29"}}},"commit":{"parents":[{"commit":"01fd53a63f8b179192c6d1c0ff8a1f2c72ac47f4","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01fd53a63f8b179192c6d1c0ff8a1f2c72ac47f4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-20 23:28:59.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fca1fcd7df462b4682e8221cb4d362ad5319f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1fca1fcd7df462b4682e8221cb4d362ad5319f46"}]},"branch":"refs/heads/master"},"13890ceb31ea5813f2a156b1b266fbac0058d11b":{"kind":"REWORK","_number":30,"created":"2024-09-23 00:06:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/30"}}},"commit":{"parents":[{"commit":"59e9316eee851009ed2b49c475a2e63154253394","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59e9316eee851009ed2b49c475a2e63154253394"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-23 00:06:27.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13890ceb31ea5813f2a156b1b266fbac0058d11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13890ceb31ea5813f2a156b1b266fbac0058d11b"}]},"branch":"refs/heads/master"},"79cc71763bf1750046f7d1c70f5b77da08991013":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-09-23 23:21:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/31"}}},"commit":{"parents":[{"commit":"8e2fddfd4d5ec8b0c1e9a85b2f8e3e631ba3a2dc","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e2fddfd4d5ec8b0c1e9a85b2f8e3e631ba3a2dc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-23 23:21:32.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79cc71763bf1750046f7d1c70f5b77da08991013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79cc71763bf1750046f7d1c70f5b77da08991013"}]},"branch":"refs/heads/master","description":"Rebase"},"4cf1745f2d9d51280c1b998cfcbccb1ea1cfd2c2":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-09-24 15:49:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/32"}}},"commit":{"parents":[{"commit":"d59e4c7fde92d7b7535418f5bbdc7e07cc7c4be0","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d59e4c7fde92d7b7535418f5bbdc7e07cc7c4be0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-24 15:49:16.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cf1745f2d9d51280c1b998cfcbccb1ea1cfd2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cf1745f2d9d51280c1b998cfcbccb1ea1cfd2c2"}]},"branch":"refs/heads/master"},"36c87461e25cad925ba871aa7b713ccae3339043":{"kind":"REWORK","_number":33,"created":"2024-09-27 23:23:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/33"}}},"commit":{"parents":[{"commit":"0fb66c71224da84335a0c2c0c76226077189d4e6","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fb66c71224da84335a0c2c0c76226077189d4e6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-27 23:22:33.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36c87461e25cad925ba871aa7b713ccae3339043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36c87461e25cad925ba871aa7b713ccae3339043"}]},"branch":"refs/heads/master"},"ffe6c26d67868ab64986e93c38a1aba0d4ee5117":{"kind":"REWORK","_number":34,"created":"2024-09-27 23:33:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/34"}}},"commit":{"parents":[{"commit":"0fb66c71224da84335a0c2c0c76226077189d4e6","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fb66c71224da84335a0c2c0c76226077189d4e6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-27 23:32:48.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffe6c26d67868ab64986e93c38a1aba0d4ee5117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffe6c26d67868ab64986e93c38a1aba0d4ee5117"}]},"branch":"refs/heads/master"},"5e239378de1b4db9c4f2ee04a3ac06f8ffaf3416":{"kind":"REWORK","_number":35,"created":"2024-10-01 23:18:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/35"}}},"commit":{"parents":[{"commit":"388ba709932b56639c91a41461475685bf20e3da","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388ba709932b56639c91a41461475685bf20e3da"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-01 23:17:51.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e239378de1b4db9c4f2ee04a3ac06f8ffaf3416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e239378de1b4db9c4f2ee04a3ac06f8ffaf3416"}]},"branch":"refs/heads/master"},"09b2b44a023e67cad88ebcab8b208ef2eb3eb60d":{"kind":"REWORK","_number":36,"created":"2024-10-04 20:10:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/36"}}},"commit":{"parents":[{"commit":"777e7ab7041e96f21a55d85ef2cad6f9fef378e6","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/777e7ab7041e96f21a55d85ef2cad6f9fef378e6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-04 20:09:19.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b2b44a023e67cad88ebcab8b208ef2eb3eb60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b2b44a023e67cad88ebcab8b208ef2eb3eb60d"}]},"branch":"refs/heads/master"},"9de413678e90b070cf04a3c2eb1ceec068d9ad5b":{"kind":"REWORK","_number":37,"created":"2024-10-10 21:54:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/37"}}},"commit":{"parents":[{"commit":"78b7e79ed80af5b34539a438f3eb5087591b29c1","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b7e79ed80af5b34539a438f3eb5087591b29c1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-10-10 21:54:12.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9de413678e90b070cf04a3c2eb1ceec068d9ad5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9de413678e90b070cf04a3c2eb1ceec068d9ad5b"}]},"branch":"refs/heads/master"},"6e06604eeb0a21adbb24e4cbe9035e2758120b9c":{"kind":"REWORK","_number":38,"created":"2024-10-16 23:34:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/38"}}},"commit":{"parents":[{"commit":"78b7e79ed80af5b34539a438f3eb5087591b29c1","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b7e79ed80af5b34539a438f3eb5087591b29c1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-16 23:34:01.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e06604eeb0a21adbb24e4cbe9035e2758120b9c"}]},"branch":"refs/heads/master"},"b9746fd0ee6018aa1467e746c2c0d72315866794":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2024-11-04 19:28:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/39"}}},"commit":{"parents":[{"commit":"bc81d998ba7d60b39ec322dc82026a16b7c086ea","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc81d998ba7d60b39ec322dc82026a16b7c086ea"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-11-04 19:28:55.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9746fd0ee6018aa1467e746c2c0d72315866794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9746fd0ee6018aa1467e746c2c0d72315866794"}]},"branch":"refs/heads/master","description":"Rebase"},"cb97e987a523ab7fb6ff09566f940f4f2e3f6be3":{"kind":"REWORK","_number":40,"created":"2024-11-15 23:29:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/40"}}},"commit":{"parents":[{"commit":"f07b501dfd573de7ae2413c1e122be91a0ee1d5a","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f07b501dfd573de7ae2413c1e122be91a0ee1d5a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-11-15 23:29:40.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb97e987a523ab7fb6ff09566f940f4f2e3f6be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb97e987a523ab7fb6ff09566f940f4f2e3f6be3"}]},"branch":"refs/heads/master"},"a52f03f3b385781012e7ecd60c21b341bed882aa":{"kind":"REWORK","_number":41,"created":"2025-01-21 23:42:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/55/836755/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/41"}}},"commit":{"parents":[{"commit":"1ba3ff2398a451782839aa68634c34dda3d6529c","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ba3ff2398a451782839aa68634c34dda3d6529c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2025-01-21 23:42:50.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52f03f3b385781012e7ecd60c21b341bed882aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a52f03f3b385781012e7ecd60c21b341bed882aa"}]},"branch":"refs/heads/master","description":"Rebase"},"4b75f298b8be85cabb6fa128f1c248f57dbb09de":{"kind":"REWORK","_number":42,"created":"2025-01-22 14:43:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/42"}}},"commit":{"parents":[{"commit":"45d665c488f44d9898dfc2b776a0cc319dc9a618","subject":"s3api: Additional checksums for MPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45d665c488f44d9898dfc2b776a0cc319dc9a618"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-22 12:07:48.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b75f298b8be85cabb6fa128f1c248f57dbb09de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b75f298b8be85cabb6fa128f1c248f57dbb09de"}]},"branch":"refs/heads/master"},"f696c6a6988d4534b23c331b5bbabadde717d2bb":{"kind":"REWORK","_number":43,"created":"2025-01-24 21:41:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/43"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-24 21:40:17.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f696c6a6988d4534b23c331b5bbabadde717d2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f696c6a6988d4534b23c331b5bbabadde717d2bb"}]},"branch":"refs/heads/master"},"98747b0b7994f570cfc6d2ae9bc1a844d75bc74d":{"kind":"REWORK","_number":44,"created":"2025-01-30 19:57:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/44"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-30 19:52:52.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98747b0b7994f570cfc6d2ae9bc1a844d75bc74d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98747b0b7994f570cfc6d2ae9bc1a844d75bc74d"}]},"branch":"refs/heads/master"},"b69846df27887f311c8030bbebdb332816945062":{"kind":"REWORK","_number":45,"created":"2025-01-31 02:51:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/45"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-01-31 02:50:07.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69846df27887f311c8030bbebdb332816945062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69846df27887f311c8030bbebdb332816945062"}]},"branch":"refs/heads/master"},"1ee623e6cce7fd26074a6e8a8c1111fba97d97e4":{"kind":"REWORK","_number":46,"created":"2025-02-01 02:38:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/46"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-01 02:38:07.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ee623e6cce7fd26074a6e8a8c1111fba97d97e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ee623e6cce7fd26074a6e8a8c1111fba97d97e4"}]},"branch":"refs/heads/master"},"3411d821efbb5c36bdd4be81ee90f1cebbec6252":{"kind":"REWORK","_number":47,"created":"2025-02-03 21:38:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/47"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-03 21:38:32.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3411d821efbb5c36bdd4be81ee90f1cebbec6252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3411d821efbb5c36bdd4be81ee90f1cebbec6252"}]},"branch":"refs/heads/master"},"941bc57d8b6b218261a8138307a983609808d3c3":{"kind":"REWORK","_number":48,"created":"2025-02-03 22:07:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/48"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-03 22:07:35.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941bc57d8b6b218261a8138307a983609808d3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/941bc57d8b6b218261a8138307a983609808d3c3"}]},"branch":"refs/heads/master"},"cf278059909f3bc56712eaf662d793cbe6650965":{"kind":"REWORK","_number":49,"created":"2025-02-03 22:39:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/49"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:14:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-03 22:38:59.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf278059909f3bc56712eaf662d793cbe6650965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf278059909f3bc56712eaf662d793cbe6650965"}]},"branch":"refs/heads/master"},"7539776b363b86f33e902c6da61844a9ac72291b":{"kind":"REWORK","_number":50,"created":"2025-02-07 18:31:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/50"}}},"commit":{"parents":[{"commit":"0dfa38d025bcb254849e8061af05e6214b5fdacf","subject":"docs: Fix version call-out for stale_worker_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dfa38d025bcb254849e8061af05e6214b5fdacf"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7539776b363b86f33e902c6da61844a9ac72291b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7539776b363b86f33e902c6da61844a9ac72291b"}]},"branch":"refs/heads/master"},"053f9ba7a2562a71430563ff66b6c0b91bbb339f":{"kind":"REWORK","_number":51,"created":"2025-02-13 01:03:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/51"}}},"commit":{"parents":[{"commit":"73201cc12dbd0a53bb83b0acf4ed432d5a1b3317","subject":"Merge \"Deprecate expirer options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73201cc12dbd0a53bb83b0acf4ed432d5a1b3317"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-13 01:02:01.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/053f9ba7a2562a71430563ff66b6c0b91bbb339f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/053f9ba7a2562a71430563ff66b6c0b91bbb339f"}]},"branch":"refs/heads/master"},"cbfce216643ee593efc0f1e72c1087b48bc1a4cf":{"kind":"REWORK","_number":52,"created":"2025-02-13 15:00:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/52"}}},"commit":{"parents":[{"commit":"73201cc12dbd0a53bb83b0acf4ed432d5a1b3317","subject":"Merge \"Deprecate expirer options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73201cc12dbd0a53bb83b0acf4ed432d5a1b3317"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-02-13 14:59:16.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfce216643ee593efc0f1e72c1087b48bc1a4cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbfce216643ee593efc0f1e72c1087b48bc1a4cf"}]},"branch":"refs/heads/master"},"1b38348be4bb37c9e335d794b7e52eaff1e9edfb":{"kind":"REWORK","_number":53,"created":"2025-02-13 18:27:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/53"}}},"commit":{"parents":[{"commit":"11557559b36fe566fefba49ad3f87b0b5d8b8ce3","subject":"Merge \"versioning: 411 PUTs with neither content-length nor transfer-encoding\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11557559b36fe566fefba49ad3f87b0b5d8b8ce3"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-13 18:27:24.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b38348be4bb37c9e335d794b7e52eaff1e9edfb"}]},"branch":"refs/heads/master"},"e490cc0924ffdc26f40eb850d71e554aa9ea396a":{"kind":"REWORK","_number":54,"created":"2025-02-14 02:34:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/54"}}},"commit":{"parents":[{"commit":"fc88b0e81644da2ac368526a9a5a25b0ee590fe4","subject":"Merge \"Add NotificationServer to fix swift-reload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc88b0e81644da2ac368526a9a5a25b0ee590fe4"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 02:33:32.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e490cc0924ffdc26f40eb850d71e554aa9ea396a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e490cc0924ffdc26f40eb850d71e554aa9ea396a"}]},"branch":"refs/heads/master"},"4a8acad77fdeb3adf0f5e5683ff5f4ec04c52412":{"kind":"REWORK","_number":55,"created":"2025-02-14 04:22:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/55"}}},"commit":{"parents":[{"commit":"fc88b0e81644da2ac368526a9a5a25b0ee590fe4","subject":"Merge \"Add NotificationServer to fix swift-reload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc88b0e81644da2ac368526a9a5a25b0ee590fe4"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 04:22:41.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a8acad77fdeb3adf0f5e5683ff5f4ec04c52412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a8acad77fdeb3adf0f5e5683ff5f4ec04c52412"}]},"branch":"refs/heads/master"},"030d6272a2b20c7db308e71cf0c30ef31e699944":{"kind":"REWORK","_number":56,"created":"2025-02-14 04:32:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/56"}}},"commit":{"parents":[{"commit":"fc88b0e81644da2ac368526a9a5a25b0ee590fe4","subject":"Merge \"Add NotificationServer to fix swift-reload\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc88b0e81644da2ac368526a9a5a25b0ee590fe4"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-14 04:31:22.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/030d6272a2b20c7db308e71cf0c30ef31e699944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/030d6272a2b20c7db308e71cf0c30ef31e699944"}]},"branch":"refs/heads/master"},"4185af65fac33a5ba42b2a46b40ac281646bd66d":{"kind":"REWORK","_number":57,"created":"2025-02-15 16:43:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/57"}}},"commit":{"parents":[{"commit":"427615a874342fc77d9540324dda0ceda71f51b8","subject":"Merge \"s3api: Stop requiring Content-MD5 for multi-deletes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/427615a874342fc77d9540324dda0ceda71f51b8"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-15 16:43:28.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4185af65fac33a5ba42b2a46b40ac281646bd66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4185af65fac33a5ba42b2a46b40ac281646bd66d"}]},"branch":"refs/heads/master"},"baa4a10091c5d6f841b88344db0d8adaa47721b9":{"kind":"REWORK","_number":58,"created":"2025-02-21 04:53:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/58"}}},"commit":{"parents":[{"commit":"a75329b4205e37938c15e40c7ab8b4508487762a","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a75329b4205e37938c15e40c7ab8b4508487762a"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-21 04:52:34.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baa4a10091c5d6f841b88344db0d8adaa47721b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baa4a10091c5d6f841b88344db0d8adaa47721b9"}]},"branch":"refs/heads/master"},"570a481aa77d36c1f110df42fa187990ee765fac":{"kind":"REWORK","_number":59,"created":"2025-02-26 18:24:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/59"}}},"commit":{"parents":[{"commit":"fa22e0c08429ecc9630a22e6709303ed331ac8a4","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa22e0c08429ecc9630a22e6709303ed331ac8a4"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 18:23:59.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/570a481aa77d36c1f110df42fa187990ee765fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/570a481aa77d36c1f110df42fa187990ee765fac"}]},"branch":"refs/heads/master"},"84bbbca32aa5c531024db4e97d3afce1bf1d3fef":{"kind":"NO_CHANGE","_number":60,"created":"2025-02-26 18:40:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/60"}}},"commit":{"parents":[{"commit":"e4cc228ed09eb30a4e186515f801d9cde0140c4d","subject":"Refactor some file-like iters as utils.InputProxy subclasses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4cc228ed09eb30a4e186515f801d9cde0140c4d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-02-26 18:40:40.000000000","tz":-480},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84bbbca32aa5c531024db4e97d3afce1bf1d3fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84bbbca32aa5c531024db4e97d3afce1bf1d3fef"}]},"branch":"refs/heads/master"},"dbb1ac227f5acb5f1279c2cd82cb7828aee0bae7":{"kind":"REWORK","_number":61,"created":"2025-03-11 17:41:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/61"}}},"commit":{"parents":[{"commit":"3f59cef644ce5b4a520cd818c97ed22cb73b731f","subject":"Merge \"Make FakeSwiftCall.DUMMY_VALUE a bespoke singleton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f59cef644ce5b4a520cd818c97ed22cb73b731f"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-11 17:30:50.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb1ac227f5acb5f1279c2cd82cb7828aee0bae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb1ac227f5acb5f1279c2cd82cb7828aee0bae7"}]},"branch":"refs/heads/master"},"02b2fe8565bf11f3031dc14a56b7a95d169f14fe":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2025-03-17 14:22:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/62"}}},"commit":{"parents":[{"commit":"d97da4438126ee49c6e295af24e28ed2e421132b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-17 11:36:35.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02b2fe8565bf11f3031dc14a56b7a95d169f14fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02b2fe8565bf11f3031dc14a56b7a95d169f14fe"}]},"branch":"refs/heads/master"},"d2a016a234f3602866fc7027c4c0e3ede7f26fbd":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2025-03-27 02:17:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/63"}}},"commit":{"parents":[{"commit":"40bb3e852cc9621b986a014f7411568b9ba3832d","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bb3e852cc9621b986a014f7411568b9ba3832d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-03-27 02:16:48.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a016a234f3602866fc7027c4c0e3ede7f26fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2a016a234f3602866fc7027c4c0e3ede7f26fbd"}]},"branch":"refs/heads/master"},"2ae3a6f5753b87e80e5fdd4a49b804f3adfedde4":{"kind":"REWORK","_number":64,"created":"2025-04-02 12:11:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/64"}}},"commit":{"parents":[{"commit":"40bb3e852cc9621b986a014f7411568b9ba3832d","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bb3e852cc9621b986a014f7411568b9ba3832d"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-02 11:00:28.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae3a6f5753b87e80e5fdd4a49b804f3adfedde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ae3a6f5753b87e80e5fdd4a49b804f3adfedde4"}]},"branch":"refs/heads/master"},"83584e5623adbd8b5da6c11012dd8d553818cc1a":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2025-04-02 13:32:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/65"}}},"commit":{"parents":[{"commit":"ca58eb934b282d6273ce50932dfec70d3b5a5c3b","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca58eb934b282d6273ce50932dfec70d3b5a5c3b"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-02 13:32:05.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83584e5623adbd8b5da6c11012dd8d553818cc1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83584e5623adbd8b5da6c11012dd8d553818cc1a"}]},"branch":"refs/heads/master"},"4c456fbff5fd9bdf955ede00b023d446999f11bb":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2025-04-02 20:12:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/55/836755/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/66"}}},"commit":{"parents":[{"commit":"12fded3227b3392ba614e0681afd4dc8eb02d2c8","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fded3227b3392ba614e0681afd4dc8eb02d2c8"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-02 20:04:27.000000000","tz":-300},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c456fbff5fd9bdf955ede00b023d446999f11bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c456fbff5fd9bdf955ede00b023d446999f11bb"}]},"branch":"refs/heads/master"},"05e4c09aea08b4aad35a5921711b98693e6c0771":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2025-04-03 18:46:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/55/836755/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/67"}}},"commit":{"parents":[{"commit":"832a31fab8e970901b805f1bcfc140b7216a8b0f","subject":"Add labeled metrics to proxy-logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832a31fab8e970901b805f1bcfc140b7216a8b0f"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 18:46:03.000000000","tz":0},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e4c09aea08b4aad35a5921711b98693e6c0771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05e4c09aea08b4aad35a5921711b98693e6c0771"}]},"branch":"refs/heads/master","description":"Rebase"},"335aec7417e7de560951f72c3eb26dd63c5b1144":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2025-04-04 15:06:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/68"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 14:55:56.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/335aec7417e7de560951f72c3eb26dd63c5b1144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/335aec7417e7de560951f72c3eb26dd63c5b1144"}]},"branch":"refs/heads/master"},"4f185d39acc98448ed82fc7d9616c9aca9d023e1":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2025-04-04 15:50:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/69"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 15:49:26.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f185d39acc98448ed82fc7d9616c9aca9d023e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f185d39acc98448ed82fc7d9616c9aca9d023e1"}]},"branch":"refs/heads/master"},"d9e7d0be01a8490f72ea4625465d802dfba35571":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2025-04-09 17:18:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/70"}}},"commit":{"parents":[{"commit":"a4c151b93c7491c026cdc6d95739c1fb8a8789ff","subject":"Merge \"tests: Display extra test summary info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4c151b93c7491c026cdc6d95739c1fb8a8789ff"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 16:27:13.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e7d0be01a8490f72ea4625465d802dfba35571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e7d0be01a8490f72ea4625465d802dfba35571"}]},"branch":"refs/heads/master"},"f5a50393ed353f9904353cde237fb7e7997cdf81":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2025-04-10 15:40:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/71"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-10 15:38:55.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5a50393ed353f9904353cde237fb7e7997cdf81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5a50393ed353f9904353cde237fb7e7997cdf81"}]},"branch":"refs/heads/master"},"279331fdf8b22c85f0f5e5749f0fbed827a583cb":{"kind":"REWORK","_number":72,"created":"2025-04-12 00:10:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/72"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-11 23:23:36.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/279331fdf8b22c85f0f5e5749f0fbed827a583cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/279331fdf8b22c85f0f5e5749f0fbed827a583cb"}]},"branch":"refs/heads/master"},"162e66a209cfa653d4da93205a36327816c96101":{"kind":"REWORK","_number":73,"created":"2025-04-14 00:05:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/73"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-14 00:04:44.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162e66a209cfa653d4da93205a36327816c96101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/162e66a209cfa653d4da93205a36327816c96101"}]},"branch":"refs/heads/master"},"2c67c74614b01a8e73f0c40fa05af977d0aebe61":{"kind":"REWORK","_number":74,"created":"2025-04-17 13:33:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/836755/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/74"}}},"commit":{"parents":[{"commit":"fd9ceecc506f88cd940edd7bb369574c044c5038","subject":"Merge \"s3request: refactor to introduce SigChecker classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd9ceecc506f88cd940edd7bb369574c044c5038"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-17 13:32:00.000000000","tz":60},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c67c74614b01a8e73f0c40fa05af977d0aebe61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c67c74614b01a8e73f0c40fa05af977d0aebe61"}]},"branch":"refs/heads/master"},"30afd0b7a01169f4aa605a1a982bf6e0e8387406":{"kind":"REWORK","_number":75,"created":"2025-04-21 22:24:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/75"}}},"commit":{"parents":[{"commit":"fd9ceecc506f88cd940edd7bb369574c044c5038","subject":"Merge \"s3request: refactor to introduce SigChecker classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd9ceecc506f88cd940edd7bb369574c044c5038"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-21 22:17:54.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30afd0b7a01169f4aa605a1a982bf6e0e8387406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30afd0b7a01169f4aa605a1a982bf6e0e8387406"}]},"branch":"refs/heads/master"},"a2312eb9e9aa5372843cb1c6587ebce77fc94cf9":{"kind":"REWORK","_number":76,"created":"2025-04-22 19:31:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/76"}}},"commit":{"parents":[{"commit":"1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06","subject":"Let WSGI know the length","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-22 19:30:51.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2312eb9e9aa5372843cb1c6587ebce77fc94cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2312eb9e9aa5372843cb1c6587ebce77fc94cf9"}]},"branch":"refs/heads/master"},"b236cd5fe95e49a52e5b169994f1c7898e2475af":{"kind":"REWORK","_number":77,"created":"2025-04-28 19:52:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/77"}}},"commit":{"parents":[{"commit":"1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06","subject":"Let WSGI know the length","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ca073ce1dac2d0e3554399fc8e8a9e0af5e4c06"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-28 19:15:27.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b236cd5fe95e49a52e5b169994f1c7898e2475af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b236cd5fe95e49a52e5b169994f1c7898e2475af"}]},"branch":"refs/heads/master"},"a3ad1ed45d472f3fe55a512e2a3949ed78576ff6":{"kind":"REWORK","_number":78,"created":"2025-05-12 18:43:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/78"}}},"commit":{"parents":[{"commit":"2c5bcfdf0ad017195ba0385bc67b3199bb601205","subject":"Merge \"Implement heartbeat response for COPY request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c5bcfdf0ad017195ba0385bc67b3199bb601205"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-12 18:39:07.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ad1ed45d472f3fe55a512e2a3949ed78576ff6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3ad1ed45d472f3fe55a512e2a3949ed78576ff6"}]},"branch":"refs/heads/master"},"865f12fc7280f147db786c57ee9517ccd7287c8a":{"kind":"REWORK","_number":79,"created":"2025-05-12 20:42:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/79"}}},"commit":{"parents":[{"commit":"2c5bcfdf0ad017195ba0385bc67b3199bb601205","subject":"Merge \"Implement heartbeat response for COPY request\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c5bcfdf0ad017195ba0385bc67b3199bb601205"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-12 20:42:20.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/865f12fc7280f147db786c57ee9517ccd7287c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/865f12fc7280f147db786c57ee9517ccd7287c8a"}]},"branch":"refs/heads/master"},"f9ac22971f6b08878b8b038855cd0ee7d93f7fd5":{"kind":"TRIVIAL_REBASE","_number":80,"created":"2025-05-13 19:37:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/836755/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/836755/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/836755/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/836755/80"}}},"commit":{"parents":[{"commit":"6212869399d943305c8b1f2e0a20d9cd2cd71b75","subject":"Merge \"proxy_logging: remove unnecessary condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6212869399d943305c8b1f2e0a20d9cd2cd71b75"}]}],"author":{"name":"Thibault Person","email":"thibault.person@ovhcloud.com","date":"2025-02-07 18:30:34.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 19:37:35.000000000","tz":-420},"subject":"Add support of Sigv4-streaming","message":"Add support of Sigv4-streaming\n\nThis update implements Sigv4-streaming (chunked upload) as described in\nthe Amazon S3 documentation:\nhttps://docs.aws.amazon.com/AmazonS3/latest/API/sigv4-streaming.html\n\nCloses-Bug: #1810026\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: ashnair \u003cashnair@nvidia.com\u003e\nChange-Id: I7be1ce9eb5dba7b17bdf3e53b0d05d25ac0a05b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9ac22971f6b08878b8b038855cd0ee7d93f7fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9ac22971f6b08878b8b038855cd0ee7d93f7fd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
