)]}'
{"id":"openstack%2Fswift~837043","triplet_id":"openstack%2Fswift~master~Icf7542fb3e9cc2e40e902a51fe9f1eaefe7fc570","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Icf7542fb3e9cc2e40e902a51fe9f1eaefe7fc570","subject":"diskfile: Reduce tracebacks when consolidating hashes","status":"NEW","created":"2022-04-07 23:33:20.000000000","updated":"2022-04-08 02:37:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec5afa91cbf81edc1b6a64c1914fff6205d78ecf","_number":837043,"virtual_id_number":837043,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-08 01:21:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 01:21:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae072b3f649df556b3b3ab59de199fb909740248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-07 23:33:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd892dd8c9b387a6459d8d4f1054460cb724723f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 01:21:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf54b7cb6444c11b67ed02106742314 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/c2f2e47b7764477da7b6328e1a32bf41 : SUCCESS in 1h 11m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d277ef8234d547a4bc5299296d81158f : SUCCESS in 57m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd58266e3a6c48c4952284931b7bc0a3 : SUCCESS in 59m 44s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c887266519e74dfd8aaf9bba0cc51f2e : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3f3421a38ac9472097ab7c363411b4b6 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b3b25367ab464d3d8ae9891bb9c85537 : SUCCESS in 23m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02569c51ee1f44eda451ec19554c55f9 : SUCCESS in 7m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ebb3324be31d40ec8a694b3c5caeeb7e : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/381b6f84f82e4b7e87b3794f7db0f18a : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db59087eaa214b3eb5f00df8a5a1780e : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65432ac92a77433c9fec87b54f2393fb : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce3508c9d07743c3a4152dbd9679b0d4 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ca43c74dcda4a49b596157e94ef2070 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b54a8b8301854945969d0dfe94dff3f3 : SUCCESS in 18m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/450a755cf7d8465198196ab48dfb4726 : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8fa0cc13cbd54b41976fb49ecffffa64 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/12aa9e24b1fc477592e752c36f6dd845 : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e1d5ea69f1b44ec86562023ba0a2373 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd7bd4aaba7d4bceacafa129b71d0797 : SUCCESS in 19m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/09f777a4d9ef489e9fe7988cacba1088 : SUCCESS in 54m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/76687c1bf7894841bc957022368c9977 : SUCCESS in 1h 46m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55e5b0a3d7b74da0be7b9a77b93ec1f0 : SUCCESS in 43m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f07afe8f87e248b588ac2cc36cccd25d : SUCCESS in 50m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a72f6e1435e04d458e90a55708169f70 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05272a85fa394e75bf516734e1994cdf : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c37b13a0cb5945939d84cc44fcef332f : SUCCESS in 20m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6149cccaadc5466b98ec752273be36fa : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ec5afa91cbf81edc1b6a64c1914fff6205d78ecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 02:37:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07c65b363d624e2c8b5283e502445a32 : SUCCESS in 47m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc016b7713fe4da8b8cb3721aaaa2306 : SUCCESS in 56m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/919934b2cb58471580828ab1adbb0e76 : TIMED_OUT in 3h 01m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e1a1d9ea4064eaab1281a04c05da437 : SUCCESS in 1h 15m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6612f4ff93e74e4a9c63dbd251e61dd3 : SUCCESS in 1h 14m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6f74ed97bded4d681a8d09a71c0b7bec098f440","revisions":{"b6f74ed97bded4d681a8d09a71c0b7bec098f440":{"kind":"REWORK","_number":1,"created":"2022-04-07 23:33:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/837043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/837043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/837043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/837043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/837043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/837043/1"}}},"commit":{"parents":[{"commit":"3d7ff12064709fd77eb95cd2e2c3276707d5b5d6","subject":"Merge \"Obj Auditor: Quarantine ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7ff12064709fd77eb95cd2e2c3276707d5b5d6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 23:25:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 23:33:12.000000000","tz":-420},"subject":"diskfile: Reduce tracebacks when consolidating hashes","message":"diskfile: Reduce tracebacks when consolidating hashes\n\nIf we hit an OSError while working with hashes.pkl/hashes.invalid,\nwe\u0027re committed to doing a full rehash anyway; we don\u0027t need to see\na traceback about it. The error itself should have enough context to see\nwhether we were having trouble with the pickle or invalidations file.\n\nChange-Id: Icf7542fb3e9cc2e40e902a51fe9f1eaefe7fc570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6f74ed97bded4d681a8d09a71c0b7bec098f440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6f74ed97bded4d681a8d09a71c0b7bec098f440"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
