)]}'
{"id":"openstack%2Fswift~837633","triplet_id":"openstack%2Fswift~master~I201734ae0d6232ecb1923e67864dd928f90b6586","project":"openstack/swift","branch":"master","topic":"safe-reload","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-06-09 07:17:52.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-11-03 16:36:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I201734ae0d6232ecb1923e67864dd928f90b6586","subject":"systemd: Send STOPPING/RELOADING notifications","status":"MERGED","created":"2022-04-12 22:15:03.000000000","updated":"2023-11-03 19:10:03.000000000","submitted":"2023-11-03 16:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":14,"has_review_started":true,"submission_id":"837633-safe-reload","meta_rev_id":"d4293d4e2b261570bd73a8f96d6d1bfbb6acee18","_number":837633,"virtual_id_number":837633,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-03 16:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-11-03 15:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-11-03 15:25:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 23:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 05:41:05.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2022-08-22 19:26:58.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 15:25:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"5dfd16db2cc38131a4785ba22e7231e1b899c60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-12 22:15:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a739d063fba7061069ccc030455d073dd9f283e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 23:18:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da57e01cdbd742718bd954120b59be7f : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/720fbb453ff1480a88ef7697e3535765 : FAILURE in 10m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9fc32d7538274c2bb1f6c33541e578df : FAILURE in 18m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcd64f61d1894bdb91f74a50922bf1e3 : POST_FAILURE in 18m 46s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f2c2214ec86343288f4b55bb4df4f668 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2ee0ac168bf547e99f7352bbc6f9685e : SUCCESS in 23m 04s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/33306a0b6e0b4f3595ba8a0986e96720 : SUCCESS in 25m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/35df69bbfbd540a19193491288c7d6e7 : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e2223a20ca1c4baaa3e3946c994c9f92 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6805ad1d63f4b37b5349fb828129239 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d184384168694ed8a4a893f132ba3473 : FAILURE in 17m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/65c9814d3b3b4fdbb724a3d0585f28ac : FAILURE in 18m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb9099cf79104d33962c633fd28efbfc : SUCCESS in 22m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5748d483cb8b493db3cdf624f3575290 : SUCCESS in 24m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/724b2528464e4994a64c3a7b94c15a40 : SUCCESS in 27m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8f8d9706433b4a518b4224b5f9fa7fd9 : SUCCESS in 23m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5cce4d22e61144e094524a44d7184bf2 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fda6c53b67d74331a1479e9578be3d96 : SUCCESS in 21m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c4144aaeeac412bb539663448b7cf77 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a33d839f1e1c4a1393bc2a3d5e77d975 : SUCCESS in 20m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8d5e29cc3d149869582bf15318db51d : SUCCESS in 1h 02m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f46be3c8739b4f9b9c43bf3a5ccd779a : SUCCESS in 54m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac216fe058634a30b0fab1c5f2cec65f : RETRY_LIMIT in 15m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44471902eaa742ccb5023c260488ea24 : RETRY_LIMIT in 7m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/add794aef519442293925e7bc9001a55 : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb38400f3f242778d1105c07237118a : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fce7338233ea406d837c35c7bec809cb : SUCCESS in 25m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/915369157dfb4879be25d2e5f7c5f9f1 : FAILURE in 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"32b351dc0ebbd39b8b2901e4fafa543af04a51de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 00:49:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c948cb0c53c2481bb4b9b61653a6f031 : FAILURE in 39m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e6ac55f4e4344eeb3a90972b9f07a27 : FAILURE in 32m 05s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7000123317b94657ab4e9d4796029f03 : FAILURE in 2h 32m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb2fbbaf0cce4692bc7fcb7e481bba57 : SUCCESS in 44m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cc071413f3e49598a665293b5ed9f11 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c000858a2734f63198b3e5c7ad6a3a981b4f825f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-13 01:00:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e4d56ccedda344c6588207fd0c2bfe9771840e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 02:06:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b868df676a1a4a389daea632abce0eb3 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/38e3423d04e1495ead6da286122b6977 : FAILURE in 11m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b73e9a5615d4590a080647d0a2626ec : FAILURE in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e090cee4b1e49fc986eac06d7a3a8d3 : POST_FAILURE in 11m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2f78be21f5e6414cb8160b121a759277 : SUCCESS in 32m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c3d5c578890b4bb4bbf344f58e95fbc0 : SUCCESS in 29m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/09af747dfec74ebd843df4a8e30f2c30 : SUCCESS in 33m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6941883ddc54972807233240ceab171 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/66571cc4e1e14b39ac0817d6f809f5ce : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68ffb7a12536413e9ef8f06d1970598d : SUCCESS in 24m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/220450acbd5a47febda6f558e90fc290 : SUCCESS in 29m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8674c036bd6d49b0bc3dc9486b7b398d : SUCCESS in 29m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14796a810eed4a82a9250a47efb93cf0 : SUCCESS in 27m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/218c1471c1c5479db53bbf2c6a7f1952 : SUCCESS in 32m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84c5d0f2e8f6459789a366a2c4550a94 : SUCCESS in 32m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/743ac7400ebb48d19e8938000676c2c6 : SUCCESS in 26m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/43defda6150a47b59dd27876fea76064 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f77ea056f9bc4a4584f8d8524fbd48c2 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31d8382c5d6b4a5cbb159cc70f748b5a : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69ec2f07248d4c32b7e849bd54c6d3ad : SUCCESS in 23m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ff8369cdba464437803b336193d39c3b : SUCCESS in 38m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5eb41b816224165bfb76323b819c6c3 : SUCCESS in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f381780d026644c9aa85719f7d3462dc : RETRY_LIMIT in 15m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c054db6934e84582967b4e27f7ceba05 : RETRY_LIMIT in 7m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ccaf67891e84cb28e65690f6da80cf2 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fea0b9a4f8e47faa564b005d2aae28d : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36d90483053f49d28a329251fb695a12 : SUCCESS in 15m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73bf90c5c27c41c0a8d14a617ad8c1f6 : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c01d947a08113d8b44f42eea34306b3419e39130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 04:04:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/318ed6524a2a4e6fa32af8a359d89129 : SUCCESS in 1h 09m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07105d09518b48b280d189a75aa8024d : SUCCESS in 49m 14s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/bd44adc87b24402fae1d93623e195dd5 : TIMED_OUT in 3h 01m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f144b7a7f3a1464ba494a1eaa4051b7b : SUCCESS in 57m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b731e312a1846eeaf340200a58d6ad9 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ab49c0a98ca4551e9ff30b7f6970ed059b9c9bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-13 20:43:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"934705c82f5a235dcc4b5faec232cc71d10a45fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 21:42:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c6266f16b8341b7af97d5d4965e12db : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a90f263f15704433ad9a87427863f78d : FAILURE in 51m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52a76a77a7b845cfad191823c32ebd6a : FAILURE in 18m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b99e6a9df5847a2a67fcc89adac66d9 : POST_FAILURE in 11m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bc1c8585ecfa4a3d91dc801106f0c219 : SUCCESS in 23m 40s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/41554a2e04f449caa7d8ec77b803751a : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e44c086a5ca54dd59ab58eeee53a30e5 : SUCCESS in 24m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2f62664c56c4a9693da30bb85d23e94 : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5ec058e5ac18468880f20838dd01e98f : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0cf294ae1ea84564820d8c5764d38c68 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa07fe839d844173a811b94d020c5923 : SUCCESS in 23m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce058cf2cffb48e98be7bfb8dd664356 : SUCCESS in 18m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38ce83420d0343cd808aa49feaf75073 : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07a56f0c957c489087dfb19720de57f5 : SUCCESS in 22m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f69e259366b4473e9ed9c574536b7210 : SUCCESS in 15m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9181274cb224154b3a305eb9e4cd80b : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7bca1fb50b964d169acdc807642c50b6 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/46805738d757468189e0d3b110f713fb : SUCCESS in 20m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5193378fda74594b8b8289d9cf432e8 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/adc67c5a7bbb44b080283c0e4fcb0858 : SUCCESS in 15m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c87db41dce64ecfb07e656920fad974 : SUCCESS in 51m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cadff5d13b654980a742fc432b2ab810 : SUCCESS in 52m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e01795ec485449cfb31c0fe89d627b3f : RETRY_LIMIT in 15m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d2f5d3c5b0841098999fec2ea34dfe0 : RETRY_LIMIT in 7m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beecffd15f434d289a0ac0b3cfc1eafe : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8537f241289f4148ab48026d20ca942b : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5fa4e429cae04f74a7b9fb5c7e576ee8 : SUCCESS in 20m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f9da07a2fe34d35887ccc4b9e0e1a07 : FAILURE in 11m 52s","accounts_in_message":[],"_revision_number":3},{"id":"619aafd9b6a026cbfe09b07e5ab09c60d89532a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:33:50.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/954db7a954ef409b9f86fbad9cf7f837 : SUCCESS in 45m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0eccce85ecf40d3803bd01b1e25e40a : SUCCESS in 46m 41s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/22cf9f18357d448881e15317fac7b7b5 : FAILURE in 2h 58m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02f76ffaf1bb4f0abaee64fd0b7e9b2c : SUCCESS in 53m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/02cf082025df4adea8803923ca7a0f3d : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6e28bc62662f0d6e4f2b42877acc914f957aee4f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-14 04:23:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d125ff0afd0631aa0d595411ec4c982356ceab4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:26:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd51457eb9645808392a6a7ff466370 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/ab222783d4174317bb660aafc7ceddc5 : SUCCESS in 54m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5551f19e9e2b4b7ea4598c9fd18e9ec3 : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae0c25bb5a2644aca0f184f2fc144ba6 : SUCCESS in 41m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e62f4bae4eb64e958227d261fe8d823c : SUCCESS in 25m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2229e89f1552466d82d49b7bf5ac3f9b : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4257c5101d174a1481777cd132fb8b13 : SUCCESS in 22m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84f2401b7ce34e7e86588c0a81b63f68 : SUCCESS in 7m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a25e210451044d891f48cac42b003d8 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e613e3b3414a4083a627921fa94c85da : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89e993188814413fb1f0ab3af38a229e : SUCCESS in 18m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/755e585c95d544aaa3e01efbaf5e1b58 : SUCCESS in 15m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d48ccbf47f5944b28abd75650f8e129a : SUCCESS in 25m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2bed10366b39427fb5d4a21d17e7e927 : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76a601887a6f41d6a3d80e1b0a8fa721 : SUCCESS in 24m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a84b7238cb54dbf93943970547d4cf0 : SUCCESS in 19m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6cfd9e41e58245229d6014afa0e067e9 : SUCCESS in 25m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab598c4f49e74db586df8df08824c0f8 : SUCCESS in 20m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50a8398eaa434f8fafbee2f18b28b446 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee5af1938ed54754b6caa009018df322 : SUCCESS in 20m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/507bd2fed166483f9cc098fedd702154 : SUCCESS in 56m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0affcd60810f40b98957ca72d9ea88ff : SUCCESS in 57m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f450049fb3b403bbc4e905253533cb5 : SUCCESS in 29m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/329c6a33d8ea4395b67d9dae7ee7f5c3 : SUCCESS in 41m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5efaee8496b4ddbb82fe7aa4dd331f0 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af82f866b4242b49f5dc0c02b629413 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62f5005475c949fba1aafe12fb696742 : SUCCESS in 20m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd6a298e7b0246929de41852eef715da : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":3},{"id":"d92ad71676da9e0dd57982e4103791a55b9f1d57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 07:16:43.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/67b871c0a5bd4a1ba3df01b08a6af8dc : SUCCESS in 34m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f537d7954734bd29f90dcbd2efae238 : SUCCESS in 34m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d7927423049f414dbc3885edb203e93c : FAILURE in 2h 48m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ee13cd7738a4bd79724c5de9192233b : SUCCESS in 48m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a79c4b74fd6648d8b984b6dca8612bc6 : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":3},{"id":"2958cb1c6274789b19eb01675ea5dcf209441f8d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-12 18:02:13.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dc1fcd2d82dd30d0f9a3ff9a641f7dfc249750bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-13 03:57:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e1c5fb048fcabf2f31a6b7cc303b3d5f3b6217e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 05:20:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df93a1469be94e68b1932284ce6c6e90 : SUCCESS in 53m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c924161e394f4bbbae292e1ed11a8d4f : SUCCESS in 59m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/77f3a8bbc6474974bda4251ab175a78e : SUCCESS in 51m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73534919767f48c59f9f021066f064ca : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":4},{"id":"658dceb5d28035e7a0a405022085655d730fbbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 05:30:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889fcd21a71e494685da34472b9c8b8b : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/8459ddfdd1f9485cb2a6ea7cb69120b9 : SUCCESS in 1h 02m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73b11678a39f4f938982b351bb45924c : SUCCESS in 59m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73450fb94009482fb4ecfa86159b69ea : SUCCESS in 56m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/780fc61be67a4e19856c54f3097c30e1 : SUCCESS in 26m 50s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6a3b26be531a4ec48885445084a815da : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/11654321025347cea0f6552472a7cc7e : SUCCESS in 24m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54a2822f0ab54f06aecdb0d79ed8a577 : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ac749ce46c62454abc52c4dd2b3ca9af : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83b3599e25554f88ab9cdbfdc775d115 : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f1610362aa4f99b444489cfde98e37 : TIMED_OUT in 1h 31m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e19662cc474ad0a0feb7c4391fd9c6 : SUCCESS in 15m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5db4e17a84d7491e8a5e1c535985a51b : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3051786128384765a803981f6b253a46 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c65ed2720f2740c896a357529314dcb3 : SUCCESS in 15m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a0fc2282612a44f5884fb868399289e6 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd7fa2f8a9d74f5a9561fcdba15ec95c : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bd990c22c1884b2b97c058891f5c257e : SUCCESS in 18m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2473cae3d7454440814e4c2aeec81939 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fbfd9bb75414c7dbfa85f4c75c55eb6 : FAILURE in 17m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e4cd9ec8bdad4e39962ff1b39d1c9b22 : SUCCESS in 53m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc09b844cb0e4b9ab33cacde1fdbc828 : FAILURE in 57m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f15ff8483b51476589bfa91088c4de66 : SUCCESS in 39m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ddbba7f643d647c0a04ebdc40c41a8c2 : SUCCESS in 45m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1485e6dd26b489d91f88246eb4503a3 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69d7ebff6ed4d3fbe3b02452bc427bf : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11282db5efb54b5b8e9afee91675339a : SUCCESS in 20m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c7a108611d44668adfdeabff60649af : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":4},{"id":"a3e817bd19b1992462db2c085d28ff0481f76e43","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-18 22:09:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"53438a9733465e7d5ec698dc3516c26f5b503686","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 23:06:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3202737dbad04f11886ed3bdad81d72d : SUCCESS in 53m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/725dbfe6847a47f8a58627c21af499f2 : SUCCESS in 55m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc07d4de9ffe46dc95de65e762947fc7 : SUCCESS in 53m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/de13df5d18ea42d5bc14829ad26e9586 : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":4},{"id":"4757e11a0158cbdb72d70afa41524172434c2597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 23:18:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d434b20632422a89e06a719b5957c3 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/5eeb23c3fbb54b0fa08b9f44e83209c8 : SUCCESS in 1h 01m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d1e06c7585e42dc88bb2b7853c6c400 : SUCCESS in 51m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa90232f1ae84092803761ce02474509 : SUCCESS in 56m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6692da2d427745bc8a64a363a7f331ab : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1aec1953532046619aecf158e3e9b036 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f5dbcae7e3f947dc9a12f5c6a37014f9 : SUCCESS in 31m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/140200e6c3584c3da6f83356c80cef91 : SUCCESS in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d7e1871aaaa458bab71c14d272c97e0 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc8b5cfb2dc24cdfb45eca2aacc47543 : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08bfa7e28a3d4f0a8aac796391d1aebd : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c8c05a38567434bb404413e5b384233 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fec7829b1ee5492eb1d477bb89f7e7a8 : SUCCESS in 25m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bacd629938274e119329ecaeda5aa298 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25bc7ee9714748338e55d695af8de371 : SUCCESS in 25m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65fe2b279d134f8e894dd1ca5c145156 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da5fcef8128d457a9cc32d1296063877 : SUCCESS in 14m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dcdfa20f0b964c09b600fa521726b083 : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bad01e7a28704c6b967c9b9cb820afde : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6e3193ff82f442eab052c613e364933 : FAILURE in 14m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c963ad23c3da4fddbb603a43acc35310 : SUCCESS in 55m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4451808d3fc34eb1994288eb711204a4 : SUCCESS in 1h 02m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e88a4c9e7d0f4e6bbcb43433115ecf69 : SUCCESS in 40m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1a37379d3c6f4993b800c8bc54f359f8 : SUCCESS in 40m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ed4f16edd5a4d0c9d498b78830fa814 : SUCCESS in 23m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d35f7133c24356b08fcc41fccbcaeb : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e18e58bea2d54edc8d27eff56f37d2fe : SUCCESS in 25m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c98d94fca64c47789d01677ac442ace7 : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":4},{"id":"136ea8721f96b63d6c7a6c146379aef84af111a9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-06 05:41:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"daad7b620325bc9db3c0adb52bebc75783e08b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 19:26:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76ce9586a98f773b5ce84e03355885899d37db28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:26:51.000000000","message":"Patch Set 5: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 837633,5","accounts_in_message":[],"_revision_number":5},{"id":"69c660406cc2c6179628582d8dae0b6c1045c7fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:26:52.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 837633,5","accounts_in_message":[],"_revision_number":5},{"id":"2ebbcc1b5be61e32994e3ff0907de3959a6803d2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-08-22 19:26:58.000000000","message":"Patch Set 5:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":5},{"id":"24bf17b1c3dc1f176cf88061446b739e6d980b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 20:15:53.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"6d8c123626b3e67deeba5965e9533856aca482dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 21:28:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7dd78bf5fc4505b5fab0188beb1e14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbe4c48c696486ba438b2c13c9f1565 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7fe203f31526488491d74995645c6131 : SUCCESS in 51m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec545d2fc87146a69532f17291866638 : SUCCESS in 49m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43b1f363d57d4a0a9c181a25f0c1e37a : SUCCESS in 53m 14s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f80f3d172e9e4c579d2c89cdcbb4aaf0 : SUCCESS in 28m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4669e354bbac4c4ca45172b363bdb274 : SUCCESS in 30m 01s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/496faa299ef24f80afe991781e68b402 : SUCCESS in 28m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e33e88a78f7c4a70a00f83ca398aa7a9 : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/444d05831a4d48e496a92afe9711f883 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/450f4ab0a0d542538cef9e65642e310e : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02f096ff660f4aa289274194618f90dc : SUCCESS in 20m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab25352d954343ad888effe13bbc2faa : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/54bfa39b86c24deaaeb33f22fb2b1a43 : SUCCESS in 22m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f29d2e064efd4c3fb7ae473e1ed19249 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b6c3fbf78fe49e2bb798992ba2d8342 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3efd715854f34fbbb97ef89a6299d148 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d20b7668bb2546cbb7fc1b924f996dfd : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7dcac9c6e8a549d5a258b3ce65fa2f74 : SUCCESS in 22m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74e922afb0f0454a82d2578204294b51 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46b89e9bb4b84ea7b35ec790f86aeab1 : SUCCESS in 21m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f1fefcedd8d407bbd4652b0eaf523cd : SUCCESS in 9m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3bde8a5933984811ba9e0ddaf722d737 : SUCCESS in 1h 10m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/29691917b58c4a3e98b18ba71b67221a : SUCCESS in 1h 02m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34c9f8bf69d340bd8876adeb6a9d253a : SUCCESS in 53m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d3905640895411d9515d93d3c63eebe : SUCCESS in 47m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3452044d69eb4818be5500a89b420056 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235a328dc9114653915bd19b6cd31297 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7da44a0b6b145d5b5d5c26ba3cfdda1 : SUCCESS in 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da253867b25e4687b7ad9ac3fedb7bbe : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":6},{"id":"d05741c9aa9f72c959f15ab78e1f6e71fdc8c8ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 21:45:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d7739906c64eb1b6215a1f1483b1cd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4071d2adf3c14fbba489294824aaa75e : SUCCESS in 33m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5254f4d19504c07a37af61e544672e6 : SUCCESS in 55m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfea7a92fbe3458d970f57ea7fe31ce0 : SUCCESS in 33m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c882638084ff4c6d8e14e43d35f099f3 : SUCCESS in 1h 27m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ac35afafa19d115e77ee744c99ca144d8b7358bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-08 21:16:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"800bf0a6b1830b597442de0cfac6922349cdbd69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 21:56:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a1f08b72424b3691a661affe840014\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0986aaeb87624496a1946a3bf5d8baa6 : SUCCESS in 31m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/036473180c3d47e0b04ceb857e87c7ce : SUCCESS in 31m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/96c48cea6417442d9eee38d6573ab787 : SUCCESS in 31m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8085a0b9e41a4d7bb80ea44bcd263794 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":7},{"id":"0bf2cf62e8da2e2e345da3227caba1365fb165cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 22:28:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d88858cc277f43098fd5c9d85198ae50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a222d14484424f0a998455597fc21845 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/6c121d83e6bd4e2ba3a46db8fa786db2 : FAILURE in 32m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94eaa80ff46b4c1eb34d3e2261c0772b : SUCCESS in 48m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/799e1ed1385040739b0b3ae2c9f095b3 : SUCCESS in 39m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/967f780be030411eafacf5529cd1c27b : FAILURE in 26m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07ac88c3a1b04ea282702af6e7846402 : FAILURE in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98d414d6feeb484e9376255b2dc457f3 : FAILURE in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a629725bfeec4959a6fee66d6d8e1a30 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46435602bf064766a218ba65cd922142 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34e934595225409da8081586abe28924 : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d317e7c319dc40ebad86ce6281e6f289 : SUCCESS in 15m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3f0191001114244b11b6a26b023873b : SUCCESS in 16m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e58e8fe5ecb475a935a659d990956d8 : SUCCESS in 20m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d473e138a4744f29fed4b3025f90531 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/748a6e32b9044f259056a428f493cba5 : SUCCESS in 20m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ff3cc558c56471a8e9465a5bfd81aa1 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3795b2cd65144db79a48498c1f73af1d : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83e814cf20e44a06856fc2ea70857027 : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2a5034f899064ed6b00883e49f845c20 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3977c553e05e4f1990c9d19194e5a72d : SUCCESS in 16m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/90d381ce2e80420393bdfb83aff4180f : SUCCESS in 8m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a9b0be744fd4843973a4c9b359cd6aa : SUCCESS in 1h 08m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/caef52f7784f44069c635920ed8ad574 : SUCCESS in 50m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86196b8a1c7747bdb98f31d9d8816451 : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2405eab7a68148d8821076dce5fe94c6 : SUCCESS in 32m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c258ac1b1664096a160947cb8163b16 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6425c8cced438780d96cccad9f85c3 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/480a7353bac64eed8c9249404c3fe624 : FAILURE in 20m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e01ea4d1bc694fbbbd5fddce191e3089 : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":7},{"id":"8f8eb904754b9c53a51d6dbc2a2ef30e0665f62c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-16 23:01:12.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1b203aeb5e6ba783ed1a8f5b59e4de84b809b12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-16 23:05:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a2f78e29128caee00b41ad53c8bb07317ff2aa16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:25:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8741bd7997149c2a5926423c5542e76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f9c75ba3a941beb10fafae782c2936 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/5bd8a93c9dcb40d39524c5a2426f0a75 : SUCCESS in 53m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b824794b36ad46bbb8e97dccb1d7bfaa : SUCCESS in 37m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3e52e3b0f3b44f2ab72faf5d8a27667 : SUCCESS in 51m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f43bd23f2d004674b3646e18ad851879 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a9b4387cd46461d9e41470d3ecce3c0 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a36998d00154d449b9eaa0f2e6ac4e6 : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65c2a493b32c49b69b7e42b3d62e99e0 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6952c29ae47d46769950ce1490a0777c : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba6ffb92c8b74f36bee9746ac2cb579b : SUCCESS in 16m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e736329695a44816be52c4748e9b5127 : SUCCESS in 15m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9296691df7074ea39fa3fd82159dae7c : SUCCESS in 15m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6948a17f99d4467794b1a7d744c84b0c : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dff0ace67450477da3014f28414493f1 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/61667b0c3a614056b659a0af1819e72e : SUCCESS in 19m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8fb4ae0c27a24e1a97875756ff3e1114 : SUCCESS in 19m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec139270cd3c4f68abdce0b252e8e05f : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59faa2c86f664fd9a372a82ca864d92f : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3b9ab5f7df14ebdac3c56d3a9c93ee9 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65a075d474aa42da9f8af39e4f1634d2 : SUCCESS in 7m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44fba38e212647a983501fc45eab8299 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/14d60b37fff842c69a2d5e60c8eb5e34 : SUCCESS in 9m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/09c2c4470cbd4303a495359831d76660 : SUCCESS in 49m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/77a2562b03d14c17938f0f96e89b58fa : SUCCESS in 1h 02m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/793eec42d1714e848f297f8a0bd17daf : SUCCESS in 48m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/358aacebc710457bb9cea8f1728d860c : SUCCESS in 37m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a2f9749bc34f069b9d1d110fef4721 : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5f0ae2ffc54140a57f474d661d4686 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ecd661f67d2541a7a514ff783b6fe8f3 : SUCCESS in 20m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efe7d70a7a8344d9849b5aacbd785706 : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":8},{"id":"9bcbb0d541894af3c6f508e243fed5251512348b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 00:39:26.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/805289dd51434cfeba93548a4d7dfa73\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8145a1bfb7b46e6a14ba757f93e8f67 : SUCCESS in 18m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3cdb0d4acee4973b72bf2e85d2f59ce : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfa8027a77e242b1864a772ecc70fd0b : SUCCESS in 18m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73a26f9175354ca3b09cbf110bfaf50b : TIMED_OUT in 1h 31m 02s","accounts_in_message":[],"_revision_number":8},{"id":"1d220788a903502dd4c1b23a9794ab2ee7777d94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-01 17:07:27.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"6e8e98baffe6f966ff64668ca0a30581f763e955","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 18:32:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeb747f10358442c8fb4d109bc596af7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86e88cd859f74800a5b5ab9da186e86c : SUCCESS in 47m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb0518022e3d47b29c90705e5856f095 : SUCCESS in 1h 22m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2844ab8208c740eeabc5bd0c927d2ccf : SUCCESS in 53m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/48cc280c4df04acb9c959a33cdcce36c : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":9},{"id":"829ef6ffd078264f67d642a09b9ed4d4f775ab73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:17:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5268066d1db6456e8f29f51ee55f92cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2f07acf48b4bdc99711fd927ccf0d1 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/dfc70c4ed13442dd839232d684319eac : SUCCESS in 1h 00m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/503cf232e8f247aa8bd8e1b9dcec0b5c : SUCCESS in 38m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60af8cb9fcf04e1a874203ec0359f4d1 : SUCCESS in 59m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/612992c86ba14f8ebe232ceee01ced1f : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acc29e5714474d388d8bd6c7bcc88885 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15428f321d4f49b980e072b67940d307 : SUCCESS in 18m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59f4c7e75d704128a63f060a9ae8cbf1 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b47d6aa3c684ef496e8025e24a90b47 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b416facf0c14d8ca9722d59038a1152 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/108da8e13d924c16bb4b847556f83996 : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14b2c5e74d5f4486882b76f8b06bd834 : SUCCESS in 14m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9c3b7e06af8c4aeba43ce41b19bd5386 : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f5e6fa4964e44d45ad23f57c59d98416 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ecceba18a2b1417ca488ac92911e81f8 : SUCCESS in 36m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cb40dc3224df45ef88aacdc6b2667590 : SUCCESS in 21m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e537e40079044d4a462f3c1797a7b28 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d1ff43b78474ca4948b599488e925a2 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1bac79290deb473ca86ed5fe0bf3dbb8 : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cae4acaac5d84ae5962ff9feddab2815 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44f3ffffd77b47b1825743a027aa15da : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1611b8532bdf496f9b3171d5942c28c5 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/145bdad2e0274118aa751c3192438734 : SUCCESS in 52m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c9435798aa242dd9e6d7b30dbdd63c9 : SUCCESS in 56m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b0df4bdb11945b8adb465164421ba92 : SUCCESS in 1h 01m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea22cd582c834aca9e3d4e8278714057 : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7f63116bcf14b0fb7288bd982ab0586 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/addac71032364917a01d758a5f8f52ce : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39b6ce7bbb0441b8ac332b6b098aeefa : SUCCESS in 22m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51aa867845844b0c8045308d439dac45 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":9},{"id":"b531eaf144893a6ef3c90687937f48643993e12c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 03:53:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fd8baad5b43be5d5dcd7fffab87d97000e4ed478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:57:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38db578d5ec44d6d8b9f998b83d13bf5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b61fc9b464c44ef2b48e956e065b65d6 : SUCCESS in 17m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c1a7df71d3b48308ebecc0fcaaef20e : SUCCESS in 1h 01m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21a283a41f3045af96087d2bbc5cfd27 : SUCCESS in 22m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/419f7a5d8e544e3ab899f8ac82cd1d8f : SUCCESS in 49m 15s","accounts_in_message":[],"_revision_number":10},{"id":"d917acb988966824b5fd2cefbbd88b84e033c4af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:02:02.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d1c10a147b407799d7535b4102a907\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974c8242ebb14342ba27e0657b042c4f : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/00dd5462d76f4b1d8bc0081693870c10 : SUCCESS in 51m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47f5042d5adb40a1ad33c7b16cb8600b : SUCCESS in 46m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32e5b3ca01b8499ab50717acb0b58e49 : SUCCESS in 43m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/949839e972e942a2875d5752511e8243 : RETRY_LIMIT in 4m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/520676925b8246f9b4afbc883caff2ee : RETRY_LIMIT in 4m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f9cbdc9515d4cde82f2bcec8fd64d90 : RETRY_LIMIT in 5m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce46c34cd0b7470181ea9256d48ffa4a : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9fb1436ef0e41eeac514260ced5943e : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66383d8ca2a24c058bfb1c89d462c828 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae24eab54abf49e59637299c6b0b7a08 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e101accedb8445f290900b3529395e9c : SUCCESS in 16m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/226142a757a54c6e8fe5c047d8e75095 : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab46099df808401c81554cbccbd83a60 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c3be8982e78480f928d3be42d8cdaa6 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8957fc5dfde94b12a1a0c09c00070ead : SUCCESS in 19m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/440d8f91c1d74a8ba0c7934fd32f4fb4 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3fc654982b664a0e99356d1ee75d73c1 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/94ef304c810348c98d483c34960c8ac2 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f93875bd38e8475da5c6b796ab307842 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d9a328a291e4e8a8cc8ffad1431215b : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4b774eedefd42958a25f584fbfa7c02 : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34beeddb28cc4bd995b6dbabdce4369d : SUCCESS in 55m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c3419b868ac453fa9ca59525fbec3ce : SUCCESS in 1h 00m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c8760377c1542e8b555065306c2f590 : SUCCESS in 48m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a88ebdfec84d49219e725a2d4f7660be : SUCCESS in 51m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a73426b4a3364cc2b87e7e5aaee0be26 : SUCCESS in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246a06a60de9402a8d77539dc510f8ae : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5151b9268ae84cda8bb500744563fff1 : SUCCESS in 20m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45aeee1b7b034dd394e539cf8c859150 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":10},{"id":"e5fabb9542928c8f51a0f6015752379b190954b5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-06-09 07:17:52.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aafb15fbaf6ab52bb8a0b6f0c61e7355256d34cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 20:47:22.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d119131a710fbd3c086ce42128bf61a34c37be9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 21:12:52.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71461b667ded484cacb2b7e144a38011\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/832fb38589414398a164fc5a6da33a68 : SUCCESS in 19m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd260b8a26db4c6bb612aea6eea14073 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ed74995abb94b38af8a70fbd4f3f971 : SUCCESS in 24m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5190053e6dd4f14afd1c888fd3566f8 : NODE_FAILURE Node request 300-0021510539 failed in 0s","accounts_in_message":[],"_revision_number":11},{"id":"41f29865dbda814dae9b7a24fea7c06972fa2864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 22:28:42.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6e5434ebb354b24bad641be8582deed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78363a0587064ec2bba5cbb46c66c8ac : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/80e42a6ff2e44168a906b2372dd54d7c : SUCCESS in 1h 02m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf6734c7794b43a9aaf6420c83409bbc : SUCCESS in 35m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/049b439cae644f30a14fe633034d3717 : SUCCESS in 55m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7ffbda3e50645d9bbb98519c1b53f65 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2545c99343694885a3f27147167229be : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10ca480524f84f47b869fffd0b173939 : SUCCESS in 18m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b0e6b40bfad44c7bc78da92aa7b8dab : POST_FAILURE in 38m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7e48c78d14484ac4bb6100dc8b99157b : POST_FAILURE in 36m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dae69789fdb7436682782fb3f47a8741 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/edb6556cdcb54adea0babc998a9fd39c : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a28a425ea55549d699600d85adbc92ea : SUCCESS in 15m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f950f0b4183647e4b9f33c03cff39e8e : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/799673848b534a44a710d3a2929b0070 : SUCCESS in 19m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8615b0382ad4f48b94e2c42dc3cbc30 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4ec95c78e494c82bbec7f98d24f7e94 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/943f66e44d304dc4a103c367834f3490 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35f0448af49d4f42898db23c39bd9708 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6fa4f09581394387af064292b69c6570 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffb44ec4e96b4b6ebacb4bdcb6d3509a : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a68619c70a843d797e784f8567c59a4 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4802926bf5f4088831450ff5530e990 : SUCCESS in 10m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a934faba8ffa4443acba12ba306b609d : SUCCESS in 55m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aa694fe0684643e7bd1eebbc78388a54 : SUCCESS in 59m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e10a796e7a044b4ac90ee1997d3349a : POST_FAILURE in 58m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ebf88eeb54584723bc1bfcc621556c95 : SUCCESS in 54m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e11345d68740e28b1eb4dc626a2841 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3008680f0cab4c13ae490935da1e5c6d : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/720a5fb05eda4b1e85be4187ab5c58f3 : SUCCESS in 22m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b640675e76ac4d4dad14b06ca80fd451 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":11},{"id":"60f7b261c4ad8cc26a91a31e0f9c47e46ea8af4d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-21 22:09:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fa756dad006054ef23e66d66050719bd116c8b6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:30:43.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ce7f68e9204cdfb852121cb99114b3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2408d4d2f5646659ebc5f698be94056 : SUCCESS in 16m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b58a0125b3e046cbb4f7bfd3f0da6a10 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f971e4b70140480f94bb445dd4c28633 : SUCCESS in 20m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/83a0daa955224c1fa9171d3570fe3573 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":11},{"id":"4fb5d18a217163d84ee3cd4e174ee5e5366f829a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:20:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b269c29d2274903b8761eb93af1c3b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008bd017f7184202b85f202766f07b20 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/0486c3ef7a5f444ca9f9c205b5398133 : SUCCESS in 1h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f10081faed4f41449b2929c9e4b6a97a : SUCCESS in 34m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/438c9b5a0f14442fbb70cb580fc410ad : SUCCESS in 46m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c817f6ae69bf408f893cdb58566906ac : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fde1b137640b41db868152345b5a065e : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97e1ef2d6d324cc185cfec86e3a85eef : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aac40636b9c8438397f6b6c7da86ec0e : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e871d303ab84e9c91a320735eabc152 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf75406063364a50a67721098fa01352 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0a25793bf04f659fe2a2f722563288 : SUCCESS in 20m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6d971f01ac246869e7e95fe621b07d0 : SUCCESS in 15m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/80a23ee20f3d4d96b3a1d11fe3c3b03a : SUCCESS in 15m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/300dd3633bd348919e832bf1181d74d9 : SUCCESS in 15m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/233d22f21544427e9c76ce451f0ba777 : SUCCESS in 13m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f789a8d6faac442ead357214001e32ac : SUCCESS in 14m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88163562d61c490a80fa1c101632e5bb : SUCCESS in 15m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ac6af30d5ffd4596ac30b678c64b2d31 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ba75f7681ac5401db1c25781146ee0d0 : SUCCESS in 11m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff7f68ceb1df4f9b8ae8ed9143a73dbc : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37865292ae0b4d58a898096bc0f5cd29 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1736b312148e4516a7bb6604600c46e4 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c2593bb22fc43c4ad4ccf74fa1f15ed : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed15179d98454566b6cb1c817f663d52 : SUCCESS in 59m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/af6aba590e9d4a18852d38102ebd9826 : SUCCESS in 1h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c61b807195d402eba03009ba4712879 : SUCCESS in 46m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c82868688b4422589e1a730b2158ebf : SUCCESS in 43m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab828908b9424631b1713bc5c5e9f95e : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed5f9410b1844758cb2be22ede57ff4 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fded70db31964eebbb23a34f202f5045 : SUCCESS in 18m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7912b4aa4e340a8b715419739ce0fe9 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":11},{"id":"9600c8ec18102d24baab0dc2b9e3d94e897acc6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-10-16 22:44:14.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"a2693bf1b25f11c688874ca8346b9c41805dff6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 23:58:36.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/626597646f464479906330a4b6934b5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734391c9ce654a689ac31947265d14e1 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/e5b545a7895044c4961a0f279d29cc7c : FAILURE in 48m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b4485b81de804a15969cc4f80db315b1 : SUCCESS in 42m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84d8bf128cce4f78ad08353d581dc872 : SUCCESS in 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2230564af96f43099973e481dcdec856 : SUCCESS in 1h 03m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25e23995d6fe464ab49a36b8c208e886 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45680cc1e8f946029281b54ff98ea0cc : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/940d6f25576e48a8874f2a540a92b71f : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a7dd9347e234ce4ad0f45020f231e66 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96858cb07a7e4ba5a478fb0027357c34 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45364b8616814781bc3ca0d08a1ba611 : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8d3dd9341843e38c5bf6804ef10311 : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22196ebca1e74c7886743af150ae2bd9 : SUCCESS in 16m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/37240501c2d94812b88ccfd90a90b9fd : SUCCESS in 15m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7cb758f434544ce69b0966415c202564 : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9bafd81bb134e38abd314b5f5b94efd : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df77a079dce246d59f0ab36c93e6f2aa : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e7fdd96a44844df9e3b310e3c0c9446 : SUCCESS in 16m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/64abd8429547489f8c07bb6356ec81d3 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4162abf6e79048469044c0cc64f57c61 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d287f80a465543afad783d2966f0728f : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7efc0ccbcdab49e9876e977eb7d71f0d : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a780582e2444c04a2126ae9adfde4b4 : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/553a6ad586e343b4a9f9757d45b19568 : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/483ef818939c4f2c88b0f9efb62bdedb : SUCCESS in 1h 00m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9c82c3e1a88471991e43e0ed1e4e061 : SUCCESS in 59m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4e0e34646cc4fe7b6c3901c2d33f554 : SUCCESS in 41m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3839d73b95944e0b4a9902ec594d2fa : SUCCESS in 54m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/333bbfc63091432d98eaa4b36044328b : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c019ed24dc0c444f9ff8f2216d40bab7 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c67c40619f364722a66fa3f1b55c96f6 : SUCCESS in 22m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74c2e5a15cde470fa638336ac50b5947 : SUCCESS in 1h 02m 17s","accounts_in_message":[],"_revision_number":12},{"id":"89584253f2addb5bbecaf42f80f8808fb7d1c0e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 03:26:15.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1c9fa4f62494acc984f2cfcbf547bb1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96744c734ba94546be2b72c82625ac0b : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4192833876844fa1ad8ec153e2035409 : SUCCESS in 1h 00m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e377175ff5ae4e079dcf2eba861722c1 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1e1352a25f84e06a7c12bab658643fe : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":12},{"id":"01502ee65cf6ddfcb772c154349f05e25ff59e3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-10-17 17:11:28.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"4170e6fb4df7f9015f9b9f8aabb54648dd5e3bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:36:19.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e891386e5d4373b13b084ee904e0bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31ede1d55e547ab8e1c46ec26760a39 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a0180d424cf04647aea9c16510fab56a : SUCCESS in 1h 05m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b50c80e9406642c9b40a8a813a88c147 : SUCCESS in 1h 02m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd4af16ffb9341cd84caa014de547e72 : SUCCESS in 35m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb609b10458643549f3744bd5e6d868e : SUCCESS in 51m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eae9181da54d41cbb8bc0067e34ccb01 : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da8005f2205847a8ab4fd51fe922af4b : SUCCESS in 23m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97b33fecca9043c9a51b498d834b5313 : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3d7cf59ae274a4291b7864162287c10 : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c745768c7f1427fab8ffbbe200aedf5 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85297338d5254d3b85fc02b694ddc5ce : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1be92a2321c74f768da5bae0f36f0cba : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f7eb0c114774b888d79507d81fd4e34 : SUCCESS in 17m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e09a3a744d534f2c83ae56f52c6e3c1a : SUCCESS in 16m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/be6283186e6645d18e1f8bc0bbc79e43 : SUCCESS in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a15edaef1d6e48faa80e46528062310c : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f8c2b854a6447019f2844c9687df8af : SUCCESS in 16m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02dfbd3696164294954239f822505fdd : SUCCESS in 21m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b831271abaeb41d9947ec0a4559c8b94 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/167738c9f3e14a2d8b616861a4cba656 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c6ef3080b0f489e88d3de9cac79517b : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f9485438dfb498f822cacb8b86893ac : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a91f29acc8224cc69846a290b19c3ac7 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/613b4b5637524f9ca471f51714337a50 : SUCCESS in 9m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8e2fb9e37564d8a97fdc9ae4d06286e : SUCCESS in 54m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2daa3a4d40194ec5bc207ae4604d61b5 : SUCCESS in 1h 06m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1b008c5b1e34043b0fc29af23b5abd4 : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58bbfb27effc4ba0b5bed8ee09652688 : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20e3bdc848f84b1fabb2084a0bb4b3c6 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b0f5acc1814fe3ade6242002420adb : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03a921afed614a5783d8e45b7af278b2 : SUCCESS in 22m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41b2e3b7bfaa41c48f111d952b035810 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":12},{"id":"af038989ce4cc1b8fc814b969b0b6236114c7f23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 19:06:06.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c67223049a74a06962b5c7bcc1504b4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/774422f1ac8f40e68c4fa87b83b7f1dc : SUCCESS in 1h 14m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d62dbe74c2784626a01f5fe347dc07e9 : SUCCESS in 1h 19m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8991c12336c48c3a7872195e28d59f5 : SUCCESS in 1h 26m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f80269b129b642d18d78c14a25d4cffe : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":12},{"id":"9a97dc8ef57f93d6a1ea93c58428a05f4d997a96","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-03 15:25:48.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(15 comments)","accounts_in_message":[],"_revision_number":12},{"id":"44494f489ab39d6237476e597e5a1d7d097020ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 15:26:22.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"a4ba3691b84727108c6c1687469a75677bae93e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:36:45.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a810c172f9304f7f9e42c4b953cf67fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c620ac474f0341e398bf72fb001e369e : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/fbfeabd8b31c408bbf9e0c75956e6cff : SUCCESS in 1h 04m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0d19dd6abe24de7a96f659bd876b2e2 : SUCCESS in 1h 00m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/398e4a918b444310a366b15ddb2ef8c2 : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b296d8632a24731be5be38250c836e0 : SUCCESS in 1h 04m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e7a3a04d63fb4ae4a2ff4c895b9464bf : SUCCESS in 9m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/89c17d08ee574fcf8ce487b030e82195 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23ed390dc8eb42f19eaa20fe63a5838a : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48671f6f93484d8c976017335acc51c0 : SUCCESS in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aec1b832bc7a4838956348ba710de2cf : SUCCESS in 16m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/25150330b8af4f8aa29fd57781c0c54e : SUCCESS in 19m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0ce0722a09544a63b9363037692a1491 : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf74fdd51e7c4c909ccb1382ebf6c50c : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d423e5ea100b4372b258a58ea05a799b : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fb98ca2b5034691901d3087cd5dbaae : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5766ea96a7d401e92d8bee30e31666d : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8eb69eb715744c14953d265733b985ca : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9e26aa62d7f144c3b1cb138caa79907b : SUCCESS in 19m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c8d1b1e682846aca2c05af0a0731085 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5fac8603838445e6b987cddc5ea195ab : SUCCESS in 6m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca5f296c1b264813924b6be79a1f5a4e : SUCCESS in 48m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/496ccea2010b4f7191b6dfd185b00129 : SUCCESS in 1h 05m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd7460c7c6e7439f81a91fe4b7eff2d1 : SUCCESS in 54m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8642fa35c6d4c13b71d796d86348fa1 : SUCCESS in 59m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/722a63133e24463f91e92dee9da8cb76 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e337b5c8d671425dbceb1e08b769ae5f : SUCCESS in 5m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/198df55bcbca4642ae4a4b20f29e0bce : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":12},{"id":"ce2ccbd2f5df071d04a5a54aea4b9a0ed4d059db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:36:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"2f9369a47b5956f501cb7d81407b4beed8f4bf24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:38:17.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9995c6bbcbcf4d85acf7cbba3e7511af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9c0728e9a24bac8c2404e36ed4c522 : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9b121a1e989f4e04adb23a479a18f3fd : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9daa84aa562e4760b085d4863650901d : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fa65f39628334ded3b6f8274008e94d53dbf5f45","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-03 18:59:30.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d4293d4e2b261570bd73a8f96d6d1bfbb6acee18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-03 19:10:03.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"55f7833d86502a2c6c4d526df8883c25719b1c45","revisions":{"6cb55327dd6fddad6b79a2b0cb8f379eb688f2a6":{"kind":"REWORK","_number":1,"created":"2022-04-12 22:15:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/1"}}},"commit":{"parents":[{"commit":"3d7ff12064709fd77eb95cd2e2c3276707d5b5d6","subject":"Merge \"Obj Auditor: Quarantine ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7ff12064709fd77eb95cd2e2c3276707d5b5d6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 22:14:57.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6cb55327dd6fddad6b79a2b0cb8f379eb688f2a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6cb55327dd6fddad6b79a2b0cb8f379eb688f2a6"}]},"branch":"refs/heads/master"},"e24dccb7d244d4f2c10ec308adf0bcc14fa72b15":{"kind":"REWORK","_number":2,"created":"2022-04-13 01:00:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/2"}}},"commit":{"parents":[{"commit":"4a68bb2bf731f6947d5d268475df06724ddb377c","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a68bb2bf731f6947d5d268475df06724ddb377c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 00:59:09.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24dccb7d244d4f2c10ec308adf0bcc14fa72b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e24dccb7d244d4f2c10ec308adf0bcc14fa72b15"}]},"branch":"refs/heads/master"},"bfc4adb6b98699d670d857c18206110e646173a5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2022-04-13 20:43:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/3"}}},"commit":{"parents":[{"commit":"796335ea0ef5a9d69d30ee888f479ca1a3f35a3e","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/796335ea0ef5a9d69d30ee888f479ca1a3f35a3e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 20:37:36.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc4adb6b98699d670d857c18206110e646173a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc4adb6b98699d670d857c18206110e646173a5"}]},"branch":"refs/heads/master"},"069334d1dd4dd2d8c1029f4b2ab6347783c91918":{"kind":"REWORK","_number":4,"created":"2022-05-13 03:57:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/4"}}},"commit":{"parents":[{"commit":"3ba4d53afcf5b3bb817f901001e24c33eb2f365a","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba4d53afcf5b3bb817f901001e24c33eb2f365a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-13 00:51:49.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069334d1dd4dd2d8c1029f4b2ab6347783c91918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/069334d1dd4dd2d8c1029f4b2ab6347783c91918"}]},"branch":"refs/heads/master"},"ddc99e1cf3d9a82a905f5d1e707ef1bd6a19b257":{"kind":"REWORK","_number":5,"created":"2022-08-22 19:26:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/5"}}},"commit":{"parents":[{"commit":"3ba4d53afcf5b3bb817f901001e24c33eb2f365a","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba4d53afcf5b3bb817f901001e24c33eb2f365a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-22 19:25:37.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddc99e1cf3d9a82a905f5d1e707ef1bd6a19b257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddc99e1cf3d9a82a905f5d1e707ef1bd6a19b257"}]},"branch":"refs/heads/master"},"88a95704135310170c17b9847fefef9a9bcddd78":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-08-22 20:15:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/6"}}},"commit":{"parents":[{"commit":"c270eada8503c88bd8eaaa6f4e6822585a838815","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c270eada8503c88bd8eaaa6f4e6822585a838815"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-22 20:10:53.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a95704135310170c17b9847fefef9a9bcddd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a95704135310170c17b9847fefef9a9bcddd78"}]},"branch":"refs/heads/master"},"0d97c0b029b094b82dd5ed39aa26cad6fc4086ec":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-11-08 21:16:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/7"}}},"commit":{"parents":[{"commit":"7341982ca36c74541317f40c9075accfa673e647","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7341982ca36c74541317f40c9075accfa673e647"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-08 20:51:30.000000000","tz":-480},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d97c0b029b094b82dd5ed39aa26cad6fc4086ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d97c0b029b094b82dd5ed39aa26cad6fc4086ec"}]},"branch":"refs/heads/master"},"9dd412d115d6b382c3215b1ab13319390b6bfa6e":{"kind":"REWORK","_number":8,"created":"2023-02-16 23:05:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/8"}}},"commit":{"parents":[{"commit":"161294d374d29044345002f907ac8dcd0c251e4b","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/161294d374d29044345002f907ac8dcd0c251e4b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-17 06:34:48.000000000","tz":-480},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dd412d115d6b382c3215b1ab13319390b6bfa6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dd412d115d6b382c3215b1ab13319390b6bfa6e"}]},"branch":"refs/heads/master"},"48e7062fc47ad60e4213d4d3426d3e74b47f1b1f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-03-01 17:07:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/9"}}},"commit":{"parents":[{"commit":"ffce82c1e47b5a3b5b5ebac97b2f624d62b4e36b","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffce82c1e47b5a3b5b5ebac97b2f624d62b4e36b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-01 17:07:16.000000000","tz":-480},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e7062fc47ad60e4213d4d3426d3e74b47f1b1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e7062fc47ad60e4213d4d3426d3e74b47f1b1f"}]},"branch":"refs/heads/master"},"2a5669c6babf434e00c145648a7d5bdf5a724eb8":{"kind":"REWORK","_number":10,"created":"2023-04-06 03:53:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/10"}}},"commit":{"parents":[{"commit":"9164095268087dc413b1ee5bc14b170a6f3b6ef4","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9164095268087dc413b1ee5bc14b170a6f3b6ef4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-06 03:51:46.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a5669c6babf434e00c145648a7d5bdf5a724eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a5669c6babf434e00c145648a7d5bdf5a724eb8"}]},"branch":"refs/heads/master"},"96e934e0b055f38479470bf6fa8b1c25822c17e6":{"kind":"REWORK","_number":11,"created":"2023-06-21 20:47:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/11"}}},"commit":{"parents":[{"commit":"5bfeafa7db8e091ea98b656d26b73090b97aa8f3","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bfeafa7db8e091ea98b656d26b73090b97aa8f3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 20:42:38.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e934e0b055f38479470bf6fa8b1c25822c17e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e934e0b055f38479470bf6fa8b1c25822c17e6"}]},"branch":"refs/heads/master"},"55f7833d86502a2c6c4d526df8883c25719b1c45":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-10-16 22:44:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/837633/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/837633/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/837633/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/837633/12"}}},"commit":{"parents":[{"commit":"212525118c520606ba94895c522ba21ed0373d2f","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/212525118c520606ba94895c522ba21ed0373d2f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-12 21:00:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-10-16 22:44:06.000000000","tz":-420},"subject":"systemd: Send STOPPING/RELOADING notifications","message":"systemd: Send STOPPING/RELOADING notifications\n\nSee https://www.freedesktop.org/software/systemd/man/sd_notify.html#Description\nfor more information.\n\nNote that this requires that we keep the NOTIFY_SOCKET env var\naround for more than just the first READY message, so we want to be\ncareful about when we\u0027re sending the default \"READY\u003d1\".\n\nUpgradeImpact\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nSince prior versions of Swift would unset the NOTIFY_SOCKET env var,\nservices must be fully restarted (rather than seamlessly reloaded) to\nemit the new messages.\n\nRelated-Change: Ice224fc2a6ba0150be180955037c13fc90365479\nChange-Id: I201734ae0d6232ecb1923e67864dd928f90b6586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55f7833d86502a2c6c4d526df8883c25719b1c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55f7833d86502a2c6c4d526df8883c25719b1c45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
