)]}'
{"id":"openstack%2Fswift~837773","triplet_id":"openstack%2Fswift~master~Ibe514a7ab22d475517b1efc50de676f47d741a4c","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-14 10:57:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe514a7ab22d475517b1efc50de676f47d741a4c","subject":"s3api: Use constant-time string comparisons in check_signature","status":"MERGED","created":"2022-04-13 22:42:54.000000000","updated":"2022-04-14 10:59:05.000000000","submitted":"2022-04-14 10:57:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837773","meta_rev_id":"4b9c3f7fdde7fc4408f72717416e60495b24f803","_number":837773,"virtual_id_number":837773,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 10:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-04-13 23:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2022-04-14 09:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2022-04-14 09:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-13 23:40:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-04-14 00:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-14 09:15:01.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6f94a4d2e68a2c693f613e5c4d23ead532fc7f34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-13 22:42:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0e4cf6d7525db9b365fe5460bfbabb470e12482","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-13 23:40:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba6f9c45612fc1a6da28fcd43b02bc638d2c7faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 00:04:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804fb6a1f37b4cc7ab3511783ae7fa7e : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a866c2a7fd9c4c3db33a3871c2a01d47 : SUCCESS in 53m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d12003c07c9458ebd8c3fccf9df6a7b : SUCCESS in 36m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ca877aa98e94bdc9a6df5cafb9563fa : SUCCESS in 58m 59s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/34bc5b8ec63d4b16908ceac026df6af6 : SUCCESS in 31m 16s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1035b356c4934f12adb42224fe83728f : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1f575adba79c47a484a40f8dcf469d5b : SUCCESS in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4c59cef9a1e409bada0236fbdb77c08 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f67b6ac2c7274c179b9261788d95898f : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6320ece1be274b2bac27a0795119a36d : SUCCESS in 25m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/156719df2a334e3383cbb387af0160a8 : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0dc699b20bd048e3ac76f631e4fb2838 : SUCCESS in 15m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/953e15f8cf964d12aa9bc9c9574dace8 : SUCCESS in 28m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88ad364761c34c518622a3d26379a9b7 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f73f781ca38f4e4e8d5306049dc2ca80 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aee42af5c5bb465090c49ec3a6bd95c1 : SUCCESS in 28m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6de6af1c732c4bc2aaabcb863e2eb953 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9d042cc27294effb0d78a3c1e7c3332 : SUCCESS in 19m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d1f86f22f1a41b2ad23526df169657e : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76019ff5b51247b7baa48ca176723270 : SUCCESS in 15m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd84f0aad4844998a1c560d0989bb4f1 : RETRY_LIMIT in 2m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/65d1052eeddd4c9a92b43e4f914d2986 : SUCCESS in 1h 18m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f030610bdb14b57a5aa3650d6fc6738 : SUCCESS in 40m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0289f35faf04b0996fa697f362602b2 : SUCCESS in 40m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca20fd97267a4ae1a1a512591bbd5451 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206bee6ccbe54f3193e943807514a593 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5cf2616885f84a3a81dc4e0c8a79e919 : SUCCESS in 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cacad9728a474dd89306457f0f2f39e0 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4d28f7b3c6b0126181a1aff5b30550fc5d9b4375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 01:55:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fe521983e3a4f9198bc16c2a757402c : SUCCESS in 59m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9fffb1111a14274a6899c61ab441229 : SUCCESS in 40m 46s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/92fb607d16514c8ba6e8c15b9a9b566d : FAILURE in 2h 46m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1a166001efd48beb01f47368bb48eb2 : SUCCESS in 48m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/13fec7d1499a41a6945a332efbd3b3e6 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"23f82803a6d46e4c480ca0fb9b3155e53d351aff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-14 04:22:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b3d3071f91dc7d4309b6d113cc8bedf60387110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:28:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ea215f69ee47089bd08a799d5ee233 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c6a846bc688344c59944c57de4ad3898 : SUCCESS in 1h 00m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3bb99ca367de437eb02dfedd80dbe40a : SUCCESS in 58m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dacfa875186342eebd2549f51b296241 : SUCCESS in 56m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e4e2d7f52493490994d3388f55ced941 : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3b5fcb933b02448b835c218daac4f430 : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1caa318550ce4266a593184f11b8d566 : SUCCESS in 29m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20bb84e2e56b4676aade7519fa5bd149 : SUCCESS in 7m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/738d605bed544e30ba8bc5dd22cf79c0 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/937159c3eacc490a8ca99a00f263a013 : SUCCESS in 16m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cdfc75800d864558b5fc88b2f4685cff : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d3789c491ba647efb7a388ad0d580083 : SUCCESS in 22m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd0e02132a7e40fc8252d71bbdd0ca0a : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2dd9aebe4ef34b8d97f9405a438f9248 : SUCCESS in 24m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25a1c37a1391473b8515e7efc35a72d7 : SUCCESS in 20m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e53ccff8dc44a928844a0aef2754b16 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0a600e92ddc49c3a777f05ad4b532e5 : SUCCESS in 24m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/20ca0ee139ec442a88827e60eb6a6367 : SUCCESS in 20m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ca9c826c28a41cda41ef18b3bdb60b5 : SUCCESS in 8m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e9cd9d967a541e2a971a55300616de9 : SUCCESS in 19m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfb3d3d083fb4073b04767d0d7f3e77d : SUCCESS in 52m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec95d5339c754b21b7da01869aac578a : SUCCESS in 53m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1cd3e2e6bb0541b3a912ba390d352843 : SUCCESS in 42m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1c5dfaae1184d9eb311f1d27c0a6140 : SUCCESS in 41m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ea064f9f1da4f539f492074525800a7 : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b943d55c21e48bd9fe54b7e5cb3d0a0 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92b8dbb5efb54b89aeb362fcc79a211b : SUCCESS in 19m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32a979a863c2468c96d75c6e9161c832 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"64fce213bc4bbf208d82b7660bb16aa813a0b62f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 07:17:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5690d4dac11343149e5711e0b30ff9a9 : SUCCESS in 46m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/422499880d004f4f8ade18778f4bbbdd : SUCCESS in 41m 14s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/3f2c64b62a784886ae951940e8d3b735 : FAILURE in 2h 52m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/69dcecb38cd34f2894c340882bc98890 : SUCCESS in 45m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/84cee4caca1346598f16d5affb74f7d5 : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":1},{"id":"af77d3ccd9d8bbd573d99e55cd821903857d4dd4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-14 09:15:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa9624f7ad398e161aae0dd131408dd3585f20cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 09:16:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa30aa98e4504387847e6e8226a7067d60ac78db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:57:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610ae9490799484788e813ddc14bc6da : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/192203f83a2b4294b0f1510476316ac3 : SUCCESS in 46m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbedbf0675024cfd86547f50600e8b02 : SUCCESS in 52m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95a292e1c9c342b897e1b4b26fc7f645 : SUCCESS in 54m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/acf0fed52ceb4de19882966dfa76ab7b : SUCCESS in 8m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2c7dffa61524413a9a761ec35b8a6f83 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e7fefa41adf444ea2126c447a475029 : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d6cd42d21934da39b6c3c3b596e1071 : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f7884db0c3c45998c252eb134780e12 : SUCCESS in 17m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cf6b786dde0d4434b5f0bd48fae06003 : SUCCESS in 22m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/510dbb1c0b6946ee9a11e544cdaf54c2 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f84f7a70995410389aacdb05ee82cf1 : SUCCESS in 23m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9754608ef8c4df39fa621a96f7f0ccb : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ed8a1f523154cd6adfd68bd2dabd1d5 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99aa8f6da4d04d80ac06c605117ff049 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83e91fddd3244e25ba524df28de42248 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d8547298ff2b4b1c83fa80e1439f8110 : SUCCESS in 53m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/19753ea8a1944760bbed80cc839a02fd : SUCCESS in 55m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a905fa97f2ae45c5b4f5bc6bc66cdcf0 : SUCCESS in 44m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c6835db157747f995f21d012ecea91e : SUCCESS in 41m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/326e11c6acda46259aa2fe07ad119f4e : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0644148a8ab457e8d3a90bb3de2989b : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a9935ab76a545cd86b54ed34a383519 : SUCCESS in 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4e10cba0b2345289f27685cd4a76719 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1f79b7943a74255d4b4b3cf830b18cf1c14ba305","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:57:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b9c3f7fdde7fc4408f72717416e60495b24f803","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:59:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c487e0cdb348de9be77bd021d5486a : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f478861b1d244573bbd54298421bcbb3 : SUCCESS in 1m 15s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/53e42ccfef9446139a1dfdb9a47ad13c : SUCCESS in 1m 10s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6142ce88cc71037ba0cd23113eb6082fa91346ac","revisions":{"6142ce88cc71037ba0cd23113eb6082fa91346ac":{"kind":"REWORK","_number":1,"created":"2022-04-13 22:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/837773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/837773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/837773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/837773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/837773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/837773/1"}}},"commit":{"parents":[{"commit":"1cee51d5269c2b04f13623593710d2be0045781d","subject":"doc: also add reverse option to pagination doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cee51d5269c2b04f13623593710d2be0045781d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:31:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:42:30.000000000","tz":-420},"subject":"s3api: Use constant-time string comparisons in check_signature","message":"s3api: Use constant-time string comparisons in check_signature\n\nChange-Id: Ibe514a7ab22d475517b1efc50de676f47d741a4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6142ce88cc71037ba0cd23113eb6082fa91346ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6142ce88cc71037ba0cd23113eb6082fa91346ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
