)]}'
{"id":"openstack%2Fswift~837811","triplet_id":"openstack%2Fswift~master~I838dbba3c791fffa6a36ffdcf73eceeaff718373","project":"openstack/swift","branch":"master","topic":"no-stats-update-before-CLEAVED","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-02-02 04:07:55.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-12-13 12:45:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-02-01 22:48:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I838dbba3c791fffa6a36ffdcf73eceeaff718373","subject":"Sharding: No stat updates before CLEAVED state","status":"MERGED","created":"2022-04-14 02:05:46.000000000","updated":"2023-02-02 04:09:25.000000000","submitted":"2023-02-02 04:07:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":59,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"837811-no-stats-update-before-CLEAVED","meta_rev_id":"fa4d54c39bfd6239c8055cfbeb6655cbaf90ed67","_number":837811,"virtual_id_number":837811,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-02 04:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-12-13 12:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-01 22:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-01 22:48:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 03:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-03 05:03:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-08-05 17:52:43.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-12-08 18:07:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"5e90109085f009f314f05e4443405467dcbdf83c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-14 02:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49f119b2531f516439a3cd0523adb66213cef928","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-14 02:07:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cb994a78ccbe92523f672eb338415dd95af88cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 03:42:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a71a561a7b4b20b7f4ea7e9c4c17e4 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/6dce168db515481394e540e9b92d7616 : SUCCESS in 1h 03m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9afa015dd384e2eb71f6c008303c28d : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43c7df9f911d47dbaa067fa4909b1779 : SUCCESS in 1h 00m 16s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7bca9d4d64844d0b906f824a2fbf96a0 : SUCCESS in 22m 16s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1f1a510dc997472bbcb0fcf363bca781 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4d743f60986f44e1b52aad5fe184dbe1 : SUCCESS in 25m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02520d97ef3049f681573efabe2de7fd : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a91ce370829a42f285d3586bfa362b65 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b32543b6db6422e935bf8a2a6c6eca0 : SUCCESS in 17m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/263ec5869ba84206b43c0967fcff5494 : SUCCESS in 17m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a80585e876f4b2f88921b8ede70948a : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ce0dd9687bb4f3d8adc1220a7a0094e : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d6716fd05a04c3fb3bdf4a9304de680 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4055a1a188c4deba8a50b6fad4ad80d : SUCCESS in 17m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d8663cbdf9c0429a979a1180e062d65f : SUCCESS in 45m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0d8203eadf9041129db11d9110b8b936 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39053c9cd7834218ad775ede5137e80f : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28a0adf487434e849ffc00b832376825 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a77cd20c73e425b9a502a7e75d60e11 : SUCCESS in 17m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/46cb6b4653414f188202a04b793690cf : SUCCESS in 52m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3269169dd6094ac68289604dadf84795 : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3df8344cbe344c2faef5c2ef4f7336b9 : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2242d421729b4300a905a521cf807a50 : SUCCESS in 48m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30e665ed82764740acccab9839287699 : SUCCESS in 18m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75904724e0845399a0bbd3ac1ddbb5a : SUCCESS in 7m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e37344fa6fd43acbf326e8fbdd7d0ad : SUCCESS in 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39718abdbe6646fd9ed82a16fe95fe43 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ff84ebafc767630f6da239ba404c303b84db7622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 05:16:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e4e3f486b414eaab805ca2e80ab128f : SUCCESS in 44m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c847cf116b9b46a7b9164d24acc0bdff : SUCCESS in 56m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/45f3dc6b71f34b91be64003afa97f50d : FAILURE in 2h 45m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdc81bdfef8b49ed8a7a79b786d1324a : SUCCESS in 57m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/194ae92361e54ff69f470f3411aa7b29 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f618c66049b6e0cfe75d5f4e62f9d1aebf97b202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-26 03:00:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ba75d44242ce34d6d51a9476d0a91520e4602804","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 03:47:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5c769b1ea0468c9631a405597b4eca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5527b73b6ba94df182a834ce8272e924 : SUCCESS in 45m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71a5d3000de84717891c6857a162c041 : SUCCESS in 43m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/64c9c31b581940289b938be1c80c7f93 : SUCCESS in 43m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f886052c98f41538b2fcbce8acaf046 : SUCCESS in 43m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ae5aeb579b595a74a169657bb453cabbc33efc9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 04:11:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842c4fb16c0d4018a8541eed23ce08c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35536ed2c7df4c03b2d939deb4b6050f : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/9b98b2b03aaf479bb84fdf873b121130 : SUCCESS in 1h 08m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b5703ad767e4b83b0b0905ace07433a : SUCCESS in 56m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ce630f3ebcc45fdbe1d650717f051e8 : SUCCESS in 1h 01m 34s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8e75bfa377e14225bf7c7fe80642389a : RETRY_LIMIT in 5m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/62eeec8fc7604e23afbbbafba62705ad : RETRY_LIMIT in 4m 19s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b4e5eb711d4946418db9fbba345842fc : RETRY_LIMIT in 4m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/038e438a12cd431d9270b4892ef94665 : FAILURE in 4m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6739f523e94c412e809b990572ef1249 : FAILURE in 8m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d2da8c9bfd342ecb45666c2f61feaa6 : SUCCESS in 21m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b45feb5fe2143749f174d75fe8f87a6 : SUCCESS in 13m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b10983105e174cd68269aad317e798ed : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdc939807f53486c899c5e585d4dddf8 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0794c573f6d14afa8e0108c7181b634b : SUCCESS in 24m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ce4566649134831ae7a640aa6637c41 : SUCCESS in 19m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0ed63aa7f7e74b71add3d451a847101b : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8df93a57c65f40f288d9195a1583ed97 : SUCCESS in 21m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c63f34c95ba4f2cb9d0e277c702b213 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24c7a824232f457abcf33df645f69ca3 : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b94645fbb7c45faa82415dd0f4d124c : SUCCESS in 23m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f8c4ce5b6b0a4dd6be2c5450ab83e54a : SUCCESS in 13m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ea13a0fb73345ed9b49e9b2350cdb6c : SUCCESS in 44m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9612b1c8fdb4291b77a8d6b0b621106 : SUCCESS in 1h 00m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08f142ead7a14a12bafa93620bd081c3 : SUCCESS in 49m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ad09115b9b54adb84cb25b86f3c2a8b : SUCCESS in 37m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c81c09d5f81041eab3b2d8b7240fe9d2 : SUCCESS in 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee874d0a5ea45b19c703e0d87e0d78a : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/128f0843fcb84fa8a3211203713654a0 : SUCCESS in 34m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/676b6e0aa1da4eddb8219d8cac784023 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8ccd0f2672102dbf45c5399e43fa70be7fdc0c1a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-03 05:03:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a5f46d9491d1a08048f840a0eacc0f2703f1a3a6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-03 07:37:57.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0d9483142c9cf1bdb1e7015017c4c691fe047b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-03 07:38:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0eb54664dde9fd8264efaf2226796ef69350217","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 08:47:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce7ac52d42f24b82a12cdd68d760729b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d330835285e49f9ae879a817c00cc74 : SUCCESS in 1h 07m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3958fd47ef694979bbfbd665491741f2 : SUCCESS in 38m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c399846526e845c4aecf4837a23ae939 : SUCCESS in 38m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/156ade5025c945f596e4503b57c3e7d9 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":3},{"id":"24130adcb34d2ea4e6b211392f08823a9441b7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 08:49:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bc1159770b74cbcab8a05b4b34f5a6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff0b42b92cbb49e591923dc02b223605 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4b180bd53773472a8600c00accbc15a9 : SUCCESS in 53m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd0c3aa9dcd94592b0a59bd18814340a : SUCCESS in 44m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d2dd9e82b884f7eb9dd4783d6145f6a : SUCCESS in 47m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8448fe73482c491a9f86c7684e871105 : SUCCESS in 26m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/84df22df6c3f4e1d9a6b7c5b8e6459c3 : SUCCESS in 26m 53s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/19ac498e9bfa43ff8d46ef731bece76c : SUCCESS in 25m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a07279f014c44909b6ced8b7ad6199a9 : FAILURE in 4m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7f613cc0511041fd800fbf1bde133f8f : FAILURE in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df7352fd0a464e1b99f6128f48f3eff6 : SUCCESS in 22m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fd4644b11f947fe9f795ffcc96294bb : SUCCESS in 21m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/00918dd44f714ba0aafcfb614fe3ee2e : SUCCESS in 22m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75482f1a272e4323af12fa268e3da77d : SUCCESS in 27m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/23681d6b0a3d4fbc9a9327c94d4cb0a2 : SUCCESS in 25m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8eec078bda84c1d83ddb30380025ac3 : SUCCESS in 25m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/98797716377b4319a1f4f8f0a82c3193 : SUCCESS in 23m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d8848c0f3fd14ef2839b4d06fa05c5e0 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/284c2b77aa774f8999faba730ff4fe96 : SUCCESS in 25m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06d9963c58b54a68be60c7506464d55d : SUCCESS in 9m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fa80acc7cfb4ef485198a7d9b50487c : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/370a42e651d641dd8ee4ddbd4edb870b : SUCCESS in 8m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/473ddf11a0f64e2683e9465423678a6d : SUCCESS in 56m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9056f0d731a4abface7ec2955087f59 : SUCCESS in 1h 09m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a5cb9f518044d8e87b7e0f9436cb806 : SUCCESS in 46m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a51c63194daa4048ace5ea819d06de8d : SUCCESS in 44m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cac1bdcb2a044d2fbf8bd609059462e9 : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2afc5b828c34311b7754eb939d04aec : FAILURE in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/389eb912bb1e4644aea9ad12a6c159f0 : SUCCESS in 30m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05ee62b8b4754305a2661aa1ee3c3d67 : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":3},{"id":"8674dee96b3d8905c0065636d8d173db45c739e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-05 05:38:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"595ce29a67074c014eb1c28081574d00be4384e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 06:13:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f91099d9df946719b58bb1c20ff35a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c42430177214ea6b96586429019d8a7 : SUCCESS in 29m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2f1c84eb405446ea45e489c66d846cd : SUCCESS in 28m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/861dc748255148b397b814e6b4f7c8ab : SUCCESS in 29m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4b7594adc0a446fab595d25c3015fe8 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":4},{"id":"9e232acbf67153c7d1e46e1309ecab10f5b298e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 06:42:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d68859aaef7b4cc28bef31d1f0ca787e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21a66bc51a6478bac7de262354cef7f : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/e4d4343260d94faa9ffc4d680e9c4c6c : SUCCESS in 59m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/106bda3e6a05435096b5b843cb883251 : SUCCESS in 44m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e372b2e71284883b338a32783dd42c0 : SUCCESS in 1h 00m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/613371fb85144f5c81c35b8f47b95dec : SUCCESS in 26m 14s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c59a62ff66a54039b8af6e99b5de4488 : SUCCESS in 26m 27s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d6b5dee72e7e48cab75f4f27b570fd77 : SUCCESS in 26m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7a911d8331e463ba8d1e71096059ce3 : FAILURE in 4m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/406be59822884ba5ae1c47750fa9ff98 : FAILURE in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/389dc3c39bb446b680c0c9a6db5fe150 : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f978397a26d4355b2b52dbf9618c23c : SUCCESS in 19m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/157068156098409b886ee39a3223ed48 : SUCCESS in 20m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac10cc6e0d694dc4b28c0c92c57cc7b5 : SUCCESS in 22m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/703c271233e4458ab1406ef648ec285d : SUCCESS in 23m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bcf36088ec55498ab17ea27ee94d9332 : SUCCESS in 25m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9e96cd2a2b3440d934828fac1ed2fe7 : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8988ff306ba45d7a791e60067f9bb19 : SUCCESS in 21m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/27ba4181df6c46438c21dcef7c2b82e5 : SUCCESS in 24m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ab904d1d308416d881654ff2dff1b2e : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2e661dc166348b6bdb2dcdc522e14b9 : SUCCESS in 22m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2752e920d9ed4c3db03c4cc136a2d3cb : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c49bebbdf024d3e82f87f98ab46862b : SUCCESS in 53m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/adba5cacc22c42bd88b8530ad10a1098 : SUCCESS in 59m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5a0f4d17b3c4b438855f153a7d033f6 : SUCCESS in 51m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08957e21e37d4ccda6a4f668c4569d11 : SUCCESS in 47m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/504bd51a9a474e5b8b225dec16870725 : SUCCESS in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ba296d2c9743b1a5520cd3026078bb : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11c29142236c41a8ad805ffba5d964ed : SUCCESS in 30m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7147f37086f64671a5e473964274ede6 : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":4},{"id":"98acb69c5ee4b8e8b3c137f6177c5b179cf77fb8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-05 17:52:43.000000000","message":"Patch Set 4:\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fd975d6baa4d4e4f759483faac3dd8fc0a9cb73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-08 06:31:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ad325ae3dde617dca1c54f60f5fe39c952e2bf9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 07:31:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e62e8f5255b54b09add1c69c73145ce3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c69e6f57d434511939e5968c6de3274 : SUCCESS in 41m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ea6afec2b8540869b26874e7755b13e : SUCCESS in 48m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/42d0ed7c85034563812d2467dab856ff : SUCCESS in 38m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/10607c5586964233be0d55ad0abf906d : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":5},{"id":"8434528c53161da58b4ac7cb343fde18ffff70ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 07:40:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc653905d1048a29d4de3f8ce962c74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f193d37540b44878dbf9b6d190995fd : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/63761f22660c496781a0ecc65ed919e9 : SUCCESS in 1h 07m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0cc46b80ba394505a1f3b236b70ea471 : SUCCESS in 44m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bacc3d3c26fc4b7aac2e4efed532501e : SUCCESS in 50m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/93b430c9ef9d413cb58b4dec49684ae3 : SUCCESS in 28m 58s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/93f82b8505fe4456902715a8b595af06 : SUCCESS in 30m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b235c5d312394caab8c4003708659065 : SUCCESS in 24m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a38e0010a31347d8baa11ae27608ed22 : FAILURE in 4m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b4756b5a5ec744738b089ea8793c8333 : FAILURE in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93164a9332ff4de89640fffa7bd789c4 : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bdfc820cecc44b3817b1c70d4bcf3b7 : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6bcbc3a980f4eea839ddaf295a88de9 : SUCCESS in 13m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/810bfdb621df4600b4b495f792cce25f : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/563e285a399f461c891d64df002288dd : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38bb6506aa1b4f28b217f50cfe53d3f1 : SUCCESS in 14m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e46eae3d36e24bfc99323e529b237ed7 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/86572c10208342a197743860e2a12ecc : SUCCESS in 12m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ba3b966f83d241f0b9f9376bd42b2a19 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9268f80f6457467ab5d37771525f1a40 : SUCCESS in 8m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e5f14280847d40d0a6bc162213756660 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e26443c162574cb5b6bf64687ccf3b1c : SUCCESS in 9m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c811c64f8a5c46929ced7569d600f984 : FAILURE in 44m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/10bcdb0dd2d147ee85aaa8c5e396093b : FAILURE in 46m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c25b53f67a2545bfa6baae76c7ff3018 : SUCCESS in 33m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08a7bbace00e4fd99d50425f78c37a3b : SUCCESS in 41m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2caf33f44324d8cbf55f6811e1b0726 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711ca6e2c2f242ffabbf6fd2dbf5882d : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62cdaa532a0640d0880c0b96552a0625 : SUCCESS in 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b825d7bcdeb4adfb02dbff303f4251f : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":5},{"id":"1212a399275991f87db9578ee7a037683c0bc7dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-21 08:29:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6f0e0dd61a7764abea886bfb2d7f2a31800e6a81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 09:14:34.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2982050a6e4a4c9f2f1a04dbca4ea1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f51c8f80e3054da0a8f622382ab03a31 : FAILURE in 32m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b183888358774321a06cfb099b5bae0f : FAILURE in 34m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9114e49f933a416689aa1edbb9b03dce : SUCCESS in 34m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/21b101d9b14e4f80a8376c8334ba3ea4 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":6},{"id":"1843a37c5a238ac70fd176805efc589bff449aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 09:36:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d81b1554bfbd4285ac980a3b47ae67ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1df6dbf2d2a44f22bd5f7b726b35b140 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ab3e7ebb36a64b96a595b1651dfbd067 : SUCCESS in 51m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/785588cad06a41b6b4df5458003013b4 : SUCCESS in 41m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb3cb4a4eaa44a2b912af38912835dbe : SUCCESS in 50m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b8240522d994bf0bca1f99df3a23afc : FAILURE in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e98dfa4475d4f5b9c2e508491e95e3e : FAILURE in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60120238704646b68127d6df2e1d832c : FAILURE in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3e8a462be5a4e1baad84a88fd2c4e87 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db4df876b0b342b4b4d21c61894bc32f : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6af3b3d068484087ae716c5da62bec79 : FAILURE in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d37acd942396497e8fc26392d56f9322 : FAILURE in 16m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/903853b588a94592870737e48f8f2c43 : FAILURE in 16m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30683cd26474495eb9bbee8bc1320be3 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b01f860cd6b425c875469103b266a2d : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc13ba74b11e438286f2edd7c557d707 : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5219e906f4f94343ba2061e3bb6f9f89 : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f28af939f7b9407fbdc869665cfd60a7 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4603fb08fd9c420d9e1c5ba0b0f98f53 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77a2acc19e064de192b833883f3c1981 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc956ebdc83f466b808716ca8cc908f5 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b7f68cbf00541adae132339d4efa44c : SUCCESS in 6m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/98bc28cdc2fc4eb692393c86a48d316d : FAILURE in 54m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef067c25e66242d1b9e670276f251d88 : FAILURE in 56m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/168dab7b4944471b932c8e7223634384 : SUCCESS in 31m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6d2586b07f743f2a791c3aaa0fc779e : SUCCESS in 47m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e45000be64a42dabfd7b5fa8c186203 : FAILURE in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13b71699e2d459493af65afb452f8ba : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfecc85c60664350b739f2d466a16038 : FAILURE in 17m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e42e3a185b34fbabb8101ea823e72f5 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":6},{"id":"e977917259873bd19a2ac13202c6eb2d07879b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-28 22:56:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4af46f94a1e14829c0aa61be76eea97d12da0a66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 23:55:14.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b2e109b0694080a07456d88767e50c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ef088f59732474ebfa0cb1fd98dece8 : FAILURE in 47m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8479822758814a4384865d395d8a000e : FAILURE in 44m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee94c443331940ca8d0b7b0316a8e5ec : SUCCESS in 56m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eb3691a0eb64d30bdb32c9930263794 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":7},{"id":"93a075166d7fb0de0b7d24a645b5074a85246165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 00:17:18.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/869e22d325914c08a4f2aa640a6ffaee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14daa1ed31d4514aa148aa969ad184d : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/bbacafa85c3b470f992b073ea00e4c99 : SUCCESS in 54m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1652c3e0be245468561ee22c9036224 : SUCCESS in 44m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88a6f8c91fd847eaacf3aad259682830 : SUCCESS in 59m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0b92837002343e59e367e7d71f85e52 : FAILURE in 18m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a189a9c739844ba9eee6ffbee9a62cb : FAILURE in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82447a10037849ce81634d433574f16b : FAILURE in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7f9098187214ccf8aa5f9246c0b3cef : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/48feb14fb11a4484a5cf9c05b3f05c31 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/632bc3e254934840ab9e7671960a609a : FAILURE in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba325860dc64cfd9c59524024120cac : FAILURE in 19m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5861628be8ae4a09808cd3754a262797 : FAILURE in 14m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4331c65f00146589de667d336be1027 : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c6b9662608145a6b02bdb17774009ec : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8ff9e2cb51e4797b3c86fcfb77a20ba : SUCCESS in 21m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7cf80b9eec14e598d09b1d660d6543d : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed7496389fbf4b1a897fb0aa654273a6 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57fec930f66442fa9cee16b27e6abacd : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/51a0b81a894a45da972b604da54dc225 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73e2bddd81854fc9b082d09002ce01a1 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/89ce4ce6283e4688bf5e38d0d7fd1178 : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db3b320f9be14d5f991dc927639170bd : FAILURE in 1h 03m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8fac4259edb4a3b994bc84d82644c08 : FAILURE in 1h 09m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3afce40233394c21a4f0f03704d4a45e : SUCCESS in 43m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17387d52763549f3b0ab2393adb80de4 : SUCCESS in 41m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa32ce9ffb07411b9ab3c73f9a375b2d : FAILURE in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbdbd9c65e614f25835d14901046662c : FAILURE in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc90b55d38e144d5ba93b1d11499bc7f : FAILURE in 19m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abfc0987a714468283701c1b76ab586f : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":7},{"id":"ad90a59a60b69504658bd9c23029cfa5c3cd7a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-29 07:40:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"a2dcca17a1e7242e4a1522c402ff1d9e274b3f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 08:43:53.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/413d8841734d457f96502d7ca7726914\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3647a5464cd493ea317508b6cf1702e : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a131bb0447b3453fb6ba704e5056daf5 : SUCCESS in 52m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f341939c25a941599c0475217e392209 : SUCCESS in 42m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/628ef41269ec484c9e26a0ac654affd0 : SUCCESS in 49m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff2659bca46d4e78befe51e1c1d9e221 : FAILURE in 26m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbcf544def044a6da0057cbb3ad06ea7 : FAILURE in 22m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc44b7ceeff345bbad0bc1cdeaa37ca0 : FAILURE in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9b816824866434c80cd6e8e26d2288f : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4fae69e07ff143d6b1ca07a541cf4b20 : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6c79a190cb44f0abe69cb76b39829c0 : FAILURE in 16m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1697835815f542e8a12652f25c83a2c8 : FAILURE in 13m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32a32bf91b934260b2aca2fb09e629ad : FAILURE in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d9fb76b4d5e49aeafc5847bc6636b68 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97532dc9fbfe49f9b32a66d3169c0a31 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c988c772ac1a4b5b9957d0e8b8b9f3c6 : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f06b8add2a24057b19d3b24692a6e13 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5df7602ce0264b45bcf0739b256406e0 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f2612bf17a5b4da2a438f96f0f617c7e : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb0b77c1a5b24686a0f22c10d1b679a6 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f77412aa95a41ad8b6223bbc213564f : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d4ccf16230c4e3f9b849f6c6bb15b54 : SUCCESS in 8m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3f83f7ae8ac452caf0163ffaa9fe8a3 : FAILURE in 58m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/07ec5983d5e74dfb97f4577dbc21b06a : FAILURE in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/961cd58d0e4a411398cedbf23f74ee15 : SUCCESS in 29m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9785b12856bb42bab26677db2cac3588 : SUCCESS in 41m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cd2a834680d495dbeadc952595bfdf6 : FAILURE in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00496b9ebec04e7190fe29d8d648d499 : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ead6aaf6bd83405c988f7fa67f4744cb : FAILURE in 26m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41d03585ec094f60a5053e82eea2999a : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":8},{"id":"e67cfeeb6a4437eab86931567bcca9877e9c3719","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 08:59:53.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a003d831c77f411cacefa67a51f386a6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/34b81f861d574120b925e560a53b0886 : FAILURE in 1h 16m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae11881108cf4b9eb4ae2d62c24c0114 : FAILURE in 1h 15m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb2cd14e2da949578270ceac0e219210 : SUCCESS in 49m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e90a35bb57f54669978d9e1ac6e0a9f3 : SUCCESS in 56m 12s","accounts_in_message":[],"_revision_number":8},{"id":"24fb50e2f5713328b08ba19ab0686470042fb7d8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-30 00:37:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fc3129618c1ba4c6f7e5eb7faf93b1c97a627c06","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-30 01:32:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d0b55e00d5e2ae02035d5f272217a9e784856979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-11-30 01:42:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ebad079c6634fd6dc853bbe8786671279f03d345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 02:47:47.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d44debf7c24f65b5389524b484ba46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8950ac4fa3c043c09028a78a39f8aa14 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/dd89526079e34a0bb2ce28afe063221d : SUCCESS in 53m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3083d68a13b64c70956e0f6ab584ee2e : SUCCESS in 43m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4398bd93bc43455e851f0ad1170ee29b : SUCCESS in 51m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cc1d73cd1fd47fe89913cdf7d0e7b08 : FAILURE in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b083d36f261454da4197830757427e7 : FAILURE in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ab2a0b6d704494ea61a8c0eb888bd91 : FAILURE in 22m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/624e862371454e298bac0957a1cf94ce : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/be339e31dd6443fc88ddd06ed72b15a4 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d92b9cd25894515bdf406960874375b : SUCCESS in 32m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e89c9be252cf41dc819e89340bc16ff0 : SUCCESS in 16m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0602872e8bba4e18b972e6c2ec0c1d31 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67f54fa9ede74841bd67b101841da190 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8512e7d5f3645629dc18c0714ab97bd : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70e9ff61b9114cc9bac5a88aa60bff40 : SUCCESS in 20m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/afa4feb1357f4101943515b90ca45e5a : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8215724586f04424a00897e36624da49 : SUCCESS in 22m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b7e8096c1ba4155a0ec4ff9ba6ed34b : SUCCESS in 18m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6544556552a4aaa8e0ea8e88671abff : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20ee3337b5ac42a38ab61a953a18069c : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/09eb0c7b89bf46739702c5d4046a718d : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/03517d7afa1c4d58b60f034c184cd48f : SUCCESS in 56m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/44541b43888c418bbea6f6421cab2c67 : SUCCESS in 1h 02m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e42692dadcdc4cc18605ead9b72191fd : SUCCESS in 40m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae94569e519745069bd11ad1314746e5 : SUCCESS in 39m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45d8be005f9b41feaa9594aafb78ecd2 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062d1837f9fc4bb89d1be9c0634d6fcd : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad2c5d9fc396490db889e2d4aa3347d4 : FAILURE in 28m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0acd85fc2628457d913246541e4e2943 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":9},{"id":"3b0113537b65261118b604f6f86ea14a6c65e4ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 03:16:29.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07750904287543188af09482bed0a010\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21ad17fca90e45c6b1179509f3ef1eb5 : TIMED_OUT in 1h 30m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/047d88580ffa4ac8b0c3b9ef7d9ac1c9 : SUCCESS in 1h 00m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb616991afed43ada7362e7554dbb97d : SUCCESS in 53m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/39b82cead9934097b5ff29706600d3d2 : TIMED_OUT in 1h 31m 17s","accounts_in_message":[],"_revision_number":9},{"id":"c77ee395851ae19e2a52ebe139375d5b67e0d9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-01 03:21:43.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"e01e89206bda0380f7022d44c56bc8ade3ad0c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 04:27:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344e3b4d73aa4131ac51037aae6ce39f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1d170c4a5b419faa2c88c3f281fda6 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/4d113d60e1cd4c619dd75a3bbf4679dc : SUCCESS in 50m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63a4becda6784472894b63ecd6089774 : SUCCESS in 37m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5570e13e8c04231a011ef3e4f8355ff : SUCCESS in 46m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e1b0bf47a0e4dc5a200c641c2438814 : FAILURE in 16m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18f5ef77404f4cbb8a225311fd4fc4e0 : FAILURE in 15m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e828187af3af41e49cecf126e1b70ded : FAILURE in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d4547275a5e41e888de75ab2a48ac8f : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9e1d96e01a1b43efa333dcc7144b824e : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07bc07a3161448e0bb634b9feb445b89 : SUCCESS in 13m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c63bff16dbe43769b93dc005608bf5b : SUCCESS in 13m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f5c08df6b8c845359c18056a2f53eb82 : SUCCESS in 11m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5534d4f4ccc495a94bbfe953e425b1b : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/001af0b83df041ab98a485e06ea962cf : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fc50e187bf947a494bdab6f2b4ba93d : SUCCESS in 16m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2093abcd79204a66851021a51bf29921 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7581ebd4e1f84d418da20f17b2e93b6f : SUCCESS in 12m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7626557d93694d159be8516c8f9fe6d9 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f98bbf3ba4f4cfca91e32872521fdae : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d16b398dedd1471787e12c22980ffd39 : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fa93af2692a4eb6ae08ccdc96e3409e : SUCCESS in 6m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca359352aa1f4c7d9cdfba32c81851e0 : SUCCESS in 1h 00m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b2b16b8da9a4ecfbbd87d0d19f5d7f6 : SUCCESS in 46m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/157d0dad16e64160a298f84e0e3ae498 : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bcf8167d64544301add9888f569055d7 : SUCCESS in 29m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ed9e2d930e74f59af43c67744e72f51 : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514970d9d2944da7912933019b08f7dd : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1572863572904b20920bcb4e00aad6e1 : FAILURE in 19m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4efe86ef1b545c48ecf0ac24c2930c5 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":10},{"id":"20acb5d05c537fa11e8ebf6b5a9c61d2e2899a62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 05:06:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7dac853367421b89092a699379251a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/888e4bed41c1409ea13ab2c6c919fd71 : SUCCESS in 1h 14m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a657d50dc6874e998fc2d5dc4d050c72 : SUCCESS in 1h 19m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5031d1f1ff4a47c49ffabee8b2e3d789 : SUCCESS in 57m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8de3a6cbcf9a47a88514670d406e6849 : SUCCESS in 1h 03m 23s","accounts_in_message":[],"_revision_number":10},{"id":"5505c401cda3fc236f8e71b37f69e4fc8dbc88a6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-06 03:59:47.000000000","message":"Patch Set 10:\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"99c604b3faaa3adfb230ef6454f5950cb077cd84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-06 06:17:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7b9f68972b80f7925b914b8c5e63d55da2a2461c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 07:20:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7440a7ccb54f41d9b0971f714ddeb116\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f61b461daba4a089c4cd5d89eebaa2b : SUCCESS in 54m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8a9500ef10c481ca99c3781299902a3 : SUCCESS in 1h 01m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/68d3f2b3a37e459395b06f11954d7e1b : SUCCESS in 1h 00m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4ab4b2713e04a57821e8e826364cb86 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":11},{"id":"2784bfdcf3b826a61406a8d3be0c4a4f654e1945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 07:30:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd7a3b0ab0d49b1a40e6ee3f6e9efc8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e6b580b515457d839c897c676926d8 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/13e7b0b7f60a4bc2864744ed8f786349 : SUCCESS in 40m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6b4cd4dfdd35449f86677c38c04e5be8 : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2289e7e56e0b4481a79d83d20a5c95f1 : SUCCESS in 40m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17d17a81a0944274b61ea232145ef59e : FAILURE in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7699411afa1642b8a49990bee2dde218 : FAILURE in 13m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd419345b43244e28244267925a8f3b3 : FAILURE in 14m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/018bccfc4315427e912176bc9cc9e7dd : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/abe5410218e24dc4b393d5afeaec36e3 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4da7f834210944d7affeaf0809c81e0d : SUCCESS in 12m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98eadae7f2644f2e8052d42a8792be27 : SUCCESS in 14m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b505464e25040c480d7131d93417d55 : SUCCESS in 11m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5df3dc307e454d67a24dd4984bad67fc : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e56fac27a1dc426aa2ad15c3e9f9ef7f : SUCCESS in 12m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e375d304c569455baf33291a15e54d71 : SUCCESS in 14m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81ed46b440e24b24ad7eaef63f47e5b8 : SUCCESS in 11m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/753690f308e14a12bbd75a60eb684530 : SUCCESS in 12m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2f4de01910534185afe09070c01fd30a : SUCCESS in 21m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ffb3266ea4c4f349b2b2989cdf4935c : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aca2d8cce02c4f689706cce9fb27e346 : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e882e44ddf84978a45610c34520dae1 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ceacd82f6a284d54b6c85c934123e66b : SUCCESS in 49m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd6ae656047a429bbf23d957beb5e038 : SUCCESS in 1h 07m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/edd093609b8549df8315a41eeab2459c : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac89c2e2f76a410eafa581b77b16aabe : SUCCESS in 48m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1723fa03a774750b394fcd3c658ea3e : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d84955727a1443eb0edf05bd9df27e5 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/733b993bdf824a10ae99b95e4d25f1e9 : FAILURE in 19m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d15f657c26247bdb65dd006c141b3db : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":11},{"id":"dbac0b989a6bfafe0dd616b3e5f5095f96f0db99","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-08 18:07:10.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"dbb8e6d8a92a43b1bfeb705ed995c8260aeb646b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-09 02:34:00.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"79ee0d0e379622cf1f26c0458c5a6f871b5c53d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-09 02:35:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3b9174066c388d2993ad7fe4eda21b6d2315d91e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 03:55:15.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72209fb6926c4ed9a25dd824acf9ba1f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514ec4013dc24ae681016c55e822c3ef : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/acc1180d70d743eab054a25d1d091e6d : FAILURE in 36m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be3d9de3fa9b4bcdbd00cc94f5c1abd2 : SUCCESS in 56m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64f5490b1ab84c758f139f6b0eb5e2e8 : SUCCESS in 1h 10m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cc5cf15421b4fb4a32b0b4940e10e4d : FAILURE in 23m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f266f3fc67244688e7be28e4df6b047 : FAILURE in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b52b56490e774af1825ee4365ad8c974 : FAILURE in 35m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aca712b52df94ac8a1cb5e9fe3059f90 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c48599c1328400f992fd09c195cf24e : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/590c72ac178c4d0f8444aefd30516e45 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba110266962443597e4ca29cb1778f6 : SUCCESS in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dfbd72ace345443aad70b4a84ab96c7c : SUCCESS in 14m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad10ece6ab1644d2b2403ed76ea96986 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/069e302e357d41cf9502f9aef1460f9d : SUCCESS in 16m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d2e0410bbb404c1d816dec1604a7eed6 : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/94bab4b7beab4e799ad6e71826295b2b : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a0db5dceeff4efba4d8dc746583a3d6 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4784cb3342a84e2ca876b95e29d33c12 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c6f0dc19b904c1e8645b14f19abc0fe : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25ec49d7cf544558bbe5c04040b8350f : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4becf4b8c7d4c3397cd8dc88af45649 : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9f65478e3622456592ced9895a7c0d84 : SUCCESS in 50m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/658e8410b8024d64880ef2aa826371d6 : SUCCESS in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53db30b3c92b4b45871d0e0279b5e7e2 : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6154c59b9d24221a5a5a1bf7be68723 : SUCCESS in 42m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29fe4e7ac2084c88a6db654e6401d56e : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9e8668ac4e4f299efb7c8c8a9f33e2 : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebc4e3bf1fda4526ab7114137b4b39bf : FAILURE in 22m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c21d269190d14230a569d4972b77ecf8 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":12},{"id":"d47d1e504efe9190a326687151b724a55f14823a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 04:33:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afe359650cc4794a4522762de889a3e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4d2944538e849b68411ec50a59f1cde : SUCCESS in 58m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edeb185ffe1749bba959e087f076f5d0 : SUCCESS in 1h 20m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/540671b90b8c4339a44d8bc4d2b179b7 : SUCCESS in 1h 00m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f66e4dec44644e2b6ff523f79241070 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":12},{"id":"02569842f310b91c41d2288b9cd78d14ba3db2e8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-09 11:49:28.000000000","message":"Patch Set 12: Code-Review+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b2cb3604b651eca69a0e926d9005aa5509b4d33d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-12 06:42:13.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c7b723a679deb80b23402ae42e97117399862ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-12 06:42:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a098738e0c355d53a714ab5b73329509b1fda73d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 07:46:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf3b173d73542bbac5096b085662349\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d26d62b161b84e8f868b3ff9922b815e : SUCCESS in 46m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff955843745d48b8bcb0ae539f53e4d0 : SUCCESS in 44m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/538a751fcd2046b6adcc80b031890c1b : SUCCESS in 1h 02m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56dfa3ac045b449f93d3fa578f8f1532 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":13},{"id":"6a67867b2f4cd1610e717a6f7093bbc52f9b3d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:12:23.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3789f6be5f874a2b81804a863f6a3144\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd6d53298b542eab841c0d2170b0b33 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/6ba19377f5e1462ebf065230c2d02b9d : SUCCESS in 51m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fcf6a996c07e4c14a90734cd59145dad : SUCCESS in 38m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f493e48883d1478ab4f4aac3292511b3 : SUCCESS in 49m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/394f2fe79e7b4bf19c0fb990dbbb104d : FAILURE in 22m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01b94a1ba39d4a0da7b0146845a0e287 : FAILURE in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eebb4fad6194884b8f5cbcf5b471df1 : FAILURE in 24m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fced51afe7b1483f82114c7ee2fe487a : SUCCESS in 8m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3ae40e0ed924cb4921fecd3a0077ff2 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a4274ec69104e34869ea73070fafa14 : SUCCESS in 19m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf46180217aa4feb8854eb95fcadd7bc : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15bbff88cab84d7486802be2f5a3962a : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a10334fbc8734254a8b6c6402e34da82 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/943350fa235b47f99f7ce6f6a5482c8d : SUCCESS in 19m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5c47ca46481e49ea881219aca9ec1ff0 : SUCCESS in 14m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be00b5f4333d4570a80cb0f8b1486c5b : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3fe47cf9a7f249c5ab862805b1ff46a9 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2b1f02e13e6646f28f507e84eedceae9 : SUCCESS in 10m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01c34a74cb1b4221bcd12a1b2923a2ba : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88fd00671033460dacd95cc418a91ee3 : SUCCESS in 14m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6ea485c1b11b460c83e6c2235a8d921b : SUCCESS in 6m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/46c38c15598a40d2927ea94fde9c8e0f : SUCCESS in 48m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cdf8c35e5c7447cdad9402d4db72b6bd : SUCCESS in 58m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/034adfee31034c7992d0ab37a72b495f : SUCCESS in 31m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4985a6ce8699493e81ba2d060cb8500b : SUCCESS in 31m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b81694f84daa4692887356be2073a732 : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b885e1f7c00499eaed719a3d27538c4 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f2284e51bbd4648bffc65569a790df1 : FAILURE in 17m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/005be66002754ba5a1ad59443b25bcc6 : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":13},{"id":"e5d34a4e957a84c826909384790baed8ad179855","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-13 12:45:39.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"48cd21075b20ba47c8674dd907dc9a074c9979ec","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-01 22:48:50.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0ce612345a5e92a30577cbdb7dd9dc613991ae83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 22:49:55.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"e27d3514260f1c411021b356378e4df487e8f461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:07:54.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6899729176c84cf9be97d30668b8b977\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e4b8b47da649f394f3e129fbb84622 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/73951e48cb6b4306a699d35ebd685442 : SUCCESS in 1h 03m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/467aa6e91a754d5da37a8bdb3635484b : SUCCESS in 1h 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a17a291147284869aeec6ffa85ca09ac : SUCCESS in 58m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cc1968ffee5d49a3a85d55dccf721ef2 : SUCCESS in 10m 15s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2f90ce19af42493f99bf315542134601 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a35bcd3e9e134ce1af329bf1dc9820b5 : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5a04524c22e42d28dceb4d1c7a6228b : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd3e1370790a4274bd241c4fd65e6f7d : SUCCESS in 17m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/31f95d6805284492a0d786c34b69fb2a : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9be389c541b74d92a4c3cf183fb7419c : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d75c37fd07f247eca7840063e6fd28c9 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd21220532f2414daf9ac1e0304c848d : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b70bbaf5178e4da8b7ce97d8de5062e3 : SUCCESS in 14m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/31b64c81dc1e4f89a84f184438529591 : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8b8244644c24d24ac31702a288c1bc7 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3de5d8fe0b54ab894b547c877dc2597 : SUCCESS in 9m 32s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/283b3f9af86d4cb6b924b1435be3708e : SUCCESS in 10m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/26685f5675534701bb92594f9b23faf7 : SUCCESS in 1h 00m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fca54593628f422abf94a3d3a5803299 : SUCCESS in 1h 07m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/441046e24d8049ba9e45de6a791da9bd : SUCCESS in 45m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aef86773e9164fe9b7552eb8549160fc : SUCCESS in 1h 23m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9247a2ec2577462ebbe2b023ec63a594 : SUCCESS in 39m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361bd482cee14466826013a8b0c2048e : SUCCESS in 5m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6467419833b9423d8590969ca80c356e : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":13},{"id":"3d021af789d75f1a86612d03d7c7631dac11af5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:07:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"fa4d54c39bfd6239c8055cfbeb6655cbaf90ed67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-02 04:09:25.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb49d99b78547308fd2a08f2cb2b356\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9dc57c4ace43d58db1715b9a0dc295 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/641362b368164353b8287127c5d9f012 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e08b666b18094010823161f752870a22 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"ece4b04e8215f80b55c367abc523290b503efc76","revisions":{"1be971df020fb769d065747b92d268327241c5c8":{"kind":"REWORK","_number":1,"created":"2022-04-14 02:05:46.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/1"}}},"commit":{"parents":[{"commit":"3d7ff12064709fd77eb95cd2e2c3276707d5b5d6","subject":"Merge \"Obj Auditor: Quarantine ENODATA\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d7ff12064709fd77eb95cd2e2c3276707d5b5d6"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-12 07:07:08.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-14 01:55:26.000000000","tz":600},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, until a shard gets to the CLEAVED state, that it holds the\nobject data for it\u0027s namespace, the number to returns is incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d SHARD_LISTING_STATES + [ShardRange.SHARDED,\n                                                       ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleaders who split the namespace is concerned.\n\nThis could be used as to allow a small replica to know if it has an\nundersized portion of the required namespace and wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1be971df020fb769d065747b92d268327241c5c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1be971df020fb769d065747b92d268327241c5c8"}]},"branch":"refs/heads/master"},"a0ce85a8c4a84a4c7cff4b461e3864837cdf81ed":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-26 03:00:29.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/2"}}},"commit":{"parents":[{"commit":"f7cd2d905ac42d55d53a94eb6fa1a613dd74abef","subject":"Merge \"Drain and close response in StaticLargeObject.get_slo_segments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7cd2d905ac42d55d53a94eb6fa1a613dd74abef"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-12 07:07:08.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-26 02:42:16.000000000","tz":600},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, until a shard gets to the CLEAVED state, that it holds the\nobject data for it\u0027s namespace, the number to returns is incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d SHARD_LISTING_STATES + [ShardRange.SHARDED,\n                                                       ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleaders who split the namespace is concerned.\n\nThis could be used as to allow a small replica to know if it has an\nundersized portion of the required namespace and wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0ce85a8c4a84a4c7cff4b461e3864837cdf81ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0ce85a8c4a84a4c7cff4b461e3864837cdf81ed"}]},"branch":"refs/heads/master"},"bd7e0b14989994da559fd3f90177f885753da664":{"kind":"REWORK","_number":3,"created":"2022-08-03 07:38:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/3"}}},"commit":{"parents":[{"commit":"f7cd2d905ac42d55d53a94eb6fa1a613dd74abef","subject":"Merge \"Drain and close response in StaticLargeObject.get_slo_segments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7cd2d905ac42d55d53a94eb6fa1a613dd74abef"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-12 07:07:08.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-03 07:38:31.000000000","tz":600},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, until a shard gets to the CLEAVED state, that it holds the\nobject data for it\u0027s namespace, the number to returns is incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d SHARD_LISTING_STATES + [ShardRange.SHARDED,\n                                                       ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleaders who split the namespace is concerned.\n\nThis could be used as to allow a small replica to know if it has an\nundersized portion of the required namespace and wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7e0b14989994da559fd3f90177f885753da664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd7e0b14989994da559fd3f90177f885753da664"}]},"branch":"refs/heads/master"},"c4d615b64b3b07126f7d25f6545e253aa89a2e09":{"kind":"REWORK","_number":4,"created":"2022-08-05 05:38:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/4"}}},"commit":{"parents":[{"commit":"f7cd2d905ac42d55d53a94eb6fa1a613dd74abef","subject":"Merge \"Drain and close response in StaticLargeObject.get_slo_segments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f7cd2d905ac42d55d53a94eb6fa1a613dd74abef"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-12 07:07:08.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-05 05:21:28.000000000","tz":600},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, until a shard gets to the CLEAVED state, that it holds the\nobject data for it\u0027s namespace, the number to returns is incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d SHARD_LISTING_STATES + [ShardRange.SHARDED,\n                                                       ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleaders who split the namespace is concerned.\n\nThis could be used as to allow a small replica to know if it has an\nundersized portion of the required namespace and wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4d615b64b3b07126f7d25f6545e253aa89a2e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4d615b64b3b07126f7d25f6545e253aa89a2e09"}]},"branch":"refs/heads/master"},"33e11539a07ef9c3b6604f715bc893e62c3fea07":{"kind":"REWORK","_number":5,"created":"2022-08-08 06:31:42.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/5"}}},"commit":{"parents":[{"commit":"fe39596194433ce983d19224a03ff2b8773ad1c9","subject":"WIP sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe39596194433ce983d19224a03ff2b8773ad1c9"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33e11539a07ef9c3b6604f715bc893e62c3fea07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33e11539a07ef9c3b6604f715bc893e62c3fea07"}]},"branch":"refs/heads/master"},"a04246e9a61b475d9579a72ac8619f3f52f7fe0e":{"kind":"REWORK","_number":6,"created":"2022-11-21 08:29:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/6"}}},"commit":{"parents":[{"commit":"29e9dace0cd075e72c166474704046dbbcdf1f09","subject":"WIP sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29e9dace0cd075e72c166474704046dbbcdf1f09"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-21 08:24:37.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a04246e9a61b475d9579a72ac8619f3f52f7fe0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a04246e9a61b475d9579a72ac8619f3f52f7fe0e"}]},"branch":"refs/heads/master"},"76ee79ca3e7730c73df1cde909624b13e121cb18":{"kind":"REWORK","_number":7,"created":"2022-11-28 22:56:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/7"}}},"commit":{"parents":[{"commit":"b6664829d6eeac59951f7123dc2aed285324f965","subject":"WIP sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6664829d6eeac59951f7123dc2aed285324f965"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-28 22:53:14.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76ee79ca3e7730c73df1cde909624b13e121cb18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76ee79ca3e7730c73df1cde909624b13e121cb18"}]},"branch":"refs/heads/master"},"6aa8f27bd6e8bf54c7f79daf9f281ed2953a8a18":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-11-29 07:40:53.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/8"}}},"commit":{"parents":[{"commit":"b53b8984e39b146ac5509de069fb60f3d90d263b","subject":"WIP sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b53b8984e39b146ac5509de069fb60f3d90d263b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-29 07:40:39.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6aa8f27bd6e8bf54c7f79daf9f281ed2953a8a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6aa8f27bd6e8bf54c7f79daf9f281ed2953a8a18"}]},"branch":"refs/heads/master"},"d7a83454619bfc0a993ed2a06fc0a6f16de6506a":{"kind":"REWORK","_number":9,"created":"2022-11-30 01:42:42.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/9"}}},"commit":{"parents":[{"commit":"1ee33cf5047e3b36d4d93d598ca4d7f0210e7cfe","subject":"WIP sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ee33cf5047e3b36d4d93d598ca4d7f0210e7cfe"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-11-30 01:30:27.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nAs we don\u0027t want to update the OSR stats and the meta_timestmap, also\nmove tombstone updates to only happen when in a SHARD_UPDATE_STAT_STATES\nstate.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a83454619bfc0a993ed2a06fc0a6f16de6506a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a83454619bfc0a993ed2a06fc0a6f16de6506a"}]},"branch":"refs/heads/master"},"c09844bd682a3e6ab05c917703ea532bfc393ac9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-12-01 03:21:43.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/10"}}},"commit":{"parents":[{"commit":"adfbee107c0fee591799b703de2e9e3abef45277","subject":"sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adfbee107c0fee591799b703de2e9e3abef45277"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-12-01 03:21:06.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nAs we don\u0027t want to update the OSR stats and the meta_timestmap, also\nmove tombstone updates to only happen when in a SHARD_UPDATE_STAT_STATES\nstate.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c09844bd682a3e6ab05c917703ea532bfc393ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c09844bd682a3e6ab05c917703ea532bfc393ac9"}]},"branch":"refs/heads/master"},"70e57f8059c28b43934323bde1f44f8685ddda11":{"kind":"REWORK","_number":11,"created":"2022-12-06 06:17:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/11"}}},"commit":{"parents":[{"commit":"001d931e6a77e7ee9167562a51f8c87ed76e559a","subject":"sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/001d931e6a77e7ee9167562a51f8c87ed76e559a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-12-06 06:16:21.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nAs we don\u0027t want to update the OSR stats and the meta_timestmap, also\nmove tombstone updates to only happen when in a SHARD_UPDATE_STAT_STATES\nstate.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70e57f8059c28b43934323bde1f44f8685ddda11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70e57f8059c28b43934323bde1f44f8685ddda11"}]},"branch":"refs/heads/master"},"1a47dd863c3857803779b4f030bb6708f511e766":{"kind":"REWORK","_number":12,"created":"2022-12-09 02:35:19.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/12"}}},"commit":{"parents":[{"commit":"001d931e6a77e7ee9167562a51f8c87ed76e559a","subject":"sharder: update own_sr stats explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/001d931e6a77e7ee9167562a51f8c87ed76e559a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-12-09 02:00:52.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container\ncan use these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it holds\nenough information for it\u0027s namespace, so before this the number it returns\nis incorrect.\nFurther, when we find and create a shard, it starts out with the number\nof objects, at the time, that are expected to go into them. This is\nbetter answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nThis also has the advantage of a cleaving shard to have an idea of the\nnumber of objects expected to be cleaved into it, at least from the\nleader who split the namespace is concerned.\n\nThis could be used to allow a small replica to know if it has an\nundersized portion of the required namespace and so wait until the other\nreplicas to cleave before moving on.\n\nAs we don\u0027t want to update the OSR stats and the meta_timestmap, also\nmove tombstone updates to only happen when in a SHARD_UPDATE_STAT_STATES\nstate.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a47dd863c3857803779b4f030bb6708f511e766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a47dd863c3857803779b4f030bb6708f511e766"}]},"branch":"refs/heads/master"},"ece4b04e8215f80b55c367abc523290b503efc76":{"kind":"REWORK","_number":13,"created":"2022-12-12 06:42:26.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/837811/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/837811/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/837811/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/837811/13"}}},"commit":{"parents":[{"commit":"9063ea0ac72a4be3c3a4e099423458f25a9783db","subject":"Merge \"sharder: update own_sr stats explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9063ea0ac72a4be3c3a4e099423458f25a9783db"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-08-08 06:25:55.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-12-12 06:34:53.000000000","tz":660},"subject":"Sharding: No stat updates before CLEAVED state","message":"Sharding: No stat updates before CLEAVED state\n\nOnce a shard container has been created as part of the sharder cycle it\npulls the shards own_shard_range, updates the object_count and\nbytes_used and pushes this to the root container. The root container can\nuse these to display the current container stats.\n\nHowever, it is not until a shard gets to the CLEAVED state, that it\nholds enough information for it\u0027s namespace, so before this the number\nit returns is incorrect.  Further, when we find and create a shard, it\nstarts out with the number of objects, at the time, that are expected to\ngo into them. This is better answer then, say, nothing.\n\nSo it\u0027s better for the shard to send it\u0027s current own_shard_range but\ndon\u0027t update the stats until it can be authoritive of that answer.\nThis patch adds a new SHARD_UPDATE_STAT_STATES that track what\nShardRange states a shard needs to be in in order to be responsible,\ncurrent definition is:\n\n    SHARD_UPDATE_STAT_STATES \u003d [ShardRange.CLEAVED, ShardRange.ACTIVE,\n                                ShardRange.SHARDING, ShardRange.SHARDED,\n                                ShardRange.SHRINKING, ShardRange.SHRUNK]\n\nAs we don\u0027t want to update the OSR stats and the meta_timestmap, also\nmove tombstone updates to only happen when in a SHARD_UPDATE_STAT_STATES\nstate.\n\nChange-Id: I838dbba3c791fffa6a36ffdcf73eceeaff718373\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ece4b04e8215f80b55c367abc523290b503efc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ece4b04e8215f80b55c367abc523290b503efc76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
