)]}'
{"id":"openstack%2Fswift~838563","triplet_id":"openstack%2Fswift~master~I5bebca91821552d7df1bc7fa479b6593ff433925","project":"openstack/swift","branch":"master","topic":"p-s3-inventory","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-06-02 04:47:29.000000000","reason":"Tim Burke replied on the change"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-06-02 07:53:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5bebca91821552d7df1bc7fa479b6593ff433925","subject":"s3api tests: allow AWS credential file loading","status":"MERGED","created":"2022-04-19 17:30:20.000000000","updated":"2022-06-02 07:55:02.000000000","submitted":"2022-06-02 07:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"838563-p-s3-inventory","meta_rev_id":"23e431872861baf9855fabeabd398127a1629f3e","_number":838563,"virtual_id_number":838563,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 07:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2022-06-02 07:55:02.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 04:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-02 04:47:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 18:32:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 20:02:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-05-31 12:14:28.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"a3b158748db90aeaebc90b9be5053c96cf57e6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-19 17:30:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d9a7f719577a50d50d759de469c8662bbf67595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:32:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0dd5ca4d2f4b70bc31183f351b2525 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdd4d5a59728487a9646612f83abbb0c : SUCCESS in 56m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2b14d9814885422c9c80bbc713e5766c : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2b3de4d6d2b14797b94c86774d6be871 : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f6b93fe8d7564de089a59d33011d8515 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4534c19b12624b07ac8b7729e6d33d93 : SUCCESS in 7m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/30523ed9166e425c96a9bf9a16d80abd : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c119665978e4481976eadb1ec5f9285 : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d235c7a677b4b56a0ea86e9ba270d22 : SUCCESS in 19m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd27e3063594f08abefad3aa32ce647 : SUCCESS in 18m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e3736ae55594b71a1827073ba3c412c : SUCCESS in 20m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8287756db9074effa5a146866cfb5208 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7aeee14815cc4779b323f587a528304b : SUCCESS in 20m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/589ffcab493a45ada14d961fd4faaedb : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/99b2467bd6be48dbbc6b4a11fe50b1ef : SUCCESS in 19m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/459e5d338d48468a8fc6e172d80c2c89 : SUCCESS in 16m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b5716299a53493aa2d2e9eb39b5d0a3 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d634c63a0e784a50aea3186bda28498e : SUCCESS in 18m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a151683a722b4e83a8885035819aad93 : FAILURE in 52m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/83e25440fc804a6395b59d5a62c08cf4 : SUCCESS in 55m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/495433d3c27645ed9f11a562fbeda3d6 : SUCCESS in 40m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7092cffe714c4213b6c089db04e6747b : SUCCESS in 42m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4db9585ec9144e7fa15504c374f12976 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7777bfa93a24c2d86c92d910e93a8cf : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f08bfb874f84d2baee92f5eb741d9ed : SUCCESS in 18m 38s","accounts_in_message":[],"_revision_number":1},{"id":"01233f4f90ad35d5ab04685786336944ac54bf66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 22:52:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83b036b799fb479d9301f779595ec30f : SUCCESS in 1h 23m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2e4edb668cb41b7a2fe5313fd386717 : SUCCESS in 45m 11s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/93f561ed601640bd88529f7d63ce1d0f : TIMED_OUT in 3h 01m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74097e90f81340b795a6f9357f482d04 : SUCCESS in 1h 14m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/40722b5d8df94f4e9f7da89b38871bf9 : SUCCESS in 1h 27m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b0cd93bb4088a914a350aeb9591bd18736cd55ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-26 02:51:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2fcf6d752576c2a4f525b45a1ecceaceb7cb2b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 04:53:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7984c1b3d634bbab4a5cc99087a7d55 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b7bb32aa9db4dc2b61a9f381e96439f : SUCCESS in 48m 56s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/768b917a5aa54394af66f0b80ea6c85a : SUCCESS in 26m 00s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e7b317b5af10423182b3e16f70839c5a : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/582a32b777814752ab22a7f346349c19 : SUCCESS in 25m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/099394deedd341c1bcc419128db67f81 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/407358633c5c4038b5e5878f640385b4 : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5574391114a846aba95a704d4b1430d5 : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8fed6a9d1471407bba2756ca19d8f1bd : SUCCESS in 13m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f6356dc3952048c792ab0512fc916a15 : SUCCESS in 15m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d7d03e1ee9940ba984c8bc6d1c96140 : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed38b30083dc4ec09e7598a16b3d194c : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fa3a28f088df47579b3c15aac8f92739 : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8bfc23f7960944e8ade5de2aa96af614 : SUCCESS in 23m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db2df93e95de450e9fddae1a99ca90d7 : SUCCESS in 22m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cda95e47d46e42309110807aae6e1483 : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08377589f8c5417997ef7320dbb3a108 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ad0e2aff3d03403b8aee06f60978652d : SUCCESS in 16m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c83f50cedb144769c1823f7d95c4dfe : SUCCESS in 53m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/95c6e711875e4a5e9a33099e562c9922 : SUCCESS in 1h 03m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66a3d1e2d5cd4264b06b14180c0565f3 : SUCCESS in 44m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80d7f330add34a348bf2a546ef584747 : SUCCESS in 44m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6949a5f94efe4e6199ce55848c87a454 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cd2747cab846b5a734070bc335ab60 : SUCCESS in 8m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f49a1e3edb1480889161bd15d32c15a : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9015e79afe4c20d68ce2ba921374fe6d0a48828f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 07:46:32.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e89f8ae4ec14544bbf8123172eb225a : SUCCESS in 53m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86df2ea1cf0e469397d39b1b9e62fc24 : SUCCESS in 52m 47s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/f84fcccdbdaa4e1fb5cf43b9794762b7 : TIMED_OUT in 3h 01m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfb5eacc92ae4d25a811c065d07c97c4 : SUCCESS in 44m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f6762076a2d4880bc8a1b81d75432a7 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8302d2b867a7d22908e5e6cccee2e6dd3838e0ee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-27 20:02:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e3f6962d27871d99b65f4217ec08c35a8ec8ca43","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-05-31 12:14:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"163c8a05adc05289df7046df5b3a77ab27c9c7b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-01 22:11:42.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c0cc991f8a04ba94e2679801cd6536313fc921ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-02 04:46:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4503fe7b5b1b493e88f515f91ee7035dd05c3143","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-02 04:47:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"44432bcb8ef413bf6ce378db45168bfb7fd4cc4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:37:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/475cccc816b147debdb78348e3662859 : SUCCESS in 38m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34dd7486f129411d8dbf4d9b0349f197 : SUCCESS in 38m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b17decb59e0344debffb2da7c40165a3 : SUCCESS in 45m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eae6cfe97b3a4b77a99e20b5f666e9a3 : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ae2fa6e200d0ce5f8db0caa264e8859b12b32ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:51:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30982300dca945209d92b9abb1c1aebe : SUCCESS in 7m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5660a5f1f0af48909e441b128d4399d0 : SUCCESS in 42m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/efdc44aea1d744e19bf52374187c95dd : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4b1965e514cd47b0b782f9ee5ee3dac8 : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/070a8d4bd80342b0bc77401789e3a306 : SUCCESS in 23m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23595ec0f57c4df6835ad7c2e357ba25 : SUCCESS in 6m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e2c5586c6d50407794927b2693b02778 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f5e1196621a470188a5049806283789 : SUCCESS in 13m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf49e7e74dc456f80fde2fdebc22b1d : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8268ee07443847ebbcae3e2b42391e4c : SUCCESS in 12m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/66b2059e1d8545da9bb90d782950decc : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8757b2d82254400cb39fb29ed433ed55 : SUCCESS in 20m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a756a335bdf74ee78da121eac44a25d1 : SUCCESS in 15m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6b76fadae9e0402fa64249805d8cfc1d : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/408263e62f0f48b593028acf1e553894 : SUCCESS in 11m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/58b90bc508e8455f9b4bb3973ffdeff6 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d98f1ebf8f34c63b86c8c29e6febfd4 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2c64b10fc334e218b6105d0cb3317b8 : SUCCESS in 12m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/09a34f5ee8bc4f81abd905de84b0796a : SUCCESS in 1h 03m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/195f9fe688df4166b2a93902f7c178fe : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b82a9b7048e40b5b263d0731f61388d : SUCCESS in 36m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e151003f699740419748f7029b7b7357 : SUCCESS in 48m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/291d919ff6734bc1be90c23f840a3626 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b573473eb4a461f97f864dd1dc73c0e : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65d31732db7f43938a4d4a660fa58e9b : SUCCESS in 25m 31s","accounts_in_message":[],"_revision_number":3},{"id":"acc69ca945354c89af78e6a40303acbf44e4a214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:52:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f050af5b37205cfdc2693b63fce325e91de40f7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 07:53:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3466e4264ba44eab8e384aba9f3b59c9 : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3516eb23a45846278bd250083a0abb51 : SUCCESS in 50m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/588bedc878af41a2a5fa6539e3f389a3 : SUCCESS in 7m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/de2fdac2d82142cfb68956c4b084fbdd : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d309957b47674c71bbab1be1ba333b11 : SUCCESS in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41d5f9ef1d474342a12a2abc53c5f0a9 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a39289a5d61549f4b52bab7f22dd028a : SUCCESS in 20m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bbac80a1021442d97d0548a3a582240 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67656752bc454554a23aab07a5458a68 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c38243f764c14085b51e715bf1e456d8 : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3ebd56135f14419b9ff30579dcb4e43b : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ca136c0ad564f529e3e45cf4559beb3 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54c4c8633d8b46c19cb73d44c3e83ef0 : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f66341823e4247fb85e6d5f8803ba93a : SUCCESS in 6m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a7282b6ec3894898b342d640c4e87da0 : SUCCESS in 45m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/673fe061db574ec6bbcd152ae5b2dc49 : SUCCESS in 49m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/caf5795c2be841dab032905f56b5f870 : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6adacbf74637414397f0d5b0db8cf26e : SUCCESS in 33m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e91f79436f2542f2a91be19d8f5450eb : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea4b367ad16469c85934d5945d38cde : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1899758f0dca4d9ca45631d64f7fb0c8 : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2ea0f6917a014712aaaf549f45bbebc2a7f0ad48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 07:53:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"eda98aabd89330625098e3ddf839770a1fdd180d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 07:55:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b4260d7273441bbe137ce4561942da : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ca9582814e534332b1fd3fbfb472d048 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ad559a854f584f57abce17c7863afc7f : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5d9f1f009c96e3049941d804b848f3166caac747","revisions":{"c857346f459393083f1c26ca64454b3c1dfc2280":{"kind":"REWORK","_number":1,"created":"2022-04-19 17:30:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/838563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/838563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/838563/1"}}},"commit":{"parents":[{"commit":"bb220f6f41487ca70bb0a96038782599f892f94e","subject":"Doc: Update links in associated projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb220f6f41487ca70bb0a96038782599f892f94e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-19 13:03:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-19 17:26:15.000000000","tz":60},"subject":"s3api tests: allow AWS credential file loading","message":"s3api tests: allow AWS credential file loading\n\nWhen switching the s3api cross-compatibility tests\u0027 target between a\nSwift endpoint and an S3 endpoint, allow specifying an AWS CLI style\ncredentials file as an alternative to editing the swift \u0027test.conf\u0027\nfile.\n\nChange-Id: I5bebca91821552d7df1bc7fa479b6593ff433925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c857346f459393083f1c26ca64454b3c1dfc2280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c857346f459393083f1c26ca64454b3c1dfc2280"}]},"branch":"refs/heads/master"},"d758c897bcdbb23cde73b598fca2a9a8bd44898a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-26 02:51:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/838563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/838563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/838563/2"}}},"commit":{"parents":[{"commit":"089608562e2d5d63c57cc96146da34290fe17a7b","subject":"WIP: s3api: Make multi-deletes async","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/089608562e2d5d63c57cc96146da34290fe17a7b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-19 13:03:39.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-26 02:49:54.000000000","tz":600},"subject":"s3api tests: allow AWS credential file loading","message":"s3api tests: allow AWS credential file loading\n\nWhen switching the s3api cross-compatibility tests\u0027 target between a\nSwift endpoint and an S3 endpoint, allow specifying an AWS CLI style\ncredentials file as an alternative to editing the swift \u0027test.conf\u0027\nfile.\n\nChange-Id: I5bebca91821552d7df1bc7fa479b6593ff433925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d758c897bcdbb23cde73b598fca2a9a8bd44898a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d758c897bcdbb23cde73b598fca2a9a8bd44898a"}]},"branch":"refs/heads/master"},"5d9f1f009c96e3049941d804b848f3166caac747":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-02 04:46:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/63/838563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/838563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/838563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/838563/3"}}},"commit":{"parents":[{"commit":"0c8a0886649f86570f935abce9feec9ab2261fcd","subject":"Merge \"pytest: explicitly set system logger to DEBUG\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8a0886649f86570f935abce9feec9ab2261fcd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-19 13:03:39.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-02 04:46:41.000000000","tz":-420},"subject":"s3api tests: allow AWS credential file loading","message":"s3api tests: allow AWS credential file loading\n\nWhen switching the s3api cross-compatibility tests\u0027 target between a\nSwift endpoint and an S3 endpoint, allow specifying an AWS CLI style\ncredentials file as an alternative to editing the swift \u0027test.conf\u0027\nfile.\n\nChange-Id: I5bebca91821552d7df1bc7fa479b6593ff433925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d9f1f009c96e3049941d804b848f3166caac747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d9f1f009c96e3049941d804b848f3166caac747"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
