)]}'
{"id":"openstack%2Fswift~838578","triplet_id":"openstack%2Fswift~stable%2Fvictoria~Ibe514a7ab22d475517b1efc50de676f47d741a4c","project":"openstack/swift","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-20 05:53:05.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Ibe514a7ab22d475517b1efc50de676f47d741a4c","subject":"s3api: Use constant-time string comparisons in check_signature","status":"MERGED","created":"2022-04-19 18:52:52.000000000","updated":"2022-04-20 17:22:04.000000000","submitted":"2022-04-20 17:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838578","meta_rev_id":"341c743f4c08fd4f5e49322c81b67f38c1e7bd07","_number":838578,"virtual_id_number":838578,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 17:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-20 14:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-20 14:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 19:54:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4c566b87fe52a07032419342343fea72646ea9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 18:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4076e950349e238e30b603072c8f506cb5780777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 19:54:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c13f37379954d088968edee4e35e1c1 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b906090745214439b9f7f40e2ed144ef : FAILURE in 11m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e1b3ff46cf6407cb20c1436a87a6c57 : SUCCESS in 53m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7ef6767ec374f7c9fa72d1e7713cae5 : SUCCESS in 39m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9334b06b8da04a01bc6ebe8ce52da7d6 : FAILURE in 7m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/337950ea4e094590bc5f4c7d63dccd5b : FAILURE in 4m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bad6ee3f46ed4c06990681e64ff21d9f : FAILURE in 14m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/184ae59eefcf4295a81d3037a4a522a5 : FAILURE in 11m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/96e26f59f53840b2b4ba7e7f944ae5d1 : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/edaf5ea2e35640b5810e3b71d8f7d461 : FAILURE in 11m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77b95c68610e4470b447250ec55b1852 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bccffb8e39fa4e2f916d535a1f430413 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fe82a60d54a4561ae4d9307ed8f9e36 : SUCCESS in 17m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/82959f9cc2eb43e69db3eaea0b2e00cf : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/21064c14efbe4107bd056d9e973d732d : SUCCESS in 18m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5ffe06509d844eecbd19e6d7aa5c8475 : SUCCESS in 16m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61e54d6589e64fa393a20bba2716ae18 : FAILURE in 5m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/45c1898e4b6446e0b8cd50a03d839b3e : SUCCESS in 38m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93bc9dd7e5d941198e89435939e043bd : SUCCESS in 41m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92e725fde8a54e57b739a8ccf2162909 : SUCCESS in 29m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3a529c4e8844e8b65595e30179a9f0 : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b03131deaf4c469f5c1e5c65bff446 : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c25cf3944e44fc28804e21d623baac7 : FAILURE in 8m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b7c24e8d225426fac84528f05ad0ea3 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9cd835e2573bee5903be9ccdc62e816ac36a3ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 04:28:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a948e98fa752738365939d8cd89fbfc351be36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 05:49:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be1266ad0ff8412faf0de71a8e9833e9 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c7edc6a0a64e41d4b537934998cbedd9 : FAILURE in 9m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8b4084bd0b24c97aa06f12c1ddc5bfb : SUCCESS in 36m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f07c172520184c12bccb82e915303511 : SUCCESS in 45m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b47207587663461fb743b5d93494fd78 : FAILURE in 6m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e5921fad3114f989d3659b2e5d7971a : FAILURE in 4m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f637bb16ae84809b2d129edcb05aca7 : SUCCESS in 13m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6471912f3f44bd8aff50d528a5f43cd : SUCCESS in 13m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8dd340ab38a46068f0f9988f9adb271 : SUCCESS in 15m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6a4ba505244e998c6f8fe7819e5566 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e9dc2a986d6499b814157d2b66cb1c2 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f2f346d328424ca0b2c9d4432a62b9d7 : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ec9ffc0199242bc89b180c2b282b1a6 : SUCCESS in 20m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc430c62395745c28f21dbfde3e4bbb4 : SUCCESS in 23m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c1aea725e676438184f56146a0688573 : SUCCESS in 19m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4a9e9a97fb554fafa53dfbd2c05b7133 : SUCCESS in 21m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7c0c69c03dc42dc9416d5c761715f17 : FAILURE in 6m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d36be6c6cc848b0966a9ef880f64e26 : SUCCESS in 44m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e273c678c2064e0a90558cbeb4f59376 : SUCCESS in 36m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/096f5aa7233a4734b8a3aef56f4c8a3b : SUCCESS in 35m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0c051e31c7a4c538d25112c1edec1b9 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32e3a27331c4605a4132e11c9c0e9fc : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d87c9a42adcf411295df856a4670515f : FAILURE in 7m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/016b1f3809f34168940b51214bb5c06b : SUCCESS in 1h 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"bf624a770b4736ed576612e4bf7291f971977aee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 05:53:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64bf10a0102243ce5fc460e1b6dadc7af6d535d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 07:02:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e052850636f74014b191f6137e48a87b : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/a2ae506a4ec44260bddb39cf0d54603f : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eac0ba9c977a40aaa32474397ba48521 : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73ef5e92ed4a4300acb4ee4dc44145b5 : SUCCESS in 1h 01m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8a04f33d179412d9bd63ccfa89517d8 : FAILURE in 7m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c5ea2e5b50f4a889528dcfa0d07f610 : FAILURE in 4m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f45e1b69db094ae8b20d7cb986deb431 : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04d9733a7ca143c6a05c671f14ed5563 : SUCCESS in 13m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ace6f96ccb74a41a81b09ba181ffe50 : SUCCESS in 14m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3bad6a3d0384b6795a294bd9a2834d9 : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d5dd1cbcf8a44988b0c68ae43f97d8d : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a8c7b217ef940c28c9e4d99602d0205 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0da4f5cb6bd544ff841466dc93d55b9d : SUCCESS in 19m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3d56afa25235492fa65036d503a10df2 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1a5bd504a28f464ca714e7903515b2b0 : SUCCESS in 21m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c7bfef3067f249abaf451a20f49e2cf3 : SUCCESS in 17m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9cac3cfa870464f8b8e3e260ba786a7 : FAILURE in 7m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8d48cbd4fa2b454abeb8a395cb81db91 : SUCCESS in 45m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb9f00e1b7a64b4191ff369f5a0bf390 : SUCCESS in 40m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb9c5e0a14f447c5a15ac9e1de6c779d : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8d0c9ca510e41d38e31bbf6c42b8358 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f88b4a6055924ab18297a3bfcc70dc66 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/273f061901f44314b323373b648dbd49 : FAILURE in 8m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/380c8426276141c8aae0ef4c88de4134 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":2},{"id":"71c6277278797647b86611f04a1ef1e2e5e23cee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 14:59:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c695f169c254b7aa90130835fb996c65f05a9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 15:01:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6267b6c9a49d2951a8f7168e97c1b9be5b5b8b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcd0786bbf04a2f8d63d4cea3bf7f41 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1de7f2a8a65b47038b5def14b5d1bf5f : SUCCESS in 1h 04m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a052bcb13374e43bac01f4d1583473a : SUCCESS in 50m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b9421766ade46a891fd42bda8e51c48 : SUCCESS in 58m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b6cc51535f2a4a14b26d44a81bca6c12 : POST_FAILURE in 10m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1a6cb9b6efb4454c8ba5a275ae0d0d9d : POST_FAILURE in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5af0109312d94f6abb4d4a2c360d3138 : SUCCESS in 26m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94660190caa5460b83453369aadf84a5 : SUCCESS in 23m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6280e1aad9694a2ba160bdcbe0f46311 : SUCCESS in 24m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e96c2528bdad4fab9600e3a821f14311 : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8b4cae51afc4c5f9d586bd2f8891999 : SUCCESS in 37m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11df9f6bd9354a95941ced2139cb64d6 : SUCCESS in 38m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/209f91023e904d9f8125086431fe1d19 : SUCCESS in 34m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bb33621fc28d4afa90b62989917f2c39 : SUCCESS in 36m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e424eb79a6b4c788f211b24213bc51c : SUCCESS in 24m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9a5463577ec84da9aace19fb3d7dcccb : SUCCESS in 1h 04m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bf776ed2cac4742b4ef7bf559323833 : SUCCESS in 53m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/395947ea4dc34cf5b7bfd16f62a40206 : SUCCESS in 56m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d49b85f47ee449deb86ac354d702cee8 : SUCCESS in 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33207b7d6d3c405ca481fff2403e9b26 : SUCCESS in 7m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/464dd2b78ce54f6080ae70c19197c906 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4483c6471053fe70ac92ceac68c7696d0b349b05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:18:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"341c743f4c08fd4f5e49322c81b67f38c1e7bd07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:22:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e39246b862af4d2fa4cf9e75f21b7f9c : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cc52e961f1db49c3b841fe23ac0939ce : FAILURE in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0d79780d784f4736a5ddd0bfb0d8bfae : FAILURE in 56s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ca7bdf93d471435278e0c1a3b1486f18fd67c0c","revisions":{"d38889860fd98b91151bf8339f718d05b5c04d06":{"kind":"REWORK","_number":1,"created":"2022-04-19 18:52:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/838578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/838578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/838578/1"}}},"commit":{"parents":[{"commit":"fe99a19129ac2ce9087debca3b6ab5b1ffda70c0","subject":"unit test S3 SigV4Request X-Amz-Content-SHA256 handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe99a19129ac2ce9087debca3b6ab5b1ffda70c0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:31:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 18:52:35.000000000","tz":-420},"subject":"s3api: Use constant-time string comparisons in check_signature","message":"s3api: Use constant-time string comparisons in check_signature\n\nChange-Id: Ibe514a7ab22d475517b1efc50de676f47d741a4c\n(cherry picked from commit 6142ce88cc71037ba0cd23113eb6082fa91346ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d38889860fd98b91151bf8339f718d05b5c04d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d38889860fd98b91151bf8339f718d05b5c04d06"}]},"branch":"refs/heads/stable/victoria"},"4ca7bdf93d471435278e0c1a3b1486f18fd67c0c":{"kind":"REWORK","_number":2,"created":"2022-04-20 04:28:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/838578/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/838578/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/838578/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/838578/2"}}},"commit":{"parents":[{"commit":"f1c110be5ea58230da6974fa3ea6a1d523df1def","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c110be5ea58230da6974fa3ea6a1d523df1def"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:31:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-20 04:28:37.000000000","tz":-420},"subject":"s3api: Use constant-time string comparisons in check_signature","message":"s3api: Use constant-time string comparisons in check_signature\n\nChange-Id: Ibe514a7ab22d475517b1efc50de676f47d741a4c\n(cherry picked from commit 6142ce88cc71037ba0cd23113eb6082fa91346ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ca7bdf93d471435278e0c1a3b1486f18fd67c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ca7bdf93d471435278e0c1a3b1486f18fd67c0c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
