)]}'
{"id":"openstack%2Fswift~838582","triplet_id":"openstack%2Fswift~master~I7eed34ab19a97b2c561e4aac8c95a8a5e1689266","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-20 00:45:47.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I7eed34ab19a97b2c561e4aac8c95a8a5e1689266","subject":"ceph tests: Register output/ceph-s3-summary.log as a job output","status":"MERGED","created":"2022-04-19 19:16:47.000000000","updated":"2022-04-20 04:08:33.000000000","submitted":"2022-04-20 04:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838582","meta_rev_id":"f1731cdd0cc89d245ca3e815c47c2272b61ed683","_number":838582,"virtual_id_number":838582,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 04:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-20 02:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-20 02:47:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 20:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af9010eb7846059b7c60476229be2d09a5ea59f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 19:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8abd504b57a558822a8fa914fbb9a296e43be065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 20:30:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8865dc5bcdf94fbf907b1bc18e0f780e : FAILURE in 4m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/66c911a580e846d18ba2f4371772da60 : SUCCESS in 1h 12m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61de633dd7cc49879915d54f09af674f : SUCCESS in 1h 01m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25c8e857ba3949ba8ae82030d466c08f : SUCCESS in 55m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/13e486e6223e41e38ce33006cdffb505 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d678db4c7e4a4a91b875682fd0912d36 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a0d56a18673a4260adb2a5b83f4b94ae : SUCCESS in 23m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4ce5e1bd3f44a5a86e6710b002c1998 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dcec1ffc6ada4605af41b7244e1ce323 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c51643165b4f4684ad0f0fe45d9048bf : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1a2d3d11731429e8eb41fc5578a9300 : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/568cbd116f184d638dfd7d9b1c05a7e4 : SUCCESS in 18m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07eef07840e64b27bcd0a10e8fd08ef9 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9588187276a6417f83d9570e727c3c0d : SUCCESS in 21m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc7f6fd5632d40b68e4a0fec3f0d095c : SUCCESS in 20m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2064e7f715c04082be17d8d9ccdc53dd : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6549af12e67422b8e2e472719a61616 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/079175f717e34490a63a561149bb1a45 : SUCCESS in 20m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f660a48a3bfc4e10b34ad2a66f2e4e3b : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9ad4497f2f27477881998ea345c0a207 : SUCCESS in 18m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea05983f5eaa48ce9b6523bc8fe9205d : SUCCESS in 50m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9db027e974684301a88a7c06d9f44d54 : SUCCESS in 53m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59ad9146a9ff43ed8dc5cd86aebffa71 : SUCCESS in 34m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c6d376b8712406f9297cb5c13caf5a8 : SUCCESS in 49m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e51d5acede3436bba58291a45ddc67a : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5342df7c054446bb6735e6965ac0ff7 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c13701caa7384109ab46d1a547e9f474 : SUCCESS in 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb417224498845f7bb35ef84ad5b2bf3 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4ea22a66ee15e838dd23042e362361ae26c783d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 00:33:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/072d93ddc7474c6b88ce946f790a5ec9 : SUCCESS in 53m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4a293ddb38d4b0191e6badde853726a : SUCCESS in 47m 06s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/30a3185f79894ad3970212cb278d4700 : TIMED_OUT in 3h 01m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa6113ec8adc4b7483b45788083c9fbf : SUCCESS in 55m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2e28ead46d8413fbd9f68185a5aed00 : SUCCESS in 1h 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"38280c313354d89327784d6dfd1e3d8dfd0f534e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 00:45:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3733a56be86a1512a23ebc35a6b04b363780e5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 01:55:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa452c27e04438089e5a2b22572346e : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/34b12e8ca99b4125ab5707f4c6baf000 : SUCCESS in 1h 04m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b16f329e2fd42649103b48f1e065bb0 : SUCCESS in 57m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/673de11b889345deba7b8e3a1069e588 : SUCCESS in 53m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c9937c30177c40d891980fc638d1d924 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f6ab133fcf2b46b3bfa9828c9bebe936 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fc6126d27f144b1fa22943308f1f7fd0 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e61ca110fc5e40b499f17a1974081103 : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a1620186c6749739b7b19d9b5d1eddd : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc41211a23a44678b144787b91550197 : SUCCESS in 21m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47be9b8d6a0147fb921851ac9a85aee7 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2436194afa54499ba47b725ed8c19d3 : SUCCESS in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8e0dd95a1d049a3a951a4c8de343415 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54ccd89952d64f239d22fa4ae365350f : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d9b0bae27564369915c5006a16f6651 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e33d772d43f94bbba6811d08dcfd097e : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be5b1ef026f34c968aa0203609d8cd95 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/593599241ece49479a83ae3aa32555c4 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2525507c5932457589a5b3207ceeee17 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f93ff4e8c2e94bf7963488665220a22e : SUCCESS in 17m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b50138bbd2f4c97b1827afb31a78a3e : SUCCESS in 47m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec8fbc54bb3e46b0ab67c211b5f5b8c3 : SUCCESS in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2af0524ded264c6c8f4b0b3b2938a9d9 : SUCCESS in 44m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d2214fda8584e9a89dfa1fddf462ab0 : SUCCESS in 42m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4b567e4aec2486aae9a2803daa58c9c : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8c2f69b2c24e6ab9dabc887f027bbb : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc04df99712c4573b6e4da62114df812 : SUCCESS in 21m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e22cf8da0b6493fa9ad02a117f736da : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e3c982b47e1f572366cda55646e6f7e69fc2c94d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 02:47:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21326ad2ac267de5018058f657408d54540eae07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 02:48:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ff5446558f891ef515e131e56d8a5fd51aaadc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 03:53:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d48c6295b4d64beb9d3540390059e07a : SUCCESS in 54m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea0449b97291488ea6f758aeee1e347d : SUCCESS in 46m 09s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c3ab7d1ac5eb4e44880ae439be0aac4a : TIMED_OUT in 3h 01m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/26bfda4ed4b548d992ffd0ed55274ab0 : SUCCESS in 55m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/621cb0625a64466aa17b9f77711e80a6 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e15fc81c9b86a474736337a11dac903eb3d16d18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 04:07:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50490b8e2dd44aee8b51002e01bc9569 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b1ec32523535422cb39489bc6536d93e : SUCCESS in 1h 04m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31fe54d05b964fdbaeac60ca70564ef8 : SUCCESS in 55m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3726d9d731e743cdbd74692146c040cd : SUCCESS in 53m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1c3bd230fd3a412d99f623fb72f1cd76 : SUCCESS in 10m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d02b48db0eb4463caa85574796fc3420 : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f53bd3db015a49f385fbf3da9ad4bec8 : SUCCESS in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/baaa3a321c3e4e2c9dca645381a3296c : SUCCESS in 14m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/236999a1967c4035baed70446c300e07 : SUCCESS in 14m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e816b80f16b44b129cacc849e68bffbb : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/12c0ccb0b6214b6cade4f85de38c081b : SUCCESS in 17m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5aa5a450c5b34ca79b3ba5e99f359ff2 : SUCCESS in 20m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0c93af7f5a934b62854fd872db5f30ed : SUCCESS in 21m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e9ee2f3b0ad64e21ae82983f2acc6fe4 : SUCCESS in 20m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/067f12e48bcc435b923b0dd49a887af8 : SUCCESS in 28m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d37cf121c6d944dc9bbdb99415752427 : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f99ca86cce340dc9565e43c9bad6d9e : SUCCESS in 52m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/34b39a605a3b4085afb67cce3427849d : SUCCESS in 59m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f3a914b536b4c2890742d6048b94e8e : SUCCESS in 50m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/070aa3fa746d4234ac3a18adc56703df : SUCCESS in 46m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0818f296a13e4b6e92b677b7a5425c85 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a2c70362d94ab1bfd6e47477262780 : SUCCESS in 7m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c06f6bb957514b9298c216721fffcd95 : SUCCESS in 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/476a44870ff245b9a717f6e0ed3fbf7c : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":1},{"id":"bf529f414c63b5c83c87dec79123285da2301299","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 04:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f1731cdd0cc89d245ca3e815c47c2272b61ed683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 04:08:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41cd81a05af43a395523078bedf16a0 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0bcfcd1d2ecf4a39aafcb0878a4816ca : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/63bb087789014f79bae3a7b62f5dc5cc : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a","revisions":{"0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a":{"kind":"REWORK","_number":1,"created":"2022-04-19 19:16:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/838582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/838582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/838582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/838582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/838582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/838582/1"}}},"commit":{"parents":[{"commit":"bb220f6f41487ca70bb0a96038782599f892f94e","subject":"Doc: Update links in associated projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb220f6f41487ca70bb0a96038782599f892f94e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 19:16:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 19:16:33.000000000","tz":-420},"subject":"ceph tests: Register output/ceph-s3-summary.log as a job output","message":"ceph tests: Register output/ceph-s3-summary.log as a job output\n\nChange-Id: I7eed34ab19a97b2c561e4aac8c95a8a5e1689266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
