)]}'
{"id":"openstack%2Fswift~838587","triplet_id":"openstack%2Fswift~stable%2Fvictoria~Ia5857c70e51d8af3e42ecaced95525be578db127","project":"openstack/swift","branch":"stable/victoria","hashtags":[],"change_id":"Ia5857c70e51d8af3e42ecaced95525be578db127","subject":"Quote paths before sending them to swob.Request.blank","status":"MERGED","created":"2022-04-19 20:07:40.000000000","updated":"2022-04-19 22:25:04.000000000","submitted":"2022-04-19 22:23:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838587","meta_rev_id":"402cc11f3f72f3a176a0d87572ea54db00f33b2e","_number":838587,"virtual_id_number":838587,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 22:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-19 21:25:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-19 21:25:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 21:14:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29045dc6057d098c72e820a91b43aaa40020e733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 20:07:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6e8959273040bb6f18efbde7ce895df63db85c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 21:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc50207d3da6466ea27161583a671172 : SUCCESS in 8m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dadf3b2b40d4b5789cdb9f972d08dae : SUCCESS in 54m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f7ec6fdfb1e4c71b6a5ba45169c8a0d : FAILURE in 6m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b30445ad3f8446c685cef52e26cdd61f : FAILURE in 4m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e9481e4e1e04ccb928ceb12040c6b21 : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f5c41c4533b495484f87c82367f84b1 : SUCCESS in 18m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfac491d3020427ab21cb3d6ae17697f : SUCCESS in 19m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/397ed08ed2c64383ad1a67176bc3a085 : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ec43797a80642feb412d8d0986aedbf : SUCCESS in 26m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bc90215daa824eaa8c9dcfe2fcd53ffe : SUCCESS in 15m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8521a0af9154466add44b5234294aa1 : SUCCESS in 16m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a67710e1ace74ed3b7f7ca96d36deac6 : SUCCESS in 25m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/47551fb28463415eac0576e0cbf12b87 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c88d4c474dc47db895df0aa422eeebb : SUCCESS in 27m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43e4d9a6c22943c3af47dd3708f8c68b : SUCCESS in 34m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2adbd4ea585e414fb771745d2c074a70 : SUCCESS in 1h 01m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/493c049c0a684fd881aff3a05a2c1033 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a7a4dbf77c4fdeac4017e82d353a81 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a82acb98279404789cc14d4c19eb0f6 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6595f4228623812385827770ed9a4138bacb7785","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 21:25:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"003095bd527c524469dec4d3a4556b9cad01dac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 21:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a66e0006a2c482be2abb85be2496dd9fdf4eee6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 22:23:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4843123f0a2340cf83939385fb935245 : SUCCESS in 8m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b732d96ff10a420a944e104535e905f3 : SUCCESS in 56m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5c5cf639f80645b7a118fa67f68683cf : POST_FAILURE in 8m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/826a298322b14036bb34c85068590087 : POST_FAILURE in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95e83b689251460296dfd2955460b01b : SUCCESS in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51116bdc5a79437585b634dad52e54b0 : SUCCESS in 21m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e914049abeac408893b64ecf80b83ec5 : SUCCESS in 19m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4079e89f61e54daaae183adbe666ef1a : SUCCESS in 18m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/68464de499bd490488ae120106d3bb55 : SUCCESS in 29m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40b89233497c40a382bf88fe2c6cbb44 : SUCCESS in 27m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/48938e7f824145bdace42a8f04266444 : SUCCESS in 21m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bf4ad257a8e144a9a1f93cf480b943e4 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f32254a3881d4ff2a8c6fd5d7f0cf270 : SUCCESS in 16m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4d0adb6498b4240aaaf2ec5650c32af : SUCCESS in 44m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4a8d88543904edf962c4e766ee768d6 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15b583fb7afb4983aa2113868697a785 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f5ee4ddbdf4c1596915b41f1bc77f0 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"31ed0a81f56dfc312f02aec23bcb460c1281ad92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 22:23:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"402cc11f3f72f3a176a0d87572ea54db00f33b2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 22:25:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dc1de25fab4f8498a257b8f5f138dc : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5bd42c2d15bb4ef9ab511bff8fa913f7 : FAILURE in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/ad75e3c6840142d0869a19bbf9025410 : FAILURE in 52s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1c110be5ea58230da6974fa3ea6a1d523df1def","revisions":{"f1c110be5ea58230da6974fa3ea6a1d523df1def":{"kind":"REWORK","_number":1,"created":"2022-04-19 20:07:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/838587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/838587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/838587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/838587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/838587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/838587/1"}}},"commit":{"parents":[{"commit":"f53500522ae52df3c07bdcf775d11b9adccc817c","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f53500522ae52df3c07bdcf775d11b9adccc817c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 19:22:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:07:17.000000000","tz":-420},"subject":"Quote paths before sending them to swob.Request.blank","message":"Quote paths before sending them to swob.Request.blank\n\nFollowing the fix for https://bugs.python.org/issue43882, our py39 unit\ntests started failing. This was because swob.Request.blank calls\nstdlib\u0027s urlparse, which now strips out newlines. Since Request.blank\n*also* always unquotes, just make sure we always quote the newlines we\nwant to use while testing.\n\nChange-Id: Ia5857c70e51d8af3e42ecaced95525be578db127\n(cherry picked from commit 2b5853f4196e4e3725d1ab55ae7528c41b180a58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c110be5ea58230da6974fa3ea6a1d523df1def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1c110be5ea58230da6974fa3ea6a1d523df1def"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
