)]}'
{"id":"openstack%2Fswift~838588","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ia5857c70e51d8af3e42ecaced95525be578db127","project":"openstack/swift","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-20 16:23:05.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Ia5857c70e51d8af3e42ecaced95525be578db127","subject":"Quote paths before sending them to swob.Request.blank","status":"MERGED","created":"2022-04-19 20:08:09.000000000","updated":"2022-04-20 18:58:22.000000000","submitted":"2022-04-20 18:56:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838588","meta_rev_id":"000fd439780cedb8833296a124dd81a9df71404c","_number":838588,"virtual_id_number":838588,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 18:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-20 14:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-20 14:47:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 20:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36ac1e855bde281ad90cb7b93a79fd77492662a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 20:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b48e480c108244adbd5961c76876e3053afbb7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 20:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe42f1fbc494052af936bf5a8a412e8 : SUCCESS in 8m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26d38ebabdf9448ba3706f58e949eadb : POST_FAILURE in 22m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e457ed7b69854344bff639134e7b168f : FAILURE in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6812612b3d554ae49ad318144536b25a : FAILURE in 4m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c00ac64e88646fe8f624303f3732e08 : SUCCESS in 12m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d61672e9636a4dffa854db822c389fb7 : SUCCESS in 18m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c470c4ff6b234225a9f39ae1b270ac3b : SUCCESS in 13m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c360c42096bb492783e410daa495234d : SUCCESS in 14m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/464008a20e9b4a13bbfde96545f0eb65 : SUCCESS in 9m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35b32ebdc8ad4b0ba8054b2b40392d13 : SUCCESS in 11m 54s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7d03e036dd7a49cfba65363a523cf81d : SUCCESS in 9m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9acc7148095040f4927eb2929f6e4b56 : SUCCESS in 11m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/5945f8dfef8643c78b36618c8af016e4 : SUCCESS in 14m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ad544fe2320f4dbfa7c287f0f4413239 : SUCCESS in 10m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2b0f29aa98314eebaa9a6980d4fe8ae1 : SUCCESS in 16m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5b6e856e8fd64b3d8a1a2d9b860ca066 : SUCCESS in 12m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/acf235c35e9a4d97970a4f9b772f3492 : SUCCESS in 10m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/008cacba601f449b92eeb97aec681efd : SUCCESS in 19m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58ed40ab276448b19c0679dd0b5de6a3 : RETRY_LIMIT in 17m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fa874c776c234f3ebc1f3acb1549b0a3 : RETRY_LIMIT in 16m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d2ce463037a47a7a59c67a0b6dee7e2 : RETRY_LIMIT in 15m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/950e0259cc1342618a1b0331a3695b1d : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a20911a4bb6442b914c7d09308fbfd9 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52145838754d40dfb2cd01a2b497800f : FAILURE in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd3daab8e28ba3f70dbcb1621c81f68d559d8533","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 21:30:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcf9afa1ad84f210db3135c71e60a78f325fd6f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 06:29:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"525c6fc7c56e85494de5421a5089b60ed2982aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 07:26:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94548a561eb54e279c2ac13d060ae5b8 : SUCCESS in 8m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c173706faad42d6a101c7e2637135ee : SUCCESS in 55m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c1b44a40abc465d8da4dbbda3aedb4f : FAILURE in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f4c138141904f3ea8c8fd7f4a50b70d : FAILURE in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/61ccc8be6f764cf09979e99f54c702b9 : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9b91b08ea5894f2aab793e87d963419f : SUCCESS in 14m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f70675874cd14d3394ea1b7972139200 : SUCCESS in 14m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0741260e564b415bb68928ea3e3a897e : SUCCESS in 19m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cffe08df2f3b4c2b8b54e5a31740c388 : SUCCESS in 9m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f5d900fef6f4405aa8bff73e02be276 : SUCCESS in 12m 49s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/06d34706dd484488bbed7d19657c71b9 : SUCCESS in 10m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1acd1bd45274f8e89d0e2ac8e0994f7 : SUCCESS in 13m 19s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/13698bb7d2f149e3b5145157aad5dde6 : SUCCESS in 14m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9f7461343af24761a0dca0d771e03c3b : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9cb4c28f518641c88af4995047e2d3b7 : SUCCESS in 13m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b8c453b366b54c7dba6386bbb3f242b5 : SUCCESS in 13m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a9a17637c004d6183b3a906146886d0 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/485a8d041e184cdaa2e936d80f6406af : SUCCESS in 14m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d52c45448534c4580517cfd8652af62 : SUCCESS in 36m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/58897639ed9640dfb64287eabf696b07 : SUCCESS in 28m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a838364db97441dbac85db27a329078 : SUCCESS in 32m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8fdf56d1c804a1b9baeef0048292537 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2343ef337605429caec72ceff703a29a : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f90f55b7d474e95a94f90016720084a : FAILURE in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"118a9d53d12518ea4209c2339687bc946e8be0fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 14:47:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e1117e8407bac6ad433ff1e190c08f1fd935dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 14:48:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"baf07e1279b4fd236dd87a2189d604f48f9446ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 16:21:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc41ea8822a4112a34888653e65ee9e : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3440d825249847ec893a1c4530fc21dd : SUCCESS in 54m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7da77aec39384bbe8046afe5887c4d2e : POST_FAILURE in 10m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c5a057956ba14d54a49b95c22f1aea93 : POST_FAILURE in 7m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3bb961862b3466288714d368fbaa28a : SUCCESS in 17m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a8b7437e3654b0bb3dc5416f728e124 : SUCCESS in 24m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55e01f430e584aa2b5e76a5989b1e621 : SUCCESS in 13m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/255dfbcdb4054fef9c6b6c98f2c3eb5b : SUCCESS in 19m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c56d373259d4a3e8129da94b7aa1a1c : FAILURE in 9m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31c35b23ebb34a0d96fab827dacfe2c1 : SUCCESS in 15m 15s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/aa04fcd53249450cb12dc19483546ad3 : SUCCESS in 10m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1cfaa65d9f7d4f1b90ad4c0673605996 : SUCCESS in 11m 23s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ac16bb8c97384d0db791c18a2ff0ae29 : SUCCESS in 14m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d9cfa9c2611743e0b9ffa98b704cb4f5 : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/80f0f93ea5bf499991745d93d7572cd0 : SUCCESS in 12m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c112f17d2ba14c5c8b75e96f7da24b67 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/983c724d5ae942be8ef65eefea4ab826 : SUCCESS in 15m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37851127416a4f75af4a324f6042cac8 : SUCCESS in 31m 43s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/89ebd916d1ec49958ae3a4dfc46aea3f : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bda4caf1b1ea4037ae451acc3f3b2202 : SUCCESS in 32m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48e152c213d04a2a8820e321cf1c341b : SUCCESS in 15m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c4f4a1839f4f5b8e601c305c62cd4f : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2d7655252582f618e201404a464e8a84302554ce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 16:23:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"970f492e088c08e657dbf1689055d5d43d98164c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:24:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0da0b1680954ec38c92a8724974f41a : SUCCESS in 8m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9689c00f8890446fa56dd4e84a314da9 : SUCCESS in 56m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/699dc802bdc0418bb024d6d899ac0ae7 : FAILURE in 7m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5254b383102a479488f6b897b1c45eb9 : FAILURE in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b0a5d913d0b48ea84fd0d9f087e294f : SUCCESS in 14m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39c1b56d3e434de1b419570f8f8b6c8a : SUCCESS in 27m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/331cd29abf9c48208fb2f7579a2f476a : SUCCESS in 13m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/da210961f664415fa95c1f3186601bde : SUCCESS in 12m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6822195364e245498301a195a715d7ee : SUCCESS in 8m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ffa22eb9c91c4741a9799ab0e95d567b : SUCCESS in 10m 38s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f7edadc52b374beaa4df41801080311a : SUCCESS in 8m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8084cc0c6c8c42b5983e2e6b93e56b36 : SUCCESS in 10m 06s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/10ab2a9d7a91419384b4db4b1f883d9f : SUCCESS in 13m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e8a312d8bce240778693465ccbc308c7 : SUCCESS in 10m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d87b2c02e640410d84d195922ce7f4cc : SUCCESS in 12m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f27566cb386d4191801932e1b0e5fff6 : SUCCESS in 10m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a6acd48dd35d49649b3a28909056508a : SUCCESS in 9m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3af406f9afdf4a53bb85d28bc08aa316 : SUCCESS in 15m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ba6dbd4541d46b292b064fe95d1a437 : SUCCESS in 23m 23s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d6fadad501754e42b7e46467a6ef2c0e : SUCCESS in 22m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ecc87e4d6f849abbc9883dbe2188dcd : SUCCESS in 22m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a52a7291c64448ea8b087667cbf90f70 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c2329b4cde453380115931bdcd79f9 : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de6e2d4176fc40549de2cb1e3e644d54 : FAILURE in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f6012430141f1802d27faf754b9384b3f133ea0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:25:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf90d44992c3f3a4c31da3c20aac282a0baef18b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:56:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946be7bbbe7446ed9dd90e1036ef654b : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17e18cd4ce754a60b31e5f0e730078e9 : SUCCESS in 50m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a46c1ddf471745cf81fab2da45ae3190 : POST_FAILURE in 10m 56s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8d51e0dc860a4a9eb209d50a99c512c4 : POST_FAILURE in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b0161561ebd64049961ffa28289868dd : SUCCESS in 21m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b717f8b1a59e4ae2a3da6e8b5bc31216 : SUCCESS in 22m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd8256439a747738cb8c550954e13b1 : SUCCESS in 13m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/969f75c84276498ab90b74149df4c053 : SUCCESS in 21m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/436042ad22e7466b9da021ab48d9cd46 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1d899d0a179e44669fd7fc1c4cf94632 : SUCCESS in 24m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a9b570874dd24519874304dc4b4bb945 : SUCCESS in 9m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1460eea0a15a41018ad681c88d7d64cb : SUCCESS in 10m 52s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/98f6afaf22fe4b68813441f502673d25 : SUCCESS in 13m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20c650facba645f090be3819c281d1bb : SUCCESS in 13m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fc34b3fdf082498c901605adb6cb40d2 : SUCCESS in 10m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ba85b203ac664da99a0a9a63c5042925 : SUCCESS in 11m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d73d4ae932b049c3a09dd6d4c13f23a7 : SUCCESS in 13m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/881eda9af592486388f98ccd676a959d : SUCCESS in 21m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8759bc6da86441f39749ba39e39a6106 : SUCCESS in 22m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/feaa2e2c8b2b44bda1e26fd28215bb12 : SUCCESS in 21m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a85150fd87a047e69fbee89466e8f692 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43dea397ee34588bd9b3e68dabca126 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"54cb6abc0d715e1567db4d4b97fe83e3d6f84d6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:56:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"000fd439780cedb8833296a124dd81a9df71404c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:58:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bf2f7dbb26e41648b8a359036d22b69 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9b8f3e99e77d43cb93444df4d17f8419 : FAILURE in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8184e61789ad4479bdffc34da79ffe92 : FAILURE in 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffcb1ff0016e7dde1d32dedad92091a132d0f88d","revisions":{"ffcb1ff0016e7dde1d32dedad92091a132d0f88d":{"kind":"REWORK","_number":1,"created":"2022-04-19 20:08:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/838588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/838588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/838588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/838588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/838588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/838588/1"}}},"commit":{"parents":[{"commit":"ade6b52f5058fb5f99f77645a468841abb3fc892","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ade6b52f5058fb5f99f77645a468841abb3fc892"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-27 19:22:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:08:02.000000000","tz":-420},"subject":"Quote paths before sending them to swob.Request.blank","message":"Quote paths before sending them to swob.Request.blank\n\nFollowing the fix for https://bugs.python.org/issue43882, our py39 unit\ntests started failing. This was because swob.Request.blank calls\nstdlib\u0027s urlparse, which now strips out newlines. Since Request.blank\n*also* always unquotes, just make sure we always quote the newlines we\nwant to use while testing.\n\nChange-Id: Ia5857c70e51d8af3e42ecaced95525be578db127\n(cherry picked from commit 2b5853f4196e4e3725d1ab55ae7528c41b180a58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffcb1ff0016e7dde1d32dedad92091a132d0f88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffcb1ff0016e7dde1d32dedad92091a132d0f88d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
