)]}'
{"id":"openstack%2Fswift~838598","triplet_id":"openstack%2Fswift~stable%2Frocky~Idb67010de835dc2dbcbf87c40def934568a0dcf5","project":"openstack/swift","branch":"stable/rocky","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-20 03:30:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idb67010de835dc2dbcbf87c40def934568a0dcf5","subject":"Py3: Use collections.MutableMapping rather than UserDict.DictMixin","status":"ABANDONED","created":"2022-04-19 20:58:35.000000000","updated":"2022-04-20 03:30:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"306b7fe8a691374e66b4dca76c456fff0494674d","_number":838598,"virtual_id_number":838598,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-19 22:31:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 20:58:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2022-04-19 22:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f4011eb37905f25e582654562ebde774087852b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 20:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd9e61ea228df7711f741b468774a907860976c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 21:18:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19851cd129c34fe89fe3a4dae233795dc0ee621f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 22:31:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f2aee7ef08ac49dd9bcd3029f80b6403 : FAILURE in 4m 33s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bdc7c2d185f34574b5584e1eed1eb06f : SUCCESS in 14m 17s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/8d64a579f47b4fbfb7bbfde76a4a5e59 : SUCCESS in 5m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9fcafadd6b2748b684f7cc14e2297acc : SUCCESS in 10m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6a3ff4ef228475db3919d15583d886d : SUCCESS in 12m 32s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/f3ad5b6ea4804983b012a0f30e535900 : SUCCESS in 10m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/89511653ae054e988d01d39add387b99 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9af8e002749544cab2cca3c184932ea7 : SUCCESS in 14m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/49cf5bd0160641d2a1d4c7f675bd0bad : RETRY_LIMIT in 3m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce952a25308b41ceaf47b6afd547ff04 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed8f30f55d1e44f88f7d82994705877c : SUCCESS in 5m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68f768b60b70450584e021dd58777fb9 : SUCCESS in 25m 51s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/eafeb2ef3d2943b8ac7393ff0026b85f : SUCCESS in 1h 10m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/987554bb84ef44d0b5be0454d3f057e0 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":2},{"id":"306b7fe8a691374e66b4dca76c456fff0494674d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 03:30:46.000000000","message":"Abandoned\n\nIncluded as part of https://review.opendev.org/c/openstack/swift/+/838612","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99a4459a46822cfc0fc1ef2a33304bd4dbe7b0e8","revisions":{"9b6ae489f5a2191d44ffce41bcfee2fe7fd40595":{"kind":"REWORK","_number":1,"created":"2022-04-19 20:58:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/838598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/838598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/838598/1"}}},"commit":{"parents":[{"commit":"b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2018-12-01 17:34:08.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:58:27.000000000","tz":-420},"subject":"Py3: Use collections.MutableMapping rather than UserDict.DictMixin","message":"Py3: Use collections.MutableMapping rather than UserDict.DictMixin\n\nTaken from https://docs.python.org/2/library/userdict.html#UserDict.DictMixin:\n\n\"Starting with Python version 2.6, it is recommended to use\ncollections.MutableMapping instead of DictMixin.\"\n\nChange-Id: Idb67010de835dc2dbcbf87c40def934568a0dcf5\n(cherry picked from commit c3cbe69da893b6c72da6d7630a5cb7b7da2cc29b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b6ae489f5a2191d44ffce41bcfee2fe7fd40595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b6ae489f5a2191d44ffce41bcfee2fe7fd40595"}]},"branch":"refs/heads/stable/rocky"},"99a4459a46822cfc0fc1ef2a33304bd4dbe7b0e8":{"kind":"REWORK","_number":2,"created":"2022-04-19 21:18:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/838598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/838598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/838598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/838598/2"}}},"commit":{"parents":[{"commit":"b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2018-12-01 17:34:08.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 21:17:48.000000000","tz":-420},"subject":"Py3: Use collections.MutableMapping rather than UserDict.DictMixin","message":"Py3: Use collections.MutableMapping rather than UserDict.DictMixin\n\nTaken from https://docs.python.org/2/library/userdict.html#UserDict.DictMixin:\n\n\"Starting with Python version 2.6, it is recommended to use\ncollections.MutableMapping instead of DictMixin.\"\n\nChange-Id: Idb67010de835dc2dbcbf87c40def934568a0dcf5\n(cherry picked from commit c3cbe69da893b6c72da6d7630a5cb7b7da2cc29b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99a4459a46822cfc0fc1ef2a33304bd4dbe7b0e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99a4459a46822cfc0fc1ef2a33304bd4dbe7b0e8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
