)]}'
{"id":"openstack%2Fswift~838612","triplet_id":"openstack%2Fswift~stable%2Frocky~I56f10d8e25e57da59f3da57fe797f615433d70a7","project":"openstack/swift","branch":"stable/rocky","hashtags":[],"change_id":"I56f10d8e25e57da59f3da57fe797f615433d70a7","subject":"Fix docs-building on old stable branches","status":"MERGED","created":"2022-04-19 23:58:29.000000000","updated":"2022-04-20 04:16:17.000000000","submitted":"2022-04-20 04:16:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838612","meta_rev_id":"f0ad9ed2b7bb6858030203260d64b567a5f197bc","_number":838612,"virtual_id_number":838612,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 04:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-20 02:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-20 02:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 01:51:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d45abf8d9774910609eb3e1a5d618e94d6d890b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 23:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"163258f368cda465c7586a08bdbafc485a615299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-19 23:59:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b9bb644dd97c6e709ab687953da000be5880f1a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 00:25:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ccc5e2fc0db4e8d6386c2bff83dae587700b27ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 01:51:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9ba72e351e5043bbb000c5566f68fd08 : SUCCESS in 5m 42s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38e532e5734a40368e17764e3cd57e22 : SUCCESS in 9m 46s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/2c412cd5d5da43878296d29d9f2fdae8 : SUCCESS in 5m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ffe93c12d7e340189b232d71d853e306 : SUCCESS in 8m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76632b1d668447eab03125c758d06417 : SUCCESS in 8m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/df1e0e68ed664c538e07d061407b2d4b : SUCCESS in 8m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0428521e52f444a59b4714d007eb6475 : SUCCESS in 8m 45s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/44696fa7b09a40a6bec32b21c3a68b41 : SUCCESS in 12m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/708e84e6734a4f51bf038c07f1956fcd : RETRY_LIMIT in 3m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2df9b7a7bece4b6d9cb8d07242ea6725 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35ddd453e464739a5f650d88a51fca0 : SUCCESS in 5m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69f6387c13c347ab90a62437d491adbc : SUCCESS in 30m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/36b5b9174c274920b52009ad6d0032e4 : SUCCESS in 1h 20m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e515c9a4ec7b49aca28e650371915684 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":3},{"id":"292e2a7ee4d6bfe88c08d75de5128e0330df0383","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-20 02:47:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"75785d3c53048db2f1f2081f56ac75b6ae480ed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 02:50:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a69501206591eaddb4dc29538d0cb83723ff14f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 04:16:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ad27b4cfbeb54f8badc9f4c281e6fd7e : SUCCESS in 8m 29s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d869d77e4e54f9189dde06bab60524b : SUCCESS in 12m 10s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/dbeb89e07ee94e06828bcae8ac25bbce : SUCCESS in 7m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10a5330b77404ded880ad1a242f24c1e : SUCCESS in 10m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31abd69d78f648b39ba5cb6df0f940c8 : SUCCESS in 12m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/459953a4432b49a2a43830e8bc44d385 : SUCCESS in 12m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73ba731821c9424c94303cd258756c91 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/eab4af9bc2c142a38c91b1bdfff89263 : SUCCESS in 14m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4940141417a44d9da301777e419b6e28 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430761688b0e49a890eb989312670827 : SUCCESS in 7m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed51ed77a93f4492bd133c8d65dc431b : SUCCESS in 30m 42s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/61d06ff44d594e0290b2f98f97d1e17b : SUCCESS in 1h 24m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6e5117a0c8c44f28076307121bbd7d4 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f0ad9ed2b7bb6858030203260d64b567a5f197bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 04:16:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76dd373a3dd3d32b7bf8ae65f920c51a1f117ffd","revisions":{"9bf477c6df5733ecedd6773cbbb1df46982ace34":{"kind":"REWORK","_number":1,"created":"2022-04-19 23:58:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/838612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/838612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/838612/1"}}},"commit":{"parents":[{"commit":"b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:36:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 23:53:12.000000000","tz":-420},"subject":"Fix docs-building on old stable branches","message":"Fix docs-building on old stable branches\n\nNot sure why, but the docs build is running under py35 (even though we\nask in multiple places for it to run under py27).\n\nMostly just need to fix some imports for xprofile.\n\nRelated-Change: Ie8d28218b974a1b6b7b7b691f786ff1d6bdded05\n\nPick up another commit, as well, as s3api uses a funky removed-in-py3 mixin:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nPy3: Use collections.MutableMapping rather than UserDict.DictMixin\n\nTaken from https://docs.python.org/2/library/userdict.html#UserDict.DictMixin:\n\n\"Starting with Python version 2.6, it is recommended to use\ncollections.MutableMapping instead of DictMixin.\"\n\nRelated-Change: Idb67010de835dc2dbcbf87c40def934568a0dcf5\n(cherry picked from commit c3cbe69da893b6c72da6d7630a5cb7b7da2cc29b)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nChange-Id: I56f10d8e25e57da59f3da57fe797f615433d70a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bf477c6df5733ecedd6773cbbb1df46982ace34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bf477c6df5733ecedd6773cbbb1df46982ace34"}]},"branch":"refs/heads/stable/rocky"},"1dfd56316d3de9cdc748617eb91eba4a73aa1e1f":{"kind":"REWORK","_number":2,"created":"2022-04-19 23:59:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/838612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/838612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/838612/2"}}},"commit":{"parents":[{"commit":"b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:36:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 23:59:18.000000000","tz":-420},"subject":"Fix docs-building on old stable branches","message":"Fix docs-building on old stable branches\n\nNot sure why, but the docs build is running under py35 (even though we\nask in multiple places for it to run under py27).\n\nMostly just need to fix some imports for xprofile.\n\nRelated-Change: Ie8d28218b974a1b6b7b7b691f786ff1d6bdded05\n\nPick up another commit, as well, as s3api uses a funky removed-in-py3 mixin:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nPy3: Use collections.MutableMapping rather than UserDict.DictMixin\n\nTaken from https://docs.python.org/2/library/userdict.html#UserDict.DictMixin:\n\n\"Starting with Python version 2.6, it is recommended to use\ncollections.MutableMapping instead of DictMixin.\"\n\nRelated-Change: Idb67010de835dc2dbcbf87c40def934568a0dcf5\n(cherry picked from commit c3cbe69da893b6c72da6d7630a5cb7b7da2cc29b)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nChange-Id: I56f10d8e25e57da59f3da57fe797f615433d70a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dfd56316d3de9cdc748617eb91eba4a73aa1e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dfd56316d3de9cdc748617eb91eba4a73aa1e1f"}]},"branch":"refs/heads/stable/rocky"},"76dd373a3dd3d32b7bf8ae65f920c51a1f117ffd":{"kind":"REWORK","_number":3,"created":"2022-04-20 00:25:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/838612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/838612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/838612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/838612/3"}}},"commit":{"parents":[{"commit":"b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd","subject":"Fix cname_lookup test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4551a424b9db98d5fbb1b1c9621fac5ec9dd6fd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-19 20:36:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-20 00:24:23.000000000","tz":-420},"subject":"Fix docs-building on old stable branches","message":"Fix docs-building on old stable branches\n\nNot sure why, but the docs build is running under py35 (even though we\nask in multiple places for it to run under py27).\n\nMostly just need to fix some imports for xprofile.\n\nRelated-Change: Ie8d28218b974a1b6b7b7b691f786ff1d6bdded05\n\nThere\u0027s also something stupid about \"unexpected indent\" from an upstream\ncpython docstring.\n\nRelated-Change: I2a354e2624c763a68fcea7a6404e9c2fde30d631\n\nPick up another commit, as well, as s3api uses a funky removed-in-py3 mixin:\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nPy3: Use collections.MutableMapping rather than UserDict.DictMixin\n\nTaken from https://docs.python.org/2/library/userdict.html#UserDict.DictMixin:\n\n\"Starting with Python version 2.6, it is recommended to use\ncollections.MutableMapping instead of DictMixin.\"\n\nRelated-Change: Idb67010de835dc2dbcbf87c40def934568a0dcf5\n(cherry picked from commit c3cbe69da893b6c72da6d7630a5cb7b7da2cc29b)\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nChange-Id: I56f10d8e25e57da59f3da57fe797f615433d70a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76dd373a3dd3d32b7bf8ae65f920c51a1f117ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76dd373a3dd3d32b7bf8ae65f920c51a1f117ffd"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
