)]}'
{"id":"openstack%2Fswift~838773","triplet_id":"openstack%2Fswift~master~Idb0eaead9757887502a3ea13e39e9aad36fed62c","project":"openstack/swift","branch":"master","topic":"fix-logging","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-09-21 19:32:27.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-07-27 21:38:14.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-09-21 19:32:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idb0eaead9757887502a3ea13e39e9aad36fed62c","subject":"Fix py3 logging._lock patching","status":"ABANDONED","created":"2022-04-20 18:37:34.000000000","updated":"2022-09-21 19:32:27.000000000","total_comment_count":17,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"3dfd4b3a18bbf6a38734e2cf751ae37e871d43e1","_number":838773,"virtual_id_number":838773,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-02 20:34:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-05-02 19:43:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-20 18:54:55.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-04-20 20:08:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 19:43:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-07-27 19:58:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"9bab17d3cb10d3c046b37733c934243f78334c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-20 18:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aab695abe35a60f10beedf38f149f75bfb8df0dc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-04-20 18:54:55.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a6453c5f2d47a93c3bd8fcc3a53c6a57e8337a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 20:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb64d11ce7e4c88a1d90c1c116861b3 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7902cc33aa7c4555ab8149d5376ccb9d : SUCCESS in 1h 24m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab54651c94e343ea90f28583e9c6f760 : SUCCESS in 51m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2fa4d476567476ba8ad20adfc79bab2 : SUCCESS in 53m 55s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3b27bb37047d4fd09d721922afbf6c23 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6f12ea355cf64e3aa6870c1ef224e25e : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5468e02469064dbca2199ef1698b7669 : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17adc48c5e2a4ecf8e4153f61f044520 : SUCCESS in 8m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0bd412cc69b54dd89a959a65924c5feb : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e7605a017ab4ab99a719d2753a4a9d3 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f8b770763f24d5e991ec1043a1ec68c : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4ddbaa3ec14b417f9fff9073affd4304 : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5cd0f59872044912b625a9232d7bee1f : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3affa0b5d5cb4b039308ce11af34188d : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/34f150dd20974e63b69bad5cd1d84f10 : SUCCESS in 17m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1ca88151f0642059cb3274414355255 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a38976afa2774798b8da514d5250301b : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fc800843ad664fbb9a3bfe4fee7caf98 : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9fe8990f4f924cfb85df53a4a423a3b0 : SUCCESS in 8m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/199fe5fde6a042a983aca1d125c9fa1f : SUCCESS in 15m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4d25a321608248b7b49f386fd8397fa7 : SUCCESS in 54m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5933a3a1825f49fa87a50c3494d7002a : SUCCESS in 56m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad63a9f3132b4865aa55b3674d630504 : SUCCESS in 47m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8c8ab23e12e4fd197fd3e4c3bda9490 : SUCCESS in 43m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/374187a9f78145eabb00d84b3f4406ec : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e69efd8f7e1480e85fc74d575e196e6 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3b952a84adc474280d27547c21a4284 : SUCCESS in 22m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d273648e1bba484ea59c29c90f6fc1ae : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2e0ef045cf7eee6e0681dfae68714a97921cf223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 21:32:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b8a27e79aa24564a3838c8949d0ab38 : SUCCESS in 47m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/805d9c981eee4a60bb32933c6c150ebb : SUCCESS in 42m 44s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/163fd765e0a5434caddd269780b640f2 : FAILURE in 2h 53m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fcfe0a906c54388922e51c5a3c6172e : SUCCESS in 44m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e41b0c6991c4446b217f11690229432 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"83105b3a91d18e1179019e319bc16a06b449e3ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 11:52:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"044e60ece057876b466085d05251161dd5bee203","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 11:53:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"12fff3d74b2a8ccb059215bc5d2d1aeb2a2436ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 12:04:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f67b9018ba524dfda452f8b806a466e560f46dec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 12:15:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"201f9953324dc1a134c910527f64925d8d3f5ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:09:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849393d9711d412da56d63f5896458f0 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/88acd196bd384d89be2430b2fda895ea : SUCCESS in 54m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19d310f655b44e7688b5c30b236fd142 : SUCCESS in 43m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea5648261b4347f2b5780ea7feabc9be : SUCCESS in 41m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c42d04b454de4854a2c7217baf40b957 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d3489bbc82224b4ba13a7486c0014e03 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6313bf66465c49e685bee011021a2788 : SUCCESS in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ad577feb34f4f509c75f73377267046 : SUCCESS in 7m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3502372b9c64739a47712681337e7e3 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cfb29760d77243e58d4f245023a6fcff : TIMED_OUT in 1h 31m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73744279a5be43fcba19fd339026590f : TIMED_OUT in 1h 31m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9f31e7e5f37741878cea8299ff9276eb : TIMED_OUT in 1h 31m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3790c63377f472da81a9def5a6e662e : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f804ef27eb9141d09a5950d3d0bf1b86 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/018e61fa0c7345d2a932fa666ebfa7b7 : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9454db3df3ef4fccbc405d1bb05a5ba5 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c6a87a2ef614e21b1cc6f07a025af82 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e89e6befa51462788cfe90a9e066eab : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d17fc40b514c4eb5bbf7be280135ed4f : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23fd42ceee4746a6946ca37f686467db : SUCCESS in 15m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d9390ad581545a5af3b1e16b64eec0a : TIMED_OUT in 2h 01m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4bdb52b49e6543f4825984f8d69f2ad0 : SUCCESS in 1h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b39f3e32efd64c6e9ce0284ef004a939 : SUCCESS in 33m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4842879dea9f48289d8c7193c08bbcfb : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce229334b595473896dff1a63f351319 : TIMED_OUT in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3eb81357b246caae0472172c2d23c4 : FAILURE in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49ef7354d5f9472fbc1e151ea1fb3cbe : FAILURE in 42m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85799fa026049ccb3e96ba4982936ff : SUCCESS in 49m 36s","accounts_in_message":[],"_revision_number":3},{"id":"0006ce14bf4a3a3fac4701a10ded189e5a9853ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 15:02:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e5880bb0e3942b1b6510bfbcb82d8f1 : TIMED_OUT in 1h 31m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc079b69bfa44257aa596c031e2a4603 : TIMED_OUT in 1h 31m 29s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/a5b882c1e50e4ea0bee5a8a24fbe2104 : FAILURE in 2h 55m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ab72ef52e534a6dbfd40ab3ee06c701 : SUCCESS in 39m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e4412a038944b30ad270648b566e333 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":3},{"id":"7fd52e57ec6fae51d1e372f34d2ffdf6fa252677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 16:14:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3bd95e898b39c3604b8bb0c599452334aa113429","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 17:44:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c78c13ae4557029b20112ae56fd2b3567e7a4b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 17:46:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fcec970b01561e1d06c029b92f5a7f14815f5fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-21 17:49:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2ca90b871b4d46b4a25e8df57d503ecee52c7286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:51:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac7ac6b5993487dbb0e4d3564c62b8e : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/6bcea37951ab463b915cb599e2f043b4 : SUCCESS in 54m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c4ee25337b9e43eb98403e9afbaea7ab : SUCCESS in 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf174401ee924f98b794984ec6a55fd4 : SUCCESS in 53m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3734633539454ff38dedab8be6f9abb4 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2c0bd779606844b59b834c93260cafda : SUCCESS in 22m 46s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ffd60f46c26041afb523f390880cb59f : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a92fd0f31a0246ea8c4d324f9ee17e90 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6c3ccf65b3946d4bc6d0a55c859e038 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c00c9dfbef7e4a6493411c76744c480f : SUCCESS in 21m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35a541ad9e0e4e809e7d8bad11574530 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/def005e57cc34865ac302d3e29f9967b : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61f24b51b08042feb4b23f27016f4b65 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db5fac4cd872437cab665bbf47ba2575 : SUCCESS in 20m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1dbc0f9e8ad4bafb44138883c7c26ac : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/af1819cb8f8e49b0a9abcf2d0fb61a7d : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d24f5a4dfc5045a293ac6181597d466d : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dc1cd4e2f8f74cd5844f6ab31b5c8a04 : SUCCESS in 19m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/42adb953e2634ae898ca470bc4972ffe : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2ec5f32a28a459b8c263dfea09ac8de : SUCCESS in 15m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a15d1c4b94aa41ea859114bb9ff7f332 : SUCCESS in 48m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/aff6a25d290d448abc81e91ee776165a : FAILURE in 55m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a95618da352445ab12aa1db7ba50008 : SUCCESS in 44m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e760f3c58b0d4cf7a12e22a05e5791cc : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef4765ccc311450e9ede2bd672fb27f5 : SUCCESS in 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff630c9bf1f4711ab3bd05c0cc017a2 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aea896aa93ea42f29bed706c71c288af : SUCCESS in 28m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c2d7d8d66394db099fb2cbb7cd2d654 : SUCCESS in 50m 00s","accounts_in_message":[],"_revision_number":6},{"id":"fbddef14affd3a34bf145859f149d108d2498e63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 23:32:01.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0b19a2b7d8643d29e2aeaa89792795f : SUCCESS in 58m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/916c6e40c28c483fa23718f56798a978 : SUCCESS in 45m 17s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/51b4e19d13f049d281dec16a3753752f : TIMED_OUT in 3h 01m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f681f745ff44c8193d054e3c7cf0c22 : FAILURE in 1h 25m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da1946515c9d4c80b1876b91c63be234 : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":6},{"id":"e766ea903dbc3bb713ae46a90cc1df315dbb5ae1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-22 09:40:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a8ad62769055212713e9e4e8adf620af1bb1b7e5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-22 09:50:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"58ebf7158e29a179cf25443a1bd4f6507a16dafb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-22 09:56:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0bb94c0cb85089d986c7bd839a9e2b2a29fe75df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:57:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904acbaaf1a74278989d3333d78c3cdd : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3b9a6e8363b541c1a2e495f9cd8f4971 : SUCCESS in 57m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1515e12b3ef54981906f6a58fc0d0255 : SUCCESS in 45m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f91200731164c33a6ebd931635bf0b7 : SUCCESS in 47m 13s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6b79cd2ecf2f4a6ca7bc8d7625d9ac70 : SUCCESS in 22m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fb79870b9705409a8fa6b867907dac1f : SUCCESS in 23m 14s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a6f48d4430924fbd9bf2875d03885685 : SUCCESS in 32m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec428c00d0ae40f0807c31c13a2dbd59 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57d7d8322c5a4363b189d2613398c510 : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d07a234d9b24392a890025660aa7288 : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64b424d2395e46ebb5ec5ee09b2210e0 : SUCCESS in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68e72eb41423433fbb589dc1cf9a12c3 : SUCCESS in 15m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0bd29a91d1fb4e9abef6c15238d44484 : SUCCESS in 22m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d815bee07a1147a796b01c889930b599 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4abcebafe6a744199fb7de2e698fd8ad : SUCCESS in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f5febb3f4e4493199ae7eea43c1f64d : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3da6213bf4e14246a2a3e07c771936f4 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3bebb1d38bcd4acbaf41be59fa1996be : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a1012026a924114b1c79b86093bd6e1 : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ed3b87e1bda44059982e11ff11f39df : SUCCESS in 15m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a57a280162124c7492f5a18bc4bc840c : SUCCESS in 56m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d2f2e972225483fbd885f42b2a4ae64 : SUCCESS in 51m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96a20aaba9a34d6d99cbc61388f03d49 : SUCCESS in 39m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9eb6068593ac49169653c98ba044a685 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83dedee43d1040d5a52f748ad86e4820 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49288783b4e14331b0f0db00a55093d7 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2151c6d19234421398f4fbe9a1bfe89b : SUCCESS in 25m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91f5902b22014a51a3bf438f410a55d8 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":7},{"id":"562f590e72ea1d4cd991d980f939e46eb1a62498","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 13:02:45.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a29826f59c95470ea0af1d969e7e0923 : SUCCESS in 1h 17m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a68df7411354435ba13feba017c2f56e : SUCCESS in 1h 22m 21s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/81b3a0ca19db4374ab4620ff2b2ca080 : TIMED_OUT in 3h 01m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bbcaa7669c842c486b7eee6d1ce7f56 : SUCCESS in 1h 16m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a2df229e29b4d6692ddcbd0e0a62abb : SUCCESS in 1h 15m 50s","accounts_in_message":[],"_revision_number":7},{"id":"ccf5474949213ca2fa6f1f7d008a4728d62390c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 19:43:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"77c10bdfa4a62fb7c3d36f55f7166ade8c30fe2c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 19:43:30.000000000","message":"Patch Set 8: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a4c372b954d0c01c20494c1e34864a136adba9fc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 19:52:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d8775ed94afb2976cb60e6dc5fba2b5cdbeb3881","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 20:32:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acb67c5997c54cdb80a69a746df57014 : SUCCESS in 46m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/496738ff379b430186d5892d4a3f0ca8 : SUCCESS in 39m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fd469eea38b4db586b733e6308604a7 : SUCCESS in 42m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe6f26fe2c3246b0bcc770bee4eeb110 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":8},{"id":"8eddedd2ad5dfd3bcee30d17b998eaa027d0d0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 20:34:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a8f72a2f74492f9108990f5a5221c2 : SUCCESS in 7m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0714e65f9b4e41c28bfbac0483574f75 : SUCCESS in 41m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3cb8f13043884c98912479bdafb00543 : SUCCESS in 41m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8142d8b2262b4a13b90174a045112b27 : SUCCESS in 37m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/326659881ede4fd88f574328218dafc7 : SUCCESS in 39m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/556bad89dc8a41399a5360b9bd0ea23f : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dc62ee8f157844ae80306c55fb332212 : SUCCESS in 22m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/026505978dfb4bcfa99dd278abf07150 : SUCCESS in 6m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bdb6d3daf5f9446290fdf0a9fe60ef49 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a3586d37da04fd19a8b3ebcfb81b7e7 : SUCCESS in 15m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b20738ac0a24d608e08a459c582b518 : SUCCESS in 17m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36872fbdad8a4efdb7b1c8b4acf7c6c1 : SUCCESS in 12m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/231556388da44bd5925e98470579aaa9 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1d76fa029dab4416a15920a836e115c2 : SUCCESS in 21m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a90ff10c10a342bf825badd314438d08 : SUCCESS in 37m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/67396e4e40684c7db63a8d6466a00dc7 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5a36c31693ca4c40abd73b89796c9d59 : FAILURE in 3m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3857539ac9747f592853fbfd7900148 : FAILURE in 3m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82f8bd9fce6d47ac99ea252c72d0896f : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7365af73f9b94871b728b950f652020f : SUCCESS in 13m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d310378057de4310b3ea8c5388fe4376 : SUCCESS in 40m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0008f052f6634724b780a93f362ebc0d : SUCCESS in 49m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6306b8161a54449d839f603feffba7c3 : SUCCESS in 30m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8564ee293f444a8a869cb3c41ade8a49 : SUCCESS in 41m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6ad85e91362458a9dd26a7dde4ce06e : SUCCESS in 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c77d098aead450f8418d48aa375e595 : FAILURE in 3m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a27b4ad7b9f4b9e8df0415eb0a9e030 : SUCCESS in 15m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf1a178644eb432392ce4e71e9af6fa3 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":8},{"id":"4d6fc54645272bb82fe80f4dd538106b51335d48","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 21:38:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c3c8566ea1bd51e7e6c032def5368d4296e3083d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-27 19:58:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cea7d10497595a828520a4ded30ca0f9a80c03ce","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-07-27 21:38:14.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3dfd4b3a18bbf6a38734e2cf751ae37e871d43e1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-21 19:32:27.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/457110","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"476986654eb3248b25303ab4c2d9fda1bdeee75d","revisions":{"e1b42ce6ebecd4e52ff5ab314154d44ad7581990":{"kind":"REWORK","_number":1,"created":"2022-04-20 18:37:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/1"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nI\u0027m not sure that the existing logging._lock is in fact patched by\neventlet, unless the logging module is reloaded. The existing lock is\nnot tracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b42ce6ebecd4e52ff5ab314154d44ad7581990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1b42ce6ebecd4e52ff5ab314154d44ad7581990"}]},"branch":"refs/heads/master"},"0dbe7d73db1f4924696906118c185c63e2b1618a":{"kind":"REWORK","_number":2,"created":"2022-04-21 11:52:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/2"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-21 11:37:42.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dbe7d73db1f4924696906118c185c63e2b1618a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dbe7d73db1f4924696906118c185c63e2b1618a"}]},"branch":"refs/heads/master"},"5acd758c414be4fdc010a916460e128915e6ca33":{"kind":"REWORK","_number":3,"created":"2022-04-21 12:04:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/3"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-21 12:04:28.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5acd758c414be4fdc010a916460e128915e6ca33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5acd758c414be4fdc010a916460e128915e6ca33"}]},"branch":"refs/heads/master"},"72983afd0447835438d7d08724c2d00de107a7ee":{"kind":"REWORK","_number":4,"created":"2022-04-21 16:14:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/4"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-21 16:14:21.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72983afd0447835438d7d08724c2d00de107a7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72983afd0447835438d7d08724c2d00de107a7ee"}]},"branch":"refs/heads/master"},"c981d63033ce773cb6b5ad53df4ab5a7b76fc0d9":{"kind":"REWORK","_number":5,"created":"2022-04-21 17:46:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/5"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-21 17:39:53.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c981d63033ce773cb6b5ad53df4ab5a7b76fc0d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c981d63033ce773cb6b5ad53df4ab5a7b76fc0d9"}]},"branch":"refs/heads/master"},"3981b0cb097dc8c466908a398d982e1ad85c824f":{"kind":"REWORK","_number":6,"created":"2022-04-21 17:49:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/6"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-21 17:49:40.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3981b0cb097dc8c466908a398d982e1ad85c824f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3981b0cb097dc8c466908a398d982e1ad85c824f"}]},"branch":"refs/heads/master"},"3a7ffc9614187b9ab4d01988b356aff23b8edf14":{"kind":"REWORK","_number":7,"created":"2022-04-22 09:56:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/73/838773/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/7"}}},"commit":{"parents":[{"commit":"e2b2a2c3062fa78236bdda868542b8b9923e726c","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2b2a2c3062fa78236bdda868542b8b9923e726c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-22 09:55:45.000000000","tz":60},"subject":"DNM: verify logging._lock patching","message":"DNM: verify logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7ffc9614187b9ab4d01988b356aff23b8edf14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7ffc9614187b9ab4d01988b356aff23b8edf14"}]},"branch":"refs/heads/master"},"476986654eb3248b25303ab4c2d9fda1bdeee75d":{"kind":"REWORK","_number":8,"created":"2022-05-02 19:43:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/73/838773/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/838773/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/838773/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/838773/8"}}},"commit":{"parents":[{"commit":"d09faee2627ad689c223d7142c806211fdd22af0","subject":"Don\u0027t monkey patch logging on import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09faee2627ad689c223d7142c806211fdd22af0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-04-20 18:35:39.000000000","tz":60},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-02 19:42:35.000000000","tz":-300},"subject":"Fix py3 logging._lock patching","message":"Fix py3 logging._lock patching\n\nWith py3, the existing logging._lock is not patched by eventlet,\nunless the logging module is reloaded. The existing lock is not\ntracked by the gc so would not be found by eventlet\u0027s\ngreen_existing_locks().\n\nChange-Id: Idb0eaead9757887502a3ea13e39e9aad36fed62c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/476986654eb3248b25303ab4c2d9fda1bdeee75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/476986654eb3248b25303ab4c2d9fda1bdeee75d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
