)]}'
{"id":"openstack%2Fswift~838811","triplet_id":"openstack%2Fswift~stable%2Fussuri~Ibe514a7ab22d475517b1efc50de676f47d741a4c","project":"openstack/swift","branch":"stable/ussuri","hashtags":[],"change_id":"Ibe514a7ab22d475517b1efc50de676f47d741a4c","subject":"s3api: Use constant-time string comparisons in check_signature","status":"MERGED","created":"2022-04-21 04:33:37.000000000","updated":"2022-04-21 16:35:13.000000000","submitted":"2022-04-21 16:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838811","meta_rev_id":"9bb9ca595038a7f5849269fd5dec244485a09aeb","_number":838811,"virtual_id_number":838811,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 16:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-21 14:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-21 14:53:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 05:52:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86417b350a52956852e8cba3df4ef716c94fbdff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-21 04:33:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71ccbcd6251fd4b1341dd5f850a7c93a3df11079","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 05:52:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1838b13ac044a7ae6b70fee07cc8c0 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/358d3fca949543658444f12277a26d2b : SUCCESS in 1h 14m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61e111aba8074aa1b6af953dc8e80247 : SUCCESS in 45m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3df7c43a10b4fcababd1fe183b271a2 : SUCCESS in 56m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e78e4bc11e424554b3ba916f99413048 : FAILURE in 7m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d98ce8cee303440d8fe167f8c4abecf6 : FAILURE in 4m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c79bb3b473c54b99b1e420d4c85fe96d : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7ccddc0f774a87a2bed10c44b9073c : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6dfd3b8c6ff41f2b19b8fea9266acfe : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3df7d7576f9c46ba9f6992c9312d5249 : SUCCESS in 14m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a47959bf0e9b4128a8140554bae07f26 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cbca144727f4f80b8471d912edd8d29 : SUCCESS in 16m 11s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/ae49cf8d562a42eba38a7f029152c5a1 : SUCCESS in 9m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6880756a369a4298b04bac0e6955fbd5 : SUCCESS in 13m 03s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/20e77df2f8224dc6b18c96786b55a1b0 : SUCCESS in 20m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f0d59a579c2444b7a0baff8dbf6707a4 : SUCCESS in 9m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/49f94c1d6a3045be8d482aa15ad3b18f : SUCCESS in 16m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a97b4928fbb246948d41cde909e074e9 : SUCCESS in 13m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/137f9344b72d428689014b6e2cf166a0 : SUCCESS in 10m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8917262d21984877872ef3b731ffe0ab : SUCCESS in 21m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cdc90222a48c482a89527a0b22e0efe1 : FAILURE in 7m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/09f82cefd0be4ffc918797d0829dca36 : SUCCESS in 41m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0a3f35552b74f15af1a9efa7a75045c : SUCCESS in 28m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a9a48e49aa0f4dd39cffd6cffa6e57e6 : SUCCESS in 28m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b118c7c2d45c4fae851cddf5026da337 : SUCCESS in 33m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3cb293c920a421a9e935b06ce1f9647 : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca183cf73d142b0b13415002751d5da : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/178e4f98400643eb94ed3991accc0f43 : FAILURE in 7m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4f7d803392047deb9ca50c34c6a75b1 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7e8f39f7b7ab0287be5c29b3a67df3817dc37d32","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-21 14:53:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71ce7b9db8fc12cc33b5426668e180d7a5a3b854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6d9500e17732736bee731fac5613b158b2c6b2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414ebe83722447199fd5aecc24d6df1a : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/6fed3758272c430f9ee7ef6564320014 : SUCCESS in 1h 07m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ad36b5193464ad0b6d29b1a521d9be9 : SUCCESS in 46m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/279c777f96b14100ba070ecf500e4f9c : SUCCESS in 57m 52s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7399e24e380e4d22919945fd1300a03e : POST_FAILURE in 9m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/59b487d4cad04071993ac6bf08f8c4ee : POST_FAILURE in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a176904c4934c9eb2706136e2c8f774 : SUCCESS in 21m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/edb50b0362e649bfa597097ba7060678 : SUCCESS in 16m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/97031a64d3d54fec96db3fa0f85ff116 : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/78cb132bcbef426da4712281d490c3a5 : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b9a67d758b2446cbd64225963c810e5 : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59c47a9ed4e94e1e9a93b4188b6c6b2b : SUCCESS in 16m 21s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/02edf7bd329f4d78b759ca3e3a1c0ac9 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7d04ae904e34710b014019ae307f56f : SUCCESS in 14m 58s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/2f2c87ea271e45e49c280503407a00c0 : SUCCESS in 15m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4dd77ae4cf264e3ba2f596132cd2b0ee : SUCCESS in 13m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/54c90050c1704d478ef8ebe05c65c4ee : SUCCESS in 10m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bf06d2a2d9004dc7ad202dc62683d817 : SUCCESS in 13m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6c160e9a30a649f2be4fd43df7d8dcb7 : SUCCESS in 15m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3183b2293d5047b0b0bd75996c383bc9 : SUCCESS in 47m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b22bc682eaf5426e9ebb24ba6c049377 : SUCCESS in 34m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/30492771d4c0485abcc3f96d1a9073fc : SUCCESS in 32m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af74d476506f4adeb16fd2c64aeeec51 : SUCCESS in 30m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fce5eba64774b0d9cf5fb83837679ee : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67fb64fe5bd448b1acf634064a856266 : SUCCESS in 6m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bdeb2ab728d453f98f91a8cfd967abc : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bfcdf2c8b040dcdddaeb3d0d2f335a83c7a40f42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9bb9ca595038a7f5849269fd5dec244485a09aeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:35:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11b4e9a19854bd8b46b0721a87baf01 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9b1679a1e13b424a91d4acccd005fdc9 : FAILURE in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1612132a477c49d1996d781d2ac7d498 : FAILURE in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"430bb7f5681f80ad96c7d7aca2b40023508d6843","revisions":{"430bb7f5681f80ad96c7d7aca2b40023508d6843":{"kind":"REWORK","_number":1,"created":"2022-04-21 04:33:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/838811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/838811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/838811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/838811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/838811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/838811/1"}}},"commit":{"parents":[{"commit":"ffcb1ff0016e7dde1d32dedad92091a132d0f88d","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffcb1ff0016e7dde1d32dedad92091a132d0f88d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:31:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-21 04:28:18.000000000","tz":-420},"subject":"s3api: Use constant-time string comparisons in check_signature","message":"s3api: Use constant-time string comparisons in check_signature\n\nChange-Id: Ibe514a7ab22d475517b1efc50de676f47d741a4c\n(cherry picked from commit 6142ce88cc71037ba0cd23113eb6082fa91346ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/430bb7f5681f80ad96c7d7aca2b40023508d6843"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
