)]}'
{"id":"openstack%2Fswift~838820","triplet_id":"openstack%2Fswift~stable%2Ftrain~Ibe514a7ab22d475517b1efc50de676f47d741a4c","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"Ibe514a7ab22d475517b1efc50de676f47d741a4c","subject":"s3api: Use constant-time string comparisons in check_signature","status":"MERGED","created":"2022-04-21 05:41:47.000000000","updated":"2022-04-21 16:35:17.000000000","submitted":"2022-04-21 16:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838820","meta_rev_id":"402be9df484328c95aaf02e221e773a49c0c0d4d","_number":838820,"virtual_id_number":838820,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 16:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-21 14:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-21 14:53:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 06:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4eac5a3d8b678000791f789707cce899fe273f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-21 05:41:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e0d1b810144f1d58404226c5d90d0c4686b27e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 06:47:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6866ced156e4c1497bda005a86a9eb7 : SUCCESS in 5m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/7a7c2f156b7640fe8e4f1bbb140efb85 : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6124b739077a4944b66868cae5d9d3b0 : SUCCESS in 45m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9078789350c24f258695fed0896547ec : SUCCESS in 58m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1f4fd9be613491e85a3a83c0192fea6 : FAILURE in 7m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9616cfb7561f4975b6b88c6f072d62a6 : SUCCESS in 12m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac4c904d1d984bbe8cf1ce538db6c651 : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3eab33f1ded249308389284d1eadd247 : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b829f31cd53045e2b1bc2066907d4b0b : SUCCESS in 10m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b2b631abed04d3ab49ca17e1aa8b80f : SUCCESS in 10m 14s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bf54ab0c3bb748c3b76131ce458c886d : SUCCESS in 9m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1cfc86e8e01a4cef978b3cca3f43b89d : SUCCESS in 10m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/9f26736bc11f481e80fc6929c7873e5f : SUCCESS in 13m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/04c89b4efeff43fe9a90e4702d803473 : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bb7ae8213a8f4c299d2b220a841a19b3 : SUCCESS in 11m 00s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7ca74cb73dd647bdad6d452725feffee : SUCCESS in 9m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7e70aee4a0454c5fa4c0d59ae876c446 : SUCCESS in 10m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4b35264f5764435da81c17d903ab52d6 : SUCCESS in 13m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5444e1a4fd64426380ca75828b20dd2b : FAILURE in 7m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/148b55255a534eb6b5c6b1e64f9fc7fa : SUCCESS in 42m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2d417e0a2b0439ba041ca4fa52f2d8c : SUCCESS in 31m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9d3e7c8ac0b842029d3d02a59ca51bc8 : SUCCESS in 35m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea5134b2f6cf4ce694921a37c1330496 : SUCCESS in 35m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba178dd3f68a4cb3846ac14b3f776b0e : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c06f11a45e40c38e02ba530af6aadd : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/398d3a35aac541c0a80eb38832735e71 : FAILURE in 7m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a955c6d74cf446a599d2aca3e6f8e307 : SUCCESS in 58m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2adb8307077a4ca2bd971a3013417122 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"247ede24d11439a8929e3d00586bc13c3c039780","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-21 14:53:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18e8de4d342f40d0c2f59023a781cf45523abd8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:54:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"565d8105ffa6b6da530cef5948a16264d04074b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62cdac0261da488183b9b5b5c404030e : SUCCESS in 6m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/cbcdf1db856342fbb3596486bc4e03cc : SUCCESS in 1h 02m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/537ac049d40d4d5b958ba019318deda7 : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4847c9295ee4ed285f0a7a9b7ae6792 : SUCCESS in 54m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e27afeba57784bed980be1816180fc87 : POST_FAILURE in 10m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5fe2598f9f1412388904b6ef26e5edf : SUCCESS in 12m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db9c333282e2408195b836cd1021fc50 : SUCCESS in 14m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7b4940a368948f4bbbe4c115c659f72 : SUCCESS in 16m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10346ff7f34e48069690fcc227a0b74c : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3fb9f75ced914ac59552ff0bf01a9a4c : SUCCESS in 9m 59s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/1a03f1b97b1a48f78c6a248ef1ee5484 : SUCCESS in 10m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe1e123eadf14d79bf493ddf59883972 : SUCCESS in 10m 21s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d6bc18c232af4c66a546f87d7e5934d4 : SUCCESS in 12m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6cc38dee90f040d198134c2849220e95 : SUCCESS in 10m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3ef1b42d295f4fc89c743a9b73c1cda1 : SUCCESS in 10m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1878236d66054c38a10ed14918d52477 : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/fafe9884c4d34699b5744f17a72d137d : SUCCESS in 20m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/720eca50894d4d86b4de936f5053ed14 : SUCCESS in 40m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c88c34ec3deb452d839369f9b62e69ba : SUCCESS in 34m 06s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/438b8a5e00ba456d83a174406d30e686 : SUCCESS in 34m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95ae20967c424b3a9c48ebab3f72c75f : SUCCESS in 33m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a97a533668f4d8dbada02c220b542e4 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3b7caac2dc4227ad87d3e72c0c2f84 : SUCCESS in 6m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b512b36055a34df6bfb79171d45f6dea : SUCCESS in 57m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b056872a33c54c1f9b60db15b331dcb9 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d8918da270da15b321889ce2f06655521a656bfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:33:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"402be9df484328c95aaf02e221e773a49c0c0d4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:35:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb5f8ea97d84b27b8450f50b7aac991 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/89b6525a4b1548d98003f74acf3159f4 : FAILURE in 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2829194bbf78058f1543819cad63fe596d0e358f","revisions":{"2829194bbf78058f1543819cad63fe596d0e358f":{"kind":"REWORK","_number":1,"created":"2022-04-21 05:41:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/20/838820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/838820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/838820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/838820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/838820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/838820/1"}}},"commit":{"parents":[{"commit":"c0fb627c4a83e668a532276fd5437c0c6f8f390d","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0fb627c4a83e668a532276fd5437c0c6f8f390d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-13 22:31:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-21 05:41:40.000000000","tz":-420},"subject":"s3api: Use constant-time string comparisons in check_signature","message":"s3api: Use constant-time string comparisons in check_signature\n\nChange-Id: Ibe514a7ab22d475517b1efc50de676f47d741a4c\n(cherry picked from commit 6142ce88cc71037ba0cd23113eb6082fa91346ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2829194bbf78058f1543819cad63fe596d0e358f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2829194bbf78058f1543819cad63fe596d0e358f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
