)]}'
{"id":"openstack%2Fswift~839268","triplet_id":"openstack%2Fswift~master~I951242ca76a9d1fd8bf4a2d885d0e260dd425788","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-25 22:44:30.000000000","reason":"Clay Gerrard replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I951242ca76a9d1fd8bf4a2d885d0e260dd425788","subject":"logging: Check for BadStatusLine before OSError","status":"NEW","created":"2022-04-25 22:33:10.000000000","updated":"2022-04-26 01:33:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"bc05b05e14e249047112c104e6a3733663725e9f","_number":839268,"virtual_id_number":839268,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-25 23:49:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-25 22:44:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-04-25 23:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5be5c53740b758722668c79a7033d9a2550bfab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-25 22:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e766a0bc3733f4852ede097124252332a1041bcd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-04-25 22:44:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1f3cb043ce2183f6c3fa6a33cf3055c2bafe2ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 23:49:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb04127152a3458ea372da705b1b041d : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/f4f1506c7a504bcdb73a53d3815d7e50 : SUCCESS in 1h 05m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ceb63313a4c4f2aadf5414c945cf37e : SUCCESS in 52m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a520341dbd1a42ab9edeac0bf0f9b18e : SUCCESS in 58m 07s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/819911d6d4264e3085a04af4ab1547ac : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fe46f939cabc4e9ea9ff9e8cd7ee5b44 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7de3114ef6ef4f2c8a1b65020c8fcac7 : SUCCESS in 23m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a12de383b1f46e59b7cfd5e597ea794 : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e533002081184734b0ae681abf0e3b3c : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d2ced2685164929b772df864d23487a : SUCCESS in 28m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6601b7498c4946dfb50dc24260dc793f : SUCCESS in 20m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6fa5fd837a9547ec9d4ff82067271948 : SUCCESS in 14m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28ae16759e004f5bb3e4ce39cd908dba : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/84b3505780404c368945040588498f4e : SUCCESS in 31m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/689bc4f0de1c4d21bc484385e168581d : SUCCESS in 16m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8820892d0b5844bbb545be47fb89dc27 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f051403b22f3481eaf1125233110a20f : SUCCESS in 29m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67af11444b0947e18a3292448e4316b6 : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/134030ceabd941bfb1b59bc8a2771b5c : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fa1501867e3b45349b15bc1898bd2c96 : SUCCESS in 14m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3dc794b260534049805e6c6ffd90eb5d : SUCCESS in 1h 14m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b84d3eb24aef45dea32a00a8d099dc14 : SUCCESS in 1h 00m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8b302fa06634f22aef84c9a12a5d513 : SUCCESS in 41m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09647512545b419e8c7f353ddc69612a : SUCCESS in 38m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2c8354033bd411981f9a2c2ecd8f499 : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec238894413442818b23eee1b3523e5b : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a25aed0ebf514b69aacf929b10f5fdfc : SUCCESS in 31m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5909b5c19c4042268aaebcedb3bb4a83 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"bc05b05e14e249047112c104e6a3733663725e9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 01:33:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41a27d79aee54719965a078fb84fed15 : SUCCESS in 52m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf8103dc745146a8849d1e97fb9e53e7 : SUCCESS in 43m 36s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c750784be22b4327ae809b1274e25801 : FAILURE in 2h 58m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de63b9c2622d45d7a7f045b063222bdd : SUCCESS in 1h 28m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8151ba32832448d2a1a4cd471ff98e8f : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3536be46d390bb54731d7e5fda5eefc1329ff65","revisions":{"a3536be46d390bb54731d7e5fda5eefc1329ff65":{"kind":"REWORK","_number":1,"created":"2022-04-25 22:33:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/839268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/839268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/839268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/839268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/839268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/839268/1"}}},"commit":{"parents":[{"commit":"0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a","subject":"ceph tests: Register output/ceph-s3-summary.log as a job output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bf5474bfa2f41a07e7e3d8c68419b9dbd0f816a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-25 22:31:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-25 22:32:03.000000000","tz":-420},"subject":"logging: Check for BadStatusLine before OSError","message":"logging: Check for BadStatusLine before OSError\n\nOtherwise you find yourself logging a traceback for a RemoteDisconnected\nerror (which inherits from *both*), when it really doesn\u0027t warrant one.\n\nChange-Id: I951242ca76a9d1fd8bf4a2d885d0e260dd425788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3536be46d390bb54731d7e5fda5eefc1329ff65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a3536be46d390bb54731d7e5fda5eefc1329ff65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
